)]}'
{"id":"openstack%2Fglance~795262","triplet_id":"openstack%2Fglance~master~I48ce88f4f398e05818457b00cfc8dda90f3c9c91","project":"openstack/glance","branch":"master","topic":"policy-quotes","hashtags":[],"change_id":"I48ce88f4f398e05818457b00cfc8dda90f3c9c91","subject":"Fix generated policies and samples.","status":"ABANDONED","created":"2021-06-08 09:29:36.000000000","updated":"2022-04-22 14:40:27.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"0b202f4d8bff6870405f3c4e0e9291cb218e7da1","_number":795262,"virtual_id_number":795262,"owner":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2021-06-08 18:54:08.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-06-10 09:57:39.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]}],"REVIEWER":[{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-08 10:51:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-08 11:21:33.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-06-10 09:57:39.000000000","updated_by":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"reviewer":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"state":"REVIEWER"}],"messages":[{"id":"84fb48cf7b86ff851ab70efb222d2b6ef97edc03","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"date":"2021-06-08 09:29:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3f9206c0b350d79daacad483e715434304df893a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-08 10:51:04.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/2e398bf3304843668e8e8b99c52f6bf6 : FAILURE in 33m 53s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d9bf1014a1844c4a8ff3dbc470dde3d9 : FAILURE in 22m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e74d86f260364c0583bfa849292923d3 : FAILURE in 6m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd3f037cc2f046aebad31382318cdf5b : SUCCESS in 5m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fe5fc56b6fe6418ab53ed4c3480f7c61 : FAILURE in 6m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e0e6795df804467c9ab584f44d96338b : FAILURE in 6m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1e51df0037b74fa1ad840c971c05260e : FAILURE in 9m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c73c773b699848e5b48cc642c108d02e : SUCCESS in 5m 56s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7d537dfcbe3c41a4b39a71d3913152b3 : SUCCESS in 8m 12s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/ddbb26b7650a409badcb8a8a9e38a158 : SUCCESS in 8m 13s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/6c7118257f9e4807aec517b41ea4907b : SUCCESS in 7m 38s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/bffad2947fb74a1c8328b65ffa8372af : FAILURE in 26m 57s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/02e6aef86d224fa0b021f2d49d7d433f : POST_FAILURE in 16m 14s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/e3d825ff088d4464995b5e33d12c2c83 : SUCCESS in 1h 11m 42s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/f1ab75076c854ad1ba62bdd5edf5fa4d : FAILURE in 24m 54s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/69049b2da2644231884292fe3887b4cc : SUCCESS in 49m 42s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/6ecedd9ba87446b495d6161e5005d1e8 : FAILURE in 1h 20m 13s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/0d6c75bf3eff4e1d829c112063285b33 : SUCCESS in 29m 39s","accounts_in_message":[],"_revision_number":1},{"id":"cbff9e5fbb9bed603321778f6ebe98834b0c52c4","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-06-08 11:21:33.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/795262/795262/1/check-not-vote/ext-glance-zuul/c67ceca : FAILURE in 17m 38s","accounts_in_message":[],"_revision_number":1},{"id":"14bf9ff7a4218c2bc2f272d3d29023e370b2e260","author":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"date":"2021-06-08 11:51:09.000000000","message":"Patch Set 1:\n\nrecheck\n\nCDN outage is over","accounts_in_message":[],"_revision_number":1},{"id":"a4ee6a620415fe84f5ff339e39e1136782b1752f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-08 13:13:55.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/3174ad6670cc402e9e2cfde989ae993b : SUCCESS in 58m 13s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9f0df4e664944a8789df926d3edfe099 : SUCCESS in 1h 12m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a9d06e3ef7804a2ebbeced5e8beecc1f : FAILURE in 7m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ee0a126837146faa055c9681da5e42c : SUCCESS in 5m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e366263e12d54776ad1a57ec3726379d : FAILURE in 6m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5cffdf538b8e436c8fb1793473505797 : FAILURE in 6m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9071209046bf47cdb4f1545ebe9c0ae1 : FAILURE in 6m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d831d82c8ad24101831811c36e521030 : SUCCESS in 6m 54s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8372ac09836844aca6b9b93f4a6a9701 : SUCCESS in 10m 20s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/76d97979cf344e8c9f2c7b608d0d1ba8 : SUCCESS in 7m 30s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/6a3f6f02939a4c479efff27c474ce056 : SUCCESS in 5m 54s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/85ddcdbc848c440695da22e367b5d125 : SUCCESS in 1h 19m 57s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/f7d4e8f352764f0fa79233edc36a6015 : SUCCESS in 1h 13m 51s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/d71e0b3c55e34100a3f5bdf5a1480c7e : SUCCESS in 1h 20m 04s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/7fa0753b981d409cb81a9949f129b9fa : SUCCESS in 1h 18m 39s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/74926e56cf1d40838ac947fecc150c15 : SUCCESS in 55m 05s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/20e05a20e79240c1a9d42dfc21cd482f : SUCCESS in 1h 16m 06s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/d6dd700073e743309fcb0db93d62060d : SUCCESS in 28m 27s","accounts_in_message":[],"_revision_number":1},{"id":"7d09fb881b37c4a668d89f013268fc5ef3b20e45","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-06-08 14:45:33.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/795262/795262/1/check-not-vote/ext-glance-zuul/d350024 : FAILURE in 16m 34s","accounts_in_message":[],"_revision_number":1},{"id":"6190df48595d40b64756de557e583797b13ec49c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"date":"2021-06-08 17:17:42.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7427c18a495d05e7c57ba0864149033319f523e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-08 18:54:08.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/dbf6a06e1fad4af5b1ad9dfe8e5b9d0f : SUCCESS in 1h 01m 17s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a3bc2c13cc074687a3074057551343f9 : SUCCESS in 1h 16m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/76f93a547f4543fa8dc45355e0589f50 : SUCCESS in 7m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/02386e7ba55a44c0aa3a086831ea7891 : SUCCESS in 5m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/92ce4e0a46aa4aa0bde2ee45b9cb5170 : SUCCESS in 6m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d4661c7db0cf492d9888418230f77430 : SUCCESS in 5m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fd97dda09724477db4b309b1f3765d97 : SUCCESS in 6m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d88a6a90189b4dbd93f83c5989187bad : SUCCESS in 6m 43s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e73b1bca015441e5a0a35216802bdc33 : SUCCESS in 8m 42s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/1aba8b34ac374d65b621cd3b13b848ed : SUCCESS in 7m 55s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/ad5ab8d30e9d4bfd97e9b1b483e1a5e0 : SUCCESS in 5m 34s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/4711cb97248747d7b3dc902d3fdaaefd : SUCCESS in 1h 32m 07s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/989ac88b8eeb49e281d03f0e7b01f214 : SUCCESS in 1h 14m 14s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/352a76ad2bee420f9bdd0506e5a7e055 : SUCCESS in 1h 31m 36s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/05814e4d29264c5b864ddccae31035ea : SUCCESS in 1h 17m 17s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/af201686757e4a07bdb4ad4dc62923d0 : SUCCESS in 53m 30s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/627d0a380bcd44308b113f964a9dbc22 : SUCCESS in 1h 24m 21s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/ba1d09b15118492a85c69b191fd9e2e3 : SUCCESS in 32m 27s","accounts_in_message":[],"_revision_number":2},{"id":"00c0005485b189a04903fcd6d217c0be0d2dec80","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-06-08 20:46:17.000000000","message":"Patch Set 2:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/795262/795262/2/check-not-vote/ext-glance-zuul/44961fb : FAILURE in 17m 30s","accounts_in_message":[],"_revision_number":2},{"id":"bbce8f04421e210b75c7af00da2782867c489962","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2021-06-10 09:57:39.000000000","message":"Patch Set 2: Code-Review+1\n\nThanks for patch Pavlo! Would be nice to add checks to oslo-policy-generator script that will allow to fail when invalid yaml/json occurs","accounts_in_message":[],"_revision_number":2},{"id":"0b202f4d8bff6870405f3c4e0e9291cb218e7da1","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"date":"2022-04-22 14:40:27.000000000","message":"Abandoned\n\nsuperseded by https://review.opendev.org/c/openstack/glance/+/810319","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ebebb3bbb7e88219458912b6ccc1ad08e4f3cce1","revisions":{"8da311f3bcfda129b4abd605ff31e14290402f52":{"kind":"REWORK","_number":1,"created":"2021-06-08 09:29:36.000000000","uploader":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"ref":"refs/changes/62/795262/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/62/795262/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/62/795262/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/62/795262/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/62/795262/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/62/795262/1"}}},"commit":{"parents":[{"commit":"07893401a8211d2a9fc9719ed3e287de77a8a0a4","subject":"Merge \"Cinder Store: Update legacy images tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/07893401a8211d2a9fc9719ed3e287de77a8a0a4"}]}],"author":{"name":"Pavlo Shchelokovskyy","email":"shchelokovskyy@gmail.com","date":"2021-06-08 09:23:53.000000000","tz":0},"committer":{"name":"Pavlo Shchelokovskyy","email":"shchelokovskyy@gmail.com","date":"2021-06-08 09:29:25.000000000","tz":0},"subject":"Fix generated policies and samples.","message":"Fix generated policies and samples.\n\ndue to escaping/quoting issues, currently default policy file\nproduced for glance with oslopolicy-policy-generator is not a valid JSON/YAML\n(same for generated policy sample if one uncomments actual rules).\n\nThis patch swaps the single and double quotes in the definition of some\npolicies so that they are rendered correctly.\n\nChange-Id: I48ce88f4f398e05818457b00cfc8dda90f3c9c91\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8da311f3bcfda129b4abd605ff31e14290402f52"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8da311f3bcfda129b4abd605ff31e14290402f52"}]},"branch":"refs/heads/master"},"ebebb3bbb7e88219458912b6ccc1ad08e4f3cce1":{"kind":"REWORK","_number":2,"created":"2021-06-08 17:17:42.000000000","uploader":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"ref":"refs/changes/62/795262/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/62/795262/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/62/795262/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/62/795262/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/62/795262/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/62/795262/2"}}},"commit":{"parents":[{"commit":"07893401a8211d2a9fc9719ed3e287de77a8a0a4","subject":"Merge \"Cinder Store: Update legacy images tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/07893401a8211d2a9fc9719ed3e287de77a8a0a4"}]}],"author":{"name":"Pavlo Shchelokovskyy","email":"shchelokovskyy@gmail.com","date":"2021-06-08 09:23:53.000000000","tz":0},"committer":{"name":"Pavlo Shchelokovskyy","email":"shchelokovskyy@gmail.com","date":"2021-06-08 17:17:29.000000000","tz":0},"subject":"Fix generated policies and samples.","message":"Fix generated policies and samples.\n\ndue to escaping/quoting issues, currently default policy file\nproduced for glance with oslopolicy-policy-generator is not a valid JSON/YAML\n(same for generated policy sample if one uncomments actual rules).\n\nThis patch swaps the single and double quotes in the definition of some\npolicies so that they are rendered correctly.\n\nChange-Id: I48ce88f4f398e05818457b00cfc8dda90f3c9c91\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ebebb3bbb7e88219458912b6ccc1ad08e4f3cce1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ebebb3bbb7e88219458912b6ccc1ad08e4f3cce1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
