)]}'
{"id":"openstack%2Fglance~796066","triplet_id":"openstack%2Fglance~master~I92d3792602a69922078d109095ad8ac9afc89d14","project":"openstack/glance","branch":"master","topic":"bp/policy-refactor","hashtags":[],"change_id":"I92d3792602a69922078d109095ad8ac9afc89d14","subject":"Add a member field to Image when appropriate","status":"MERGED","created":"2021-06-11 21:39:01.000000000","updated":"2021-08-05 06:41:09.000000000","submitted":"2021-08-05 06:39:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"796066-bp/policy-refactor","meta_rev_id":"cbc37315043e06286623b730f1070b7e37d779a5","_number":796066,"virtual_id_number":796066,"owner":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-05 06:39:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"recommended":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"all":[{"value":1,"date":"2021-08-04 20:39:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":2,"date":"2021-08-05 06:41:09.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2021-08-05 06:41:09.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":1,"date":"2021-08-05 05:17:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":1,"date":"2021-08-05 06:41:09.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"}],"reviewers":{"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]}],"REVIEWER":[{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-11 23:01:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-24 06:46:01.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"CC"},{"updated":"2021-06-24 06:46:01.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"state":"REVIEWER"},{"updated":"2021-06-25 10:47:50.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2021-07-28 17:33:48.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-08-02 18:19:59.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2021-08-04 20:39:16.000000000","updated_by":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"reviewer":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"state":"REVIEWER"}],"messages":[{"id":"8e5ea683a2952ab78f699bdd0245c98a55ff67b1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-06-11 21:39:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0d7ce1454f9d6590f04b4eb7aa04c6fb9059ad73","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-11 23:01:40.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/9ff0a63fda7948f092159d1ec9bff17b : SUCCESS in 1h 05m 27s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7d121dbdfcc741dcb610f9a358d0e1fa : SUCCESS in 1h 07m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ab2b866e46a14bbc99c417dcba4cc8b7 : SUCCESS in 7m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ff43d69ce2646d0a098758c798eead2 : FAILURE in 6m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2bfef075120647be8e509496c0eb4798 : SUCCESS in 7m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b77024a36c2d4be58689d039704ab833 : SUCCESS in 6m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/13f44ef9e9344f85b144b6da9467dbbe : SUCCESS in 7m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38637b112e724f28875bf4bff601750d : SUCCESS in 7m 09s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/bbe968e366b7449c9895ea54f60a55db : SUCCESS in 10m 46s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/200b631cb0544dbfbecba61e7b347cd4 : SUCCESS in 11m 39s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/ccf112cd456e4b53afc2a7be0094b396 : SUCCESS in 6m 03s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/c405a27f1a6a47b4aac6788713b759e3 : SUCCESS in 1h 20m 37s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/32e54d2fa8484b38bd2ce8e56affc339 : SUCCESS in 1h 10m 34s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/0aa6200bc6bc496c859e8bc683e44513 : SUCCESS in 1h 08m 28s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/e76b68d1d4444c17a6c8ee6288929505 : SUCCESS in 1h 10m 32s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7aed3e31aa0d41159d5254a82c86b348 : SUCCESS in 50m 06s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/8f600b96584b4bf894801524365811bc : SUCCESS in 1h 17m 43s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/2d29f22014e64d4baa4de6c20b854709 : SUCCESS in 31m 05s\n\nWarning:\n  Comments left for invalid file glance/api/v2/images.py","accounts_in_message":[],"_revision_number":1},{"id":"831137a9706bdcf32dcd5a473da9d7c94e8a6a0b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-06-16 23:59:23.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ef0ae836c0cbc48021b59a70cedffefcee0e73ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-17 01:34:01.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/cfa25d7358ef4c17be340b7c1604b2a6 : SUCCESS in 1h 30m 57s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/72d1c7a371654fd6a151bcb42a36787d : SUCCESS in 55m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a5ca9c7ee2304ff58bbf6caa3ae84517 : FAILURE in 6m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ddf5dba71baa41968ae71ff153141cb5 : SUCCESS in 6m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/98f02c4c646a4ae987f9ca4868564f30 : FAILURE in 6m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d87a8983886e4e7b8d2137365c043f3b : FAILURE in 8m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e1b576faa1f6496589a751ba6c980782 : FAILURE in 6m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b84caacaecc4e338aad734f5ea3a6e7 : SUCCESS in 7m 51s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e435e60de8e548c8b44e77a351a53078 : SUCCESS in 11m 11s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/76df7acc493d4ff2b01f10de7983643e : SUCCESS in 10m 42s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/ecb6033d01694c32aa4a342676cd792f : SUCCESS in 6m 44s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/bc51047873cf495eab10df1824e00180 : SUCCESS in 1h 00m 06s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/70b63824c0994b09b8f3784d25b6f720 : SUCCESS in 1h 07m 59s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/193b26ada4484d0c9700611cfd70b816 : SUCCESS in 1h 03m 14s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/190f5306d7c442829e79728821e0620d : SUCCESS in 1h 04m 33s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8b62dfe925404623bf762391671edb92 : SUCCESS in 47m 08s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/ebf79a4056e14c5d814bedac82782fc8 : SUCCESS in 1h 22m 03s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/95a275de7fe0429db9da2d6f3f5dfe74 : SUCCESS in 24m 41s","accounts_in_message":[],"_revision_number":2},{"id":"e988ae7034984953879f44fefbf756219bd2c3a8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-06-23 17:07:38.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"a631acd293aa90bcd91993ec5520190db35d02c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-23 19:35:41.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/59577988650343e9a88a734be427c486 : SUCCESS in 56m 03s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/72ef1e8264944adca22a94324627b88b : SUCCESS in 1h 07m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ddf19c4f37cb490196844503bfbb5152 : SUCCESS in 10m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/41010855fb6f42a8a801a63901ec0d0f : SUCCESS in 6m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a678335acf9043df808fc0d23b3b573d : SUCCESS in 6m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9968eaf4054240569be10cbe334f51f8 : SUCCESS in 7m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cfd65816f8464c54ab7918d94ab5a333 : SUCCESS in 8m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4007ce9663964738a699fd96a48d260d : SUCCESS in 7m 19s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b29ad3b7927a4f448cd6fbfdd24d0f7b : SUCCESS in 10m 18s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/f3578e25923d4d468f2e38e93f8a4975 : SUCCESS in 10m 18s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/c01d7e219ef048cd8c983235bc6369be : SUCCESS in 7m 40s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/981413b0c2b8467bba6f4af2879128b5 : SUCCESS in 1h 22m 31s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/3e1e97b2cb2b47a7bf8248a584100996 : SUCCESS in 1h 20m 29s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/71e87652b02d4c52b120fb5cde7705f1 : SUCCESS in 1h 03m 12s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/502fbcd43d4d475ea277f7aa1a8c3c39 : SUCCESS in 1h 11m 04s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/385a774db3704906a10843ef65a55b7e : SUCCESS in 53m 36s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/e6f7a140b41a45299652ab1b32b17258 : SUCCESS in 1h 16m 38s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/cdd731132ed943d1999b2b147e99a635 : SUCCESS in 37m 24s","accounts_in_message":[],"_revision_number":3},{"id":"f4e121434b521c3030ac9f91ea1fb8c1434f53f4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-06-25 10:47:50.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"74e726f42e57866c74b6d5ac5820ad5ca52635dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-25 12:15:24.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/8c6eb88773c44b9889083320698d8910 : SUCCESS in 1h 07m 44s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/34804c94cc6c410ab6643818aea0b6b5 : SUCCESS in 1h 01m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/853b64d213ea49eb9a91006817821cf9 : SUCCESS in 10m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d378b2323b645a1bd508c36e53cbe09 : SUCCESS in 7m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/63997d84438a45d7b7c7c0ae1bb22972 : SUCCESS in 8m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ca9d8aa1b50f47e18d1eb6d8e7c031d2 : SUCCESS in 8m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e3f7674c81ac4d09b88a95e1ff05ce55 : SUCCESS in 9m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f55ef47fb3824a44a3fc3a871a7bf688 : SUCCESS in 8m 49s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2c6fc1107b4a4620838eb8e9b241fad1 : SUCCESS in 10m 24s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/d414855abdfd4eb7b8d765ab32fa979b : SUCCESS in 12m 30s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/8c2685bfad834c2ca2d11e009178ed8a : SUCCESS in 8m 16s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/92eb16f05635415ba2e202c001af8a9d : SUCCESS in 1h 12m 55s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/b992f2a6c3e542879df18e96e64e546b : SUCCESS in 1h 15m 17s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/c2188ed727964dce8bcd9dbde45ce996 : SUCCESS in 1h 15m 35s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/f30a9e35e54343b2b377a2bfd7e4524a : SUCCESS in 1h 21m 08s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/585e2bf2513d44b39bcaecd3981a1d53 : SUCCESS in 50m 33s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/03e7f90023d44071b224873612e384e1 : SUCCESS in 1h 13m 12s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/d6319e2a79444480a5320e0bbc485b68 : SUCCESS in 34m 25s","accounts_in_message":[],"_revision_number":4},{"id":"6883ff4b896047f1dc1fdca15ba4d7cb97ff19c7","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-06-28 05:37:52.000000000","message":"Topic delete-policy-poc removed","accounts_in_message":[],"_revision_number":4},{"id":"63d8d7b2a7cbe6310ebdabbb2cffb8c6c5263bf6","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-06-28 05:38:01.000000000","message":"Topic set to policy-poc","accounts_in_message":[],"_revision_number":4},{"id":"6c96de77dd2ca553bfd9bb6b7e67a18f83346eab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-06-28 17:48:24.000000000","message":"Uploaded patch set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"9ccf22817332056be3696380e890430e6f29631e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-28 19:07:52.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/3e5898b1c0b7430e94a3e36be9514e78 : SUCCESS in 1h 03m 03s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/14dbafb1a66f482695430a99265bd8c7 : SUCCESS in 1h 08m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dcfc78bb38cf46cdbb68e550fc0386e5 : SUCCESS in 9m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c55d316e9be45e1a2ecfe20d90bd93d : SUCCESS in 7m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fce79731f7da494aa87fa4595145e8c5 : SUCCESS in 7m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6dd28d20392944de87062384c82f4971 : SUCCESS in 8m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cde398ee8ac3435c9e64e395e81a421b : SUCCESS in 8m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/156a1d65491242ffb52ff33a2d3fcc66 : SUCCESS in 8m 52s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4d0b581bf667466a99e4372f2ed1bf1f : SUCCESS in 10m 15s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/e5bb6c0510c244fd9ba48d64cd648802 : SUCCESS in 12m 38s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/86e952f4898144fab1334c7c22021ad3 : SUCCESS in 8m 24s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/4d4afcf18d964b9797a33ced5c3c0ae7 : SUCCESS in 1h 13m 33s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/83499ffe61fe4521bcb39cb76dd101f2 : SUCCESS in 1h 03m 57s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/286f0b7180d549c885afbd93eb553764 : SUCCESS in 1h 06m 19s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/c0055def6e394ab2a36121525fe9dc49 : SUCCESS in 1h 10m 55s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cad6acecf97d4f54b1ec49126ce8d03d : SUCCESS in 50m 47s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/c9439f98a8ef464d90c39d542c7b9a3e : SUCCESS in 1h 14m 40s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/07ea9fea57f94c7e8acb1579f2dfeb0b : SUCCESS in 30m 16s","accounts_in_message":[],"_revision_number":5},{"id":"42beb885d6ba44213d0607c5acbe67c7077a2525","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-07-02 14:49:29.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"6d22b4984bcb8d218bae60fa2e5cb2d868ef22c4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-07-02 15:51:23.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.","accounts_in_message":[],"_revision_number":7},{"id":"52ea43f17d9b596f93bc2bdcb964484fc8a92b0f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-02 18:12:33.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/6214a0f8810b489f8182deb20a430160 : SUCCESS in 1h 05m 32s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0e365feecd724bfba06a3a35c4865c97 : SUCCESS in 1h 07m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff264dbd22ed4208ae1ca6fef7d0b674 : SUCCESS in 7m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/11c761d549a44a6d8929383b116d0989 : SUCCESS in 5m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3e4dd8f1448a48e9bcd2e5e590cfbac8 : SUCCESS in 6m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/72749db3177e40c5884f4898df6b3b87 : SUCCESS in 7m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff4e1687f2524915856d2de0c46c58ff : SUCCESS in 6m 47s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/33c90e32922f43c789cddbeef6daeed2 : FAILURE in 14m 51s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/2b83dfef87b04b15884df059d3382f5e : FAILURE in 9m 25s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/7759a2690bec4c5fa7a9cbda403ffdc7 : FAILURE in 8m 07s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/82ea8b5cb8bb4026a5b5ecca62e28cff : SUCCESS in 5m 48s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/bd6df75435434f5f84f354e51d447101 : SUCCESS in 1h 10m 55s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/c45d0e847ca842c0a47acf900f884d75 : SUCCESS in 1h 02m 42s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/92facb3b4b5e40608995291bc51d33ba : SUCCESS in 1h 07m 33s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/fd53f9d1162f426998cabe4b38ff7740 : SUCCESS in 1h 09m 32s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/209159a5c6374d7899f55a0495dc80ee : SUCCESS in 44m 06s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/dcc591858b5f4e01b209609761fd5f49 : SUCCESS in 1h 18m 23s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/465182b99a6540ca9635141eefb5ceca : SUCCESS in 32m 47s","accounts_in_message":[],"_revision_number":7},{"id":"2a4854f7faefd1c6cc75418955339adb14370704","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-07-06 10:51:18.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.","accounts_in_message":[],"_revision_number":8},{"id":"26ae1effab456baaa6e3c35f605afdfc1ca9f60e","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-07-06 11:08:19.000000000","message":"Topic metadef-policy-poc removed","accounts_in_message":[],"_revision_number":8},{"id":"53c0d980b7650e9dfbb925a572c8394da2a7ebc0","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-07-06 11:08:27.000000000","message":"Topic set to policy-poc","accounts_in_message":[],"_revision_number":8},{"id":"2d36b917d94bc1d047eb7a624709b83f54a4a094","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-06 12:18:32.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/84bf1b3eaaeb4c57af7b7fcc083f7b7c : SUCCESS in 1h 02m 02s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2f572da6ceb44a1eb510ae0ecd792db3 : SUCCESS in 1h 11m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ad6dd1c5e584665b8ba1920865539f8 : SUCCESS in 6m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/13d2293ed0064c37bfffb01dfa594d57 : SUCCESS in 11m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/427d48c34e9440f9a363d07891172b37 : SUCCESS in 6m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1db8b33bdc5744c0a73ddebb73cd8a17 : SUCCESS in 7m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc3a33fa2d204e19b3f4f51b9f59f39a : SUCCESS in 7m 29s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/bb8f237fd81642d19b38b64327ddd334 : FAILURE in 14m 07s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c60eec28bbe54b278566a6a3a2ebe367 : FAILURE in 9m 55s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/224dc08b500348e587d5806e4827ecdf : FAILURE in 10m 11s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/0479da9b889c476293fdaf56859484e2 : SUCCESS in 6m 53s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/b2a6c9eef73c4addab11734b57df91f7 : SUCCESS in 1h 10m 02s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/b34f6e558e724ec1811105808c18f327 : SUCCESS in 1h 15m 15s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/e0f151378bd647e38676272b2c851511 : SUCCESS in 1h 18m 02s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/6d669c784af54a2084702a0f2fda940c : SUCCESS in 1h 12m 40s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/21ad58932954453f88a7a8a5edafd61a : SUCCESS in 51m 22s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/9e211309b03b4d9fbfd44e703ff50c12 : SUCCESS in 1h 24m 06s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/423eac278e4941f1aa76ccd3e54b3872 : SUCCESS in 35m 39s","accounts_in_message":[],"_revision_number":8},{"id":"ed8c78adc8f6123951fa6b2af6783b624bc3960f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-07-06 18:00:34.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.","accounts_in_message":[],"_revision_number":9},{"id":"d374038caf1ed5823b9a1209dfb28f43c71384a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-06 19:26:26.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/02b8d14597a54c2caa8d389193d882b2 : SUCCESS in 1h 13m 04s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/26ca4b91474a40b79b48937c388f6aca : SUCCESS in 1h 08m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ba7ccc8842c4956a2a942294e55f8ee : SUCCESS in 6m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f82bda2c00144d6c847ff6313481e6f9 : SUCCESS in 6m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/070b62792de94358a76bc0f328f90256 : SUCCESS in 6m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3df76f0f4ea241d4be22b16db43f51ce : SUCCESS in 6m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f7386dfeef446989b35c45831e6a940 : SUCCESS in 7m 14s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0f9ff3aa035d4ca0bfe0e94daffd3772 : SUCCESS in 9m 20s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/aeac499319134a8aae6682d6dadd5e69 : SUCCESS in 8m 46s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/823ae71c7bde491cb7974d671a971a9c : FAILURE in 9m 03s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/4e9488b790cb4cf4ad91b2ff7ce42703 : SUCCESS in 5m 43s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/a7409df6912c47e4b5a5bc4d1ae72b79 : SUCCESS in 1h 12m 36s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/d31726e7a21b40cbb2a1a91585aebf7b : SUCCESS in 1h 06m 11s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/4539e76a99494fc6b09943beb415cc4e : SUCCESS in 1h 15m 58s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/af2afd36026e4fbf9349292840ca9795 : FAILURE in 1h 10m 57s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/29027f58f9e54fea8de669e6ce4386f2 : SUCCESS in 55m 46s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/d0e2bb9fa97e442083592c4e347a5310 : SUCCESS in 1h 18m 33s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/b74df6dcbd494c3d92206fcded0a1b45 : SUCCESS in 36m 28s","accounts_in_message":[],"_revision_number":9},{"id":"fa9f4391d21ce4b1b603da05c59c6de6d94c1046","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-07-14 19:32:52.000000000","message":"Uploaded patch set 10: Patch Set 9 was rebased.","accounts_in_message":[],"_revision_number":10},{"id":"412ed8649d6108d8576b9c763d63a6652f7ed4ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-14 20:59:52.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/74b9c102e54e4c03aee914a3fc641095 : SUCCESS in 1h 09m 00s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/fd17c10f7b3a430ea7b841f15f76f82c : SUCCESS in 58m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/062b54a8f1e14b95946387f683908a38 : SUCCESS in 6m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2aa4126957a94780b910676003f57ff9 : SUCCESS in 6m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1be96d21c92442f9ae6520b78d7a04ab : SUCCESS in 6m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/48f6cba90148474c898f8d33f094c003 : SUCCESS in 6m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3769ec92629640d5b69dcd45cbe0a1b5 : SUCCESS in 7m 42s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5b37c1951121451490500cee0c0193d8 : SUCCESS in 9m 48s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/cf533f28501643dd9a666121f87402d4 : SUCCESS in 10m 09s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/e4cf461ceb89439791a9b6093705ba8a : SUCCESS in 8m 53s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/524916066afa455a96be6d9ed55efb4c : SUCCESS in 5m 55s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/1a15f5ff26eb4e2b9781367d2e0a11f3 : SUCCESS in 1h 16m 57s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/d06bb3f7fcbc43309b4c0206bee55b93 : SUCCESS in 1h 19m 46s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/e03ce05cc04f443a9b039d6887d309d1 : SUCCESS in 1h 08m 26s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/8a358022fa4f4f04b8df36ccd6a6c943 : FAILURE in 1h 09m 05s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/26d02373ca8d4ef2a5007e630012d1c3 : SUCCESS in 46m 52s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/d3a43dee185644f59c83a279bc7f9c69 : SUCCESS in 1h 12m 26s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/f916bc47083a4cef938c38f9d3e882d3 : SUCCESS in 31m 38s","accounts_in_message":[],"_revision_number":10},{"id":"370e97c26545a3891a896fe41351b7db53aeede3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-07-16 13:56:40.000000000","message":"Uploaded patch set 11: Patch Set 10 was rebased.","accounts_in_message":[],"_revision_number":11},{"id":"ab5c8e4b8c85271da728797c85995f792997483f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-07-16 15:28:57.000000000","message":"Uploaded patch set 12: Patch Set 11 was rebased.","accounts_in_message":[],"_revision_number":12},{"id":"6a64f993a72fb9f4394fb71d52c13b70abc795a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-16 16:58:57.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/32ab5bac71f94f49a282b5f8e90ef6bf : SUCCESS in 1h 12m 47s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a8389c94a89642b58d24b07ec69b34b2 : SUCCESS in 1h 00m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f296da58c464b7a865a55a29fcd6a48 : SUCCESS in 6m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d1745c69e2014085b421946942f782cc : SUCCESS in 7m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fd5dc0e3cecc4cac92091f16533c8110 : SUCCESS in 6m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/283bd9f697964c339888c6e5652d8c1b : SUCCESS in 6m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c2cf3e4a5f3489fb67d7fe16b04151c : SUCCESS in 7m 55s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/94237aad7d88408cb89edf5010391e91 : SUCCESS in 8m 43s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/aae10b3c46104638add2521d141e9084 : SUCCESS in 9m 41s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/a04fe2507ced41009c2ce1d729135d14 : SUCCESS in 9m 06s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/6807644fc29d4ac4a9dc8c373bc253e5 : SUCCESS in 5m 47s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/3b2abf04d2ac4516a68558c12f6795a1 : SUCCESS in 1h 11m 22s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/b3c354b252154bf0be31f547d201f588 : SUCCESS in 1h 18m 32s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/2eb14a63b2f04ae99f384b671b81779a : SUCCESS in 1h 09m 18s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/2e60e9c3d29d476d90638b53526a7019 : SUCCESS in 1h 15m 28s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2490134817dd48588136be46a0cfde76 : SUCCESS in 58m 14s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/121869ece39d49de95fd7dda9051b9f3 : SUCCESS in 1h 16m 22s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/9c8c28f31bd641a88efcecdd74558d0c : SUCCESS in 37m 08s","accounts_in_message":[],"_revision_number":12},{"id":"6593cfc059e3eddd933a440e927363f3037fed02","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-07-16 17:49:55.000000000","message":"Uploaded patch set 13: Patch Set 12 was rebased.","accounts_in_message":[],"_revision_number":13},{"id":"cf1b0227eda61c7691bd98309010ffb761ebee04","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-07-16 18:29:40.000000000","message":"Uploaded patch set 14: Patch Set 13 was rebased.","accounts_in_message":[],"_revision_number":14},{"id":"3a3d72af57df0121837c437bc4fe3166c6f1be99","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-16 19:48:29.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/d6adaeaeef684f52ad32eb3fa99a1480 : SUCCESS in 1h 10m 22s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/45f857da57774592a9b38d1bfc8b7d0c : SUCCESS in 1h 03m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b73c7c191604ad5b9de51ad3aec7feb : SUCCESS in 6m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0f1ab3ed6c8d499cb557f82454413053 : SUCCESS in 6m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/67b53247964f429cb56036558ee0643e : SUCCESS in 8m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ca5bf3fa68d1447ca3b6ac2fa58c4e9f : SUCCESS in 7m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2396748ab8af47639cb0cff1df3f5ac2 : SUCCESS in 7m 10s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d1f7b2ce3e4947718c3d3825564a7700 : SUCCESS in 8m 38s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/a288864dc38143209b511f14723ceeb9 : SUCCESS in 9m 20s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/8d9aa365cb5642bbb572b34562ff59bc : SUCCESS in 8m 34s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/877d3170b1344db38f5bd357d7c77c58 : SUCCESS in 5m 51s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/2ddb004b21da44e78ea413335ee879de : SUCCESS in 1h 13m 24s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/cac7e9f20c0f4793b3cbb01cc42ea575 : SUCCESS in 1h 09m 27s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/370dec1522754122bc6a482cbb5f0ede : SUCCESS in 1h 16m 21s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/500d09d876ef4558badbbaa03cbb3454 : FAILURE in 1h 05m 09s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9717f836ed8b4f8b880225a33d9877f6 : SUCCESS in 51m 47s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/6f1eb8d96344460a83f811742b633e41 : SUCCESS in 1h 09m 26s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/d06cc4f4e3934c019343cbdb49ed9eb2 : SUCCESS in 33m 42s","accounts_in_message":[],"_revision_number":14},{"id":"194c30ab2225ab08f6ca8188a0a80e3f39ace394","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-07-19 19:37:51.000000000","message":"Uploaded patch set 15: Patch Set 14 was rebased.","accounts_in_message":[],"_revision_number":15},{"id":"a17a9eb4018e331fe57dd13be9d73148f8783e4d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-19 21:05:59.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/b167c2355af94d068340cb61015f10fa : SUCCESS in 53m 14s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9e7ab2ea501044ada8e54fabbb336fb9 : SUCCESS in 57m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b540431f68344cee8de71bb6c975af4c : SUCCESS in 5m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1531bf1a25b946199ad19e202ddafee9 : SUCCESS in 7m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/be3ce0ab514b49b6b3d15344a03d2e0e : SUCCESS in 6m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/71fa2754805540b3a98a40871f18c44e : SUCCESS in 7m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5313ba0cb28c4b8389ea75a931277d49 : SUCCESS in 7m 15s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b2be345e68544b3fa1fac00ac249e190 : SUCCESS in 8m 51s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/2d3d0edfd3674fa8afa466b1f173ea54 : SUCCESS in 9m 04s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/3b3a023e605e4b8a8e0c49ffd0dc6bb0 : SUCCESS in 9m 20s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/2b09a97291734382af2f1820664b62a8 : SUCCESS in 6m 03s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/6ac8837ecca24d0bbcea554ffbda1432 : SUCCESS in 1h 00m 46s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/e2354906b20945b4be03f1a1d42b8167 : SUCCESS in 1h 03m 39s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/5eb1ce92e14a492db97550302c331a74 : SUCCESS in 1h 05m 02s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/3ae07ed8e0ee4ee28bc69f4eeddc59db : SUCCESS in 1h 10m 40s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1d2565c647604664bb32232c63918edb : SUCCESS in 56m 21s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/8ebfd30984db4e3e8e9fae9d56089ea6 : SUCCESS in 1h 23m 15s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/daf54e3f1e6a4eadb7fb689a000b1eb4 : SUCCESS in 35m 25s","accounts_in_message":[],"_revision_number":15},{"id":"965d9bb873783132b5b711890cfe8f0e85744720","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-07-20 16:30:14.000000000","message":"Uploaded patch set 16: Patch Set 15 was rebased.","accounts_in_message":[],"_revision_number":16},{"id":"7e1733c0c3e658b5865df5001c085f4ba6b06a80","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-20 17:56:48.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/d420db879bf241c885f8b72c41e65790 : SUCCESS in 54m 16s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1c4f7de3cf664699b5e6d0eeab8130e1 : SUCCESS in 1h 17m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ccac2ca1d33c44799626d66864d4f3d7 : FAILURE in 5m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6594cc6529ef4a8bbc2d9a46d77f2919 : SUCCESS in 6m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/60ba9d2b17d5447782fe3ed27d65e970 : SUCCESS in 7m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6c9ad7955651449f9d502c5bf1bd61f7 : SUCCESS in 6m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c723b95f70384971950f1379d0db02f2 : SUCCESS in 7m 21s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/580ecb2d40f6464482bfba8321c79c10 : SUCCESS in 9m 09s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/48f1f9c85b744d26b7fce77657c63682 : SUCCESS in 9m 17s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/e8e524b817734355a9c4b1a316dcdc06 : SUCCESS in 8m 50s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/cd5b70c34556450a87f1ddc5068babe2 : SUCCESS in 5m 24s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/aa4354af732a4528a6526b7620d51d16 : SUCCESS in 1h 13m 45s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/4a378ac5b121436fbf7c97d6a5ca8db5 : SUCCESS in 1h 04m 11s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/8bfcd56511a747a58bfcad3473c9cb5e : SUCCESS in 1h 09m 44s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/161ce1a6c8cb4600a2d8b7d6c28abc71 : SUCCESS in 1h 20m 01s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/369b77d0766a4458b49f9a20fe74d4e2 : SUCCESS in 56m 37s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/34f28e5b808741e1af0594a2b8512f27 : SUCCESS in 1h 21m 36s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/567cdd86e9ac4d4796a2d6fe9ceabf51 : SUCCESS in 37m 50s\n\nWarning:\n  Comments left for invalid file glance/tests/functional/v2/test_images_api_policy.py","accounts_in_message":[],"_revision_number":16},{"id":"35099de51bd02dc0cee32ce134264cdf789f234a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-07-20 18:34:26.000000000","message":"Uploaded patch set 17: Patch Set 16 was rebased.","accounts_in_message":[],"_revision_number":17},{"id":"440356d519bf9c38bbed261d743939a0032b09e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-20 19:57:48.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/26196f9e8aed4f5ba82ac4ad421bdc06 : SUCCESS in 1h 09m 42s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/31908f2932d9427b94748826c8d11bf4 : SUCCESS in 1h 05m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cbfa059195734dbab0881a4356bee893 : SUCCESS in 5m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8e5383bd0b77423ab5fc8a2c2c81b2e1 : SUCCESS in 6m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4bdf6f71982f42c8ab12756d651dcea4 : SUCCESS in 6m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bacd4c6127df4975b13e41a55024712b : SUCCESS in 6m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a3a99368b4f435b8a4cb2b76886c168 : SUCCESS in 7m 17s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b9ea644354a646a58e28ccdd25da2881 : SUCCESS in 9m 53s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/0ac3c09db9ad4439800d3e6caf5ee158 : SUCCESS in 12m 32s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/70b13d71cbb74a2f84a32d0981edfd6c : SUCCESS in 9m 50s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/55038bf5ccd848598a4d9a26ca47f279 : SUCCESS in 5m 46s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/0d1ba2203f3f448d8d012d033d54b08b : FAILURE in 29m 24s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/f520b65bf42741db9128d5c6d5943263 : SUCCESS in 1h 04m 07s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/f85811d25d6f40e8b77be8cb4d447a99 : FAILURE in 53m 55s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/ac316787524d448b854d90bfdbb4f10a : SUCCESS in 1h 17m 26s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/01b6b052f99b46de9592f6c2f222f95f : SUCCESS in 51m 01s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/62c67b6e3e4747c994735778a029631c : FAILURE in 32m 10s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/8245545ec4fd494ab14870e1f0ebaf75 : FAILURE in 21m 00s","accounts_in_message":[],"_revision_number":17},{"id":"2a88f2becdd7959c45d0d48ade5af64847fca20d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-07-26 19:49:00.000000000","message":"Uploaded patch set 18: Patch Set 17 was rebased.","accounts_in_message":[],"_revision_number":18},{"id":"9164b8f43f4031f15a07702c614ee642cde1e41e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-26 21:11:01.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/2fbc833ffc514120b9a5e05d311e3bba : SUCCESS in 59m 03s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9abb20fa61d04c9685c2d0adb31104ec : SUCCESS in 1h 04m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a4e88611fb74c799b11aa1782f0f389 : SUCCESS in 5m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/204641d3d7444c73aaa27c67958e2103 : SUCCESS in 6m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e6bf9de7c46e41e8b306b044a1e98299 : SUCCESS in 6m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d261c0ed549440639543fa971a7ae8b7 : SUCCESS in 7m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e717eed5b71443c2a2ef558e9b960836 : SUCCESS in 7m 26s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/eb774030ecda4683bb98fcc50c4b762f : SUCCESS in 9m 04s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/0d1610c15c7640e09af2d518746ba701 : SUCCESS in 10m 48s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/3912238bb91b4569b41e12b12ad24bec : SUCCESS in 10m 30s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/ac70c424f168480687ec3f1e8c405c6c : SUCCESS in 4m 59s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/5e783735adb444d3a4bdbf6c3dd74182 : SUCCESS in 1h 16m 16s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/08be6a252f32473eb38f3057c4bff57d : SUCCESS in 1h 04m 43s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/a6ac001406164a3cadd54075e259f731 : SUCCESS in 1h 15m 15s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/07b69f624e194d34b7d55741baf26d9c : FAILURE in 1h 02m 02s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/67891e2cfdb144abab5c72bcb46ce83a : SUCCESS in 41m 00s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/060d81d812414637991b9df530de550b : SUCCESS in 1h 13m 04s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/4b8faf8545ce472a9b3fbe72430b0a26 : SUCCESS in 34m 00s","accounts_in_message":[],"_revision_number":18},{"id":"3b134f0d2cdd67af2279c2509e03db184f3ac8e8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-07-27 18:34:02.000000000","message":"Uploaded patch set 19: Patch Set 18 was rebased.","accounts_in_message":[],"_revision_number":19},{"id":"9f48323f4ee48ed78d1458ec3a5a91a2eb229b61","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-27 20:00:09.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/962f50b757f345ab9524a7f56df06b00 : SUCCESS in 59m 53s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e67c89cee1e04d7589451ab0efa05a0e : SUCCESS in 1h 05m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b6e1ab87e30f40758c6231e1ce77b98e : SUCCESS in 6m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/695ba33d705b486b99a4501e7891057c : SUCCESS in 7m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f10cb2579e3e4ecfad9fdff3da7df42f : SUCCESS in 7m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e5482590e18d4c5d9e15f7b89b2caef5 : SUCCESS in 8m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ddc704924de74c3499beb509ac002cb4 : SUCCESS in 7m 02s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/81c33c627b474e008a637df74ff7972f : SUCCESS in 10m 21s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/56a03a2701d2460397fbae43f3714723 : SUCCESS in 9m 45s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/8e49ad9d469d4786b6d0c50b75635ce5 : SUCCESS in 9m 16s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/be0a934e270e4b8981f542ef4a4ebbe2 : SUCCESS in 6m 24s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/063db2d0ad51463aad94077792d3750f : SUCCESS in 1h 20m 55s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/0b22f5f488dc4a0bbc9cdeead2464e0a : SUCCESS in 1h 02m 51s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/45b2568c9ce94d28bd95b019d2c3f6c7 : SUCCESS in 1h 03m 16s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/54feb677b8354457af1c351c7ff4c1da : SUCCESS in 1h 19m 51s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7f0678a9054d40c3a8542593795e7d8f : SUCCESS in 51m 34s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/e0783c0da1ab424c8dde5ec5aa9d618e : SUCCESS in 1h 22m 26s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/124ef1251b144972be517cb333230512 : SUCCESS in 31m 18s","accounts_in_message":[],"_revision_number":19},{"id":"24fdeeee516d62a79caff139a4cc61e7b047de96","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-07-28 13:33:44.000000000","message":"Patch Set 19:\n\nThis patch will be crucial for Image Members policy refactoring. Will help a lot to get rid of read-only checks in auth layer.\n\nCould you please point out remaining things here?","accounts_in_message":[],"_revision_number":19},{"id":"f8022fe2a8e1bffc4b0deeee846ad365675f1ffb","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-07-28 13:58:35.000000000","message":"Patch Set 19:\n\n\u003e This patch will be crucial for Image Members policy refactoring. Will help a lot to get rid of read-only checks in auth layer.\n\nYeah, I\u0027ve been kicking this further in the stack since it\u0027s not needed by the lower ones. I guess I could pull it out if you\u0027re planning to do the member work.\n\n\u003e Could you please point out remaining things here?\n\nTests? :)\n\nI think it\u0027d be good to make sure that it behaves in a variety of situations, members, no members, no owner, public, etc. Also, I think there\u0027s simple/db implementation that needs to be done probably.","accounts_in_message":[],"_revision_number":19},{"id":"ddd30f35ccc34862d47478f5a8712e2620153abc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-07-28 15:13:52.000000000","message":"Uploaded patch set 20.","accounts_in_message":[],"_revision_number":20},{"id":"d82eeeb9c5bb43504bb0eaacd5f2c4400825be38","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-07-28 17:33:48.000000000","message":"Patch Set 20:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/796066/796066/20/check-not-vote/ext-glance-zuul/afcc467 : SUCCESS in 36m 29s","accounts_in_message":[],"_revision_number":20},{"id":"9118316c7189faf9c6efbb9805db1af0e783a1e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-28 21:12:57.000000000","message":"Patch Set 20: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/4e6aa8896143473393dfdcfb8452a688 : SUCCESS in 1h 07m 25s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/cb44c28f76b44e6caa11c6271c7f1475 : FAILURE in 49m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1bb0ddad85aa49408c41a6aec15a588a : SUCCESS in 5m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cc3e437e9e234f4c94be06a1a26b6418 : SUCCESS in 6m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/39cb22c28cc64d52a515f4d21d67d049 : SUCCESS in 6m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/184d22b1cc1e4eee9cb2e6d0a58d75d6 : SUCCESS in 7m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2d407777798406ea322e1bb3b724a4b : SUCCESS in 8m 11s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/feeee122843b4e37b21f5dcebfcd04ca : SUCCESS in 9m 57s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/ce3317cfc9c1408c9436baf0c0439763 : SUCCESS in 9m 54s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/76713a5db32b427b9a9dcf4c74f12706 : SUCCESS in 10m 15s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/9f12e3417427472e871aa398e86d1d8d : SUCCESS in 6m 04s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/7a44c9d0ecf64f7a9c4df92205c3f54f : SUCCESS in 1h 08m 25s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/f5ebbcde8f0a4a0893720c3c7539e8e3 : SUCCESS in 1h 03m 28s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/d6a2d8b1490141bf934717957aa38720 : SUCCESS in 59m 16s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/941e7ca78edf4204a73595f75ee202c4 : SUCCESS in 1h 07m 21s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/765e26da6532407cb6af7af93107b206 : SUCCESS in 44m 31s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/424e9ab1662843cd915fe023c5fd6e64 : SUCCESS in 1h 05m 55s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/6e602435d9264bdb8aaed9bb50b9ba64 : SUCCESS in 29m 09s","accounts_in_message":[],"_revision_number":20},{"id":"9b72e302f7dd5b9fedeea0d3df0b12da0c8b480c","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-07-29 05:15:06.000000000","message":"Patch Set 20:\n\nrecheck","accounts_in_message":[],"_revision_number":20},{"id":"cb03f12ad2705d50d0415fd03c08e04c38451da6","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-07-29 05:55:00.000000000","message":"Patch Set 20:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/796066/796066/20/check-not-vote/ext-glance-zuul/0bcb2ef : SUCCESS in 38m 57s","accounts_in_message":[],"_revision_number":20},{"id":"051924fe5729116e11eaa4fddd459ce223646a5f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-29 06:42:23.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/5a0a536e0fe14801b2688689b5904ef4 : SUCCESS in 1h 11m 03s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f20264edd6e34d15b16e5693e02d14b9 : SUCCESS in 1h 07m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/134a7f67074f4626aef4177c89fac8c9 : SUCCESS in 5m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/195ff87755a24f4ea5066ce2a05ba775 : SUCCESS in 6m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5748d05134754ecca04d545ddc58b0b2 : SUCCESS in 5m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b3ff38f00bd743d4a42b83a4f22bf2bd : SUCCESS in 6m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/882d46080a4d496baea499cd88d628bf : SUCCESS in 6m 21s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/cf51ba6d58ab429bbff5f6f88bbbf4e8 : SUCCESS in 9m 08s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/43f06dc1e4e8476f9333946fd521f93a : SUCCESS in 8m 28s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/59efadcdc7114d21bb51af1977d50211 : SUCCESS in 8m 25s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/c0c5aa66f7ea46d187e60e932fa5d7ff : SUCCESS in 5m 08s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/8c3df6c43a4046d4b829620c1f8662b7 : SUCCESS in 1h 17m 14s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/870420d91aee488e88d5ce656c6024df : SUCCESS in 1h 10m 22s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/76089132cec242c59b720e8c00dd467f : SUCCESS in 1h 16m 54s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/341c3a8049b94e338f1b94b6fb08f4e9 : SUCCESS in 1h 20m 37s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/23bbd1ca1593444fa673e2b8e1892594 : SUCCESS in 58m 19s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/46b82ff7eb154517928661019e08e6e0 : SUCCESS in 1h 21m 11s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/e53ea9b01f0b429bb654bc86f7ee75a0 : SUCCESS in 36m 17s","accounts_in_message":[],"_revision_number":20},{"id":"3d5dbdd4d01793264f879e826e9ded2cf5731677","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-07-29 17:51:27.000000000","message":"Patch Set 20: Code-Review+1\n\n(1 comment)\n\nLooks good to me, just one question inline.","accounts_in_message":[],"_revision_number":20},{"id":"188f01d434890d12c858caa668fef70eccaaea8c","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-07-29 18:01:49.000000000","message":"Patch Set 20: Code-Review+2\n\nMay be it is just not flagged in every get call so some calls might be visiting auth repo.","accounts_in_message":[],"_revision_number":20},{"id":"939fe4aaf70dc5673203f284ad6c6f0b7fe2272c","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-07-29 23:59:36.000000000","message":"Patch Set 20:\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"0fa32a6529d75e1b17bfe6e35035be06025a52bf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-08-02 17:22:08.000000000","message":"Uploaded patch set 21: Patch Set 20 was rebased.","accounts_in_message":[],"_revision_number":21},{"id":"e8365e49a96ef516f9454657923a706dc6cf2c06","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2021-08-02 18:19:59.000000000","message":"Patch Set 21: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":21},{"id":"8eeae3aeb07f3476b110a75584121f8080a0975c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-02 19:02:52.000000000","message":"Patch Set 21: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/c6f4e876bfc241169b36adf08d1e24f2 : SUCCESS in 1h 13m 29s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0a3eb0db51e341788e8d5ad1d0307c13 : SUCCESS in 1h 17m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/52d4e3bc74f24082b36bfe348d9cf926 : SUCCESS in 5m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/130e311def2a4c6b8b19bf918f9ff7b1 : SUCCESS in 7m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5da52ef9afe14e02a569a289540528ce : SUCCESS in 6m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b0034c76b6d641588445b9ffca6713be : SUCCESS in 6m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/730bf58864434549a538a3d1146ca786 : SUCCESS in 7m 07s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9580c3b9b26c463a98ae0a7a9ce48b36 : SUCCESS in 9m 34s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/a83393549607476aa1da1a262fb1bc8a : SUCCESS in 9m 31s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/2227afe46238401993cac70287570940 : SUCCESS in 9m 28s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/510d1c14a37a472681a1d266120c971c : SUCCESS in 5m 38s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/ee6dad78da804db89c2dcc0ec5ab0afc : SUCCESS in 1h 20m 01s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/9dc05e0979d145efafb40efadd3e383d : SUCCESS in 1h 10m 11s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/cee1bda74ae34047b35118bc21528ff1 : SUCCESS in 1h 09m 36s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/8370d78b9c65403f94aa0f79fb04cd01 : SUCCESS in 1h 23m 15s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2fc6c84b4ffb45a1999b2a42eb33243e : SUCCESS in 52m 28s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/401e39ca62144d6d82c70a53357b5920 : SUCCESS in 1h 34m 16s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/10b44d0da53047ccb1e8f2a1f85160c5 : SUCCESS in 35m 42s","accounts_in_message":[],"_revision_number":21},{"id":"732cc36a5f8b902643a26a3deaba9641d3109862","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-08-03 18:36:00.000000000","message":"Uploaded patch set 22: Patch Set 21 was rebased.","accounts_in_message":[],"_revision_number":22},{"id":"c8a00dcd979a0e26d5d64bbbfb9cea1327c18a39","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-03 20:04:59.000000000","message":"Patch Set 22: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/f37654869c114e518a545cee0724ad8b : FAILURE in 48m 47s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/07233b314cee45eaade10f9ad85240ae : SUCCESS in 1h 09m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/933dd6c704d6482182362277dbfce4bd : SUCCESS in 5m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ea7c681973d04b03a4574a272c0c0971 : SUCCESS in 6m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/40086031aa77477e93ef05be75cf1273 : SUCCESS in 6m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/400bc88a1cb646308f86d0c84c7154a6 : SUCCESS in 6m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b129ed12fae4ead916b0100e87e251d : SUCCESS in 6m 17s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/fff884778ef84c928a908c6b34376322 : SUCCESS in 9m 57s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/55de76671c6b4264b6c7a6a602f7f0d4 : SUCCESS in 9m 03s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/358e424c65f34f3489a1d64d6d1ce78e : SUCCESS in 8m 53s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/ff8e65e818684f38912a2c5a2de90491 : SUCCESS in 5m 08s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/c4e27d5f271240fdb170a1b8ec860705 : SUCCESS in 1h 19m 08s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/cba9acb2ae2942699c10492924ba4e93 : SUCCESS in 1h 11m 14s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/841e6e01da05438ba0a135f8e63f76dc : SUCCESS in 1h 09m 38s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/7b492eed482c4bf0baf57604eff3ccb9 : SUCCESS in 1h 19m 35s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/aa2441a083c64871b56c10ccdf472feb : SUCCESS in 55m 05s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/77d7e49189cc4a7e8a6d15d6b53b8d48 : SUCCESS in 1h 08m 06s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/a2b61c68dbbc427b9d1bc91e4d2bdd4f : SUCCESS in 35m 25s","accounts_in_message":[],"_revision_number":22},{"id":"18234187f3d6fcbb37bae923dcd18eb1870b590a","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-04 06:10:27.000000000","message":"Patch Set 22: Review-Priority+1\n\nrecheck unrelated failure","accounts_in_message":[],"_revision_number":22},{"id":"f099082fe1d52a768d3cb57db186864546cab972","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-04 08:19:29.000000000","message":"Patch Set 22: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/49fa939c145a4ed08fbd290bbfc63fb6 : SUCCESS in 1h 00m 49s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5ed242c70e3d45b19e53b6c324cca58a : SUCCESS in 1h 09m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/419f2d92a3f947f0b418a7d4f46feb8c : SUCCESS in 6m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/626e761a80d44b558a636ad578dccb50 : SUCCESS in 6m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/32194ca4b18a446ebd95b73b12d3c0e3 : SUCCESS in 7m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ffdd2d4d39a84890a6d2bd6716500f9b : SUCCESS in 7m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f96008643475495cb002f79fcb8d14c9 : SUCCESS in 7m 47s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6bb404845c3443f6a54ec30c4b7e4a2b : SUCCESS in 9m 16s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/e41a7538cdc04e90a50e39fdf7ab7cbe : SUCCESS in 8m 29s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/68799d79cc0d4b72bbf89dd0294fd43b : SUCCESS in 7m 47s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/08d3b385a3574bd39e6a5583a454f455 : SUCCESS in 7m 08s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/9bf3fa96e60845ebb46ce1d141de6a18 : SUCCESS in 1h 10m 26s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/90d115f6a92d4624a4f72a7da5c44e2c : SUCCESS in 1h 10m 54s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/5e4aeb400c074632a4873bb0f1e36eef : SUCCESS in 1h 05m 14s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/cc80797029f847669022f1242eb52a06 : SUCCESS in 1h 27m 37s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/87dac7953e7a4032a9207ba2c9a4338f : SUCCESS in 48m 58s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/44b25e5c179b4aa9ac618a0b3940ca51 : SUCCESS in 1h 09m 37s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/1d3b1e9d14af47caa596164da4ac834e : SUCCESS in 34m 50s","accounts_in_message":[],"_revision_number":22},{"id":"7ce113020f3649b0824f95e4243285e0a6a87630","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-08-04 13:54:02.000000000","message":"Uploaded patch set 23: Patch Set 22 was rebased.","accounts_in_message":[],"_revision_number":23},{"id":"50da3822e62cd037b2e3dfd1254fe267ae3fd1ca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-04 15:36:10.000000000","message":"Patch Set 23: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/dc6da2b7c113417eb94ad7bdb2d7597a : SUCCESS in 1h 14m 18s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9116b71a81cd498eaf9513ee8d6541bf : SUCCESS in 1h 17m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b761ef9508c46bb9f90cbb8caf375dc : SUCCESS in 7m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e7cdcfc4291d4a7ab0b83bc6abe4b3c2 : SUCCESS in 5m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1753a681e1f1492ca3a27e39fe8145ac : SUCCESS in 6m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b04e198b8aa749ef8f5281d4028232f4 : SUCCESS in 6m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8de2dfa6c5a54ab1a0dd45a75a9bcdc8 : SUCCESS in 6m 33s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a28f23bb2c9e42ad89b0b112e069d561 : SUCCESS in 8m 53s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b4d7c5a359be4eacb0bfb3a288122156 : SUCCESS in 8m 56s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/8fb4aefd9f0a40a8aaa2ee5f2712c90c : SUCCESS in 8m 22s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/812129dfd93540149f8c12c1d6b2185c : SUCCESS in 5m 09s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/29386c44289c40ec84c52285872505f6 : SUCCESS in 1h 08m 46s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/6419df594e3d4066b748ff20141ee41f : SUCCESS in 1h 13m 53s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/46d1208c81384408af423edc44f96997 : SUCCESS in 1h 07m 00s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/3401b01ac1244f57be20741f0b2c20ac : SUCCESS in 1h 35m 40s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d7a1b0c18ce34ac18b1a964b6b504210 : SUCCESS in 53m 30s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/5c748056187a480bb899ca1e2f7bf21f : POST_FAILURE in 45m 08s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/2eafcd56b164482a9cf4476055c3d9d7 : SUCCESS in 30m 33s","accounts_in_message":[],"_revision_number":23},{"id":"f6cef2d37a4919b70171a667de0a2a1b56e4be8d","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-04 15:38:50.000000000","message":"Patch Set 23:\n\nrecheck unrelated failure","accounts_in_message":[],"_revision_number":23},{"id":"6c236a49f948343444f8f439cc3e171381f07cd5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-04 17:11:00.000000000","message":"Patch Set 23: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/3517ff8b75e94d75af7fdb0ebb9296c2 : SUCCESS in 1h 02m 20s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/db0ed6fac6244a28985cc13b9afc5329 : SUCCESS in 1h 04m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c0ffe0e35ddb46a5b22e4d0e5a7dfff7 : SUCCESS in 6m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c64f8c86af914bf7a39dbf524cf78725 : SUCCESS in 7m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cbb90d3abc0644f4a57f77df80ed0f6e : SUCCESS in 6m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4d6e3bf358f1436c806427c6b31a5a49 : SUCCESS in 7m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/efa145599b204a75aff1219140dc0c02 : SUCCESS in 9m 48s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/edffd0e98b234dd795e627bbb46c1753 : SUCCESS in 8m 41s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/14e1e312b12041f3bb237bf686268b31 : SUCCESS in 9m 07s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/6ae99d20dcc44e0584213c6517f3e3d7 : SUCCESS in 8m 43s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/7bd567a0e8774b95b3d657c78bea10e5 : SUCCESS in 5m 56s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/98a609bd9593459a8d48bebc51ee5d69 : SUCCESS in 1h 07m 12s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/e0a10b812fc4400ba05f07918b089539 : SUCCESS in 1h 05m 32s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/309dd75ca0e74fe0b8a21a57e7f14988 : SUCCESS in 1h 08m 23s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/52f73229cb424f5eb5ca9d810d943cae : SUCCESS in 1h 08m 58s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e306462db94c49969e23bc2691a39d6f : SUCCESS in 1h 08m 57s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/7c25d2f503214d7097a6bfccfb0f281c : SUCCESS in 1h 23m 47s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/64f4924d95cd4c6f82343e5ceaaa1ab8 : SUCCESS in 36m 45s","accounts_in_message":[],"_revision_number":23},{"id":"6861ab52ca1479302100968c410a6fd0f8578efc","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-08-04 20:39:16.000000000","message":"Patch Set 23: Code-Review+1","accounts_in_message":[],"_revision_number":23},{"id":"c86d39c99f9e874653470749a3fb235b2864f896","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-05 05:17:00.000000000","message":"Patch Set 23: Workflow+1\n\nThis is good to go now!!","accounts_in_message":[],"_revision_number":23},{"id":"a7b48fda2d57653ecfc6e94cca029c12a5b98737","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-05 05:17:30.000000000","message":"Patch Set 23: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":23},{"id":"92736b403e122df738c2e91b028aa1c0d3d3c75a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-05 06:39:02.000000000","message":"Patch Set 23: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/9a7f24f7c6c547b5b328784d3ae7c080 : SUCCESS in 1h 04m 11s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/57e3b25dede24ec5a4426ffb388f1624 : SUCCESS in 1h 11m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/57d4742a041b4247857fd9458877d39b : SUCCESS in 5m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/022173c448c140c89ee0b648652b424a : SUCCESS in 5m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/346e81724c4a499dbff1e5f84f18d3fe : SUCCESS in 5m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98767c5922b04e2790228cc55f6327ce : SUCCESS in 6m 25s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/cbb9d3473d6a4c2fa927855e6da85768 : SUCCESS in 8m 29s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/5348e96877494518a7b5ebef7a782b09 : SUCCESS in 9m 46s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/0586eba67a954692846a10c6f22f66c7 : SUCCESS in 1h 18m 02s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/7012c5a66bae4a63a2a5ff2a39484275 : SUCCESS in 1h 05m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a77a4ddca1324ec4a85dc85aa04eb401 : SUCCESS in 47m 06s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/09c6284296d84ae980eec9ec578fd8bd : SUCCESS in 1h 10m 57s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/642c98c325cc4d3a88b0aa1df54b92e1 : SUCCESS in 56m 25s","accounts_in_message":[],"_revision_number":23},{"id":"635374753e6abd8be1eb5fc6889a09890b9cd1d4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-05 06:39:03.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":23},{"id":"777eca1a193bf4155a780f5bb100351df1ad287e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-05 06:41:09.000000000","message":"Patch Set 23:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c75b49e0c3654451815d483e49e22c97 : SUCCESS in 1m 45s","accounts_in_message":[],"_revision_number":23}],"current_revision_number":23,"current_revision":"8ddbdb95268bf923f0b1ba941f11097d17e84eb4","revisions":{"c483d68115ac562afcd3cd469c1ac76cd2a1b770":{"kind":"REWORK","_number":1,"created":"2021-06-11 21:39:01.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/66/796066/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/66/796066/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/796066/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/796066/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/796066/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/66/796066/1"}}},"commit":{"parents":[{"commit":"063ec0f9031a255a620289772bda2a42cd13c5a3","subject":"WIP: Make image update check policy at API layer","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/063ec0f9031a255a620289772bda2a42cd13c5a3"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-06-11 20:21:28.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-06-11 20:51:14.000000000","tz":-420},"subject":"WIP: Make ImageTarget calculate our member","message":"WIP: Make ImageTarget calculate our member\n\nAs noted by lbragstad, we need to make ImageTarget contain a member\nfield so that we can generically apply policies and be able to\nproperly include images for which we are a member. This was hacked\ninto place for ImageRepoProxy.get() but in order to apply it generally\nto listing and other ops, we need to formalize that.\n\nNOTE: This is far from done at the moment; just enough to work on test\nfailures in the next patch.\n\nChange-Id: I92d3792602a69922078d109095ad8ac9afc89d14\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c483d68115ac562afcd3cd469c1ac76cd2a1b770"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c483d68115ac562afcd3cd469c1ac76cd2a1b770"}]},"branch":"refs/heads/master"},"a9f6b708900eb8141364aa5110a187d7d939fc0e":{"kind":"REWORK","_number":2,"created":"2021-06-16 23:59:23.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/66/796066/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/66/796066/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/796066/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/796066/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/796066/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/66/796066/2"}}},"commit":{"parents":[{"commit":"52e7b464d64d605be048a04714e11a2bc7e6f16c","subject":"WIP: Make image update check policy at API layer","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/52e7b464d64d605be048a04714e11a2bc7e6f16c"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-06-11 20:21:28.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-06-16 15:34:34.000000000","tz":-420},"subject":"WIP: Add a member field to Image when appropriate","message":"WIP: Add a member field to Image when appropriate\n\nAs noted by lbragstad, we need to make ImageTarget contain a member\nfield so that we can generically apply policies and be able to\nproperly include images for which we are a member. This was hacked\ninto place for ImageRepoProxy.get() but in order to apply it generally\nto listing and other ops, we need to formalize that.\n\nNOTE: This is far from done at the moment; just enough to work on test\nfailures in the next patch.\n\nChange-Id: I92d3792602a69922078d109095ad8ac9afc89d14\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a9f6b708900eb8141364aa5110a187d7d939fc0e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a9f6b708900eb8141364aa5110a187d7d939fc0e"}]},"branch":"refs/heads/master"},"4949fbf8f2bac64a83f8420b95bc300b0a332aca":{"kind":"REWORK","_number":3,"created":"2021-06-23 17:07:38.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/66/796066/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/66/796066/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/796066/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/796066/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/796066/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/66/796066/3"}}},"commit":{"parents":[{"commit":"1811d2bfbe99dd0654374efde856b922cb15ccb4","subject":"WIP: Make image update check policy at API layer","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1811d2bfbe99dd0654374efde856b922cb15ccb4"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-06-11 20:21:28.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-06-23 17:09:43.000000000","tz":-420},"subject":"WIP: Add a member field to Image when appropriate","message":"WIP: Add a member field to Image when appropriate\n\nAs noted by lbragstad, we need to make ImageTarget contain a member\nfield so that we can generically apply policies and be able to\nproperly include images for which we are a member. This was hacked\ninto place for ImageRepoProxy.get() but in order to apply it generally\nto listing and other ops, we need to formalize that.\n\nNOTE: This is far from done at the moment; just enough to work on test\nfailures in the next patch.\n\nChange-Id: I92d3792602a69922078d109095ad8ac9afc89d14\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4949fbf8f2bac64a83f8420b95bc300b0a332aca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4949fbf8f2bac64a83f8420b95bc300b0a332aca"}]},"branch":"refs/heads/master"},"3e4782edb0a4e95b2196ce337e2c3895a561a560":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2021-06-25 10:47:50.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/66/796066/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/66/796066/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/796066/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/796066/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/796066/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/66/796066/4"}}},"commit":{"parents":[{"commit":"ce264f7968ec4c1e33d63fbc51892de2af3aded5","subject":"WIP: Make image update check policy at API layer","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ce264f7968ec4c1e33d63fbc51892de2af3aded5"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-06-11 20:21:28.000000000","tz":-420},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-06-25 06:54:33.000000000","tz":0},"subject":"WIP: Add a member field to Image when appropriate","message":"WIP: Add a member field to Image when appropriate\n\nAs noted by lbragstad, we need to make ImageTarget contain a member\nfield so that we can generically apply policies and be able to\nproperly include images for which we are a member. This was hacked\ninto place for ImageRepoProxy.get() but in order to apply it generally\nto listing and other ops, we need to formalize that.\n\nNOTE: This is far from done at the moment; just enough to work on test\nfailures in the next patch.\n\nChange-Id: I92d3792602a69922078d109095ad8ac9afc89d14\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3e4782edb0a4e95b2196ce337e2c3895a561a560"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3e4782edb0a4e95b2196ce337e2c3895a561a560"}]},"branch":"refs/heads/master"},"53cb5eeffeb12d19fad2e3203925621c3ab66e0b":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2021-06-28 17:48:24.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/66/796066/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/66/796066/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/796066/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/796066/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/796066/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/66/796066/5"}}},"commit":{"parents":[{"commit":"f8b471ad957ce31f64bf01ccbed1f8dbe8a04455","subject":"POC: Make image update check policy at API layer","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f8b471ad957ce31f64bf01ccbed1f8dbe8a04455"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-06-11 20:21:28.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-06-28 17:50:33.000000000","tz":-420},"subject":"POC: Add a member field to Image when appropriate","message":"POC: Add a member field to Image when appropriate\n\nAs noted by lbragstad, we need to make ImageTarget contain a member\nfield so that we can generically apply policies and be able to\nproperly include images for which we are a member. This was hacked\ninto place for ImageRepoProxy.get() but in order to apply it generally\nto listing and other ops, we need to formalize that.\n\nNOTE: This is far from done at the moment; just enough to work on test\nfailures in the next patch.\n\nChange-Id: I92d3792602a69922078d109095ad8ac9afc89d14\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/53cb5eeffeb12d19fad2e3203925621c3ab66e0b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/53cb5eeffeb12d19fad2e3203925621c3ab66e0b"}]},"branch":"refs/heads/master"},"344fa559bb6ab1b29a455a19621613e2cc17640f":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2021-07-02 14:49:29.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/66/796066/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/66/796066/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/796066/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/796066/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/796066/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/66/796066/6"}}},"commit":{"parents":[{"commit":"3e34eb9702c7fafc7fe94327a819756a79d2d282","subject":"POC: Make image update check policy at API layer","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3e34eb9702c7fafc7fe94327a819756a79d2d282"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-06-11 20:21:28.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-07-02 15:33:55.000000000","tz":-420},"subject":"POC: Add a member field to Image when appropriate","message":"POC: Add a member field to Image when appropriate\n\nAs noted by lbragstad, we need to make ImageTarget contain a member\nfield so that we can generically apply policies and be able to\nproperly include images for which we are a member. This was hacked\ninto place for ImageRepoProxy.get() but in order to apply it generally\nto listing and other ops, we need to formalize that.\n\nNOTE: This is far from done at the moment; just enough to work on test\nfailures in the next patch.\n\nChange-Id: I92d3792602a69922078d109095ad8ac9afc89d14\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/344fa559bb6ab1b29a455a19621613e2cc17640f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/344fa559bb6ab1b29a455a19621613e2cc17640f"}]},"branch":"refs/heads/master"},"278becc2def866ed61ce0ff78bd721f3451737e5":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2021-07-02 15:51:23.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/66/796066/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/66/796066/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/796066/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/796066/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/796066/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/66/796066/7"}}},"commit":{"parents":[{"commit":"93fba05e3c12fc8af0dc11d6f9caf148338b82f7","subject":"POC: Make image update check policy at API layer","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/93fba05e3c12fc8af0dc11d6f9caf148338b82f7"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-06-11 20:21:28.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-07-02 16:51:30.000000000","tz":-420},"subject":"POC: Add a member field to Image when appropriate","message":"POC: Add a member field to Image when appropriate\n\nAs noted by lbragstad, we need to make ImageTarget contain a member\nfield so that we can generically apply policies and be able to\nproperly include images for which we are a member. This was hacked\ninto place for ImageRepoProxy.get() but in order to apply it generally\nto listing and other ops, we need to formalize that.\n\nNOTE: This is far from done at the moment; just enough to work on test\nfailures in the next patch.\n\nChange-Id: I92d3792602a69922078d109095ad8ac9afc89d14\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/278becc2def866ed61ce0ff78bd721f3451737e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/278becc2def866ed61ce0ff78bd721f3451737e5"}]},"branch":"refs/heads/master"},"62a1438eb3c224528c230a3f1dd320a1a88c9678":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2021-07-06 10:51:18.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/66/796066/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/66/796066/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/796066/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/796066/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/796066/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/66/796066/8"}}},"commit":{"parents":[{"commit":"b7baa63ae17e22b91729efc9f6ffc07c71f8f070","subject":"POC: Make image update check policy at API layer","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b7baa63ae17e22b91729efc9f6ffc07c71f8f070"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-06-11 20:21:28.000000000","tz":-420},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-07-05 08:56:54.000000000","tz":0},"subject":"POC: Add a member field to Image when appropriate","message":"POC: Add a member field to Image when appropriate\n\nAs noted by lbragstad, we need to make ImageTarget contain a member\nfield so that we can generically apply policies and be able to\nproperly include images for which we are a member. This was hacked\ninto place for ImageRepoProxy.get() but in order to apply it generally\nto listing and other ops, we need to formalize that.\n\nNOTE: This is far from done at the moment; just enough to work on test\nfailures in the next patch.\n\nChange-Id: I92d3792602a69922078d109095ad8ac9afc89d14\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/62a1438eb3c224528c230a3f1dd320a1a88c9678"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/62a1438eb3c224528c230a3f1dd320a1a88c9678"}]},"branch":"refs/heads/master"},"ac1f23227250f7e1881f9a4b2b77151affb36ac9":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2021-07-06 18:00:34.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/66/796066/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/66/796066/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/796066/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/796066/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/796066/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/66/796066/9"}}},"commit":{"parents":[{"commit":"1732193f19adf9916cad34f61fff10c4fafd0597","subject":"POC: Make image update check policy at API layer","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1732193f19adf9916cad34f61fff10c4fafd0597"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-06-11 20:21:28.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-07-06 18:57:36.000000000","tz":-420},"subject":"POC: Add a member field to Image when appropriate","message":"POC: Add a member field to Image when appropriate\n\nAs noted by lbragstad, we need to make ImageTarget contain a member\nfield so that we can generically apply policies and be able to\nproperly include images for which we are a member. This was hacked\ninto place for ImageRepoProxy.get() but in order to apply it generally\nto listing and other ops, we need to formalize that.\n\nNOTE: This is far from done at the moment; just enough to work on test\nfailures in the next patch.\n\nChange-Id: I92d3792602a69922078d109095ad8ac9afc89d14\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ac1f23227250f7e1881f9a4b2b77151affb36ac9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ac1f23227250f7e1881f9a4b2b77151affb36ac9"}]},"branch":"refs/heads/master"},"0349fb6b7b8e6e284b5bd1560630ee1e9e3d5ba1":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2021-07-14 19:32:52.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/66/796066/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/66/796066/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/796066/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/796066/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/796066/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/66/796066/10"}}},"commit":{"parents":[{"commit":"2b537dc56e5ee941677479320e78d26163c295ec","subject":"POC: Make image update check policy at API layer","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2b537dc56e5ee941677479320e78d26163c295ec"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-06-11 20:21:28.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-07-14 20:28:33.000000000","tz":-420},"subject":"POC: Add a member field to Image when appropriate","message":"POC: Add a member field to Image when appropriate\n\nAs noted by lbragstad, we need to make ImageTarget contain a member\nfield so that we can generically apply policies and be able to\nproperly include images for which we are a member. This was hacked\ninto place for ImageRepoProxy.get() but in order to apply it generally\nto listing and other ops, we need to formalize that.\n\nNOTE: This is far from done at the moment; just enough to work on test\nfailures in the next patch.\n\nChange-Id: I92d3792602a69922078d109095ad8ac9afc89d14\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0349fb6b7b8e6e284b5bd1560630ee1e9e3d5ba1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0349fb6b7b8e6e284b5bd1560630ee1e9e3d5ba1"}]},"branch":"refs/heads/master"},"0a2703d36b46aaf819b00788d272d66eec462ae1":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2021-07-16 13:56:40.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/66/796066/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/66/796066/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/796066/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/796066/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/796066/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/66/796066/11"}}},"commit":{"parents":[{"commit":"d94384f00cedfdcaedd60ee4f19fbaec0ead3dc1","subject":"POC: Make image update check policy at API layer","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d94384f00cedfdcaedd60ee4f19fbaec0ead3dc1"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-06-11 20:21:28.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-07-16 14:58:46.000000000","tz":-420},"subject":"POC: Add a member field to Image when appropriate","message":"POC: Add a member field to Image when appropriate\n\nAs noted by lbragstad, we need to make ImageTarget contain a member\nfield so that we can generically apply policies and be able to\nproperly include images for which we are a member. This was hacked\ninto place for ImageRepoProxy.get() but in order to apply it generally\nto listing and other ops, we need to formalize that.\n\nNOTE: This is far from done at the moment; just enough to work on test\nfailures in the next patch.\n\nChange-Id: I92d3792602a69922078d109095ad8ac9afc89d14\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0a2703d36b46aaf819b00788d272d66eec462ae1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0a2703d36b46aaf819b00788d272d66eec462ae1"}]},"branch":"refs/heads/master"},"13b4b435a7029c18385906554a35f84e0443760a":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2021-07-16 15:28:57.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/66/796066/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/66/796066/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/796066/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/796066/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/796066/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/66/796066/12"}}},"commit":{"parents":[{"commit":"7dc43b61b54637b0c215ba88dbfe1fa4081f4091","subject":"POC: Make image update check policy at API layer","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7dc43b61b54637b0c215ba88dbfe1fa4081f4091"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-06-11 20:21:28.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-07-16 16:31:07.000000000","tz":-420},"subject":"POC: Add a member field to Image when appropriate","message":"POC: Add a member field to Image when appropriate\n\nAs noted by lbragstad, we need to make ImageTarget contain a member\nfield so that we can generically apply policies and be able to\nproperly include images for which we are a member. This was hacked\ninto place for ImageRepoProxy.get() but in order to apply it generally\nto listing and other ops, we need to formalize that.\n\nNOTE: This is far from done at the moment; just enough to work on test\nfailures in the next patch.\n\nChange-Id: I92d3792602a69922078d109095ad8ac9afc89d14\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/13b4b435a7029c18385906554a35f84e0443760a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/13b4b435a7029c18385906554a35f84e0443760a"}]},"branch":"refs/heads/master"},"cec2ff5e5c2721d1a9ebec7aa4b11b620d2ddca1":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2021-07-16 17:49:55.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/66/796066/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/66/796066/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/796066/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/796066/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/796066/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/66/796066/13"}}},"commit":{"parents":[{"commit":"e25d0f49ab0e86e5ad5d989758d80e049cc58212","subject":"Make image update check policy at API layer","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e25d0f49ab0e86e5ad5d989758d80e049cc58212"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-06-11 20:21:28.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-07-16 18:37:28.000000000","tz":-420},"subject":"POC: Add a member field to Image when appropriate","message":"POC: Add a member field to Image when appropriate\n\nAs noted by lbragstad, we need to make ImageTarget contain a member\nfield so that we can generically apply policies and be able to\nproperly include images for which we are a member. This was hacked\ninto place for ImageRepoProxy.get() but in order to apply it generally\nto listing and other ops, we need to formalize that.\n\nNOTE: This is far from done at the moment; just enough to work on test\nfailures in the next patch.\n\nChange-Id: I92d3792602a69922078d109095ad8ac9afc89d14\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/cec2ff5e5c2721d1a9ebec7aa4b11b620d2ddca1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/cec2ff5e5c2721d1a9ebec7aa4b11b620d2ddca1"}]},"branch":"refs/heads/master"},"6dd76743ccce34a35d4a65095bdc2870df9cfe10":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2021-07-16 18:29:40.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/66/796066/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/66/796066/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/796066/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/796066/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/796066/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/66/796066/14"}}},"commit":{"parents":[{"commit":"da567802505b30a53871ef69562fe80ac3d45cd1","subject":"Make image update check policy at API layer","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/da567802505b30a53871ef69562fe80ac3d45cd1"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-06-11 20:21:28.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-07-16 19:31:43.000000000","tz":-420},"subject":"POC: Add a member field to Image when appropriate","message":"POC: Add a member field to Image when appropriate\n\nAs noted by lbragstad, we need to make ImageTarget contain a member\nfield so that we can generically apply policies and be able to\nproperly include images for which we are a member. This was hacked\ninto place for ImageRepoProxy.get() but in order to apply it generally\nto listing and other ops, we need to formalize that.\n\nNOTE: This is far from done at the moment; just enough to work on test\nfailures in the next patch.\n\nChange-Id: I92d3792602a69922078d109095ad8ac9afc89d14\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6dd76743ccce34a35d4a65095bdc2870df9cfe10"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6dd76743ccce34a35d4a65095bdc2870df9cfe10"}]},"branch":"refs/heads/master"},"bc774c266d20570e093cea0013f9b9c97f6b70fa":{"kind":"TRIVIAL_REBASE","_number":15,"created":"2021-07-19 19:37:51.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/66/796066/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/66/796066/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/796066/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/796066/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/796066/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/66/796066/15"}}},"commit":{"parents":[{"commit":"9f37434412bf0fd1f7454d1b8456c0d0c9d6ce4d","subject":"Make image update check policy at API layer","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9f37434412bf0fd1f7454d1b8456c0d0c9d6ce4d"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-06-11 20:21:28.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-07-19 20:40:05.000000000","tz":-420},"subject":"POC: Add a member field to Image when appropriate","message":"POC: Add a member field to Image when appropriate\n\nAs noted by lbragstad, we need to make ImageTarget contain a member\nfield so that we can generically apply policies and be able to\nproperly include images for which we are a member. This was hacked\ninto place for ImageRepoProxy.get() but in order to apply it generally\nto listing and other ops, we need to formalize that.\n\nNOTE: This is far from done at the moment; just enough to work on test\nfailures in the next patch.\n\nChange-Id: I92d3792602a69922078d109095ad8ac9afc89d14\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/bc774c266d20570e093cea0013f9b9c97f6b70fa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/bc774c266d20570e093cea0013f9b9c97f6b70fa"}]},"branch":"refs/heads/master"},"1a740ed0af9753a1c55c2ea44a5e60e395be6068":{"kind":"TRIVIAL_REBASE","_number":16,"created":"2021-07-20 16:30:14.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/66/796066/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/66/796066/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/796066/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/796066/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/796066/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/66/796066/16"}}},"commit":{"parents":[{"commit":"1fd9681388754f9e47fc2d2dcd153a6cd4e408be","subject":"Check get_image(s) in the API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1fd9681388754f9e47fc2d2dcd153a6cd4e408be"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-06-11 20:21:28.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-07-20 17:32:07.000000000","tz":-420},"subject":"POC: Add a member field to Image when appropriate","message":"POC: Add a member field to Image when appropriate\n\nAs noted by lbragstad, we need to make ImageTarget contain a member\nfield so that we can generically apply policies and be able to\nproperly include images for which we are a member. This was hacked\ninto place for ImageRepoProxy.get() but in order to apply it generally\nto listing and other ops, we need to formalize that.\n\nNOTE: This is far from done at the moment; just enough to work on test\nfailures in the next patch.\n\nChange-Id: I92d3792602a69922078d109095ad8ac9afc89d14\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1a740ed0af9753a1c55c2ea44a5e60e395be6068"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1a740ed0af9753a1c55c2ea44a5e60e395be6068"}]},"branch":"refs/heads/master"},"99e131ec562db1ef7646f160c763691f25d796c8":{"kind":"TRIVIAL_REBASE","_number":17,"created":"2021-07-20 18:34:26.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/66/796066/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/66/796066/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/796066/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/796066/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/796066/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/66/796066/17"}}},"commit":{"parents":[{"commit":"46d9f27cadf07bee52c35a082b6e4b256a4c04b7","subject":"Check get_image(s) in the API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/46d9f27cadf07bee52c35a082b6e4b256a4c04b7"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-06-11 20:21:28.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-07-20 19:36:42.000000000","tz":-420},"subject":"POC: Add a member field to Image when appropriate","message":"POC: Add a member field to Image when appropriate\n\nAs noted by lbragstad, we need to make ImageTarget contain a member\nfield so that we can generically apply policies and be able to\nproperly include images for which we are a member. This was hacked\ninto place for ImageRepoProxy.get() but in order to apply it generally\nto listing and other ops, we need to formalize that.\n\nNOTE: This is far from done at the moment; just enough to work on test\nfailures in the next patch.\n\nChange-Id: I92d3792602a69922078d109095ad8ac9afc89d14\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/99e131ec562db1ef7646f160c763691f25d796c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/99e131ec562db1ef7646f160c763691f25d796c8"}]},"branch":"refs/heads/master"},"60be8298ec6aecc5da8f6c2d6d02bf80811cdb27":{"kind":"TRIVIAL_REBASE","_number":18,"created":"2021-07-26 19:49:00.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/66/796066/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/66/796066/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/796066/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/796066/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/796066/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/66/796066/18"}}},"commit":{"parents":[{"commit":"cf3995cebf4f89e6bff68380adfc7cd225f9285f","subject":"Check get_image(s) in the API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/cf3995cebf4f89e6bff68380adfc7cd225f9285f"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-06-11 20:21:28.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-07-26 20:36:24.000000000","tz":-420},"subject":"POC: Add a member field to Image when appropriate","message":"POC: Add a member field to Image when appropriate\n\nAs noted by lbragstad, we need to make ImageTarget contain a member\nfield so that we can generically apply policies and be able to\nproperly include images for which we are a member. This was hacked\ninto place for ImageRepoProxy.get() but in order to apply it generally\nto listing and other ops, we need to formalize that.\n\nNOTE: This is far from done at the moment; just enough to work on test\nfailures in the next patch.\n\nChange-Id: I92d3792602a69922078d109095ad8ac9afc89d14\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/60be8298ec6aecc5da8f6c2d6d02bf80811cdb27"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/60be8298ec6aecc5da8f6c2d6d02bf80811cdb27"}]},"branch":"refs/heads/master"},"192b275383c1112808d8160e87d2b6be6bf8d87b":{"kind":"TRIVIAL_REBASE","_number":19,"created":"2021-07-27 18:34:02.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/66/796066/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/66/796066/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/796066/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/796066/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/796066/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/66/796066/19"}}},"commit":{"parents":[{"commit":"76cef3fb92dbb0fcf785cfe70b31e3de2af4fa86","subject":"Check get_image(s) in the API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/76cef3fb92dbb0fcf785cfe70b31e3de2af4fa86"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-06-11 20:21:28.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-07-27 19:26:04.000000000","tz":-420},"subject":"POC: Add a member field to Image when appropriate","message":"POC: Add a member field to Image when appropriate\n\nAs noted by lbragstad, we need to make ImageTarget contain a member\nfield so that we can generically apply policies and be able to\nproperly include images for which we are a member. This was hacked\ninto place for ImageRepoProxy.get() but in order to apply it generally\nto listing and other ops, we need to formalize that.\n\nNOTE: This is far from done at the moment; just enough to work on test\nfailures in the next patch.\n\nChange-Id: I92d3792602a69922078d109095ad8ac9afc89d14\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/192b275383c1112808d8160e87d2b6be6bf8d87b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/192b275383c1112808d8160e87d2b6be6bf8d87b"}]},"branch":"refs/heads/master"},"bdfbcaf0883112ce0d94b7f0bb0363684e381690":{"kind":"REWORK","_number":20,"created":"2021-07-28 15:13:52.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/66/796066/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/66/796066/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/796066/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/796066/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/796066/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/66/796066/20"}}},"commit":{"parents":[{"commit":"76cef3fb92dbb0fcf785cfe70b31e3de2af4fa86","subject":"Check get_image(s) in the API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/76cef3fb92dbb0fcf785cfe70b31e3de2af4fa86"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-06-11 20:21:28.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-07-28 16:14:03.000000000","tz":-420},"subject":"Add a member field to Image when appropriate","message":"Add a member field to Image when appropriate\n\nAs noted by lbragstad, we need to make ImageTarget contain a member\nfield so that we can generically apply policies and be able to\nproperly include images for which we are a member. This was hacked\ninto place for ImageRepoProxy.get() but in order to apply it generally\nto listing and other ops, we need to formalize that.\n\nPartially-Implements: blueprint policy-refactor\n\nChange-Id: I92d3792602a69922078d109095ad8ac9afc89d14\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/bdfbcaf0883112ce0d94b7f0bb0363684e381690"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/bdfbcaf0883112ce0d94b7f0bb0363684e381690"}]},"branch":"refs/heads/master"},"32db51d203136d78997ccf57eb9461cae4a54ecd":{"kind":"TRIVIAL_REBASE","_number":21,"created":"2021-08-02 17:22:08.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/66/796066/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/66/796066/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/796066/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/796066/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/796066/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/66/796066/21"}}},"commit":{"parents":[{"commit":"b14ddca049e7a5210a4941284ea0dd97a8e3158d","subject":"Check get_image(s) in the API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b14ddca049e7a5210a4941284ea0dd97a8e3158d"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-06-11 20:21:28.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-08-02 18:23:38.000000000","tz":-420},"subject":"Add a member field to Image when appropriate","message":"Add a member field to Image when appropriate\n\nAs noted by lbragstad, we need to make ImageTarget contain a member\nfield so that we can generically apply policies and be able to\nproperly include images for which we are a member. This was hacked\ninto place for ImageRepoProxy.get() but in order to apply it generally\nto listing and other ops, we need to formalize that.\n\nPartially-Implements: blueprint policy-refactor\n\nChange-Id: I92d3792602a69922078d109095ad8ac9afc89d14\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/32db51d203136d78997ccf57eb9461cae4a54ecd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/32db51d203136d78997ccf57eb9461cae4a54ecd"}]},"branch":"refs/heads/master"},"9611dcc8ffb8a518ae1c12f1519943e96e48f97f":{"kind":"TRIVIAL_REBASE","_number":22,"created":"2021-08-03 18:36:00.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/66/796066/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/66/796066/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/796066/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/796066/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/796066/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/66/796066/22"}}},"commit":{"parents":[{"commit":"1c3e1003e7b3b84a699ad6550b6fc52a07d8974c","subject":"Check get_image(s) in the API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1c3e1003e7b3b84a699ad6550b6fc52a07d8974c"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-06-11 20:21:28.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-08-03 19:04:22.000000000","tz":-420},"subject":"Add a member field to Image when appropriate","message":"Add a member field to Image when appropriate\n\nAs noted by lbragstad, we need to make ImageTarget contain a member\nfield so that we can generically apply policies and be able to\nproperly include images for which we are a member. This was hacked\ninto place for ImageRepoProxy.get() but in order to apply it generally\nto listing and other ops, we need to formalize that.\n\nPartially-Implements: blueprint policy-refactor\n\nChange-Id: I92d3792602a69922078d109095ad8ac9afc89d14\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9611dcc8ffb8a518ae1c12f1519943e96e48f97f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9611dcc8ffb8a518ae1c12f1519943e96e48f97f"}]},"branch":"refs/heads/master"},"8ddbdb95268bf923f0b1ba941f11097d17e84eb4":{"kind":"TRIVIAL_REBASE","_number":23,"created":"2021-08-04 13:54:02.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/66/796066/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/66/796066/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/796066/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/796066/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/796066/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/66/796066/23"}}},"commit":{"parents":[{"commit":"ba37ea322777bb1efb210f4cbb72cae2f2c541d2","subject":"Check get_image(s) in the API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ba37ea322777bb1efb210f4cbb72cae2f2c541d2"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-06-11 20:21:28.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-08-04 14:56:05.000000000","tz":-420},"subject":"Add a member field to Image when appropriate","message":"Add a member field to Image when appropriate\n\nAs noted by lbragstad, we need to make ImageTarget contain a member\nfield so that we can generically apply policies and be able to\nproperly include images for which we are a member. This was hacked\ninto place for ImageRepoProxy.get() but in order to apply it generally\nto listing and other ops, we need to formalize that.\n\nPartially-Implements: blueprint policy-refactor\n\nChange-Id: I92d3792602a69922078d109095ad8ac9afc89d14\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8ddbdb95268bf923f0b1ba941f11097d17e84eb4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8ddbdb95268bf923f0b1ba941f11097d17e84eb4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
