)]}'
{"id":"openstack%2Fglance~797117","triplet_id":"openstack%2Fglance~master~If8798525aff5195e77f451dfce6c45cfaf16eaf4","project":"openstack/glance","branch":"master","topic":"docs/group-management","attention_set":{},"removed_from_attention_set":{"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2022-07-07 16:53:08.000000000","reason":"Change was submitted"},"4393":{"account":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"last_update":"2022-07-07 16:53:08.000000000","reason":"Change was submitted"},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2022-05-30 20:23:31.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"If8798525aff5195e77f451dfce6c45cfaf16eaf4","subject":"Doc: Glance group management","status":"MERGED","created":"2021-06-18 14:25:14.000000000","updated":"2022-07-07 16:54:15.000000000","submitted":"2022-07-07 16:53:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"797117-docs/group-management","meta_rev_id":"15623663cc6fabeb1ae7d74369470a6254bdbc7f","_number":797117,"virtual_id_number":797117,"owner":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-07 16:53:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"recommended":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"all":[{"value":1,"date":"2022-05-31 12:06:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":2,"date":"2022-05-30 20:23:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":1,"date":"2022-07-07 15:31:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"}],"CC":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-18 14:31:44.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"CC"},{"updated":"2021-06-18 15:58:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-18 15:00:19.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"CC"},{"updated":"2022-05-30 20:23:11.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2022-05-31 12:06:05.000000000","updated_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"reviewer":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"state":"REVIEWER"},{"updated":"2022-07-01 06:40:01.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"}],"messages":[{"id":"7e7313c15f9d040fc92dd9c865a8359601fbf5a3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-06-18 14:25:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"df8e135fb500c313898c183e4bef815b3957fe13","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2021-06-18 14:31:44.000000000","message":"Patch Set 1:\n\nGreat to get these things explicitly written down.\n\nNot sure if sphinx is going to like having the two TOC entries for the same page. Will wait to see what the job reports. But assuming that is happy, this looks good to me.","accounts_in_message":[],"_revision_number":1},{"id":"e00384d9b64a1d328e40a2b9d5205856ec74605e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-18 15:58:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd3db5e77c164460b4c4c7f51ee033eb : SUCCESS in 5m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38115592679341ac937157e3b8c56f2e : SUCCESS in 7m 19s","accounts_in_message":[],"_revision_number":1},{"id":"2bf70ec0a639dc6f578222163d8996d2495503bc","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2022-05-18 15:00:19.000000000","message":"Patch Set 1:\n\nThis is nice, but the links are very long and not always broken down in the HTML output, thus forcing other columns to become very small and difficult to read. Could you use pretty links in the last column?","accounts_in_message":[],"_revision_number":1},{"id":"ba4af206e387d59f2a62bf45b1a9187ff3ff95e9","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-05-23 06:04:37.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6c1d86e27e622d55ddf3cef02e4524c7939412d2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-05-30 14:09:42.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"4986ddddc51962a2f4a64911b738cc82b3e8c999","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2022-05-30 14:20:56.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1d1029d0b83177341349568b6819073c8891d891","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-30 14:21:23.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1457c59635784ff79a363f9ea749d670 : SUCCESS in 5m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/108689af7c8141d6a2f061ace58ebd38 : SUCCESS in 8m 01s","accounts_in_message":[],"_revision_number":2},{"id":"4621fdd920d6f76c73faf31c79848795ae5283e5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-05-30 14:30:57.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"38a72d1f710f7ba8b7a456cc873c396152d71870","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-30 14:39:14.000000000","message":"Patch Set 3: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b2555e6840ef4ffca7aa89d5375edc4c : FAILURE in 5m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7861687c83540f2a8de6a2a5d375e60 : FAILURE in 5m 37s","accounts_in_message":[],"_revision_number":3},{"id":"bb721a3341bda17970f516ad935912f29a02968e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-05-30 14:44:13.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"9a772f04a6bab8613afcc7307a45449d7db9b118","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-30 14:58:51.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d183b15e5cd14d38b63565b5a1b69311 : SUCCESS in 5m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ae3ea39f5f545a38c6170e8742b2044 : SUCCESS in 8m 34s","accounts_in_message":[],"_revision_number":4},{"id":"722ec8e2155285d5b4408739980e613348eb56e2","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2022-05-30 20:23:11.000000000","message":"Patch Set 1: Code-Review+2\n\nNow it looks amazing, thanks!","accounts_in_message":[],"_revision_number":1},{"id":"828f9bc54aba14e7ad2ab47c6fafeb24bb84e53c","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2022-05-30 20:23:31.000000000","message":"Patch Set 4: Code-Review+2\n\nNow it looks amazing, thanks!","accounts_in_message":[],"_revision_number":4},{"id":"e898ad111650530abb18df2a3475bb8321708a1a","author":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"date":"2022-05-31 12:06:05.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"7c8ed2ec9db97e88c2e11560df6a11f06c929138","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2022-07-01 13:37:13.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"fc6d24ad17991c7f2d657da5240c5be9677d4386","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-07-01 14:09:39.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"bbd38c460b6e22a989ae414c2db509eec3eeafbe","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2022-07-07 15:31:08.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"9a74bd63366da32c321e0c79cc478e66c37489bf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-07 15:31:46.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"0f602bc0c4a11e67f32dc935effeb723e67c0e52","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-07 16:53:08.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c8f46291c15c43faad8b8af190959373\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/644e7d8539304bea9690aee5fbae7c51 : SUCCESS in 4m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/124720945d804d598ae997daa5cb2483 : SUCCESS in 5m 44s","accounts_in_message":[],"_revision_number":4},{"id":"a304f2d06674a193190534e6d142bb1f11afb9d9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-07 16:53:08.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"15623663cc6fabeb1ae7d74369470a6254bdbc7f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-07 16:54:15.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a862e55bd1564aeaa8d7563d7e767bce\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b2cf433d6ce4bbd8615b07129c306fa : SUCCESS in 45s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"e1e8e9de401d083e8565ff5440fe4873105120c5","revisions":{"1b6d837184139bb13beafedf78f0cd3b96c956f0":{"kind":"REWORK","_number":1,"created":"2021-06-18 14:25:14.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/17/797117/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/17/797117/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/17/797117/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/17/797117/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/17/797117/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/17/797117/1"}}},"commit":{"parents":[{"commit":"049a8cd447be442efe294347b0ad440939aef31b","subject":"Merge \"Allow member creation when using db.simple api\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/049a8cd447be442efe294347b0ad440939aef31b"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-06-18 14:23:29.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-06-18 14:23:29.000000000","tz":0},"subject":"Doc: Glance group management","message":"Doc: Glance group management\n\nChange-Id: If8798525aff5195e77f451dfce6c45cfaf16eaf4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1b6d837184139bb13beafedf78f0cd3b96c956f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1b6d837184139bb13beafedf78f0cd3b96c956f0"}]},"branch":"refs/heads/master"},"1ee4b0cda24979f1d9dfcf48255613cbc8c85d52":{"kind":"REWORK","_number":2,"created":"2022-05-30 14:09:42.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/17/797117/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/17/797117/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/17/797117/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/17/797117/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/17/797117/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/17/797117/2"}}},"commit":{"parents":[{"commit":"049a8cd447be442efe294347b0ad440939aef31b","subject":"Merge \"Allow member creation when using db.simple api\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/049a8cd447be442efe294347b0ad440939aef31b"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-06-18 14:23:29.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2022-05-30 14:09:42.000000000","tz":0},"subject":"Doc: Glance group management","message":"Doc: Glance group management\n\nChange-Id: If8798525aff5195e77f451dfce6c45cfaf16eaf4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1ee4b0cda24979f1d9dfcf48255613cbc8c85d52"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1ee4b0cda24979f1d9dfcf48255613cbc8c85d52"}]},"branch":"refs/heads/master"},"dc0528cbd1de050875f74e4764f88e64439af2f6":{"kind":"REWORK","_number":3,"created":"2022-05-30 14:30:57.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/17/797117/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/17/797117/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/17/797117/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/17/797117/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/17/797117/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/17/797117/3"}}},"commit":{"parents":[{"commit":"049a8cd447be442efe294347b0ad440939aef31b","subject":"Merge \"Allow member creation when using db.simple api\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/049a8cd447be442efe294347b0ad440939aef31b"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-06-18 14:23:29.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2022-05-30 14:30:57.000000000","tz":0},"subject":"Doc: Glance group management","message":"Doc: Glance group management\n\nChange-Id: If8798525aff5195e77f451dfce6c45cfaf16eaf4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/dc0528cbd1de050875f74e4764f88e64439af2f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/dc0528cbd1de050875f74e4764f88e64439af2f6"}]},"branch":"refs/heads/master"},"e1e8e9de401d083e8565ff5440fe4873105120c5":{"kind":"REWORK","_number":4,"created":"2022-05-30 14:44:13.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/17/797117/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/17/797117/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/17/797117/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/17/797117/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/17/797117/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/17/797117/4"}}},"commit":{"parents":[{"commit":"049a8cd447be442efe294347b0ad440939aef31b","subject":"Merge \"Allow member creation when using db.simple api\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/049a8cd447be442efe294347b0ad440939aef31b"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-06-18 14:23:29.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2022-05-30 14:44:12.000000000","tz":0},"subject":"Doc: Glance group management","message":"Doc: Glance group management\n\nChange-Id: If8798525aff5195e77f451dfce6c45cfaf16eaf4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e1e8e9de401d083e8565ff5440fe4873105120c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e1e8e9de401d083e8565ff5440fe4873105120c5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
