)]}'
{"id":"openstack%2Fglance~799496","triplet_id":"openstack%2Fglance~master~I37d67457efdd8572d41cd839fb00c84f1f197892","project":"openstack/glance","branch":"master","topic":"bug/1934673","hashtags":[],"change_id":"I37d67457efdd8572d41cd839fb00c84f1f197892","subject":"Fix the policy deprecation message","status":"MERGED","created":"2021-07-05 13:41:33.000000000","updated":"2021-07-06 04:18:45.000000000","submitted":"2021-07-06 04:17:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"799496-bug/1934673","meta_rev_id":"f3bc27cd105952f4f347c8add26456ce13b0f7f1","_number":799496,"virtual_id_number":799496,"owner":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-06 04:17:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2021-07-05 17:55:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2021-07-06 00:57:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":1,"date":"2021-07-06 00:57:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]}],"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-05 15:08:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-05 17:37:02.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-07-05 17:55:17.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2021-07-06 00:57:13.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"}],"messages":[{"id":"3e15f22c22f833b42e32bf0aefccf6c6556d9f7c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2021-07-05 13:41:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9896c068072acd2c8b1526fb776222ddb30ea1ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-05 15:08:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/11f3c14af6994c89b229de9036739b7c : SUCCESS in 1h 13m 09s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/05d30fa830ac45389a642eae03aad0ec : SUCCESS in 1h 16m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f80f1500a03d4e38bd344afb6d3ee698 : SUCCESS in 6m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c6a413de269547acae164c3c5045dbfd : SUCCESS in 6m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ad50c5feb4d5451d8c96a6cac0985ca6 : SUCCESS in 6m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1a224bf8384c42b9aeb76052d15e3955 : SUCCESS in 6m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ff386e362134295ad0eb1183f39273e : SUCCESS in 6m 25s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4ebb6260735b47de8084386e316d5193 : SUCCESS in 10m 04s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/64b5417ed2214923aba7fa3b56f835dc : SUCCESS in 9m 16s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/497c4cb3fbff4031a66dfbaaca323abc : FAILURE in 8m 07s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/8020e64856004fbe8d8a989d41f5fe38 : SUCCESS in 5m 39s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/ca50dbb98e6848e28c49dbdea2936f6b : SUCCESS in 1h 05m 27s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/9ab0224a2bb64474b2dcd9877e598263 : SUCCESS in 1h 02m 05s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/e9dbddd2c9334f3997cf33482630e357 : SUCCESS in 1h 08m 06s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/ed827c1a700c49059895c513c614a552 : SUCCESS in 1h 21m 01s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6aaf93ecbf5c41979ff5e6b1a4788f23 : SUCCESS in 51m 19s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/6b1c7c91a4434d03b84d943c371b5939 : SUCCESS in 1h 19m 36s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/b300da9aa7964718afeec0805e8ffb8f : SUCCESS in 33m 09s","accounts_in_message":[],"_revision_number":1},{"id":"5df2e2c3b17f997c9ef80bc6166c92f2328fa56d","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-07-05 17:37:02.000000000","message":"Patch Set 1:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/799496/799496/1/check-not-vote/ext-glance-zuul/19be98e : SUCCESS in 37m 16s","accounts_in_message":[],"_revision_number":1},{"id":"6adf608be5d1e52f935bc69aa151441c5ed39fd8","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-07-05 17:55:17.000000000","message":"Patch Set 1: Code-Review+2\n\nGood catch!!","accounts_in_message":[],"_revision_number":1},{"id":"54796207586ed3d479ba6cb6161a37cd5e8a531c","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2021-07-06 00:57:13.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nOnly one line to review, perfect!","accounts_in_message":[],"_revision_number":1},{"id":"a514ae02b3dbcd22823fe5db443f5e46599ddd42","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-06 00:57:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"37af28021b92478a9cb4bb9df956b68205cf9c04","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-06 04:17:12.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/6bd5fe652603464aa80e53391f457b13 : SUCCESS in 1h 04m 57s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7af03ee58b31490d904670760b097520 : SUCCESS in 1h 17m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8be306c0a5eb4af39c5f8dcf8e3136f7 : SUCCESS in 5m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9a9499776f884c988a9e2abdab61b0ce : SUCCESS in 6m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dadbe483c3b044369d9ef964198f30ef : SUCCESS in 6m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13cc04cdea1b473dbe1040a1e979a4a5 : SUCCESS in 6m 21s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/781a043b92bf45fea4bdf191d122b5b8 : SUCCESS in 8m 44s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/f23d80839caa4ec2ab1a35ec18c7a524 : SUCCESS in 9m 11s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/c6d9e2a743a04459bdf387d6fb5d69fb : SUCCESS in 1h 07m 08s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/793a3ad1891548e18e3a1e050affab18 : SUCCESS in 1h 19m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f007ac3304a54cc6b188c14a8d14258a : SUCCESS in 51m 59s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/c8cca8fa55964defb8433ad5c96340b8 : SUCCESS in 1h 15m 40s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/60e51c4f52184c56bcc201f5ee2ed369 : SUCCESS in 33m 41s","accounts_in_message":[],"_revision_number":1},{"id":"d3854bc36f8bcbcb304646e7869ad7bb03f458a9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-06 04:17:13.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"f3bc27cd105952f4f347c8add26456ce13b0f7f1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-06 04:18:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d43f054e9b714f0dbe3683aaeefea253 : SUCCESS in 1m 15s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2d2817375296c7eff5594a092a8463dd31dbf4e4","revisions":{"2d2817375296c7eff5594a092a8463dd31dbf4e4":{"kind":"REWORK","_number":1,"created":"2021-07-05 13:41:33.000000000","uploader":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"ref":"refs/changes/96/799496/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/96/799496/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/96/799496/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/96/799496/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/96/799496/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/96/799496/1"}}},"commit":{"parents":[{"commit":"da771fce258585d60393bdcfc2f910070bf2af63","subject":"Add unified quotas documentation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/da771fce258585d60393bdcfc2f910070bf2af63"}]}],"author":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2021-07-05 13:39:48.000000000","tz":60},"committer":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2021-07-05 13:41:19.000000000","tz":60},"subject":"Fix the policy deprecation message","message":"Fix the policy deprecation message\n\nRemoved the defaulting to roles part of the deprecation message.\n\nChange-Id: I37d67457efdd8572d41cd839fb00c84f1f197892\nCloses-Bug: #1934673\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2d2817375296c7eff5594a092a8463dd31dbf4e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2d2817375296c7eff5594a092a8463dd31dbf4e4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
