)]}'
{"id":"openstack%2Fglance~799632","triplet_id":"openstack%2Fglance~master~If9e23cbded044b3600960a847c17737d5d23a451","project":"openstack/glance","branch":"master","topic":"policy-refactor","hashtags":[],"change_id":"If9e23cbded044b3600960a847c17737d5d23a451","subject":"Refactor gateway auth layer for metadef APIs","status":"MERGED","created":"2021-07-06 10:51:18.000000000","updated":"2021-08-10 17:52:57.000000000","submitted":"2021-08-10 17:48:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"799632-policy-refactor","meta_rev_id":"fb474dc281c1c4fd08ddaf88d9ece740cece070e","_number":799632,"virtual_id_number":799632,"owner":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-10 17:48:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"all":[{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2021-08-10 15:40:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":2,"date":"2021-08-09 19:44:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"all":[{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":1,"date":"2021-08-10 15:40:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]}],"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-06 12:40:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-06 18:00:34.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2021-07-22 18:41:10.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-07-29 09:35:08.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"state":"REVIEWER"},{"updated":"2021-08-09 19:44:48.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"}],"messages":[{"id":"0bbfc7810b00effcc18ada1a1f79244031fe3925","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-07-06 10:51:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"32f54626a3a538bc8130aec776e0838ffbb6e078","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-06 12:40:17.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/b4f0ba6734624e7d93f3674a54f3dde9 : SUCCESS in 1h 11m 27s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/97cc318fe8a2415f80c6f7170b78ec33 : SUCCESS in 1h 09m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/44a67faec8b842e3b0c088fce768bb63 : SUCCESS in 7m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eb769df586af4555ad909049b27531ef : SUCCESS in 9m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e85f612e914c42e2a42c6ce18c6d2e4f : SUCCESS in 9m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3680fe48a311401a911f85d9eeaaab53 : SUCCESS in 9m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b990fe69033542fd9a8a2509a9acaf3a : SUCCESS in 7m 38s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f8c03eb1c2994d5e94a4c5577a5aaa2c : FAILURE in 9m 57s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/952efc82147f441683cbc6c273700471 : FAILURE in 9m 29s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/a2455bd895284fe4a913dfab0f9480d5 : FAILURE in 10m 42s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/6de2238356274dcc9eb7c8d4af513569 : SUCCESS in 6m 37s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/67ff4a34253542d0809405225593b5b9 : SUCCESS in 1h 06m 54s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/992646b028974e0f86bc1705c81c98e6 : SUCCESS in 1h 01m 49s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/6af2ff1939724aacbda2035935a89d2c : SUCCESS in 1h 00m 03s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/78513d200ae34cd0b70aa72d1ef9af3f : SUCCESS in 1h 09m 40s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/066a227b740b4eea8a36cf5c5f93484f : SUCCESS in 44m 08s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/bd193a5ae4384490a0eda8d2b1107e15 : SUCCESS in 1h 31m 12s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/ea867f8c95244aa0914d06184b6f50ca : SUCCESS in 48m 17s","accounts_in_message":[],"_revision_number":1},{"id":"6dcda9e9f9450e7648e7a062dc5a0bfd53b41226","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-07-06 18:00:34.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"38ff3bafe05ec5d7abb72e11e10648711a1aed37","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-06 19:28:48.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/be94fc3ed5564f358c4dc4cadc5c6ba8 : SUCCESS in 1h 01m 27s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/47b6830e0de84be28de64f2bd065ffab : SUCCESS in 1h 02m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e1caf5151344cf6a65714b80147ae52 : SUCCESS in 7m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cf45436f8c744cf68d82cfd42a7a5642 : SUCCESS in 6m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0e793b2ea40a4b13b235cb41d43d2ed9 : SUCCESS in 6m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a7571f9b8558441db377d8b38345bd4c : SUCCESS in 7m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/728980b610874b37b49c3aeb59e23f1d : SUCCESS in 6m 54s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3f52d43d929d4ca1b843f5bee3296c8b : SUCCESS in 9m 50s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/ef890ae8c42d44f59225601c0775d770 : SUCCESS in 10m 16s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/cc2c67914b834f7b9c534a0f79177f79 : FAILURE in 8m 56s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/4f1c14786ae8438eb14fcb71177a5801 : SUCCESS in 6m 24s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/a9b8742d16fb40b3b56ce328249c92f4 : SUCCESS in 1h 15m 49s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/3e12ba689329427f9493c34816c6b121 : SUCCESS in 1h 08m 31s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/9916460cb8ab4f5794514f4e1a162777 : SUCCESS in 1h 06m 33s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/595d375ba8cc4b10802b24775cde1233 : SUCCESS in 1h 09m 03s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/40b39eebca0744bd8d4454b3266c1d69 : SUCCESS in 53m 25s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/bfb1538951604c9fa67f543b985694c2 : SUCCESS in 1h 23m 02s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/08f7e8e825c14fe9bbe955aa257ebf76 : SUCCESS in 37m 20s","accounts_in_message":[],"_revision_number":2},{"id":"d63ed48cddc7b1e88818f83c38ad134b32f024b6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-07-14 19:32:52.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"c03acbc26e8adbced5669e4223010b5a69d709b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-14 21:03:10.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/21de3e019bc84e06836a500254f641dd : SUCCESS in 1h 02m 31s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2d2d3147e5c143b0b3125e2b861289a0 : SUCCESS in 1h 03m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8103473e944b4294abaeca03ae3d6e91 : SUCCESS in 5m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e8341dbe6acb46b9a552c4c8ee02c257 : SUCCESS in 7m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7715993609ad41a4959ea7eedacfe798 : SUCCESS in 6m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4bdba52bb2574edcbb1681490848e3bb : SUCCESS in 7m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33c5080cee9d4a068377811f103606c6 : SUCCESS in 6m 34s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c39e045f4fed44f39dba811af60db119 : SUCCESS in 8m 47s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/f892c348f33040429f3f0d7aec40eb4e : SUCCESS in 10m 00s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/4e8fe5d8379141099246e8026f09d196 : SUCCESS in 9m 18s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/08afd355d0ae414184ca409c31854e40 : SUCCESS in 5m 38s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/5d7cfda50a834946b9b6eac56d4d0325 : SUCCESS in 1h 25m 49s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/08f60101e643410780346909eddb1c50 : SUCCESS in 1h 03m 44s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/f83c9660268843b585a37b9a6984d9d2 : SUCCESS in 59m 23s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/9e595ea1b2c249d680546cc5dda868cf : SUCCESS in 1h 15m 46s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c92efd99e471459f980198add35c7f34 : SUCCESS in 40m 05s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/5808ce14765946dcbd444b43e90ca2af : SUCCESS in 1h 14m 42s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/20a972eccc2e482282a2b092e5348607 : SUCCESS in 28m 47s","accounts_in_message":[],"_revision_number":3},{"id":"86278a0b28dbc1e6ed7f58e5c7d516b51e84671f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-07-16 13:56:40.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"fc15ad8e261b0b8713cc07bd33337d4470ac20e6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-07-16 15:28:57.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"c2c6f2b3ea787cc901a4647a5fdebb9d63d1f446","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-16 15:29:51.000000000","message":"Patch Set 4: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":4},{"id":"22540b2ef91d1d1602f74a603ab79b4035021dca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-16 16:59:55.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/51297038b8f14528b8264905ea8aa165 : SUCCESS in 1h 08m 52s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/cfd47540bcc44cf5a1121143fced7878 : SUCCESS in 1h 15m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/40d5d02c292940239fec9319a2d93c27 : SUCCESS in 6m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/62bbfe79040a471597ac15b8d164bebc : SUCCESS in 7m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d88c7df540d849c1b3ed1afa35392dc8 : SUCCESS in 7m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/53996dc289764d8791dcf1836d4567be : SUCCESS in 7m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/abd5813472044906b37d1f443a2d3824 : SUCCESS in 7m 59s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0076fdb4c17a4aa28edb0e35e4e16d20 : SUCCESS in 10m 02s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c77fde9064bf41e986d859badef867eb : SUCCESS in 10m 08s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/1401ddb427f84fb9a3c26d2251c2cf05 : SUCCESS in 10m 57s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/045c2ec5d7a24e89ae32f2c3ee2b95f5 : SUCCESS in 7m 16s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/9f8e629bde5744eea95876f15a962143 : SUCCESS in 1h 17m 57s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/cd081175f7c04b69ad316cba97e2ca9a : SUCCESS in 1h 26m 11s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/25fe6c8b7a8243e78dd654f06fe19957 : SUCCESS in 1h 04m 51s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/8d40ac424a58407dbde1d1f7113b0f54 : SUCCESS in 1h 09m 24s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d655b5301852466a81c2345300846d49 : SUCCESS in 55m 15s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/724e0bee0b634835a8cfadf06deb68b7 : SUCCESS in 1h 19m 05s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/0a2c8ac8d2294175a86336ae9b3de43d : SUCCESS in 37m 21s","accounts_in_message":[],"_revision_number":5},{"id":"5971554b1563a9c5f566eddae2a7e1b3a9924185","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-07-16 19:55:49.000000000","message":"Abandoned\n\nIncluded changes in this patch, https://review.opendev.org/c/openstack/glance/+/799633","accounts_in_message":[],"_revision_number":5},{"id":"c9182e0a0d64c7fd525180976a9a021630ef6674","tag":"autogenerated:gerrit:restore","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-07-22 09:45:46.000000000","message":"Restored","accounts_in_message":[],"_revision_number":5},{"id":"ae5ac73fd57a080ef6958a9c2cc064ce7c7b2b66","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-07-22 09:46:07.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"dac6867c661aa78aa62b3dad6af1e8f27304e673","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-22 09:46:20.000000000","message":"Patch Set 5: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":5},{"id":"133091b779be8415b20f880b6f51aace7e0dd169","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-07-22 09:48:09.000000000","message":"Topic resource-policy-refactor removed","accounts_in_message":[],"_revision_number":6},{"id":"8777bfc2eb944574fd424d03847656009ff5058a","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-07-22 09:48:12.000000000","message":"Topic set to policy-refactor","accounts_in_message":[],"_revision_number":6},{"id":"93445e07e37a32589befae80e4dff3e99f73dfd1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-22 11:01:01.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/725a08ad876d4dbdb856616c2dfbdce2 : SUCCESS in 1h 00m 09s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a20b80a981ea4c1cbb35f3493d402226 : SUCCESS in 58m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/772331bc87da49358328f69293a6557a : SUCCESS in 5m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2473925008f7425ebf02738096f7b698 : SUCCESS in 5m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/16300d2a52194499a60c3f769b419cb5 : SUCCESS in 5m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a222e94a20dc4a6d80a94696dcf04266 : SUCCESS in 6m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a1f1faca56244af8847e08cf4e95b7f : SUCCESS in 6m 09s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ccb231941f18408299deca7d87a30944 : SUCCESS in 8m 26s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/1d9795460b4f4202a66388f68282b78f : SUCCESS in 9m 10s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/451054c8a0ec471e92eb130e0729ae38 : SUCCESS in 8m 21s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/ad70afa8c4404f56a808858ffe034d0d : SUCCESS in 5m 39s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/52125a2f926a4dc89681d3bf9d7e4f29 : SUCCESS in 1h 09m 38s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/12eb1df85bde4463a5cf03fc2ba3c8ff : SUCCESS in 1h 01m 48s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/62ac30b2720c4dd0909183d2164b46b5 : FAILURE in 53m 47s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/e286f56d5e7b46c4bb1dc30514a8829f : FAILURE in 1h 02m 48s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c45c897d863d4d3b96167a1eefb1bad3 : SUCCESS in 43m 48s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/ecb46851be8d47e38e800c78557a13d8 : SUCCESS in 1h 12m 38s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/ed519ede419c4e3793877409ce723d96 : SUCCESS in 28m 40s","accounts_in_message":[],"_revision_number":6},{"id":"88d86e1de793d5192b977c29ace53cfb12e5ede2","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-07-22 12:41:32.000000000","message":"Patch Set 6:\n\nrecheck unrelated issue","accounts_in_message":[],"_revision_number":6},{"id":"77f01b9080f7ed97d89ab4eff2df8217ad01707c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-22 14:20:19.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/399530f4588b4a149c21cd8f1f98a274 : SUCCESS in 57m 31s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/52ef7dfb8759493a82727dfbfcbeb549 : SUCCESS in 1h 15m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/02d918c188da4ede8a317b66884a4705 : SUCCESS in 5m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7454dea8474f475e82b2aa41256946da : SUCCESS in 5m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/beedc33e1b024c2fa8a933d6f628ccaa : SUCCESS in 5m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3ee1da87b1d94d398a626426254c2c33 : SUCCESS in 6m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aaa8621bbd3d4e2093195131a0570ece : SUCCESS in 7m 25s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/05261505249d4d4da8ca1b14372c1425 : SUCCESS in 9m 13s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/6c2d876910ff471db287094407e6a11b : SUCCESS in 10m 32s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/b5640e1ce9a449f7b2cb1d643c5e8c71 : SUCCESS in 10m 06s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/72fde3c40c5a4b5d8f9c8725dc3920c4 : SUCCESS in 5m 39s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/586ffe9470c34fcda2577a3549e54fa0 : SUCCESS in 1h 20m 05s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/b8faa93dd73b47219379574a16f17399 : SUCCESS in 1h 06m 45s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/0dfdbade3dd14134bbc24444c16d9614 : SUCCESS in 1h 12m 13s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/fa2377c5dd794e21835537dd61b563a2 : SUCCESS in 1h 10m 47s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/03c5c28717a0472aa1d7a9c30d9ffaa4 : SUCCESS in 54m 44s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/f1d51047640e43808e1bb21dfd46beb0 : SUCCESS in 1h 36m 42s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/b2eab2801c12404d82f478be3241be18 : SUCCESS in 30m 23s","accounts_in_message":[],"_revision_number":6},{"id":"e51d733c4c4efc314296e8ccf6b88344939e87e2","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-07-22 14:39:46.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"2089a8d62c357c8ca88bfe965cb3020576b791af","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-07-22 18:13:40.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.","accounts_in_message":[],"_revision_number":7},{"id":"90d876730001aa3bbd1655b75a851237243a5b7d","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-07-22 18:41:07.000000000","message":"Topic namespace-policy-refactor removed","accounts_in_message":[],"_revision_number":7},{"id":"7ce3559802c72f4a28b4dcea79e7e750ee24e860","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-07-22 18:41:10.000000000","message":"Topic set to policy-refactor","accounts_in_message":[],"_revision_number":7},{"id":"9e3ec90536f36305d7f745ae37efa92a30937363","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-07-22 18:41:10.000000000","message":"Patch Set 6:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/799632/799632/6/check-not-vote/ext-glance-zuul/c63e165 : SUCCESS in 37m 29s","accounts_in_message":[],"_revision_number":6},{"id":"88caca8311628393ab83589d6159230b8f9e1367","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-22 19:27:45.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/f4e38e3410194180a78eaaa3aff3d935 : SUCCESS in 1h 01m 26s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9e5131049f0442319435378d035eb835 : SUCCESS in 1h 05m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0836767bd2324c1988d02692e05021b9 : SUCCESS in 6m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0bd6e4355445452c8ec52ce3e08d939c : SUCCESS in 6m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/50c07928da0a4e5083cef0ce2e8f4f3c : SUCCESS in 6m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d6403fb5a47242418e8fa08583d29987 : SUCCESS in 6m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c5420356b7c4552a72e3fa146e87091 : SUCCESS in 7m 07s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/36de832e7a374d588068d54f9b3c6799 : SUCCESS in 10m 15s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/d7e7c523c63144c78de27ad9888143ad : SUCCESS in 8m 27s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/b784d161af11466cbfafb64631e49a85 : SUCCESS in 9m 50s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/4b512325b10f47c3bb53983dfc12324c : SUCCESS in 6m 02s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/d50e609f5cf64ff2a83713109056b56a : SUCCESS in 1h 09m 49s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/41f73d0e98514a4b8af7a300844b3d22 : SUCCESS in 1h 01m 39s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/2192c84bf9f44a67926ce8af38e2453c : SUCCESS in 54m 44s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/10fea1f423ca490f9ba254818469620b : SUCCESS in 58m 51s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d2cd4a5b34c4486587652f10362c4a96 : SUCCESS in 49m 13s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/d66d1ead10aa4d39b338e8df4e28b9ef : SUCCESS in 1h 11m 07s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/999342d7ace34a0fb06d51049c545504 : SUCCESS in 28m 52s","accounts_in_message":[],"_revision_number":7},{"id":"c6071a416e958d5d560cd1c6b1f1e79db0b4debd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-02 08:46:26.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.","accounts_in_message":[],"_revision_number":8},{"id":"7ee2b73eae80048e71ee15c7d3aa1363924ba23c","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-08-02 09:26:50.000000000","message":"Patch Set 8:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/799632/799632/8/check-not-vote/ext-glance-zuul/657d042 : SUCCESS in 39m 48s","accounts_in_message":[],"_revision_number":8},{"id":"b4dddb6660b6e0464a9fafc8e1c5d0a550349545","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-02 10:24:02.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/269cf482fc754218a0f048351eb938bb : SUCCESS in 1h 01m 39s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f3c225c3ebd9443cae75e89b373965d9 : SUCCESS in 1h 07m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f4913397d414464ac0307377cf764c1 : SUCCESS in 6m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/122cfa1f9e894e088a85d33a67eae5ed : SUCCESS in 6m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/988fa27ea12d41ef8a0576cc8a9d3f0f : SUCCESS in 7m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5732480ce5144102978b081ffaa61b9b : SUCCESS in 7m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e847df453c342fead3bd3bc01b54570 : SUCCESS in 6m 33s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c577aadb7c15409ebb1a814f2abe8ae8 : SUCCESS in 9m 23s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/cc55595166a54c5a8bd1cf903ac2ce94 : SUCCESS in 9m 04s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/d6a465a23b1f4b07a582fff6212d4e50 : SUCCESS in 9m 39s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/731feeec2a974648bc91381cdaa89f85 : SUCCESS in 5m 44s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/ce08294b9245423cb292e4254c72842d : SUCCESS in 1h 18m 13s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/95a586dea303436888cb1dfe945d3f28 : SUCCESS in 1h 09m 08s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/9f4a975769b64b6598ffb088614b5d50 : SUCCESS in 1h 11m 21s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/7d2c87ca856b47e9a84f36d7d9ee6169 : SUCCESS in 1h 20m 21s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/214d1ee3535f490baaa8019c1cfadc77 : SUCCESS in 45m 33s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/44739e476e5447df9ab8b0ef0d943804 : SUCCESS in 1h 26m 43s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/c13f2eafa9884a0995127a48319a41f3 : SUCCESS in 29m 47s","accounts_in_message":[],"_revision_number":8},{"id":"c42c55904443754a034573f7c1b7d92e2d9e8522","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-04 15:39:38.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"d345b4c82916641abd0925739ce5d6cfbe48a9cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-04 17:04:32.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/0ed46939df7f40f7a996ce117e05dde4 : SUCCESS in 1h 01m 58s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b653cd46f1424dde846affd7b80d5636 : SUCCESS in 1h 03m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a523fa6f673e4734b5178f6525ec5bd5 : SUCCESS in 5m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/065860e29caf4221b1baf9420819c4f8 : SUCCESS in 6m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5a6d7c6ad68a4bd5a73305904229bc26 : SUCCESS in 6m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/48a6f9fd0c824a2fbc0ccd2347540d9b : SUCCESS in 6m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0953ae2f86c94185bd4ce58519492e44 : SUCCESS in 6m 24s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1c90f9335fe244b0ab8e1cde4bee5f44 : SUCCESS in 12m 40s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/7c5700af78cd465e8fbd3aecbb76797b : SUCCESS in 8m 35s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/44850d8a05f346c2bf7edab5a9d6e248 : SUCCESS in 8m 34s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/92f01a43c14a4aa48d0ae08f3807dd93 : SUCCESS in 5m 27s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/fb82bfbea375447cbec513bae381358e : SUCCESS in 1h 13m 14s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/981e310c066e497b9665ac3e6b8e17d3 : SUCCESS in 1h 17m 01s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/f8e5ebf690b740fea830fdbaf7c13adb : SUCCESS in 1h 09m 12s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/8feb099306df408b944584452dd62d3a : SUCCESS in 1h 19m 39s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8918809efcaf491b9ac6623b7fe0c407 : SUCCESS in 48m 51s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/9ed8a392450d4d5b914f1600be9ee0c4 : SUCCESS in 1h 15m 38s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/25cabe50fe514f18992439dedf91907e : SUCCESS in 32m 49s","accounts_in_message":[],"_revision_number":9},{"id":"ca95d9e4ddfd7185ed52083e68121f111cb54554","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-08-04 18:20:15.000000000","message":"Patch Set 9: Code-Review+1\n\nLooks consistent with the other change for the ImageRepo that already landed.","accounts_in_message":[],"_revision_number":9},{"id":"5a8e8ae2273fd7e5bb39c294b9ca066d9718e957","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-08-05 06:48:30.000000000","message":"Patch Set 9:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/799632/799632/9/check-not-vote/ext-glance-zuul/4e8167a : SUCCESS in 36m 41s","accounts_in_message":[],"_revision_number":9},{"id":"99dcf626e158d5bd99939fb928ce43fa5a9ea18a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-05 11:34:56.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"6f32b9f9cb8ec3d12d2958459969a877c5f94c5f","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-05 17:15:20.000000000","message":"Patch Set 10: Workflow-1","accounts_in_message":[],"_revision_number":10},{"id":"a37b48e620a2d42fb856ce447c2f6e0e960aaf45","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-05 19:02:40.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"5a882ef9afbab99fe8072b846ab30980f27a69b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-05 20:42:42.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/6f7d51d1e1e4489aba4da7b096bfaa37 : SUCCESS in 51m 59s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/aa1429b61329444ca5f4d5135b7938bd : SUCCESS in 1h 07m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a94c765635549508e56ec0c388b5bbf : SUCCESS in 5m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4a81ef88e3ee4bf98902acba19fdb2f1 : SUCCESS in 6m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1df0f298dcab43198b50063663ff2591 : SUCCESS in 6m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0a73486499b74afa8f429b3d78ddf9aa : SUCCESS in 6m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53228389e3104b1992a84fe504150a82 : SUCCESS in 6m 32s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c95e0b4dad4343de8275b8c4bf98d1bd : SUCCESS in 8m 01s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/3fa07785b84c407cac7f1977bcd9c1f7 : SUCCESS in 8m 17s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/e61174c82b544c8bb88a90203ccb9d77 : SUCCESS in 13m 46s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/da08eec8b3114106851424c429066d76 : SUCCESS in 5m 58s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/58dda5ac60634731b9cd01f1b931fa44 : SUCCESS in 1h 06m 31s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/3e5e102bff7649ccae4548040668ca4c : SUCCESS in 1h 36m 06s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/bff63bf6913649609d0b6e51c9bcfaf3 : SUCCESS in 1h 02m 58s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/445565c8ca5f465f865443789d6b1bb8 : FAILURE in 1h 10m 53s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c77b1d9600ea4211a5ed2a4aaa89b676 : SUCCESS in 46m 52s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/ae8249821aab443dbf8dc3f9f8fbb9a1 : SUCCESS in 1h 10m 18s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/e9415ab59c734837933ced12ecf4fca6 : SUCCESS in 30m 15s","accounts_in_message":[],"_revision_number":11},{"id":"4da8bd8484a50156554d48babdc30b3001259a02","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-08-06 14:43:22.000000000","message":"Patch Set 11:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/799632/799632/11/check-not-vote/ext-glance-zuul/dc7ed80 : SUCCESS in 39m 17s","accounts_in_message":[],"_revision_number":11},{"id":"b54be0cca73653a4b3a946557bc0c7782a6dff24","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2021-08-09 19:44:48.000000000","message":"Patch Set 11: Code-Review+2","accounts_in_message":[],"_revision_number":11},{"id":"6d74b05d84529dea9cf572bf3fbab59afdd076de","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-08-10 15:40:27.000000000","message":"Patch Set 11: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":11},{"id":"0c499e57bce3301d5d8e0118cc54de6d6def757c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-10 15:41:02.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"df98bce27a544c843f024d57bdc2dfdf097aaf44","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-10 17:48:40.000000000","message":"Patch Set 11: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/2ab22c4f837f41b787227fe62f4dc6c7 : SUCCESS in 59m 27s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/30220d1226324856b1a82a6e838db3ad : SUCCESS in 1h 23m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1371769183504b72acda514f59e50c4a : SUCCESS in 6m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8c2d839d5d7c45ec97c5f969380f2900 : SUCCESS in 7m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dfcb13d6e22040fdb7ecd9e77260fb5c : SUCCESS in 7m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d32eeb77ba6b4efc812ff65e9389346b : SUCCESS in 8m 23s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3694ae031a02464caadaa70d8ac6da97 : SUCCESS in 9m 16s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/e845b8b34c54451abdc2541525f8d21e : SUCCESS in 9m 50s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/f621d18ef8a4499983a11de823513c6c : SUCCESS in 1h 09m 57s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/90cfa184f0794490a10f00a178cc8c29 : SUCCESS in 1h 06m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e4ef17686e0a47c9b6ca20daf238b9df : SUCCESS in 1h 02m 29s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/e98518c3de084300a81d05019e6e1dc2 : SUCCESS in 1h 23m 19s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/2bfea40552064b05944feed91a169685 : SUCCESS in 32m 13s","accounts_in_message":[],"_revision_number":11},{"id":"3a75e4d59ffac240b865f569f09a6cd9304c06e6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-10 17:48:40.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":11},{"id":"fb474dc281c1c4fd08ddaf88d9ece740cece070e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-10 17:52:57.000000000","message":"Patch Set 11:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee093780698d4ea5809228d708fa2556 : SUCCESS in 1m 29s","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"2943741c8600bd1bd1e91c3310fb36630a368d23","revisions":{"ae4a06be9e60d22983bdef79762f59b294c80ec5":{"kind":"REWORK","_number":1,"created":"2021-07-06 10:51:18.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/32/799632/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/32/799632/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/32/799632/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/32/799632/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/32/799632/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/32/799632/1"}}},"commit":{"parents":[{"commit":"739c0d46d64927b0ae9c7b6c0ab4dc1b325d92ab","subject":"POC: Check deactivate, reactivate policy in the API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/739c0d46d64927b0ae9c7b6c0ab4dc1b325d92ab"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-07-05 07:39:07.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-07-05 08:58:03.000000000","tz":0},"subject":"POC Refactor gateway auth layer for metadef APIs","message":"POC Refactor gateway auth layer for metadef APIs\n\nChange-Id: If9e23cbded044b3600960a847c17737d5d23a451\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ae4a06be9e60d22983bdef79762f59b294c80ec5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ae4a06be9e60d22983bdef79762f59b294c80ec5"}]},"branch":"refs/heads/master"},"88827835ecc3610c6d26683e0e96a64f80610bd4":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-07-06 18:00:34.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/32/799632/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/32/799632/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/32/799632/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/32/799632/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/32/799632/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/32/799632/2"}}},"commit":{"parents":[{"commit":"8a3db67da1332634602d79f4efe600d5ae9591e0","subject":"POC: Check deactivate, reactivate policy in the API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8a3db67da1332634602d79f4efe600d5ae9591e0"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-07-05 07:39:07.000000000","tz":0},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-07-06 19:02:22.000000000","tz":-420},"subject":"POC Refactor gateway auth layer for metadef APIs","message":"POC Refactor gateway auth layer for metadef APIs\n\nChange-Id: If9e23cbded044b3600960a847c17737d5d23a451\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/88827835ecc3610c6d26683e0e96a64f80610bd4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/88827835ecc3610c6d26683e0e96a64f80610bd4"}]},"branch":"refs/heads/master"},"37aab818b076cc438136079eb0b307c17d79fd25":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-07-14 19:32:52.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/32/799632/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/32/799632/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/32/799632/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/32/799632/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/32/799632/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/32/799632/3"}}},"commit":{"parents":[{"commit":"e29cc0615727d4f307540d75d8926dddcda4022a","subject":"POC: Check deactivate, reactivate policy in the API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e29cc0615727d4f307540d75d8926dddcda4022a"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-07-05 07:39:07.000000000","tz":0},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-07-14 20:29:01.000000000","tz":-420},"subject":"POC Refactor gateway auth layer for metadef APIs","message":"POC Refactor gateway auth layer for metadef APIs\n\nChange-Id: If9e23cbded044b3600960a847c17737d5d23a451\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/37aab818b076cc438136079eb0b307c17d79fd25"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/37aab818b076cc438136079eb0b307c17d79fd25"}]},"branch":"refs/heads/master"},"f585fef0335c6728b6eee6529c5bc8c2aceb67f6":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2021-07-16 13:56:40.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/32/799632/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/32/799632/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/32/799632/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/32/799632/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/32/799632/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/32/799632/4"}}},"commit":{"parents":[{"commit":"944c2f28f64d958a309ca5840fb9ed3277a8da55","subject":"POC: Check deactivate, reactivate policy in the API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/944c2f28f64d958a309ca5840fb9ed3277a8da55"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-07-05 07:39:07.000000000","tz":0},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-07-16 14:58:47.000000000","tz":-420},"subject":"POC Refactor gateway auth layer for metadef APIs","message":"POC Refactor gateway auth layer for metadef APIs\n\nChange-Id: If9e23cbded044b3600960a847c17737d5d23a451\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f585fef0335c6728b6eee6529c5bc8c2aceb67f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f585fef0335c6728b6eee6529c5bc8c2aceb67f6"}]},"branch":"refs/heads/master"},"1fe03887679c8c37479f4bb0d550f01c4007d9cc":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2021-07-16 15:28:57.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/32/799632/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/32/799632/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/32/799632/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/32/799632/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/32/799632/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/32/799632/5"}}},"commit":{"parents":[{"commit":"d748041deee2e698aaf04a516923a54a69c95e76","subject":"POC: Check deactivate, reactivate policy in the API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d748041deee2e698aaf04a516923a54a69c95e76"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-07-05 07:39:07.000000000","tz":0},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-07-16 16:31:07.000000000","tz":-420},"subject":"POC Refactor gateway auth layer for metadef APIs","message":"POC Refactor gateway auth layer for metadef APIs\n\nChange-Id: If9e23cbded044b3600960a847c17737d5d23a451\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1fe03887679c8c37479f4bb0d550f01c4007d9cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1fe03887679c8c37479f4bb0d550f01c4007d9cc"}]},"branch":"refs/heads/master"},"e8542b36577473fc9115aa65b9c5b5f13a0a5fb4":{"kind":"REWORK","_number":6,"created":"2021-07-22 09:46:07.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/32/799632/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/32/799632/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/32/799632/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/32/799632/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/32/799632/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/32/799632/6"}}},"commit":{"parents":[{"commit":"acda526fe94932ae99fb2ef015e942c825a64e7c","subject":"Add base policy check module","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/acda526fe94932ae99fb2ef015e942c825a64e7c"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-07-22 05:17:16.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-07-22 05:17:16.000000000","tz":0},"subject":"Refactor gateway auth layer for metadef APIs","message":"Refactor gateway auth layer for metadef APIs\n\nThis makes us able to optionally not get the authorization layer on\ntop of the stack. It also don\u0027t insert the policy layer in the\nmiddle of the stack.\n\nRelated blueprint policy-refactor\nChange-Id: If9e23cbded044b3600960a847c17737d5d23a451\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e8542b36577473fc9115aa65b9c5b5f13a0a5fb4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e8542b36577473fc9115aa65b9c5b5f13a0a5fb4"}]},"branch":"refs/heads/master"},"a6f3adf910c05262d301e9569e8da989ead704bc":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2021-07-22 18:13:40.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/32/799632/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/32/799632/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/32/799632/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/32/799632/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/32/799632/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/32/799632/7"}}},"commit":{"parents":[{"commit":"485677a3f383489bbedf7c184ed992f1bcc53b24","subject":"Add base policy check module","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/485677a3f383489bbedf7c184ed992f1bcc53b24"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-07-22 05:17:16.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-07-22 17:20:24.000000000","tz":0},"subject":"Refactor gateway auth layer for metadef APIs","message":"Refactor gateway auth layer for metadef APIs\n\nThis makes us able to optionally not get the authorization layer on\ntop of the stack. It also don\u0027t insert the policy layer in the\nmiddle of the stack.\n\nRelated blueprint policy-refactor\nChange-Id: If9e23cbded044b3600960a847c17737d5d23a451\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a6f3adf910c05262d301e9569e8da989ead704bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a6f3adf910c05262d301e9569e8da989ead704bc"}]},"branch":"refs/heads/master"},"00f11f5a7c2584c4c611491c399aface56c37559":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2021-08-02 08:46:26.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/32/799632/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/32/799632/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/32/799632/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/32/799632/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/32/799632/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/32/799632/8"}}},"commit":{"parents":[{"commit":"429f16124b7cd69c85e6d764bcf1e47687780286","subject":"Remove dead 403-\u003e404 code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/429f16124b7cd69c85e6d764bcf1e47687780286"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-07-22 05:17:16.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-08-02 04:53:06.000000000","tz":0},"subject":"Refactor gateway auth layer for metadef APIs","message":"Refactor gateway auth layer for metadef APIs\n\nThis makes us able to optionally not get the authorization layer on\ntop of the stack. It also don\u0027t insert the policy layer in the\nmiddle of the stack.\n\nRelated blueprint policy-refactor\nChange-Id: If9e23cbded044b3600960a847c17737d5d23a451\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/00f11f5a7c2584c4c611491c399aface56c37559"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/00f11f5a7c2584c4c611491c399aface56c37559"}]},"branch":"refs/heads/master"},"0cc8190d0edef73ffa7ee64c527e2d083cbb8bd1":{"kind":"REWORK","_number":9,"created":"2021-08-04 15:39:38.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/32/799632/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/32/799632/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/32/799632/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/32/799632/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/32/799632/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/32/799632/9"}}},"commit":{"parents":[{"commit":"8767489f7b8f8c98d402ab77e8a34a6445e25a04","subject":"Refactor gateway get_repo auth layer","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8767489f7b8f8c98d402ab77e8a34a6445e25a04"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-07-22 05:17:16.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-08-04 14:25:07.000000000","tz":0},"subject":"Refactor gateway auth layer for metadef APIs","message":"Refactor gateway auth layer for metadef APIs\n\nThis makes us able to optionally not get the authorization layer on\ntop of the stack. It also don\u0027t insert the policy layer in the\nmiddle of the stack.\n\nRelated blueprint policy-refactor\nChange-Id: If9e23cbded044b3600960a847c17737d5d23a451\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0cc8190d0edef73ffa7ee64c527e2d083cbb8bd1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0cc8190d0edef73ffa7ee64c527e2d083cbb8bd1"}]},"branch":"refs/heads/master"},"1acf6e6faaf95d1359a15bf042d826b3332db710":{"kind":"REWORK","_number":10,"created":"2021-08-05 11:34:56.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/32/799632/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/32/799632/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/32/799632/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/32/799632/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/32/799632/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/32/799632/10"}}},"commit":{"parents":[{"commit":"96524ce80e59a4f2a3bd07d67279f110ad92a99d","subject":"Merge \"Add a member field to Image when appropriate\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/96524ce80e59a4f2a3bd07d67279f110ad92a99d"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-07-22 05:17:16.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-08-05 06:40:58.000000000","tz":0},"subject":"Refactor gateway auth layer for metadef APIs","message":"Refactor gateway auth layer for metadef APIs\n\nThis makes us able to optionally not get the authorization layer on\ntop of the stack. It also don\u0027t insert the policy layer in the\nmiddle of the stack.\n\nRelated blueprint policy-refactor\nChange-Id: If9e23cbded044b3600960a847c17737d5d23a451\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1acf6e6faaf95d1359a15bf042d826b3332db710"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1acf6e6faaf95d1359a15bf042d826b3332db710"}]},"branch":"refs/heads/master"},"2943741c8600bd1bd1e91c3310fb36630a368d23":{"kind":"REWORK","_number":11,"created":"2021-08-05 19:02:40.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/32/799632/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/32/799632/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/32/799632/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/32/799632/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/32/799632/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/32/799632/11"}}},"commit":{"parents":[{"commit":"96524ce80e59a4f2a3bd07d67279f110ad92a99d","subject":"Merge \"Add a member field to Image when appropriate\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/96524ce80e59a4f2a3bd07d67279f110ad92a99d"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-07-22 05:17:16.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-08-05 18:07:12.000000000","tz":0},"subject":"Refactor gateway auth layer for metadef APIs","message":"Refactor gateway auth layer for metadef APIs\n\nThis makes us able to optionally not get the authorization layer on\ntop of the stack. It also don\u0027t insert the policy layer in the\nmiddle of the stack.\n\nRelated blueprint policy-refactor\n\nChange-Id: If9e23cbded044b3600960a847c17737d5d23a451\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2943741c8600bd1bd1e91c3310fb36630a368d23"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2943741c8600bd1bd1e91c3310fb36630a368d23"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
