)]}'
{"id":"openstack%2Fglance~799633","triplet_id":"openstack%2Fglance~master~I59cd3bbd98065457bf0c9cfaf6888b7f56f7617c","project":"openstack/glance","branch":"master","topic":"policy-refactor","hashtags":[],"change_id":"I59cd3bbd98065457bf0c9cfaf6888b7f56f7617c","subject":"Move metadef namespace policy checks in the API","status":"MERGED","created":"2021-07-06 10:51:18.000000000","updated":"2021-08-11 22:15:14.000000000","submitted":"2021-08-11 22:13:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":76,"unresolved_comment_count":23,"has_review_started":true,"submission_id":"799633-policy-refactor","meta_rev_id":"00f1f5d3eb6a2aa55d78d42652e3d6bd5971a6b4","_number":799633,"virtual_id_number":799633,"owner":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-11 22:13:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2021-08-11 21:02:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":1,"date":"2021-08-11 21:02:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-06 12:51:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-06 18:00:34.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2021-07-17 00:50:24.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-07-21 16:48:49.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2021-08-04 19:12:43.000000000","updated_by":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"reviewer":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"state":"CC"}],"messages":[{"id":"36d2b478bd367efc5c27d631b043df71899a9b57","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-07-06 10:51:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7ffa2cc56b4f9d2f58d3e4c5d14700e168d71bec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-06 12:51:23.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/4ce97d4956194f08b23864db0f2c02dc : SUCCESS in 1h 03m 53s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0422c74a63c447ff9ef91f00b1417963 : SUCCESS in 1h 08m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/895b44e422ec4caf919bbc4648893d78 : SUCCESS in 6m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/99d2d582732a469ea12626ab867bf052 : SUCCESS in 7m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2b5a8f9fa8e6475a8ce1cb1bf38b3e78 : SUCCESS in 6m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c85c12c0b415469c80b88de72b875472 : SUCCESS in 7m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b8818eafda84fb4a15af48d17c1787a : SUCCESS in 7m 39s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c378b1ff851b423080c0c7e121201313 : FAILURE in 9m 54s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/06dbf57de2cd4a9e8d95111b5ea647de : FAILURE in 8m 51s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/00a4654f240b4021abd01679b4b3706c : FAILURE in 10m 03s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/4ac50f806017403892cdb52eb3970aa0 : SUCCESS in 5m 45s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/a714040c335f4847b452e24ba51e3154 : SUCCESS in 1h 17m 54s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/93542e9bdaf94b23b8106062edf7adbe : SUCCESS in 1h 35m 25s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/599a3e5d3d4244f1bbd981ee4e010dca : SUCCESS in 1h 17m 02s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/b1d4fd14dbc04dae9c7ef25edc002885 : SUCCESS in 57m 41s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/812f3d79e69e4edfa9c5d02cb51cb48a : SUCCESS in 1h 10m 10s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/824024875b9145829e5a6120f6f2ffb2 : SUCCESS in 1h 31m 08s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/7fccc18d65824eba98e99721bfe5e3e8 : SUCCESS in 49m 38s","accounts_in_message":[],"_revision_number":1},{"id":"eec8fa891f4a1fffa031e55b99f2cf3754ecf61e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-07-06 18:00:34.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"18e31707dbabe4119d4a810d60111c74484b5331","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-06 19:26:13.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/10536c339cb543e988ecd23ce54ce579 : SUCCESS in 55m 38s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/566537f6b5a9468dad6292e2a5f11942 : SUCCESS in 1h 09m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/44ceb36858374638a76d24ab369ccf55 : SUCCESS in 6m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/875edee462254b00b61bbff73d295502 : SUCCESS in 7m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c7a37564b006492c9f4db273b94adaeb : SUCCESS in 7m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b3da4d98dd5c4e1b82ec13399692e1ba : SUCCESS in 7m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f836025f869648ecbc96c45d6d0ab528 : SUCCESS in 7m 43s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/dcfb7db499c8482faa82c0997b769dda : SUCCESS in 9m 32s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/64610106d0da42bd865748b8021450a8 : SUCCESS in 10m 22s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/77849d297fd04ce68360f6a1fef19f55 : FAILURE in 9m 08s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/0a668e003968480c88b69d7ba19a0235 : SUCCESS in 5m 48s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/ba6b4ca97ac74fd7b709552eea1d09ea : SUCCESS in 1h 18m 10s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/35363f62fdbe4bedaa5802fd3db7356f : SUCCESS in 1h 12m 57s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/473ac0376efd4ef4a6a236c706aedbe3 : SUCCESS in 1h 09m 49s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/36a32e5fbb1c45898dc437bff3a34f87 : SUCCESS in 1h 08m 04s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a3129a1300a34ed5aeb85be1125b8c8b : SUCCESS in 44m 16s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/decadd43c26c465d80b188075c0c4fd2 : SUCCESS in 1h 22m 14s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/607c3d6932b548938d6d7141f0d97ec1 : SUCCESS in 29m 43s","accounts_in_message":[],"_revision_number":2},{"id":"03b7b5e17e70ec7e70fe5b69fdc887446361c66d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-07-14 19:32:52.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"0383dfe6f5589f2c5539d36925f5540007e1d764","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-14 20:55:52.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/571abb16197e4f92a5dd820874395df7 : SUCCESS in 1h 08m 30s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1427331a8e234394a335c6348fefb8d3 : SUCCESS in 58m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/729b3c15aafd48a5bc8ebc5062abbdc9 : SUCCESS in 5m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5992ce949c92493fad60370579c2bafa : SUCCESS in 6m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/222877b70fff4fd29db29e01736d44e6 : SUCCESS in 6m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/46159ef1c7854c38af844dcd447c65ea : SUCCESS in 7m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73acecfe609d46d59657a57fa4950207 : SUCCESS in 7m 17s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7fba63584369452bbd45e2fb28fdb346 : SUCCESS in 9m 13s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/2f78c4e2da59427ca9d930b4316bb8de : SUCCESS in 8m 41s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/d03481b0605f4059934f336cab7a3c7d : SUCCESS in 9m 12s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/847d816fd1c640b08b631ea3e1650723 : SUCCESS in 5m 58s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/9fdab67b7ee742608f84a9244a9ea389 : SUCCESS in 1h 14m 19s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/dfa3f66810df42f1afd3f4db0fe13d77 : SUCCESS in 1h 15m 14s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/fbb99301c4f94ccdac8fa2620db476c6 : SUCCESS in 1h 02m 29s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/cb1b4d9804c1446f994a1584bb3e872b : SUCCESS in 1h 11m 57s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2da61904e9ea4dd1bc95b511205c381a : SUCCESS in 55m 08s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/11cf28efd93f4ab6be98e4e6d1ef4f11 : SUCCESS in 1h 06m 29s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/efdd5140542749d9915b153b96a92f3e : SUCCESS in 33m 57s","accounts_in_message":[],"_revision_number":3},{"id":"a11afb9a3c0b889e47ea76a7d4f635168056b5f5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-07-16 13:56:40.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"43c26d357f61c948c0afbacf101e4cd4f8fbaedf","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-07-16 14:16:21.000000000","message":"Patch Set 4:\n\n(8 comments)","accounts_in_message":[],"_revision_number":4},{"id":"cefb8f4b117bd4d6467150a85c90732cfb688774","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-07-16 14:40:34.000000000","message":"Patch Set 4:\n\n(6 comments)\n\nThank you for review Dan, added inline replies for your comments.","accounts_in_message":[],"_revision_number":4},{"id":"868dbe5c9649bea0991d0a0378f8c59c12609e51","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-07-16 15:28:57.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"8b794e5a8c59a49adad7799e8bdfff9d333c4e92","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-16 15:30:08.000000000","message":"Patch Set 4: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":4},{"id":"1748ecf3808137925e62ca3f3f3c78d7b39310c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-16 16:56:21.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/8b521d175a2a4378801ba8466b131e53 : SUCCESS in 1h 03m 35s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/cc6a7a8b4f6742428ac558d929f68bde : SUCCESS in 1h 09m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/be5c79ad1af34de089649c43b4409ce2 : SUCCESS in 6m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ef93aea15be94f179606b5875b36bcb0 : SUCCESS in 7m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a544c2cb29434e03906630525b462ad1 : SUCCESS in 7m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8205bd80dd1440459a9f254ae9acc172 : SUCCESS in 8m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2e4002069fb416a85e899765321e14d : SUCCESS in 8m 23s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/12c159be349240f9a21667d5a2e98b98 : SUCCESS in 10m 00s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/0d990550f3454b73bdd77be2e5ae5805 : SUCCESS in 8m 28s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/9f56b39c748643e59012e960a73e8dd2 : SUCCESS in 10m 38s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/ea4d94cb39fb4db0bf78feb165ddffb6 : SUCCESS in 7m 20s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/91ddf73117e74ae0bd2366137d727a89 : SUCCESS in 1h 21m 26s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/3df7e19d6c84476da08b1c8b898c81ab : SUCCESS in 1h 12m 52s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/bbec83dca5254dcd82cf2fad38254893 : SUCCESS in 1h 06m 56s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/8329872aa3544ad3aa7c56cc07ec1f4f : SUCCESS in 1h 10m 34s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0a31e2fc95a6442da6901f287b9ffac7 : SUCCESS in 48m 22s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/fedc63bdbefa40a0b891850fad4a404a : SUCCESS in 1h 08m 31s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/daa831f0d34a41ff96a6171770e20c74 : SUCCESS in 36m 51s","accounts_in_message":[],"_revision_number":5},{"id":"57912da764868adbf3c793e8f0198b7c92cba176","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-07-16 19:48:43.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"993d0e0d74e587eca6eb915c768639fa9973f112","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-16 22:32:13.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/f8973a46acaf4829a87c82a5b91f4201 : SUCCESS in 57m 25s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/088107a4e7424602b914077003d444ec : SUCCESS in 1h 02m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3fc28e2e79724e0cb74d0d4b6cab2b71 : SUCCESS in 6m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e69f1f493797447c8208dddf5eab1af0 : SUCCESS in 8m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fa8c79da5df4406bbc4a1d2323dadfa2 : SUCCESS in 6m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d16ba9613a094e9bafb1c8d213ec5a67 : SUCCESS in 7m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72eda719f17847b69a471e0e03cfb4f1 : SUCCESS in 7m 45s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/50698f63c1ce4b3e9cccf4af2d12e469 : SUCCESS in 10m 17s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/1a170a670ccd4e62b6615be5d1e468a7 : SUCCESS in 8m 46s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/a24a7fdd24134028bf5c905ac005bdd9 : SUCCESS in 8m 42s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/472fc5048ce349b0b644d2505429b0f1 : SUCCESS in 5m 49s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/733163ee1ee04684b3dde4654ff8b9c0 : SUCCESS in 1h 09m 57s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/939fb0aaa59a49e09659fe39b094f1ce : SUCCESS in 56m 39s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/0efe10bf39da47358ec7ffce0b2bde61 : SUCCESS in 1h 12m 09s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/b5ee9bd82184486aabd0d755fb802c92 : SUCCESS in 1h 08m 40s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bff7dfbeed8b412e81024f9467338df5 : SUCCESS in 40m 51s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/8af08f95d2b5408fbc1e1d40959ec706 : SUCCESS in 1h 10m 59s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/575c9216b1d34d0399a7345fac4878cf : SUCCESS in 29m 57s","accounts_in_message":[],"_revision_number":6},{"id":"6596c86ce91dc6ac628dca382cb4483191853158","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-07-17 00:50:24.000000000","message":"Patch Set 6:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/799633/799633/6/check-not-vote/ext-glance-zuul/e1467b3 : SUCCESS in 36m 49s","accounts_in_message":[],"_revision_number":6},{"id":"b2c39131753e610b041b2585f445f19b2e009b38","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-07-19 18:32:45.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"50b9800905cfdf515c88c87d00d5d5593297cb3e","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-07-19 18:33:53.000000000","message":"Topic tags-policy-refactor removed","accounts_in_message":[],"_revision_number":7},{"id":"3ba24840d92868eb70945ab19833c1a78412930d","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-07-19 18:33:56.000000000","message":"Topic set to policy-refactor","accounts_in_message":[],"_revision_number":7},{"id":"6aa1112a7171d1d851b380ece8c2d7854a5c2cf8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-19 19:49:43.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/878c849cd5134052a8edf80c14829540 : SUCCESS in 59m 25s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/89aa814c6c764d81bb4a19f77f093ca7 : SUCCESS in 58m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/149653bf4a1041458b9bf56e6a12b08d : SUCCESS in 5m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0436874e15b84b2190da6b9adb162b92 : SUCCESS in 7m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4f5eafa165084c20928475a3e03db637 : SUCCESS in 6m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8e1354362c6241f69f4e0aa9b536d6b0 : SUCCESS in 6m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64584541afc44b64920b1ae2c466e25f : SUCCESS in 7m 50s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7a186c24031a4fcab299e99dc607cc0f : SUCCESS in 9m 48s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/e5a02b1fb9c54c009d91b497c32c2dbf : SUCCESS in 9m 42s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/448da4ccbc37438f8cd77e3bd77dbbe6 : SUCCESS in 13m 08s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/06d60a74265e4fea8d84ce137c94b970 : SUCCESS in 4m 58s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/693e969fa7d442ab8b48ca28380dd20b : POST_FAILURE in 1h 00m 08s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/8ad96f29f14942e79220bacae800352c : SUCCESS in 1h 07m 33s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/22636be7c0aa49acbf10a805067c5384 : SUCCESS in 1h 08m 44s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/84440e1cbeea439b8f96c6eb21e2a9da : SUCCESS in 1h 14m 30s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/86bfc8b3d31c443ebd4e1bd41ae2d670 : SUCCESS in 44m 37s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/1eafee0bddcf48e5b184c7f32da1e7d5 : SUCCESS in 1h 12m 55s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/ccfd4f6b2c1046eaaf95f489ef50192b : SUCCESS in 36m 31s","accounts_in_message":[],"_revision_number":7},{"id":"d586560fe1cccb9fe16913d3f3b14336928c0e75","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-07-21 16:46:17.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"784bc4eb69186775f1ec8aa88623d5b5cb506e8f","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-07-21 16:48:23.000000000","message":"Topic namespace-policy-refactor removed","accounts_in_message":[],"_revision_number":8},{"id":"90580fd200dce3b539bc6618abaa406aee671a7d","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-07-21 16:48:31.000000000","message":"Topic set to policy-refactor","accounts_in_message":[],"_revision_number":8},{"id":"a70500545990f4b626b8195a8d693642157b3e7e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-21 18:01:13.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/2f5e96a435634c8e93947fee55f48762 : SUCCESS in 57m 19s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8ed60530e804478ca28ff79b1f186cd0 : SUCCESS in 58m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e9f178e193ca491dae1266105ab284cb : SUCCESS in 5m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4417042df0d642b1806f09e8b676efdc : SUCCESS in 8m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/42c63a7593ae4db9bc4ba465840bc797 : SUCCESS in 5m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fcdb0e01256f40f6b2b232d07f512921 : SUCCESS in 6m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1125b0e418894611bdc75330d5e21a4a : SUCCESS in 6m 01s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/606b79ea9e474ff6aaaeb7da4a584660 : SUCCESS in 9m 56s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/f2257cc7bd2f4394bb280d78de946240 : SUCCESS in 8m 06s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/7d522b35e7dd4dc8bf46c6a648be9f2b : SUCCESS in 9m 14s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/407a237b57b54206ade4be08860cbf09 : SUCCESS in 4m 48s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/7d2ec8f139634edcbfd386825a3cbaca : FAILURE in 24m 30s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/96521d14110a4546a47d37f0e6b7bc53 : SUCCESS in 58m 51s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/2e529af176234360acc546944265bc66 : SUCCESS in 1h 08m 18s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/d0e4afbe67634cd1974ead11d2c3fe3f : SUCCESS in 1h 12m 43s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/baef0601200f448dbac6be1fe929459f : SUCCESS in 42m 46s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/bed76e31f34a499ab16175e259be8edb : FAILURE in 23m 47s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/97407e597dc54fbb8653ab3d5ba98af9 : SUCCESS in 27m 05s","accounts_in_message":[],"_revision_number":8},{"id":"0b4523bffa2099219c6eed430f2011a23deba0d4","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-07-21 18:39:20.000000000","message":"Patch Set 8:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/799633/799633/8/check-not-vote/ext-glance-zuul/5492dc6 : SUCCESS in 38m 56s","accounts_in_message":[],"_revision_number":8},{"id":"359c333f4b051d21e2d586651fdccbd6603db4bf","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-07-21 18:50:49.000000000","message":"Patch Set 8: Code-Review+1\n\n(4 comments)\n\nSee my comment about retrying the operations after disabling the policy being different.\n\nBut otherwise, looks good I think. It\u0027s a lot of stuff packed into one thing, but all related.","accounts_in_message":[],"_revision_number":8},{"id":"830fd193262103b54fd720b4cedda215e53e6566","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-07-21 19:15:13.000000000","message":"Patch Set 8:\n\n(4 comments)\n\nMay be I can separate out gateway_repo changes to make it little bit shorter.\n\nThank you Dan for your review!","accounts_in_message":[],"_revision_number":8},{"id":"9008cc240798ed29a9959ea3ef3125fef991b1ee","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-07-22 05:46:47.000000000","message":"Patch Set 8:\n\n(3 comments)\n\nMade changes as per your suggestions!","accounts_in_message":[],"_revision_number":8},{"id":"4035784966aff9612fcf21791d70ac3f0795ae9f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-07-22 09:46:07.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"93b0d336b3d91a44e5606964c1a5458fbd8594f5","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-07-22 09:48:21.000000000","message":"Topic resource-policy-refactor removed","accounts_in_message":[],"_revision_number":9},{"id":"69fa86afd89a78c8955fcc8c00a98a8a57c2c629","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-07-22 09:48:25.000000000","message":"Topic set to policy-refactor","accounts_in_message":[],"_revision_number":9},{"id":"5a833c702795a259d01c2dd7f15cd49912dffa88","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-22 11:02:09.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/6a45a3f79dfc46669d1912ea20d36703 : SUCCESS in 1h 02m 23s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/845a5f6fee1c4ed68f1a6b0653264bd0 : SUCCESS in 1h 00m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e1e8c29db4a2435281e78fa29534715a : SUCCESS in 6m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/708652613c3b44aa81fc76f2922c49b9 : SUCCESS in 6m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/881ae991b7a74ae6bf933e0b1890f5fb : SUCCESS in 6m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/24f24ad5111c433c83826180327248c8 : SUCCESS in 7m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/843f6b348ea44b228729b17d5632ec87 : SUCCESS in 7m 12s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2e05002fcea545d1ada3c17e4228e209 : SUCCESS in 9m 46s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/2b8ad043dcf447418b63f39ed1bfefcf : SUCCESS in 9m 11s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/1084ad7719ff419ba07ce77aa29160b0 : SUCCESS in 10m 05s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/749df9518c6f4a7f8509ada955f0c99a : SUCCESS in 6m 24s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/8dcbeb1ea3e0484790cc517708e7cd89 : SUCCESS in 1h 10m 06s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/d377fe894e414075806b75f0ec665325 : SUCCESS in 1h 03m 08s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/32288bd732704debb15bbcfc59002dda : SUCCESS in 58m 23s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/c1b30388a9bb4294bf168ea72996f9b1 : SUCCESS in 1h 11m 53s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ad8cdc919dae417f84ae0a55bffafa4d : SUCCESS in 51m 13s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/cbb7f571e68c437e800bce12447235c6 : SUCCESS in 1h 13m 36s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/004318051fc44b4fb3fc75cd797a751f : SUCCESS in 32m 39s","accounts_in_message":[],"_revision_number":9},{"id":"b20c66a6eb0b92519a2cbc1bbfc1765078ef3072","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-07-22 14:43:34.000000000","message":"Patch Set 9: Code-Review-1\n\n(1 comment)\n\nJust one thing that I think we should fix before merge, to avoid modifying globals and potentially introducing weird test behavior. Otherwise, the test changes look good, thanks for doing that!","accounts_in_message":[],"_revision_number":9},{"id":"8ba7b5efd79d32551136234dfda5a646b9390141","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-07-22 14:45:48.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"85ce8e036a76baf50148a2f18dd260b53080f2d0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-07-22 18:13:40.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"0fae1239806587944f1b7fd5375c4eb2deaff921","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-07-22 18:14:49.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"7c563e07dc6f0ee0fd4fc3f14b960b6d0d3efc63","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-07-22 18:20:27.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"45459ae5f1119ac3ed1cec6e17b92a03dbf7a169","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-07-22 18:40:57.000000000","message":"Topic namespace-policy-refactor removed","accounts_in_message":[],"_revision_number":11},{"id":"193809c47919c5710ae319cfcd71a5250c4ac25d","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-07-22 18:41:01.000000000","message":"Topic set to policy-refactor","accounts_in_message":[],"_revision_number":11},{"id":"e4395e235223ddd88ce940348719cee1a53750e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-22 19:49:48.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/0385685b88774f13853b7d5fadbde559 : SUCCESS in 1h 02m 10s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f423f9e0aa7b478799caa43328d00f38 : SUCCESS in 1h 04m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ed7043dc6d4416b8574ce849ca63cf6 : SUCCESS in 5m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aae3db7fdf69494390a469119a4baf65 : SUCCESS in 5m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0883b05f91be4450b4ba0e4cd4de27c2 : SUCCESS in 6m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cf82870727464fe9ae4e6f815ad5a4c3 : SUCCESS in 6m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bffffda9b61e49ea906f10989a69a358 : SUCCESS in 6m 44s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f3ce1b9da43b48539e41f6ccb4cef340 : SUCCESS in 9m 28s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/f1eb77d83f844cbc93c919cc115483b9 : SUCCESS in 9m 53s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/2ce10b939a2c495c947a4528390a8a0a : SUCCESS in 8m 43s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/3b71cf8d624846ca89cbe17ba87c9e8c : SUCCESS in 5m 16s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/7e9110b09fd14e93b85ca5421e91bcfc : SUCCESS in 1h 04m 29s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/e47e6346090f4ef69ed280fd64c4b1bb : SUCCESS in 1h 12m 34s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/d8607ccc4c464f3881de60719b9bf303 : SUCCESS in 1h 03m 35s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/1f74bd97292845bfa1d4c716a7dd4e78 : SUCCESS in 1h 20m 39s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/eecb753871434d87b30d3cbed2013e47 : SUCCESS in 50m 44s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/ace1b2e119464b238f9a45d03e89f4d9 : SUCCESS in 1h 22m 58s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/0b0f3d3ff4cd47768e076216ec5a610b : SUCCESS in 44m 36s","accounts_in_message":[],"_revision_number":11},{"id":"55ce34d96e533c90f734c41599954c8b951db40c","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-07-23 09:46:10.000000000","message":"Patch Set 11:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/799633/799633/11/check-not-vote/ext-glance-zuul/be28651 : SUCCESS in 38m 02s","accounts_in_message":[],"_revision_number":11},{"id":"2613c829c27796674ef4ee6e507bd90d12fc4afd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-07-24 10:02:03.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"a7bb5a5af5dc857d3eafbde29ba3e7ef08a829c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-24 11:19:13.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/6a62dec4962a428db35986c9799e6788 : SUCCESS in 1h 09m 48s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e8b170b22f814aec99efc567b06c7237 : SUCCESS in 1h 07m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/226f4787e2ea453fba3440319a9df56a : SUCCESS in 5m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/860d9e90936a4c6b8651444ab2c288a3 : SUCCESS in 6m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dab140f4811f405d8dd5742b8507a2e8 : SUCCESS in 6m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/37bde646ffb941c8bc8ed53d6546b06a : SUCCESS in 6m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b38b839a39e04d7a8bb52b794bb86c49 : SUCCESS in 7m 43s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2ec43dc5d4474451ae7a9edee523694f : SUCCESS in 9m 02s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/bb8037dab3474a23a4d0a8ce338cfb6e : SUCCESS in 8m 39s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/625e8cf62f19407d938368f9f8a1168e : SUCCESS in 9m 26s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/6fde5380d4e24063823d8eddf2b9a427 : SUCCESS in 5m 30s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/b5acb9b53b94405ead9e5e130ea4a712 : SUCCESS in 1h 09m 44s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/6d6c644182344a95b90ef465cc601977 : SUCCESS in 1h 05m 50s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/adbeb7892cd343858da3efa03dcc8bf5 : SUCCESS in 1h 07m 07s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/c3bb80ba38634ab7975a73c5bac88d2a : SUCCESS in 1h 11m 29s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2f3f051d36474d3092e139b8895f40f8 : SUCCESS in 46m 31s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/2de796304b454ea0a859f684da6d55eb : SUCCESS in 1h 12m 38s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/840b35deb1234d0f83268c1d158ed434 : SUCCESS in 32m 48s","accounts_in_message":[],"_revision_number":12},{"id":"12eda9ff5ca0aa99bda5f949b027bb94204cdc09","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-07-24 12:34:07.000000000","message":"Patch Set 12:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/799633/799633/12/check-not-vote/ext-glance-zuul/fa11e8f : SUCCESS in 33m 33s","accounts_in_message":[],"_revision_number":12},{"id":"ffe41f18e6b17994adc2494d9a4181b295525cd3","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-07-28 17:40:24.000000000","message":"Patch Set 12: Code-Review+2","accounts_in_message":[],"_revision_number":12},{"id":"b04099f44678171b1c65e282bc2f2e3f3776ee67","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-07-30 17:19:36.000000000","message":"Patch Set 12: Workflow-1\n\n(1 comment)\n\nSetting workflow -1 to confirm my comment!","accounts_in_message":[],"_revision_number":12},{"id":"9074aa7f9c7f9b9ca571d90df7869e7b07b520fb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-02 08:46:26.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"a57833b28605fb37d06b95b8f40a576077f8fe4f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-02 10:18:08.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/d4c8b7d3df384f7290e491664f8bb19b : SUCCESS in 1h 05m 07s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3c445371451446659fba3fd5e74b57a3 : SUCCESS in 1h 13m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4717df52c1a34c1ba85aaf9f9b2fc618 : SUCCESS in 5m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0beb27acc9734b2e9639ef0f31c9a34b : SUCCESS in 7m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8ed7ca7fd7b7447981f47b340b3fbcfa : SUCCESS in 6m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/45b012dac32f4785b62672c3535367c2 : SUCCESS in 7m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ddec41c3fda47b6a338111a2a0883c1 : SUCCESS in 8m 36s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e9a252d19e284cd8a2c24e9fa024db79 : SUCCESS in 14m 07s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/4238e24622d14384b770cb89c24090d8 : SUCCESS in 9m 48s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/bcc3175ae6dc404e87bd42932b64d9db : SUCCESS in 14m 35s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/e3daebcfb59f46e38f74836468ac622d : SUCCESS in 5m 39s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/f41fc71fe6c34c31a55258baa35e1a6e : SUCCESS in 1h 07m 45s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/1b35668d35d74c58ab83b54bac8ef639 : SUCCESS in 1h 21m 41s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/b4c6271180834b94adcc3c859f357ef0 : SUCCESS in 1h 11m 39s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/f80fc2f3bcd64c5ca5c0686d6c8b2b6b : SUCCESS in 1h 15m 50s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8be099ea75f043208caf7582bd0219bb : SUCCESS in 51m 44s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/94f66b8dc377489bab8fd43b6739a16b : SUCCESS in 1h 28m 29s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/1ece6e39507940c7bc19033a01dba061 : SUCCESS in 34m 44s","accounts_in_message":[],"_revision_number":13},{"id":"cfda1d73fc425470830532298c328f0def38c53b","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-08-02 14:51:55.000000000","message":"Patch Set 13: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"0d0c1ae754518f6bc04c857a42c8b0bf07badb46","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-02 18:11:39.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"e236e0efb5a42da8350db3b4a3983f815a644b3a","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-08-02 18:47:23.000000000","message":"Patch Set 14: Code-Review+2","accounts_in_message":[],"_revision_number":14},{"id":"ecb4e8e28c1958b005e2e7d650cab36de83ebb22","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-08-02 18:51:47.000000000","message":"Patch Set 14: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"da375e778957b1f04146308b134630e2cd9579a1","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-02 19:04:12.000000000","message":"Patch Set 14:\n\n(1 comment)\n\nwill fix it in next review :D","accounts_in_message":[],"_revision_number":14},{"id":"f4341a465e04d3c75d4667a35680ef510930a432","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-02 19:37:56.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/668665bdb7304e88aaff075c1a49719a : FAILURE in 50m 33s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f9f0f857e37546a683369c47ae86a699 : SUCCESS in 1h 12m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e10472ad4766482384538679e06aac35 : SUCCESS in 6m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7275c6d2f9c249a6a64f57aad7a9d0d7 : SUCCESS in 6m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b630b9f123a849ce9ba42a8db7ccddfc : SUCCESS in 7m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9247909c86fa4b8185662c30a8f62345 : SUCCESS in 7m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cab28a11d02c433faeb5cc5a8f911655 : SUCCESS in 7m 18s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/53ab45c5d7674a3797980275433f04d0 : SUCCESS in 9m 40s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/1906f4be40d5471cb24de6f7419438a1 : SUCCESS in 8m 52s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/23221a8633a643a7b45d0b8313df2b1d : SUCCESS in 11m 32s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/022b201e12734d46bc4e8f28a5e44e71 : SUCCESS in 6m 33s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/8588fa7c7ca34a2688a569280c74eaf2 : SUCCESS in 1h 24m 10s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/e479a7196c764f9597106ef7ca200a76 : SUCCESS in 1h 12m 00s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/c9e96d7a9ef94bc3b85ca4591d65780d : SUCCESS in 1h 07m 17s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/462e6fc5cb93428c999b19d9fce18c89 : SUCCESS in 1h 05m 35s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/62a8b4f2da214541944e43ae97f09161 : SUCCESS in 41m 01s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/ea3dfc06a61f4929859daa99f58c1f75 : SUCCESS in 1h 07m 32s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/d95708b2f0d349ea8a7eecd6ee8939a4 : SUCCESS in 26m 10s","accounts_in_message":[],"_revision_number":14},{"id":"64186f0553b59ad4a0acf33aa9627610eef6c0bd","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-03 08:04:19.000000000","message":"Patch Set 14:\n\nrecheck bug 1938763","accounts_in_message":[],"_revision_number":14},{"id":"327e8a0b5821d72f113f0ab895a14a3473c27e6d","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-08-03 08:43:19.000000000","message":"Patch Set 14:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/799633/799633/14/check-not-vote/ext-glance-zuul/9f99ab8 : SUCCESS in 38m 24s","accounts_in_message":[],"_revision_number":14},{"id":"728063e3dd21b9f8331afe326bc1bd013ae87e05","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-03 09:59:49.000000000","message":"Patch Set 14:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/3127263565bb435d80abfb2ead39a23a : FAILURE in 41m 42s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/70a8af0100a94c73a32b7b05a69639a7 : SUCCESS in 1h 00m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ee6c79ab6344a3494c1241ca54dfc1b : SUCCESS in 5m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a2bc6bfd2b7141318fd8421bab89c97f : SUCCESS in 7m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c7e77fb9f6064c2a9ebe4e403882c853 : SUCCESS in 7m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8d27e4193fb8484d944eee8f35fe1f38 : SUCCESS in 7m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83dbfee8d1e34e899bdeba5eb775ca31 : SUCCESS in 7m 27s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/aa6cab45a397467cbd83b469e446279b : SUCCESS in 9m 55s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/5344bfd559204aa69426bcf0a9c50e3a : SUCCESS in 11m 11s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/3ab3a36f69be4515b3bc96c02ee81aa7 : SUCCESS in 10m 35s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/03ace77a7b134bce9938f47cf9a65e5e : SUCCESS in 6m 59s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/539096eeab9644a1b33f617eca0d17cd : FAILURE in 1h 22m 47s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/4a47b8bcfbd241f281d43cb5fa36b943 : SUCCESS in 1h 03m 34s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/db12bed917ec4fdc8c57b14f6390b569 : SUCCESS in 1h 08m 00s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/34177f0118a44874968ff5abbfe40c1c : FAILURE in 1h 37m 56s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/81b2165e92144ce9889951b4aff0e16a : SUCCESS in 42m 52s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/98f5279c827b42a58b7c275c60c57e9d : SUCCESS in 1h 03m 12s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/c35b3fbf49174c65aaab776d214900af : SUCCESS in 50m 25s","accounts_in_message":[],"_revision_number":14},{"id":"a1ca9d6803fd76500730e5a58d8e2ae87d78952f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-03 14:54:49.000000000","message":"Uploaded patch set 15: Commit message was updated.","accounts_in_message":[],"_revision_number":15},{"id":"78713ae72df525b20d5f1808f5ca1411a2df8a1e","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-08-03 15:06:47.000000000","message":"Patch Set 15: Code-Review+2","accounts_in_message":[],"_revision_number":15},{"id":"2089d5d48b841b4e9db3bd3024f5f75c68cd21d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-03 16:24:35.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/eccd735277014ab0b9146dafb066f722 : SUCCESS in 1h 01m 38s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/67d3927083b1417286fa4f52651a6026 : SUCCESS in 1h 10m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/40976aeb1a9d4ae4ac7867322ca9b323 : SUCCESS in 6m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/047671fb3d54478a9bdf965e109dd509 : SUCCESS in 6m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/af946d87e16a43298c4f9effe96545f1 : SUCCESS in 7m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e9518625bf874923a73ecb2b54eb2ba6 : SUCCESS in 7m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90e1b9e694fe496e975e5ea12509ef05 : SUCCESS in 7m 44s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c210b70fe16d439983022551aea634d7 : SUCCESS in 13m 43s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/f55f4820c4bc4c38b692bd7451293e07 : SUCCESS in 14m 23s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/4c14f50aeff14b4492f3587f17fcadae : SUCCESS in 11m 13s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/56ce9d97a658461abc3e0be728fc30fc : SUCCESS in 6m 34s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/d87ad392654444b89bdf1877ae4931a4 : SUCCESS in 1h 27m 01s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/93a24a7df4634c05824733826f1673fc : SUCCESS in 1h 02m 47s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/c3b70dc96c8d4010afe927f32efca60d : SUCCESS in 1h 03m 23s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/720b99eb64ed4db591ac212a5b0c5f34 : SUCCESS in 1h 14m 27s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4b2ad218cc494c8f93060456adf471a3 : SUCCESS in 45m 14s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/89af4423885447e998a0fe0758925ab4 : SUCCESS in 1h 10m 49s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/fb277fa8bd454d42a04669d5f1b1a85f : SUCCESS in 37m 49s","accounts_in_message":[],"_revision_number":15},{"id":"0ece02e98c609d9f3a485244b23a4d383807145c","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-08-03 23:10:57.000000000","message":"Patch Set 15:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/799633/799633/15/check-not-vote/ext-glance-zuul/a3f65ae : SUCCESS in 36m 43s","accounts_in_message":[],"_revision_number":15},{"id":"462179cf832356a9068337fe9480353d9c4fbf24","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-04 15:39:38.000000000","message":"Uploaded patch set 16: Patch Set 15 was rebased.","accounts_in_message":[],"_revision_number":16},{"id":"7a13ea02604d64d788632eab649389a918c04033","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-04 17:23:50.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/93098ae7e607426db640a4eb44600423 : SUCCESS in 1h 01m 09s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/24141b14583a48169371ab396cf12902 : SUCCESS in 1h 03m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e3103adc128a448aab5800910b9323e4 : SUCCESS in 7m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e2c1086b9d324f43b29b030648433152 : SUCCESS in 6m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ad1fa27dde444e3bad0443c9837becbf : SUCCESS in 5m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9f59503aa4974523bbd3ffa334db44d4 : SUCCESS in 6m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f40b5cec5204670b43ebf1167e44e8c : SUCCESS in 6m 44s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7375816a7c33442daa41aa19ad0a35e9 : SUCCESS in 13m 56s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/1c9c419f92944e759356c0ba7d5bf56e : SUCCESS in 8m 42s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/26fd6d395bcf42b6be3b58b620b0b64e : SUCCESS in 8m 47s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/9e0d6a8e25d54a7095d29d9aaa158fd2 : SUCCESS in 5m 32s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/d2bd562ec2ca46d8b16587e7bd83d743 : SUCCESS in 1h 40m 42s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/1fe737d2c4af4da09135875359411383 : SUCCESS in 1h 03m 15s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/51059af5ee1344efb4392da071525c10 : SUCCESS in 55m 22s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/7227551f58de4b6caf502a7c982283c8 : SUCCESS in 1h 09m 22s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ccc4d793d30641e28e080d227f8b25fa : SUCCESS in 1h 00m 27s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/c41d580c12b7435eb6eabe1324ec4c0a : SUCCESS in 1h 23m 48s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/5f97f9903484483ebb7bb52fbe2de636 : SUCCESS in 32m 46s","accounts_in_message":[],"_revision_number":16},{"id":"840471edd01f852158e6e135da6933a0fb088986","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-08-04 19:12:43.000000000","message":"Patch Set 16:\n\n(4 comments)","accounts_in_message":[],"_revision_number":16},{"id":"00c84a6e67d12d861b2cb3e1744eb7119ef6068a","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-04 19:22:00.000000000","message":"Patch Set 16: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"423b8cae0c5ad81816579213b15c15c8377d6479","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-04 19:29:41.000000000","message":"Patch Set 16:\n\n(3 comments)\n\nLance, thank you for review, added answers for your questions.","accounts_in_message":[],"_revision_number":16},{"id":"7e9e7576c2766e8bc24d4587c465f796966d6e92","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-08-04 20:07:56.000000000","message":"Patch Set 16:\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"4cabca439c3f59f3e0059ae09ddf7d19bb170fec","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-08-04 20:20:03.000000000","message":"Patch Set 16:\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"03667a717bb405d9a863065617b3dba3e89f73bf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-05 11:34:56.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"15ea9204ecc499ececaf9c55833c1dc72cfa0834","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-05 12:52:29.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/485a7905762849afa39bd578d3e3eedd : SUCCESS in 1h 06m 36s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a7acfa258f9a4fdd953bcf0b6fa135ca : SUCCESS in 1h 01m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c4523897fdc649af9ae112d02e409717 : SUCCESS in 6m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/39b4dc04ee864574ade7f21526603a6a : SUCCESS in 6m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/12eed3503d494fd688163a0b471b6306 : SUCCESS in 6m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/875d64d27a6b4b8d95e1ee147fafe6ff : SUCCESS in 6m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07ed1ea325da4adf8f3117a63a30e0ed : SUCCESS in 7m 00s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/eee216f9db3047e69643ab7e0589329b : SUCCESS in 8m 10s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/7cfc40afc93549009ba656ac7c5bbdc7 : SUCCESS in 8m 24s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/ccc2308a1e354eefb4dc1cafc368425f : SUCCESS in 7m 57s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/12a85b3ec846463aadb38569db40e9c7 : SUCCESS in 5m 08s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/341e4691a2544f1188cb53886ba931e2 : SUCCESS in 1h 09m 38s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/d1f0708926f742ddb5c85dde48794151 : SUCCESS in 1h 11m 04s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/41d6c87e428846bb8053f8779c65c477 : SUCCESS in 1h 13m 15s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/a864a06f2f794971b005f376882579d5 : SUCCESS in 1h 12m 53s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f95f7a01b4b143eaa45b3c5a8e0cce30 : SUCCESS in 50m 08s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/06eae005d01446728c4592d9c86d46bf : SUCCESS in 1h 08m 07s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/3bcda91aaccd4fbdbc904440d972560e : SUCCESS in 32m 30s","accounts_in_message":[],"_revision_number":17},{"id":"d85d95b0c49d49c8780adf4b4d9acbd7320604e1","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-05 16:01:10.000000000","message":"Patch Set 17: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"d3ea696179598a87b9eaa0976aedbffdf457ca4f","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-05 17:12:16.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"2a6f51a2560ca7b98523416ca3b37ac4296e9de2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-05 19:02:40.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"fe06000c6f1a1c29757addbc83469671a4ee1b42","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-05 20:21:20.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/a76c2f4c38c14696b13f6c8f77c7b016 : SUCCESS in 53m 27s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5d6fd8c2a5064993b7707932a3954501 : SUCCESS in 1h 00m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e50a57788d7402785436fc0e8146152 : SUCCESS in 6m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/99e5aeaa0ad44a4a9351884f781efd27 : SUCCESS in 6m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b93887452ad44499bc46bbbd6c1bd897 : SUCCESS in 6m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ac7906159f2b412483c022460c24f928 : SUCCESS in 7m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e3d9ffdfd674a7e8049d97f2a3759c3 : SUCCESS in 7m 14s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/087f9b78ee404bf98d0d94e2351dd6a6 : SUCCESS in 7m 29s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/fb130f50504c449096b0e45ac9f8ebef : SUCCESS in 9m 43s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/f08d9fb48cbb492a8f4826db181854da : SUCCESS in 9m 35s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/7c44ebee929e4413bd601fb63a62b445 : SUCCESS in 6m 09s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/b00502f212b0444aaefec47fa2e06154 : SUCCESS in 1h 00m 02s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/e42cfecdbbe947e5be22a1832f33962c : SUCCESS in 1h 02m 05s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/c8a10755b94a428f8d7656ab53bb3f59 : SUCCESS in 59m 21s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/af4ec1be00d04a4d8672be5d8b1ddcce : SUCCESS in 1h 12m 31s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7b05a5d0a2994d8d8e4a7008dc9164d8 : SUCCESS in 44m 57s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/3188e0a421a04650927bcaec5fa2c8a7 : SUCCESS in 1h 00m 31s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/6895928754674cd6b9119c0d32c4d4ee : SUCCESS in 32m 25s","accounts_in_message":[],"_revision_number":18},{"id":"18c3068db1f961609fef4dd06beec723f52a0700","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2021-08-09 20:16:20.000000000","message":"Patch Set 18: Code-Review+2","accounts_in_message":[],"_revision_number":18},{"id":"6a8d8e86f782f8de9f6ada9daf9a6011cae22aa0","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-08-10 16:06:16.000000000","message":"Patch Set 18:\n\n(7 comments)\n\nSorry, a couple things I didn\u0027t notice last time, in addition to some comments on the new things. Maybe all of these are expected, but I want to ask first...","accounts_in_message":[],"_revision_number":18},{"id":"0a2800a611e18e0923bb509c7fc25515aab8c213","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-10 16:46:44.000000000","message":"Patch Set 18:\n\n(5 comments)\n\nHi Dan,\nthank you for review, few of the suggestions make sense and worth fixing, adding comments to others.\nKindly have a look and let me know what you think of those.","accounts_in_message":[],"_revision_number":18},{"id":"200a573b3b8e7813914a708274c7643a5ee6444c","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-08-10 17:02:02.000000000","message":"Patch Set 18:\n\n(2 comments)","accounts_in_message":[],"_revision_number":18},{"id":"70ef8bf5c5a44c692e6ecba2d75dbc467f2b13f1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-10 18:51:54.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"70435b93bf59b717168de4bacf4d0deb03ad99ee","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-10 19:19:29.000000000","message":"Patch Set 19:\n\n(4 comments)\n\nThank you Dan, made changes as per suggestion!","accounts_in_message":[],"_revision_number":19},{"id":"1c0250d97dab2432b4d3b3a07bdaeb4f24325dd7","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-08-10 19:28:02.000000000","message":"Patch Set 19:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/799633/799633/19/check-not-vote/ext-glance-zuul/4b60bc8 : SUCCESS in 35m 38s","accounts_in_message":[],"_revision_number":19},{"id":"66ac003144b829a2e8fb9147ef3ca135b5e32014","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-08-10 19:39:14.000000000","message":"Patch Set 19:\n\n(3 comments)","accounts_in_message":[],"_revision_number":19},{"id":"8188ad33e4e854d1e04be7a077babcc4527fd7cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-10 20:12:31.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/444e398ac6244d2d92f333fca658529a : SUCCESS in 1h 18m 51s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/436a80c3a08647bea9def83388ecc0c9 : SUCCESS in 1h 09m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/81b55a81936c42f9a42cf9fd77df1302 : SUCCESS in 5m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/71c43e5aa7d24abd808a4657567d5e4c : SUCCESS in 6m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0c6939e3c1eb474fbd6b0a91feacfede : SUCCESS in 5m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9e8e199709a440cb83b6f3baf74b37a5 : SUCCESS in 6m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e808ee8f57604b299d4758ad352d38f7 : SUCCESS in 7m 47s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a1065a60a8ee429d9b95952a542204c3 : SUCCESS in 9m 11s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/a59e6828bdd648bca8958872418c38fb : SUCCESS in 7m 34s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/b20375e437b94cde9ac71f9fa5210508 : SUCCESS in 8m 12s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/db37785a62aa47e2afb162a029a88ad4 : SUCCESS in 5m 13s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/1b9401f9acc94426a9e893c6e969501d : SUCCESS in 1h 05m 24s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/bca6d4771eef4263a9c3838ff6be25b9 : SUCCESS in 1h 00m 15s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/a383ad4260a3443ca70869e5d22b692e : SUCCESS in 1h 04m 06s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/953181e52d8e4613b94fa5490ec6ed09 : FAILURE in 52m 36s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/43c058462c7c4e4ba0bd1330fbe061e6 : SUCCESS in 44m 21s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/9b02c2e431354e15b9b6437ba66fc248 : SUCCESS in 1h 00m 20s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/02f6193cc3864f829eadc685824389b7 : SUCCESS in 28m 34s","accounts_in_message":[],"_revision_number":19},{"id":"04924797b7badbb717d3c6afae9b51be3e204a4f","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-08-10 21:27:23.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"3c787cd6957a3bd2e5dc3f537aa099d2d6d6550d","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-10 21:47:42.000000000","message":"Patch Set 19:\n\n(2 comments)","accounts_in_message":[],"_revision_number":19},{"id":"1a128bb84c3c0dd74d9cb072da9e73e475c90134","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-10 21:58:09.000000000","message":"Uploaded patch set 20.","accounts_in_message":[],"_revision_number":20},{"id":"7c07c04a0b09fe39661be9e384f05fab173f8c42","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-08-10 22:34:30.000000000","message":"Patch Set 20: Code-Review+2","accounts_in_message":[],"_revision_number":20},{"id":"4dcaecc203c4083d1c0674f1e002159e9c051dca","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-08-10 22:36:19.000000000","message":"Patch Set 20:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/799633/799633/20/check-not-vote/ext-glance-zuul/e475d6b : SUCCESS in 37m 41s","accounts_in_message":[],"_revision_number":20},{"id":"a9847824822bd43dc0bff3e8e58c43113d36438d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-10 23:51:22.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/1daebbc9988f4787aaeca24c1ef4f346 : SUCCESS in 52m 52s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/20d4d3b21887444e89963d93b779ac3c : SUCCESS in 1h 09m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42eb4d06cddf45738e1f3bb4cf32dd8b : SUCCESS in 5m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/73191de6dc3d422eadd76c5d1ab0e0bf : SUCCESS in 6m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/42d11321e8f4404c89991318a8e64329 : SUCCESS in 6m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/365e6af0e8974051854d4190a1de0a18 : SUCCESS in 6m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e404507b8904393bb52d9f9237d3854 : SUCCESS in 6m 29s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9dc59d17acc2430c8e71dc7678a97d80 : SUCCESS in 9m 19s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/52c3e0530cbe4ac19184340968b2d06b : SUCCESS in 9m 45s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/57ddbadfd317423cb796047de689173c : SUCCESS in 7m 53s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/dd2cdf2355dd4038ab647e657de6395d : SUCCESS in 5m 46s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/54b30a1e539d4a73b62f6fd1e1ed51d6 : SUCCESS in 1h 16m 39s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/054adc2b52b04f45a3fe13cb79523384 : SUCCESS in 54m 27s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/badd28526ec44791a80de553a0d05fdf : SUCCESS in 1h 08m 44s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/451a7fabc71f43bdbee809fd3176280c : SUCCESS in 1h 24m 39s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ee516bd35cfb4627ab25cbb6a4d1a9cc : SUCCESS in 1h 06m 59s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/06ffafe510af4eff8dbda4d3b88369f5 : SUCCESS in 1h 05m 41s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/617dc4e5b6c74588835b552885fdf64f : SUCCESS in 26m 29s","accounts_in_message":[],"_revision_number":20},{"id":"38e4fbb5a694bbb7e2a3b7bd15dabd9a995ce0ad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-11 15:20:42.000000000","message":"Uploaded patch set 21.","accounts_in_message":[],"_revision_number":21},{"id":"634517e52135843ce1c9c80daa888a0f653a2705","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-11 15:49:43.000000000","message":"Patch Set 21:\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"053bad074a9bfa35175f42c18e23eaa2dd5de98f","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-08-11 16:15:57.000000000","message":"Patch Set 21: Code-Review+2\n\n(1 comment)\n\nI don\u0027t like the new nested try..try in this version. But, we\u0027ve been around too many times so I won\u0027t nit out on it. But I definitely liked the previous arrangement better. Was there some reason to change it?","accounts_in_message":[],"_revision_number":21},{"id":"7b8d8af830cbb3d09a5d6e59f32a754e8a0bc009","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-11 16:31:11.000000000","message":"Patch Set 21:\n\n\u003e Patch Set 21: Code-Review+2\n\u003e \n\u003e (1 comment)\n\u003e \n\u003e I don\u0027t like the new nested try..try in this version. But, we\u0027ve been around too many times so I won\u0027t nit out on it. But I definitely liked the previous arrangement better. Was there some reason to change it?\n\nPreviously I was catching both forbidden and not found and raising Notfound only, in inner layer I just need to catch forbidden and Not found will be caught directly in the chain below.\n\nIf you have strong objection then I will revert it back to original.","accounts_in_message":[],"_revision_number":21},{"id":"6401de0713afd70d56414e42ca027a59d2a332f1","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-08-11 16:36:05.000000000","message":"Patch Set 21:\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"940cd30faeb9017028a585f6bf918d8e3b9ae0c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-11 16:41:42.000000000","message":"Patch Set 21: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/561a967ab4a9436194397ceefcd8dae5 : SUCCESS in 53m 09s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d65db4737dca4d0d8762a63331b60ff5 : SUCCESS in 59m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a8c5fc9ffcf34ba8b7a33b22f8096912 : SUCCESS in 6m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7c25fac694fc474bb06ea2af2e988f98 : SUCCESS in 7m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1d0fb87c6ec2418cabdad0b4af9f0a02 : SUCCESS in 7m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5a1ca7dd379049528e4f53dffc210884 : SUCCESS in 7m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e65a567cbecf40edbf25f10be02742f5 : SUCCESS in 6m 55s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/fcffa675663f418991bf9a2b4b2c6744 : SUCCESS in 8m 49s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/5aa434c5c8a04ba08306a5482366aa29 : SUCCESS in 8m 19s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/7ceaed1172ab444fa88fe6279705349f : SUCCESS in 8m 25s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/70ab6dfb83ca47049d4277af93bf40aa : SUCCESS in 5m 09s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/828c4dfdd47a428aa42b9e346a7ee18e : SUCCESS in 1h 14m 38s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/090e1b556ae948478e371c2bc05432de : SUCCESS in 1h 11m 32s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/6210830ff22d4c85b4531b1d3081f032 : SUCCESS in 1h 12m 04s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/187104a0536b4d8da793bf5358ae291e : SUCCESS in 1h 16m 45s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0ac2ab5e974046c99050b842f92c2147 : SUCCESS in 44m 43s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/90823a70277d461d82316e762b9889f0 : SUCCESS in 1h 06m 08s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/a1f6d8cc456d4fdab562ae866e4f167c : SUCCESS in 33m 44s","accounts_in_message":[],"_revision_number":21},{"id":"3a925dff2bd92b5bd7d618e1a441096286b9f7f6","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-08-11 16:45:04.000000000","message":"Patch Set 21:\n\n(1 comment)\n\n\u003e Previously I was catching both forbidden and not found and raising Notfound only, in inner layer I just need to catch forbidden and Not found will be caught directly in the chain below.\n\u003e \n\u003e If you have strong objection then I will revert it back to original.\n\nI just hate deep nesting and I think it makes it difficult to understand and lends itself to accidental bugs. I think it\u0027s much better to write the code (when possible) to avoid very deep nesting so that you can eye-parse and verify that all the hidden \"else\" or \"no exception\" cases are covered.\n\nBut no, you don\u0027t have to change it for me, it just seemed like a step in the wrong direction. We have other stuff to worry about, so no need to nit out on this.","accounts_in_message":[],"_revision_number":21},{"id":"e5856ea7ed84a8c5bc274e1026bc003ecd44840f","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-11 16:47:44.000000000","message":"Patch Set 21:\n\n\u003e Patch Set 21:\n\u003e \n\u003e (1 comment)\n\u003e \n\u003e \u003e Previously I was catching both forbidden and not found and raising Notfound only, in inner layer I just need to catch forbidden and Not found will be caught directly in the chain below.\n\u003e \u003e \n\u003e \u003e If you have strong objection then I will revert it back to original.\n\u003e \n\u003e I just hate deep nesting and I think it makes it difficult to understand and lends itself to accidental bugs. I think it\u0027s much better to write the code (when possible) to avoid very deep nesting so that you can eye-parse and verify that all the hidden \"else\" or \"no exception\" cases are covered.\n\u003e \n\u003e But no, you don\u0027t have to change it for me, it just seemed like a step in the wrong direction. We have other stuff to worry about, so no need to nit out on this.\n\nAck, Anyway I need to push a followup patch to reuse code in functional tests, I will correct this at that time, added to my ToDo list.","accounts_in_message":[],"_revision_number":21},{"id":"badb4f811c8331bcc40a3408fac33eb852997a12","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2021-08-11 18:20:31.000000000","message":"Patch Set 21:\n\n(2 comments)\n\nYou said on IRC you\u0027d do push one more set, so I decided to nitpick a little :)","accounts_in_message":[],"_revision_number":21},{"id":"18c921087d0302eaeb54dc36b133cf510e24452b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-11 19:03:14.000000000","message":"Uploaded patch set 22.","accounts_in_message":[],"_revision_number":22},{"id":"7b5ac9e4abd54ca971952b35c0b569acaee3651d","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-11 19:06:31.000000000","message":"Patch Set 22:\n\n(2 comments)\n\nThank you Dan and Cyril,\naddressed suggestions given by both!!","accounts_in_message":[],"_revision_number":22},{"id":"1c841775ce2eab574eb17a950b5b528769850d20","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-11 20:25:59.000000000","message":"Patch Set 22: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/81a987bdeead4c38b0047c4e17f3d3aa : SUCCESS in 1h 08m 14s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a30ebfeee3434728b04ab4b97efe62fe : SUCCESS in 1h 13m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a8d669cbb8e4426d8ea9975e934a5821 : SUCCESS in 5m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8be2cba287a04f02b9d7dcc1949dbbd8 : SUCCESS in 5m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/42f1dc87bbf74b21b80b0dde6b366e81 : SUCCESS in 12m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/457b0d43a71d4b06a47256945672646f : SUCCESS in 6m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8cdd25e3dcda403f9d588820e5adef2f : SUCCESS in 7m 00s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7ab8e1ef1e604850bdcd4c8223f042aa : SUCCESS in 8m 32s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/11c36db03b444d71b245b61c776e36dd : SUCCESS in 8m 29s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/7aa13f9ec29b476d8e167dda47678e80 : SUCCESS in 8m 46s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/fa166051b6aa46ca9b62f97a177b36f3 : SUCCESS in 10m 27s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/997c1110867b4ce7a65bd96753c9de43 : SUCCESS in 1h 18m 45s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/bbc3bf4f21df4e43b8b5c7a77e275428 : SUCCESS in 1h 11m 08s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/4c27233b962840459eb899d995d3e5da : SUCCESS in 1h 10m 25s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/c111ad19e3ae43b6ba8b94997e14776c : SUCCESS in 1h 14m 16s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b668576201f14338a8e8039435c7fb42 : SUCCESS in 45m 42s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/48ec803613b54d1a818027aeb5e81fc8 : SUCCESS in 1h 12m 50s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/07691f0c995c49319f41bfeac3b7bf4f : SUCCESS in 32m 26s","accounts_in_message":[],"_revision_number":22},{"id":"94af0ed5f88c760ebd1c1d97b3667ff8e21d5f76","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-08-11 21:02:06.000000000","message":"Patch Set 22: Code-Review+2","accounts_in_message":[],"_revision_number":22},{"id":"2968add38c401cfcd22098311ce660386c3f1719","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-08-11 21:02:37.000000000","message":"Patch Set 22: Workflow+1\n\nActually, I\u0027m going to +W this, because Cyril\u0027s things were (self-admitted) nits, and they\u0027ve been done here to the letter, AFAICT.","accounts_in_message":[],"_revision_number":22},{"id":"4cf67dce1a4aa29a84a2afbc320699b83f8ac013","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-11 21:03:09.000000000","message":"Patch Set 22: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":22},{"id":"52b262d51483614178e8ae11ce85b31144eb5d2b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-11 22:13:35.000000000","message":"Patch Set 22: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/98d10cde26234b15ba1505424a3643ce : SUCCESS in 1h 05m 02s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e33acc4428f44a19ae5fe4d823b1f5fb : SUCCESS in 1h 02m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ef61ad4390b408b8f64cf1a2c88e129 : SUCCESS in 5m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1151b52426034107ba0712dfedb07e17 : SUCCESS in 6m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b8120a74006240ec973bfb952f14c852 : SUCCESS in 5m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/94c3bc1702ad4ec38f51fbbdf0c1115b : SUCCESS in 6m 55s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e574d3fb758d4f4e934cdfb61be7dd30 : SUCCESS in 10m 37s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/0fadf82e841b4ac583eeb13ee55c9d2f : SUCCESS in 9m 44s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/72a6a7d8de794cfbb54dfda88f19e6a9 : SUCCESS in 57m 53s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/346a85eb20744e0ca2d883676e3549ef : SUCCESS in 1h 05m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d88f3e73d1524dacb0700e5e3efcd41e : SUCCESS in 47m 50s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/8622d071a148437f95a123f1afa54c6b : SUCCESS in 1h 01m 22s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/5b7c5443b20f4a3fba1a3753748c1e36 : SUCCESS in 35m 00s","accounts_in_message":[],"_revision_number":22},{"id":"75132e737fe5d09b50597aebb0902cb4b89b75eb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-11 22:13:35.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":22},{"id":"00f1f5d3eb6a2aa55d78d42652e3d6bd5971a6b4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-11 22:15:14.000000000","message":"Patch Set 22:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da7cdf179ab04353b5e8ff4accad3ce3 : SUCCESS in 1m 20s","accounts_in_message":[],"_revision_number":22}],"current_revision_number":22,"current_revision":"99331d7efacb556effc9dda2a7db06bdc6af0319","revisions":{"a82d1d887203b201e69b5c151da71205e0040bf3":{"kind":"REWORK","_number":1,"created":"2021-07-06 10:51:18.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/33/799633/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/33/799633/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/33/799633/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/33/799633/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/33/799633/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/33/799633/1"}}},"commit":{"parents":[{"commit":"ae4a06be9e60d22983bdef79762f59b294c80ec5","subject":"POC Refactor gateway auth layer for metadef APIs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ae4a06be9e60d22983bdef79762f59b294c80ec5"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-07-05 09:41:35.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-07-05 18:01:09.000000000","tz":0},"subject":"POC: Move metadef namepsace policy checks in the API","message":"POC: Move metadef namepsace policy checks in the API\n\nNOTE: This has changes to make existing tests pass, but likely needs\nadditional coverage for full verification\n\nChange-Id: I59cd3bbd98065457bf0c9cfaf6888b7f56f7617c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a82d1d887203b201e69b5c151da71205e0040bf3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a82d1d887203b201e69b5c151da71205e0040bf3"}]},"branch":"refs/heads/master"},"65d5fc8694b079733a12e61851f01defa76c02e7":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-07-06 18:00:34.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/33/799633/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/33/799633/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/33/799633/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/33/799633/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/33/799633/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/33/799633/2"}}},"commit":{"parents":[{"commit":"88827835ecc3610c6d26683e0e96a64f80610bd4","subject":"POC Refactor gateway auth layer for metadef APIs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/88827835ecc3610c6d26683e0e96a64f80610bd4"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-07-05 09:41:35.000000000","tz":0},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-07-06 19:02:22.000000000","tz":-420},"subject":"POC: Move metadef namepsace policy checks in the API","message":"POC: Move metadef namepsace policy checks in the API\n\nNOTE: This has changes to make existing tests pass, but likely needs\nadditional coverage for full verification\n\nChange-Id: I59cd3bbd98065457bf0c9cfaf6888b7f56f7617c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/65d5fc8694b079733a12e61851f01defa76c02e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/65d5fc8694b079733a12e61851f01defa76c02e7"}]},"branch":"refs/heads/master"},"a2bbe7d845dbd9012e433ad13eedb286c8d3fac2":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-07-14 19:32:52.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/33/799633/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/33/799633/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/33/799633/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/33/799633/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/33/799633/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/33/799633/3"}}},"commit":{"parents":[{"commit":"37aab818b076cc438136079eb0b307c17d79fd25","subject":"POC Refactor gateway auth layer for metadef APIs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/37aab818b076cc438136079eb0b307c17d79fd25"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-07-05 09:41:35.000000000","tz":0},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-07-14 20:29:01.000000000","tz":-420},"subject":"POC: Move metadef namepsace policy checks in the API","message":"POC: Move metadef namepsace policy checks in the API\n\nNOTE: This has changes to make existing tests pass, but likely needs\nadditional coverage for full verification\n\nChange-Id: I59cd3bbd98065457bf0c9cfaf6888b7f56f7617c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a2bbe7d845dbd9012e433ad13eedb286c8d3fac2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a2bbe7d845dbd9012e433ad13eedb286c8d3fac2"}]},"branch":"refs/heads/master"},"095c05b3741a57b688be34e8c475b274e63ad917":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2021-07-16 13:56:40.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/33/799633/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/33/799633/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/33/799633/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/33/799633/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/33/799633/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/33/799633/4"}}},"commit":{"parents":[{"commit":"f585fef0335c6728b6eee6529c5bc8c2aceb67f6","subject":"POC Refactor gateway auth layer for metadef APIs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f585fef0335c6728b6eee6529c5bc8c2aceb67f6"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-07-05 09:41:35.000000000","tz":0},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-07-16 14:58:47.000000000","tz":-420},"subject":"POC: Move metadef namepsace policy checks in the API","message":"POC: Move metadef namepsace policy checks in the API\n\nNOTE: This has changes to make existing tests pass, but likely needs\nadditional coverage for full verification\n\nChange-Id: I59cd3bbd98065457bf0c9cfaf6888b7f56f7617c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/095c05b3741a57b688be34e8c475b274e63ad917"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/095c05b3741a57b688be34e8c475b274e63ad917"}]},"branch":"refs/heads/master"},"cff45d428debaf711ec4528c357e28426ea984ab":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2021-07-16 15:28:57.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/33/799633/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/33/799633/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/33/799633/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/33/799633/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/33/799633/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/33/799633/5"}}},"commit":{"parents":[{"commit":"1fe03887679c8c37479f4bb0d550f01c4007d9cc","subject":"POC Refactor gateway auth layer for metadef APIs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1fe03887679c8c37479f4bb0d550f01c4007d9cc"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-07-05 09:41:35.000000000","tz":0},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-07-16 16:31:07.000000000","tz":-420},"subject":"POC: Move metadef namepsace policy checks in the API","message":"POC: Move metadef namepsace policy checks in the API\n\nNOTE: This has changes to make existing tests pass, but likely needs\nadditional coverage for full verification\n\nChange-Id: I59cd3bbd98065457bf0c9cfaf6888b7f56f7617c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/cff45d428debaf711ec4528c357e28426ea984ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/cff45d428debaf711ec4528c357e28426ea984ab"}]},"branch":"refs/heads/master"},"c1baca60c6d3fa2289effb7858cdc2d60269eb0a":{"kind":"REWORK","_number":6,"created":"2021-07-16 19:48:43.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/33/799633/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/33/799633/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/33/799633/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/33/799633/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/33/799633/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/33/799633/6"}}},"commit":{"parents":[{"commit":"c0c31971924f4ad8d4f3814a5549a9a9c6174089","subject":"Add base poloicy check module","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c0c31971924f4ad8d4f3814a5549a9a9c6174089"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-07-16 19:44:29.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-07-16 19:44:29.000000000","tz":0},"subject":"POC: Move metadef namepsace policy checks in the API","message":"POC: Move metadef namepsace policy checks in the API\n\nNOTE: This has changes to make existing tests pass, but likely needs\nadditional coverage for full verification.\n\nThis makes us able to optionally not get the authorization layer on\ntop of the stack. It also don\u0027t insert the policy layer in the\nmiddle of the stack.\n\nPartially implements: blueprint policy-refactor\nChange-Id: I59cd3bbd98065457bf0c9cfaf6888b7f56f7617c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c1baca60c6d3fa2289effb7858cdc2d60269eb0a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c1baca60c6d3fa2289effb7858cdc2d60269eb0a"}]},"branch":"refs/heads/master"},"fafacac534478e32d7bc500a87299030bc46b3de":{"kind":"REWORK","_number":7,"created":"2021-07-19 18:32:45.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/33/799633/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/33/799633/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/33/799633/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/33/799633/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/33/799633/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/33/799633/7"}}},"commit":{"parents":[{"commit":"205880dc7698f4c0020f6b9d99c14c2728315326","subject":"Add base poloicy check module","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/205880dc7698f4c0020f6b9d99c14c2728315326"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-07-16 19:44:29.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-07-19 16:58:45.000000000","tz":0},"subject":"POC: Move metadef namepsace policy checks in the API","message":"POC: Move metadef namepsace policy checks in the API\n\nNOTE: This has changes to make existing tests pass, but likely needs\nadditional coverage for full verification.\n\nThis makes us able to optionally not get the authorization layer on\ntop of the stack. It also don\u0027t insert the policy layer in the\nmiddle of the stack.\n\nPartially implements: blueprint policy-refactor\nChange-Id: I59cd3bbd98065457bf0c9cfaf6888b7f56f7617c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/fafacac534478e32d7bc500a87299030bc46b3de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/fafacac534478e32d7bc500a87299030bc46b3de"}]},"branch":"refs/heads/master"},"d3a32eae038d11894c3edd729dce22419d0c3e9c":{"kind":"REWORK","_number":8,"created":"2021-07-21 16:46:17.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/33/799633/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/33/799633/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/33/799633/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/33/799633/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/33/799633/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/33/799633/8"}}},"commit":{"parents":[{"commit":"963e28fff04bfc2810fa48bd4ad38486c8a3d1fa","subject":"Add base policy check module","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/963e28fff04bfc2810fa48bd4ad38486c8a3d1fa"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-07-16 19:44:29.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-07-21 16:43:05.000000000","tz":0},"subject":"Move metadef namepsace policy checks in the API","message":"Move metadef namepsace policy checks in the API\n\nThis makes us able to optionally not get the authorization layer on\ntop of the stack. It also don\u0027t insert the policy layer in the\nmiddle of the stack.\n\nThis change also includes change in index method for \u0027GET\u0027 all\nnamespaces to not include properties, objects, tags, associations\netc. in the response if policies of either one is disabled.\n\nPartially implements: blueprint policy-refactor\nChange-Id: I59cd3bbd98065457bf0c9cfaf6888b7f56f7617c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d3a32eae038d11894c3edd729dce22419d0c3e9c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d3a32eae038d11894c3edd729dce22419d0c3e9c"}]},"branch":"refs/heads/master"},"0963a78749f3e4f93f5dad502c876d5d2af1e565":{"kind":"REWORK","_number":9,"created":"2021-07-22 09:46:07.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/33/799633/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/33/799633/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/33/799633/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/33/799633/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/33/799633/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/33/799633/9"}}},"commit":{"parents":[{"commit":"e8542b36577473fc9115aa65b9c5b5f13a0a5fb4","subject":"Refactor gateway auth layer for metadef APIs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e8542b36577473fc9115aa65b9c5b5f13a0a5fb4"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-07-16 19:44:29.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-07-22 05:47:18.000000000","tz":0},"subject":"Move metadef namepsace policy checks in the API","message":"Move metadef namepsace policy checks in the API\n\nThis change also includes change in index method for \u0027GET\u0027 all\nnamespaces to not include properties, objects, tags, associations\netc. in the response if policies of either one is disabled.\n\nPartially implements: blueprint policy-refactor\nChange-Id: I59cd3bbd98065457bf0c9cfaf6888b7f56f7617c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0963a78749f3e4f93f5dad502c876d5d2af1e565"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0963a78749f3e4f93f5dad502c876d5d2af1e565"}]},"branch":"refs/heads/master"},"89b1b09f0fb93b23eed77562065f216e17cc4416":{"kind":"REWORK","_number":10,"created":"2021-07-22 18:13:40.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/33/799633/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/33/799633/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/33/799633/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/33/799633/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/33/799633/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/33/799633/10"}}},"commit":{"parents":[{"commit":"a6f3adf910c05262d301e9569e8da989ead704bc","subject":"Refactor gateway auth layer for metadef APIs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a6f3adf910c05262d301e9569e8da989ead704bc"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-07-16 19:44:29.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-07-22 18:12:33.000000000","tz":0},"subject":"Move metadef namepsace policy checks in the API","message":"Move metadef namepsace policy checks in the API\n\nThis change also includes change in index method for \u0027GET\u0027 all\nnamespaces to not include properties, objects, tags, associations\netc. in the response if policies of either one is disabled.\n\nPartially implements: blueprint policy-refactor\nChange-Id: I59cd3bbd98065457bf0c9cfaf6888b7f56f7617c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/89b1b09f0fb93b23eed77562065f216e17cc4416"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/89b1b09f0fb93b23eed77562065f216e17cc4416"}]},"branch":"refs/heads/master"},"0c4f7c9cdfa19fce98832e859bf0774d568bb919":{"kind":"REWORK","_number":11,"created":"2021-07-22 18:20:27.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/33/799633/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/33/799633/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/33/799633/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/33/799633/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/33/799633/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/33/799633/11"}}},"commit":{"parents":[{"commit":"a6f3adf910c05262d301e9569e8da989ead704bc","subject":"Refactor gateway auth layer for metadef APIs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a6f3adf910c05262d301e9569e8da989ead704bc"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-07-16 19:44:29.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-07-22 18:19:40.000000000","tz":0},"subject":"Move metadef namepsace policy checks in the API","message":"Move metadef namepsace policy checks in the API\n\nThis change also includes change in index method for \u0027GET\u0027 all\nnamespaces to not include properties, objects, tags, associations\netc. in the response if policies of either one is disabled.\n\nPartially implements: blueprint policy-refactor\nChange-Id: I59cd3bbd98065457bf0c9cfaf6888b7f56f7617c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0c4f7c9cdfa19fce98832e859bf0774d568bb919"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0c4f7c9cdfa19fce98832e859bf0774d568bb919"}]},"branch":"refs/heads/master"},"05d3b6c4bf75515b05afe795e8edf5acd4bc5964":{"kind":"REWORK","_number":12,"created":"2021-07-24 10:02:03.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/33/799633/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/33/799633/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/33/799633/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/33/799633/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/33/799633/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/33/799633/12"}}},"commit":{"parents":[{"commit":"a6f3adf910c05262d301e9569e8da989ead704bc","subject":"Refactor gateway auth layer for metadef APIs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a6f3adf910c05262d301e9569e8da989ead704bc"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-07-16 19:44:29.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-07-24 06:53:44.000000000","tz":0},"subject":"Move metadef namepsace policy checks in the API","message":"Move metadef namepsace policy checks in the API\n\nThis change also includes change in index method for \u0027GET\u0027 all\nnamespaces to not include properties, objects, tags, associations\netc. in the response if policies of either one is disabled.\n\nPartially implements: blueprint policy-refactor\nChange-Id: I59cd3bbd98065457bf0c9cfaf6888b7f56f7617c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/05d3b6c4bf75515b05afe795e8edf5acd4bc5964"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/05d3b6c4bf75515b05afe795e8edf5acd4bc5964"}]},"branch":"refs/heads/master"},"8a19a3a3277024cc78d7f58f4fafb37912cf6670":{"kind":"REWORK","_number":13,"created":"2021-08-02 08:46:26.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/33/799633/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/33/799633/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/33/799633/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/33/799633/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/33/799633/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/33/799633/13"}}},"commit":{"parents":[{"commit":"00f11f5a7c2584c4c611491c399aface56c37559","subject":"Refactor gateway auth layer for metadef APIs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/00f11f5a7c2584c4c611491c399aface56c37559"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-07-16 19:44:29.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-08-02 08:11:57.000000000","tz":0},"subject":"Move metadef namepsace policy checks in the API","message":"Move metadef namepsace policy checks in the API\n\nThis change also includes change in index method for \u0027GET\u0027 all\nnamespaces to not include properties, objects, tags, associations\netc. in the response if policies of either one is disabled.\n\nPartially implements: blueprint policy-refactor\n\nChange-Id: I59cd3bbd98065457bf0c9cfaf6888b7f56f7617c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8a19a3a3277024cc78d7f58f4fafb37912cf6670"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8a19a3a3277024cc78d7f58f4fafb37912cf6670"}]},"branch":"refs/heads/master"},"b8b8cdbbc3ba2bb66c1f2472723ef46a627c1c4e":{"kind":"REWORK","_number":14,"created":"2021-08-02 18:11:39.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/33/799633/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/33/799633/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/33/799633/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/33/799633/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/33/799633/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/33/799633/14"}}},"commit":{"parents":[{"commit":"00f11f5a7c2584c4c611491c399aface56c37559","subject":"Refactor gateway auth layer for metadef APIs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/00f11f5a7c2584c4c611491c399aface56c37559"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-07-16 19:44:29.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-08-02 17:12:44.000000000","tz":0},"subject":"Move metadef namepsace policy checks in the API","message":"Move metadef namepsace policy checks in the API\n\nThis change also includes change in index method for \u0027GET\u0027 all\nnamespaces to not include properties, objects, tags, associations\netc. in the response if policies of either one is disabled.\n\nPartially implements: blueprint policy-refactor\n\nChange-Id: I59cd3bbd98065457bf0c9cfaf6888b7f56f7617c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b8b8cdbbc3ba2bb66c1f2472723ef46a627c1c4e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b8b8cdbbc3ba2bb66c1f2472723ef46a627c1c4e"}]},"branch":"refs/heads/master"},"41d9ea77142e2d199016d75e2a49e5057c4d89c1":{"kind":"NO_CODE_CHANGE","_number":15,"created":"2021-08-03 14:54:49.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/33/799633/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/33/799633/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/33/799633/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/33/799633/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/33/799633/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/33/799633/15"}}},"commit":{"parents":[{"commit":"00f11f5a7c2584c4c611491c399aface56c37559","subject":"Refactor gateway auth layer for metadef APIs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/00f11f5a7c2584c4c611491c399aface56c37559"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-07-16 19:44:29.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-08-03 14:50:48.000000000","tz":0},"subject":"Move metadef namespace policy checks in the API","message":"Move metadef namespace policy checks in the API\n\nThis change also includes change in index method for \u0027GET\u0027 all\nnamespaces to not include properties, objects, tags, associations\netc. in the response if policies of either one is disabled.\n\nPartially implements: blueprint policy-refactor\n\nChange-Id: I59cd3bbd98065457bf0c9cfaf6888b7f56f7617c\nDepends-On: https://review.opendev.org/c/openstack/grenade/+/803317\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/41d9ea77142e2d199016d75e2a49e5057c4d89c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/41d9ea77142e2d199016d75e2a49e5057c4d89c1"}]},"branch":"refs/heads/master"},"9857fb65e944ab87d718f2948583574baade320d":{"kind":"TRIVIAL_REBASE","_number":16,"created":"2021-08-04 15:39:38.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/33/799633/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/33/799633/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/33/799633/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/33/799633/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/33/799633/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/33/799633/16"}}},"commit":{"parents":[{"commit":"0cc8190d0edef73ffa7ee64c527e2d083cbb8bd1","subject":"Refactor gateway auth layer for metadef APIs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0cc8190d0edef73ffa7ee64c527e2d083cbb8bd1"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-07-16 19:44:29.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-08-04 14:58:04.000000000","tz":0},"subject":"Move metadef namespace policy checks in the API","message":"Move metadef namespace policy checks in the API\n\nThis change also includes change in index method for \u0027GET\u0027 all\nnamespaces to not include properties, objects, tags, associations\netc. in the response if policies of either one is disabled.\n\nPartially implements: blueprint policy-refactor\n\nChange-Id: I59cd3bbd98065457bf0c9cfaf6888b7f56f7617c\nDepends-On: https://review.opendev.org/c/openstack/grenade/+/803317\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9857fb65e944ab87d718f2948583574baade320d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9857fb65e944ab87d718f2948583574baade320d"}]},"branch":"refs/heads/master"},"ba608deb7daa8d309a02e85cd50da8f8ea6dce05":{"kind":"REWORK","_number":17,"created":"2021-08-05 11:34:56.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/33/799633/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/33/799633/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/33/799633/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/33/799633/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/33/799633/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/33/799633/17"}}},"commit":{"parents":[{"commit":"1acf6e6faaf95d1359a15bf042d826b3332db710","subject":"Refactor gateway auth layer for metadef APIs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1acf6e6faaf95d1359a15bf042d826b3332db710"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-07-16 19:44:29.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-08-05 10:29:15.000000000","tz":0},"subject":"Move metadef namespace policy checks in the API","message":"Move metadef namespace policy checks in the API\n\nThis change also includes change in index method for \u0027GET\u0027 all\nnamespaces to not include properties, objects, tags, associations\netc. in the response if policies of either one is disabled.\n\nPartially implements: blueprint policy-refactor\n\nChange-Id: I59cd3bbd98065457bf0c9cfaf6888b7f56f7617c\nDepends-On: https://review.opendev.org/c/openstack/grenade/+/803317\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ba608deb7daa8d309a02e85cd50da8f8ea6dce05"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ba608deb7daa8d309a02e85cd50da8f8ea6dce05"}]},"branch":"refs/heads/master"},"1e4c9a890297a5da794d960f3e73b05daaa2a23a":{"kind":"REWORK","_number":18,"created":"2021-08-05 19:02:40.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/33/799633/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/33/799633/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/33/799633/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/33/799633/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/33/799633/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/33/799633/18"}}},"commit":{"parents":[{"commit":"2943741c8600bd1bd1e91c3310fb36630a368d23","subject":"Refactor gateway auth layer for metadef APIs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2943741c8600bd1bd1e91c3310fb36630a368d23"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-07-16 19:44:29.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-08-05 18:08:09.000000000","tz":0},"subject":"Move metadef namespace policy checks in the API","message":"Move metadef namespace policy checks in the API\n\nThis change also includes change in index method for \u0027GET\u0027 all\nnamespaces to not include properties, objects, tags, associations\netc. in the response if policies of either one is disabled.\n\nPartially implements: blueprint policy-refactor\n\nChange-Id: I59cd3bbd98065457bf0c9cfaf6888b7f56f7617c\nDepends-On: https://review.opendev.org/c/openstack/grenade/+/803317\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1e4c9a890297a5da794d960f3e73b05daaa2a23a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1e4c9a890297a5da794d960f3e73b05daaa2a23a"}]},"branch":"refs/heads/master"},"caeb34a7804e1cd5f36fbd30d1e0bb6f482db792":{"kind":"REWORK","_number":19,"created":"2021-08-10 18:51:54.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/33/799633/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/33/799633/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/33/799633/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/33/799633/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/33/799633/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/33/799633/19"}}},"commit":{"parents":[{"commit":"ccac6b6c6ba054a9f51deb0b7db9e93118621458","subject":"Merge \"Check delete_image policy in the API\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ccac6b6c6ba054a9f51deb0b7db9e93118621458"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-07-16 19:44:29.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-08-10 18:46:48.000000000","tz":0},"subject":"Move metadef namespace policy checks in the API","message":"Move metadef namespace policy checks in the API\n\nThis change also includes change in index method for \u0027GET\u0027 all\nnamespaces to not include properties, objects, tags, associations\netc. in the response if policies of either one is disabled.\n\nPartially implements: blueprint policy-refactor\n\nChange-Id: I59cd3bbd98065457bf0c9cfaf6888b7f56f7617c\nDepends-On: https://review.opendev.org/c/openstack/grenade/+/803317\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/caeb34a7804e1cd5f36fbd30d1e0bb6f482db792"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/caeb34a7804e1cd5f36fbd30d1e0bb6f482db792"}]},"branch":"refs/heads/master"},"fa18ecaa4656c342a48882e0e60d1036dac0a62f":{"kind":"REWORK","_number":20,"created":"2021-08-10 21:58:09.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/33/799633/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/33/799633/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/33/799633/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/33/799633/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/33/799633/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/33/799633/20"}}},"commit":{"parents":[{"commit":"ccac6b6c6ba054a9f51deb0b7db9e93118621458","subject":"Merge \"Check delete_image policy in the API\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ccac6b6c6ba054a9f51deb0b7db9e93118621458"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-07-16 19:44:29.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-08-10 21:52:28.000000000","tz":0},"subject":"Move metadef namespace policy checks in the API","message":"Move metadef namespace policy checks in the API\n\nThis change also includes change in index method for \u0027GET\u0027 all\nnamespaces to not include properties, objects, tags, associations\netc. in the response if policies of either one is disabled.\n\nThe MetadefAPIPolicy object will raise Forbidden for cases where an\noperation is forbidden, but the namespace would be otherwise visible to\nthe user, and NotFound otherwise to obscure the existence of namespaces\nthat the user can not otherwise see.\n\nPartially implements: blueprint policy-refactor\n\nChange-Id: I59cd3bbd98065457bf0c9cfaf6888b7f56f7617c\nDepends-On: https://review.opendev.org/c/openstack/grenade/+/803317\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/fa18ecaa4656c342a48882e0e60d1036dac0a62f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/fa18ecaa4656c342a48882e0e60d1036dac0a62f"}]},"branch":"refs/heads/master"},"7e919e8f9cd8cada42364207ea0aebd3a671a289":{"kind":"REWORK","_number":21,"created":"2021-08-11 15:20:42.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/33/799633/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/33/799633/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/33/799633/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/33/799633/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/33/799633/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/33/799633/21"}}},"commit":{"parents":[{"commit":"ccac6b6c6ba054a9f51deb0b7db9e93118621458","subject":"Merge \"Check delete_image policy in the API\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ccac6b6c6ba054a9f51deb0b7db9e93118621458"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-07-16 19:44:29.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-08-11 15:16:23.000000000","tz":0},"subject":"Move metadef namespace policy checks in the API","message":"Move metadef namespace policy checks in the API\n\nThis change also includes change in index method for \u0027GET\u0027 all\nnamespaces to not include properties, objects, tags, associations\netc. in the response if policies of either one is disabled.\n\nThe MetadefAPIPolicy object will raise Forbidden for cases where an\noperation is forbidden, but the namespace would be otherwise visible to\nthe user, and NotFound otherwise to obscure the existence of namespaces\nthat the user can not otherwise see.\n\nPartially implements: blueprint policy-refactor\n\nChange-Id: I59cd3bbd98065457bf0c9cfaf6888b7f56f7617c\nDepends-On: https://review.opendev.org/c/openstack/grenade/+/803317\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7e919e8f9cd8cada42364207ea0aebd3a671a289"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7e919e8f9cd8cada42364207ea0aebd3a671a289"}]},"branch":"refs/heads/master"},"99331d7efacb556effc9dda2a7db06bdc6af0319":{"kind":"REWORK","_number":22,"created":"2021-08-11 19:03:14.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/33/799633/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/33/799633/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/33/799633/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/33/799633/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/33/799633/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/33/799633/22"}}},"commit":{"parents":[{"commit":"d9b8023410348e769f8a69c168461126ce7ff6fc","subject":"Refactor gateway auth layer for member APIs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d9b8023410348e769f8a69c168461126ce7ff6fc"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-07-16 19:44:29.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-08-11 18:59:01.000000000","tz":0},"subject":"Move metadef namespace policy checks in the API","message":"Move metadef namespace policy checks in the API\n\nThis change also includes change in index method for \u0027GET\u0027 all\nnamespaces to not include properties, objects, tags, associations\netc. in the response if policies of either one is disabled.\n\nThe MetadefAPIPolicy object will raise Forbidden for cases where an\noperation is forbidden, but the namespace would be otherwise visible to\nthe user, and NotFound otherwise to obscure the existence of namespaces\nthat the user can not otherwise see.\n\nPartially implements: blueprint policy-refactor\n\nChange-Id: I59cd3bbd98065457bf0c9cfaf6888b7f56f7617c\nDepends-On: https://review.opendev.org/c/openstack/grenade/+/803317\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/99331d7efacb556effc9dda2a7db06bdc6af0319"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/99331d7efacb556effc9dda2a7db06bdc6af0319"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
