)]}'
{"id":"openstack%2Fglance~799699","triplet_id":"openstack%2Fglance~master~I0084350ebb09cc1cb3752b45165e49f166bfdf91","project":"openstack/glance","branch":"master","topic":"bp/policy-refactor","hashtags":[],"change_id":"I0084350ebb09cc1cb3752b45165e49f166bfdf91","subject":"Remove dead 403-\u003e404 code","status":"MERGED","created":"2021-07-06 18:00:34.000000000","updated":"2021-07-29 08:44:19.000000000","submitted":"2021-07-29 08:42:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"799699-bp/policy-refactor","meta_rev_id":"594164b4d69bc9fc577687431260dfc544d7ff1e","_number":799699,"virtual_id_number":799699,"owner":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-29 08:42:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2021-07-29 08:44:19.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2021-07-28 18:46:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":1,"date":"2021-07-28 18:46:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2021-07-29 08:44:19.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-06 19:24:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-16 13:44:56.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2021-07-19 15:43:18.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"CC"},{"updated":"2021-07-20 17:07:31.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-07-28 18:46:30.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"}],"messages":[{"id":"1b061e4aef4af7c51dd3ef97708e7337033dc0a8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-07-06 18:00:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f3954ad761ca2aacb5a41f21bb68348d88b97dac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-06 19:24:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/c738b992125249689078843a90667cfc : SUCCESS in 1h 10m 00s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3e0837e0d0494ebb8afa0f7c1e49fe9d : SUCCESS in 1h 05m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7af27152cebc411aa5ef6f863956c87f : SUCCESS in 6m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eabb9076a9834035b42554d302ad938d : SUCCESS in 6m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f620c9de8cbd4c02a89f97a2a46afd3e : SUCCESS in 6m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fd1c4af6668649eca397419399115ecd : SUCCESS in 7m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/159454334e6b4b7aa58f6970b91f298e : SUCCESS in 6m 44s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c14a6c36e79640f59f8667a9db628673 : SUCCESS in 9m 50s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/aec18cd2c37e4eff9d4f84681541b0be : SUCCESS in 11m 24s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/b8edc3eff144424484338be546e06bec : FAILURE in 8m 10s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/d0c8b5e4fe294770bad995b7ed9d4fbc : SUCCESS in 6m 10s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/1939ffb099c54fe4be68dd4b4fe9e639 : SUCCESS in 1h 08m 52s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/7973dbae89f041458b8053cc92fa528a : SUCCESS in 1h 11m 26s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/12a96678ab564160a9ed0a640651fc18 : SUCCESS in 1h 16m 10s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/44057bd59d4b47c4aa504526a4ecf196 : SUCCESS in 1h 09m 35s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4d005a9e362d4d90b9e33399fef3555a : SUCCESS in 55m 01s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/c1729b8850fb4a84aa5a00dc30c2a461 : SUCCESS in 1h 13m 17s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/9972dab406b04e278af19d587f073815 : SUCCESS in 34m 20s","accounts_in_message":[],"_revision_number":1},{"id":"43cae8672a761e7aa4d83ffc33f16d4767eadb87","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-07-14 14:03:38.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"9ae4f1dd5e790c424d607050f2de44252a3b1435","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-14 15:31:20.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/3ff47c10cf0d4c4e93a40771160b808d : SUCCESS in 1h 05m 57s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/da41e673fcd6476baf5b9ae427dba055 : SUCCESS in 1h 01m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/19ac258a614d42faa28d5c590a7e6247 : SUCCESS in 5m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5983fe40223f46b29a3097dcec334bea : SUCCESS in 6m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2ec909de43164c67857798a9d2d56512 : SUCCESS in 5m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8cb188334aeb44c997d6f0ea433f6802 : SUCCESS in 6m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9f43436d9f848b087ffd814330333ec : SUCCESS in 7m 15s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/72fae0121c184750b74cf20f13e64f73 : SUCCESS in 9m 23s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/f938f5bde89a40ad917a718d8d74519d : SUCCESS in 8m 46s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/b474a16f769d4efebebbef25be46eaf7 : FAILURE in 8m 50s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/2b7c8dedb28543bda6f5de00c161f83d : SUCCESS in 6m 21s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/1f9094f85133490e9ef2956b1567a517 : SUCCESS in 1h 22m 41s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/39e692dd92d94c8581a4eeec408ddd39 : SUCCESS in 1h 14m 18s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/489875a81c7f4cab9bdca139beb4a2f2 : SUCCESS in 1h 07m 25s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/26a615078e7742db9f45ecb1a1b4bc1c : SUCCESS in 1h 15m 54s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/45181380a7714bed93555cb57a8fbc16 : SUCCESS in 53m 25s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/ae472f3c711042cb9d37330e81d536dc : SUCCESS in 1h 15m 56s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/b2b266dfe77b4cc0987a5c1c955229b6 : SUCCESS in 37m 03s","accounts_in_message":[],"_revision_number":2},{"id":"884422e3d9805d574cc5523e3f7fa5143fd7fc4e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-07-14 19:32:52.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"9d86f7c22ac7c68814be19628a49281cab7c5d1f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-14 21:02:06.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/ab8f1601b8c14737bc146c2034a710dd : SUCCESS in 1h 20m 36s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5a6e7f366f6b42ea9086516025e3c165 : SUCCESS in 1h 04m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/574a35d9c5c54d599795c30927ae399e : SUCCESS in 5m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3f44b9b38edb4b0ba5dc9c2643f7d2c0 : SUCCESS in 6m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c45816ca74d24b3d8b85a2f4a13d2076 : SUCCESS in 6m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5afd897555404dba88357835aeb72bf7 : SUCCESS in 7m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bca716c13d90436b86a7a9dca9da8511 : SUCCESS in 7m 55s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a431a11f428e44fca2fc0e510a42b587 : SUCCESS in 9m 23s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/86a182306a7a4740b8ac782b8a4cbad4 : SUCCESS in 9m 23s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/2ec9d5580d9f488aa38b1e31a49450e0 : SUCCESS in 8m 32s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/e0963dbe495840de80d5a9a2b9bed6d8 : SUCCESS in 5m 51s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/60eec82aed654a0e87fcce4773cac0ff : SUCCESS in 1h 07m 34s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/fea498e54ab54e66b8f9dc793437db85 : SUCCESS in 1h 02m 00s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/52c4c167341646a6b7669ba7d559b316 : SUCCESS in 1h 19m 46s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/a069aac316cb40929119d9015c0ea364 : SUCCESS in 1h 11m 25s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/08e23dc00c2d4809b790dc7daf5fbc21 : SUCCESS in 43m 13s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/477e3a43ce7640229f4c52f25a2981df : SUCCESS in 1h 19m 50s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/60126bfe3be54cb7bcf62988d2bea4e1 : SUCCESS in 30m 52s","accounts_in_message":[],"_revision_number":3},{"id":"722e81a2cb14616ad37e4c38b3686d5c853f8fb1","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-07-16 13:44:56.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)\n\nLooks good, one suggestion inline.","accounts_in_message":[],"_revision_number":3},{"id":"049d6a9fdb0dc50c96dbad0f7c6b1325aef6146a","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-07-16 13:53:59.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f5bdbea132a57af4761e20337cc0e9dfe9908e3c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-07-16 13:56:40.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"ac5c3b973f401a6b7d5dbe7a78ca6db536a1b2dc","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-07-16 14:02:13.000000000","message":"Patch Set 4: Code-Review+2\n\nLooks perfect, Thank you!!","accounts_in_message":[],"_revision_number":4},{"id":"58eb53c8674fa08571d8a9bf2dfe7074315a0a2a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-16 15:24:27.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/a0271f65f3144fce86100c504805bb03 : SUCCESS in 1h 03m 32s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/862ef83ef47144bf9a3679819f5a5208 : SUCCESS in 1h 02m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d159f5bef8ca4a57b17a721a1d6c90cd : SUCCESS in 6m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d46fd8aae7f64a1caecfefe7a0b84275 : FAILURE in 6m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/18d9076e23a347d980f9fecb8ab9d85e : FAILURE in 6m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e4a6970d76244ce6bd4203bbca9a36db : FAILURE in 7m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/207e6fed5b274bee9f64d08c48d6b32b : SUCCESS in 7m 26s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6ffc01d87aa94a2d9de95ef4147ed212 : SUCCESS in 9m 03s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/1314af028a75404aaa104b80e9e4b99a : SUCCESS in 10m 44s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/d3089c0881704a6cbc3b20f9f87b7d45 : SUCCESS in 9m 36s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/b7ad7a4974e94a15a90f0b6a73f7ac17 : SUCCESS in 8m 26s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/90081cdbe8bc4803a927f42f22a97d26 : SUCCESS in 1h 08m 42s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/e9b1dcf9c3df4c7e828680b09783b71f : SUCCESS in 1h 23m 20s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/60073be096284483bb521eb034c4270c : SUCCESS in 1h 06m 34s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/2ab24267396d4576b7fee98802b22c9b : SUCCESS in 1h 20m 36s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/778817dc9e4e46beb17e7ba0ba0d3f03 : SUCCESS in 49m 24s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/a0c4c026f2b64a8583eba203831c71d9 : SUCCESS in 1h 17m 45s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/7191531b5f664d97a7b58e2aec63a506 : SUCCESS in 32m 39s","accounts_in_message":[],"_revision_number":4},{"id":"fb52c5734e6a4468b2320ed39c8803149b5210e5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-07-16 15:28:57.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"19145c4b6698f9c701f217e8bcde670f31ddc7ca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-16 17:05:38.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/56ffefaace87407daf69b5953c82dec1 : SUCCESS in 1h 04m 53s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8a10ad39afd54463ad16f0e07317a782 : SUCCESS in 1h 30m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7671f57c1ed8413295e0ecc8a48ae20e : SUCCESS in 6m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7d9908a3965747a88d99c855db023820 : SUCCESS in 8m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c02f259b1f8b4fb3940a2025e37bb027 : SUCCESS in 7m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a78752c8d5af4eeab65b0bebb971dea8 : SUCCESS in 8m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c09fc28d54342b399632202a1ee67f5 : SUCCESS in 9m 11s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/cbcd9c325cd348c2b7e8fe1c63e22034 : SUCCESS in 10m 39s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/612b9c7eee134845a8cbda7bc5b6dfa2 : SUCCESS in 9m 15s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/b6a1c473ee934092aa76a9318d6f1a1a : SUCCESS in 10m 00s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/6dbfe52c2e844ed882ef23c02be39e57 : SUCCESS in 6m 19s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/cf8549981dfc4f988c1ad7ad28159356 : SUCCESS in 1h 14m 17s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/56b6beef7ef041ef99481da6a1707ea7 : SUCCESS in 1h 22m 20s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/1bd79907ff3f416b855bc8d46f9616cd : SUCCESS in 1h 22m 49s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/eb0da553ef7d4442aff950af114696d1 : SUCCESS in 1h 14m 06s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bc73a8644aa845319d9487bf21027e71 : SUCCESS in 53m 06s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/3b57f38c8b8446189d80ece69dac7e20 : SUCCESS in 1h 13m 11s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/b5f7cfaf532a4df1ade23944c315d2d4 : SUCCESS in 34m 25s","accounts_in_message":[],"_revision_number":5},{"id":"143cbe3c3227b3d44aeba198dcfcb2a614ae37f2","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2021-07-19 15:43:18.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"ddd51b190fb79d0785470bee9d628b6cfd09fb7c","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-07-19 15:48:55.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"6a0ffae9eb73ae0f52100e8975d334d03a5a82ff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-07-20 16:30:14.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"183c174698aa1f342a4d265b9765dcf9491a27ad","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-07-20 17:07:31.000000000","message":"Patch Set 6:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/799699/799699/6/check-not-vote/ext-glance-zuul/8430f3d : SUCCESS in 36m 41s","accounts_in_message":[],"_revision_number":6},{"id":"d057a4eca1edc426c25ef5676a403357ab110ab5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-20 17:59:17.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/f6e3d25f4e7944a9855016a4dd454230 : SUCCESS in 1h 12m 40s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/61baa7a99095477f9ce2cff4c61bcb90 : SUCCESS in 1h 04m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ebb5b44db48b454f87ad03143d4819e3 : SUCCESS in 6m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5f55062226a641d0a4bb9ec887fe7ef7 : SUCCESS in 7m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c1cdfb99ce8c462ea5404f3a57c7fba9 : SUCCESS in 11m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c52d643a15ea4142be69c56cfe9efe66 : SUCCESS in 7m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87fd482a0d3d412dbc62cb8191cf7e9c : SUCCESS in 6m 29s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2472a8e5cd044911865afa8d7615e1aa : SUCCESS in 9m 11s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/83770d054ad7440ebd33754a92faeace : SUCCESS in 9m 18s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/710695c1f15441eba52fac07a4c26ee1 : SUCCESS in 9m 18s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/88900b611b8548469b0f8ae6ca3f3fff : SUCCESS in 5m 38s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/911c35102d134e578ae295c67884517a : FAILURE in 1h 15m 52s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/f9f1654f5c8b4834a038739ef1203184 : SUCCESS in 58m 45s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/41331fef76bb4a20b0d922accaa3d161 : SUCCESS in 1h 17m 19s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/37d2f82fc2a44980a6519f9de960ecbb : FAILURE in 1h 06m 05s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fbbfae1d774340e188a06a64e4220740 : SUCCESS in 59m 20s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/618f615fb5214d4689b8ffef479bdcde : SUCCESS in 1h 22m 05s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/8c142493d5a2433b8b38c1915c153a6c : SUCCESS in 28m 27s","accounts_in_message":[],"_revision_number":6},{"id":"2b4ddc023ca4084a30309ed2d7b3f97dfcf4f829","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-07-22 14:33:15.000000000","message":"Patch Set 6: Code-Review+2 Review-Priority+1\n\nThank you Dan,\nI have added comment on previous patch set to explain why change in simple db is needed.","accounts_in_message":[],"_revision_number":6},{"id":"49b48df9339bf5061f7ad7f8659d3ce3d55b2802","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-07-26 19:49:00.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.","accounts_in_message":[],"_revision_number":7},{"id":"facc30538dad122a56921e70e31ce3b61e2b9d23","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-26 21:40:54.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/ccb7472de85d4a439a019c892c326647 : SUCCESS in 55m 34s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/320d2a16cc454d80b0f35948ceff85dd : SUCCESS in 1h 21m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db084b44e75145468938d2357c50d1c3 : SUCCESS in 5m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/63f89f8ff5c948c48103a21c61358db9 : SUCCESS in 6m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e856f21a92e94403a9e7654905a17d1f : SUCCESS in 5m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/434fc15b6f8847be908093a7684eedbf : SUCCESS in 7m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10044fdda797441ea6fb63e618b2cc6a : SUCCESS in 6m 13s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/bc9cf07bd65549549511f68f46650284 : SUCCESS in 9m 12s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/aca654bea6544fc28b3141aff3a438d0 : SUCCESS in 9m 01s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/355b43928ec94fb097137a00bee061a5 : SUCCESS in 9m 17s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/4d2120166d5746708ab49f52e7e0da37 : SUCCESS in 5m 02s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/22c1f619f2a945c4bb38ea4ddece3cf1 : SUCCESS in 1h 18m 02s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/71667e826adc4d5386b870a924d795e6 : SUCCESS in 56m 45s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/fbd9367be11d4c80a6fda84c3a1173a9 : SUCCESS in 1h 14m 20s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/12b9ac8767384411b5c4f14d5124e745 : SUCCESS in 1h 03m 26s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/435af93a061f43b8b22b2975385dcd4f : SUCCESS in 51m 59s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/c4ab7cbfebbb486eadb777f0a4ab4aa1 : SUCCESS in 1h 47m 42s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/97ffb16da2e5483f95c78fd8b10d7fbf : SUCCESS in 28m 47s","accounts_in_message":[],"_revision_number":7},{"id":"644601bd69eda2303d7abb6d1de2b2d883327a3b","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-07-26 22:37:56.000000000","message":"Patch Set 7:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/799699/799699/7/check-not-vote/ext-glance-zuul/4e89d15 : SUCCESS in 54m 20s","accounts_in_message":[],"_revision_number":7},{"id":"59e9d6d4dc7bdefee2c9b28ebd542f1403efe78d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-07-27 18:34:02.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.","accounts_in_message":[],"_revision_number":8},{"id":"264809b8457a07952d74481c99ef6769127050d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-27 20:06:36.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/71ec4b88a115437bad9863e1e3036fa6 : SUCCESS in 1h 06m 56s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/45cf2ac142a64aefb4a0b9cf624c66b8 : SUCCESS in 1h 15m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e949840573f04b04a311d31e50329863 : SUCCESS in 7m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cc96dc5a30f34ef196fe9c0665164ad6 : SUCCESS in 7m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a544ee484e26480db848dabab709820a : SUCCESS in 6m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/62fcdadd02d74980a5665dc2a2ad4e2b : SUCCESS in 7m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c48fc64662504ae091e2552a756978c7 : SUCCESS in 8m 38s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3178fbb513fb41a4a84f0a2cfbd967be : SUCCESS in 9m 25s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/ff4fcbb34f884de499f7b29163828d41 : SUCCESS in 11m 32s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/56e63ee7b9454e8d9fa29dcd7c91d0b9 : SUCCESS in 9m 48s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/c9c4ad741a82458a88cdd4dd060e5138 : SUCCESS in 6m 23s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/0c00b24d9ce2442880911bdf2808e406 : SUCCESS in 1h 25m 53s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/943e3a9c51164a6dbac5ed9d88ac4a41 : SUCCESS in 1h 09m 45s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/16eeb6b4d9ef4a8a93928047e1f87597 : SUCCESS in 1h 03m 03s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/92f73a05efe5409db9adda93bb5eb281 : SUCCESS in 1h 24m 36s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ad5fe688d1864fb5aa6dc6aeb858ab8d : SUCCESS in 51m 44s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/06daa7632e254b428940131bcfbf8a98 : SUCCESS in 1h 12m 38s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/c1b24d195c6940d29c518306f5d532aa : SUCCESS in 37m 02s","accounts_in_message":[],"_revision_number":8},{"id":"a62c0521afcacb85421712578ac2512339cad86f","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2021-07-28 18:46:30.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"f1c2f21c759ec6531c96dd7b38709778f165f48f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-28 18:47:07.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"520da90d825f2b631c77013908272a03af6ba1d1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-28 19:56:48.000000000","message":"Patch Set 8:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"a53c96358ea8ad9392143dca698bca9cf1218d65","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-28 22:01:21.000000000","message":"Patch Set 8: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/1e9da820ce75487182a38f99b63aab18 : SUCCESS in 52m 47s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c50b27b04f1b4fd49a5564daa9f975b9 : FAILURE in 59m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/848bd4b0681442eb8dc15b409ed2a94d : SUCCESS in 5m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f5b757917ed6434bbf79128bae30b71e : SUCCESS in 6m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bc6c9ed4270d489cb7ed9d8e7d9c09f8 : SUCCESS in 6m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b38f16b9aae14723a7700991408eca57 : SUCCESS in 7m 06s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/892b13f4a84848659d0b7105c74e929e : TIMED_OUT in 31m 02s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/1bff23fb851349789150668c0932a5bc : SUCCESS in 8m 49s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/681287da47d44b8fa320298d32aceffe : SUCCESS in 1h 01m 50s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/993a3034e7264a7b80ce172863538953 : SUCCESS in 1h 10m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/eea296ef231b40b0a1435e92af02ce3f : SUCCESS in 46m 52s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/82b60e25f2e94bf8af060677e84d917e : SUCCESS in 1h 11m 35s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/1b402af9692446d3b778d3f92377c2d6 : SUCCESS in 34m 03s","accounts_in_message":[],"_revision_number":8},{"id":"bce55ed982382d270716c8850dcff262f9267a18","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-07-29 05:58:37.000000000","message":"Patch Set 8:\n\nrecheck unrelated error","accounts_in_message":[],"_revision_number":8},{"id":"7d11684049a006e6b44547710dee13dd8340f25c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-29 07:13:01.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/619095a921264ff0b7ed5e6219698caa : SUCCESS in 1h 01m 33s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0f7ec2a261d641c18875d93f338d43a9 : SUCCESS in 1h 02m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5b43338c0e0a4fc6a23ef8eb32c1c6d4 : SUCCESS in 5m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/64d87aff66ec4695a1a8cc6eb0b3c404 : SUCCESS in 5m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6462eb8b2e0442a49228409e36a2d048 : SUCCESS in 6m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e5f6898c4f9b4841978a18aceab8e395 : SUCCESS in 7m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad403dccc6af4239888093ee544dd956 : SUCCESS in 7m 13s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7d23023af7e04bde853a7c8ee9f733dd : SUCCESS in 9m 36s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c2c9193f6d78446fb41866c9fabc9728 : SUCCESS in 9m 50s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/fcafea8eb08845989e9f209a26314e30 : SUCCESS in 10m 10s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/4d3551f79982481a91ca14ced011fef5 : SUCCESS in 5m 38s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/56ad89830a74422ca41420ce8b59b1e7 : SUCCESS in 1h 06m 16s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/d9b999d15a364575b8662593b25e4584 : SUCCESS in 1h 07m 01s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/76eb2a11da214d97809e8101b17b6a31 : SUCCESS in 1h 07m 15s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/c115cc70a0bf40939438cd7907f7b18c : SUCCESS in 1h 12m 57s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/22a6b141670a427fb807fd1846e5b087 : SUCCESS in 51m 52s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/432630b993f14b648fe22201a3d3964f : SUCCESS in 1h 09m 38s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/db7d1f09c02145edbade39588ac851ab : SUCCESS in 31m 37s","accounts_in_message":[],"_revision_number":8},{"id":"1ca010913d08726e41362312e372cb6a810179b8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-29 07:13:33.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"260aa456aa2778bfe30e7709ec80f7eefcfc374e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-29 08:42:18.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/f83bdb64b9384223b7a92d2954d5f62d : SUCCESS in 59m 29s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/16d58d937ed24606a494a62bd2cbce28 : SUCCESS in 1h 03m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca5dbcc98ea44defb9539220c1663ce2 : SUCCESS in 5m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a6dfd4ffb25d452caf1b447a69ddaed1 : SUCCESS in 6m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e62753c98d1c44998baee29659cb16f1 : SUCCESS in 6m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7af778c343f24007b1da32752f9642fe : SUCCESS in 6m 50s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c46a3ad681304e57bea865be895ee07f : SUCCESS in 9m 55s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/d1f97ab58dc04ef88cab0326250e933c : SUCCESS in 8m 18s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/ee21c62f9f884964b5a82f70d7de2c71 : SUCCESS in 1h 03m 34s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/6bcc7680dd874557a329e7f0bce9d040 : SUCCESS in 1h 00m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/87ecbed0cb16451dbd5cde3cd1598f10 : SUCCESS in 57m 48s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/3264b95f895b434b9851cbd00a111da3 : SUCCESS in 1h 08m 01s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/0efeb60236bb40e6b7ce87dd3d0a5c79 : SUCCESS in 29m 41s","accounts_in_message":[],"_revision_number":8},{"id":"865224010444173c96b3b5aa1bfe693e3586e258","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-29 08:42:18.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":8},{"id":"d8644a56b17671307a1664e354964dc98ff53156","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-29 08:44:19.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9830ff5d44d84e0cb383456e5de78c80 : SUCCESS in 1m 40s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"429f16124b7cd69c85e6d764bcf1e47687780286","revisions":{"e7243e9f7bf127c4736060a9e8631e2abf998089":{"kind":"REWORK","_number":1,"created":"2021-07-06 18:00:34.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/99/799699/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/99/799699/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/99/799699/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/99/799699/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/99/799699/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/99/799699/1"}}},"commit":{"parents":[{"commit":"cc1bcbf8f8024701e87ccc39556d61ab36922b0a","subject":"Move lazy store update to locations layer","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/cc1bcbf8f8024701e87ccc39556d61ab36922b0a"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-07-06 18:21:15.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-07-06 18:57:36.000000000","tz":-420},"subject":"Remove dead 403-\u003e404 code","message":"Remove dead 403-\u003e404 code\n\nThe db.ImageRepo.save() operation covers up a Forbidden error with\nNotFound in the case of trying to update an image that the user does\nnot own. This is actually never hit in reality as the authorization\nlayer would have caught it before, and returned Forbidden. The API is\nthe layer that should be deciding to hide images for which the user\ncannot see, to avoid things like being able to show an image, but get\nback a 404 on update. In order to do that, we need the lower layers to\nreport the situation faithfully and let the upper layers decide how\nto expose that to the user.\n\nSpecifically, for the policy refactor, we have tests that assert the\nForbidden behavior, and after disabling the authorization layer, will\nbreak (and rightfully so) because they get NotFound when they hit the\nlower-layer check. Because it is hidden so deep, even the API can\u0027t\ndistinguish between the two.\n\nI imagine this check was added long ago to provide the \"if you can\u0027t\nsee it, return NotFound instead of Forbidden, which would expose the\nfact that it exists\" behavior which is desired. However, the\nauthorization layer means we never get here anymore.\n\nThis patch is provided without any test changes to prove that it does\nnot actually alter the real behavior. The test_permissions functional\ntest asserts the proper behavior before and after this change, as well\nas after subsequent refactor patches.\n\nChange-Id: I0084350ebb09cc1cb3752b45165e49f166bfdf91\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e7243e9f7bf127c4736060a9e8631e2abf998089"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e7243e9f7bf127c4736060a9e8631e2abf998089"}]},"branch":"refs/heads/master"},"8f82b666bc0ebdcc0361ed5a0b93f6cd80398426":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-07-14 14:03:38.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/99/799699/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/99/799699/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/99/799699/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/99/799699/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/99/799699/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/99/799699/2"}}},"commit":{"parents":[{"commit":"3197a49cd2495ca834b6967b7e5b70acf510c78d","subject":"Move lazy store update to locations layer","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3197a49cd2495ca834b6967b7e5b70acf510c78d"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-07-06 18:21:15.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-07-14 15:05:43.000000000","tz":-420},"subject":"Remove dead 403-\u003e404 code","message":"Remove dead 403-\u003e404 code\n\nThe db.ImageRepo.save() operation covers up a Forbidden error with\nNotFound in the case of trying to update an image that the user does\nnot own. This is actually never hit in reality as the authorization\nlayer would have caught it before, and returned Forbidden. The API is\nthe layer that should be deciding to hide images for which the user\ncannot see, to avoid things like being able to show an image, but get\nback a 404 on update. In order to do that, we need the lower layers to\nreport the situation faithfully and let the upper layers decide how\nto expose that to the user.\n\nSpecifically, for the policy refactor, we have tests that assert the\nForbidden behavior, and after disabling the authorization layer, will\nbreak (and rightfully so) because they get NotFound when they hit the\nlower-layer check. Because it is hidden so deep, even the API can\u0027t\ndistinguish between the two.\n\nI imagine this check was added long ago to provide the \"if you can\u0027t\nsee it, return NotFound instead of Forbidden, which would expose the\nfact that it exists\" behavior which is desired. However, the\nauthorization layer means we never get here anymore.\n\nThis patch is provided without any test changes to prove that it does\nnot actually alter the real behavior. The test_permissions functional\ntest asserts the proper behavior before and after this change, as well\nas after subsequent refactor patches.\n\nChange-Id: I0084350ebb09cc1cb3752b45165e49f166bfdf91\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8f82b666bc0ebdcc0361ed5a0b93f6cd80398426"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8f82b666bc0ebdcc0361ed5a0b93f6cd80398426"}]},"branch":"refs/heads/master"},"efc9e7cd6adc7c39cec691af85a5d2a781b7d6c0":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-07-14 19:32:52.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/99/799699/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/99/799699/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/99/799699/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/99/799699/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/99/799699/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/99/799699/3"}}},"commit":{"parents":[{"commit":"dc34b57f52a8efd56da5eca25b4ce903f897f569","subject":"Move lazy store update to locations layer","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/dc34b57f52a8efd56da5eca25b4ce903f897f569"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-07-06 18:21:15.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-07-14 20:28:33.000000000","tz":-420},"subject":"Remove dead 403-\u003e404 code","message":"Remove dead 403-\u003e404 code\n\nThe db.ImageRepo.save() operation covers up a Forbidden error with\nNotFound in the case of trying to update an image that the user does\nnot own. This is actually never hit in reality as the authorization\nlayer would have caught it before, and returned Forbidden. The API is\nthe layer that should be deciding to hide images for which the user\ncannot see, to avoid things like being able to show an image, but get\nback a 404 on update. In order to do that, we need the lower layers to\nreport the situation faithfully and let the upper layers decide how\nto expose that to the user.\n\nSpecifically, for the policy refactor, we have tests that assert the\nForbidden behavior, and after disabling the authorization layer, will\nbreak (and rightfully so) because they get NotFound when they hit the\nlower-layer check. Because it is hidden so deep, even the API can\u0027t\ndistinguish between the two.\n\nI imagine this check was added long ago to provide the \"if you can\u0027t\nsee it, return NotFound instead of Forbidden, which would expose the\nfact that it exists\" behavior which is desired. However, the\nauthorization layer means we never get here anymore.\n\nThis patch is provided without any test changes to prove that it does\nnot actually alter the real behavior. The test_permissions functional\ntest asserts the proper behavior before and after this change, as well\nas after subsequent refactor patches.\n\nChange-Id: I0084350ebb09cc1cb3752b45165e49f166bfdf91\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/efc9e7cd6adc7c39cec691af85a5d2a781b7d6c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/efc9e7cd6adc7c39cec691af85a5d2a781b7d6c0"}]},"branch":"refs/heads/master"},"c8e01e8275d575d68eff88b2169a72973bfc9b6f":{"kind":"REWORK","_number":4,"created":"2021-07-16 13:56:40.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/99/799699/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/99/799699/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/99/799699/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/99/799699/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/99/799699/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/99/799699/4"}}},"commit":{"parents":[{"commit":"2cf5df86adffc367240775280f3dc38f87aa568f","subject":"Move lazy store update to locations layer","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2cf5df86adffc367240775280f3dc38f87aa568f"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-07-06 18:21:15.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-07-16 14:58:46.000000000","tz":-420},"subject":"Remove dead 403-\u003e404 code","message":"Remove dead 403-\u003e404 code\n\nThe db.ImageRepo.save() operation covers up a Forbidden error with\nNotFound in the case of trying to update an image that the user does\nnot own. This is actually never hit in reality as the authorization\nlayer would have caught it before, and returned Forbidden. The API is\nthe layer that should be deciding to hide images for which the user\ncannot see, to avoid things like being able to show an image, but get\nback a 404 on update. In order to do that, we need the lower layers to\nreport the situation faithfully and let the upper layers decide how\nto expose that to the user.\n\nSpecifically, for the policy refactor, we have tests that assert the\nForbidden behavior, and after disabling the authorization layer, will\nbreak (and rightfully so) because they get NotFound when they hit the\nlower-layer check. Because it is hidden so deep, even the API can\u0027t\ndistinguish between the two.\n\nI imagine this check was added long ago to provide the \"if you can\u0027t\nsee it, return NotFound instead of Forbidden, which would expose the\nfact that it exists\" behavior which is desired. However, the\nauthorization layer means we never get here anymore.\n\nThis patch is provided without any test changes to prove that it does\nnot actually alter the real behavior. The test_permissions functional\ntest asserts the proper behavior before and after this change, as well\nas after subsequent refactor patches.\n\nChange-Id: I0084350ebb09cc1cb3752b45165e49f166bfdf91\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c8e01e8275d575d68eff88b2169a72973bfc9b6f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c8e01e8275d575d68eff88b2169a72973bfc9b6f"}]},"branch":"refs/heads/master"},"7175bb7daa070ac86bab232e8433f4c9100a0ccd":{"kind":"REWORK","_number":5,"created":"2021-07-16 15:28:57.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/99/799699/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/99/799699/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/99/799699/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/99/799699/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/99/799699/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/99/799699/5"}}},"commit":{"parents":[{"commit":"2cf5df86adffc367240775280f3dc38f87aa568f","subject":"Move lazy store update to locations layer","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2cf5df86adffc367240775280f3dc38f87aa568f"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-07-06 18:21:15.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-07-16 16:31:04.000000000","tz":-420},"subject":"Remove dead 403-\u003e404 code","message":"Remove dead 403-\u003e404 code\n\nThe db.ImageRepo.save() operation covers up a Forbidden error with\nNotFound in the case of trying to update an image that the user does\nnot own. This is actually never hit in reality as the authorization\nlayer would have caught it before, and returned Forbidden. The API is\nthe layer that should be deciding to hide images for which the user\ncannot see, to avoid things like being able to show an image, but get\nback a 404 on update. In order to do that, we need the lower layers to\nreport the situation faithfully and let the upper layers decide how\nto expose that to the user.\n\nSpecifically, for the policy refactor, we have tests that assert the\nForbidden behavior, and after disabling the authorization layer, will\nbreak (and rightfully so) because they get NotFound when they hit the\nlower-layer check. Because it is hidden so deep, even the API can\u0027t\ndistinguish between the two.\n\nI imagine this check was added long ago to provide the \"if you can\u0027t\nsee it, return NotFound instead of Forbidden, which would expose the\nfact that it exists\" behavior which is desired. However, the\nauthorization layer means we never get here anymore.\n\nThis patch is provided without any test changes to prove that it does\nnot actually alter the real behavior. The test_permissions functional\ntest asserts the proper behavior before and after this change, as well\nas after subsequent refactor patches.\n\nChange-Id: I0084350ebb09cc1cb3752b45165e49f166bfdf91\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7175bb7daa070ac86bab232e8433f4c9100a0ccd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7175bb7daa070ac86bab232e8433f4c9100a0ccd"}]},"branch":"refs/heads/master"},"88587a8620d6eb1660d529a24cacf46a9436310a":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2021-07-20 16:30:14.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/99/799699/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/99/799699/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/99/799699/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/99/799699/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/99/799699/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/99/799699/6"}}},"commit":{"parents":[{"commit":"62334aa4645546011b97664c5329e5539a444676","subject":"Merge \"Move lazy store update to locations layer\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/62334aa4645546011b97664c5329e5539a444676"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-07-06 18:21:15.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-07-20 16:37:40.000000000","tz":-420},"subject":"Remove dead 403-\u003e404 code","message":"Remove dead 403-\u003e404 code\n\nThe db.ImageRepo.save() operation covers up a Forbidden error with\nNotFound in the case of trying to update an image that the user does\nnot own. This is actually never hit in reality as the authorization\nlayer would have caught it before, and returned Forbidden. The API is\nthe layer that should be deciding to hide images for which the user\ncannot see, to avoid things like being able to show an image, but get\nback a 404 on update. In order to do that, we need the lower layers to\nreport the situation faithfully and let the upper layers decide how\nto expose that to the user.\n\nSpecifically, for the policy refactor, we have tests that assert the\nForbidden behavior, and after disabling the authorization layer, will\nbreak (and rightfully so) because they get NotFound when they hit the\nlower-layer check. Because it is hidden so deep, even the API can\u0027t\ndistinguish between the two.\n\nI imagine this check was added long ago to provide the \"if you can\u0027t\nsee it, return NotFound instead of Forbidden, which would expose the\nfact that it exists\" behavior which is desired. However, the\nauthorization layer means we never get here anymore.\n\nThis patch is provided without any test changes to prove that it does\nnot actually alter the real behavior. The test_permissions functional\ntest asserts the proper behavior before and after this change, as well\nas after subsequent refactor patches.\n\nChange-Id: I0084350ebb09cc1cb3752b45165e49f166bfdf91\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/88587a8620d6eb1660d529a24cacf46a9436310a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/88587a8620d6eb1660d529a24cacf46a9436310a"}]},"branch":"refs/heads/master"},"28f1aeb5c692bbb80981ed46378207645e5ac2b8":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2021-07-26 19:49:00.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/99/799699/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/99/799699/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/99/799699/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/99/799699/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/99/799699/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/99/799699/7"}}},"commit":{"parents":[{"commit":"485677a3f383489bbedf7c184ed992f1bcc53b24","subject":"Add base policy check module","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/485677a3f383489bbedf7c184ed992f1bcc53b24"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-07-06 18:21:15.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-07-26 20:13:22.000000000","tz":-420},"subject":"Remove dead 403-\u003e404 code","message":"Remove dead 403-\u003e404 code\n\nThe db.ImageRepo.save() operation covers up a Forbidden error with\nNotFound in the case of trying to update an image that the user does\nnot own. This is actually never hit in reality as the authorization\nlayer would have caught it before, and returned Forbidden. The API is\nthe layer that should be deciding to hide images for which the user\ncannot see, to avoid things like being able to show an image, but get\nback a 404 on update. In order to do that, we need the lower layers to\nreport the situation faithfully and let the upper layers decide how\nto expose that to the user.\n\nSpecifically, for the policy refactor, we have tests that assert the\nForbidden behavior, and after disabling the authorization layer, will\nbreak (and rightfully so) because they get NotFound when they hit the\nlower-layer check. Because it is hidden so deep, even the API can\u0027t\ndistinguish between the two.\n\nI imagine this check was added long ago to provide the \"if you can\u0027t\nsee it, return NotFound instead of Forbidden, which would expose the\nfact that it exists\" behavior which is desired. However, the\nauthorization layer means we never get here anymore.\n\nThis patch is provided without any test changes to prove that it does\nnot actually alter the real behavior. The test_permissions functional\ntest asserts the proper behavior before and after this change, as well\nas after subsequent refactor patches.\n\nChange-Id: I0084350ebb09cc1cb3752b45165e49f166bfdf91\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/28f1aeb5c692bbb80981ed46378207645e5ac2b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/28f1aeb5c692bbb80981ed46378207645e5ac2b8"}]},"branch":"refs/heads/master"},"429f16124b7cd69c85e6d764bcf1e47687780286":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2021-07-27 18:34:02.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/99/799699/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/99/799699/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/99/799699/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/99/799699/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/99/799699/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/99/799699/8"}}},"commit":{"parents":[{"commit":"2fef2e6c4f9fbbac795740ddfec57ba193029d03","subject":"Add api_patch() to SynchronousAPIBase","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2fef2e6c4f9fbbac795740ddfec57ba193029d03"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-07-06 18:21:15.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-07-27 14:59:17.000000000","tz":-420},"subject":"Remove dead 403-\u003e404 code","message":"Remove dead 403-\u003e404 code\n\nThe db.ImageRepo.save() operation covers up a Forbidden error with\nNotFound in the case of trying to update an image that the user does\nnot own. This is actually never hit in reality as the authorization\nlayer would have caught it before, and returned Forbidden. The API is\nthe layer that should be deciding to hide images for which the user\ncannot see, to avoid things like being able to show an image, but get\nback a 404 on update. In order to do that, we need the lower layers to\nreport the situation faithfully and let the upper layers decide how\nto expose that to the user.\n\nSpecifically, for the policy refactor, we have tests that assert the\nForbidden behavior, and after disabling the authorization layer, will\nbreak (and rightfully so) because they get NotFound when they hit the\nlower-layer check. Because it is hidden so deep, even the API can\u0027t\ndistinguish between the two.\n\nI imagine this check was added long ago to provide the \"if you can\u0027t\nsee it, return NotFound instead of Forbidden, which would expose the\nfact that it exists\" behavior which is desired. However, the\nauthorization layer means we never get here anymore.\n\nThis patch is provided without any test changes to prove that it does\nnot actually alter the real behavior. The test_permissions functional\ntest asserts the proper behavior before and after this change, as well\nas after subsequent refactor patches.\n\nChange-Id: I0084350ebb09cc1cb3752b45165e49f166bfdf91\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/429f16124b7cd69c85e6d764bcf1e47687780286"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/429f16124b7cd69c85e6d764bcf1e47687780286"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
