)]}'
{"id":"openstack%2Fglance~800100","triplet_id":"openstack%2Fglance~stable%2Fwallaby~I37d67457efdd8572d41cd839fb00c84f1f197892","project":"openstack/glance","branch":"stable/wallaby","hashtags":[],"change_id":"I37d67457efdd8572d41cd839fb00c84f1f197892","subject":"Fix the policy deprecation message","status":"MERGED","created":"2021-07-08 19:17:19.000000000","updated":"2021-08-06 14:32:16.000000000","submitted":"2021-08-06 14:30:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"800100","meta_rev_id":"3302631d686a1bac821d2b355520c4452cf42b64","_number":800100,"virtual_id_number":800100,"owner":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-06 14:30:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2021-08-06 11:45:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2021-08-05 14:41:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2021-08-06 11:45:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2021-08-05 14:41:59.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-08 21:00:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-05 14:41:59.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2021-08-06 11:45:39.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"}],"messages":[{"id":"9c2898df5d0cb0047e2b21cd0b760b987c1b37eb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2021-07-08 19:17:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"59fe2b8311236f3bf1460ad6b356a944ac9f8b4a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-08 21:00:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/b40ec44001ac4acb8ae0cac865035114 : SUCCESS in 1h 14m 02s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8a26ba3e7c054943ae9b893125653935 : SUCCESS in 1h 23m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4c516a0db4ef473eb68f4b263dc9b348 : SUCCESS in 7m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1bf998682bd6492683fcee969fa11902 : SUCCESS in 5m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ef75707da9c048368c68c9da255cc71d : SUCCESS in 6m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3527b4c487f54c68bb9df6d7f524fd26 : SUCCESS in 5m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f4777f981d444befa36cb6a20506ebcb : SUCCESS in 10m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b77986faa0ea49a6996810dfda75f5b7 : SUCCESS in 6m 44s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c59de3f4b5294bb5b32552f21e472eee : SUCCESS in 13m 51s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/ae42078745014446879af8a8e7e2576c : SUCCESS in 8m 09s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/74bfd33957e84da7bd1dd81cb6dfb8cc : SUCCESS in 5m 04s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/bccd3df0e4324c2c8e6d63d910c04ce3 : SUCCESS in 1h 15m 53s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/28261400bd1f4b1f8cc8afb6797f8bcf : SUCCESS in 1h 40m 42s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/494d2a2cc486453fa543b0b45b6dfabb : SUCCESS in 1h 04m 52s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/fada1eb1aee54105a10545f371687dd3 : SUCCESS in 1h 35m 23s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/702d10d26def452e84638adeb73ffe7c : SUCCESS in 53m 24s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/173c7375ddd6481e858f768f32e9b0d5 : SUCCESS in 1h 16m 04s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/3b9276b00f834ebf815ac51c2f1d7525 : SUCCESS in 45m 22s","accounts_in_message":[],"_revision_number":1},{"id":"cd283dca1dfdcd98088ca4a0ebc57762b239c1c8","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-05 14:41:59.000000000","message":"Patch Set 1: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"efb2b2c0cbae0cfb92cf64a7e290e2425b93c9b2","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2021-08-06 11:45:39.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"62b899a16563f126575773845eedc7ef5edfa008","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-06 11:46:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a09b8d6f428c493b9d3af9c4b2371d7db5f0b1a8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-06 14:30:15.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/4212879017444556a6b36151c4141629 : SUCCESS in 1h 22m 11s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/cf184947d11e4333bb784ea97fa7c5d2 : SUCCESS in 1h 38m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/11d8eb98fd8d4f88b53e503f3c7fd9e2 : SUCCESS in 7m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b97c824f5ee4ab79b179877a3529c24 : SUCCESS in 6m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/802506fe5d8443dcaeca1875b3021102 : SUCCESS in 5m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a67a5214a79748bdadd04c34561e7011 : SUCCESS in 6m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac89a2c8abf744cf9b44a842ccbc65aa : SUCCESS in 6m 45s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/91e62758196344668cecc8cd08e846bb : SUCCESS in 9m 11s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/31531b68e3524bf99964483ba0cd49d5 : SUCCESS in 9m 24s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/b6aef0aaee1d439a99c7ee8f39ebabf6 : SUCCESS in 1h 14m 26s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/82b69fdb719d47a4816208d475687c84 : SUCCESS in 1h 22m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/741d221506864fce9dcc4205b828e145 : SUCCESS in 53m 49s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/76ccefc36d654b4aab54351235c40a16 : SUCCESS in 1h 15m 45s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/a6869d347d4d4191a98c9e35522476cf : SUCCESS in 34m 59s","accounts_in_message":[],"_revision_number":1},{"id":"668b27e961866bd55a301eab3dcfbf8f8e8e1d58","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-06 14:30:15.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"3302631d686a1bac821d2b355520c4452cf42b64","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-06 14:32:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ab3500e8d5c41668b59eeadc27b7051 : SUCCESS in 1m 32s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"297258379d84f6b2b25f2c51f9ef71d0c92f8894","revisions":{"297258379d84f6b2b25f2c51f9ef71d0c92f8894":{"kind":"REWORK","_number":1,"created":"2021-07-08 19:17:19.000000000","uploader":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"ref":"refs/changes/00/800100/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/00/800100/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/00/800100/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/00/800100/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/00/800100/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/00/800100/1"}}},"commit":{"parents":[{"commit":"b5437773b20db3d6ef20d449a8a43171c8fc7f69","subject":"Update TOX_CONSTRAINTS_FILE for stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b5437773b20db3d6ef20d449a8a43171c8fc7f69"}]}],"author":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2021-07-05 13:39:48.000000000","tz":60},"committer":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2021-07-08 19:16:30.000000000","tz":60},"subject":"Fix the policy deprecation message","message":"Fix the policy deprecation message\n\nRemoved the defaulting to roles part of the deprecation message.\n\nChange-Id: I37d67457efdd8572d41cd839fb00c84f1f197892\nCloses-Bug: #1934673\n(cherry picked from commit 2d2817375296c7eff5594a092a8463dd31dbf4e4)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/297258379d84f6b2b25f2c51f9ef71d0c92f8894"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/297258379d84f6b2b25f2c51f9ef71d0c92f8894"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
