)]}'
{"id":"openstack%2Fglance~800102","triplet_id":"openstack%2Fglance~stable%2Fvictoria~If12a013404296486dc387b099477d1608b24ba63","project":"openstack/glance","branch":"stable/victoria","topic":"victoria-trusts","hashtags":[],"change_id":"If12a013404296486dc387b099477d1608b24ba63","subject":"Revert \"Remove all usage of keystoneclient\"","status":"MERGED","created":"2021-07-08 19:21:22.000000000","updated":"2021-09-09 23:05:18.000000000","submitted":"2021-09-09 23:03:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"800102-victoria-trusts","meta_rev_id":"437cf8ff3145295a3bd4facd7faa60d0ec7ddbea","_number":800102,"virtual_id_number":800102,"owner":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-09 23:03:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2021-09-09 19:58:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":2,"date":"2021-09-09 19:02:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2021-09-09 19:58:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"date":"2021-09-09 14:30:34.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-08 21:05:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-09 14:20:21.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"CC"},{"updated":"2021-09-09 19:02:31.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2021-09-09 19:02:31.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"}],"messages":[{"id":"287c1800afff0251b9a6232fd4e82639e9e07d24","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2021-07-08 19:21:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0bae1da181fd2896dc09045cd31b51edd071e485","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2021-07-08 19:21:59.000000000","message":"Patch Set 1: Workflow-1\n\nWallaby not merged yet","accounts_in_message":[],"_revision_number":1},{"id":"b04e4af0473c5904d2ed12564f1c357997a6d355","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-08 21:05:35.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/9495b14480e1412b814d128f738cb38c : SUCCESS in 6m 02s\n- grenade https://zuul.opendev.org/t/openstack/build/dfcf5677ada1417c92d64b509a8412c5 : SUCCESS in 52m 28s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ef5329417caa41ffa124e6d06c908c5b : FAILURE in 59m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bddeba3ad77a440bbeb8dc329e579a2f : SUCCESS in 7m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d8b281004b8c44b7bc0cdfd01c774738 : SUCCESS in 8m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e32ccd694dd544959e58a4fd5e2cfe17 : SUCCESS in 9m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4c41c8d60254d56b6615e0213392095 : SUCCESS in 6m 39s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/dadd5d4124e142c9a6910b850cdf5f40 : SUCCESS in 7m 44s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/e733c4cb35f44671b1e84b0ad8d3d370 : SUCCESS in 7m 53s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/840302447020445ea64539c59e2b9547 : SUCCESS in 5m 05s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/54965263389a492aba60b5f8ae9c65b1 : SUCCESS in 1h 17m 05s (non-voting)\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/42dfddddc9a749e5952f0725c328ef95 : SUCCESS in 1h 17m 42s\n- tempest-integrated-storage-wsgi-import https://zuul.opendev.org/t/openstack/build/190d007d655e4a988b5b8dfb2d8b0305 : SUCCESS in 1h 12m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2511cf8071f947c7be1950dafb9047ae : SUCCESS in 1h 04m 30s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/1c9a0e3bfb544869a12fce3951e8d2ee : SUCCESS in 1h 37m 22s","accounts_in_message":[],"_revision_number":1},{"id":"849eb7a478fb79ea5c7bdc4de00453b5caa12360","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2021-08-11 11:12:59.000000000","message":"Patch Set 1: -Workflow\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"64dd8a801696102cc3f2451d2a54806135457b47","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-11 12:29:48.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/c1f21758dcf14fb088ef7ebd44b2bda0 : SUCCESS in 4m 31s\n- grenade https://zuul.opendev.org/t/openstack/build/030ae777d6954c60bcd45b8a2e7a9619 : SUCCESS in 58m 37s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4d83936e304b48128f2db16c2e43aaee : SUCCESS in 1h 02m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87df522f8b374fd5b24abc3c9e5d9b86 : SUCCESS in 7m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7861f43496234b80965ea6ab56e5861e : SUCCESS in 5m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d73f809002464fdf93d95cdbe9e3846f : SUCCESS in 7m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00363b79f04541e7835c13f6bbbefe08 : SUCCESS in 6m 39s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/107f7628f8e74910bbd0445822bdcc7b : SUCCESS in 10m 50s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/693d1d823f554247a1f1e244821f931c : SUCCESS in 9m 35s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/7509f4470c7a4f829dcdfbad856e88c3 : SUCCESS in 5m 34s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/5585d65294b746e4bc617ae7b787fd51 : SUCCESS in 1h 14m 04s (non-voting)\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/ddcdf60704e14cfc8575fa08e9e062be : SUCCESS in 1h 05m 44s\n- tempest-integrated-storage-wsgi-import https://zuul.opendev.org/t/openstack/build/3ebb6677c735417791e03bce2a8e450c : SUCCESS in 1h 08m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/425952ac692344a99ce1a0c6c40d2875 : SUCCESS in 48m 11s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/99905922124b486e9b5478aebfbea1f5 : FAILURE in 1h 11m 51s","accounts_in_message":[],"_revision_number":1},{"id":"b35cb03fc131594e72a549d78d8404dd860a290e","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2021-08-12 11:56:33.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ce7dfd2d0f26b1ef1c0320474f800894bf8f47df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-12 13:19:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/2c98d07a706a4b68b435f21f46bf1308 : SUCCESS in 3m 45s\n- grenade https://zuul.opendev.org/t/openstack/build/107a51b835184bb286d6ed025a0adccf : SUCCESS in 1h 03m 56s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b3ed38b6171e496da569aba7bf4a0f78 : SUCCESS in 1h 14m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/adcb8ae8202a4906a01ee493fe406bfb : SUCCESS in 5m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/38f32c2f26544e70a2371e020bebd3e5 : SUCCESS in 5m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f3605b86e6334c38872542e1ff953a5d : SUCCESS in 5m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/951de91c503247f49fa7b7d25c4c65f8 : SUCCESS in 6m 28s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/74ff41d2c8fa401a947acccd3a4ea1fb : SUCCESS in 8m 08s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/64d294a99fb342ed975e906098e181de : SUCCESS in 8m 01s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/8a0373f0159240e1807d02126b2b34e0 : SUCCESS in 5m 07s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/10560a4449494d70b21f4b6b1764d2af : SUCCESS in 1h 19m 03s (non-voting)\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/472a6a22c57946dca85742a4a817c05c : SUCCESS in 1h 16m 04s\n- tempest-integrated-storage-wsgi-import https://zuul.opendev.org/t/openstack/build/056c486f91ad444db8e65fc1fbc12dc6 : SUCCESS in 1h 15m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7a3f56fb95d24829ae6759f210da25f2 : SUCCESS in 1h 02m 39s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/42002a5d50124b19b652390d6a796a23 : SUCCESS in 1h 12m 25s","accounts_in_message":[],"_revision_number":1},{"id":"0b6f52389f8691d61d172b7fe2fbc0a779e03150","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-09-09 14:20:21.000000000","message":"Patch Set 1:\n\nI think this should be cherry-picked from stable/wallaby and not master branch, ideally this should have two cherry-pick id\u0027s in commit message.","accounts_in_message":[],"_revision_number":1},{"id":"a593e6ca99fae3017c27fc39e5c19eb3fe7ba281","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-09-09 14:21:16.000000000","message":"Patch Set 1:\n\nsimilar to this;\nhttps://review.opendev.org/c/openstack/glance/+/803619","accounts_in_message":[],"_revision_number":1},{"id":"7c6c5a208946f2b95154701eda60b6b41762e404","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2021-09-09 14:30:34.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1:\n\u003e \n\u003e I think this should be cherry-picked from stable/wallaby and not master branch, ideally this should have two cherry-pick id\u0027s in commit message.\n\nhttps://docs.openstack.org/project-team-guide/stable-branches.html#proposing-fixes\n\n\"\"\"git cherry-pick -x $master_commit_id\"\"\"","accounts_in_message":[],"_revision_number":1},{"id":"3f159af5c3be20b9531d9f77c5ffe30b3b8beb51","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-09-09 19:02:31.000000000","message":"Patch Set 1: Code-Review+2\n\nPatch merged in stable/wallaby so I guess this is good to go.","accounts_in_message":[],"_revision_number":1},{"id":"6d78bb9358b6093801794ddf726a3822fd1fa6cc","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2021-09-09 19:58:39.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nIn Wallaby and master, make sense to backport.","accounts_in_message":[],"_revision_number":1},{"id":"2fcad703f1299f598b72228de911fe7054ea9264","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-09 19:59:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7afb7876526ce822939e250245c02b6bc3860588","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-09 23:03:23.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/c42d1e0287fd486ab4990335fa84e987 : SUCCESS in 4m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/46fa9c5893784959bed1866b0dc95faf : SUCCESS in 1h 06m 16s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1b6aeebc00fa40f28c9b79590e663e7b : SUCCESS in 57m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a455d626cd2d4a0a91951c1a2dfbd119 : SUCCESS in 5m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/95efe4e00f99477d91f12bc76644de54 : SUCCESS in 6m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8093541c0c9142389a85d06d9384f078 : SUCCESS in 6m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99a861a2e8b848dfb91683536ba477fe : SUCCESS in 7m 22s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f39721fe044340e4b66cb47fbc74fef4 : SUCCESS in 8m 57s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/e60ebf4bc30f4a4186d2e5b8b7bb39d0 : SUCCESS in 10m 49s\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/1c7763cff9934b0d966f857b733dcac2 : SUCCESS in 1h 28m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2b899db87f0d4a48b3c52e184b5cda3f : SUCCESS in 53m 25s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/545b6b57add04c049a2b6942d239cc65 : SUCCESS in 1h 24m 19s","accounts_in_message":[],"_revision_number":1},{"id":"9446987a3e87645c63a8b326de9466b756fbfed1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-09 23:03:24.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"437cf8ff3145295a3bd4facd7faa60d0ec7ddbea","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-09 23:05:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1611bc99e5544f1a4ab5fed827c54fb : SUCCESS in 1m 33s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3139ef39cb34c411d340572b098a790c92cfdf7e","revisions":{"3139ef39cb34c411d340572b098a790c92cfdf7e":{"kind":"REWORK","_number":1,"created":"2021-07-08 19:21:22.000000000","uploader":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"ref":"refs/changes/02/800102/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/02/800102/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/02/800102/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/02/800102/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/02/800102/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/02/800102/1"}}},"commit":{"parents":[{"commit":"958840f29a6294bd0e1932b0a3298282bdc2d8b6","subject":"Merge \"Add housekeeping module and staging cleaner\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/958840f29a6294bd0e1932b0a3298282bdc2d8b6"}]}],"author":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2021-06-10 12:51:49.000000000","tz":60},"committer":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2021-07-08 19:21:05.000000000","tz":60},"subject":"Revert \"Remove all usage of keystoneclient\"","message":"Revert \"Remove all usage of keystoneclient\"\n\nThis reverts commit 810417df865c7b4eb71cee4dfee843d65972852b.\n\nThe Swift driver not being able to use Trusts had nothing to do\nabout the store driver itself nor that keystoneauth1 would have\nbroken the feature, but rather it not having the functionality\nin the first place and us not catching that on reviews.\n\nWe should figure out how to test this before we try to replace\nthis code again.\n\nChange-Id: If12a013404296486dc387b099477d1608b24ba63\nCloses-Bug: #1916052\n(cherry picked from commit 9b683678b2a19bb1069aace58e054150a78b5f6f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3139ef39cb34c411d340572b098a790c92cfdf7e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3139ef39cb34c411d340572b098a790c92cfdf7e"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
