)]}'
{"id":"openstack%2Fglance~800103","triplet_id":"openstack%2Fglance~stable%2Fussuri~If12a013404296486dc387b099477d1608b24ba63","project":"openstack/glance","branch":"stable/ussuri","topic":"ussuri-trusts","hashtags":[],"change_id":"If12a013404296486dc387b099477d1608b24ba63","subject":"Revert \"Remove all usage of keystoneclient\"","status":"MERGED","created":"2021-07-08 19:24:42.000000000","updated":"2021-09-10 18:12:57.000000000","submitted":"2021-09-10 18:11:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"800103-ussuri-trusts","meta_rev_id":"96ffb745379945e87676157b09dd450adff3b99b","_number":800103,"virtual_id_number":800103,"owner":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-10 18:11:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2021-09-09 19:59:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":2,"date":"2021-09-09 19:03:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2021-09-10 14:22:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-08 20:53:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-09 19:03:25.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2021-09-09 19:03:25.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"}],"messages":[{"id":"864618546e504e04fed97ae6ac946e5e5e62272b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2021-07-08 19:24:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5e798120bafbde07ac2181ddc63ff83a03bb336b","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2021-07-08 19:25:23.000000000","message":"Patch Set 1: Workflow-1\n\nVictoria patch not merged yet","accounts_in_message":[],"_revision_number":1},{"id":"ec75f4c154c6e650739a04d3fdf11f71b71dfe90","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-08 20:53:50.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/dc6158241a564d1d9b698b3011eea8bc : RETRY_LIMIT in 3m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/1749e276b3e049dcb1e5070b654de46d : SUCCESS in 1h 22m 23s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c5056dbf7bcc47249501df38d82a5d26 : SUCCESS in 1h 24m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/745f8f1528be451ba868e861079c56a2 : SUCCESS in 5m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bfc17aa4f8e04f5bb673fec64ca55f95 : SUCCESS in 5m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/59796354749444a7a14268b36de02090 : SUCCESS in 5m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80e5ab34fe0242a9a749ff7e06e9f8dc : SUCCESS in 7m 01s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/cde31b7fdbca4a84be19ea56f43aaa37 : SUCCESS in 7m 48s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/668808af2611473889981dd258e06411 : SUCCESS in 7m 48s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/730d4ef54ee148c3931671ae09886f63 : RETRY_LIMIT in 3m 41s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/86ae4c4a8464451594398ecc028b3864 : SUCCESS in 1h 19m 20s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b8948845b1c6412db03631a73a5b53a5 : SUCCESS in 1h 05m 50s","accounts_in_message":[],"_revision_number":1},{"id":"56f8b85ee6342fcb1fdfba5d7004e38c14c9de7d","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-09-09 19:03:25.000000000","message":"Patch Set 1: Code-Review+2\n\nputting my +2 but this needs to approve only after workflow set to 0 by Erno!!","accounts_in_message":[],"_revision_number":1},{"id":"477e1f8b51dc0c035fc457e579ed903b50d7cc48","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2021-09-09 19:59:11.000000000","message":"Patch Set 1: Code-Review+2\n\nMerging to victoria. Good to go for ussuri once that happens.","accounts_in_message":[],"_revision_number":1},{"id":"a39df252fc87473439939b80ea2193ffc9e1a925","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2021-09-10 14:22:19.000000000","message":"Patch Set 1: Workflow+1\n\nFlipping the WF, 2x +2s just waited the previous branches to merge.","accounts_in_message":[],"_revision_number":1},{"id":"f058a8c40171afdba87d70e552d180f07a4e3f84","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-10 16:08:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/461511d375284be18a9bafc150146b2f : SUCCESS in 4m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/aee194a74b1d48c488e7d480b3c0618b : SUCCESS in 1h 36m 42s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7c49a5735fdd43e5848e8cb1e1b1bdff : SUCCESS in 1h 12m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/52408e1282894123aad9039c72a6f6dd : SUCCESS in 6m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c28ff7ab58934983b0bf18821ce41a70 : SUCCESS in 6m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/42c83684b22442e2a9ae1f5bc476bce5 : SUCCESS in 6m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e63c764b38445bba502f036188ff1be : SUCCESS in 7m 10s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b4c150ea8cc34deea227732a8a715c60 : SUCCESS in 8m 41s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/0c718d8768ed46fda2c9e423ad23d29c : SUCCESS in 8m 56s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/2398c1449d4141deb5b648c85f403f22 : SUCCESS in 5m 49s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/cf8c85ebdb4a473c91844653bbe6f842 : SUCCESS in 1h 17m 41s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/994c639f21f0437e8b2e4bebb71856fb : SUCCESS in 1h 01m 08s","accounts_in_message":[],"_revision_number":1},{"id":"e7ba97747eb0e3d561ac0712aedf6c4a59ff30de","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-10 16:14:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b967bd4926905be97901cf1a5663e1ed92abd6f6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-10 18:11:11.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/ed24efb43fa949dc86ccc1e46e419e20 : SUCCESS in 5m 57s\n- grenade https://zuul.opendev.org/t/openstack/build/e52af0eb5a6d48e4a582eea0aaa02769 : SUCCESS in 1h 03m 47s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/709822fc254545859339c9bd82ac0262 : SUCCESS in 1h 25m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc846fd4b20f49aaa0caeba7b9ea4d8f : SUCCESS in 9m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e0f0c870bdfd48e79e0d93f0985e29c6 : SUCCESS in 7m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d16f91e7967b4f308527ad15196ae3e8 : SUCCESS in 8m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a27bbe985673429fad6e79ed4bf5aa25 : SUCCESS in 8m 21s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d3de81dd873e4c0e920ac2bca6de01f4 : SUCCESS in 12m 14s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/d28b4e02968b43eaa90a090589a390f1 : SUCCESS in 11m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b3885c057ab54bfa928e2c37edd8768e : SUCCESS in 1h 03m 12s","accounts_in_message":[],"_revision_number":1},{"id":"0454ea98d9485fe978060639160f93f9cb17abbe","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-10 18:11:12.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"96ffb745379945e87676157b09dd450adff3b99b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-10 18:12:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3490465173274470a3938fd0e56a0e5c : SUCCESS in 1m 19s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f213c1bf44388fde5802bd707cf2850f88a26d5e","revisions":{"f213c1bf44388fde5802bd707cf2850f88a26d5e":{"kind":"REWORK","_number":1,"created":"2021-07-08 19:24:42.000000000","uploader":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"ref":"refs/changes/03/800103/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/03/800103/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/03/800103/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/03/800103/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/03/800103/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/03/800103/1"}}},"commit":{"parents":[{"commit":"cc22468d74cc83a24c198cc45a827f0ed14b3227","subject":"Drop lower-constraints jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/cc22468d74cc83a24c198cc45a827f0ed14b3227"}]}],"author":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2021-06-10 12:51:49.000000000","tz":60},"committer":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2021-07-08 19:23:53.000000000","tz":60},"subject":"Revert \"Remove all usage of keystoneclient\"","message":"Revert \"Remove all usage of keystoneclient\"\n\nThis reverts commit 810417df865c7b4eb71cee4dfee843d65972852b.\n\nThe Swift driver not being able to use Trusts had nothing to do\nabout the store driver itself nor that keystoneauth1 would have\nbroken the feature, but rather it not having the functionality\nin the first place and us not catching that on reviews.\n\nWe should figure out how to test this before we try to replace\nthis code again.\n\nmerge-conflict on requirements jsonschema.\n\nChange-Id: If12a013404296486dc387b099477d1608b24ba63\nCloses-Bug: #1916052\n(cherry picked from commit 9b683678b2a19bb1069aace58e054150a78b5f6f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f213c1bf44388fde5802bd707cf2850f88a26d5e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f213c1bf44388fde5802bd707cf2850f88a26d5e"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
