)]}'
{"id":"openstack%2Fglance~800221","triplet_id":"openstack%2Fglance~master~I79683c81233b35f2399119128a63d33d69c50eeb","project":"openstack/glance","branch":"master","topic":"bug/1885928","hashtags":[],"change_id":"I79683c81233b35f2399119128a63d33d69c50eeb","subject":"\u0027community\u0027 images need to be treated as public","status":"MERGED","created":"2021-07-09 12:50:02.000000000","updated":"2021-09-05 16:57:39.000000000","submitted":"2021-09-03 00:42:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":5,"has_review_started":true,"submission_id":"800221-bug/1885928","meta_rev_id":"cb0cd8a100ff76d1723651dfb826b1f9fcc4071e","_number":800221,"virtual_id_number":800221,"owner":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-03 00:42:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2021-09-02 17:54:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2021-08-31 14:52:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2021-09-02 17:54:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-09 13:24:12.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-07-09 14:16:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-15 17:24:44.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2021-08-31 13:47:25.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"}],"messages":[{"id":"8040e1ad3d7fa6a77596edd2b0e616216dba7d96","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2021-07-09 12:50:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"574d3c661672f81fefd641ed18fdb6d98ab11f09","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-07-09 13:24:12.000000000","message":"Patch Set 1:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/800221/800221/1/check-not-vote/ext-glance-zuul/d877ec8 : SUCCESS in 33m 42s","accounts_in_message":[],"_revision_number":1},{"id":"83d69d0944a8bbf72d300abab65af9402a3a67a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-09 14:16:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/1e0677bcfba34c4c9d3ef8e6836f2310 : SUCCESS in 1h 00m 23s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/17fdb1515b59455d82c8d90d53d8f7ba : SUCCESS in 1h 06m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/12ba2dd756a54b84969b7a3440948018 : SUCCESS in 6m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/13c4282116f041559b9617325316df7d : SUCCESS in 8m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0c64ff34550b432fa9a8b806be396722 : SUCCESS in 6m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c69b79ba67594289870efb44c01bcb41 : SUCCESS in 6m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/adb801cc589c4088a51dff529c009ced : SUCCESS in 7m 19s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/95b061eeac604e34940bfcb52ae1d833 : SUCCESS in 9m 04s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/0bfc2951f96c4673b88263e96ab8da96 : SUCCESS in 10m 22s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/742bc5f0c79b42f2a228306bd0f8c379 : FAILURE in 10m 05s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/b7005f106dd0484c9d22ebfde6c33f8b : SUCCESS in 5m 28s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/3c8176808d7547b7886bb8fc0b32bcf9 : SUCCESS in 1h 17m 35s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/cfb25ed2fe1f44b8957b2a2932cd5ea8 : SUCCESS in 1h 07m 10s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/d60280bd729840fa9fb265b1b41da88e : SUCCESS in 1h 06m 24s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/57f9f42ae2b44ec49726b57a834bfd38 : SUCCESS in 1h 08m 48s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b3bcc21047cd4598aae56b0e2cd23b98 : SUCCESS in 47m 06s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/f41d42bb34a148359efab4fed9a13646 : SUCCESS in 1h 08m 42s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/83d9ed90015745ada3c41cf7fa21c4e9 : SUCCESS in 26m 25s","accounts_in_message":[],"_revision_number":1},{"id":"98b8aec7b32e1f45736ea2696f0437d40402762e","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-07-15 17:24:44.000000000","message":"Patch Set 1: Code-Review+1\n\n(2 comments)\n\nProbably better to have unit test to check we are calling set_acls_for_multi_store for community image.","accounts_in_message":[],"_revision_number":1},{"id":"6e365aa7bf215871dfe9743b0c851915f40af12a","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2021-07-16 10:59:03.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"1c1861eb39a548b5bd14248517acc084872b7cd8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2021-08-16 11:53:35.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a4e95d4580b78ebf1798c6f36d685dd171d76bac","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-08-16 12:53:42.000000000","message":"Patch Set 2:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/800221/800221/2/check-not-vote/ext-glance-zuul/d9a087b : SUCCESS in 30m 28s","accounts_in_message":[],"_revision_number":2},{"id":"0e08dfd22de06a17cc5bc95d13ccceeb5e4c3282","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-16 13:07:35.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/ec1694c4637148aea160b00bb707ec78 : SUCCESS in 56m 18s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c4fea1d6941240cfa6fde651ed7ef4fa : SUCCESS in 59m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3cdd3c8b269045e493ebbe5a3ff39b56 : SUCCESS in 5m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7badaa9719a94a83a6df0e4d76256f36 : SUCCESS in 6m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2f3e2a7deff04b87902d1df06bcee973 : SUCCESS in 5m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e4032cb0b3424ca6845d278c113062d8 : SUCCESS in 6m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e28ee88dc9b54b4da7d0d632433eb6c6 : SUCCESS in 7m 24s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/35ca417cc33c4dfc91e6149694755bc5 : SUCCESS in 8m 53s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/756a6f034da74033b3908832d8a6629e : SUCCESS in 9m 50s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/86ec1fb1c11142e9ac2583b3e8eb2d1e : SUCCESS in 8m 38s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/279fd5b06c0547ca85ed3e34a5e63483 : SUCCESS in 4m 47s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/bd12fb31bdac4044b1bd5187fe4017d1 : SUCCESS in 1h 00m 33s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/3c0a4b0e79b54a31aff1b6198f9333ca : SUCCESS in 1h 06m 37s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/5662e95aaf8449b39c2f45b1c2c72e36 : SUCCESS in 1h 05m 01s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/f0de43af0d284e7d856bedff050dd6f1 : SUCCESS in 1h 11m 55s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d5f7e7671189475f8cbe32a8c1d7f7c5 : SUCCESS in 1h 01m 34s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/5835781635534eb0afa0b31981a672c3 : SUCCESS in 1h 09m 02s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/1b26787d7c5d44bfbf8169c39fd3d048 : SUCCESS in 32m 47s","accounts_in_message":[],"_revision_number":2},{"id":"07da4db5013d71635c044f6b766803b40aee1274","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-16 15:31:17.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"83d0499523daf00234f4888dfe32dfbc6cd107b3","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-08-31 14:46:58.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"aed63fbea813b3baf48183a983ac2a47bd8b8ab7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-08-31 14:47:49.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"8eea9ea2c63d100fef7377321b0d2fd83ec12bd7","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-08-31 14:52:11.000000000","message":"Patch Set 3: Code-Review+2\n\nJust fixed commit message grammar and expanded the tests to cover public as well.","accounts_in_message":[],"_revision_number":3},{"id":"63673beb32770ff0f7cf11f9c5686d0b93b31859","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-31 18:41:24.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/39d665e915c248898d7958567455ccfb : SUCCESS in 1h 12m 08s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d046e56dfd684af79aff46681a31827e : SUCCESS in 1h 12m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8350605615594a658deef9a9e1aed091 : SUCCESS in 7m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b46d7ac9af0b43cd84a567287d5c6107 : SUCCESS in 6m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6922c15c7e8e47c38d3516641566d34a : SUCCESS in 6m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2132d883d5234b4d9ef34fb40a8678cd : SUCCESS in 8m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/94a8ff0526574b69a8206fac5fd317d1 : SUCCESS in 8m 37s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/dc85339fc98b46f29f497e448f9982b5 : TIMED_OUT in 31m 49s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/65133d72d6f342289fe453c36643eaf2 : SUCCESS in 9m 04s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/b6e870377bf24978a54bbdf8a6a5ec07 : SUCCESS in 10m 05s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/1b3a25bd27a2417db057c772947cec51 : SUCCESS in 5m 41s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/2aba8c04aef045998bffa2eb77a153d7 : SUCCESS in 1h 11m 39s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/91a9516f2b314225bde2d2db7bfbe771 : POST_FAILURE in 24m 07s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/8bf9100987024c669cc05201fc98f272 : SUCCESS in 1h 12m 31s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/614d57ee6fe94a85bc9d5cc937906f43 : SUCCESS in 1h 13m 54s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/064600f9c46c45919347fd2855205c18 : SUCCESS in 50m 15s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/b5eaa18a3521485395bc4a054f6502dd : SUCCESS in 1h 25m 32s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/7b92d16b194e45028187c8277ac2fb8e : SUCCESS in 33m 53s","accounts_in_message":[],"_revision_number":3},{"id":"92e5c8ac883b82ecd443cdb230a24c0592d4fb0b","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-09-02 09:40:14.000000000","message":"Patch Set 3:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/800221/800221/3/check-not-vote/ext-glance-zuul/9344411 : SUCCESS in 35m 09s","accounts_in_message":[],"_revision_number":3},{"id":"6d5ae658a6c7d3510594d658412b7dcc60aeb88d","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-09-02 17:06:55.000000000","message":"Patch Set 3:\n\nrecheck tempest post failure","accounts_in_message":[],"_revision_number":3},{"id":"b4c87a31704c771456d56ba8227d2d485060721c","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-09-02 17:54:55.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nLooks good, thank you Dan for additional coverage!","accounts_in_message":[],"_revision_number":3},{"id":"4ca3b2e318cbbabd02cf42f986f35894e3dae4f9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-02 19:10:20.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/f1f9cb3da4a64d7abd356ed8b39d55b8 : SUCCESS in 59m 50s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2debfb70584e4b1c8f9c4c87bca07496 : FAILURE in 57m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff0d15dde2014bb5a8cf9d3f51ab7d1c : SUCCESS in 6m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dea9247c75de4a90b3ad0538824a6c9d : SUCCESS in 8m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f19e9ef1f9cf44eb982fa761468d6220 : SUCCESS in 7m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e001a0b305f240d2ba4a173b98664ab1 : SUCCESS in 7m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1503319f9bf4864a6aec5d7f3d36f83 : SUCCESS in 7m 50s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/bba09c554ac945b080c49d154f5979e0 : SUCCESS in 9m 41s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/8ab8d48f963b4aa2b7a40d4dbf4a2443 : SUCCESS in 10m 48s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/a076470a503b403485f218d56f5ebc52 : SUCCESS in 10m 05s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/7844cac65db147c7b87234cfd83ecaeb : SUCCESS in 6m 30s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/f944dc03f4b74bf4bb101d6b63d05f0f : SUCCESS in 1h 14m 55s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/494bde347e974c5192f542a5c7ea1bf4 : SUCCESS in 1h 07m 10s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/472a9849a9c84b078a8d62e0f84fc383 : SUCCESS in 1h 10m 29s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/5cb141d3af324bca80f957db281eff30 : FAILURE in 1h 45m 56s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/549a7e2a87284918bf6f9093337819c6 : SUCCESS in 49m 51s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/ac66320e8c1046c1ba9874d34db400f4 : SUCCESS in 1h 05m 46s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/fcd584ad498246e3ba30b9d570fb0f38 : SUCCESS in 31m 02s","accounts_in_message":[],"_revision_number":3},{"id":"1266a929b3114e39facb00dcfdee94479fcac51f","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-09-02 19:46:35.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"7e0c2bf955374b617cf4a2cf21f411ca32780d9c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-02 21:13:59.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/b2457cc306864710ae04ef57264cc730 : SUCCESS in 1h 17m 25s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/dd3ffc81f9854bd1bf5a3516c415df76 : SUCCESS in 1h 04m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee8141945a6241349fcbcda32cbc0efa : SUCCESS in 5m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c796feb331aa4a0d81c81c27b06cd7fe : SUCCESS in 7m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/df0d93691cbc41a18a04ac2f0f82336d : SUCCESS in 6m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ff6c19ed18cc4c8d8e2f96756a96d197 : SUCCESS in 6m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e242ac68429744a387f325f9ff07922e : SUCCESS in 8m 02s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/53a1927138d946ee8e2854fc539ba19c : SUCCESS in 10m 57s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/ec80851125cc47c293359b8585606965 : SUCCESS in 9m 06s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/08e35d6086354764b2b09813e6ce379a : SUCCESS in 9m 38s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/26f5e903e35544f1848973496ddc7ae8 : SUCCESS in 5m 34s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/c39632f4ad93471babb6cb889f1a109a : SUCCESS in 1h 17m 25s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/b63ef86346e74e31a3047c7768c9edfe : SUCCESS in 1h 04m 15s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/6d8b0f4fa89c462c85f4bbfcf4947960 : SUCCESS in 1h 03m 34s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/096d69dec27149d69c2fa4bcb557bae2 : SUCCESS in 1h 14m 36s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c4f3c6134d464495af0dbfb3a6c0e28d : SUCCESS in 48m 24s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/412ea46cfcb84f9f9be67c785bed0b7e : SUCCESS in 1h 06m 51s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/25e38418908542009421301151ab3344 : SUCCESS in 31m 01s","accounts_in_message":[],"_revision_number":3},{"id":"88a746c1869a96f3b35da9915761dd111d544aae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-02 21:14:41.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"fec4185a3f1abd52b90a7723fdba32819c7cea08","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-03 00:42:03.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/16b27c72b5584dd3bb38cfa1c70ff9e1 : SUCCESS in 57m 46s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3eb89d5459474b378e1811479c394e79 : SUCCESS in 1h 05m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/621082eb81844c97bfdbe7a4618ccd6a : SUCCESS in 6m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e9a048d04ec442b8ba226696924b78c1 : SUCCESS in 7m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/50e521f21b1f484d9132dfeb3baa8fdc : SUCCESS in 6m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/537359c26c8c4e7a8e0f21f61835d7c6 : SUCCESS in 7m 03s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f2905752ad71443f910dabe3f4f9be45 : SUCCESS in 9m 03s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/a73b8407b44844bcbb1fbdc10d8c7d53 : SUCCESS in 12m 26s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/4993db250b0747c382b669901786e675 : SUCCESS in 1h 04m 31s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/5b499c049e364a129b474bcd5833a077 : SUCCESS in 1h 00m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/423ebc26f22a4924bef6f5dd4a082e45 : SUCCESS in 57m 20s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/fc80310ef0e64e579a0e3eab469fbdc3 : SUCCESS in 1h 11m 07s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/ec39a1076aef4ff8ad35a924e0d929a9 : SUCCESS in 25m 53s","accounts_in_message":[],"_revision_number":3},{"id":"dfb2d40a14ce1c9e46f1fec02b3c985df15c0045","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-03 00:42:03.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"dd83ce6bbe4a713c507d2e3f9656252323d23d76","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-03 00:44:11.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e32be71f6e4142a5b3199ff42d7ec2e8 : SUCCESS in 1m 30s","accounts_in_message":[],"_revision_number":3},{"id":"cb0cd8a100ff76d1723651dfb826b1f9fcc4071e","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-09-05 16:57:39.000000000","message":"Patch Set 3:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/800221/800221/3/check-not-vote/ext-glance-zuul/2ddb694 : SUCCESS in 40m 13s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"f0d891a3edbf9978f8c427df05e8c912fce54cf4","revisions":{"b3fc120881ca0ee6d2e8d950bbdcda34a5a3bedb":{"kind":"REWORK","_number":1,"created":"2021-07-09 12:50:02.000000000","uploader":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"ref":"refs/changes/21/800221/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/21/800221/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/21/800221/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/21/800221/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/21/800221/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/21/800221/1"}}},"commit":{"parents":[{"commit":"77be0e3e230aa0087a4fec64565b71aff16df9bb","subject":"Merge \"Revert \"Remove all usage of keystoneclient\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/77be0e3e230aa0087a4fec64565b71aff16df9bb"}]}],"author":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2021-07-09 12:48:45.000000000","tz":60},"committer":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2021-07-09 12:48:45.000000000","tz":60},"subject":"\u0027community\u0027 images needs to be treated as public","message":"\u0027community\u0027 images needs to be treated as public\n\nThe file acls on community images needs to be treated as if the\nimage was public.\n\nChange-Id: I79683c81233b35f2399119128a63d33d69c50eeb\nCloses-bug: #1885928\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b3fc120881ca0ee6d2e8d950bbdcda34a5a3bedb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b3fc120881ca0ee6d2e8d950bbdcda34a5a3bedb"}]},"branch":"refs/heads/master"},"177d8586b9d2b1b7f7203beb7a4963824cb754ed":{"kind":"REWORK","_number":2,"created":"2021-08-16 11:53:35.000000000","uploader":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"ref":"refs/changes/21/800221/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/21/800221/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/21/800221/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/21/800221/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/21/800221/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/21/800221/2"}}},"commit":{"parents":[{"commit":"77be0e3e230aa0087a4fec64565b71aff16df9bb","subject":"Merge \"Revert \"Remove all usage of keystoneclient\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/77be0e3e230aa0087a4fec64565b71aff16df9bb"}]}],"author":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2021-07-09 12:48:45.000000000","tz":60},"committer":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2021-08-16 11:51:18.000000000","tz":60},"subject":"\u0027community\u0027 images needs to be treated as public","message":"\u0027community\u0027 images needs to be treated as public\n\nEven the \u0027community\u0027 images re not listed by default their\nbehaviour is like public images otherwise. This means that\nthe image data needs to be available for everyone and thus\nthe acls for the file/object should be like public too.\n\nChange-Id: I79683c81233b35f2399119128a63d33d69c50eeb\nCloses-bug: #1885928\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/177d8586b9d2b1b7f7203beb7a4963824cb754ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/177d8586b9d2b1b7f7203beb7a4963824cb754ed"}]},"branch":"refs/heads/master"},"f0d891a3edbf9978f8c427df05e8c912fce54cf4":{"kind":"REWORK","_number":3,"created":"2021-08-31 14:47:49.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/21/800221/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/21/800221/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/21/800221/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/21/800221/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/21/800221/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/21/800221/3"}}},"commit":{"parents":[{"commit":"77be0e3e230aa0087a4fec64565b71aff16df9bb","subject":"Merge \"Revert \"Remove all usage of keystoneclient\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/77be0e3e230aa0087a4fec64565b71aff16df9bb"}]}],"author":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2021-07-09 12:48:45.000000000","tz":60},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-08-31 15:49:38.000000000","tz":-420},"subject":"\u0027community\u0027 images need to be treated as public","message":"\u0027community\u0027 images need to be treated as public\n\nEven though \u0027community\u0027 images are not listed by default their\nbehaviour is like public images otherwise. This means that\nthe image data needs to be available for everyone and thus\nthe acls for the file/object should be like public too.\n\nChange-Id: I79683c81233b35f2399119128a63d33d69c50eeb\nCloses-bug: #1885928\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f0d891a3edbf9978f8c427df05e8c912fce54cf4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f0d891a3edbf9978f8c427df05e8c912fce54cf4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
