)]}'
{"id":"openstack%2Fglance~800957","triplet_id":"openstack%2Fglance~master~I3f42e37e783fdeefe60a25211a97d54600c97021","project":"openstack/glance","branch":"master","topic":"migration-version-bump","hashtags":[],"change_id":"I3f42e37e783fdeefe60a25211a97d54600c97021","subject":"Change database migration version to xena","status":"MERGED","created":"2021-07-15 15:13:35.000000000","updated":"2021-07-15 21:39:12.000000000","submitted":"2021-07-15 20:51:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"800957-migration-version-bump","meta_rev_id":"3e6999ee8b250c74e26506b60f35209d19a36a52","_number":800957,"virtual_id_number":800957,"owner":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-15 20:51:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"all":[{"value":2,"date":"2021-07-15 17:55:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"all":[{"value":1,"date":"2021-07-15 17:55:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]}],"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-15 15:44:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-15 15:46:45.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-07-15 17:55:59.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"}],"messages":[{"id":"b103d050acbe08f8c4e3dd01e00530e40a029ca5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-07-15 15:13:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e6c99bb48667689cd38b9a128959d5537bcae254","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-15 15:44:24.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ace2dd0549749eab373191ee9d0dbf7 : SUCCESS in 5m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bcdf7349a77f42a69f90bbd25e0fce08 : SUCCESS in 6m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2c1761d4345b4c12a2cfcc92aec5ec8b : SUCCESS in 6m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/81a197aa7cae422bb75e4bd6b785cc08 : SUCCESS in 7m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/384855621ae14a13a25fc55982af223b : SUCCESS in 6m 47s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/617327e0939d4f52a709eb09b9199eb0 : SUCCESS in 9m 07s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/f1e51ea8d08e4019be7bb6a8d55a997f : SUCCESS in 9m 43s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/3dad7381dbd14944a82b244ed8c4f8b5 : FAILURE in 10m 10s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/13c6c898e6ed4a27ab6362cbb3df8155 : FAILURE in 5m 22s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/3005e7d3238647bfbbaba97486f6fe8a : SUCCESS in 29m 22s","accounts_in_message":[],"_revision_number":1},{"id":"b011dc67b6c19a6b107af391f5a177c5475fb547","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-07-15 15:46:45.000000000","message":"Patch Set 1:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/800957/800957/1/check-not-vote/ext-glance-zuul/08d764b : SUCCESS in 32m 43s","accounts_in_message":[],"_revision_number":1},{"id":"4c4bc7179ee8748d1ea78ada033eef3befaff73d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-07-15 17:53:17.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"dfec5da479c12681a9829d6539c8448098d9e1df","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-07-15 17:55:18.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"71dfcd8d3bc8b4a503f4de8b15cee1b5e127c1ef","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-07-15 17:55:59.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"9933f647c0fd9d5db9ce0976c2a5f71c0af088a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-15 19:18:50.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/5c9d4bd100da4e5d86578c136412dc5f : SUCCESS in 59m 33s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3e57e951a25d478a8967c996fdadab8a : SUCCESS in 1h 03m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b6b5bf5b145448c9ce37eb240fe60e8 : SUCCESS in 5m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/26dc4958d337451cba709432a1000cb6 : SUCCESS in 6m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bb505deda9c44a1196d15d96d59c88d6 : SUCCESS in 5m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/46c91441fe604ec79eedfaf866757a07 : SUCCESS in 6m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/05b213d078694a11aa35d76c7f0ef3fd : SUCCESS in 6m 04s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6602802529914eacaf223e9c6da5a764 : SUCCESS in 8m 05s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c94fc09928124c138abf0a3419c4bd1a : SUCCESS in 8m 38s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/e4e9f554d27c41baaf153655e09d7578 : FAILURE in 7m 30s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/ccc27144d33c44d7bf58b41cddbd1ad3 : SUCCESS in 5m 47s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/59420a3b1f8f4087b76a68962383df14 : SUCCESS in 1h 00m 53s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/e42d02c10a8d474093007218abadc7b3 : SUCCESS in 1h 02m 12s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/c768066a282f4f39ba71a1a23379e8f5 : SUCCESS in 58m 17s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/b347ba781d254db9a7712bb029de1797 : SUCCESS in 1h 17m 38s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/28f8915cf9f34539b1a466bf47e00e6a : SUCCESS in 54m 41s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/595d83019d6f4a78bf82436c3cf03da9 : SUCCESS in 1h 09m 44s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/38b0f73cc01c4756b19406df7ce297e5 : SUCCESS in 32m 49s","accounts_in_message":[],"_revision_number":3},{"id":"e9ba866340d465897ea8371c1be1f5f929551a29","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-15 19:19:53.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"6ce519bb66783cf40b51a7d56ab72bea037a5c34","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-15 20:51:18.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/a0956c94e92c4410ab2a77fe86435968 : SUCCESS in 53m 02s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/060e0827d5094be58beddf0f63b1c0e3 : SUCCESS in 1h 04m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/877840fc47614f15bf35c57c26a9145a : SUCCESS in 7m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0f2db26ca70c4746ab4cbd0101e61bcd : SUCCESS in 6m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e329fcb8ecb046009bc78f15b8846f87 : SUCCESS in 7m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79cc76aadbe241db8f3f4d2c8367e323 : SUCCESS in 7m 52s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a6c7f21976bd4617b02a1ade9b155fa8 : SUCCESS in 10m 53s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/863832add19f49458ec480f34c72050a : SUCCESS in 9m 47s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/511cb3e322e147ee94542dbd456b43bb : SUCCESS in 56m 20s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/bdddb701845043abafb344fa1a8e21af : SUCCESS in 1h 02m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c01d57ef9ed04a1c83cd72ef5cbda1ac : SUCCESS in 37m 47s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/253340ad307b42bab711278e1a675314 : SUCCESS in 1h 29m 37s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/7409706af4f942debc5a866007ebf3c0 : SUCCESS in 25m 30s","accounts_in_message":[],"_revision_number":3},{"id":"7563c74b66c0f79cf6e56829a796dc0b649ae4a5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-15 20:51:19.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"80b169d63312959f4084aff09b480891856f564d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-15 20:53:06.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cee3ca87716d4a0491b6251b58e1f6fc : SUCCESS in 1m 12s","accounts_in_message":[],"_revision_number":3},{"id":"3e6999ee8b250c74e26506b60f35209d19a36a52","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-07-15 21:39:12.000000000","message":"Patch Set 3:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/800957/800957/3/check-not-vote/ext-glance-zuul/bd0b50a : SUCCESS in 33m 59s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"e33e0debb125bd72b2a631266a51620149712d1b","revisions":{"9f8416713a3f36798ef9eecd26c2ffc17487eaed":{"kind":"REWORK","_number":1,"created":"2021-07-15 15:13:35.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/57/800957/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/57/800957/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/57/800957/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/57/800957/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/57/800957/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/57/800957/1"}}},"commit":{"parents":[{"commit":"2b62197f91cf52566d7acbce6edf68c2e7477a17","subject":"Merge \"Refresh Glance example configs for Xena milestone 2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2b62197f91cf52566d7acbce6edf68c2e7477a17"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-07-15 15:12:30.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-07-15 15:12:30.000000000","tz":0},"subject":"DNM: To test migration constant check after milestone release","message":"DNM: To test migration constant check after milestone release\n\nChange-Id: I3f42e37e783fdeefe60a25211a97d54600c97021\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9f8416713a3f36798ef9eecd26c2ffc17487eaed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9f8416713a3f36798ef9eecd26c2ffc17487eaed"}]},"branch":"refs/heads/master"},"2b26a1f2efb35f01384f410070fd4c360af289e8":{"kind":"REWORK","_number":2,"created":"2021-07-15 17:53:17.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/57/800957/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/57/800957/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/57/800957/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/57/800957/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/57/800957/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/57/800957/2"}}},"commit":{"parents":[{"commit":"2b62197f91cf52566d7acbce6edf68c2e7477a17","subject":"Merge \"Refresh Glance example configs for Xena milestone 2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2b62197f91cf52566d7acbce6edf68c2e7477a17"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-07-15 15:12:30.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-07-15 17:51:52.000000000","tz":0},"subject":"Change database migration version to xena","message":"Change database migration version to xena\n\nChange-Id: I3f42e37e783fdeefe60a25211a97d54600c97021\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2b26a1f2efb35f01384f410070fd4c360af289e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2b26a1f2efb35f01384f410070fd4c360af289e8"}]},"branch":"refs/heads/master"},"e33e0debb125bd72b2a631266a51620149712d1b":{"kind":"REWORK","_number":3,"created":"2021-07-15 17:55:18.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/57/800957/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/57/800957/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/57/800957/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/57/800957/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/57/800957/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/57/800957/3"}}},"commit":{"parents":[{"commit":"2b62197f91cf52566d7acbce6edf68c2e7477a17","subject":"Merge \"Refresh Glance example configs for Xena milestone 2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2b62197f91cf52566d7acbce6edf68c2e7477a17"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-07-15 15:12:30.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-07-15 17:55:13.000000000","tz":0},"subject":"Change database migration version to xena","message":"Change database migration version to xena\n\nChange-Id: I3f42e37e783fdeefe60a25211a97d54600c97021\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e33e0debb125bd72b2a631266a51620149712d1b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e33e0debb125bd72b2a631266a51620149712d1b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
