)]}'
{"id":"openstack%2Fglance~802054","triplet_id":"openstack%2Fglance~master~If2001ef652e21f54048db0fc67893337681a2b6f","project":"openstack/glance","branch":"master","topic":"secure-rbac","hashtags":[],"change_id":"If2001ef652e21f54048db0fc67893337681a2b6f","subject":"Implement project personas for metadef objects","status":"MERGED","created":"2021-07-23 16:34:12.000000000","updated":"2021-09-03 02:11:15.000000000","submitted":"2021-09-03 02:08:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":35,"unresolved_comment_count":10,"has_review_started":true,"submission_id":"802054-secure-rbac","meta_rev_id":"1526924fceb806b3ffb131b651170058bcd59141","_number":802054,"virtual_id_number":802054,"owner":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-03 02:08:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"all":[{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2021-09-02 22:05:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"all":[{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":1,"date":"2021-09-02 22:05:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-23 17:41:13.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2021-07-23 18:39:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-24 03:41:51.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-07-28 14:51:33.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2021-08-04 07:02:00.000000000","updated_by":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"reviewer":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"state":"REVIEWER"},{"updated":"2021-08-04 07:02:00.000000000","updated_by":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"}],"messages":[{"id":"5185ac3266eca1b6967ab023763782f5b0c082c7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2021-07-23 16:34:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b936dac832a6bbe0021e6865dee1e09290e02174","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2021-07-23 17:02:22.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"7a2b7def0294aee1f7913ea6e59bce133a1744fc","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-07-23 17:41:13.000000000","message":"Patch Set 2: Code-Review+1\n\n(2 comments)\n\nFew suggestions inline.","accounts_in_message":[],"_revision_number":2},{"id":"76f64bc1dfb07101aa737f206b62c248f6e23339","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-23 18:39:35.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/057663136dfc4badaa72ab96901e5f92 : SUCCESS in 1h 07m 22s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5ad6ee88b9ec48c583bd3d7397f0b019 : SUCCESS in 1h 08m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac085923413349469af378209ff3f166 : SUCCESS in 6m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bf89c23da9f6472e8438e505d256cf85 : SUCCESS in 5m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cb3b63b5365f440287698962a2c723c7 : SUCCESS in 6m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/79066023ec7345ffbd49ea4aafab5810 : SUCCESS in 7m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18752a7e96b64f5db096015b7013fde8 : SUCCESS in 6m 54s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b3944c75d47641a4ae467230b370ef25 : SUCCESS in 9m 17s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/89f62e6a34984b8093c7f6f70436db42 : SUCCESS in 8m 57s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/2d72af233f3048f7ae26fa183cc15861 : SUCCESS in 9m 35s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/a428db260a864a9c9918bbf6098cc48e : SUCCESS in 5m 16s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/9af14a6c09614b9c8830f92943021330 : SUCCESS in 1h 19m 26s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/1f06e12094a04a51a09b4c8ab55f48e7 : SUCCESS in 1h 11m 44s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/4f914599c36d4982ac28a6ce4d7d1e47 : SUCCESS in 1h 01m 47s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/fedc014c11774bd2a8f3e438cac7083b : SUCCESS in 1h 22m 48s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7324a934baa04ab9909a029eb8c01dec : SUCCESS in 50m 39s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/c0fd4639528748f38ed2f83d944998a5 : SUCCESS in 1h 16m 14s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/fa3b18b1814f44d2a53ca47b86f02e36 : SUCCESS in 25m 04s","accounts_in_message":[],"_revision_number":2},{"id":"7fe10f79936aa7c1e74e6e6d99023e2e5043ee65","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-07-24 03:41:51.000000000","message":"Patch Set 2:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/802054/802054/2/check-not-vote/ext-glance-zuul/fb28a7e : SUCCESS in 35m 43s","accounts_in_message":[],"_revision_number":2},{"id":"4d61859d06cf5f9a49631d9b89c0e7ccca3d8767","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2021-07-25 19:36:32.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"a699cbe08d2c0aafa2e74341a720cc6a3e6ac900","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-25 21:02:12.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/cc26c8bd7745490f8be9d02e0594dfbd : SUCCESS in 59m 16s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/166357d23e43468bb22f74164c335b1d : SUCCESS in 59m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42b5e623c71b465b9fe44c4abf61ba9c : SUCCESS in 5m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/244a6b4554114a4594e828fca26984b6 : SUCCESS in 5m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8d8bf7fd697149a58c33fbb3d54628c5 : SUCCESS in 6m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e6a1c587de3e4cf8af2042ceaada5107 : SUCCESS in 6m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6fa72d56cc85485caa9921f66bdf2ad0 : SUCCESS in 7m 09s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3116949487bd474aaaf808f671edd25d : SUCCESS in 9m 05s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/5a8b2094835a4fafb9f3ffd93210d32f : SUCCESS in 9m 01s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/e4b4c793fbc64b3694eb5a1ffb2daaf3 : SUCCESS in 8m 44s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/ee577975dd4a48818965982147f15315 : SUCCESS in 5m 47s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/e0db844e474142819d63ea2fd87e5ef6 : SUCCESS in 1h 19m 23s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/f9319ef42929487b8a8ac7005f4dc623 : SUCCESS in 59m 52s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/3de80bd093da4fddaa59a0efcd90be24 : SUCCESS in 1h 05m 28s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/c1e24be34ae64732a45eac78dbb526e7 : SUCCESS in 1h 12m 35s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f3d15c8f020546168f9bca5fa4d4809c : SUCCESS in 43m 05s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/d7279f59300246b48140884b41c38a36 : SUCCESS in 1h 08m 36s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/b0ea588808c345c09e7571a8a6c16cf6 : SUCCESS in 31m 32s","accounts_in_message":[],"_revision_number":3},{"id":"5308b425fbdee08830722de6077696167d7829f0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2021-07-26 10:16:28.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"00704220371e67519d4b22f7b9395f8f0b2016bb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-26 11:38:40.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/295c44cf184d46b1addcfc1b62480f49 : SUCCESS in 57m 47s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/46f6f008c4da4f9fa8d400d1ebabb2e4 : SUCCESS in 1h 06m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8018460ddd944031a39223e7b68e621a : SUCCESS in 6m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/573d8826245a400c8a55b912069318b4 : SUCCESS in 6m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/88ab700c5bf141eba888ab19dc570143 : SUCCESS in 6m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/607dc10afd8340eda885eab38d707a0f : SUCCESS in 6m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e6298a4f82443379ab27e4f6a34bbe6 : SUCCESS in 7m 47s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f5aba77fd67d44e1abbd850d14112fd1 : TIMED_OUT in 31m 28s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/84038b2d6512427a832f278327a3de27 : SUCCESS in 11m 30s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/d987c98c071b4bf0b9423fd42ffa5636 : SUCCESS in 11m 46s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/2dc693367d094dc7af752fcfa5427202 : SUCCESS in 6m 16s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/f0b44c01eba94c19bfc7924edf4a8543 : SUCCESS in 1h 13m 04s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/c5712099a4da4139856e4d6423126e9d : SUCCESS in 1h 10m 33s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/861c8b81cdae48a99f04748d401a93d4 : SUCCESS in 1h 17m 03s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/4f956aef1c544a119c7752717bf43fba : SUCCESS in 1h 18m 56s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/95e8ac3e35a241d2a5c20794c069fbd4 : SUCCESS in 43m 36s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/03697ae9aace41b884b14fba658e997f : SUCCESS in 1h 04m 41s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/366819a5de8042a3b5c8d70b6d47ab33 : SUCCESS in 31m 18s","accounts_in_message":[],"_revision_number":4},{"id":"2903c444bb4f17f519c9a3c2633231bf67914e7a","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-07-26 11:40:24.000000000","message":"Patch Set 4:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/802054/802054/4/check-not-vote/ext-glance-zuul/d6a966b : SUCCESS in 33m 46s","accounts_in_message":[],"_revision_number":4},{"id":"6354cbc641fc11858fe0d2c30eb70d59657fed33","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-07-27 08:24:52.000000000","message":"Patch Set 4: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"5155b15f7d69c08840815973b0c00c559e5cf9e6","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2021-07-28 14:51:33.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"f116d6c55ab11a6b153db086309ac363a3c90fee","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2021-07-28 18:30:46.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"fdfd836f88970ecd7cf3b83b61fe4ff28ca82c8e","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-07-28 18:41:25.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"37ce6b4b6390da4c085d53301c950376f31a5a71","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2021-07-29 11:05:07.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"be28b0261b7af5b3c38c9cb47abf7a61892beb26","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-29 12:37:26.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/240fbff9e9c54ce2a1d5a863b0abba1e : SUCCESS in 1h 00m 48s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/60e4061a5b0b4a49895b3c59bcba93df : SUCCESS in 1h 10m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e18fe5e9385411bb7c768a9fd5f0e3a : SUCCESS in 6m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/efda4b2c1bb74bb594722e6d56d5469b : SUCCESS in 6m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/729d48be047a4589bea729503aaec38e : SUCCESS in 6m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fb90d59016ee4e92a8ab352433a7d332 : SUCCESS in 8m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c9f3735f7ba470b893d02b5bba4dfda : SUCCESS in 7m 39s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/22ab27e81aa14c5a88045737bf35496f : SUCCESS in 9m 11s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/d378e04780ca45cbac01e4db4a2f5069 : SUCCESS in 10m 03s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/b253e262a5234feeaa46408b175bb363 : SUCCESS in 8m 59s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/43974978e4ab4d2a96c73827b151c2be : SUCCESS in 5m 31s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/5b026ae4e7d144d49e0b530f76c449cf : SUCCESS in 1h 21m 13s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/70fe42a0a93d4ef18e5ca449f905bd86 : SUCCESS in 1h 07m 57s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/1d0b1b82b98143fda5ed5b8711012337 : SUCCESS in 1h 10m 32s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/787fb7d75ecc472fb904ed02793d64e6 : SUCCESS in 1h 23m 55s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c7cc20945f1a4cc5a7b37e77120d65bb : SUCCESS in 54m 53s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/ee08b18485ea43ffae87663afee056d7 : SUCCESS in 1h 19m 02s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/8e6ca83a8d73433fa31b67ae6e77a0fb : SUCCESS in 36m 26s","accounts_in_message":[],"_revision_number":5},{"id":"429192e2bcd3088466b3806ae151f4368e7d259b","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2021-07-29 12:38:22.000000000","message":"Patch Set 5:\n\n(3 comments)","accounts_in_message":[],"_revision_number":5},{"id":"fe1b5646074c43cfe79bce1eabc3b2e6135c7407","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-07-29 16:03:31.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"6f33bb5c3ebb61cda94a10aa10acaa72afb580eb","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-07-29 16:04:24.000000000","message":"Patch Set 5:\n\nAs commented on parent patch will move to +2 if other reviewers are ok with it.","accounts_in_message":[],"_revision_number":5},{"id":"e5b7e63ca19744c66d772514a9c025ebcb0ad77e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2021-07-30 11:26:59.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"872ed0789b6acf038473117daf0c39aba91209bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-30 12:49:11.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/fe2e2e35a3ba47b39c584807481bf57d : SUCCESS in 58m 11s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4a222d4fb17a4a8ebb59d432bf9f15e5 : SUCCESS in 55m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/214a83e8160d4b6699865600ad2d2539 : SUCCESS in 5m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/78435b40d471412a88eaaad5d0831a0c : SUCCESS in 6m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9f66a6ed96f04f2989436818292f268f : SUCCESS in 6m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a60075865bbd4d828ab23299037a4e82 : SUCCESS in 6m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52a7f7af13b24213b03c130a44793149 : SUCCESS in 6m 56s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e20322ab6aa647bea7da9828f42708c2 : SUCCESS in 9m 49s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/45c85272e55b423a9ed40173361fa7d4 : SUCCESS in 8m 43s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/da916f981d614b7ba7ed71206d2a68ff : FAILURE in 8m 12s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/46309f78ccf8405389256bd02e9ebc82 : SUCCESS in 7m 05s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/19fc1587326d488a9d12d22832df9f7b : SUCCESS in 1h 16m 01s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/a6e51b42916d414caf5a11d4ab6a993d : SUCCESS in 1h 02m 16s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/601c3e397cd7485fae9f642f5f813252 : SUCCESS in 1h 20m 24s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/582fe6f7ab8b40d983fa186f29318ddc : SUCCESS in 1h 14m 08s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9d5c794a859b4bda91fae8db26475f05 : SUCCESS in 53m 21s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/1203ec3fc7004ead94465304492557f0 : SUCCESS in 1h 18m 29s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/2433fef24a4d417aae09f09bdb5df66b : SUCCESS in 37m 54s","accounts_in_message":[],"_revision_number":6},{"id":"fdb1a8ea02b0c19b7e94e0a7ff49beee60d08b4f","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-07-30 15:31:26.000000000","message":"Patch Set 6:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/802054/802054/6/check-not-vote/ext-glance-zuul/6c76b35 : SUCCESS in 36m 05s","accounts_in_message":[],"_revision_number":6},{"id":"f99b5b8a4c2ef11181ed17b63114f8df5c15007f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2021-08-03 05:31:34.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"bb946016ac879bddbbdef0dc4078b388907b06b8","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-03 05:50:02.000000000","message":"Patch Set 7: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"b2545ec9a1438b4f086f1894a1c1872c0ebd21b1","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-08-03 06:08:56.000000000","message":"Patch Set 7:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/802054/802054/7/check-not-vote/ext-glance-zuul/a37fe12 : SUCCESS in 36m 15s","accounts_in_message":[],"_revision_number":7},{"id":"4e08fa6abb10977ed0fea1779ba60dcf4ae12603","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-03 06:59:52.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/f64db83b8dd94ae2a5e8df74457a5e4f : FAILURE in 47m 23s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/483f05a1479f465fb3976b588dc90935 : SUCCESS in 1h 16m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/872cad0a26894ddc9fa28aea8912fc0f : SUCCESS in 5m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/53be598f472449deb7def5a03df0e97f : SUCCESS in 5m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aa40f680cced494fb29e9b90fc037994 : SUCCESS in 5m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c3b63b1459744cc3a8c0e99e16920e2c : SUCCESS in 6m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1051d258a7af42508841bdc313f4cc6a : SUCCESS in 6m 04s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/190978d58ddd4c0e950515d6254028e7 : SUCCESS in 8m 02s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/8e3161144f2d4751a2103010da1b3348 : SUCCESS in 8m 39s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/851616b6daf9400fa9454ef27e4e3e83 : SUCCESS in 8m 39s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/31be9775ed8643ba89014a2a8e09e62f : SUCCESS in 5m 44s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/01aefc397a0746ae9189f459c8d499c1 : SUCCESS in 1h 15m 59s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/35c1c1916a204d0e981b3b7756ad920e : SUCCESS in 1h 14m 32s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/5a7a5209dbdd4f0ab964712e0181217b : SUCCESS in 1h 13m 38s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/17b7c80b089e4c49b154b77ab465dfa2 : SUCCESS in 1h 22m 14s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3bf01069386944f6b640130515755e5d : SUCCESS in 52m 47s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/bf1d6037f5164372a317a5df743157f3 : SUCCESS in 1h 09m 32s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/f63b0142832b462f83c7ebd84aab8a00 : SUCCESS in 34m 18s","accounts_in_message":[],"_revision_number":7},{"id":"35322afc4e9787dcdb6cbd5d89fd03836f77f754","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2021-08-03 09:00:47.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"8718cd8a619b28337b2b4864fc21125ed2b8e479","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2021-08-03 09:05:26.000000000","message":"Patch Set 8:\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"f1399c1b1564ae0cee2722ca62529a079700ee3f","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-03 09:21:52.000000000","message":"Patch Set 8: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"6b306c5a71718bda400742ae8ee9051d8b28cf5c","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-08-03 09:38:33.000000000","message":"Patch Set 8:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/802054/802054/8/check-not-vote/ext-glance-zuul/23ba8d7 : SUCCESS in 37m 00s","accounts_in_message":[],"_revision_number":8},{"id":"b6b4dfdc9d5c257e860857bd12d7ee5675ada42a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-03 10:36:51.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/7c2bfbe433134121b5ec10dc3203c3d2 : FAILURE in 49m 56s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4f7ef2dd736a4990a2649a56681cddd7 : SUCCESS in 1h 16m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/48b9bfd4b6ad41599b894806f94517ad : SUCCESS in 5m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f6fa4a9d70214076adc2901c08363119 : SUCCESS in 6m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2a6474ebf5184e1c923813cdd345dda7 : SUCCESS in 5m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e73e8b68fd5a4691a9d76835399c91b9 : SUCCESS in 7m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d7ad08794a0c4e00b0b317d4800edb38 : SUCCESS in 7m 07s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2e6bcd7be0c947e89037ff0b1f473ed3 : SUCCESS in 9m 25s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/d0513ca729034eecbe7398494426e505 : SUCCESS in 9m 31s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/152a43bdabd14ef68d3eeb1d2726c9fd : SUCCESS in 9m 12s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/c6b5c6bee3f84eefa5f63a7e095459ff : SUCCESS in 5m 27s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/bd899dd64fe64e9ab06f5aeb429e5030 : SUCCESS in 1h 24m 12s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/b1c70274e4e041328d57f26efa53199a : SUCCESS in 1h 11m 08s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/d133aed0e5a243efa595e496344770f6 : SUCCESS in 59m 16s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/5ea5902279c84176a408d773e060240e : SUCCESS in 1h 17m 43s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/09fa030ffaf249c9965d0678d792fd67 : SUCCESS in 1h 05m 41s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/bbd88ab2e51449d1a6f824b1d234a57f : SUCCESS in 1h 04m 58s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/8f3842d1513c4d6e858d61def9c36ac2 : SUCCESS in 36m 33s","accounts_in_message":[],"_revision_number":8},{"id":"640178fb1d8a7dbec8c67d3bd3bb07d3c92b29a2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2021-08-03 11:22:18.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"4eafa4597b07dc6989a2b79e689ea908a111952e","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2021-08-03 11:26:41.000000000","message":"Patch Set 9:\n\n(2 comments)","accounts_in_message":[],"_revision_number":9},{"id":"b9d3418a6c69ce79e6d008b88110180fec545f8f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-03 12:52:15.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/d1848441863c4d60b5f0c336b0e552e9 : FAILURE in 52m 21s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/578f582b8ca646e0a6dfb71da8b3c660 : SUCCESS in 59m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e3f09e3ad1394377bbfb3ceb46abf6b3 : SUCCESS in 5m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b59668653a5445de97fbfa8864789107 : SUCCESS in 6m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/51dff6bb49fc4e9eba344ed5616514bc : SUCCESS in 6m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d44b4ae4e9894fef93f2079ca75e92da : SUCCESS in 6m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0270b5b2f7c64faca7e546bb013dcdf7 : SUCCESS in 6m 50s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/39fd81dffa36404da0b29d46cb302fc2 : SUCCESS in 9m 07s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/da51132632514f358cf3f965148d9cf2 : SUCCESS in 8m 41s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/90280cd236f640c6ae0d95c2dd62149a : SUCCESS in 8m 57s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/6d2b3a7254ae475c80cbad0cd51502d7 : SUCCESS in 5m 34s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/4f3037a32283474c820c1212661e1f71 : SUCCESS in 1h 17m 52s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/006a3e6f3c84437390881d77bd294be9 : SUCCESS in 1h 14m 05s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/7a4ad9100fdf4af7add5b0077453a879 : SUCCESS in 1h 16m 27s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/78797760fbbf45518ee569f1f0f0320c : SUCCESS in 1h 21m 12s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ebf631e15aea43dca7a0a04622ed6785 : SUCCESS in 48m 23s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/cb08cb1f57bf4610a05f13ea80dce08e : SUCCESS in 1h 12m 03s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/7970bd557a60410790b714d639cf0280 : SUCCESS in 40m 22s","accounts_in_message":[],"_revision_number":9},{"id":"a94548600246626a2388fc01ae357f9fc66f5279","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-08-03 14:14:12.000000000","message":"Patch Set 9:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/802054/802054/9/check-not-vote/ext-glance-zuul/223423f : SUCCESS in 37m 30s","accounts_in_message":[],"_revision_number":9},{"id":"9238ff627ce8ba271b8021476166de46b24b4de5","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2021-08-04 05:23:29.000000000","message":"Patch Set 9:\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"cbcc730a3e5eb411b20c12816363197ec469d9a3","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2021-08-04 05:24:17.000000000","message":"Topic secure-rbac-tags removed","accounts_in_message":[],"_revision_number":9},{"id":"e74586ca02d7a7148be9431c1807aee98d9f6ccd","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2021-08-04 05:24:25.000000000","message":"Topic set to secure-rbac","accounts_in_message":[],"_revision_number":9},{"id":"0c43b29966d09fad04f3fceddf36a94325115f91","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-04 06:39:46.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/65ff0dd13b924e8281e220825fb36daa : SUCCESS in 1h 10m 29s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/160c506cdeeb4a8ba9694e86913d5e9a : SUCCESS in 1h 01m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa2ec809d72a414a8f89e1ef10876e8b : SUCCESS in 5m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8fba54b576454cdc93e4a3d29865c3bf : SUCCESS in 7m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/51f03fb94dee4f44a5537b44135e319f : SUCCESS in 5m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/728975e729d64ef4a7caaa65465023e1 : SUCCESS in 6m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7d8156aac2f4578bfe0fc1026d9ee7c : SUCCESS in 6m 32s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/cc2a6e518f8c4db98830a7e2e3b3eaa6 : SUCCESS in 8m 16s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/123751724eba40b7904c56668e22317f : SUCCESS in 9m 32s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/e8dc56e6c7ee49fd8fa062ebdd1a76cf : TIMED_OUT in 31m 06s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/cc1e058b1bb04c60a98edc98b1800619 : SUCCESS in 5m 05s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/07a3c11ee7d24216b4de3c2be43fece1 : SUCCESS in 1h 10m 42s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/48005410950b461086999564111c72e5 : SUCCESS in 1h 07m 00s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/6783216765fe4542a529eeb93ac995cd : SUCCESS in 1h 01m 47s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/bfa9766e533a4fe088ef480a02e6f2ee : SUCCESS in 1h 10m 52s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/856a0e0cbae14eed9d92a45fa8f445db : SUCCESS in 43m 11s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/6bc30fcb9b6246528fd9f3a8f87fd4a3 : SUCCESS in 1h 14m 45s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/6af893cbfdde4dff8b23980d49efdc5d : SUCCESS in 33m 48s","accounts_in_message":[],"_revision_number":9},{"id":"7197c1b1fc9ed3c622c5b212162bc6b30226659f","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-08-04 10:14:12.000000000","message":"Patch Set 9:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/802054/802054/9/check-not-vote/ext-glance-zuul/fcc3a6d : SUCCESS in 35m 46s","accounts_in_message":[],"_revision_number":9},{"id":"1d762b0d385c6a221243f6cc7b39c3ecb07d4d16","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2021-08-16 14:07:12.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"eaf8d69aaf9bfdba4b87a7c0eaabc30c3c190e3b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-16 15:27:37.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/e483d268dd9646a4aed2696156ae6e09 : SUCCESS in 1h 00m 36s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a048a2902df54f219ce3e0dda1bd953e : SUCCESS in 1h 04m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a538986f75ac40b68b6e14ed4f79eabf : SUCCESS in 5m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cdff2b20813a4aefa922c85c37092b98 : SUCCESS in 6m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7bf9f23149e44717b7914ea888838a93 : SUCCESS in 7m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5e8c893860004e56abdf237a0a420c61 : SUCCESS in 6m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ef58662df224b499dae4d795231c640 : SUCCESS in 7m 27s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/552eb364d60b448282952b3730fd1b37 : SUCCESS in 9m 18s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/4db8269986344528b95d201d747a0991 : SUCCESS in 12m 11s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/2974dcda37094f119be78add26372b91 : SUCCESS in 8m 38s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/e4ade4cd128241f2bce8bf487e63f2e1 : SUCCESS in 6m 14s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/e70e3e40c65946ef8d857ba9fdecf825 : SUCCESS in 1h 12m 18s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/562604d0feef48b480e8ab283d8125ca : SUCCESS in 1h 12m 10s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/5b316e6231ba4655b0e44d5da5b6ad35 : SUCCESS in 1h 02m 45s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/d66afbf2dfe646148ec4ce7f47a50037 : SUCCESS in 1h 15m 17s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ffceffc2779948f999f4cd33ec4d533d : SUCCESS in 49m 05s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/d5689f7d8f43414ab94e92e2acd1bee7 : SUCCESS in 1h 16m 38s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/7153d371525b4838bc4b34aa62b1be8d : SUCCESS in 40m 01s","accounts_in_message":[],"_revision_number":10},{"id":"1c5d6e10a14718d02ef2a30dd8178216b716f8ec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2021-08-23 08:53:17.000000000","message":"Patch Set 11: Patch Set 10 was rebased","accounts_in_message":[],"_revision_number":11},{"id":"e0e4a5aadeb5b37a157945841f90ae917bf3329a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-23 10:12:31.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/4841a07cbc22475e885de19b54ef6cf3 : SUCCESS in 55m 09s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/45efc7a073c04c8b9a6526ba01ee7ba1 : SUCCESS in 1h 13m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/43728dae5f4a4fbcac8ef82be77b3566 : SUCCESS in 5m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/57d9a6ec40794999b7fa0fc2072b0c22 : SUCCESS in 6m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d983ed371ae54ee18d049e965c0f301a : SUCCESS in 10m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d9ecbfae1eeb491e9c13a73e6636fb70 : SUCCESS in 7m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eefe3cc174df4cf8b8f2bf239a0a40e9 : SUCCESS in 7m 13s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3cd07776b5044f97a48cee853003785b : SUCCESS in 11m 00s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c4e0e776c0904618a029fd4cfd98f891 : SUCCESS in 9m 35s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/f897e906e447417698948a3289565a85 : SUCCESS in 8m 35s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/9977835fb6274564aaf9131013c07a25 : SUCCESS in 5m 42s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/d515003e4c03412db7e0a2c5f93f3a72 : SUCCESS in 1h 10m 19s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/d3974970a4364c93a3eebb08ee2094e0 : POST_FAILURE in 20m 50s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/0c61ee3e39f44ff0863203ee8e81a1bc : SUCCESS in 1h 05m 40s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/b2926985cb274788a096545527342495 : SUCCESS in 1h 13m 01s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dfd1c78aa8034d8898777c48832cffc1 : SUCCESS in 45m 54s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/d220038f7d26441dabf681f5aa6d7f52 : SUCCESS in 1h 15m 38s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/dc25bd20206f4b5c870e22637f68bc6f : SUCCESS in 25m 12s","accounts_in_message":[],"_revision_number":11},{"id":"320f8ab41a8d827cc784a98a5531bf6888c2b424","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-08-23 11:44:07.000000000","message":"Patch Set 11:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/802054/802054/11/check-not-vote/ext-glance-zuul/4700840 : SUCCESS in 32m 46s","accounts_in_message":[],"_revision_number":11},{"id":"8462fdc25334aa1008c3169f7b94d288c685b669","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-26 04:02:48.000000000","message":"Uploaded patch set 12: Patch Set 11 was rebased.","accounts_in_message":[],"_revision_number":12},{"id":"e211b23ed002dfbe24ad50b8ac0d6e5f5c8f5988","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-26 05:27:40.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/8e80382f27a14faab02e2761540e4530 : SUCCESS in 1h 08m 15s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c71f14e7f9874121b0ef70566195b05b : SUCCESS in 1h 08m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d478ca9708844c2a74ea95a696db4a6 : SUCCESS in 6m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2e2fb91e20764849ad26af2c962109f1 : SUCCESS in 6m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c8c3e4ab2bf5451883fc693678774746 : SUCCESS in 6m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/901833d8cd294433a39377066f6f4ee1 : SUCCESS in 6m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9674e279c6c94785908af68ed1b4faa5 : SUCCESS in 6m 39s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b45e2ac2ad9f4200bc13eea04df5aba6 : SUCCESS in 9m 04s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/f30d6de6626d453da986ae9eba9d5153 : SUCCESS in 9m 57s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/ad6a3e6fc8794b0c8fddf9b076c0970c : SUCCESS in 8m 21s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/de2c42c83cdb4cf4a73cdbfc0d8524f6 : SUCCESS in 5m 41s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/85379fba206f409ead5ddd1e32e82851 : FAILURE in 1h 19m 00s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/7c6c0ab457274425a83340697048a2be : SUCCESS in 1h 11m 43s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/0b212413030843ebaa5ba11f2a2af33c : SUCCESS in 1h 17m 25s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/e8355b62c74c44d491d380aee4394e85 : SUCCESS in 1h 15m 43s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cfac706aab894ef689bf4e28bd193d81 : SUCCESS in 49m 14s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/c835c859d53c4fc2bc1692065b42fe16 : SUCCESS in 1h 06m 36s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/78f91d462a41402bb3b71f6a75bb6103 : SUCCESS in 34m 51s","accounts_in_message":[],"_revision_number":12},{"id":"b239331cefe7d0f279464218b56366e4474abca4","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-08-30 16:30:48.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"7e2742153f7d7cde7359c9a14d437d1b48498bad","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-30 17:08:15.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"d168bae5fdd408222c2277b10f7bd3be3a00134b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2021-08-30 19:47:26.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"a15268acd2be45614490ab50d95ede64a3ac7696","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-08-30 20:51:00.000000000","message":"Patch Set 13:\n\n(2 comments)","accounts_in_message":[],"_revision_number":13},{"id":"6e0a51f002b05e2b253ff03d8401939cb7987530","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-30 21:06:20.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/cde4cd3fbc1d45f3ad7dbb7c43c15bf3 : SUCCESS in 55m 16s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ea3e7938d2894dd2940c28b473b1003d : SUCCESS in 1h 12m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a2b49ce4d3694d8cb67e5983f349946c : SUCCESS in 5m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3ffcfdf4618d4fccb5714e7705e028ba : SUCCESS in 7m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/64baad2cfcf44e3cbecee9874bd214cc : SUCCESS in 5m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a5116254bc1c45bc8b88b472ca65b754 : SUCCESS in 8m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2739f694aee3467fa7f82eb4fdd9f07d : SUCCESS in 7m 43s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a76963042e39422f8ee2b5a7437e5714 : SUCCESS in 8m 14s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/72b6a7d649a049a39fb9a4eb36255836 : SUCCESS in 10m 11s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/5287d7ca3667455ca231dbb9917e12d8 : SUCCESS in 10m 26s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/02018b684cca44c088a9503e1d150b28 : SUCCESS in 4m 57s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/92352223bf9240e6af30250859103886 : SUCCESS in 1h 03m 10s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/1e18796d0416462b895b3710a5361754 : SUCCESS in 1h 12m 59s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/a912b9ef2eda419984460dabb1f898b4 : SUCCESS in 1h 09m 07s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/ac729bb9df6041fe904ff816463da6bf : SUCCESS in 1h 01m 27s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2dc274b770fa45f6a94e54c41d7c5e85 : SUCCESS in 50m 57s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/cf6300be36a548c2abe86ea768249168 : SUCCESS in 1h 14m 28s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/453d42b49bd342f0b2b18d12305a8e5f : SUCCESS in 24m 23s","accounts_in_message":[],"_revision_number":13},{"id":"8b682df1abea2d24d2c62aa9a55ef2de04a86c27","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2021-08-31 06:08:04.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"f128b33e20a78d83f77f7852dae1f6fc7609fc24","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2021-08-31 06:09:30.000000000","message":"Patch Set 14:\n\n(2 comments)","accounts_in_message":[],"_revision_number":14},{"id":"5fcb0c922265db3b48f76514a19a40ae6b162548","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-31 08:50:17.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/ab0730fd4b7d4094aa561b2e3a17bdc0 : SUCCESS in 1h 06m 34s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a9c0cfb734b54b68aac8108b32d088b8 : SUCCESS in 1h 10m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f0741d3263147c7be4776b689839365 : SUCCESS in 7m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b08606dd7cc8455199ba17b559c3aaee : SUCCESS in 9m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6e7eed85c89a4c5db1026d822f0dea0e : SUCCESS in 8m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/187e3790812b481a82ae62f56c8cb59e : SUCCESS in 8m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6d92ddc14a9479aa59f721ca8a61850 : SUCCESS in 9m 22s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f747e344b2e54eac9bb45d25fa382ef6 : SUCCESS in 11m 16s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/7659cf9b4edb4c81aaa1434776cd0c42 : SUCCESS in 11m 05s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/b7e1a564bd8844eb8f6fc06455781ecb : SUCCESS in 11m 14s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/973d3ae8be2a4bda8601029db526f8ba : SUCCESS in 7m 34s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/109e34f2c0ab4edb83f80710f58d38c0 : SUCCESS in 1h 28m 10s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/2348c4d2cbee453990ade9f106b78920 : SUCCESS in 1h 15m 54s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/2b8adec091d440bd95f8aa0503143bfc : SUCCESS in 1h 06m 26s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/9fecb5bfc92544f98ae146b28589e63c : FAILURE in 1h 03m 51s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8d47e2c609bc4ca5b290fff2b1dc1694 : SUCCESS in 50m 21s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/078d6fd96e4a4867b15f919fd0fe0856 : SUCCESS in 1h 06m 19s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/c1a69cd4a3504fd295f0be3396731863 : SUCCESS in 35m 11s","accounts_in_message":[],"_revision_number":14},{"id":"bab9b6dabc8a758885b3e3ecdf042a53f16cb541","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2021-08-31 20:04:52.000000000","message":"Uploaded patch set 15: New patch set was added with same tree, parent, and commit message as Patch Set 14.","accounts_in_message":[],"_revision_number":15},{"id":"7ede0d735685e7b3434b8749a12792cde2e37ed9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2021-08-31 20:20:51.000000000","message":"Uploaded patch set 16: Patch Set 15 was rebased.","accounts_in_message":[],"_revision_number":16},{"id":"6b5ea1a7276c85a66c494bc50373d672107fad7f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-31 21:46:56.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/1443c4b898424c05b84be8953e576e8a : SUCCESS in 56m 36s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d340b9a085df4aa0b56be01b82664c36 : SUCCESS in 1h 12m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d6a890bc18947678bc7521332485ac2 : SUCCESS in 6m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4af0cb2b98974d4a86ca5a4f9b240639 : SUCCESS in 8m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1bdf17906e1641a1a27885d53f1cb393 : SUCCESS in 6m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cd202ec339574230b46b8681967d3a24 : SUCCESS in 7m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/869a5f0bb3eb4f02b2d7b0de5cf7efa4 : SUCCESS in 7m 27s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9b66636813504593a6b444c2bd679934 : SUCCESS in 9m 11s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/88eee14ab8124ce29a88ebc43c55ffc3 : SUCCESS in 10m 09s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/45a4d04f3440424fa8f283fb47dc49fd : SUCCESS in 8m 54s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/84d26a980a6d4199bd62746f7af5a9c1 : SUCCESS in 5m 04s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/1915df4a77ac4b94bd4b1a01522d58f8 : SUCCESS in 1h 10m 58s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/e273aa7cec1a48df9364fc55a0baa219 : SUCCESS in 1h 13m 15s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/687cf712be864ec1ba2b873c6c9bed45 : SUCCESS in 1h 06m 45s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/8db7aded1c5e4fba8547dd8427d9a99b : SUCCESS in 1h 19m 53s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/97ae78876f1e4d9b830eb17b380762bd : SUCCESS in 47m 29s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/f51e4f1ceb5d4c5f8d8a5b28dd03fa75 : SUCCESS in 1h 05m 13s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/d0f80736337f40918d4ebd8826baa628 : SUCCESS in 32m 11s","accounts_in_message":[],"_revision_number":16},{"id":"cbaa2a6163f6c0a00522f91f1ab1a5e1b77917dc","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-09-01 17:19:44.000000000","message":"Patch Set 16:\n\n(3 comments)","accounts_in_message":[],"_revision_number":16},{"id":"4eef0800a1b766cffda578d7b7031fa97a13731f","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-09-02 09:14:05.000000000","message":"Patch Set 16:\n\n(3 comments)","accounts_in_message":[],"_revision_number":16},{"id":"b40ceac209e6693faefaf56f67e8c6b3d7f1796b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-09-02 09:20:55.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"d5ad032d0d9362a97b1dfc61d4782868e313eac1","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-09-02 09:28:52.000000000","message":"Patch Set 17: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"567884a651556d17f6173d500b55fd485cd62dad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-02 10:56:21.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/56ef6a1a94694e4da760f403065eeb3c : SUCCESS in 1h 05m 36s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/bec27cb3cd8e4f168cf1eb03d85ead80 : SUCCESS in 1h 02m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/abaad27a845544baba628febc56a37a0 : SUCCESS in 6m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aea80b64f9e94b67af376e57fdf84d67 : SUCCESS in 7m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5d2843f44e6e4f70a5e0a461bd0a4d48 : SUCCESS in 10m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9c8bb65c8e154346b18cd3f39ad64959 : SUCCESS in 10m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48d7362da61443dd8a6afc3ae654b222 : SUCCESS in 9m 04s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/480bfd75392d4bb0a04361d0a7d1aac7 : SUCCESS in 11m 49s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/6e8f1f6cf78c49bc98ed0c3f3c4fabe6 : SUCCESS in 17m 20s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/9a1e6a3f30ac41539380da24d4da3b68 : SUCCESS in 12m 42s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/1ba264d945c540d98fa9d7ee32424d6e : SUCCESS in 5m 51s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/d80e53480bee4755aa73c04facdaa53c : SUCCESS in 1h 25m 28s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/3453a2e6c0a7493383d5efd4232552cc : SUCCESS in 1h 07m 05s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/5cbf5d5cb5a04c2a96f98f0030040b4c : SUCCESS in 1h 03m 36s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/633c5862d4304715a42eb75bea143114 : SUCCESS in 1h 19m 35s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/157885bcb7dd4c76b39c4402e35c8389 : SUCCESS in 50m 17s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/0c36beb3035d4f90a9707fe411801b2c : SUCCESS in 1h 29m 07s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/fd871e687a974d6aae4072d28e134dd2 : SUCCESS in 34m 34s","accounts_in_message":[],"_revision_number":17},{"id":"4b2aeb5fb66c0535f14a253558ebb584c377a4f4","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2021-09-02 12:17:27.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"7b2e12496734dd21d0077eb8dc275d775bea3c4a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2021-09-02 12:18:28.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"f5a922e0d83e86d99ea96318e0410533860c3e29","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-02 13:51:20.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/8f9e260ca9bd4980bf9d650f078008d5 : SUCCESS in 1h 01m 51s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/dd8124d7d5ee40bd84c2037783133a47 : SUCCESS in 1h 16m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fbb7d81956a94a3786c96f1c1c8ee06d : SUCCESS in 7m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c924ccddf40041fcaa63dcf7f6293dbb : SUCCESS in 6m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a4a0d66c459847d0aa08545117b6dc01 : SUCCESS in 9m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/201c65ca2baf4ac7834ecf2485951377 : SUCCESS in 6m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6fae5cf909424c0fa05a178f988bb235 : SUCCESS in 6m 54s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5488b3a6ea204c4b90e10559faaba004 : SUCCESS in 9m 41s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/d20133a8ac7b4f7f9c004edfd7cae8f6 : SUCCESS in 11m 51s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/ab1ca837325642d1990be97f0c0269b9 : SUCCESS in 8m 59s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/c9fbe7d036774dbb9c769d7ff1653356 : SUCCESS in 6m 40s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/a56395495d964cb0a71969a51df468b9 : SUCCESS in 1h 13m 13s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/35a716f4cd884558a8b6f8b49a7b4601 : SUCCESS in 1h 06m 56s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/7ff8af09d0fa4f20a93915f0169239d5 : SUCCESS in 1h 23m 43s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/549ef547f632450ca0c3ea2724ed717a : FAILURE in 1h 20m 42s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/074ea5b5118e4a098b4717dcd41eda8b : SUCCESS in 52m 06s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/e146c55abae34ffdab244b0326d61df8 : SUCCESS in 1h 17m 51s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/8d55714b0faa421bab137222e989ce3b : SUCCESS in 32m 58s","accounts_in_message":[],"_revision_number":18},{"id":"48d70282530f3aee5f24979de3457ee5127c5471","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-09-02 16:57:22.000000000","message":"Patch Set 18: Code-Review+2","accounts_in_message":[],"_revision_number":18},{"id":"874326271383c23cea45f4e43a74df2d8fd77171","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-09-02 18:34:22.000000000","message":"Patch Set 18: Code-Review+2\n\n(1 comment)\n\nI think this is good, but probably best to leave it until lance can ack it before we +W","accounts_in_message":[],"_revision_number":18},{"id":"b9c4662c1c96a9596493609c9bdc9040916faaa5","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-09-02 20:19:47.000000000","message":"Patch Set 18: Code-Review+1","accounts_in_message":[],"_revision_number":18},{"id":"1cd318ac7dfc1137e56591db279608b07855c118","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-09-02 20:56:06.000000000","message":"Patch Set 18: Workflow+1","accounts_in_message":[],"_revision_number":18},{"id":"f87c839493b28aa9b248873af2a3172332adb369","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-02 20:57:26.000000000","message":"Patch Set 18: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":18},{"id":"56e8b5dd4f235d2c90c7b0b1b68f2053d081b94b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-02 21:28:10.000000000","message":"Patch Set 18: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":18},{"id":"a9041bffc14b927f6b6e7a920ee42f728393fef8","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-09-02 21:49:30.000000000","message":"Patch Set 18: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"28e555642ba4dc3bf44d9f783ff17764029f77c1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-09-02 22:03:30.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"aec1197168fc911d8843138bd6f769d1545903fb","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-09-02 22:05:29.000000000","message":"Patch Set 19: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":19},{"id":"56b45b4db1c85920b3a168606c76c728bf4b2dcb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-02 23:35:35.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/9ac28cb95e684ae5afa83478e0ad0f8f : SUCCESS in 1h 03m 48s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/44e157944e0843e2909549976a42c55a : SUCCESS in 1h 08m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b16eddfe34a42b58a617b46a11b198c : SUCCESS in 5m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3ec07b4b74624731a6c91979b155f485 : SUCCESS in 6m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b0291f1140af44e18efabc62563b65b0 : SUCCESS in 6m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3f9e9b30622b4bed85ffdd8b08b61d3f : SUCCESS in 6m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e727ccfe6e7e4e75bf4dcd818ef92403 : SUCCESS in 6m 56s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/86273d40f5724189ad3a9a9f8182c0e0 : SUCCESS in 9m 33s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/210095367137456e85d36951e83b13bf : SUCCESS in 9m 15s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/9e9877da93984c5ea38201c4996dee62 : SUCCESS in 11m 18s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/7801a16c6e3848d580b5057118249845 : SUCCESS in 5m 33s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/91ad4b6a99c44bb3bef4fd52e38d4d3f : SUCCESS in 1h 08m 52s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/757301e211f44c4cba2077524fc7c2d4 : SUCCESS in 1h 05m 19s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/c7dc8de8a4794c91805781ea4f02f4ef : SUCCESS in 1h 06m 36s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/ae6e1f82132e4b1d850dc6bf2e095dfb : SUCCESS in 1h 16m 20s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d8d5c7b7153e4525b6c4fc0cb1668fec : SUCCESS in 52m 38s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/4e546ee39214432fb8c0f1e381845b8d : SUCCESS in 1h 12m 08s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/0303b0aaab6040b0b46730372bf7e0bf : SUCCESS in 35m 08s","accounts_in_message":[],"_revision_number":19},{"id":"86c7695c0bbfb9a4a5e87dfeacea423c455f2bc9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-02 23:36:23.000000000","message":"Patch Set 19: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":19},{"id":"46573b208dfc6e264cb5cab64c9d7feb8a3c53a4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-03 02:08:24.000000000","message":"Patch Set 19: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/ab1c7c195dbc4442b9e2bb68d0818522 : SUCCESS in 1h 03m 01s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0318447a81734a748f78d34fae2720b9 : SUCCESS in 1h 15m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dcfb018ad2ff474b9535157e922413b9 : SUCCESS in 5m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ed15245dd3c74f5f964f8a826d18d9fd : SUCCESS in 6m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/314dd68873774b8e8656971c09e97270 : SUCCESS in 6m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8accf068647c47b1b1410c96d71edcfe : SUCCESS in 6m 44s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e127b94b27074d538a4abb0113681b4c : SUCCESS in 9m 25s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/ad31fbc12837466ebce221eb443c41fa : SUCCESS in 9m 02s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/75a6b9b7f502498b98e606b3abc9ff93 : SUCCESS in 1h 32m 43s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/f51db9bc769b4d3f8bdda4a084ea5ad3 : SUCCESS in 1h 13m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a64e5282d663489d851e6661aabd7a84 : SUCCESS in 48m 31s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/e2fc364555054ddbbcd0b21008651b76 : SUCCESS in 1h 07m 03s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/2d662bdc0240487393737fea2294a96e : SUCCESS in 31m 31s","accounts_in_message":[],"_revision_number":19},{"id":"52ea1c77bf6d4519e11ff6830d57642ca8fdc6cd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-03 02:08:24.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":19},{"id":"1526924fceb806b3ffb131b651170058bcd59141","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-03 02:11:15.000000000","message":"Patch Set 19:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58bcb5be294041d1a737df347a0d3080 : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":19}],"current_revision_number":19,"current_revision":"035e714f83effde844425db30423117dced50688","revisions":{"8afbe6a83accf2ad57a2f51d2a420f882ae74a76":{"kind":"REWORK","_number":1,"created":"2021-07-23 16:34:12.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/54/802054/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/54/802054/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/54/802054/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/54/802054/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/54/802054/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/54/802054/1"}}},"commit":{"parents":[{"commit":"6ed73ec57a88a6b8145a809d84748359b9168cea","subject":"Implement project personas for metadef resource-types","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6ed73ec57a88a6b8145a809d84748359b9168cea"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2021-07-23 12:18:07.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2021-07-23 12:18:07.000000000","tz":0},"subject":"Implement project personas for metadef objects","message":"Implement project personas for metadef objects\n\nThis commit updates the policies for metadef objects to\nuse default roles available from keystone.\n\nImplements: blueprint secure-rbac\nChange-Id: If2001ef652e21f54048db0fc67893337681a2b6f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8afbe6a83accf2ad57a2f51d2a420f882ae74a76"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8afbe6a83accf2ad57a2f51d2a420f882ae74a76"}]},"branch":"refs/heads/master"},"7a605b07c95919fe7936c2dad9b608f39248dd33":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-07-23 17:02:22.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/54/802054/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/54/802054/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/54/802054/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/54/802054/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/54/802054/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/54/802054/2"}}},"commit":{"parents":[{"commit":"a8a1916f24327688546727b0f00bc452fa6147ac","subject":"Implement project personas for metadef resource-types","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a8a1916f24327688546727b0f00bc452fa6147ac"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2021-07-23 12:18:07.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2021-07-23 17:02:22.000000000","tz":0},"subject":"Implement project personas for metadef objects","message":"Implement project personas for metadef objects\n\nThis commit updates the policies for metadef objects to\nuse default roles available from keystone.\n\nImplements: blueprint secure-rbac\nChange-Id: If2001ef652e21f54048db0fc67893337681a2b6f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7a605b07c95919fe7936c2dad9b608f39248dd33"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7a605b07c95919fe7936c2dad9b608f39248dd33"}]},"branch":"refs/heads/master","description":"Rebase"},"f2cd2460a1a57986a6a07eecffaf182e36544737":{"kind":"REWORK","_number":3,"created":"2021-07-25 19:36:32.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/54/802054/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/54/802054/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/54/802054/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/54/802054/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/54/802054/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/54/802054/3"}}},"commit":{"parents":[{"commit":"326945bc0a6715f6e5c8af52fd2455de385c1e0d","subject":"Implement project personas for metadef resource-types","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/326945bc0a6715f6e5c8af52fd2455de385c1e0d"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2021-07-23 12:18:07.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2021-07-25 18:16:09.000000000","tz":0},"subject":"Implement project personas for metadef objects","message":"Implement project personas for metadef objects\n\nThis commit updates the policies for metadef objects to\nuse default roles available from keystone.\n\nImplements: blueprint secure-rbac\nChange-Id: If2001ef652e21f54048db0fc67893337681a2b6f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f2cd2460a1a57986a6a07eecffaf182e36544737"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f2cd2460a1a57986a6a07eecffaf182e36544737"}]},"branch":"refs/heads/master"},"fc6604635431382e33491f4f37f31a21fc1c019f":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2021-07-26 10:16:28.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/54/802054/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/54/802054/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/54/802054/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/54/802054/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/54/802054/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/54/802054/4"}}},"commit":{"parents":[{"commit":"12207a9752781f4d8984e50b9f54d72d44801184","subject":"Implement project personas for metadef resource-types","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/12207a9752781f4d8984e50b9f54d72d44801184"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2021-07-23 12:18:07.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2021-07-26 10:16:28.000000000","tz":0},"subject":"Implement project personas for metadef objects","message":"Implement project personas for metadef objects\n\nThis commit updates the policies for metadef objects to\nuse default roles available from keystone.\n\nImplements: blueprint secure-rbac\nChange-Id: If2001ef652e21f54048db0fc67893337681a2b6f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/fc6604635431382e33491f4f37f31a21fc1c019f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/fc6604635431382e33491f4f37f31a21fc1c019f"}]},"branch":"refs/heads/master","description":"Rebase"},"b5ec11842534d6308951cd7dd45707ba4fa33226":{"kind":"REWORK","_number":5,"created":"2021-07-29 11:05:07.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/54/802054/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/54/802054/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/54/802054/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/54/802054/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/54/802054/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/54/802054/5"}}},"commit":{"parents":[{"commit":"5e1c13ed2ed8b25a0bd3d6f6097bbc97afd2be95","subject":"Implement project personas for metadef resource-types","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5e1c13ed2ed8b25a0bd3d6f6097bbc97afd2be95"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2021-07-23 12:18:07.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2021-07-29 09:49:39.000000000","tz":0},"subject":"Implement project personas for metadef objects","message":"Implement project personas for metadef objects\n\nThis commit updates the policies for metadef objects to\nuse default roles available from keystone.\n\nImplements: blueprint secure-rbac\nChange-Id: If2001ef652e21f54048db0fc67893337681a2b6f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b5ec11842534d6308951cd7dd45707ba4fa33226"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b5ec11842534d6308951cd7dd45707ba4fa33226"}]},"branch":"refs/heads/master"},"44e7ddbd1c2d0cad78d76536f8dca8007201809b":{"kind":"REWORK","_number":6,"created":"2021-07-30 11:26:59.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/54/802054/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/54/802054/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/54/802054/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/54/802054/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/54/802054/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/54/802054/6"}}},"commit":{"parents":[{"commit":"716b4d058e4a2fc790ec762e6eca6ae085318c5e","subject":"Implement project personas for metadef resource-types","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/716b4d058e4a2fc790ec762e6eca6ae085318c5e"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2021-07-23 12:18:07.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2021-07-30 10:28:19.000000000","tz":0},"subject":"Implement project personas for metadef objects","message":"Implement project personas for metadef objects\n\nThis commit updates the policies for metadef objects to\nuse default roles available from keystone.\n\nImplements: blueprint secure-rbac\nChange-Id: If2001ef652e21f54048db0fc67893337681a2b6f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/44e7ddbd1c2d0cad78d76536f8dca8007201809b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/44e7ddbd1c2d0cad78d76536f8dca8007201809b"}]},"branch":"refs/heads/master"},"c1861fc57bbd287de2a31d11a80b265a13cd28fc":{"kind":"REWORK","_number":7,"created":"2021-08-03 05:31:34.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/54/802054/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/54/802054/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/54/802054/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/54/802054/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/54/802054/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/54/802054/7"}}},"commit":{"parents":[{"commit":"beacf51401685b42f960f3dc8026568dcabf067b","subject":"Implement project personas for metadef resource-types","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/beacf51401685b42f960f3dc8026568dcabf067b"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2021-07-23 12:18:07.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2021-08-02 19:20:12.000000000","tz":0},"subject":"Implement project personas for metadef objects","message":"Implement project personas for metadef objects\n\nThis commit updates the policies for metadef objects to\nuse default roles available from keystone.\n\nImplements: blueprint secure-rbac\nChange-Id: If2001ef652e21f54048db0fc67893337681a2b6f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c1861fc57bbd287de2a31d11a80b265a13cd28fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c1861fc57bbd287de2a31d11a80b265a13cd28fc"}]},"branch":"refs/heads/master"},"a2fb137f03858c8a2161e7b86b3b076b0208fd44":{"kind":"REWORK","_number":8,"created":"2021-08-03 09:00:47.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/54/802054/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/54/802054/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/54/802054/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/54/802054/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/54/802054/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/54/802054/8"}}},"commit":{"parents":[{"commit":"beacf51401685b42f960f3dc8026568dcabf067b","subject":"Implement project personas for metadef resource-types","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/beacf51401685b42f960f3dc8026568dcabf067b"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2021-07-23 12:18:07.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2021-08-03 06:33:48.000000000","tz":0},"subject":"Implement project personas for metadef objects","message":"Implement project personas for metadef objects\n\nThis commit updates the policies for metadef objects to\nuse default roles available from keystone.\n\nImplements: blueprint secure-rbac\nChange-Id: If2001ef652e21f54048db0fc67893337681a2b6f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a2fb137f03858c8a2161e7b86b3b076b0208fd44"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a2fb137f03858c8a2161e7b86b3b076b0208fd44"}]},"branch":"refs/heads/master"},"6c5999aa2f53b18866dc0f756d3e901eca547459":{"kind":"REWORK","_number":9,"created":"2021-08-03 11:22:18.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/54/802054/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/54/802054/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/54/802054/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/54/802054/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/54/802054/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/54/802054/9"}}},"commit":{"parents":[{"commit":"beacf51401685b42f960f3dc8026568dcabf067b","subject":"Implement project personas for metadef resource-types","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/beacf51401685b42f960f3dc8026568dcabf067b"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2021-07-23 12:18:07.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2021-08-03 10:51:17.000000000","tz":0},"subject":"Implement project personas for metadef objects","message":"Implement project personas for metadef objects\n\nThis commit updates the policies for metadef objects to\nuse default roles available from keystone.\n\nImplements: blueprint secure-rbac\nChange-Id: If2001ef652e21f54048db0fc67893337681a2b6f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6c5999aa2f53b18866dc0f756d3e901eca547459"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6c5999aa2f53b18866dc0f756d3e901eca547459"}]},"branch":"refs/heads/master"},"77916628935ee67fefeea4c1e4a5a60e1cca7277":{"kind":"REWORK","_number":10,"created":"2021-08-16 14:07:12.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/54/802054/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/54/802054/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/54/802054/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/54/802054/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/54/802054/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/54/802054/10"}}},"commit":{"parents":[{"commit":"d53fb2631a6333dca322cd2d211cd14390d92e0a","subject":"Implement project personas for metadef namespaces","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d53fb2631a6333dca322cd2d211cd14390d92e0a"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2021-07-23 12:18:07.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2021-08-16 11:53:57.000000000","tz":0},"subject":"Implement project personas for metadef objects","message":"Implement project personas for metadef objects\n\nThis commit updates the policies for metadef objects to\nuse default roles available from keystone.\n\nImplements: blueprint secure-rbac\nChange-Id: If2001ef652e21f54048db0fc67893337681a2b6f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/77916628935ee67fefeea4c1e4a5a60e1cca7277"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/77916628935ee67fefeea4c1e4a5a60e1cca7277"}]},"branch":"refs/heads/master"},"28f9afc09d6069e29c3eeb8d06658a931a60e910":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2021-08-23 08:53:17.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/54/802054/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/54/802054/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/54/802054/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/54/802054/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/54/802054/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/54/802054/11"}}},"commit":{"parents":[{"commit":"9759e0a7d7555c94b436b194919a7512bfa49ae7","subject":"Implement project personas for metadef namespaces","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9759e0a7d7555c94b436b194919a7512bfa49ae7"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2021-07-23 12:18:07.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2021-08-23 08:53:17.000000000","tz":0},"subject":"Implement project personas for metadef objects","message":"Implement project personas for metadef objects\n\nThis commit updates the policies for metadef objects to\nuse default roles available from keystone.\n\nImplements: blueprint secure-rbac\nChange-Id: If2001ef652e21f54048db0fc67893337681a2b6f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/28f9afc09d6069e29c3eeb8d06658a931a60e910"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/28f9afc09d6069e29c3eeb8d06658a931a60e910"}]},"branch":"refs/heads/master","description":"Rebase"},"5b91830b3037e133273ba341014b1c613f125cfe":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2021-08-26 04:02:48.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/54/802054/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/54/802054/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/54/802054/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/54/802054/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/54/802054/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/54/802054/12"}}},"commit":{"parents":[{"commit":"81b7942736f52b1b70746dd8aa3de1a59b8ddfed","subject":"Implement project personas for metadef namespaces","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/81b7942736f52b1b70746dd8aa3de1a59b8ddfed"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2021-07-23 12:18:07.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-08-26 03:56:33.000000000","tz":0},"subject":"Implement project personas for metadef objects","message":"Implement project personas for metadef objects\n\nThis commit updates the policies for metadef objects to\nuse default roles available from keystone.\n\nImplements: blueprint secure-rbac\nChange-Id: If2001ef652e21f54048db0fc67893337681a2b6f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5b91830b3037e133273ba341014b1c613f125cfe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5b91830b3037e133273ba341014b1c613f125cfe"}]},"branch":"refs/heads/master"},"1e7b14ad0a705c8167406924d25dd6400b7e9abb":{"kind":"REWORK","_number":13,"created":"2021-08-30 19:47:26.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/54/802054/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/54/802054/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/54/802054/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/54/802054/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/54/802054/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/54/802054/13"}}},"commit":{"parents":[{"commit":"70f0ddcf0b327d18097334c6dc827f005a5892c7","subject":"Implement project personas for metadef namespaces","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/70f0ddcf0b327d18097334c6dc827f005a5892c7"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2021-07-23 12:18:07.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2021-08-30 19:06:00.000000000","tz":0},"subject":"Implement project personas for metadef objects","message":"Implement project personas for metadef objects\n\nThis commit updates the policies for metadef objects to\nuse default roles available from keystone.\n\nImplements: blueprint secure-rbac\nChange-Id: If2001ef652e21f54048db0fc67893337681a2b6f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1e7b14ad0a705c8167406924d25dd6400b7e9abb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1e7b14ad0a705c8167406924d25dd6400b7e9abb"}]},"branch":"refs/heads/master"},"913382cc7330d534a89eb5252ba39f5030d59ce8":{"kind":"REWORK","_number":14,"created":"2021-08-31 06:08:04.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/54/802054/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/54/802054/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/54/802054/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/54/802054/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/54/802054/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/54/802054/14"}}},"commit":{"parents":[{"commit":"1321baaaea3c50a201ddd78c39d6f9cb8720be24","subject":"Implement project personas for metadef namespaces","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1321baaaea3c50a201ddd78c39d6f9cb8720be24"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2021-07-23 12:18:07.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2021-08-31 05:35:25.000000000","tz":0},"subject":"Implement project personas for metadef objects","message":"Implement project personas for metadef objects\n\nThis commit updates the policies for metadef objects to\nuse default roles available from keystone.\n\nImplements: blueprint secure-rbac\nChange-Id: If2001ef652e21f54048db0fc67893337681a2b6f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/913382cc7330d534a89eb5252ba39f5030d59ce8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/913382cc7330d534a89eb5252ba39f5030d59ce8"}]},"branch":"refs/heads/master"},"784d1b67ccfdd6ce6f6c2bda50750983d451980f":{"kind":"NO_CHANGE","_number":15,"created":"2021-08-31 20:04:52.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/54/802054/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/54/802054/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/54/802054/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/54/802054/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/54/802054/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/54/802054/15"}}},"commit":{"parents":[{"commit":"3f2bf1254cc49727f252fa40bf31d01abcfe7fa8","subject":"Implement project personas for metadef namespaces","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3f2bf1254cc49727f252fa40bf31d01abcfe7fa8"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2021-07-23 12:18:07.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2021-08-31 19:34:41.000000000","tz":0},"subject":"Implement project personas for metadef objects","message":"Implement project personas for metadef objects\n\nThis commit updates the policies for metadef objects to\nuse default roles available from keystone.\n\nImplements: blueprint secure-rbac\nChange-Id: If2001ef652e21f54048db0fc67893337681a2b6f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/784d1b67ccfdd6ce6f6c2bda50750983d451980f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/784d1b67ccfdd6ce6f6c2bda50750983d451980f"}]},"branch":"refs/heads/master"},"1f22a07ecabe821a06276e648f76ac318b575d0d":{"kind":"TRIVIAL_REBASE","_number":16,"created":"2021-08-31 20:20:51.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/54/802054/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/54/802054/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/54/802054/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/54/802054/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/54/802054/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/54/802054/16"}}},"commit":{"parents":[{"commit":"ab0558960cf4d6562ed60cc91a3ebee2692d4fd5","subject":"Implement project personas for metadef namespaces","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ab0558960cf4d6562ed60cc91a3ebee2692d4fd5"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2021-07-23 12:18:07.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2021-08-31 20:03:20.000000000","tz":0},"subject":"Implement project personas for metadef objects","message":"Implement project personas for metadef objects\n\nThis commit updates the policies for metadef objects to\nuse default roles available from keystone.\n\nImplements: blueprint secure-rbac\nChange-Id: If2001ef652e21f54048db0fc67893337681a2b6f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1f22a07ecabe821a06276e648f76ac318b575d0d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1f22a07ecabe821a06276e648f76ac318b575d0d"}]},"branch":"refs/heads/master"},"6c5c0c8dc90d16be2bb590a5e9d40e54fb48e0e2":{"kind":"REWORK","_number":17,"created":"2021-09-02 09:20:55.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/54/802054/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/54/802054/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/54/802054/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/54/802054/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/54/802054/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/54/802054/17"}}},"commit":{"parents":[{"commit":"80cfad4a677434638fadd65fcab808db70954a50","subject":"Implement project personas for metadef namespaces","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/80cfad4a677434638fadd65fcab808db70954a50"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2021-07-23 12:18:07.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-09-02 07:28:42.000000000","tz":0},"subject":"Implement project personas for metadef objects","message":"Implement project personas for metadef objects\n\nThis commit updates the policies for metadef objects to\nuse default roles available from keystone.\n\nImplements: blueprint secure-rbac\nChange-Id: If2001ef652e21f54048db0fc67893337681a2b6f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6c5c0c8dc90d16be2bb590a5e9d40e54fb48e0e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6c5c0c8dc90d16be2bb590a5e9d40e54fb48e0e2"}]},"branch":"refs/heads/master"},"ea3ef052fdb4d3cc6d53a9ed02e4575e743993bb":{"kind":"REWORK","_number":18,"created":"2021-09-02 12:18:28.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/54/802054/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/54/802054/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/54/802054/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/54/802054/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/54/802054/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/54/802054/18"}}},"commit":{"parents":[{"commit":"521e8b23618634e0e167f44589bde28214fc82a8","subject":"Implement project personas for metadef namespaces","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/521e8b23618634e0e167f44589bde28214fc82a8"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2021-07-23 12:18:07.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2021-09-02 10:52:45.000000000","tz":0},"subject":"Implement project personas for metadef objects","message":"Implement project personas for metadef objects\n\nThis commit updates the policies for metadef objects to\nuse default roles available from keystone.\n\nImplements: blueprint secure-rbac\nChange-Id: If2001ef652e21f54048db0fc67893337681a2b6f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ea3ef052fdb4d3cc6d53a9ed02e4575e743993bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ea3ef052fdb4d3cc6d53a9ed02e4575e743993bb"}]},"branch":"refs/heads/master"},"035e714f83effde844425db30423117dced50688":{"kind":"REWORK","_number":19,"created":"2021-09-02 22:03:30.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/54/802054/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/54/802054/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/54/802054/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/54/802054/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/54/802054/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/54/802054/19"}}},"commit":{"parents":[{"commit":"9d9d7abe215423d48b50dfc2691567bc56fa3dda","subject":"Implement project personas for metadef namespaces","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9d9d7abe215423d48b50dfc2691567bc56fa3dda"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2021-07-23 12:18:07.000000000","tz":0},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-09-02 23:05:18.000000000","tz":-420},"subject":"Implement project personas for metadef objects","message":"Implement project personas for metadef objects\n\nThis commit updates the policies for metadef objects to\nuse default roles available from keystone.\n\nImplements: blueprint secure-rbac\nChange-Id: If2001ef652e21f54048db0fc67893337681a2b6f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/035e714f83effde844425db30423117dced50688"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/035e714f83effde844425db30423117dced50688"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
