)]}'
{"id":"openstack%2Fglance~802055","triplet_id":"openstack%2Fglance~master~Ic93c2a8a7017e6eac16d23fb5a9224d2c11e7211","project":"openstack/glance","branch":"master","topic":"secure-rbac","hashtags":[],"change_id":"Ic93c2a8a7017e6eac16d23fb5a9224d2c11e7211","subject":"Implement project personas for metadef properties","status":"MERGED","created":"2021-07-23 16:34:12.000000000","updated":"2021-09-03 06:34:12.000000000","submitted":"2021-09-03 06:32:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":40,"unresolved_comment_count":13,"has_review_started":true,"submission_id":"802055-secure-rbac","meta_rev_id":"3eed2d0f9e050abef29d3f383b6943a35fd21f8f","_number":802055,"virtual_id_number":802055,"owner":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-03 06:32:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2021-09-02 22:05:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":1,"date":"2021-09-02 22:05:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-23 17:44:29.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2021-07-23 18:24:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-24 04:46:02.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-08-04 07:02:26.000000000","updated_by":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"reviewer":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"state":"REVIEWER"},{"updated":"2021-08-04 07:02:26.000000000","updated_by":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2021-08-04 07:02:26.000000000","updated_by":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"}],"messages":[{"id":"6ab1cf0d550aefcec0c6f8e24e44b85cbeb9d3b7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2021-07-23 16:34:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"33312b1d1b5b83d33d4b99f0e4a44a8a9236e7cc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2021-07-23 17:02:33.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"939f832ddc5f02b1bc0a3aa1abf87272e99002bf","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-07-23 17:44:29.000000000","message":"Patch Set 2: Code-Review+1\n\n(3 comments)\n\nFew suggestions inline!","accounts_in_message":[],"_revision_number":2},{"id":"310d175ba482955b15173e597a4d3245d3e3668a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-23 18:24:52.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/76bab2be789f476d88ba430ae3774ac3 : SUCCESS in 1h 12m 22s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d323fd20cbfc46c4805997a24b250d2c : SUCCESS in 1h 09m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab7b702c5bd646d7829fd9ea337c398b : SUCCESS in 5m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/83db8bba08f44b61a407c1e7d86a4d33 : SUCCESS in 6m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fef040f3b4794c1692b25505b2a54a1a : SUCCESS in 6m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/aff69a003b074d05a7905c13061de90e : SUCCESS in 6m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5ccf1aa82c34f4d83588cdde7e32ac8 : SUCCESS in 7m 13s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ce52ec43a3094bd8a53167e467067485 : FAILURE in 8m 57s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/17b8c828dda24a3d98323c3a4b4d40c2 : SUCCESS in 9m 02s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/10ee8ab4688a43bbac4fe9baf49c98bf : SUCCESS in 9m 38s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/bf41dae00e7b435ab0706a3dba4e05ec : SUCCESS in 5m 47s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/57a97667cfeb468f88294a748f24e2e1 : SUCCESS in 1h 14m 45s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/3bfc85b55d1a4f62b929bf872bf164f3 : SUCCESS in 1h 10m 36s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/58f7a104652646f7928c5778d68e06d1 : SUCCESS in 1h 11m 43s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/0308f61db2d74a2189ca085b0521543b : SUCCESS in 1h 11m 08s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dfe04f0f14b442419311b07dba757000 : SUCCESS in 49m 55s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/1081774bfeb340dc8540a0f7eac26a2e : SUCCESS in 1h 09m 23s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/bb78acab5071458da68b19703fbef43b : SUCCESS in 37m 35s","accounts_in_message":[],"_revision_number":2},{"id":"6ea777b759b3bde6d39e24df8a6ae49236cdbbcf","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-07-24 04:46:02.000000000","message":"Patch Set 2:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/802055/802055/2/check-not-vote/ext-glance-zuul/0b078af : SUCCESS in 35m 12s","accounts_in_message":[],"_revision_number":2},{"id":"3f2ee81e5daca0e3af94f6a20d22b20ef1636554","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2021-07-25 19:36:32.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"455547e2637ad605f18edcebf2ecacd252a62972","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-25 21:14:22.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/31e0078ac06c4eecb81c3b1ee32dad52 : SUCCESS in 1h 07m 15s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/eec7f55622044eb78afc525aa4a289c2 : SUCCESS in 58m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/16b6e753a5ae409085d1890a5528fa08 : SUCCESS in 5m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a749fce654ea45f1bae9156b3e97c1c3 : SUCCESS in 13m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9039c241327a4d13973699f4f45593a7 : SUCCESS in 6m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/40e18d13a34f477fbce631528ad52177 : SUCCESS in 6m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35947f07f61f4457ba58c6b6ee4d3c60 : SUCCESS in 6m 12s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6105d87521334a7698bd9cc04740d8fd : SUCCESS in 9m 22s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/858d79d40ac246bea7af6933fd579ac2 : SUCCESS in 8m 16s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/3602e18080b54f00bbd5f841d4882796 : SUCCESS in 9m 07s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/10c8133d2aa24ede8673a41f8560b9a0 : SUCCESS in 6m 16s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/43ab882dd19b4bb8afcb7435c9f8b078 : SUCCESS in 1h 31m 12s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/e6b7fe18b0e94d4cb313eacfca375fd7 : SUCCESS in 1h 00m 37s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/9e206446e3b3485698e623e23f51aa97 : SUCCESS in 1h 18m 00s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/46f2eb182ef94e56bffbae0bfb6116eb : SUCCESS in 1h 30m 20s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/317f41aa7e0546bb9d8134fc188917cf : SUCCESS in 50m 43s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/d4313384769b41c78e7bb0f3751b9549 : SUCCESS in 1h 06m 22s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/79d67cfadef5431ab0a785d082323235 : SUCCESS in 30m 06s","accounts_in_message":[],"_revision_number":3},{"id":"e051630f3da5ae89f724141fbf9a0f323c18ed61","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2021-07-26 10:16:39.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"aca9debed9f39f99e669a5e9041e69cd4dfe7526","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-26 11:32:55.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/208ca05ff1f146a1b1781c6aeaf736b2 : SUCCESS in 1h 10m 48s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/95b5314955bb4c918bcf0341b2c6b216 : SUCCESS in 1h 13m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/153efa469d8141938dfd212d5ab5f061 : SUCCESS in 5m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/75cc0501e9f34e7693c5682018f1c406 : SUCCESS in 7m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/71142bd200584691a024bf392a9eeeec : SUCCESS in 7m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4e6322ddf40d486c99f1bb86eb8f6ae2 : SUCCESS in 7m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4affa9542f7b48988a05c25ef8a38a81 : SUCCESS in 6m 38s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6b03a9cb275f4a409492f69348ad6211 : SUCCESS in 8m 59s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/f64683a8cfa2424eb9db7d7ddc7fb82b : SUCCESS in 11m 04s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/052f45dd653d4d8492fee9bdd32dbcec : SUCCESS in 11m 15s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/ca0e722bdf0549e68e592c1486051904 : SUCCESS in 5m 42s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/1b87340de94c4fa88b45b9a380cf6ee7 : SUCCESS in 1h 12m 01s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/44a0df030e5240f5a02b47fd3240c49b : SUCCESS in 1h 10m 01s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/e776ff8ca35e481da261bce5df64ba52 : SUCCESS in 1h 10m 51s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/205b11b63313455a874df41aef380816 : SUCCESS in 1h 14m 39s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4ac66e02875848549ab051d927dea1b0 : SUCCESS in 49m 35s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/6b09c12d1b2e47f58aa0a06336d14804 : POST_FAILURE in 1h 08m 40s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/b738af842aa64c8d9bc3bcaafbb1d7d1 : SUCCESS in 31m 22s","accounts_in_message":[],"_revision_number":4},{"id":"d9b7a5719f9418b87735a77c65ab2ac089c348f5","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-07-26 12:28:36.000000000","message":"Patch Set 4:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/802055/802055/4/check-not-vote/ext-glance-zuul/d145570 : SUCCESS in 32m 58s","accounts_in_message":[],"_revision_number":4},{"id":"babc933902903a587da120eb2be0cf03b890e533","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-07-27 08:25:15.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"a63d3268198ce622e216ee1c9e9ffa361db62d01","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-07-27 08:32:58.000000000","message":"Patch Set 4: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"8cc24a7e7f01a717b834976caf5c77c10c89bc7b","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-07-27 09:03:51.000000000","message":"Patch Set 4:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/802055/802055/4/check-not-vote/ext-glance-zuul/b207ae0 : SUCCESS in 37m 46s","accounts_in_message":[],"_revision_number":4},{"id":"0f41f42508f003bf014696edf25bac5501d1ce6d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-27 09:51:02.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/ab091433d7cd4ae3abc1de5c50c1c51d : SUCCESS in 1h 00m 58s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ce04760c41d54d549cfdfa10a1ab577b : SUCCESS in 1h 06m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f4e3a9f3ffd4f2492af5352d3e55893 : SUCCESS in 5m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5a05485bac0b48349a0f69a551168873 : SUCCESS in 7m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/76b5320b26624b98aca52c986b0c4668 : SUCCESS in 6m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/804f3773f7c74b44a7277e564041dcf4 : SUCCESS in 6m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/920894c9143c41d9b8dc00716849104f : SUCCESS in 8m 15s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/377777ec410c4e289b8dbaf9d77cc6bb : SUCCESS in 10m 43s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/1f8ef2b7d24644ddb53046795b28a378 : SUCCESS in 11m 48s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/7335367ddc59481ca5436a2aa92f6014 : SUCCESS in 9m 26s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/441dc90578004451847924bf2984c4ee : SUCCESS in 6m 15s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/a6d8c12fbfa64338b115fff5e96ac9d7 : SUCCESS in 1h 13m 06s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/4c1ba2af69604108a015957facd55a66 : SUCCESS in 1h 13m 37s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/b2bfa471f2fe45c987539f9c837ac08e : SUCCESS in 1h 16m 35s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/8ca3a0bb2cab44e18e0afc5703e0a67b : SUCCESS in 1h 08m 57s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/78138b1a838b4fed9ab3ab7373c41233 : SUCCESS in 49m 24s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/ce559e6aceec4ffaa8c252ed44ca765a : SUCCESS in 1h 10m 21s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/1de91a56b83b493f8c278b2ab59627cb : SUCCESS in 32m 55s","accounts_in_message":[],"_revision_number":4},{"id":"9ab4cf8ac9e44f32197891a53f26147749870578","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2021-07-29 11:05:07.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"cfc6e64f928a00f090b1880d4dc00ab8597ce0a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-29 12:31:50.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/41e48bdee91647db8211e55c00a7768d : SUCCESS in 1h 05m 33s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/617dd02b44f7441890b08dd55988994e : SUCCESS in 1h 11m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/51de3e8b9df24bc28ef84036962f3408 : SUCCESS in 5m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/75e4292f8c114d2d94476db86512a5db : SUCCESS in 6m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8e73015d0f1f47df9198e2d5c4bc732b : SUCCESS in 6m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a67130e9230a430b96d892ab326ec3e5 : SUCCESS in 7m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df43c57be31a4aff94f0596feba2eb28 : SUCCESS in 7m 48s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/58f5666449cb41bea16ebba38036b91f : SUCCESS in 8m 33s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c35452d539074309baf1b00cdd8c43fe : SUCCESS in 8m 57s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/5d4d21ffaca4481fa3d7a02f64db0bc1 : SUCCESS in 8m 39s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/841f75884bcf437a984a288377f24b20 : SUCCESS in 5m 46s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/ff22ff081a5d4a70823c23cd787ebb1f : FAILURE in 1h 10m 16s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/b0e2770d24a14ab283fcad6869d6222e : SUCCESS in 1h 09m 23s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/3c33784741ac4b019493ddb281799661 : SUCCESS in 1h 13m 41s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/93985db3fa534959b05875906b874226 : SUCCESS in 1h 16m 38s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f66965e9420b4bcbbc1347758d177f05 : SUCCESS in 57m 57s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/1e95503fa5bd478a911b698a6cc1ff20 : SUCCESS in 1h 20m 58s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/6fc47282d80540d7958c38bbf1e67612 : SUCCESS in 32m 36s","accounts_in_message":[],"_revision_number":5},{"id":"54a61973aa029eb74c6865d2aa9e22c1bf0d7e34","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2021-07-29 12:34:30.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"96dd523e0b376f26beded7d8d442c5c81e71938a","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-07-29 16:23:47.000000000","message":"Patch Set 5: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"cffef782cc63f1698a702cfea9fa996913c601ad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2021-07-30 11:26:59.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"86c061536c6d32e7674d7f2dceda51b48f56bf7b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-30 12:59:28.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/f44db2523346404688d3e1be129e3d02 : SUCCESS in 59m 40s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/947857b29ce3405caef68a3465823ffd : SUCCESS in 1h 07m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5367e5c78724f1dbf434a94a39d2bf0 : SUCCESS in 5m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ee63ea56ebbf4813a2198f109202e72e : SUCCESS in 6m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/77d416c5cb3f4b5da6558aea449f0adb : SUCCESS in 6m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3e30d551900c4667acac5cc07ed4b7a5 : SUCCESS in 7m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/829df63c07624e97be68e03a1f9eb9d7 : SUCCESS in 7m 08s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/cf83fde96eb7451d9b1d0c10f8865e28 : SUCCESS in 7m 50s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/264ffc5cdab441b0a31aa56c0fead7c5 : SUCCESS in 8m 58s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/fad7f7b5ceb74dca97b8787e458e5b8d : FAILURE in 10m 10s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/794eb56949bf48a3b301f35002f9ad3b : SUCCESS in 6m 09s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/144f6ecbe130448aa14371a7b5412a0d : SUCCESS in 1h 15m 12s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/9c1670b41b5548428202947731dc4b14 : SUCCESS in 1h 04m 26s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/e9071b37968b493ba3a46b9d51694a20 : SUCCESS in 1h 09m 13s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/c9aa9be5d8264f6192c2a0afef46097a : SUCCESS in 1h 05m 12s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8320c91f2b4843fcaf8679450e64b646 : SUCCESS in 45m 34s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/be2be89eb90747f180912c5c17bd0ad7 : FAILURE in 1h 31m 31s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/302f3fe0d1a345e3affea2a23568ab51 : SUCCESS in 33m 59s","accounts_in_message":[],"_revision_number":6},{"id":"c6b13e251544b73af10e5586bd9b962211ee2513","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2021-08-03 05:31:34.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"2202b8e5e2f9eefe00c60ad152ee368bbf5ac216","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-03 05:57:41.000000000","message":"Patch Set 7: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":7},{"id":"33083b371209c3b3ddcf4778002f0f71051fc82c","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-08-03 06:11:58.000000000","message":"Patch Set 7:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/802055/802055/7/check-not-vote/ext-glance-zuul/b9f41ad : SUCCESS in 38m 43s","accounts_in_message":[],"_revision_number":7},{"id":"7d6712c4c4fa317c3a50376709d63755744cb662","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-03 06:59:38.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/eacabfe017a840c39d4874dacc0773bc : FAILURE in 33m 48s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/42c9a7f7073a430e94c52e8c186f3728 : SUCCESS in 1h 03m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c4d2ab4471a4cb39c421e13b4c85f5c : SUCCESS in 6m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/baa61539dcc64cd087a7c7980b66e597 : SUCCESS in 7m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/55825a996ac14e67bf3b5b0c6ec9f013 : SUCCESS in 6m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d8c45f6dd8494be2b67fb39d437445ae : SUCCESS in 8m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57c1ad77940a4a0a80de6cf18a3e143e : SUCCESS in 7m 37s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b0b465d0cfe34b4e9a64cdd76da8d7e4 : TIMED_OUT in 31m 16s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/4f027510caa541f59d26086cbdf41e5f : SUCCESS in 9m 09s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/ca96f6241d0145128a1f346df18045c8 : SUCCESS in 10m 53s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/39ec5b99c40b42bfa2618d03004329a4 : SUCCESS in 5m 27s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/8f5ab152111948eb8ebb2f0fda8ee428 : SUCCESS in 1h 21m 04s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/50538a780b83434da5efd53ffbf7dad6 : SUCCESS in 1h 20m 51s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/81f400ce24764c76a74fff04b0aba10f : SUCCESS in 1h 05m 09s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/872f591fc6da4dfba2009ddbf8186845 : SUCCESS in 1h 19m 16s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1494b903cd6f45b2ae7b2491f1f63d19 : SUCCESS in 54m 35s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/031c18072ee447619cf9c11b8b23e81f : SUCCESS in 1h 07m 53s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/3c96231025774ff0aef04cf43db44e2c : SUCCESS in 37m 36s","accounts_in_message":[],"_revision_number":7},{"id":"41a5c6735ae7a59dcac4f06e59bc62dc0df1f3b0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2021-08-03 09:00:47.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"6c7735415c66e97d835e353cbebfc4fbad0af27a","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2021-08-03 09:02:52.000000000","message":"Patch Set 8:\n\n(5 comments)","accounts_in_message":[],"_revision_number":8},{"id":"83e4d7d3d1ccce1bb08e0475ac9c7fdbe361c621","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-03 09:25:30.000000000","message":"Patch Set 8: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":8},{"id":"693d28145a0ad28c728e320bab0fa6f10063f855","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-03 10:26:00.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/13a90281db354794ab1f325ec77c76ea : FAILURE in 40m 40s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8a871ed018464789af0ef81206392281 : SUCCESS in 1h 11m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0658e9bed7974030bee6ee504250b46d : SUCCESS in 5m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/67b89f9ffb1b4865a4c0cf6c26ac28d3 : SUCCESS in 6m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d7075407533c44a59f0749ac71c6a25e : SUCCESS in 6m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e7f807343065484e8dd1e301c125f634 : SUCCESS in 6m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ee2dbbb87534817bd2606a668db3113 : SUCCESS in 6m 47s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/de4211e00e384644b28731f68b97d9c9 : TIMED_OUT in 31m 11s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/40712df7e8b94eb7b9fd6e4052bdaf6f : SUCCESS in 11m 47s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/cc9d66dcfa4d41fe9a5b4f0b34ab1484 : SUCCESS in 8m 59s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/502c23ca7245494d9a7faf5b9c5622bf : SUCCESS in 9m 11s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/3681c3ac9e6c4d59bdea5d101de3adac : SUCCESS in 1h 12m 26s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/24cbe00f7e5542efb6289816746a083b : SUCCESS in 1h 21m 05s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/d40a7cd23dc246499347a43f69e229a7 : SUCCESS in 1h 20m 29s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/adb10516714d4543900aebcae051b5f1 : FAILURE in 1h 20m 17s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b730db40895c4c35b0e3d3ecfbe63e63 : SUCCESS in 48m 12s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/e881ad7c5ad54c9a9a2fb0f6b1547076 : SUCCESS in 1h 11m 28s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/2ed43b94473b4d81be31b3ee543d4a13 : SUCCESS in 36m 38s","accounts_in_message":[],"_revision_number":8},{"id":"324c9f20b05424b737771b1fc713f1e1b5c89920","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2021-08-03 11:22:18.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"4d157a72cf68fae1ca0b044a1a8578eceea31b66","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2021-08-03 11:24:14.000000000","message":"Patch Set 9:\n\n(3 comments)","accounts_in_message":[],"_revision_number":9},{"id":"28635f854c58e09e5f6bbcb5aec181437dfaa93a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-03 12:37:41.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/8187486213a940fda9878eafdbe91902 : FAILURE in 47m 39s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1f0e26e6cdcd45d48fe96c0b46fdd442 : SUCCESS in 1h 07m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6cd61cbd9a494ef38d1b8b8dbabf9fdb : SUCCESS in 5m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bf2432a905e545c7aa93a4a11131e878 : SUCCESS in 6m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bf92b1d3bdcc4aa4b853d7413e607764 : SUCCESS in 6m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f88812d4c62d4d4e890115a84d6e22b1 : SUCCESS in 6m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1798222c43f40ff9b18b7ff6812f76a : SUCCESS in 7m 07s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/024bd21708b849aa8d1166e5996bc4ae : SUCCESS in 9m 23s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/06448ab99b5f4f528e352fb5cb07a9e3 : SUCCESS in 13m 24s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/a1fab59e4b5e460083bf4f4171a93634 : SUCCESS in 12m 26s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/c04bb5676d2243529a8e014edade5afd : SUCCESS in 5m 56s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/d80246966041484ea978428fc30f024b : SUCCESS in 1h 10m 02s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/d14a49b02033451fab5784a333ddc94a : SUCCESS in 1h 10m 04s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/e483084453d74b89accafd5131289fa7 : SUCCESS in 1h 07m 59s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/5373a347b963499eaf6c99b245024370 : FAILURE in 1h 11m 54s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f214b8671994494382dbcb734a84844b : SUCCESS in 46m 24s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/d92e332d46b448d7b584f3fd4daeb867 : SUCCESS in 1h 07m 17s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/b6116b92f3314a79998aeff88073fa1c : SUCCESS in 34m 02s","accounts_in_message":[],"_revision_number":9},{"id":"41420031779459f959453798c9db4b07e36e9a0e","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2021-08-04 05:24:35.000000000","message":"Topic secure-rbac-tags removed","accounts_in_message":[],"_revision_number":9},{"id":"02a31163db6bdb8f0e327293d75f9a03c4359c5e","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2021-08-04 05:24:48.000000000","message":"Topic set to secure-rbac","accounts_in_message":[],"_revision_number":9},{"id":"28a4b282a1b8f720ddb7513cb21db6931cdd5dee","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2021-08-04 05:25:38.000000000","message":"Patch Set 9:\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"0b71e3e440ec3f67d169d6a91f9cbcc113fe15e7","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-08-04 06:40:03.000000000","message":"Patch Set 9:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":9},{"id":"2a535c20267fc5202d2c57e83394814305a73182","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-04 06:47:10.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/a830be395b3b41e1a0086f9ae563df88 : SUCCESS in 1h 04m 11s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/05fecf7a82a440a196cb4d7b3dcbefe5 : SUCCESS in 1h 12m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b01fb1678f2e41478cc5b94864b7fbe6 : SUCCESS in 5m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0ecb15a660534db8823d649766ce6fdf : SUCCESS in 6m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/11f92756773f409f882783e489b8c939 : SUCCESS in 6m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a52ca21261c649b0b50d73040a2e3205 : SUCCESS in 7m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/309dc6be962045f08d361e560e0914dc : SUCCESS in 6m 34s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3efc85cc1df5423883e34ea005591659 : SUCCESS in 9m 25s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/9b3f68dd1fc843cab294db023e04c89b : SUCCESS in 9m 28s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/4a922460313e4a589e3d27c9233f0eaa : SUCCESS in 10m 18s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/f13e99992b0248428a1088079c449ff3 : SUCCESS in 5m 14s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/69aa38c8b83e4711b334c9162e5c63d5 : SUCCESS in 1h 19m 58s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/c63bc5f136d34f0a9943aff2a13cf4ed : SUCCESS in 1h 10m 45s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/7d2c93ddd5be4136999a1edbfcddcf24 : SUCCESS in 1h 06m 35s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/5fbd247ed98144f0bb2906d55bcf55ed : SUCCESS in 1h 06m 07s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/622f2a8f6eac416eb6fced4dd17a4158 : SUCCESS in 55m 32s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/08602866ce42445ab8aba214a0c329a8 : SUCCESS in 1h 16m 41s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/f93fa89ea2324151a04019cddd7cc0d2 : SUCCESS in 36m 10s","accounts_in_message":[],"_revision_number":9},{"id":"5a0e03e519adf16482e799410384dad79459a7c7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2021-08-16 14:07:12.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"7b0adc5b6f3618525d1a7d447b931587eba1ece1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-16 15:32:57.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/4bf1a114652e432eadab190a89426daa : SUCCESS in 59m 51s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ee9437ec4d0341ff83f47a7dab8e596e : SUCCESS in 1h 12m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca076960dffa462cbe6b312ee858aaa1 : SUCCESS in 5m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/848f74d895954f48adf05e3cf11611b0 : SUCCESS in 6m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0974f1e6e5304f4ebb5cd45e6c5f8594 : SUCCESS in 7m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5e8cb71559a2452dbfb2523d09361815 : SUCCESS in 7m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ba57ba5a71847e1b3f395721f81b030 : SUCCESS in 7m 34s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e2395f7dae3243b08c73fc44c8133e44 : SUCCESS in 8m 52s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/32023d73a7f7489e814aaa27a59a4f6a : SUCCESS in 10m 07s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/c61554c52dca4825a5ac641a9f4d9457 : SUCCESS in 9m 09s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/889cd32acc634cd8b6433a1d00bbbf5e : SUCCESS in 5m 25s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/f87b759a8b7848b1a820dcfd45da8ef0 : SUCCESS in 1h 07m 37s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/93fd231c702c4b59a2c3a52c811eb907 : SUCCESS in 1h 07m 37s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/43e76e437ca44f268a6a3b646ab9e7de : SUCCESS in 1h 06m 23s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/f567e873d685477fb3842993f0abc87f : SUCCESS in 1h 17m 27s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9ba48775ff5846db85621bbc5db43483 : SUCCESS in 51m 12s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/8f8e0b1dfec94da98c48d054180035bf : SUCCESS in 1h 07m 48s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/da599546888a443e9b05af34edf82997 : SUCCESS in 34m 07s","accounts_in_message":[],"_revision_number":10},{"id":"0779a0578e75ce05af300544bac05500be05bd31","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-08-18 20:40:52.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"5f647089332b0f7c23939fddaaca6e79b1bf62ad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2021-08-23 08:53:40.000000000","message":"Patch Set 11: Patch Set 10 was rebased","accounts_in_message":[],"_revision_number":11},{"id":"4fc3368c78d4f61a7f1dba4005a0ddac0e13b3c7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-23 10:13:42.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/948dec96e6e14772adc7579da020f0cb : SUCCESS in 1h 02m 33s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f391cfa40fea400c86baa28bbd8045cc : SUCCESS in 1h 06m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/59867ed77c744296b77d1159bb6b6770 : SUCCESS in 6m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/68e3b7a278db455b84370a31ec5f9766 : SUCCESS in 7m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f531c6c469dc44f6bfd2dddffa41b023 : SUCCESS in 6m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8ef55cce3e664fb9aa37c49b339d1a96 : SUCCESS in 7m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82b1082498044322bbbebe355c13d4ea : SUCCESS in 7m 39s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/cebe1b64e14c4394aab6145b24fb9c08 : SUCCESS in 8m 46s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/d37a52fbe9944a11b965f8800fc1e097 : SUCCESS in 8m 51s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/589c6304edf74bfb99ebf745b27cb762 : SUCCESS in 11m 42s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/1335c190071f4859a9f50371ad5b78ca : SUCCESS in 5m 16s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/d029a189012a43fc8aa22c5a109a175f : SUCCESS in 1h 09m 21s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/32a35128066645b7af12a09208291c22 : SUCCESS in 1h 16m 58s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/50ba6dcb2e2047499aaf4181dd89588b : SUCCESS in 1h 08m 54s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/c6197eaefebf4b3faed85f469409bec7 : SUCCESS in 1h 08m 34s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/528cf36ca33d4f12991fc130f916049f : SUCCESS in 50m 14s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/eb9ceb28099f44319f3861848dad742c : SUCCESS in 1h 17m 56s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/e77b2175de0d43a6862fe4b65dab8345 : SUCCESS in 33m 55s","accounts_in_message":[],"_revision_number":11},{"id":"9a0a01d3cd19aaa33a85d6d171116269688dee4e","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-08-23 12:45:04.000000000","message":"Patch Set 11:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/802055/802055/11/check-not-vote/ext-glance-zuul/e8eb61f : SUCCESS in 36m 29s","accounts_in_message":[],"_revision_number":11},{"id":"bad01a7527c55dbcc9e7cc2bf317a3221559a4c0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-26 04:02:48.000000000","message":"Uploaded patch set 12: Patch Set 11 was rebased.","accounts_in_message":[],"_revision_number":12},{"id":"529f4055dfe3e62dedd6dce23af16fc23bf8839a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-26 05:26:56.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/88ce711cfe56403e8cd37d0a94ba5dac : SUCCESS in 1h 11m 30s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a1ebc6b3e5e94eeba5dfc89f8c1c0f8b : SUCCESS in 1h 17m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/26bdd3b039e941a499576affbec3c016 : SUCCESS in 5m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f4f84b41a31540c3bfbf6683820ab650 : SUCCESS in 6m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/412e852d14d44b7aa140958f9553ad58 : SUCCESS in 6m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2937fc56bdb54ec39f24f6a5d3b30581 : SUCCESS in 6m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa8d1797c11e49acb8ef64dac0482122 : SUCCESS in 7m 08s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/93d419bc1206419f8560b963122d0fc5 : SUCCESS in 8m 44s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/76fe5b0305c046b7ae98cf2ca2a4c06f : SUCCESS in 8m 45s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/5de4232270fe4839bbdc46b704326a14 : SUCCESS in 8m 41s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/b7ba8a6163ab4ea393e73c927204adf0 : SUCCESS in 4m 52s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/67dadffbf73c4bfa8965a1e59445ef94 : FAILURE in 1h 05m 41s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/b90b2b0693d34f7187a4ddca3a60f6c3 : FAILURE in 1h 09m 22s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/716d1083122d4204be81608fb50ac96e : FAILURE in 53m 29s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/be380474ab534f5b814b501392c12cc3 : FAILURE in 1h 11m 15s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c552ed87f23e4f5488d367f001583d26 : SUCCESS in 48m 45s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/1e8c103bf17845b1b510a25ebefef36a : FAILURE in 1h 19m 40s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/6b34b78ef8cf440c8d54a4a0d3cfa068 : SUCCESS in 29m 21s","accounts_in_message":[],"_revision_number":12},{"id":"622e6c15a52332ad3a0e1fdbf85dfc2598c1ba04","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-26 05:34:51.000000000","message":"Patch Set 12:\n\nrecheck","accounts_in_message":[],"_revision_number":12},{"id":"092867145cf9da975d97197cb9838dc28a99b611","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-26 07:10:19.000000000","message":"Patch Set 12:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/8b4c75a16adc45b98359018637c35d4d : SUCCESS in 1h 08m 28s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/307fcd7e501447759d7a7db5f5a1219f : SUCCESS in 1h 20m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e734dbb83e64b969c7db7c4398242b7 : SUCCESS in 6m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9105220b53fc4b0daa9ed5a9dd5bcc54 : SUCCESS in 8m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c3a81d9f135a4dc880731bab388453f7 : SUCCESS in 6m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5034209b19b1424cb4087e8c0d897efe : SUCCESS in 6m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b7bfd14a59148bcb4fe0ea8b6b7e060 : SUCCESS in 7m 13s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e2e7a5f7902d4f97b156ce506dcf9fe7 : SUCCESS in 10m 08s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/f74b8edac4ac434ca130c7a35b58c94b : SUCCESS in 8m 32s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/438239f9ab7e412db29599f9f45e25d9 : SUCCESS in 9m 32s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/476e98fa2b0c4079aaf451bca11378c1 : SUCCESS in 5m 27s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/e413b90107dd40d29712b611022bdb58 : SUCCESS in 1h 15m 33s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/644c9912024f41c28ef739e176614d47 : SUCCESS in 1h 12m 58s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/9411a9589aa64f078bdb33c32e995f28 : FAILURE in 1h 01m 37s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/f0a588a4b3e74bf3aa4f026f586835a0 : SUCCESS in 1h 29m 21s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/14f6fd759c00469d9eb8df9100b3ad59 : SUCCESS in 45m 42s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/a1d3db35512a48079a7e79a0ea10bb0d : FAILURE in 1h 12m 48s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/3169789419df48a78682a0ec6bfca3db : SUCCESS in 27m 28s","accounts_in_message":[],"_revision_number":12},{"id":"49f44ac5e0a40ee493152a59e22d05e0a12c1b75","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2021-08-26 10:51:21.000000000","message":"Patch Set 12:\n\nrecheck","accounts_in_message":[],"_revision_number":12},{"id":"0307fe95f8bd38f898e912c1fd6969a98841df86","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-26 12:37:22.000000000","message":"Patch Set 12:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/c7d4159204f0483d8be4218681f84894 : SUCCESS in 1h 42m 18s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4ff2ea1fc2c94acdba13ffef3b545c4d : FAILURE in 1h 02m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b9cb1105aad4ae5a3541da978467355 : SUCCESS in 5m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ac575651654e43a59213c0f9fa9011b0 : SUCCESS in 6m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/42ed76d7c0724e2cba267302de66f7cf : SUCCESS in 5m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8e96930efbfd42e1ab65062da55b9bbf : SUCCESS in 7m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da962bc6eb914bf4a975d242c2830e51 : SUCCESS in 8m 20s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/fe3226b12fed4f86b1b3d4e1ab7083d8 : SUCCESS in 9m 20s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/53fe981919e048a791d11ae5ec44ce64 : SUCCESS in 10m 22s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/84cd4e6301374ab2940c2ac130665e09 : SUCCESS in 9m 47s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/7167734c166143548bc10b9a91d2ae22 : SUCCESS in 5m 15s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/4c0a51f7d8b246b698c042874aaef741 : SUCCESS in 1h 03m 57s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/8070a8ee8af747a496c4c4f11933d518 : SUCCESS in 1h 00m 10s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/8340e8ad0c3147929749a66f7338cc3e : SUCCESS in 1h 14m 44s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/f4931a20fa1341008ba4990c8628ee26 : SUCCESS in 1h 16m 55s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/26aec72c5b844c23be96fdedb83bca8c : SUCCESS in 49m 40s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/2f4fe03f7f4c4a108e77cd739cc293f0 : SUCCESS in 1h 14m 02s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/1ea83a7cc2e54f43b9a22255270984c3 : SUCCESS in 31m 18s","accounts_in_message":[],"_revision_number":12},{"id":"6c899262fe427a496404be564a60672a395005b9","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-30 15:48:55.000000000","message":"Patch Set 12:\n\nrecheck","accounts_in_message":[],"_revision_number":12},{"id":"96d9dba3242eaee348c14f21929f8a5ffefca612","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-30 17:16:36.000000000","message":"Patch Set 12:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/c83d8f7813a34869917489301a847b9d : SUCCESS in 1h 03m 03s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6bad0b0e5b3d49a79db40c09ea5784f4 : SUCCESS in 1h 06m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b78948346cc34270a9ed344a07132c62 : SUCCESS in 7m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b29fd09abf0d437f939a6e69b37f6a3f : SUCCESS in 7m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/04d1d113cb464b1f91aa0edefb6d042a : SUCCESS in 6m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/324ff2b56e884338b50b7cd2a42f9e0e : SUCCESS in 7m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/45f4506cf5484ceab5a8dede6c0321e2 : SUCCESS in 8m 05s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ebf4da6cc231431c95850198b27b6a32 : SUCCESS in 10m 46s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/6b323e39f8e74c3bba851b2035c193b2 : SUCCESS in 9m 38s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/31e055e0cf494a449d1ced7b357fe649 : SUCCESS in 10m 30s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/084abfe5a1a448a6935f2f4a8f1ef545 : SUCCESS in 6m 04s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/c945734969a343e586a26ffa5aa379b2 : SUCCESS in 1h 12m 19s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/8e25f0fa99a2471798546c97762e8f60 : FAILURE in 1h 12m 19s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/be5f4fe397c44daeb26ba3c8d59dc8d9 : SUCCESS in 1h 00m 35s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/229c4d411cbc46bcaec550d5ccd5ca77 : SUCCESS in 1h 14m 05s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b9b6073811e9405588ee81decb072e79 : SUCCESS in 37m 50s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/6d1f84d3f4064c8e99ce30ee5f1f0ee4 : SUCCESS in 59m 29s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/04b77f464a834367a2540f5564298e7a : SUCCESS in 32m 35s","accounts_in_message":[],"_revision_number":12},{"id":"1a3a3f97a26a9fff4876ab65d89ec1f2fcb1c798","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2021-08-30 19:47:26.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"41da357944c09e64ff6b7275f0f77f15511016ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-30 21:10:47.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/2e0898ea31ae49e8a43980218a044cd4 : SUCCESS in 1h 12m 42s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/acf7821d007c4e7c980ce5eb2d6967c3 : SUCCESS in 1h 10m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/705435db7fb143d4aaa8ab386fb7fe66 : SUCCESS in 5m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dcadcf28a8404966bd585d10d3d5024a : SUCCESS in 6m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/593cea98042849b69b6403caf30ceb39 : SUCCESS in 5m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4537cc8ecb3a4e3c996d5fc554e140b5 : SUCCESS in 7m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e30721c7b8d4a21bf6f0e5f82c32c06 : SUCCESS in 8m 00s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a4d08d19cd1946568cc906fe0cbfa925 : SUCCESS in 9m 18s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/f2aa4697bd384041bc044c770fb78e75 : SUCCESS in 9m 00s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/dddc3855f5b14d22b70d7afc8810c340 : SUCCESS in 9m 43s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/889b120f9582435093d02ab74a57b544 : SUCCESS in 6m 58s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/a2be02cbbb134f84898d7ccd8926a990 : SUCCESS in 1h 06m 39s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/48b6b29e824743afa57de3f4e3172fde : SUCCESS in 1h 08m 06s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/595e66dd86d442fab121f72b621521cd : SUCCESS in 1h 08m 02s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/dc77d8987082403daea588d482d9b9ab : SUCCESS in 1h 17m 14s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8a38df0d3a3f45868b97ba0502848ab5 : SUCCESS in 47m 16s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/4298efda56594538be13b20a828178e8 : SUCCESS in 1h 17m 34s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/4185e8d749214d8fa43ddc4ad11b040b : SUCCESS in 41m 58s","accounts_in_message":[],"_revision_number":13},{"id":"299996a2fe7fa4861f51888cd02cfa595fc9f4c7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2021-08-31 06:08:04.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"ae9355ad429234008c8571087c0f1662bddc68e9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-31 08:52:50.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/b5fabf4b689d4831852e370db53a6058 : SUCCESS in 1h 03m 57s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/94edb23aa1994a76bb90ff38a151b0e4 : SUCCESS in 1h 05m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad737f76de2145aca92be381f93da768 : SUCCESS in 7m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/140f008ab9e646a8a8bea9d9b9cef9e7 : SUCCESS in 7m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/637dae3ad6084cba8e4c51cc0ec54696 : SUCCESS in 7m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f81eec04e00e4ee289e58d930df66873 : SUCCESS in 10m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e8cc780b3be449aaa86d61db177326e : SUCCESS in 12m 51s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3d9565d29bbc4f449a89c581beec293f : SUCCESS in 12m 39s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/982db1c3c5cf4bbe8ab94e22d002b54b : SUCCESS in 9m 59s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/9817667625e74d0cb05d6445660e7c76 : SUCCESS in 10m 10s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/a6e15345fd134c5390d9dca9ed05b426 : SUCCESS in 7m 03s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/6a86a568800541b6ab4a5b8c34b63c7c : SUCCESS in 1h 04m 23s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/a9a285543fcc41efb8ade008a2668670 : SUCCESS in 1h 24m 22s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/f07a22fc28194bf5be747ef6728fecf5 : SUCCESS in 1h 02m 24s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/243a66bb740c4188be73ac1f4fb0861d : SUCCESS in 1h 20m 53s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6471c964ec2b4d2e83b0c676bde857e4 : SUCCESS in 51m 46s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/aeddcb4ed5e64ccfbff075c8682d4b40 : SUCCESS in 1h 27m 31s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/31f8a18924a34b7d8e67e45f96cbfd6f : SUCCESS in 32m 26s","accounts_in_message":[],"_revision_number":14},{"id":"9044aab964be61b96ff127d2a1c0d49b66026d7a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2021-08-31 20:04:52.000000000","message":"Uploaded patch set 15: Patch Set 14 was rebased.","accounts_in_message":[],"_revision_number":15},{"id":"ade4d171572ae1f9f2d43aba95b4bb4af6022b97","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2021-08-31 20:20:51.000000000","message":"Uploaded patch set 16: Patch Set 15 was rebased.","accounts_in_message":[],"_revision_number":16},{"id":"3a440ad65d33968b70ba7779c5138f461956689f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-31 21:47:11.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/d5e18b30aaaf47a08b5eb18cc74f86e0 : SUCCESS in 1h 00m 25s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/56cdd7e25d5546f183cf3633c66d7e65 : SUCCESS in 59m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba66d243767a47acadc93349829c4442 : SUCCESS in 5m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/154a12f7335547a4834bdef00f3baf29 : SUCCESS in 6m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/624eda0c2e5b43969a2082ed8fb4c0b2 : SUCCESS in 7m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ecfbfe48eba64ae7a68dd18648f5cb18 : SUCCESS in 7m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2799c940d9d41b6b64417e8298f6de7 : SUCCESS in 7m 39s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9c7a7c8606c649b09e2e7bfc9c8d1feb : SUCCESS in 10m 53s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/a7e4751a305041f489158395c54ca007 : SUCCESS in 8m 00s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/c8c1cc42b90945ec9ad11399c1e57034 : SUCCESS in 10m 28s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/6607310281f94c03a3008318755ce26e : SUCCESS in 7m 21s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/4c0d0a954f51418eb57a13ea5f5feb69 : SUCCESS in 1h 24m 08s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/d48e9c6306d9408091519edc59ed2e83 : SUCCESS in 1h 14m 30s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/73aaca40fb9a4e058d751583595d858b : SUCCESS in 1h 10m 18s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/79e70bea45134c38aed197c0829b3b04 : FAILURE in 59m 47s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/59c49c24a2e04c6d85d3a59c776ba9c7 : SUCCESS in 49m 37s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/56d580db143d46f79ac9ea539b183887 : SUCCESS in 1h 07m 10s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/338bbb8e5bf2402a97d026339bd24693 : SUCCESS in 25m 44s","accounts_in_message":[],"_revision_number":16},{"id":"c0dd45303483195167e2c9f9d67a1fb03079ad72","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-09-02 08:58:29.000000000","message":"Patch Set 16: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"1932f8dbc2a9fa68f8be147b8ea954645b176c3e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-09-02 09:20:55.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"633a2d45ed8722c4a94e0eed3c27ec6853906487","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-09-02 09:38:36.000000000","message":"Patch Set 17: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":17},{"id":"befac5ced99f0544c1a153906cf1823a4ca686dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-02 11:02:39.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/e1f4789edcdc4bf8818a07bfdbe814ec : SUCCESS in 1h 01m 11s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a04c29074f7c40cab88a81e32e2426ff : SUCCESS in 1h 06m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e34d148e61a4d43a50c4af69040eedf : SUCCESS in 5m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/620a96670518489f8f7184ef8cc3fca4 : SUCCESS in 6m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/348f2b06486f4512a359bd336118c5c2 : SUCCESS in 8m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ac09ccd91f6c4d2ab3cd42a9c6b0164c : SUCCESS in 7m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/480c7f4a036d4c50952019d03fd80e34 : SUCCESS in 8m 58s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5c9d8f8ed58545a19dc8e12c9e53a30e : SUCCESS in 9m 55s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/9ff69788640c46c28755c6a023576333 : SUCCESS in 8m 57s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/65080762d3144fa98b26e0950db39c06 : SUCCESS in 11m 56s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/9af2d85b5a8645d4a2bc17fa00ef9840 : SUCCESS in 6m 10s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/3f986aeff3514e27a1ad2bca0379f048 : SUCCESS in 1h 12m 53s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/6c25fa6142274ddfa5f737e66585ca8a : FAILURE in 56m 39s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/f8710af523f5481bb35537407882edeb : SUCCESS in 1h 08m 29s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/5f5a13fc1fa34d4984eb2fc23c8d1ddc : SUCCESS in 1h 14m 36s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0c84e46347b14be5ad7718a1428f6f4e : SUCCESS in 46m 27s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/d2d7e8c3dc75450796c1e40096512e64 : SUCCESS in 1h 28m 14s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/221ee206bb2f47e5990f92ea1735d78d : SUCCESS in 29m 21s","accounts_in_message":[],"_revision_number":17},{"id":"17a019bbb1297a9fbab50a95975b1abbf929502c","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2021-09-02 12:18:10.000000000","message":"Patch Set 17:\n\n(2 comments)","accounts_in_message":[],"_revision_number":17},{"id":"ecac78f2e6e5ba49e599e239ff005d11ff688146","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2021-09-02 12:18:28.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"9b28f7701edd3f49284763ddbe8355646f08684a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-02 13:41:24.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/e77670f3da07436b864097592e02c64e : SUCCESS in 1h 08m 47s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0b5dbd20d37c459895050873dbe429ee : SUCCESS in 1h 06m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5d22fb8eb784d63ace611ec82655200 : SUCCESS in 6m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/36c85d4f8940482aab6d2acec113250b : SUCCESS in 6m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/514d10f3e6844869a829b3ee44a4d560 : SUCCESS in 6m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bf0580fe6b2849a0bcc2962928f13e99 : SUCCESS in 11m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84fec657a652472cb2879bf2ced5950f : SUCCESS in 7m 12s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/dc8c4a695e054be7b4d64fba41768554 : SUCCESS in 10m 34s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/940c78936588400f850c06d488182b49 : SUCCESS in 11m 30s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/cf697dd9fb7146028d17ec3499ce4af0 : SUCCESS in 9m 07s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/28f0dc2704834a5093e03db9dd9c85f8 : SUCCESS in 9m 45s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/c17ef645333a454a8a8593f3429e89eb : SUCCESS in 1h 14m 09s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/b80132ead5264c548162722a70f2a3a8 : SUCCESS in 1h 07m 31s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/55d7ebfc394d4443bf238f98f805851f : SUCCESS in 1h 12m 43s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/1d38492d9bdf445784562964f674f62c : SUCCESS in 1h 18m 24s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fd33a0db8ac84024a585ad94824d95aa : SUCCESS in 50m 29s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/941f9c9deafe4b3cb0385d41ee4f8a99 : SUCCESS in 1h 09m 53s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/12670f36a72941679288c44b960922aa : SUCCESS in 33m 26s","accounts_in_message":[],"_revision_number":18},{"id":"b9e267f3813a3077769f8182001e2d185a1dfadf","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-09-02 16:59:20.000000000","message":"Patch Set 18: Code-Review+2","accounts_in_message":[],"_revision_number":18},{"id":"e58987d1549e3649632c8b3c711800d3206ee774","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-09-02 19:59:05.000000000","message":"Patch Set 18: Code-Review+2","accounts_in_message":[],"_revision_number":18},{"id":"acd920691116123fb32970479069867db46bb531","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-09-02 20:32:37.000000000","message":"Patch Set 18:\n\n(3 comments)\n\nJust wondering if we\u0027re missing a test case for ensuring list properties are (or are not) restricted across private namespaces.","accounts_in_message":[],"_revision_number":18},{"id":"cfa7d2bd6a724c7a9af8c627fcef0267b4a88740","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-09-02 20:57:59.000000000","message":"Patch Set 18:\n\n(2 comments)","accounts_in_message":[],"_revision_number":18},{"id":"c4c84a03c080c62379af75dcc351c5633965ef73","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-09-02 20:59:32.000000000","message":"Patch Set 18:\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"1e1c8565f912b35df0b8234cb952cd0bcf51950d","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-09-02 21:08:26.000000000","message":"Patch Set 18:\n\n(2 comments)","accounts_in_message":[],"_revision_number":18},{"id":"a7d19cffb9b48870fb49f953e13b56366b6be310","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-09-02 21:19:14.000000000","message":"Patch Set 18:\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"ff69cfeb772072d78b3074feff740e8f9fecee39","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-09-02 21:39:04.000000000","message":"Patch Set 18:\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"2bc1a98c26f80449ffbe1ee05fe1b60e7b222e95","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-09-02 22:03:30.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"f98a215b4acbf2522edd58f4d6eb1a312304cfbf","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-09-02 22:05:45.000000000","message":"Patch Set 19: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":19},{"id":"66ca9867f765012dceed923080cf2792e0327195","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-02 23:18:48.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/814d6c1eda4d412f862d55fb8cfdd4d9 : SUCCESS in 1h 03m 07s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/67f39eb7f7344edfb95d449cf9ecf698 : SUCCESS in 1h 01m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/67615903d0344f0db2e94d9392f6298b : SUCCESS in 6m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/663cc74baebb4ab4a7939c045b92d7b0 : SUCCESS in 6m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8c221ff9303a4a1c81d988e6b1f28cd6 : SUCCESS in 7m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d230b274331e43cebbccb9c8499e231c : SUCCESS in 8m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37cf4c753ddf4093aba345d266229629 : SUCCESS in 7m 07s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/aad6351104f0490b89d1df8e845b4b9f : SUCCESS in 11m 12s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b445101767fb41d19499f3a04c395b0a : SUCCESS in 10m 38s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/4ddea6d1af4a4ced8de868dbe519e0e6 : SUCCESS in 15m 31s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/77535a89ea39464194c982751442b5f2 : SUCCESS in 5m 36s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/ebc8ba263ea94178981ea869a3146e08 : SUCCESS in 1h 09m 10s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/6ffdf0ac2ddd4c2c83aac2816af0d7b8 : SUCCESS in 1h 02m 50s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/193a9cc123574bf0850bb8b1e381ba3c : SUCCESS in 1h 00m 59s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/7b4b0c77d7734f5d9d0c26f4cd2da600 : SUCCESS in 1h 13m 13s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/97c9a0c620804efca2d50cbe3b69eb08 : SUCCESS in 43m 20s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/4d7e3c4e46674d6597ec1031d8e93a7a : SUCCESS in 1h 08m 11s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/d65aae815bd947cba654cbf3c9d89fbd : SUCCESS in 36m 39s","accounts_in_message":[],"_revision_number":19},{"id":"7cfe651d83263c6ef37d4d8a92bd6a3b421d70bd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-02 23:36:53.000000000","message":"Patch Set 19: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":19},{"id":"114e5c90de1893793a5a553a823c8c28993171b5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-03 02:08:28.000000000","message":"Patch Set 19: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":19},{"id":"f30cb250989a06fd8f92cbb12a3056e441f7d82e","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-09-03 03:23:15.000000000","message":"Patch Set 19:\n\nrecheck","accounts_in_message":[],"_revision_number":19},{"id":"462e0eede57aafe1751814b385c4c777544537f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-03 05:09:37.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/d0138b4afa284aafafb75659ef38b390 : SUCCESS in 1h 03m 30s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3eca230369a54be99c3a770480987f7b : SUCCESS in 1h 08m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/166e02101d5c4bccb523dc7a51ce407b : SUCCESS in 5m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b6cd501be4a34c9a8718a52efa763c10 : SUCCESS in 6m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f35a4af4f27f4beaa30503659265c68f : SUCCESS in 6m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1e7b8f75f2d1436d984ff8b8dfd4a16e : SUCCESS in 6m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e3b155ced8b459eb42f89cd63da4528 : SUCCESS in 7m 02s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/cb3e52da1c7042368843bd4c72a12635 : SUCCESS in 9m 56s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/89e46cfad0bb4a81bbd9e857548bf1ff : SUCCESS in 8m 40s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/f16a4e28c6924a5992775764bf1f64ea : SUCCESS in 8m 53s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/69b13a97695f48b894a5800b7f177448 : SUCCESS in 5m 34s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/eeab27ffa5fc481c9cdae87a615e7201 : SUCCESS in 1h 08m 44s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/60259d104f644367b8fb95eeab01d397 : SUCCESS in 1h 01m 36s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/286b1e1067604ffa9ffe791799bea9d3 : SUCCESS in 1h 02m 43s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/9cc8bf97af4e4aad838fe84d45b1c6bd : SUCCESS in 1h 39m 24s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3a52c0fa91254485894b302f1b2d2b3e : SUCCESS in 51m 09s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/e82a0a4ea2304cc0a7fe82b9aee87857 : SUCCESS in 1h 23m 19s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/efba443c16234148bda388b6503f38cf : SUCCESS in 30m 29s","accounts_in_message":[],"_revision_number":19},{"id":"b1ed689a9270d807b2004568753b00d71bf4a8a4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-03 05:10:10.000000000","message":"Patch Set 19: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":19},{"id":"c592bca98179e7b735f3e77fddbe1ba2127413d4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-03 06:32:14.000000000","message":"Patch Set 19: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/3925e90eddd840bda8049ffd65e0851c : SUCCESS in 1h 02m 20s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/757e9aeefc5a4d929c1795e33d37ed92 : SUCCESS in 1h 07m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba98770029574e7cadb9b70c93a8af6d : SUCCESS in 5m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7110d88965154d01a3b7a526fd01716a : SUCCESS in 8m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8f0a89cb51ba49e38a0d6a18bc6a730e : SUCCESS in 6m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78daae539a6449b593f7929a0a2489ce : SUCCESS in 7m 59s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/fcb851a0a40546ecab7842f1633981af : SUCCESS in 11m 09s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/0ad9806510654c77a894cb18eb0b714b : SUCCESS in 9m 00s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/508c5253f21a4f81bf7f772af6c51076 : SUCCESS in 1h 07m 30s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/aff78e6c7d8d476da963a70301dc17d8 : SUCCESS in 1h 03m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bea31dcf9c06483da0445f59ae8e4804 : SUCCESS in 45m 27s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/7e5209a98936417b8a25439c4cff52c5 : SUCCESS in 1h 20m 58s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/515fe9d0ca3f4ce1b467df143e6e3d61 : SUCCESS in 25m 45s","accounts_in_message":[],"_revision_number":19},{"id":"e18062b954983e59a70973d3576673fb190d1be5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-03 06:32:15.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":19},{"id":"3eed2d0f9e050abef29d3f383b6943a35fd21f8f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-03 06:34:12.000000000","message":"Patch Set 19:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dfc3307d8a8b4196b798342ee2f1b2cf : SUCCESS in 1m 32s","accounts_in_message":[],"_revision_number":19}],"current_revision_number":19,"current_revision":"9796593f31a4f0bbc62ae6e0541385b573b9ce12","revisions":{"8201990e7415837c197ec52b75863bdefd64b997":{"kind":"REWORK","_number":1,"created":"2021-07-23 16:34:12.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/55/802055/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/55/802055/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/802055/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/802055/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/802055/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/55/802055/1"}}},"commit":{"parents":[{"commit":"8afbe6a83accf2ad57a2f51d2a420f882ae74a76","subject":"Implement project personas for metadef objects","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8afbe6a83accf2ad57a2f51d2a420f882ae74a76"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2021-07-23 14:19:40.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2021-07-23 14:19:40.000000000","tz":0},"subject":"Implement project personas for metadef properties","message":"Implement project personas for metadef properties\n\nThis commit updates the policies for metadef properties\nto use default roles available from keystone.\n\nImplements: blueprint secure-rbac\nChange-Id: Ic93c2a8a7017e6eac16d23fb5a9224d2c11e7211\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8201990e7415837c197ec52b75863bdefd64b997"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8201990e7415837c197ec52b75863bdefd64b997"}]},"branch":"refs/heads/master"},"e284c9db15ce0badfa661856eb1495dc55317f89":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-07-23 17:02:33.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/55/802055/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/55/802055/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/802055/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/802055/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/802055/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/55/802055/2"}}},"commit":{"parents":[{"commit":"7a605b07c95919fe7936c2dad9b608f39248dd33","subject":"Implement project personas for metadef objects","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7a605b07c95919fe7936c2dad9b608f39248dd33"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2021-07-23 14:19:40.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2021-07-23 17:02:33.000000000","tz":0},"subject":"Implement project personas for metadef properties","message":"Implement project personas for metadef properties\n\nThis commit updates the policies for metadef properties\nto use default roles available from keystone.\n\nImplements: blueprint secure-rbac\nChange-Id: Ic93c2a8a7017e6eac16d23fb5a9224d2c11e7211\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e284c9db15ce0badfa661856eb1495dc55317f89"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e284c9db15ce0badfa661856eb1495dc55317f89"}]},"branch":"refs/heads/master","description":"Rebase"},"edf00595f255ad59fe8f24dc677bab98cb59bc5d":{"kind":"REWORK","_number":3,"created":"2021-07-25 19:36:32.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/55/802055/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/55/802055/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/802055/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/802055/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/802055/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/55/802055/3"}}},"commit":{"parents":[{"commit":"f2cd2460a1a57986a6a07eecffaf182e36544737","subject":"Implement project personas for metadef objects","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f2cd2460a1a57986a6a07eecffaf182e36544737"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2021-07-23 14:19:40.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2021-07-25 18:44:16.000000000","tz":0},"subject":"Implement project personas for metadef properties","message":"Implement project personas for metadef properties\n\nThis commit updates the policies for metadef properties\nto use default roles available from keystone.\n\nImplements: blueprint secure-rbac\nChange-Id: Ic93c2a8a7017e6eac16d23fb5a9224d2c11e7211\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/edf00595f255ad59fe8f24dc677bab98cb59bc5d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/edf00595f255ad59fe8f24dc677bab98cb59bc5d"}]},"branch":"refs/heads/master"},"91eccd8b543ff735743ba2c211d4bf8f714aa165":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2021-07-26 10:16:39.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/55/802055/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/55/802055/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/802055/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/802055/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/802055/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/55/802055/4"}}},"commit":{"parents":[{"commit":"fc6604635431382e33491f4f37f31a21fc1c019f","subject":"Implement project personas for metadef objects","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/fc6604635431382e33491f4f37f31a21fc1c019f"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2021-07-23 14:19:40.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2021-07-26 10:16:39.000000000","tz":0},"subject":"Implement project personas for metadef properties","message":"Implement project personas for metadef properties\n\nThis commit updates the policies for metadef properties\nto use default roles available from keystone.\n\nImplements: blueprint secure-rbac\nChange-Id: Ic93c2a8a7017e6eac16d23fb5a9224d2c11e7211\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/91eccd8b543ff735743ba2c211d4bf8f714aa165"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/91eccd8b543ff735743ba2c211d4bf8f714aa165"}]},"branch":"refs/heads/master","description":"Rebase"},"f5c4f1ce46951b7c30da801a691096be89250276":{"kind":"REWORK","_number":5,"created":"2021-07-29 11:05:07.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/55/802055/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/55/802055/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/802055/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/802055/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/802055/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/55/802055/5"}}},"commit":{"parents":[{"commit":"b5ec11842534d6308951cd7dd45707ba4fa33226","subject":"Implement project personas for metadef objects","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b5ec11842534d6308951cd7dd45707ba4fa33226"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2021-07-23 14:19:40.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2021-07-29 10:20:50.000000000","tz":0},"subject":"Implement project personas for metadef properties","message":"Implement project personas for metadef properties\n\nThis commit updates the policies for metadef properties\nto use default roles available from keystone.\n\nImplements: blueprint secure-rbac\nChange-Id: Ic93c2a8a7017e6eac16d23fb5a9224d2c11e7211\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f5c4f1ce46951b7c30da801a691096be89250276"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f5c4f1ce46951b7c30da801a691096be89250276"}]},"branch":"refs/heads/master"},"b80823a173757fbd60d1453b7a5c9a2c98cc1bb7":{"kind":"REWORK","_number":6,"created":"2021-07-30 11:26:59.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/55/802055/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/55/802055/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/802055/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/802055/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/802055/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/55/802055/6"}}},"commit":{"parents":[{"commit":"44e7ddbd1c2d0cad78d76536f8dca8007201809b","subject":"Implement project personas for metadef objects","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/44e7ddbd1c2d0cad78d76536f8dca8007201809b"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2021-07-23 14:19:40.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2021-07-30 11:05:00.000000000","tz":0},"subject":"Implement project personas for metadef properties","message":"Implement project personas for metadef properties\n\nThis commit updates the policies for metadef properties\nto use default roles available from keystone.\n\nImplements: blueprint secure-rbac\nChange-Id: Ic93c2a8a7017e6eac16d23fb5a9224d2c11e7211\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b80823a173757fbd60d1453b7a5c9a2c98cc1bb7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b80823a173757fbd60d1453b7a5c9a2c98cc1bb7"}]},"branch":"refs/heads/master"},"13fff0895a3e304e44fac5dcc2ab776fd9dffb34":{"kind":"REWORK","_number":7,"created":"2021-08-03 05:31:34.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/55/802055/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/55/802055/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/802055/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/802055/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/802055/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/55/802055/7"}}},"commit":{"parents":[{"commit":"c1861fc57bbd287de2a31d11a80b265a13cd28fc","subject":"Implement project personas for metadef objects","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c1861fc57bbd287de2a31d11a80b265a13cd28fc"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2021-07-23 14:19:40.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2021-08-03 05:23:31.000000000","tz":0},"subject":"Implement project personas for metadef properties","message":"Implement project personas for metadef properties\n\nThis commit updates the policies for metadef properties\nto use default roles available from keystone.\n\nImplements: blueprint secure-rbac\nChange-Id: Ic93c2a8a7017e6eac16d23fb5a9224d2c11e7211\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/13fff0895a3e304e44fac5dcc2ab776fd9dffb34"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/13fff0895a3e304e44fac5dcc2ab776fd9dffb34"}]},"branch":"refs/heads/master"},"8e405642e97c01d9649477c93b44b3d8bc8e41b5":{"kind":"REWORK","_number":8,"created":"2021-08-03 09:00:47.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/55/802055/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/55/802055/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/802055/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/802055/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/802055/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/55/802055/8"}}},"commit":{"parents":[{"commit":"a2fb137f03858c8a2161e7b86b3b076b0208fd44","subject":"Implement project personas for metadef objects","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a2fb137f03858c8a2161e7b86b3b076b0208fd44"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2021-07-23 14:19:40.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2021-08-03 08:44:35.000000000","tz":0},"subject":"Implement project personas for metadef properties","message":"Implement project personas for metadef properties\n\nThis commit updates the policies for metadef properties\nto use default roles available from keystone.\n\nImplements: blueprint secure-rbac\nChange-Id: Ic93c2a8a7017e6eac16d23fb5a9224d2c11e7211\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8e405642e97c01d9649477c93b44b3d8bc8e41b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8e405642e97c01d9649477c93b44b3d8bc8e41b5"}]},"branch":"refs/heads/master"},"61e6cbcbc4721f60eeec2fbdc783cc2fb8f63ee3":{"kind":"REWORK","_number":9,"created":"2021-08-03 11:22:18.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/55/802055/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/55/802055/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/802055/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/802055/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/802055/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/55/802055/9"}}},"commit":{"parents":[{"commit":"6c5999aa2f53b18866dc0f756d3e901eca547459","subject":"Implement project personas for metadef objects","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6c5999aa2f53b18866dc0f756d3e901eca547459"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2021-07-23 14:19:40.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2021-08-03 11:02:23.000000000","tz":0},"subject":"Implement project personas for metadef properties","message":"Implement project personas for metadef properties\n\nThis commit updates the policies for metadef properties\nto use default roles available from keystone.\n\nImplements: blueprint secure-rbac\nChange-Id: Ic93c2a8a7017e6eac16d23fb5a9224d2c11e7211\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/61e6cbcbc4721f60eeec2fbdc783cc2fb8f63ee3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/61e6cbcbc4721f60eeec2fbdc783cc2fb8f63ee3"}]},"branch":"refs/heads/master"},"068deeea149405799c032857ee83cbe1f75c87a1":{"kind":"REWORK","_number":10,"created":"2021-08-16 14:07:12.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/55/802055/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/55/802055/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/802055/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/802055/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/802055/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/55/802055/10"}}},"commit":{"parents":[{"commit":"ec31b7d40cac0655de8ef255bbad9275286caae6","subject":"Implement project personas for metadef resource-types","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ec31b7d40cac0655de8ef255bbad9275286caae6"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2021-07-23 14:19:40.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2021-08-16 13:23:41.000000000","tz":0},"subject":"Implement project personas for metadef properties","message":"Implement project personas for metadef properties\n\nThis commit updates the policies for metadef properties\nto use default roles available from keystone.\n\nImplements: blueprint secure-rbac\nChange-Id: Ic93c2a8a7017e6eac16d23fb5a9224d2c11e7211\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/068deeea149405799c032857ee83cbe1f75c87a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/068deeea149405799c032857ee83cbe1f75c87a1"}]},"branch":"refs/heads/master"},"90a68710c4d187a7f26ee32ec189c50fca1d0d8b":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2021-08-23 08:53:40.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/55/802055/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/55/802055/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/802055/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/802055/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/802055/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/55/802055/11"}}},"commit":{"parents":[{"commit":"9a77b5af90c9d84f538906749394293bb79f9517","subject":"Implement project personas for metadef resource-types","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9a77b5af90c9d84f538906749394293bb79f9517"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2021-07-23 14:19:40.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2021-08-23 08:53:40.000000000","tz":0},"subject":"Implement project personas for metadef properties","message":"Implement project personas for metadef properties\n\nThis commit updates the policies for metadef properties\nto use default roles available from keystone.\n\nImplements: blueprint secure-rbac\nChange-Id: Ic93c2a8a7017e6eac16d23fb5a9224d2c11e7211\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/90a68710c4d187a7f26ee32ec189c50fca1d0d8b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/90a68710c4d187a7f26ee32ec189c50fca1d0d8b"}]},"branch":"refs/heads/master","description":"Rebase"},"be9a2975bcc303801047a9c3415b77377c7c1501":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2021-08-26 04:02:48.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/55/802055/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/55/802055/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/802055/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/802055/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/802055/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/55/802055/12"}}},"commit":{"parents":[{"commit":"da548a183b38f02cea4da24af106b3b901335554","subject":"Implement project personas for metadef resource-types","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/da548a183b38f02cea4da24af106b3b901335554"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2021-07-23 14:19:40.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-08-26 03:57:02.000000000","tz":0},"subject":"Implement project personas for metadef properties","message":"Implement project personas for metadef properties\n\nThis commit updates the policies for metadef properties\nto use default roles available from keystone.\n\nImplements: blueprint secure-rbac\nChange-Id: Ic93c2a8a7017e6eac16d23fb5a9224d2c11e7211\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/be9a2975bcc303801047a9c3415b77377c7c1501"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/be9a2975bcc303801047a9c3415b77377c7c1501"}]},"branch":"refs/heads/master"},"b216383f52643b27e8421878fdfcbbe365207301":{"kind":"REWORK","_number":13,"created":"2021-08-30 19:47:26.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/55/802055/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/55/802055/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/802055/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/802055/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/802055/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/55/802055/13"}}},"commit":{"parents":[{"commit":"91c47d0bf7f858371b819ffbfc318bc42c4547b9","subject":"Implement project personas for metadef resource-types","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/91c47d0bf7f858371b819ffbfc318bc42c4547b9"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2021-07-23 14:19:40.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2021-08-30 19:27:39.000000000","tz":0},"subject":"Implement project personas for metadef properties","message":"Implement project personas for metadef properties\n\nThis commit updates the policies for metadef properties\nto use default roles available from keystone.\n\nImplements: blueprint secure-rbac\nChange-Id: Ic93c2a8a7017e6eac16d23fb5a9224d2c11e7211\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b216383f52643b27e8421878fdfcbbe365207301"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b216383f52643b27e8421878fdfcbbe365207301"}]},"branch":"refs/heads/master"},"273a00af1537809e7e7775a277147019ed34095e":{"kind":"REWORK","_number":14,"created":"2021-08-31 06:08:04.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/55/802055/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/55/802055/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/802055/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/802055/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/802055/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/55/802055/14"}}},"commit":{"parents":[{"commit":"503c9d9e8e9413255ac2b1a7daefdf88dda1c680","subject":"Implement project personas for metadef resource-types","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/503c9d9e8e9413255ac2b1a7daefdf88dda1c680"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2021-07-23 14:19:40.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2021-08-31 05:45:29.000000000","tz":0},"subject":"Implement project personas for metadef properties","message":"Implement project personas for metadef properties\n\nThis commit updates the policies for metadef properties\nto use default roles available from keystone.\n\nImplements: blueprint secure-rbac\nChange-Id: Ic93c2a8a7017e6eac16d23fb5a9224d2c11e7211\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/273a00af1537809e7e7775a277147019ed34095e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/273a00af1537809e7e7775a277147019ed34095e"}]},"branch":"refs/heads/master"},"0d9f639dde96261685907439b1465e6954e78c07":{"kind":"TRIVIAL_REBASE","_number":15,"created":"2021-08-31 20:04:52.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/55/802055/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/55/802055/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/802055/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/802055/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/802055/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/55/802055/15"}}},"commit":{"parents":[{"commit":"d36659c6bf34a0ebe3c2bfd6dd2ff646b9d8685c","subject":"Implement project personas for metadef resource-types","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d36659c6bf34a0ebe3c2bfd6dd2ff646b9d8685c"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2021-07-23 14:19:40.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2021-08-31 19:47:48.000000000","tz":0},"subject":"Implement project personas for metadef properties","message":"Implement project personas for metadef properties\n\nThis commit updates the policies for metadef properties\nto use default roles available from keystone.\n\nImplements: blueprint secure-rbac\nChange-Id: Ic93c2a8a7017e6eac16d23fb5a9224d2c11e7211\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0d9f639dde96261685907439b1465e6954e78c07"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0d9f639dde96261685907439b1465e6954e78c07"}]},"branch":"refs/heads/master"},"90e281e2b25096ba0f5f188c0c142a3ee11d697e":{"kind":"TRIVIAL_REBASE","_number":16,"created":"2021-08-31 20:20:51.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/55/802055/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/55/802055/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/802055/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/802055/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/802055/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/55/802055/16"}}},"commit":{"parents":[{"commit":"97856f29bf73e5fed48d5358d1d07f60a6fc70a6","subject":"Implement project personas for metadef resource-types","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/97856f29bf73e5fed48d5358d1d07f60a6fc70a6"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2021-07-23 14:19:40.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2021-08-31 20:05:29.000000000","tz":0},"subject":"Implement project personas for metadef properties","message":"Implement project personas for metadef properties\n\nThis commit updates the policies for metadef properties\nto use default roles available from keystone.\n\nImplements: blueprint secure-rbac\nChange-Id: Ic93c2a8a7017e6eac16d23fb5a9224d2c11e7211\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/90e281e2b25096ba0f5f188c0c142a3ee11d697e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/90e281e2b25096ba0f5f188c0c142a3ee11d697e"}]},"branch":"refs/heads/master"},"0289c27c44f90bffe7c86bd768403374f0d67a5b":{"kind":"REWORK","_number":17,"created":"2021-09-02 09:20:55.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/55/802055/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/55/802055/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/802055/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/802055/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/802055/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/55/802055/17"}}},"commit":{"parents":[{"commit":"d2fc573e311af1a166e3591f0ba5837fca973256","subject":"Implement project personas for metadef resource-types","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d2fc573e311af1a166e3591f0ba5837fca973256"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2021-07-23 14:19:40.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-09-02 08:50:28.000000000","tz":0},"subject":"Implement project personas for metadef properties","message":"Implement project personas for metadef properties\n\nThis commit updates the policies for metadef properties\nto use default roles available from keystone.\n\nImplements: blueprint secure-rbac\nChange-Id: Ic93c2a8a7017e6eac16d23fb5a9224d2c11e7211\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0289c27c44f90bffe7c86bd768403374f0d67a5b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0289c27c44f90bffe7c86bd768403374f0d67a5b"}]},"branch":"refs/heads/master"},"3554e7c393b1fe03d27af8c11806734961bd84dd":{"kind":"REWORK","_number":18,"created":"2021-09-02 12:18:28.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/55/802055/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/55/802055/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/802055/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/802055/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/802055/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/55/802055/18"}}},"commit":{"parents":[{"commit":"3a96d5ab326f5ca1fa90e77034a5f6ab0867cf7a","subject":"Implement project personas for metadef resource-types","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3a96d5ab326f5ca1fa90e77034a5f6ab0867cf7a"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2021-07-23 14:19:40.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2021-09-02 11:29:04.000000000","tz":0},"subject":"Implement project personas for metadef properties","message":"Implement project personas for metadef properties\n\nThis commit updates the policies for metadef properties\nto use default roles available from keystone.\n\nImplements: blueprint secure-rbac\nChange-Id: Ic93c2a8a7017e6eac16d23fb5a9224d2c11e7211\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3554e7c393b1fe03d27af8c11806734961bd84dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3554e7c393b1fe03d27af8c11806734961bd84dd"}]},"branch":"refs/heads/master"},"9796593f31a4f0bbc62ae6e0541385b573b9ce12":{"kind":"REWORK","_number":19,"created":"2021-09-02 22:03:30.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/55/802055/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/55/802055/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/802055/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/802055/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/802055/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/55/802055/19"}}},"commit":{"parents":[{"commit":"c64083badcf6261c55f01fd8098b1e47f19fbb5b","subject":"Implement project personas for metadef resource-types","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c64083badcf6261c55f01fd8098b1e47f19fbb5b"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2021-07-23 14:19:40.000000000","tz":0},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-09-02 23:06:12.000000000","tz":-420},"subject":"Implement project personas for metadef properties","message":"Implement project personas for metadef properties\n\nThis commit updates the policies for metadef properties\nto use default roles available from keystone.\n\nImplements: blueprint secure-rbac\nChange-Id: Ic93c2a8a7017e6eac16d23fb5a9224d2c11e7211\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9796593f31a4f0bbc62ae6e0541385b573b9ce12"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9796593f31a4f0bbc62ae6e0541385b573b9ce12"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
