)]}'
{"id":"openstack%2Fglance~802056","triplet_id":"openstack%2Fglance~master~I2f97c64a72e081ae2f9e558b6fa53dbfd5af2026","project":"openstack/glance","branch":"master","topic":"secure-rbac","hashtags":[],"change_id":"I2f97c64a72e081ae2f9e558b6fa53dbfd5af2026","subject":"Implement project personas for metadef tags","status":"MERGED","created":"2021-07-23 16:34:12.000000000","updated":"2021-09-06 06:32:17.000000000","submitted":"2021-09-03 17:05:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":32,"unresolved_comment_count":11,"has_review_started":true,"submission_id":"802056-secure-rbac","meta_rev_id":"41c0734da613f2466252ddb1da26cb8b9f959b63","_number":802056,"virtual_id_number":802056,"owner":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-03 17:05:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"recommended":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"all":[{"value":1,"date":"2021-09-03 14:01:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":2,"date":"2021-09-03 15:44:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2021-09-03 14:19:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":1,"date":"2021-09-03 15:44:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}],"reviewers":{"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]}],"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-23 17:46:34.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2021-07-23 18:29:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-24 05:31:29.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-08-04 07:03:10.000000000","updated_by":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"reviewer":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"state":"REVIEWER"},{"updated":"2021-08-04 07:03:10.000000000","updated_by":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2021-08-04 07:03:10.000000000","updated_by":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"}],"messages":[{"id":"c86519b119a91dde85cbb8cba98fe56ba66fed88","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2021-07-23 16:34:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6409471b484d645722c8579d631fcfd64e137754","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2021-07-23 17:02:45.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"0e3a7b906fea362d54885acb6785de4a629bdc49","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-07-23 17:46:34.000000000","message":"Patch Set 2: Code-Review+1\n\n(3 comments)\n\nFew suggestions inline!","accounts_in_message":[],"_revision_number":2},{"id":"0a3ec460438c4d21a39a9b7a6cb0f5e6d3c993dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-23 18:29:53.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/e84052b47ddf4f3cad9ee837861fd18b : SUCCESS in 59m 31s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/526beec9d92643c4a50989bd8fdff1b5 : SUCCESS in 1h 10m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5462029207b94b5fa66840e255171bcd : SUCCESS in 6m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/53292363d5254c1aa8ab1eafb10bcaa0 : SUCCESS in 6m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e29cccc265364462ad6aa6ce685ab055 : SUCCESS in 6m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d7e1f3e19fea49fba8917f701dd1c06b : SUCCESS in 7m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6fb66d2ee1a4f0f81eb36ecc3b6339d : SUCCESS in 8m 16s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c42910e6191146028a3ca9553a6fe01b : SUCCESS in 9m 34s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/4f9d648a5dc047d897370edd866a050f : SUCCESS in 10m 22s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/fa7fa8e504a045e7b152372d3d7ebba8 : SUCCESS in 9m 19s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/36f683f362764fbd848583c81d50934b : SUCCESS in 5m 24s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/18e046324e824a6b8fbfe5c2263caa9e : SUCCESS in 1h 19m 26s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/af918e805fa84e719e980d36b13a22eb : SUCCESS in 1h 03m 33s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/5923f1d904114123b821c7117feb3b81 : SUCCESS in 1h 00m 59s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/03192510112245d998b9b229531b6d41 : SUCCESS in 1h 15m 43s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/33bbf98f3b894c939d7c79fd219f0d75 : SUCCESS in 49m 05s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/3a711a589e564668b7143ecddc76528d : SUCCESS in 1h 18m 16s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/1cc8b2dc38ef48dc9ece277a5dcbff02 : SUCCESS in 37m 45s","accounts_in_message":[],"_revision_number":2},{"id":"5df75139258218a7be2334faee3b9594e61f57ba","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-07-24 05:31:29.000000000","message":"Patch Set 2:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/802056/802056/2/check-not-vote/ext-glance-zuul/6c9ba54 : SUCCESS in 33m 26s","accounts_in_message":[],"_revision_number":2},{"id":"3b1cd45c88857dbbed3919f3ed25515fdd4bd8f6","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-07-24 09:11:33.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2a601355e447c73cee3153a0462947ed6f73a525","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2021-07-25 19:36:32.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"1459f82bb37dbc2ec362a4221c907ad9fc5bbb7d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-25 21:18:20.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/68064bbb1e39424ba0cb69a1aa04b5ed : SUCCESS in 1h 33m 29s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a7d61bc13d3a4a78bdd6be374c0ed6fd : SUCCESS in 1h 11m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d1444fa27eea419c8b03b80bae48627e : SUCCESS in 4m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5ce906abaa744d1895c21b997b6df81c : SUCCESS in 6m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1549ecf744cd43fb9db9c57828bb4f05 : SUCCESS in 6m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ec47313afd064f2cafda77aed1674593 : SUCCESS in 6m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07eb7b745b2543d2ace969a9b0780bc1 : SUCCESS in 6m 35s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/672d1e9325a141aabd5bcbb88af56cfc : SUCCESS in 9m 46s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/9811babb40d04f97b4d2e7213f9dddc0 : SUCCESS in 7m 37s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/a626f59078c44098aa2e71e01a491d13 : SUCCESS in 11m 43s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/bfed5411c0e44a0d9db382e57b849c4b : SUCCESS in 5m 12s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/b7b5e2681d3a47f1a8b995365dea2839 : SUCCESS in 1h 14m 25s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/2dd7af9a366e47de9b8571a003198538 : SUCCESS in 1h 18m 59s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/ed1cde96e8454b879891f76695574508 : FAILURE in 1h 38m 45s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/2146b8a2615744c28535e20e058e1b3d : SUCCESS in 1h 11m 03s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/73ab47f1eea647f2a2fb008fc0ec5e34 : SUCCESS in 49m 08s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/3f47b8d91a8a4de9a4799aa4a2c59cf8 : SUCCESS in 1h 10m 47s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/6e30b225c5494693ae5208baca72a1a6 : SUCCESS in 33m 37s","accounts_in_message":[],"_revision_number":3},{"id":"844f2d23310056b002198b3e1655fb06928fdef6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2021-07-26 10:16:51.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"c6de9abce81a5c76afb33c4d926a7d13e5afaa3e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-26 11:53:28.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/bbe9285c521c43daaf25887a3476a535 : SUCCESS in 54m 27s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2ca3b612a0c642ad86ea0551f59ebea5 : SUCCESS in 1h 08m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/010b581dfce14dbf9e2a68d92e60cd2f : SUCCESS in 5m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6f9292f48eb04ef2895a71fb54fc99ab : SUCCESS in 7m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dab5c4dfc7a944bc822cecacca65fa54 : SUCCESS in 6m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4e467b2f033b47d797c8df8aaba0bcad : SUCCESS in 7m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa74bd34989844ad9216eb0e0860367c : SUCCESS in 6m 51s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/71eb8bfefa424018a15f6f4961fe1c79 : SUCCESS in 10m 30s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/4986457093b24738aad40159d45e0cc8 : SUCCESS in 8m 36s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/6f70b750d7f44465ad12488f9f06a0f8 : SUCCESS in 9m 19s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/e4e2b0c3a8904b298f804e8a70615f2e : SUCCESS in 5m 47s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/c795015b102a469cbd1fca1a43ef43b3 : SUCCESS in 1h 06m 48s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/cc1ed4a7849245bf80aa32e5591d6de4 : SUCCESS in 1h 13m 16s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/42bc9dbe260a44269abbeecb148a3532 : SUCCESS in 1h 16m 13s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/f0f377c4f66c4b548c6404ced05de59a : SUCCESS in 1h 30m 34s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/95ec4458dd4a4edda7c322d2f045476e : SUCCESS in 43m 55s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/10238fd532ba4a8fad7b1f705fb99ec1 : SUCCESS in 1h 23m 30s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/7bf6e09ed48542bd8c569fcba38411d5 : SUCCESS in 37m 04s","accounts_in_message":[],"_revision_number":4},{"id":"48aaeb7a1b1621fc614af1135e23973a0ec6b78a","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-07-26 12:49:44.000000000","message":"Patch Set 4:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/802056/802056/4/check-not-vote/ext-glance-zuul/4c9b172 : SUCCESS in 40m 02s","accounts_in_message":[],"_revision_number":4},{"id":"0975260288baf5865a5988bcc9bff2117c8f5bdc","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-07-27 08:37:40.000000000","message":"Patch Set 4: Code-Review+1\n\n(2 comments)\n\nSome refactoring is possible in functional test.","accounts_in_message":[],"_revision_number":4},{"id":"7f735222e074877487506cee105538e1cd3ac347","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2021-07-29 11:05:07.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"eb813bb09a5e98a99cb3f0a0a72f5e8568468398","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2021-07-29 12:24:55.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"7d7c79055dd9e4c57a0dab3f755e23f20e30ad3d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-29 12:39:42.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/f01f5b919461477daed126a0211250d0 : SUCCESS in 1h 01m 13s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c2862b7717e84752a1e162c5f599be2f : SUCCESS in 1h 02m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b0ac54559db349c0bc43dbf477c8415e : SUCCESS in 6m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e839b96322a3418890c71bc5f34c50e4 : SUCCESS in 6m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/822dab8c9163462392b056d235360f36 : SUCCESS in 6m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fadb15d439764c91add68ab548a45e81 : SUCCESS in 6m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8ddaf6e5b0e4dc4a0477c7e532f76fd : SUCCESS in 7m 08s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f02681d262744995b0d5cfb593fe71cb : SUCCESS in 8m 58s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/1847efbd5a7d43adb175ff24b25cb5cd : SUCCESS in 8m 54s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/99614a2b099849a49344113100fb9716 : SUCCESS in 9m 36s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/adadad0fcd30432587bfdac7a67dfc61 : SUCCESS in 5m 29s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/50fddbd25e0f4e9ea3c2031074fc103a : SUCCESS in 1h 06m 58s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/e58ca6de32b345f8ab09e8340ac429ef : SUCCESS in 1h 00m 52s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/158c67724a634aada9c5d389ddf348aa : SUCCESS in 1h 06m 32s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/c55bda88abf342b581ae4220f9a93415 : SUCCESS in 1h 29m 26s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8f730611ed154a3f8f3ce13976f7433c : SUCCESS in 1h 01m 07s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/dc4dfd3360da480dbdd366fc014c2853 : SUCCESS in 1h 12m 36s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/df8e8cf5f98a416fa4b256c8fd58a2a6 : SUCCESS in 33m 32s","accounts_in_message":[],"_revision_number":5},{"id":"f537af8e6d7a574ed1f36358718f0265bac328be","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-07-29 16:26:13.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"255a1b65e5dfd72c7942cc712b5077a4390020ca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2021-07-30 11:26:59.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"0019b7d69e0bd2c35661a479caa644ee5507ef4b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-30 12:45:04.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/58bd2d2f0d92439990b2f87c318185c6 : SUCCESS in 1h 05m 21s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d5d479717d024a528fcdcf4402471bbc : SUCCESS in 1h 10m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5cb5eb4d5c2413799a7d28459ef22b9 : SUCCESS in 5m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4afd9b60a5084fea9e563bd2ba2a4725 : SUCCESS in 6m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f4fbb492f5084227bb96396760df87b8 : SUCCESS in 6m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/23a5e39c98fb4aed991d78113d743582 : SUCCESS in 6m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fbe5616f6c2641bbb3dfc5578e96b24e : SUCCESS in 6m 45s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3f66efbdd22c45508f998d71e77ed934 : SUCCESS in 10m 23s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/baee2c67b9c544109f1e06ecdb9f917b : SUCCESS in 8m 17s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/bacb8714a27c49a8a089b3307c84ff46 : FAILURE in 8m 41s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/3a59462382e843be89afff4a85ffc259 : SUCCESS in 5m 32s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/26c99ccce13d46a3878f2fa183798b7d : SUCCESS in 1h 08m 01s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/43510a12f76f43bf9c5263c98304e482 : SUCCESS in 1h 06m 57s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/be13f845435f48009d2e1e4091e9d3fc : SUCCESS in 1h 09m 27s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/682dcbdc120241bfbc2dc3c5eeeb9af2 : SUCCESS in 1h 13m 40s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c7324419504f4b3d8b0c2397c6005de7 : SUCCESS in 51m 07s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/f431b95eab254c8fbb8d9c23f0d57ea9 : SUCCESS in 1h 12m 59s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/cbb61c6b168a433280b45a31ad6286a0 : SUCCESS in 35m 09s","accounts_in_message":[],"_revision_number":6},{"id":"153dfa75423b0834fa46539bbe2a8e15068b4075","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2021-08-03 06:14:48.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"a9cebf534a03a40e5c82b265aba8f3340b9730ff","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-08-03 06:52:51.000000000","message":"Patch Set 7:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/802056/802056/7/check-not-vote/ext-glance-zuul/6e20330 : SUCCESS in 37m 04s","accounts_in_message":[],"_revision_number":7},{"id":"98407d94f399b26a72bbf593161ca7f1c683eda8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-03 08:20:46.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/be0ee602201646cc91558ef34ad1b77e : FAILURE in 45m 28s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6c191007cf92431e8201f48d74c2b311 : FAILURE in 25m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/524f245eeb0a4e818e769db60271edbe : SUCCESS in 5m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/35d4ff49b3f04572bfe0745e56298a6f : SUCCESS in 6m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/49eb12c409ab4a269b8ec51c8afb95e0 : SUCCESS in 6m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b91ae01993a14d3fa767392b287b2ade : SUCCESS in 6m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fbe6f6fdbd624f6cbbe59d98510234df : SUCCESS in 8m 27s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ba4050af93d54253b75d98604c420c60 : TIMED_OUT in 31m 41s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/d363fcf671474075a09017ed590b5f53 : SUCCESS in 12m 20s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/675fda81dcd8494d92423ce7860c273f : SUCCESS in 8m 18s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/464837692e084a1d8cadbe57590337fd : SUCCESS in 5m 22s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/72d372fc5a74421fb4d02b5440708fd9 : SUCCESS in 1h 12m 17s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/3b5617b5da8e42e0ad40517926ce6cbc : SUCCESS in 1h 15m 58s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/4f9f08211e79408293e4a234596a3c7a : SUCCESS in 1h 03m 59s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/8e4eff5c3f484fad832cfc35907ed434 : SUCCESS in 1h 19m 50s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1f3c0ff83b0341aa989eaa98131cf8d6 : SUCCESS in 53m 18s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/d6ceacb3ae1548fc8b1451a426bedc5b : SUCCESS in 1h 16m 32s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/158336bdb1be4c0c8fac70c8cd73692b : SUCCESS in 30m 47s","accounts_in_message":[],"_revision_number":7},{"id":"f1a4786de55cc1af2948714483357083c91200d5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2021-08-03 09:05:58.000000000","message":"Patch Set 8: Patch Set 7 was rebased","accounts_in_message":[],"_revision_number":8},{"id":"d8d87c9332e1fe7c0d60d13b01f2489b22d55e4e","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-03 09:28:46.000000000","message":"Patch Set 8: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":8},{"id":"72a1326d928e19e8944233ca89a302f4aed281c9","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-08-03 09:46:28.000000000","message":"Patch Set 8:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/802056/802056/8/check-not-vote/ext-glance-zuul/5d57067 : SUCCESS in 36m 24s","accounts_in_message":[],"_revision_number":8},{"id":"f198dcebfadb3f79778c5fce113d20a9bdc7e06e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-03 10:18:17.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/0cb13e686d0e4f36a039f275a85e77d3 : FAILURE in 40m 47s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a0b2ae6995554b17ba24a4a29dc777f4 : SUCCESS in 1h 04m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b48fdc93533444849e1622976e17c9e9 : SUCCESS in 6m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b5d4170d88104ddf91d357811c4e9d7e : SUCCESS in 10m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2034b03f2ed54d489564d4791ea5740e : SUCCESS in 6m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/64150547a02b4dee8ef2f62e39004a77 : SUCCESS in 6m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2cd171ce5fc14a459376d2dbd798bae6 : SUCCESS in 7m 14s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6d7ef0f8930841f0a20918a8baaecea6 : SUCCESS in 8m 42s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/e7b234ec85b2436eae7b7e9f69e2ded7 : SUCCESS in 10m 25s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/0dce99f239574005af21b5e4ca322dcf : SUCCESS in 10m 34s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/0998db03e6124f9b85a8abdb1aea7c47 : SUCCESS in 6m 07s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/a3547ff411b04d82a060d865d5ed59e7 : FAILURE in 1h 05m 45s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/db6bbae9d65d487980b029104b8d4f6a : SUCCESS in 1h 03m 19s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/600b1ef12c2243ff9328ecba5c3b4c89 : SUCCESS in 1h 09m 01s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/0abec75f5925494093da1fb9a6cb1e41 : SUCCESS in 1h 10m 19s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cb2dd3a1632f41a282620333c0b99d5d : SUCCESS in 43m 42s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/34aff207fcae494a832dd1b36a8b5afd : SUCCESS in 58m 31s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/303e124fc5b14bc7b612e18e8fb9817a : SUCCESS in 28m 06s","accounts_in_message":[],"_revision_number":8},{"id":"a7aa475c1210353cad6984d9955db4c25f8d9070","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2021-08-03 11:22:18.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"0c90bf258c5bee1c69352effe0ab70b8de975b14","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2021-08-03 11:27:09.000000000","message":"Patch Set 9:\n\n(3 comments)","accounts_in_message":[],"_revision_number":9},{"id":"2c85679b3d05c088eaf39f24dcd65b5d690645b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-03 12:57:45.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/436c842035514bb8aa16c5bcf7504b2a : FAILURE in 43m 15s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/de4a6766507144c4be91b1e117ec27c6 : SUCCESS in 1h 07m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/93c22487e6094f5fa258104d6c3a451f : SUCCESS in 5m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7eabc568402d4623869244b64879bc0d : SUCCESS in 7m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8f27a76fd0994ed5a97e5401abe038c2 : SUCCESS in 6m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/27cdedd83ee6418c86ebece7ec02f6a8 : SUCCESS in 7m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c39c1a915a3c4861aa76ba86799e46ec : SUCCESS in 6m 56s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9eda87ee0ac74ec592c8ce439fdb4b8e : SUCCESS in 9m 15s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/f822b46d4f2a4850bd150e2d35c93113 : SUCCESS in 10m 19s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/e2e39a3275b64216977005bddb336799 : SUCCESS in 9m 32s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/3facc81f3dac4f9f88c16ce2c6c6ce23 : SUCCESS in 6m 14s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/14b9e5219f374bdcbb4258969202df35 : SUCCESS in 1h 09m 31s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/507178ef22bc4d94b99cfafd225dd14d : SUCCESS in 1h 02m 59s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/a0a5fca7f9f549518c7ceb3c071e9d8b : SUCCESS in 1h 11m 47s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/a22a83322f7547bf818d14cd05e1b90d : FAILURE in 1h 09m 18s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4ee92100343849b0b287841f187f9a84 : SUCCESS in 52m 39s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/48701614587940bf91a4955c38ff04dd : SUCCESS in 1h 24m 14s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/25c5c7d04eb74fb19f68f4b2b13e4e66 : SUCCESS in 32m 25s","accounts_in_message":[],"_revision_number":9},{"id":"5949ad9f481d59ac5b6f4743b4df1f0d7032d390","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2021-08-04 05:24:56.000000000","message":"Topic secure-rbac-tags removed","accounts_in_message":[],"_revision_number":9},{"id":"ab3fc9428c6841193cd58f45ebba9118a458756e","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2021-08-04 05:25:03.000000000","message":"Topic set to secure-rbac","accounts_in_message":[],"_revision_number":9},{"id":"862e852e7f2bb146bd174c226491ef82e31e430d","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2021-08-04 05:25:16.000000000","message":"Patch Set 9:\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"e774917f110b835c4164e734e391ed8e514db4de","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-08-04 06:40:01.000000000","message":"Patch Set 9:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":9},{"id":"b36d54a55a6365809a6a6bd340ec8ac2ffb55502","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-04 06:47:02.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/537e870e8d4b4c0399c7269892863248 : SUCCESS in 1h 13m 53s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a249e4986c5248fe9a3958fe9af82fd7 : SUCCESS in 1h 07m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd17ef3bb1fd42758150bfcb2bb80fe0 : SUCCESS in 5m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c046abe1108246dfb40568b135922d48 : SUCCESS in 6m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/01680ddc934f41979803cc3e5372af1c : SUCCESS in 6m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a1772b518570421c8ff66f95007d52b9 : SUCCESS in 6m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/210cc4024f5e4de6ad2ce5a7497d470e : SUCCESS in 6m 32s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9272216240f74da3a282c78b65912e1e : SUCCESS in 8m 51s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/a372186990ea4c4f8668c38d9a393f5b : SUCCESS in 8m 32s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/e87f79cf2a97431bbec341b2cf5f0a26 : SUCCESS in 8m 42s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/a0f88bfc34bc4c4e9b12bf62e0090adf : SUCCESS in 4m 56s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/409eae67072341efbee6aa242c9476a9 : SUCCESS in 1h 10m 03s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/2abef2a8629a4f28938f7f226bd6236d : SUCCESS in 1h 08m 26s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/ef48ef6761b64c919bf316c62fe96049 : FAILURE in 1h 00m 34s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/b15bcb5b146b4a8280eccfe7436499fe : SUCCESS in 1h 17m 23s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9541f96566b04cb2868797a7ca76d661 : SUCCESS in 51m 17s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/614a9529452d42538fc5d1f5e3df0d0d : SUCCESS in 1h 11m 19s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/3b9d416e9331418c87328736114a7d59 : SUCCESS in 29m 55s","accounts_in_message":[],"_revision_number":9},{"id":"baebc79405e208c41fc4f569b57fc61f0ff97698","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2021-08-04 07:02:46.000000000","message":"Patch Set 9:\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"e4f9b7bfb265c7e5fea3ed8385b9f083fc513be8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-04 08:42:28.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/34865dabc43a4169813ed049ddda4ce8 : SUCCESS in 1h 03m 18s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8a1b966ca2f241cb8d33bfdfd0b89721 : SUCCESS in 1h 01m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b08c9d1768ec4fbf917385b5b59c43ad : SUCCESS in 6m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c0ae5c2e78ea47b8b025ca51cb89f25a : SUCCESS in 6m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/18dfbff0f26a473aa200a7b9efdaa5be : SUCCESS in 7m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/132b970119ae4ecf98c9699ce4b57ee4 : SUCCESS in 6m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5bb4c3ec78974e9790aaf71d1d4be20b : SUCCESS in 6m 44s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/52eaf03b8e984445a83fdb0063ea77f7 : SUCCESS in 9m 14s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c2be42d33aae405ab11a960fb0dbe9cc : SUCCESS in 8m 32s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/d9b0aa82516d4c7aa28b8c8eccc63bde : SUCCESS in 9m 26s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/23c7601250e14cacae5ee29498d44bea : SUCCESS in 5m 23s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/01c521dfe81d40988a7ec164c0619c66 : SUCCESS in 1h 09m 19s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/0c4979fdb06a4aa7bf028bd5a336eeb0 : SUCCESS in 1h 09m 41s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/6f7dfe3c0ad447f3a34ddba00106c8f0 : SUCCESS in 59m 42s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/d898285d432f4629b1bfed3119823517 : SUCCESS in 1h 37m 47s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5d8c8c32d7b449d0ba6e0260208b9083 : SUCCESS in 46m 28s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/8d12250fc58647128b47068591b65b53 : SUCCESS in 1h 14m 29s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/2a9d4b6d9664453dbf22f44bbe6cf77b : SUCCESS in 38m 07s","accounts_in_message":[],"_revision_number":9},{"id":"c987c758edb44b27ebff11cbb76692a2e23b73a5","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-08-04 11:46:05.000000000","message":"Patch Set 9:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/802056/802056/9/check-not-vote/ext-glance-zuul/23b8101 : SUCCESS in 38m 54s","accounts_in_message":[],"_revision_number":9},{"id":"c402d22f1d79620c85e3e824a193f5e73e148d48","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2021-08-16 14:07:12.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"5ffc20099148f1cc93c6703418f5f0e00027ace6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-16 15:25:19.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/c1d963b0464644e7910873fa29e48c5f : SUCCESS in 1h 04m 11s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/448cc61ce4374681a428eab27ff22fb0 : SUCCESS in 1h 08m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3299a656933b42d0a7a12c544b17d76d : SUCCESS in 6m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f2d735fe6d664810a419f1a5c1147e77 : SUCCESS in 6m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0d6b5c5507e24704a35bb7b4cabeb632 : SUCCESS in 6m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3cc3257b6c5d41a2b784702814bb139d : SUCCESS in 7m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9112248799c34c4d8b98f13b704f4987 : SUCCESS in 8m 36s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4349384bfff747699845c01b67007340 : SUCCESS in 8m 47s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/319bc28ff2684d4592c11a1e97616404 : SUCCESS in 11m 04s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/6d748297f39240139c163a0b1f135158 : SUCCESS in 11m 02s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/fb10fc341b71445fbb002bc9dfb46705 : SUCCESS in 6m 27s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/53eef2fc7b384da6a528e1ee57fad45e : SUCCESS in 1h 10m 32s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/3d7981ddc41641a8a6753491018bce27 : SUCCESS in 1h 16m 06s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/23c157dc1d414e259a8717bc782f6482 : SUCCESS in 1h 12m 11s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/4ea2d1411d5b4080811956a4f08288e8 : SUCCESS in 1h 14m 49s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/882504d42a994809a165ba0841cbb79d : SUCCESS in 43m 47s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/569bb1ba2fd1483c82c128718cdf086b : SUCCESS in 1h 16m 20s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/e2e6f77cce3444c18053c7542023a720 : SUCCESS in 29m 28s","accounts_in_message":[],"_revision_number":10},{"id":"8a868657277f792d7173263facb497d7425784f9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2021-08-23 08:53:50.000000000","message":"Patch Set 11: Patch Set 10 was rebased","accounts_in_message":[],"_revision_number":11},{"id":"35c883e503dd5f2635bd53fead52c5fbe5fe7111","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-23 10:14:30.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/7c141005300147249422067f166a2fc5 : SUCCESS in 58m 05s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b5ed374fde5b42cf97268ddd28676f6e : SUCCESS in 1h 07m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/535985bd59b74a68bba1f0adbca6a44e : SUCCESS in 5m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/26f986a59eb94e1e9c3403dc99341336 : SUCCESS in 6m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f1d213a6e1a6470db86b45289e23deac : SUCCESS in 6m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2fa6ea54f6a1471cbd35faa48f9dd728 : SUCCESS in 7m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51ea7e963e31429c85c57717daa4c821 : SUCCESS in 6m 52s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0bc6b34b116c4709b5f5dcb26a00aaf1 : SUCCESS in 10m 12s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c35f8a9f302f423cae5d1bfb3954003c : SUCCESS in 10m 44s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/7f6d94b4ffb14d298dab8f739ba95d01 : SUCCESS in 8m 24s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/7d66ae2e69834d129c340035bd867839 : SUCCESS in 6m 11s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/d7501ae89ec84bf7a931c78bc8993fa5 : SUCCESS in 1h 09m 31s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/7ffa5cd04c5e4923b3ccb74c7150c17a : SUCCESS in 1h 09m 17s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/4ca17611330c4d5493e2661db3993a90 : SUCCESS in 1h 11m 04s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/b7f2a478fab647869a7b152d8af92693 : SUCCESS in 1h 14m 52s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0d03439657004881876ab5eb322d1a01 : SUCCESS in 52m 51s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/d536088a206b4283a3b29d3f5f3e9067 : SUCCESS in 1h 08m 27s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/c122350661114718b1828fc858426822 : SUCCESS in 32m 04s","accounts_in_message":[],"_revision_number":11},{"id":"0ec2710e78660dd7acfe08ec51fbc947b2e59137","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-26 04:02:48.000000000","message":"Uploaded patch set 12: Patch Set 11 was rebased.","accounts_in_message":[],"_revision_number":12},{"id":"a1639dfa603ba27fcf918103962e6c6047d7e682","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-26 05:28:58.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/6805405065234406b0464518c2e0903b : SUCCESS in 51m 44s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3227b60509f64b2ba132fafee18d388e : FAILURE in 54m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed51486fa803405fb6cf10b4dbe7eba6 : SUCCESS in 6m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ab54698b3ff643658d0e40d15e08d9e1 : SUCCESS in 6m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/80734125f2ae4000af600fc43d47f338 : SUCCESS in 5m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3f9303128005474e85261e175ad02d3d : SUCCESS in 6m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c6f35acf4ac4c2db0263612caa81844 : SUCCESS in 7m 00s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9249734a3e0646b49e5c72397ee27906 : SUCCESS in 9m 22s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/ae032512ba414252831a385ec456e90e : SUCCESS in 9m 06s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/a6ea4d56a2684dfda47379909d96b64f : SUCCESS in 8m 51s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/e025252c3ad14dd8bde139c2166e66dc : SUCCESS in 6m 19s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/b603b9f5dc97463ba4b300e28a8f98b2 : FAILURE in 56m 43s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/6c0a5332fb7f420db018c753e6134637 : SUCCESS in 1h 18m 11s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/8f20ef5bc1da4c44a81ea527ac9d441f : FAILURE in 47m 28s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/b3d67801128d4939ab26b342acf432cc : FAILURE in 1h 06m 38s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/883d19838fc14a06bc342e12dca40a73 : SUCCESS in 44m 39s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/4747c0bb8abc4925a26f558d1d1215d8 : FAILURE in 1h 22m 42s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/23edddd37697410f8fdbc5f7725aa52b : SUCCESS in 32m 11s","accounts_in_message":[],"_revision_number":12},{"id":"f27277c1994add65bf259d1942601ba46d27f55b","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-26 05:34:37.000000000","message":"Patch Set 12:\n\nrecheck","accounts_in_message":[],"_revision_number":12},{"id":"b91ca53df59ef3407625e9e35469e847d3c22145","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-26 06:55:41.000000000","message":"Patch Set 12:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/af683db6baa64004bdaff2566c59bf10 : SUCCESS in 1h 03m 11s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c4d416d3b5534c41aa039748e46c7dba : SUCCESS in 1h 13m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/daa1ffa074f7493d96d35a5e95a5334d : SUCCESS in 6m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4e7c91310c4442a28a13dc2736fbe680 : SUCCESS in 6m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b9974da893da48069158c33fe0c3e4c2 : SUCCESS in 6m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3ed6829b3cf5481883fa67e01bbf5737 : SUCCESS in 6m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ecb914692699452da94d54bedbca5e36 : SUCCESS in 7m 22s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b66af41fb37d47ecb5803c5efb2ae9f5 : SUCCESS in 10m 44s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/151e1b583170483d8caf13d10bf4d490 : SUCCESS in 10m 21s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/c35a73795e724946b8a7500567c2f49d : SUCCESS in 8m 16s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/a8e21312f39e412fbc00b80b1117703d : SUCCESS in 5m 09s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/3daae33a7b9643339cc9c897a8b78c58 : SUCCESS in 1h 18m 26s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/7f6f645b769f43688686519953aa992f : FAILURE in 1h 07m 51s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/5cd3ecbac55c493fa7a54f70ca791f74 : SUCCESS in 1h 06m 08s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/8da7f85e5bd44954b08f7203ee5a7278 : SUCCESS in 1h 12m 45s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a40ea5c6d52b40d09e41cea454be5896 : SUCCESS in 1h 00m 00s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/c37a85fc1e8f4bfeb93885699409025f : SUCCESS in 1h 14m 55s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/b0249683574d48cf9a47a2158d24686a : SUCCESS in 34m 29s","accounts_in_message":[],"_revision_number":12},{"id":"970b129d4cdf40c268c67747237e35334bbb41ea","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2021-08-26 10:51:47.000000000","message":"Patch Set 12:\n\nrecheck","accounts_in_message":[],"_revision_number":12},{"id":"8a090ca9c5232bab314889ad0696ea659d1c1cd3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-26 12:12:19.000000000","message":"Patch Set 12:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/19b5fc29ff2b426d9c3fba28842fcbd2 : SUCCESS in 1h 15m 33s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/126c1833ee3f4a968c961bb33fbf885b : FAILURE in 1h 02m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8201861c617d4c5796de1e0fc5ab6611 : SUCCESS in 5m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/30604949ded14e2383701e19acb311cf : SUCCESS in 6m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ef4ae9651d9d41248227e99d6e344765 : SUCCESS in 7m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b4f91e56e9604a7ca8fc38ce9ac85fc9 : SUCCESS in 6m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7554561f1fc4f6287e55dba6754cf7b : SUCCESS in 7m 13s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f09bb583d7b34e40b335936119a106b3 : SUCCESS in 10m 28s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/113e250484e04892a82eb2fc9f7f5c7d : SUCCESS in 10m 52s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/670e66a4889245be90b590566bfe168f : SUCCESS in 10m 31s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/2bb0ac5bec2743fe87aede5047d0ea0a : SUCCESS in 8m 29s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/71902db558bb427bbc37c34b91c28f08 : FAILURE in 1h 13m 17s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/6f89c9ba3b91460c932af85bc38b778d : SUCCESS in 1h 06m 30s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/de3a00ae7b4c4ec395f2abe11ca7cd7a : FAILURE in 59m 55s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/d2475354761f4beda663e721ed2d6c96 : FAILURE in 1h 10m 21s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/682e4a26315d4f95ab28242f49bfb235 : SUCCESS in 52m 47s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/f6cb13ec665a4ecc8506696af7b23a97 : SUCCESS in 1h 14m 34s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/e3813814a47b43158c24b4f31b64a54a : POST_FAILURE in 23m 47s","accounts_in_message":[],"_revision_number":12},{"id":"d9585ea3cef3b4474c19cb8bbe5bfd3e83093dbb","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-30 15:49:04.000000000","message":"Patch Set 12:\n\nrecheck","accounts_in_message":[],"_revision_number":12},{"id":"80a25691bbc2e4204399fd98ea5a8f2910b6bccb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-30 17:22:42.000000000","message":"Patch Set 12:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/5cf18b88e07d4e42ba479620f9c73361 : SUCCESS in 1h 00m 14s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c624f5a396ff4601a8e6bfcb7aaa8f0b : SUCCESS in 1h 14m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9bcf5c0c7353424da15356da3a75a70b : SUCCESS in 6m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f233de09283044a9adec3d2844f90fc4 : SUCCESS in 6m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d0a39433d3714416bdeb45e198d35597 : SUCCESS in 7m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4505e1c0bca94632aeec5051eb755593 : SUCCESS in 7m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/004e64609cf041568edde75244215599 : SUCCESS in 9m 36s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0a669d4d5d8048839e77a911e766fb84 : SUCCESS in 11m 04s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/2c2b13d257514bb8b2c196ea50a1f5f8 : SUCCESS in 9m 41s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/f68ab49e607047c68a018575b8e305ab : SUCCESS in 9m 32s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/441157547ccf476fa6a681f360518f5a : SUCCESS in 7m 17s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/c5a75084c58b4421bc5f3fc5eecdb382 : SUCCESS in 1h 02m 10s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/b5c8df2d683d47c5943338b627a80a53 : FAILURE in 1h 25m 06s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/787945073c914b8a8ef8f4ae56a907e0 : SUCCESS in 59m 55s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/3a578d81b0ae46559e373a16ed4abeaf : SUCCESS in 1h 20m 47s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4e7d9b55d2e94634b42ee4de353e7cf1 : SUCCESS in 54m 21s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/d0a5f47e3d0e4395b68a63361dd2b2ac : SUCCESS in 1h 07m 42s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/ac1f2187599f4309aa5fbcf075fd5251 : SUCCESS in 36m 19s","accounts_in_message":[],"_revision_number":12},{"id":"ed01100935ae68b81546748aba904665ba087efe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2021-08-30 19:47:26.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"6174df9efb3dbbf0cea105e943b98babdcc9238a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-30 21:12:39.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/ff6e872226f4470e83d1deae6cd45a70 : SUCCESS in 1h 04m 12s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/379a2761033242b29c110d9d08b45b4e : SUCCESS in 1h 07m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a11a0bbd847c4c16b858fe95cbc1a69c : SUCCESS in 5m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/234e876c359a48eb88dff3771e180c56 : SUCCESS in 6m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9b2dc09c08614e2b82aced2656552558 : SUCCESS in 6m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7e30e4f8729040e6963bd2ad34078ed7 : SUCCESS in 6m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3391fc43d9740e8b5abba2ddd9685e8 : SUCCESS in 6m 38s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e3cea66b6f3b4fa08e7a6fcbc382fdcc : TIMED_OUT in 31m 57s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/565785272a4643a388e2a67338c3f9e2 : SUCCESS in 9m 15s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/25972760b6884764a8c8bfba6dd615b4 : SUCCESS in 9m 12s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/5da08d7341d04960b45ba43076153dc1 : SUCCESS in 6m 59s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/8fe68a049eeb43a6b9415a81ae2eddd8 : SUCCESS in 1h 20m 37s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/66f49c41c50f4f1e9c0b383baa621249 : SUCCESS in 1h 06m 48s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/3306e7d693764ec399edb6f09a9992a0 : SUCCESS in 1h 02m 53s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/e396012b2fa5437e986e89a27c7c85b1 : SUCCESS in 1h 15m 27s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/63347c5cf0e74edd98c6a73607eb94fd : SUCCESS in 48m 50s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/edf61746e99a4ed9a0fc8f3949b8de5f : SUCCESS in 1h 19m 28s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/84b15251fee94489886d1640fa5f90ea : SUCCESS in 29m 55s","accounts_in_message":[],"_revision_number":13},{"id":"cc86bf208927f1b5650acb1ab95d0d85668ad16e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2021-08-31 06:08:04.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"8e1b5a827a8f3ee97ddd1be3522a26f742925a11","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-31 09:02:26.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/dfd2ac41c34b4fd4a7db96456f255de3 : SUCCESS in 1h 27m 42s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/dbb2ea0822c54666ba7e221a8dbaf9a8 : FAILURE in 1h 30m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1657266116104c7b96115d56a50f87d4 : SUCCESS in 6m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1c190b36231848e7a71e94b4b759b4b9 : SUCCESS in 9m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6d3635e217f24f76b35c1d0b6a543d4a : SUCCESS in 8m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/55b28e6e836b42dea8452b53f943a4fd : SUCCESS in 8m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84ad4bdcb51b4a4faf64d9a1d0fcfdf0 : SUCCESS in 8m 34s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5074fbfd823e4b2f9d017336b15f605c : SUCCESS in 11m 07s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/cc06663bc42d4671a41a4f6180b628bb : SUCCESS in 10m 56s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/c335a4d87841400b86732c5f0c3afc27 : SUCCESS in 10m 24s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/7e905e3a2f694396b7cf77a56f387b66 : SUCCESS in 6m 59s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/8ed600ebd95d41698563ea433552b6a3 : SUCCESS in 1h 32m 29s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/d8886caed52949aea035726872331cee : SUCCESS in 1h 08m 04s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/192f977b33c84d9ba2be52b4c0230218 : SUCCESS in 1h 22m 08s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/a8cc9216b89749659b45489ea62a4dbf : SUCCESS in 1h 17m 16s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9547942dad8c45bd9024056d69d16f43 : SUCCESS in 52m 41s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/4d52427886c34d0eb4c63de41b065d92 : SUCCESS in 1h 32m 42s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/a7236f301a104028b53c78d9210d17ce : SUCCESS in 27m 45s","accounts_in_message":[],"_revision_number":14},{"id":"f4fd215cecebafe734a93cdc4ffc55ca3266a2ae","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-31 09:07:16.000000000","message":"Patch Set 14:\n\nrecheck nova failure","accounts_in_message":[],"_revision_number":14},{"id":"c8025ad5c3621079ede464d1de3f7bda2f453234","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-31 10:38:28.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/fdb1a8cfd0b64b2592b185340a25e5f7 : SUCCESS in 1h 05m 13s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7480cf15a6984885b0e1f9e5755f508a : SUCCESS in 1h 13m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e03c70d278d48308cad37ad4bd3598d : SUCCESS in 5m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a5cca280b1af4045b945a400d629d6e2 : SUCCESS in 7m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/05875d75132a4f3a8fd858054fbd4b4b : SUCCESS in 7m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ebf03b2567904a16a6cba126da8521ce : SUCCESS in 7m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80493c08598c4d7b8bb57986a8950c3c : SUCCESS in 10m 46s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/35891512ef804c4f8c4f5852201c08ab : SUCCESS in 9m 18s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/8ea0dc1e92c4467fa14f331008d22b4a : SUCCESS in 8m 58s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/e87e778dfb6f4914a278145d40fc6d08 : SUCCESS in 9m 56s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/ad645cce7ead4725892047545924855d : SUCCESS in 5m 24s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/923b191acc0d41f8ad1226e9a1b93c7d : SUCCESS in 1h 19m 16s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/4a18b0e559724c2aa6f327e63d838cdc : SUCCESS in 1h 07m 27s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/ae4fba6a7f6b400da2665dcb1e3909df : SUCCESS in 1h 10m 14s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/acfd9704c041458abfe544810dd065b1 : SUCCESS in 1h 25m 22s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6ea8917d0ee54dc581eaa045afcacdaa : SUCCESS in 46m 11s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/50f25ef01f72495da0c6d942ba8ff62e : SUCCESS in 1h 13m 56s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/7d74b3327e77442fbecb1f9fbfac867d : SUCCESS in 30m 06s","accounts_in_message":[],"_revision_number":14},{"id":"470eabbce0ce5ceb9236b0f9a9792085d80c7550","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-08-31 15:08:07.000000000","message":"Patch Set 14:\n\n(2 comments)","accounts_in_message":[],"_revision_number":14},{"id":"1768b9784cc5cf8f8a094bc355b7d3fcbaeed138","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2021-08-31 20:04:52.000000000","message":"Uploaded patch set 15: Patch Set 14 was rebased.","accounts_in_message":[],"_revision_number":15},{"id":"5b7b32e070548aa9ccea80889b2f9ae555551c65","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2021-08-31 20:20:51.000000000","message":"Uploaded patch set 16: Patch Set 15 was rebased.","accounts_in_message":[],"_revision_number":16},{"id":"c7e07a56cd60ebd465dc4b1b821293693ab57f59","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-31 21:36:49.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/2c012da2127a438890620077732814ce : SUCCESS in 1h 02m 57s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7dfe0673d2394fc982983b5d41cfae47 : SUCCESS in 1h 09m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/53671f78aeec4c1a9abcc43b7957b44c : SUCCESS in 6m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4775a42aedfb4d70906a9e0212797e0b : SUCCESS in 7m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ca8ddb15ef5d4a06a72ea7c99214057e : SUCCESS in 6m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a354f1e450bc445a8497535396c7f5a7 : SUCCESS in 6m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81796ca3c2974b6c9dbc8a3c89ec4ba3 : SUCCESS in 8m 04s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/60c46939f87c4b73968a649a57195405 : SUCCESS in 9m 05s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/7f4929fdd23a41d0a0aaf5f04aac142c : SUCCESS in 9m 16s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/b8f64159ee2e42398c403a6a0563c837 : SUCCESS in 9m 21s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/aa3278276ef94d82a485ce305a852264 : SUCCESS in 6m 31s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/2f47680a49304508be90c4a426401220 : SUCCESS in 1h 02m 35s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/f7832469d70d4247bb482d26cd681018 : SUCCESS in 1h 04m 07s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/e9c7cbf0f8a7442ba6270660c074b830 : SUCCESS in 1h 05m 19s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/83ccd09de1ff485fb2a731ed6d0b00dd : SUCCESS in 1h 13m 44s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e99c7c2734fc4ee1a152da3f3011859d : SUCCESS in 48m 12s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/67256b26139641dbb3383a0028e83e27 : SUCCESS in 1h 05m 41s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/5cef865d7b6a45d2b9cfc3612863a6d5 : SUCCESS in 31m 42s","accounts_in_message":[],"_revision_number":16},{"id":"2372bce9f2b75ef25042cdd658fa272c99c1fa69","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-09-02 09:13:08.000000000","message":"Patch Set 16: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"9b74558f0740feecc513fe403dc0801f579087e2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-09-02 09:20:55.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"eb1cff49cc30282b9a9331fe5fe672c828c985a9","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-09-02 09:41:36.000000000","message":"Patch Set 17: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":17},{"id":"981e2ff82a14feaa195f64138f7f138fd8cb0f06","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-02 10:55:19.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/3b53b6132d56421b8b59cb4d9bffac32 : SUCCESS in 1h 04m 45s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c8acede9e4734ed1afd29fc963a7b6d2 : SUCCESS in 1h 19m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/618a6f4dba8342bbac0b433c92267d90 : SUCCESS in 6m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1701e3eb9d264216a542ab4e6f1cdab2 : SUCCESS in 7m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fe96c2567533424e99687e5855315d2a : SUCCESS in 7m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/be5289ac61144b6691dda45c3d729650 : SUCCESS in 7m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c273f350e01949d4876f0e657a540b21 : SUCCESS in 8m 56s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1863f18218c141a98822e7d2229a906e : SUCCESS in 14m 50s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/0e2b5f6541be479f8fdf288d6b97da77 : SUCCESS in 14m 41s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/9f604113010347cb93b6d5a90378b2c8 : SUCCESS in 13m 06s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/8fb1b0ce2eec491abe8f0c4cb0553d30 : SUCCESS in 7m 32s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/56d1e02eebff422ca3e8cd65dbc2912c : SUCCESS in 1h 05m 39s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/fff4ada647f04656a17311d5aa567ef6 : SUCCESS in 1h 13m 55s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/387949a85d6e4ee189abc4115425e2a0 : SUCCESS in 1h 09m 39s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/afed75687f3c48ebbb3fd6fb35f29e48 : SUCCESS in 1h 28m 15s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9fbbfc36d5594756a7f2f80268bbb8fd : SUCCESS in 52m 44s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/087148828c0b45fc820ef81969037d85 : FAILURE in 1h 14m 44s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/c6e752a1a5ee4bf38a2528ed4f422e6e : SUCCESS in 33m 46s","accounts_in_message":[],"_revision_number":17},{"id":"ef97348118103eb04805bcaeee9d8a3716fb7bd3","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2021-09-02 12:18:20.000000000","message":"Patch Set 17:\n\n(3 comments)","accounts_in_message":[],"_revision_number":17},{"id":"93988c37fe6024db629bda0c30c8cf61e76381e1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2021-09-02 12:18:28.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"ce1602a122fbec5b99744abd4d21231c755ff357","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-02 13:36:50.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/5026bdf335ed47c8a8760c248283d523 : SUCCESS in 1h 00m 49s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a0b67370185c458eb4932357aeb3b100 : SUCCESS in 1h 05m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce9846524adf4e50a59bd14922a021f1 : SUCCESS in 6m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8ae175a2edc645109b984f52f37f18ec : SUCCESS in 7m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/198c2e0a76f0472a99e7eec1a14c15e6 : SUCCESS in 6m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d3578e5d376b42398ca7a4bfb8a94e01 : SUCCESS in 7m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5756a1d2fba04e0d9b1683c4293c0750 : SUCCESS in 7m 22s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f8ba186d193949fd9084fe10d07ab8d6 : SUCCESS in 9m 22s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c57f85d6f4f7499ab0aeb74a009ec097 : SUCCESS in 12m 55s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/dca30155c66a4f72918b9583c3f7592b : SUCCESS in 9m 18s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/e53a598be1724ab29619bc70f052ad14 : SUCCESS in 6m 46s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/6c7fdf6a661645f48a95c7dbdea17242 : SUCCESS in 1h 10m 51s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/ffdbcb8589f4460285c19fbd8dd01853 : SUCCESS in 1h 13m 14s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/8331e88990ce46958e857c1091cc863f : SUCCESS in 1h 10m 16s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/ab60116fbb694ababda3191173cd5e73 : SUCCESS in 1h 10m 23s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f03d61f4a52a47df84fcc8b0450fbb60 : SUCCESS in 53m 19s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/635ea7207a534f598bf986a3d5d27953 : SUCCESS in 1h 08m 34s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/4f3e0ba9416a43abad4961f36841ccc7 : SUCCESS in 33m 14s","accounts_in_message":[],"_revision_number":18},{"id":"a234aa1a7e8b468e7e311cf94ea4016e3a2a52fe","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-09-02 17:00:33.000000000","message":"Patch Set 18: Code-Review+2","accounts_in_message":[],"_revision_number":18},{"id":"bcd8eb815f19627b7b78cae1f2cfa8a861daf874","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-09-02 20:01:51.000000000","message":"Patch Set 18: Code-Review-1\n\n(2 comments)\n\nA couple small things in the tests","accounts_in_message":[],"_revision_number":18},{"id":"013cac7816b6d899b9297c345330ae75813cf3ed","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-09-02 20:42:31.000000000","message":"Patch Set 18:\n\n(3 comments)","accounts_in_message":[],"_revision_number":18},{"id":"494beefb655286cdde787dc536eafd4ae88557e2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-09-02 22:03:30.000000000","message":"Uploaded patch set 19: Patch Set 18 was rebased.","accounts_in_message":[],"_revision_number":19},{"id":"2e9b7180eeafc2d5c3c30490da762408bc36fc5a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-02 23:25:50.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/0d409cd0cc5b49d3aca20028c86233d3 : SUCCESS in 1h 00m 59s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b29d2d210c1044c0ae7b63dee778ca8e : SUCCESS in 1h 03m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d885f4f271fb4daf9fd4e3cdb785d324 : SUCCESS in 8m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2cfb7cd2dafe45f79eab62d934cbb8b2 : SUCCESS in 6m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6f7a71b41979417d8347967a70c076ab : SUCCESS in 6m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6ff2cd1ee0f945208ffde95618e38c9e : SUCCESS in 6m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cfe894dbe5944c569abeefa13c18ae95 : SUCCESS in 6m 29s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/98244013be62442a8bd5d3d071dd2b26 : SUCCESS in 10m 27s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/ecd33197fff8463c81210e9add2c6727 : SUCCESS in 10m 09s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/d1a4eaacdc754a7eb763c62810656966 : SUCCESS in 11m 55s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/e26a61ceba0d42cea37aaa384e3bbcca : SUCCESS in 5m 30s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/922249e7ffe14137bff4ec2c93e50d16 : SUCCESS in 1h 06m 49s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/15071da164f447bb903d8938f885e5c5 : SUCCESS in 1h 02m 55s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/7eaf637966354cc1a3d006ee134e3674 : SUCCESS in 1h 10m 29s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/d51fa78d4b4c4a2385849c75b670ec2b : SUCCESS in 1h 12m 09s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b0046f0c6ce34861b249fcb82f400354 : SUCCESS in 47m 25s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/e787178efea346c28fe429246893532f : SUCCESS in 1h 13m 43s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/990f01ff24aa4afcaccda8aad63c705e : SUCCESS in 30m 13s","accounts_in_message":[],"_revision_number":19},{"id":"e540fd5249cf80242acf67988955c65de8084062","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-09-03 05:45:31.000000000","message":"Uploaded patch set 20.","accounts_in_message":[],"_revision_number":20},{"id":"f654de3799f8a4f2e89089dd548de6f094da603e","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-09-03 06:03:09.000000000","message":"Patch Set 20:\n\n(4 comments)\n\nThank you for reviews, added missing scenario for list in new PS.","accounts_in_message":[],"_revision_number":20},{"id":"5bdbb8a1485c31f831c928f57a49ee28e67d41bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-03 07:06:25.000000000","message":"Patch Set 20: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/a17a5d21407c4a9e9bda782d267273b2 : SUCCESS in 1h 00m 18s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/cc4b52cfe86e463d86f4aa15065fbd15 : SUCCESS in 1h 07m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c36481482fc3459dbbb786adfa6396eb : SUCCESS in 5m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5d3d3e53428e43eda7239e787bab8da1 : SUCCESS in 5m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3e1b5f52c073471ab225cac76ff45bde : SUCCESS in 6m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b7d44510c0164bce9ffcfc3429c52829 : SUCCESS in 10m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/61f6256d43b140559a7d978436a95710 : SUCCESS in 8m 56s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3bc477b816c4455bb37084e0816c85ce : SUCCESS in 9m 22s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/59853c4859bf4844a187f5e35e7b07a3 : FAILURE in 11m 09s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/ecb895cc91a943e68b292a2a3d5c890c : SUCCESS in 9m 05s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/1cf18546965c4a2cb607cc6f5fa62c70 : SUCCESS in 7m 22s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/2838c87f206b4676be3a441ab7ae3670 : SUCCESS in 1h 15m 10s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/539c3056a47f46cc93c45efcd662cff2 : SUCCESS in 1h 11m 53s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/b95c4bf2476448a88cb521be1a6b8530 : SUCCESS in 1h 08m 09s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/dc6d86c2b909464d813087283717b3c3 : FAILURE in 1h 10m 12s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bd86a549b9514ff4bcd517d99201bd5b : SUCCESS in 47m 12s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/21f3325411404c26a899a2010c6d3320 : SUCCESS in 1h 06m 27s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/4ad006f999d144a186cfb420ec44ab02 : SUCCESS in 27m 45s","accounts_in_message":[],"_revision_number":20},{"id":"67bc9914c69ebb58ed4172728065815b196081ad","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-09-03 07:09:04.000000000","message":"Patch Set 20:\n\nrecheck","accounts_in_message":[],"_revision_number":20},{"id":"9f01765f8955e9bd6a1795d25700892b31e57953","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-03 08:49:37.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/759f3016e2e643a8a78b8cd474a1c350 : SUCCESS in 1h 03m 52s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/45c87ca28a994921a84c576bc297447d : SUCCESS in 1h 11m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc041bf24bf74f60b0eba9c0d168c183 : SUCCESS in 7m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a26431adf7b141d09ff70d011193165f : SUCCESS in 10m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e643180d0bd84fd884d2f6b731912fac : SUCCESS in 7m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e4e553f482cb4b0d8a57407706013f01 : SUCCESS in 8m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09560fa9331e499f9ef22935f45daf66 : SUCCESS in 9m 27s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/068f9dd82ee74af8bbe52045ae20e170 : SUCCESS in 10m 02s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/e2e5943d44d2457689985f4911f6f5bd : SUCCESS in 10m 48s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/9ba07ffdb50d4abfaaac1cc2d41cce62 : SUCCESS in 13m 10s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/3eee289bbacc410eba34d7d122cef830 : SUCCESS in 7m 30s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/c6a750c4374d44eb87c300fa86e322b8 : SUCCESS in 1h 18m 23s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/b0307dd8eea1452993c1476a7a2148fe : SUCCESS in 1h 02m 25s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/288a305cf02c48d384077a2c5b46d2b1 : SUCCESS in 1h 09m 43s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/a910b99d78d246919382a4aede1c1975 : FAILURE in 1h 29m 34s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4149a4b781b74bd7a3edf89bcb7fd6e1 : SUCCESS in 50m 35s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/22aab88d9433404b81499ff523c3626e : SUCCESS in 1h 14m 00s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/fd8a61aac41f4d068df27cc08d5012c6 : SUCCESS in 28m 26s","accounts_in_message":[],"_revision_number":20},{"id":"dc4cc6ca22bd6c4d197721d91e92be983b7bf223","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-09-03 14:01:58.000000000","message":"Patch Set 20: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"3488c5a72a3e54ffe05d0ff9d58d0e1761578a76","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-09-03 14:19:34.000000000","message":"Patch Set 20: Code-Review+2","accounts_in_message":[],"_revision_number":20},{"id":"c9a6b3dc88bdddadbd1202f5fdfa2d70a6612a7e","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-09-03 15:44:05.000000000","message":"Patch Set 20: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":20},{"id":"607c01af7f0880b2f850fbbc5b69582151401e3c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-03 15:44:56.000000000","message":"Patch Set 20: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":20},{"id":"8b338199448603e0c3197851caecf5546e256c53","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-03 17:05:15.000000000","message":"Patch Set 20: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/9b6a17a7bbc04a1884d45bc408588b5b : SUCCESS in 1h 13m 51s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7c8156a4a9034d47ba02e4860e1c1520 : SUCCESS in 1h 07m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e938591f2924428b50694d85692185d : SUCCESS in 7m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/209131b9eb83471e8dc12092fe153673 : SUCCESS in 8m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/36aa757408154d8a9607f0dba2216b1f : SUCCESS in 9m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74f7a9d02f71449fb22c03107a7ef6c5 : SUCCESS in 9m 27s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0a22b9f17fe14105b70d0b445d71064c : SUCCESS in 12m 31s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/27595d17404446aa8cfcc691b8c129c0 : SUCCESS in 12m 48s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/c26a22c17428406e8c6cbbfb146c1a57 : SUCCESS in 1h 04m 29s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/9d0a79d528d14dabb7a66fa18184b94b : SUCCESS in 1h 09m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c0169b1fc0884dd9b86338e5b6c54ba8 : SUCCESS in 50m 52s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/59215bb58a60420bae2ad750bb85687f : SUCCESS in 1h 11m 32s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/402e3db49dfe4a428728718402312310 : SUCCESS in 34m 03s","accounts_in_message":[],"_revision_number":20},{"id":"a0f916f579c7b5fd494dcf1b91c2ccc6d7fbabf2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-03 17:05:15.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":20},{"id":"eb2c757f0e1223590288e2832f128da561b65f23","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-03 17:08:53.000000000","message":"Patch Set 20:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28b17e90c3944acbb32c41053df168c1 : SUCCESS in 1m 18s","accounts_in_message":[],"_revision_number":20},{"id":"41c0734da613f2466252ddb1da26cb8b9f959b63","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-09-06 06:32:17.000000000","message":"Patch Set 20:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/802056/802056/20/check-not-vote/ext-glance-zuul/8683c2d : SUCCESS in 33m 12s","accounts_in_message":[],"_revision_number":20}],"current_revision_number":20,"current_revision":"4473a6053644eec84552780930efe544fd503774","revisions":{"445604b74ed09592a721f25f421445f10ef1c691":{"kind":"REWORK","_number":1,"created":"2021-07-23 16:34:12.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/56/802056/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/56/802056/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/56/802056/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/56/802056/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/56/802056/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/56/802056/1"}}},"commit":{"parents":[{"commit":"8201990e7415837c197ec52b75863bdefd64b997","subject":"Implement project personas for metadef properties","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8201990e7415837c197ec52b75863bdefd64b997"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2021-07-23 14:47:22.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2021-07-23 16:27:31.000000000","tz":0},"subject":"Implement project personas for metadef tags","message":"Implement project personas for metadef tags\n\nThis commit updates the policies for metadef tags\nto use default roles available from keystone.\n\nChange-Id: I2f97c64a72e081ae2f9e558b6fa53dbfd5af2026\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/445604b74ed09592a721f25f421445f10ef1c691"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/445604b74ed09592a721f25f421445f10ef1c691"}]},"branch":"refs/heads/master"},"9b81e7534ecc3e639fd3d4fbbdbfffa313515a11":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-07-23 17:02:45.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/56/802056/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/56/802056/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/56/802056/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/56/802056/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/56/802056/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/56/802056/2"}}},"commit":{"parents":[{"commit":"e284c9db15ce0badfa661856eb1495dc55317f89","subject":"Implement project personas for metadef properties","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e284c9db15ce0badfa661856eb1495dc55317f89"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2021-07-23 14:47:22.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2021-07-23 17:02:45.000000000","tz":0},"subject":"Implement project personas for metadef tags","message":"Implement project personas for metadef tags\n\nThis commit updates the policies for metadef tags\nto use default roles available from keystone.\n\nChange-Id: I2f97c64a72e081ae2f9e558b6fa53dbfd5af2026\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9b81e7534ecc3e639fd3d4fbbdbfffa313515a11"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9b81e7534ecc3e639fd3d4fbbdbfffa313515a11"}]},"branch":"refs/heads/master","description":"Rebase"},"866397347321e7265bcbcff0bc67580efc49f667":{"kind":"REWORK","_number":3,"created":"2021-07-25 19:36:32.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/56/802056/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/56/802056/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/56/802056/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/56/802056/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/56/802056/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/56/802056/3"}}},"commit":{"parents":[{"commit":"edf00595f255ad59fe8f24dc677bab98cb59bc5d","subject":"Implement project personas for metadef properties","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/edf00595f255ad59fe8f24dc677bab98cb59bc5d"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2021-07-23 14:47:22.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2021-07-25 19:28:44.000000000","tz":0},"subject":"Implement project personas for metadef tags","message":"Implement project personas for metadef tags\n\nThis commit updates the policies for metadef tags\nto use default roles available from keystone.\n\nChange-Id: I2f97c64a72e081ae2f9e558b6fa53dbfd5af2026\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/866397347321e7265bcbcff0bc67580efc49f667"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/866397347321e7265bcbcff0bc67580efc49f667"}]},"branch":"refs/heads/master"},"55464e6ca68a8d714d623dad21aa010f8880768a":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2021-07-26 10:16:51.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/56/802056/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/56/802056/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/56/802056/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/56/802056/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/56/802056/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/56/802056/4"}}},"commit":{"parents":[{"commit":"91eccd8b543ff735743ba2c211d4bf8f714aa165","subject":"Implement project personas for metadef properties","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/91eccd8b543ff735743ba2c211d4bf8f714aa165"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2021-07-23 14:47:22.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2021-07-26 10:16:51.000000000","tz":0},"subject":"Implement project personas for metadef tags","message":"Implement project personas for metadef tags\n\nThis commit updates the policies for metadef tags\nto use default roles available from keystone.\n\nChange-Id: I2f97c64a72e081ae2f9e558b6fa53dbfd5af2026\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/55464e6ca68a8d714d623dad21aa010f8880768a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/55464e6ca68a8d714d623dad21aa010f8880768a"}]},"branch":"refs/heads/master","description":"Rebase"},"b662e7ae767a34b496446dba203cbe5e80920cd0":{"kind":"REWORK","_number":5,"created":"2021-07-29 11:05:07.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/56/802056/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/56/802056/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/56/802056/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/56/802056/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/56/802056/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/56/802056/5"}}},"commit":{"parents":[{"commit":"f5c4f1ce46951b7c30da801a691096be89250276","subject":"Implement project personas for metadef properties","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f5c4f1ce46951b7c30da801a691096be89250276"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2021-07-23 14:47:22.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2021-07-29 10:39:38.000000000","tz":0},"subject":"Implement project personas for metadef tags","message":"Implement project personas for metadef tags\n\nThis commit updates the policies for metadef tags\nto use default roles available from keystone.\n\nChange-Id: I2f97c64a72e081ae2f9e558b6fa53dbfd5af2026\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b662e7ae767a34b496446dba203cbe5e80920cd0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b662e7ae767a34b496446dba203cbe5e80920cd0"}]},"branch":"refs/heads/master"},"8e48aa7064489f09ce00891da71a61be0353cef7":{"kind":"REWORK","_number":6,"created":"2021-07-30 11:26:59.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/56/802056/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/56/802056/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/56/802056/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/56/802056/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/56/802056/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/56/802056/6"}}},"commit":{"parents":[{"commit":"b80823a173757fbd60d1453b7a5c9a2c98cc1bb7","subject":"Implement project personas for metadef properties","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b80823a173757fbd60d1453b7a5c9a2c98cc1bb7"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2021-07-23 14:47:22.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2021-07-30 11:20:23.000000000","tz":0},"subject":"Implement project personas for metadef tags","message":"Implement project personas for metadef tags\n\nThis commit updates the policies for metadef tags\nto use default roles available from keystone.\n\nChange-Id: I2f97c64a72e081ae2f9e558b6fa53dbfd5af2026\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8e48aa7064489f09ce00891da71a61be0353cef7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8e48aa7064489f09ce00891da71a61be0353cef7"}]},"branch":"refs/heads/master"},"9838b33d064828fb543b05902bffd5559bbc2f4b":{"kind":"REWORK","_number":7,"created":"2021-08-03 06:14:48.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/56/802056/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/56/802056/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/56/802056/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/56/802056/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/56/802056/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/56/802056/7"}}},"commit":{"parents":[{"commit":"13fff0895a3e304e44fac5dcc2ab776fd9dffb34","subject":"Implement project personas for metadef properties","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/13fff0895a3e304e44fac5dcc2ab776fd9dffb34"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2021-07-23 14:47:22.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2021-08-03 06:08:01.000000000","tz":0},"subject":"Implement project personas for metadef tags","message":"Implement project personas for metadef tags\n\nThis commit updates the policies for metadef tags\nto use default roles available from keystone.\n\nChange-Id: I2f97c64a72e081ae2f9e558b6fa53dbfd5af2026\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9838b33d064828fb543b05902bffd5559bbc2f4b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9838b33d064828fb543b05902bffd5559bbc2f4b"}]},"branch":"refs/heads/master"},"f150e1809269c96bd3069232efbfe245b43e144e":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2021-08-03 09:05:58.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/56/802056/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/56/802056/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/56/802056/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/56/802056/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/56/802056/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/56/802056/8"}}},"commit":{"parents":[{"commit":"8e405642e97c01d9649477c93b44b3d8bc8e41b5","subject":"Implement project personas for metadef properties","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8e405642e97c01d9649477c93b44b3d8bc8e41b5"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2021-07-23 14:47:22.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2021-08-03 09:05:58.000000000","tz":0},"subject":"Implement project personas for metadef tags","message":"Implement project personas for metadef tags\n\nThis commit updates the policies for metadef tags\nto use default roles available from keystone.\n\nChange-Id: I2f97c64a72e081ae2f9e558b6fa53dbfd5af2026\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f150e1809269c96bd3069232efbfe245b43e144e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f150e1809269c96bd3069232efbfe245b43e144e"}]},"branch":"refs/heads/master","description":"Rebase"},"5aeb8ffdd127c7bf6615f6e8086d6987b9fd70bc":{"kind":"REWORK","_number":9,"created":"2021-08-03 11:22:18.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/56/802056/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/56/802056/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/56/802056/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/56/802056/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/56/802056/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/56/802056/9"}}},"commit":{"parents":[{"commit":"61e6cbcbc4721f60eeec2fbdc783cc2fb8f63ee3","subject":"Implement project personas for metadef properties","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/61e6cbcbc4721f60eeec2fbdc783cc2fb8f63ee3"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2021-07-23 14:47:22.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2021-08-03 11:13:54.000000000","tz":0},"subject":"Implement project personas for metadef tags","message":"Implement project personas for metadef tags\n\nThis commit updates the policies for metadef tags\nto use default roles available from keystone.\n\nChange-Id: I2f97c64a72e081ae2f9e558b6fa53dbfd5af2026\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5aeb8ffdd127c7bf6615f6e8086d6987b9fd70bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5aeb8ffdd127c7bf6615f6e8086d6987b9fd70bc"}]},"branch":"refs/heads/master"},"9f861ec34a4ef4bb3a14506fe4257f72915cdc65":{"kind":"REWORK","_number":10,"created":"2021-08-16 14:07:12.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/56/802056/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/56/802056/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/56/802056/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/56/802056/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/56/802056/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/56/802056/10"}}},"commit":{"parents":[{"commit":"068deeea149405799c032857ee83cbe1f75c87a1","subject":"Implement project personas for metadef properties","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/068deeea149405799c032857ee83cbe1f75c87a1"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2021-07-23 14:47:22.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2021-08-16 13:56:21.000000000","tz":0},"subject":"Implement project personas for metadef tags","message":"Implement project personas for metadef tags\n\nThis commit updates the policies for metadef tags\nto use default roles available from keystone.\n\nChange-Id: I2f97c64a72e081ae2f9e558b6fa53dbfd5af2026\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9f861ec34a4ef4bb3a14506fe4257f72915cdc65"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9f861ec34a4ef4bb3a14506fe4257f72915cdc65"}]},"branch":"refs/heads/master"},"9ab00daed222e3ae0d73ca9fbee5a556126aa4c9":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2021-08-23 08:53:50.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/56/802056/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/56/802056/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/56/802056/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/56/802056/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/56/802056/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/56/802056/11"}}},"commit":{"parents":[{"commit":"90a68710c4d187a7f26ee32ec189c50fca1d0d8b","subject":"Implement project personas for metadef properties","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/90a68710c4d187a7f26ee32ec189c50fca1d0d8b"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2021-07-23 14:47:22.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2021-08-23 08:53:50.000000000","tz":0},"subject":"Implement project personas for metadef tags","message":"Implement project personas for metadef tags\n\nThis commit updates the policies for metadef tags\nto use default roles available from keystone.\n\nChange-Id: I2f97c64a72e081ae2f9e558b6fa53dbfd5af2026\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9ab00daed222e3ae0d73ca9fbee5a556126aa4c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9ab00daed222e3ae0d73ca9fbee5a556126aa4c9"}]},"branch":"refs/heads/master","description":"Rebase"},"709a54ce5e5f2426dc1c6bde19b5add696522736":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2021-08-26 04:02:48.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/56/802056/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/56/802056/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/56/802056/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/56/802056/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/56/802056/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/56/802056/12"}}},"commit":{"parents":[{"commit":"be9a2975bcc303801047a9c3415b77377c7c1501","subject":"Implement project personas for metadef properties","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/be9a2975bcc303801047a9c3415b77377c7c1501"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2021-07-23 14:47:22.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-08-26 03:57:32.000000000","tz":0},"subject":"Implement project personas for metadef tags","message":"Implement project personas for metadef tags\n\nThis commit updates the policies for metadef tags\nto use default roles available from keystone.\n\nChange-Id: I2f97c64a72e081ae2f9e558b6fa53dbfd5af2026\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/709a54ce5e5f2426dc1c6bde19b5add696522736"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/709a54ce5e5f2426dc1c6bde19b5add696522736"}]},"branch":"refs/heads/master"},"1b22f7c34a5d32c96ad2934302aefdeaf74193fd":{"kind":"REWORK","_number":13,"created":"2021-08-30 19:47:26.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/56/802056/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/56/802056/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/56/802056/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/56/802056/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/56/802056/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/56/802056/13"}}},"commit":{"parents":[{"commit":"b216383f52643b27e8421878fdfcbbe365207301","subject":"Implement project personas for metadef properties","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b216383f52643b27e8421878fdfcbbe365207301"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2021-07-23 14:47:22.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2021-08-30 19:37:16.000000000","tz":0},"subject":"Implement project personas for metadef tags","message":"Implement project personas for metadef tags\n\nThis commit updates the policies for metadef tags\nto use default roles available from keystone.\n\nChange-Id: I2f97c64a72e081ae2f9e558b6fa53dbfd5af2026\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1b22f7c34a5d32c96ad2934302aefdeaf74193fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1b22f7c34a5d32c96ad2934302aefdeaf74193fd"}]},"branch":"refs/heads/master"},"f942d212d74b6b5b76d20a96855857e3c6833c14":{"kind":"REWORK","_number":14,"created":"2021-08-31 06:08:04.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/56/802056/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/56/802056/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/56/802056/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/56/802056/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/56/802056/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/56/802056/14"}}},"commit":{"parents":[{"commit":"273a00af1537809e7e7775a277147019ed34095e","subject":"Implement project personas for metadef properties","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/273a00af1537809e7e7775a277147019ed34095e"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2021-07-23 14:47:22.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2021-08-31 05:53:20.000000000","tz":0},"subject":"Implement project personas for metadef tags","message":"Implement project personas for metadef tags\n\nThis commit updates the policies for metadef tags\nto use default roles available from keystone.\n\nChange-Id: I2f97c64a72e081ae2f9e558b6fa53dbfd5af2026\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f942d212d74b6b5b76d20a96855857e3c6833c14"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f942d212d74b6b5b76d20a96855857e3c6833c14"}]},"branch":"refs/heads/master"},"9ae6524b4639bcc9e26bdd10d4e5e4ab8c8c3a5e":{"kind":"TRIVIAL_REBASE","_number":15,"created":"2021-08-31 20:04:52.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/56/802056/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/56/802056/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/56/802056/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/56/802056/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/56/802056/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/56/802056/15"}}},"commit":{"parents":[{"commit":"0d9f639dde96261685907439b1465e6954e78c07","subject":"Implement project personas for metadef properties","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0d9f639dde96261685907439b1465e6954e78c07"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2021-07-23 14:47:22.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2021-08-31 19:48:30.000000000","tz":0},"subject":"Implement project personas for metadef tags","message":"Implement project personas for metadef tags\n\nThis commit updates the policies for metadef tags\nto use default roles available from keystone.\n\nChange-Id: I2f97c64a72e081ae2f9e558b6fa53dbfd5af2026\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9ae6524b4639bcc9e26bdd10d4e5e4ab8c8c3a5e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9ae6524b4639bcc9e26bdd10d4e5e4ab8c8c3a5e"}]},"branch":"refs/heads/master"},"c57b4325405861077c5958300c7fca3b4f6c4b26":{"kind":"TRIVIAL_REBASE","_number":16,"created":"2021-08-31 20:20:51.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/56/802056/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/56/802056/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/56/802056/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/56/802056/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/56/802056/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/56/802056/16"}}},"commit":{"parents":[{"commit":"90e281e2b25096ba0f5f188c0c142a3ee11d697e","subject":"Implement project personas for metadef properties","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/90e281e2b25096ba0f5f188c0c142a3ee11d697e"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2021-07-23 14:47:22.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2021-08-31 20:06:12.000000000","tz":0},"subject":"Implement project personas for metadef tags","message":"Implement project personas for metadef tags\n\nThis commit updates the policies for metadef tags\nto use default roles available from keystone.\n\nChange-Id: I2f97c64a72e081ae2f9e558b6fa53dbfd5af2026\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c57b4325405861077c5958300c7fca3b4f6c4b26"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c57b4325405861077c5958300c7fca3b4f6c4b26"}]},"branch":"refs/heads/master"},"5813879b0f70276426e17074e28095ba238dd734":{"kind":"REWORK","_number":17,"created":"2021-09-02 09:20:55.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/56/802056/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/56/802056/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/56/802056/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/56/802056/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/56/802056/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/56/802056/17"}}},"commit":{"parents":[{"commit":"0289c27c44f90bffe7c86bd768403374f0d67a5b","subject":"Implement project personas for metadef properties","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0289c27c44f90bffe7c86bd768403374f0d67a5b"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2021-07-23 14:47:22.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-09-02 09:14:27.000000000","tz":0},"subject":"Implement project personas for metadef tags","message":"Implement project personas for metadef tags\n\nThis commit updates the policies for metadef tags\nto use default roles available from keystone.\n\nChange-Id: I2f97c64a72e081ae2f9e558b6fa53dbfd5af2026\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5813879b0f70276426e17074e28095ba238dd734"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5813879b0f70276426e17074e28095ba238dd734"}]},"branch":"refs/heads/master"},"11eab5ef8a3f3d912c609bb0a51500db8ea7169b":{"kind":"REWORK","_number":18,"created":"2021-09-02 12:18:28.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/56/802056/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/56/802056/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/56/802056/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/56/802056/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/56/802056/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/56/802056/18"}}},"commit":{"parents":[{"commit":"3554e7c393b1fe03d27af8c11806734961bd84dd","subject":"Implement project personas for metadef properties","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3554e7c393b1fe03d27af8c11806734961bd84dd"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2021-07-23 14:47:22.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2021-09-02 12:08:02.000000000","tz":0},"subject":"Implement project personas for metadef tags","message":"Implement project personas for metadef tags\n\nThis commit updates the policies for metadef tags\nto use default roles available from keystone.\n\nChange-Id: I2f97c64a72e081ae2f9e558b6fa53dbfd5af2026\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/11eab5ef8a3f3d912c609bb0a51500db8ea7169b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/11eab5ef8a3f3d912c609bb0a51500db8ea7169b"}]},"branch":"refs/heads/master"},"a2f18e5554afe7316f4b750a4feb6348e9eefad7":{"kind":"TRIVIAL_REBASE","_number":19,"created":"2021-09-02 22:03:30.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/56/802056/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/56/802056/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/56/802056/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/56/802056/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/56/802056/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/56/802056/19"}}},"commit":{"parents":[{"commit":"9796593f31a4f0bbc62ae6e0541385b573b9ce12","subject":"Implement project personas for metadef properties","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9796593f31a4f0bbc62ae6e0541385b573b9ce12"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2021-07-23 14:47:22.000000000","tz":0},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-09-02 23:06:13.000000000","tz":-420},"subject":"Implement project personas for metadef tags","message":"Implement project personas for metadef tags\n\nThis commit updates the policies for metadef tags\nto use default roles available from keystone.\n\nChange-Id: I2f97c64a72e081ae2f9e558b6fa53dbfd5af2026\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a2f18e5554afe7316f4b750a4feb6348e9eefad7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a2f18e5554afe7316f4b750a4feb6348e9eefad7"}]},"branch":"refs/heads/master"},"4473a6053644eec84552780930efe544fd503774":{"kind":"REWORK","_number":20,"created":"2021-09-03 05:45:31.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/56/802056/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/56/802056/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/56/802056/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/56/802056/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/56/802056/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/56/802056/20"}}},"commit":{"parents":[{"commit":"9796593f31a4f0bbc62ae6e0541385b573b9ce12","subject":"Implement project personas for metadef properties","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9796593f31a4f0bbc62ae6e0541385b573b9ce12"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2021-07-23 14:47:22.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-09-03 05:37:05.000000000","tz":0},"subject":"Implement project personas for metadef tags","message":"Implement project personas for metadef tags\n\nThis commit updates the policies for metadef tags\nto use default roles available from keystone.\n\nChange-Id: I2f97c64a72e081ae2f9e558b6fa53dbfd5af2026\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4473a6053644eec84552780930efe544fd503774"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4473a6053644eec84552780930efe544fd503774"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
