)]}'
{"id":"openstack%2Fglance~802546","triplet_id":"openstack%2Fglance~master~I66be51f467fb9465fb479bdc823ade283a7ddaf9","project":"openstack/glance","branch":"master","topic":"bug/1938265","attention_set":{},"removed_from_attention_set":{"5202":{"account":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"last_update":"2022-01-20 13:01:59.000000000","reason":"Change was marked work in progress"},"11583":{"account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"last_update":"2022-01-20 13:01:53.000000000","reason":"Arnaud Morin replied on the change"}},"hashtags":[],"change_id":"I66be51f467fb9465fb479bdc823ade283a7ddaf9","subject":"Figure out backend from URI","status":"ABANDONED","created":"2021-07-27 15:31:44.000000000","updated":"2024-02-16 14:27:18.000000000","total_comment_count":4,"unresolved_comment_count":1,"work_in_progress":true,"has_review_started":true,"meta_rev_id":"33faff002e318f034c73d7c0ce75ebd64ad3b52d","_number":802546,"virtual_id_number":802546,"owner":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-09-03 13:28:07.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"all":[{"value":-1,"date":"2022-01-18 18:13:50.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-27 15:37:31.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"CC"},{"updated":"2021-07-27 15:37:31.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2021-07-27 16:31:09.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-07-27 17:22:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-04 15:52:38.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"CC"},{"updated":"2022-01-18 18:13:50.000000000","updated_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"}],"messages":[{"id":"60d07b6c445e8543cb7297a82fab9837f1328624","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2021-07-27 15:31:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"329623affa57d36b24926b90e49c1d967253fa6c","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-07-27 16:31:09.000000000","message":"Patch Set 1:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/802546/802546/1/check-not-vote/ext-glance-zuul/ada70a3 : SUCCESS in 34m 31s","accounts_in_message":[],"_revision_number":1},{"id":"9b3baa9f5ac950f249c9c62b3d601dc3837cd53f","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-07-27 16:39:48.000000000","message":"Patch Set 1:\n\nIt\u0027s hard to review this in much detail without some tests...","accounts_in_message":[],"_revision_number":1},{"id":"85310b92eaaf016119038ff90b8fbb5f797bd7a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-27 17:22:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/9ec8bc8be08447df9bcd4125dc3d7cb9 : SUCCESS in 1h 05m 07s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c454deb5c1944d66b4ee0648f5be9df4 : SUCCESS in 1h 10m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5f5ffd1bca24e168cd5b9bd912b7ef6 : SUCCESS in 7m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f77884b27b804ac181bf93666573fa71 : SUCCESS in 8m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/12e0d442c06e4197b0c946f5df65506c : SUCCESS in 8m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/698ec71070a0472db780fe432243fe38 : SUCCESS in 8m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99e662c41c194ea3919ccafc57b61a9b : SUCCESS in 8m 12s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3b58748ed7184e2ca3c33cf46c0ebb4f : TIMED_OUT in 31m 25s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/486844a5c4154d2e92911e0aaa0d86e8 : SUCCESS in 9m 51s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/38e4944f3fc54314905de97769d2cdb5 : SUCCESS in 10m 31s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/16fc8c99a6ca4056a78be029a4ab9ec0 : SUCCESS in 6m 03s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/f220efb56aff439ab83238431c1d1cbd : SUCCESS in 1h 20m 51s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/7304c8a714e04b388af9ccaf35cf6e97 : SUCCESS in 1h 10m 33s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/fc7e703a1b6d48e9847e806f4a31c47e : SUCCESS in 1h 16m 20s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/749c80b7d5cd4955be97c500cce3e3fc : SUCCESS in 1h 43m 31s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2e036d79dbaa48bca9462b0fe2bf528f : SUCCESS in 56m 23s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/8860f827d4024ab184f71c1d1c00439b : SUCCESS in 1h 35m 03s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/829b4caaa7a94f2c9cef9e060062612c : SUCCESS in 42m 39s","accounts_in_message":[],"_revision_number":1},{"id":"7876fc15394e67b170a163ba8b482dfb3df4b7d2","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2021-07-28 08:24:15.000000000","message":"Topic set to bug-1938265","accounts_in_message":[],"_revision_number":1},{"id":"6d6b7c2e7a6ba64a469f67ccbf73d8a1388e6f19","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2021-07-28 08:25:07.000000000","message":"Topic bug-1938265 removed","accounts_in_message":[],"_revision_number":1},{"id":"4a7c5b3acd85445feb0b362da1deffaeac327bf5","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2021-07-28 08:25:11.000000000","message":"Topic set to bug/1938265","accounts_in_message":[],"_revision_number":1},{"id":"8b1a637f2246f6d64e7ad74ba3e2c1d04b93edab","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2021-09-03 12:08:06.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"fd29337dd4d64dcb7cf9377e0472077e3c4f9c27","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-03 13:28:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/2fdeb4f6e56748cea36b324d7b929545 : SUCCESS in 1h 04m 09s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/bfcf85236ef64950a46b68b47ec86fe3 : SUCCESS in 1h 10m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c95d272295eb493aa6e3d27d8df34d04 : SUCCESS in 6m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a2facd7c3f734ee3a6d485b256677f65 : SUCCESS in 7m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/73a497635d464860b1cab4069676a174 : SUCCESS in 7m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a3e96fdca3014b8a96f03e70619853aa : SUCCESS in 8m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e5e2fd04cb194318b68b48a8c325356f : SUCCESS in 7m 48s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/297df60f6b324e14b0ac0af6e0fda160 : SUCCESS in 9m 19s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/057e5b70f8da4caeb3a5d7801258bf1c : SUCCESS in 11m 21s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/2f6185799bbf49bbb7b26d7f581e0c63 : SUCCESS in 11m 07s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/c95e167235eb4a73a06ab3ef695d1c3f : SUCCESS in 7m 26s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/868c287a0c274cbc97b7e44b57f942a6 : SUCCESS in 1h 01m 22s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/bd17513987cf48f89b554bd59d57f8e9 : SUCCESS in 58m 01s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/5b9e1e006abb40a2a586eed03e2a72a9 : SUCCESS in 1h 06m 16s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/8836c0e36ce8427d8f3bd8cf160d4ff4 : SUCCESS in 1h 15m 37s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/34631e27467143019e4d4d213a94c162 : SUCCESS in 47m 48s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/d455b1a6213e4e8ebe71808be4164e33 : SUCCESS in 1h 07m 43s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/31af4c9866ab40359758b8be223d753d : SUCCESS in 28m 06s","accounts_in_message":[],"_revision_number":1},{"id":"43dd06ff7a1c76230906dca33167fda4af7d5028","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-09-06 12:27:04.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/802546/802546/1/check-not-vote/ext-glance-zuul/d7258a2 : FAILURE in 15m 36s","accounts_in_message":[],"_revision_number":1},{"id":"3f281b220fdaf2ca5c643a7409453eafef6bd9fe","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2021-09-28 08:17:45.000000000","message":"Patch Set 1:\n\nHey team, did you had time to talk/check this issue?","accounts_in_message":[],"_revision_number":1},{"id":"1b7de261862533abe4f9f9d0e4d524c214a40a6d","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2021-11-04 15:52:38.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f87e59ff3035536d0856bf24f1f76f1d4952f719","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2022-01-18 18:13:50.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"0529288482e0925b23b7874df819cadbaf1cb571","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2022-01-20 13:01:53.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d16ae97d56ddf8b883fbdfa69eab6b8475329b28","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2022-01-20 13:01:59.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"33faff002e318f034c73d7c0ce75ebd64ad3b52d","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2024-02-16 14:27:18.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"39654d32abc9a90702ca23d809761aa873fb903f","revisions":{"39654d32abc9a90702ca23d809761aa873fb903f":{"kind":"REWORK","_number":1,"created":"2021-07-27 15:31:44.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/46/802546/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/46/802546/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/46/802546/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/46/802546/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/46/802546/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/46/802546/1"}}},"commit":{"parents":[{"commit":"485677a3f383489bbedf7c184ed992f1bcc53b24","subject":"Add base policy check module","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/485677a3f383489bbedf7c184ed992f1bcc53b24"}]}],"author":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2021-07-27 15:23:24.000000000","tz":120},"committer":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2021-07-27 15:23:24.000000000","tz":120},"subject":"Figure out backend from URI","message":"Figure out backend from URI\n\nAdd a mechanism in _check_image_location to retrieve the backend from\nURL if no backend provided in metadata.\n\nThis will avoid calling _check_location_uri with backend\u003dNone and fail\nwith an Invalid location.\n\nWe can endup in this situation when nova tries to PATCH an image by\nadding a location (nova is not populating the metadata). See [1]\n\n[1] https://github.com/openstack/nova/blob/master/nova/image/glance.py#L559\n\nChange-Id: I66be51f467fb9465fb479bdc823ade283a7ddaf9\nSigned-off-by: Arnaud Morin \u003carnaud.morin@ovhcloud.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/39654d32abc9a90702ca23d809761aa873fb903f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/39654d32abc9a90702ca23d809761aa873fb903f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
