)]}'
{"id":"openstack%2Fglance~802762","triplet_id":"openstack%2Fglance~master~I6618366e6464d569a86fb027919be2e1825e438c","project":"openstack/glance","branch":"master","topic":"cleanup","hashtags":[],"change_id":"I6618366e6464d569a86fb027919be2e1825e438c","subject":"tests: Remove use of \u0027oslo_db.sqlalchemy.test_base\u0027","status":"MERGED","created":"2021-07-28 17:06:19.000000000","updated":"2021-08-25 11:31:41.000000000","submitted":"2021-08-25 11:30:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"802762-cleanup","meta_rev_id":"92b16c26afa91f718104808aece100ed950e5fb0","_number":802762,"virtual_id_number":802762,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-25 11:30:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"all":[{"value":2,"date":"2021-08-25 09:46:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":2,"date":"2021-07-29 16:58:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"all":[{"value":1,"date":"2021-08-25 09:46:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-28 17:48:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-28 19:41:32.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-07-29 00:21:39.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"CC"},{"updated":"2021-07-29 16:58:24.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2021-08-25 09:46:31.000000000","updated_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"}],"messages":[{"id":"0c0efaa4e78740065bf4e00e82aed49eb3aee8da","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-07-28 17:06:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"30ad2d9282c8d6bc9ea334ddce91d7c18ecba5db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-28 17:48:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e018f81153144e480c97b57a7758d1b : SUCCESS in 5m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3c26ee56475b417ca8f83f5e4b900ce6 : SUCCESS in 6m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b63f1af84d294a9796ec38c0e16cfd46 : SUCCESS in 5m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/87e5632c8290452a97d2233e8c8077dc : SUCCESS in 6m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c0ce2ab167c4ea6bf12b46836a3ad82 : SUCCESS in 6m 41s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f1c4d600f2b245c799e5960d84904f73 : SUCCESS in 9m 39s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/fbf609e1c3eb41d192a9c4b2d852cdd0 : SUCCESS in 9m 23s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/21a1ac7a6e934bc18496295d50ef261d : SUCCESS in 9m 06s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/5885790a59da40b890669bf69c285c1b : SUCCESS in 5m 51s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/f142d24c21fe4ec3b175e1b8cac2e3b8 : SUCCESS in 35m 44s","accounts_in_message":[],"_revision_number":1},{"id":"4a9fc6b6f62b6d0d86a412186b05971be03d1260","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-07-28 19:41:32.000000000","message":"Patch Set 1:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/802762/802762/1/check-not-vote/ext-glance-zuul/396a306 : SUCCESS in 35m 06s","accounts_in_message":[],"_revision_number":1},{"id":"3b03da394fb63050f3c1dd446833f71b3b927f6e","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2021-07-29 00:21:39.000000000","message":"Patch Set 1:\n\nI\u0027m no oslo_db expert but it looks good to me. One question though: why do we suddenly need to inherit from glance.tests.utils.BaseTestCase?","accounts_in_message":[],"_revision_number":1},{"id":"2bff423ec03b5de4529394cebb997f44360abafa","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-07-29 09:19:37.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1:\n\u003e \n\u003e I\u0027m no oslo_db expert but it looks good to me. One question though: why do we suddenly need to inherit from glance.tests.utils.BaseTestCase?\n\nWe need to inherit from _a_ base class. The other classes there are mixins as opposed to something derived from unittest.TestCase. I could also use oslotest\u0027s BaseTestCase, but I had a look at other tests in glance.tests.functional.db and this was the first base test class I spotted 😄\n\nLet me know if I need to change anything here","accounts_in_message":[],"_revision_number":1},{"id":"5163ef5add02b6ae9bae592463fbdb1b0a6bf04d","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-07-29 16:58:24.000000000","message":"Patch Set 1: Code-Review+2\n\nI am good with using glance tests base class.\nThank you!","accounts_in_message":[],"_revision_number":1},{"id":"a40ac3a20d2ad84577aec208592572bee22e6a52","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2021-08-25 09:46:31.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nlgtm","accounts_in_message":[],"_revision_number":1},{"id":"74fb2a3ecf7a2393fc8c7b4655b8a89d1c99d482","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-25 09:47:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f102b56b78deed8a4fa9239def869b3164d008e3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-25 11:30:09.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f8972fa14464ca2abc9e682f364b3ab : SUCCESS in 8m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/de2a029c555b48ac9d96d423cd1476b7 : SUCCESS in 11m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/702d5b3aa7d246d59be03718af64ca0c : SUCCESS in 8m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8155760982640b6afed1f27e647dfe1 : SUCCESS in 9m 42s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/150354b813f549afaf096720fd93a151 : SUCCESS in 11m 05s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/28767801281047c39b6f6deac8663eb5 : SUCCESS in 11m 51s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/df74fc49c82e4658b0f37229b0653fd3 : SUCCESS in 27m 59s","accounts_in_message":[],"_revision_number":1},{"id":"91de965909ee6f79a4be16d4965b8869d465cf57","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-25 11:30:09.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"92b16c26afa91f718104808aece100ed950e5fb0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-25 11:31:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56d5ab15d67e40dfb4b6c8f716c8d4bf : SUCCESS in 1m 09s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"815922fbee8eae613b21dbe70c9ac3998bbf09a0","revisions":{"815922fbee8eae613b21dbe70c9ac3998bbf09a0":{"kind":"REWORK","_number":1,"created":"2021-07-28 17:06:19.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/62/802762/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/62/802762/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/62/802762/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/62/802762/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/62/802762/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/62/802762/1"}}},"commit":{"parents":[{"commit":"2fef2e6c4f9fbbac795740ddfec57ba193029d03","subject":"Add api_patch() to SynchronousAPIBase","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2fef2e6c4f9fbbac795740ddfec57ba193029d03"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2021-07-28 15:20:20.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2021-07-28 15:40:17.000000000","tz":60},"subject":"tests: Remove use of \u0027oslo_db.sqlalchemy.test_base\u0027","message":"tests: Remove use of \u0027oslo_db.sqlalchemy.test_base\u0027\n\nThe various test cases and mixins in \u0027oslo_db.sqlalchemy.test_base\u0027 have\nbeen deprecated since oslo.db 4.34.0 (March 2018). Remove use of these.\n\nChange-Id: I6618366e6464d569a86fb027919be2e1825e438c\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/815922fbee8eae613b21dbe70c9ac3998bbf09a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/815922fbee8eae613b21dbe70c9ac3998bbf09a0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
