)]}'
{"id":"openstack%2Fglance~802996","triplet_id":"openstack%2Fglance~master~If425d6263854e3bdb505fbdcb4eecf2e9d98d725","project":"openstack/glance","branch":"master","topic":"policy-refactor","hashtags":[],"change_id":"If425d6263854e3bdb505fbdcb4eecf2e9d98d725","subject":"Move member policy checks to API layer","status":"MERGED","created":"2021-07-30 08:41:25.000000000","updated":"2021-08-12 22:20:14.000000000","submitted":"2021-08-12 22:18:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":33,"unresolved_comment_count":7,"has_review_started":true,"submission_id":"802996-policy-refactor","meta_rev_id":"54eeb7f8d30f6823ddd0b81b5849524771295589","_number":802996,"virtual_id_number":802996,"owner":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-12 22:18:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"all":[{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2021-08-12 20:22:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":2,"date":"2021-08-12 21:16:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":1,"date":"2021-08-12 21:16:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":1,"date":"2021-08-12 22:20:14.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-30 10:21:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-30 14:42:47.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"state":"REVIEWER"},{"updated":"2021-08-06 18:50:58.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-08-06 18:53:11.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2021-08-06 18:53:22.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2021-08-10 01:51:59.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2021-08-11 14:38:59.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"}],"messages":[{"id":"c6ac413f9f033206f53d61c85d3f1852a0a5a96e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-07-30 08:41:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2c405a3cdbed12b0a94584398caf1a51c8c72ac2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-30 10:21:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/da5e5beb609d4b2491f9735f5156c679 : SUCCESS in 1h 15m 52s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7f805d601cc34d86b48c3a6254461ec0 : SUCCESS in 1h 15m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f19e071b2c5e480aa5502cfdd7a12410 : SUCCESS in 5m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8879a555023f4a7c8a1f284300bf9397 : SUCCESS in 6m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2966bb15201f49d794e6b7c471ae13a6 : SUCCESS in 6m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/eb56b995208a47e999ead0e1aaa333bf : SUCCESS in 7m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/19ecb410c4824c60a24dd72b7b10fdbd : SUCCESS in 6m 57s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/426927c048374895883eaeacbba37743 : SUCCESS in 9m 35s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/3f4fa97c098a4eccb6054d6fcb1170a0 : SUCCESS in 8m 59s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/dca12e3d1e864b48bd14ae90662fd247 : SUCCESS in 8m 29s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/eafad55362ce454ea928a4b9d4c146d4 : SUCCESS in 5m 15s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/6fd5f130bb284dbcb000ff94709d9e20 : SUCCESS in 1h 07m 55s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/c9db14297c424b299e3fed4bbcf82050 : SUCCESS in 1h 37m 37s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/c33f17663e644c428bfa30eaf906e4dc : SUCCESS in 1h 15m 58s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/b8caa3e29ca0422e8fef975c6a2fb873 : SUCCESS in 1h 25m 59s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9bd68789871e403795cd527b2bbc16e1 : SUCCESS in 47m 13s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/ef243e1c2f1c4cfdb770c17b90921573 : SUCCESS in 1h 23m 02s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/0d35cc360e474e5097fd3ea7ca4d1dcc : SUCCESS in 27m 11s","accounts_in_message":[],"_revision_number":1},{"id":"e0affdf984767fc420f349ab7857b2c71e442511","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-07-30 14:43:56.000000000","message":"Patch Set 1:\n\nMore information about this here, https://docs.google.com/spreadsheets/d/1SWBq0CsHw8jofHxmOG8QeZEX6veDE4eU0QHItOu8uQs/edit?pli\u003d1#gid\u003d1906142683","accounts_in_message":[],"_revision_number":1},{"id":"e1d056647ecc97c7c20e9f82b094e1518b1945f1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-06 16:47:05.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0d0f8b1b05ef7d4b30482c77e64dd31184e51d91","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-06 16:51:35.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nAll existing unit and functional tests are passing with this change.\nFunctional job to test with RBAC enabled is passing as well.","accounts_in_message":[],"_revision_number":2},{"id":"ac7c7da7ee1d8c3842157e20fbc04f030183962d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-06 18:13:21.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/50c193ac9e8e4f80b314d17828629916 : SUCCESS in 57m 43s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/04ec09275f3b4112a5caeee726710b7d : SUCCESS in 1h 03m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e122f91e7d640d08655be3a73789b79 : SUCCESS in 5m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1871a32ef4404b3ab1f006a54274cc9f : SUCCESS in 7m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bcbe0ee7c0bf4eaa8eef839e700d77ce : SUCCESS in 6m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c7612fffb811499486198c894b305293 : SUCCESS in 6m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d150956a9cbe4beeb04a0236a535c0b2 : SUCCESS in 6m 08s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1ec36c4b5edb42999c146d2a6a8c8ea9 : SUCCESS in 8m 59s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/f51af7b270a04540abf5130e6c1e9deb : SUCCESS in 10m 29s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/a329a6bd68c5493abef6b24ecfaef76e : SUCCESS in 8m 28s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/bf5aa910352a43528a668e89190e0f01 : SUCCESS in 5m 55s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/c9baa5f50a724c9e88658812373f204b : SUCCESS in 1h 07m 15s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/c3ad51aacddd4b1e91a38a8da101e246 : SUCCESS in 1h 04m 31s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/d8defe8a45564a72937b9b26a2c0144a : SUCCESS in 1h 16m 57s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/3f0c89db50834790b2e14272d86bc3a0 : SUCCESS in 1h 17m 14s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/58a4f257f2364c2bb804314c60ca65d0 : SUCCESS in 50m 00s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/5efe5e3b58f646c08b40aafe0b9b73bd : SUCCESS in 1h 17m 53s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/545540ddd21b4af8a8b55edb6650d50b : FAILURE in 28m 21s","accounts_in_message":[],"_revision_number":2},{"id":"51265a63fce802770bde1e097cbecd28f491624b","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-06 18:21:21.000000000","message":"Patch Set 2:\n\nKnow test failure as mentioned in the commit message.\nNeed to work on glance-tempest-plugin fix.","accounts_in_message":[],"_revision_number":2},{"id":"b9d0892c872b474dbe0a4908e957761cf4633daa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-06 18:50:18.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"2b3c3dff008ef62ffb5357d690feb499f11511a1","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-08-06 18:50:58.000000000","message":"Patch Set 3:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/802996/802996/3/check-not-vote/ext-glance-zuul/8c9728a : FAILURE in 0s","accounts_in_message":[],"_revision_number":3},{"id":"0b72ff52c371450d737cdf1c322e5eb6086cdcf0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-06 18:52:57.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"c513b0914bf9c5b30a9264a1ba00d3082ab8601f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-06 18:53:10.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n\nWarning:\n  Dependency cycle detected","accounts_in_message":[],"_revision_number":4},{"id":"88446bcf4899426e053df84fdcb5f847a898f69e","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-08-06 18:53:11.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n\nWarning:\n  Dependency cycle detected","accounts_in_message":[],"_revision_number":4},{"id":"5234a685cb5627f31371923712df5790affea13a","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-08-06 18:53:22.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n\nWarning:\n  Dependency cycle detected","accounts_in_message":[],"_revision_number":4},{"id":"9b1f632f8f0cd19c0807a539e7df1007c67eb8b9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-06 18:54:36.000000000","message":"Patch Set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"0ae9a6cbbabafe0ba58db53ccc11e03828aa69eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-06 20:38:47.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/ab7af5af6b374e09b63d4dc57e30fb3e : SUCCESS in 1h 06m 54s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/bd195d16262444758a98cce0a303be13 : SUCCESS in 1h 06m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/36a5fb3e84734001ad39ebcb8b9f6e5d : SUCCESS in 5m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/61373675497946bdae29704db87cb8fd : SUCCESS in 6m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5f43db5b71c942e18a107eb96eec9261 : SUCCESS in 5m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2ec6e5ba8ded41dd93f4a5473dfdab93 : SUCCESS in 6m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e350bdc58bf4655bd6ae67d3a41243c : SUCCESS in 6m 27s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f3d93dd738b9431da9997192d5f425da : SUCCESS in 9m 17s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/e3294204bdd5474ea601a70a1d45ea17 : SUCCESS in 8m 20s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/52c27925565844dbbc69d61a212f787d : SUCCESS in 9m 42s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/dca43e9b7c664cb2a965ac0f849e46b9 : SUCCESS in 5m 25s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/526aa75eeb974fa5a1205fb98b617811 : SUCCESS in 1h 10m 58s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/30fe4b53ca754c7fb4467210670d2ac4 : SUCCESS in 1h 09m 23s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/fb49b9d0855d4b1099b0d91343ccf26c : SUCCESS in 1h 09m 27s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/a4699aebd97942bb88bbbb95c8d71d65 : SUCCESS in 1h 37m 45s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/781f3985a17e4506832278d0394005ac : SUCCESS in 54m 56s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/9f50342c1b954d6faba44867bfd08108 : SUCCESS in 1h 30m 40s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/dde322e771ab433794b02d5502448580 : FAILURE in 33m 08s","accounts_in_message":[],"_revision_number":5},{"id":"717b08442b325ac9816bdfa5f11afc80e8b8d9c4","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-08-06 21:45:27.000000000","message":"Patch Set 5:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/802996/802996/5/check-not-vote/ext-glance-zuul/774bd49 : SUCCESS in 38m 28s","accounts_in_message":[],"_revision_number":5},{"id":"858f4149079a706aaa7eff1324463832a3a50806","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-08-06 21:56:31.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"0f0c1d11c445595ff172b1c2e576cf70b433320f","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-08-06 22:06:43.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"38aed28ac0f10705d0c89e48ac3ea4a7f8088523","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-07 05:02:10.000000000","message":"Patch Set 5:\n\n(1 comment)\n\nWill abandoned mentioned tempest plugin patch in commit message as it is not required.","accounts_in_message":[],"_revision_number":5},{"id":"ff1cd38830093fd03b99af5e477b36e200371ad0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-08 19:53:01.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"57029bb62ff51321e71cf7bdea00dc69a87e3437","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-08-08 20:30:45.000000000","message":"Patch Set 6:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/802996/802996/6/check-not-vote/ext-glance-zuul/7c7531d : SUCCESS in 37m 04s","accounts_in_message":[],"_revision_number":6},{"id":"69a128f2751e5ec5c3599bd23e49a0d74c900297","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-08 21:16:55.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/4736d25f946741f3a8cac2bb3f052d5b : SUCCESS in 56m 12s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/feaca8e8d0704b9fb156be9af9d2bb27 : SUCCESS in 1h 01m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d4b778372a8e4f9eab2aceee4498ff0e : SUCCESS in 5m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8a623c210d094152a2c117cde50521c5 : SUCCESS in 5m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ad4f5850447946d4ad98d0cb80b74693 : SUCCESS in 6m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dee4fa6dd3ac4381959db91b3fd1425d : SUCCESS in 6m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db2cb8eca53245f1936febc11e261fd2 : SUCCESS in 7m 09s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/15f91151c7d943d0911514b314afd2e5 : SUCCESS in 10m 02s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/e74dfcc8680a44348d68ec5478a83952 : SUCCESS in 9m 27s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/67b6f925a13f42b092f87b68c65efdcb : SUCCESS in 8m 36s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/e4c16aa619c44f948ba639419fa887a0 : SUCCESS in 5m 18s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/89da36a993e04ef6bb3412be6d6fd859 : SUCCESS in 1h 06m 41s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/d0960cc1b80141f6a2d224166f85913e : SUCCESS in 57m 42s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/7e5f03c30d7143ff8a3869386b8189a3 : SUCCESS in 1h 19m 17s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/acacd02ba09243f9874db2dd403faed7 : SUCCESS in 1h 10m 45s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/31c676840e0849fc8eb04c72fb94ca34 : SUCCESS in 1h 00m 48s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/0f99152f47e34787a6943025931fcb9a : SUCCESS in 1h 02m 18s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/bafa825f79d6440482769bcf20ccc4b2 : SUCCESS in 33m 01s","accounts_in_message":[],"_revision_number":6},{"id":"fa2d7eaaae8c2838effa19e9323a92a99cab4f26","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-08-09 16:08:41.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"6dee40a1690242958decfabd9c4721edf5ea4921","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2021-08-10 01:51:59.000000000","message":"Patch Set 6: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"58f078a1ce39f1d7dcf051a55cec5a9209b21b72","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-10 05:35:21.000000000","message":"Patch Set 6: Review-Priority+1\n\n(2 comments)\n\nThank you Cyril, added inline answer for your comment.","accounts_in_message":[],"_revision_number":6},{"id":"de906f310cf6d7c8635508522ba3b308513916c5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-10 05:54:33.000000000","message":"Patch Set 7: Published edit on patch set 6.","accounts_in_message":[],"_revision_number":7},{"id":"33a6ebd3aa4944506bae4c64f045410835041eee","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-08-10 05:55:06.000000000","message":"Patch Set 7:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":7},{"id":"e4856960102973d10ff84a0c647c455fd4c22dde","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-10 07:31:49.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/d365c5b228fb4035984509c296f2bb77 : SUCCESS in 1h 01m 36s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ca4ea0b2f0ae47d5bbfe285bf3d7dbcd : FAILURE in 1h 05m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8983beaeb3ca4f8da6f20169c0dc0d07 : SUCCESS in 5m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/21692536b3d84f7fa2831fd6208e7f2a : SUCCESS in 6m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8313de9ac3ac4e1495e48e55b18e1e09 : SUCCESS in 7m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bff8257b2bf94c26bba7433117111f11 : SUCCESS in 6m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62108a5fb5c14034bec3e8b94ef452d9 : SUCCESS in 8m 04s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c400fa2f9b0d4f9b977f8765d5d3cb88 : SUCCESS in 7m 55s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/27ab13457c7149fdaff3f7c338b40e3d : SUCCESS in 9m 37s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/0c916653497741f99b5043d23591d987 : SUCCESS in 9m 11s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/029cf335be004fd4a5fe1a8c79f3fe70 : SUCCESS in 5m 19s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/c3b75b881e8b444682912b9cdf3e9b91 : SUCCESS in 1h 15m 55s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/91a62fa19ec54b9c90ef98872ab4cc25 : SUCCESS in 1h 07m 55s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/9e03285f22d04224b3cf1a7aa0ded14d : SUCCESS in 1h 35m 12s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/39f0e18f64c241b7acc0aca6fb2092d2 : SUCCESS in 1h 16m 08s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/82ed284212b048139fc57a710793a308 : SUCCESS in 52m 42s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/952e83e70b6b499dba6dc21b8fc4d4c1 : SUCCESS in 1h 09m 20s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/5f29f58c084d4fee9e59e857c4654c28 : SUCCESS in 43m 59s","accounts_in_message":[],"_revision_number":7},{"id":"5276f2fc3f440c0529e27040010d74461624b3c2","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-10 07:36:40.000000000","message":"Patch Set 7:\n\nrecheck unrelated error","accounts_in_message":[],"_revision_number":7},{"id":"81202da804200e84a428868c3ef483d5cc3c21cc","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-08-10 08:13:00.000000000","message":"Patch Set 7:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/802996/802996/7/check-not-vote/ext-glance-zuul/53374e8 : SUCCESS in 35m 47s","accounts_in_message":[],"_revision_number":7},{"id":"5b6058676d34571093fd2abc15afdea98aca3e0c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-10 08:54:11.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/4f7d94e409914a3784eaf600ed54d4c9 : SUCCESS in 59m 52s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2bff17468f1744f687e57ecd67346bd3 : SUCCESS in 1h 05m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a8d0a14b532472582cfd7034feb3e24 : SUCCESS in 5m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7bd59fd357de41ec91a4603961b64722 : SUCCESS in 6m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1a3ba8b46dd240ffbc21bfa5a041b6e3 : SUCCESS in 6m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9d61e7fa54484f2b8c11d9752a534782 : SUCCESS in 6m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a494bfc4d8bb41119ed5f8396b0e14b2 : SUCCESS in 7m 02s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d8f01e439cf44bd08fc54615c687a3b5 : SUCCESS in 10m 21s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/796e7adb7e064653aa82d588bb050ae8 : SUCCESS in 8m 56s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/d33d446efe0d4a989010aa0e447c0795 : SUCCESS in 7m 44s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/2e8d27b974d94844864c226e0a12f667 : SUCCESS in 5m 45s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/964937ae41774e738796bef207c2f7cf : SUCCESS in 1h 09m 56s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/fcbb55cf01ac4f7d93ec1a1b2e755060 : SUCCESS in 1h 10m 37s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/9bcd1907bfcb4907a3b7d5198b729946 : SUCCESS in 1h 07m 26s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/52edf798c5a34705b14fddf8d41121fb : SUCCESS in 1h 15m 48s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/92370dfc8d0f46b59c2739c3905d7f1c : SUCCESS in 41m 27s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/4dc0d20e22344d0bafdf201ea9d1091c : SUCCESS in 1h 00m 34s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/4080fba801d348b7b3bed9ad82943c2a : SUCCESS in 31m 13s","accounts_in_message":[],"_revision_number":7},{"id":"c56f45249c5ea14c4f8b70f434db4a71e75797dc","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2021-08-10 14:01:21.000000000","message":"Patch Set 7: Code-Review+2\n\nThanks for the clarification and the fix!","accounts_in_message":[],"_revision_number":7},{"id":"9c2cce986fb9c56d9356c7b921634d0a24d3afcf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-11 09:11:22.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"b7601013fb0902cf3693740c7c37e7612926e561","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-11 10:29:32.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/ba8cae6c516a4c048906ed252d2009ff : SUCCESS in 1h 06m 24s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1d7748c01a624521b3353cbcdcc46a0b : SUCCESS in 1h 06m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed69a734c83f4b5f82cf597d505f423c : SUCCESS in 5m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/43bf91d317264b2d8751bc801053fb39 : SUCCESS in 7m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/79f26c1289af4b0c9721a0954c8adaf9 : SUCCESS in 6m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0e7554589efc406180368259b62deb4d : SUCCESS in 9m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/632f4edbfd2e431eb4429ce2bb12a34e : SUCCESS in 6m 30s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/91518608bec348c3ab7a2f17f375bcf2 : SUCCESS in 7m 52s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/af2515e335a944d4a5b112c281ae34ae : SUCCESS in 8m 25s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/2854f17599a149b9a67f86d81d7ebc00 : SUCCESS in 8m 21s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/7040fd82c4034a4aa822212a6c389418 : SUCCESS in 6m 35s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/64e58d34739948489a4297607957689d : SUCCESS in 1h 05m 40s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/a95e8b0155f04e1aa7fbb1608d87e4e2 : SUCCESS in 1h 04m 27s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/cd23acf3b95546bba13fb499c07c15d8 : SUCCESS in 1h 04m 51s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/e0c8d6e86c7b490f98bf77d6f2af41cc : SUCCESS in 1h 13m 23s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a09f306d8eb14da39a4fad47cb803a7d : SUCCESS in 44m 19s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/cf2b688f14b2493ba0fef3c88d4be522 : SUCCESS in 1h 10m 43s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/71d4f21bd3274e729b2ddb69e9755a01 : SUCCESS in 31m 59s","accounts_in_message":[],"_revision_number":8},{"id":"05a726dbdbd3d842a8a933a3e8345a729ef3c0d3","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-08-11 14:38:59.000000000","message":"Patch Set 8: Code-Review-1\n\n(4 comments)\n\nSome typos in comments and other comments. I need to go look at the tempest plugin to see if we\u0027re fully testing the member stuff there, but my feeling is that with some extra rbac policy items we might be missing those details in some test somewhere...","accounts_in_message":[],"_revision_number":8},{"id":"734c01cf0e3cd53637abe6ff809c1096685f5f7c","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-11 14:50:00.000000000","message":"Patch Set 8:\n\n(3 comments)","accounts_in_message":[],"_revision_number":8},{"id":"fe6568107c2c463d1fa5441e3b6499833116be2b","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-11 17:49:46.000000000","message":"Patch Set 8:\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"c34befb91b9017c5423ab68367a5cb7706dc9c77","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-12 09:31:34.000000000","message":"Patch Set 8: Workflow-1\n\nNeed some more work","accounts_in_message":[],"_revision_number":8},{"id":"f8235c1e2965f2f2c4f90957d9eb34c6eceb3572","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-12 09:48:18.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"fbe10f9641e7e0d9fd81777f181b6be29361e32e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-12 10:45:23.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"59f731440525a8a6bae338805d2076adcd26b337","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-12 12:18:32.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/7e46ba4296d348b7b324035691c7216d : SUCCESS in 1h 04m 16s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d6100a27461f4163bc7dcdf96d2753e0 : SUCCESS in 1h 20m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7098f1da1b2d4fcb9f6d6f065a5559f7 : SUCCESS in 5m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c70af827b2b640818b81a6d899f2555f : SUCCESS in 6m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c528fc9ce3284b4691cd44ae81a36ee3 : SUCCESS in 6m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8330a3cc8af04987aa7f82d70d5e28cc : SUCCESS in 6m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6eb58a62775a45e2b75602a681bcf695 : SUCCESS in 6m 42s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8518ea98cc234dad9aff2fd1f76fa4d5 : SUCCESS in 8m 30s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/37a84d0944c044c6af0ea6eb8c40d4fc : SUCCESS in 8m 51s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/301703c535ec462d94b724799a597461 : SUCCESS in 8m 33s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/dc677c1408a9401198c3d20a01108c9b : SUCCESS in 5m 09s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/95cd9ce8cfb7459e841f191bdf08cf4f : SUCCESS in 1h 12m 08s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/eb5e22349a15489ea38430ec46195644 : SUCCESS in 1h 06m 48s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/71ff9c25799040539be37dc967483f32 : SUCCESS in 1h 09m 12s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/8e063ce0de8249ee89fc93785176b151 : SUCCESS in 1h 28m 24s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0714141e97ee4ea08602531312892699 : SUCCESS in 51m 24s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/92b33c05317344e382e12be9e618eddf : SUCCESS in 1h 24m 24s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/1b2ff5ce6922412bbdf21020ee9adbf5 : SUCCESS in 33m 28s","accounts_in_message":[],"_revision_number":9},{"id":"3d40a5a4f58918a91a3ad498e5cf752afb195a69","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-08-12 16:22:49.000000000","message":"Patch Set 9:\n\n(4 comments)\n\nQuestions inline. I did not notice the get_image duplication before, so I think we might want to not do that?","accounts_in_message":[],"_revision_number":9},{"id":"2caad911b5a83e27d9a70b350ddaac6b4ef7b0a4","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-12 16:32:15.000000000","message":"Patch Set 9:\n\n(3 comments)\n\nThank you Dan!","accounts_in_message":[],"_revision_number":9},{"id":"e27d8c88004926a0c3c26fe4f6a575fbe3e728bf","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-12 16:48:48.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"398ea4ff2c0f4b5279fda45474b51e209bae5413","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-08-12 16:57:08.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"75fb8ca351364ed40da80d0c66cdd8f57e447c25","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-12 19:40:46.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"12b0deb1eec1b010d2a0f693b9427e58428986a3","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-12 19:41:20.000000000","message":"Patch Set 10:\n\n(2 comments)\n\nDone, Thank you for suggestions!!","accounts_in_message":[],"_revision_number":10},{"id":"0d4561e941e6bbe1ee5596f5f853c2d41f299a53","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-08-12 20:22:57.000000000","message":"Patch Set 10: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":10},{"id":"da1247717740049c54ab2df8083ca9cee1f3194a","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-12 20:25:42.000000000","message":"Patch Set 10:\n\n(2 comments)","accounts_in_message":[],"_revision_number":10},{"id":"75f0b9dd792f2cf81175443441560cb5c2a6dc0e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-12 20:57:28.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/8ca251833df04f38bb8533a65af1e614 : SUCCESS in 56m 02s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e8391cb3228040a28b8f1f8cf6673b49 : SUCCESS in 1h 01m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d357d17c28948288cb5ad63250c6eff : SUCCESS in 5m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cf4d039adc7842d2a0c85e72abe0f3c6 : SUCCESS in 6m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/303fbfc2f56444cea09dcc9a7b308d86 : SUCCESS in 5m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/63ed58bc4a5b4adfa95b6a848977651a : SUCCESS in 6m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f81ad0d9871c40d080fff2af2ebcd4f8 : SUCCESS in 6m 45s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/99737facc0d54be4aa057a16c2a4a6d3 : SUCCESS in 9m 47s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/ba0fa7dde56f4d1f85bc070c9e3dcd55 : SUCCESS in 10m 49s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/c5cd28ba4061441fa0abbfce68d63c26 : SUCCESS in 9m 00s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/4a93363eb7e64d9aaf98f76c276f522a : SUCCESS in 6m 19s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/edbde00682fd46dbaa0b9da4f2dd87ba : SUCCESS in 1h 06m 55s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/a5195e4dfba84685b2c60167b1ceb5d1 : SUCCESS in 1h 00m 47s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/0f0c8ce63f384ec890b3ce2aca89f02e : SUCCESS in 1h 04m 59s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/47fdcd340b714906a40fe30cc5041c25 : SUCCESS in 1h 14m 03s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8cf2e77fff8f41b5bc2f923ad6bf7842 : SUCCESS in 36m 53s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/fbbb86b051d542969e799105781ef11b : SUCCESS in 1h 06m 21s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/65feabfbf84d4e3e8d04640266687f1b : SUCCESS in 28m 01s","accounts_in_message":[],"_revision_number":10},{"id":"e875a28c08dcbd2f70568502b86e8a7e72038d05","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2021-08-12 21:16:59.000000000","message":"Patch Set 10: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":10},{"id":"c121f662776e482060ee8838feba7ea2d0a82dc5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-12 21:17:31.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"bc86e4b666d7fe9ced14f72680d0eb706e275a5d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-12 22:18:44.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/dba61d4f47944e6a81421f7793c1a2bf : SUCCESS in 59m 02s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6436c4c2b4c3499b99b85eb5186bdd22 : SUCCESS in 53m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e4beb3434454c8ea6b88e2133acceec : SUCCESS in 5m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2d29cc3022164d6db5ac2697ca4ff99a : SUCCESS in 6m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c2455393aee342a6a42bcb72ba753011 : SUCCESS in 5m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac81cab504644390bbc715f3c2629d89 : SUCCESS in 6m 24s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2c33f39d6a2e4f9492c71a746b076559 : SUCCESS in 8m 39s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/07526261f46a43e383ed8813345c3fd1 : SUCCESS in 8m 39s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/eaf170c12b4548ca91f865996ab42be9 : SUCCESS in 52m 21s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/32a3cd5c867a43a685435227994d3883 : SUCCESS in 54m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8edd420774ee4725a749d35201c90ee5 : SUCCESS in 38m 43s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/d96802d5ad8848cab2c874d58cf9a7ac : SUCCESS in 1h 00m 15s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/f640e98898024b599e56da309a47d5eb : SUCCESS in 26m 08s","accounts_in_message":[],"_revision_number":10},{"id":"aa0da90164b7dd9b3cbf623550a5c3099afe220d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-12 22:18:45.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":10},{"id":"16282b2cb193f7928a27758dbb6b2c33121ae4cf","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-12 22:20:14.000000000","message":"Patch Set 10:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ab39935388041c880bed70fc43630de : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"a1ac21e76f16a82b6e29f42884afe3c906d75759","revisions":{"664418a1976393cde8d347b5953430dc107624f6":{"kind":"REWORK","_number":1,"created":"2021-07-30 08:41:25.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/96/802996/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/96/802996/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/96/802996/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/96/802996/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/96/802996/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/96/802996/1"}}},"commit":{"parents":[{"commit":"716fb4be2a10b591271daea11933a18248379c71","subject":"Refactor gateway auth layer for member APIs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/716fb4be2a10b591271daea11933a18248379c71"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-07-27 09:05:01.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-07-30 08:36:45.000000000","tz":0},"subject":"PoC Move member policy checks to API layer","message":"PoC Move member policy checks to API layer\n\nThis change enforce member policy checks in API layer and disables\nread-only checks at auth layer and disables policy layer as well.\n\nToDo:\n1. Carefully verify read only checks at auth layer vs secure RBAC policy\nand redefine RBAC policies to include auth check\n2. Provide additional test coverage\n\nPartially implements: blueprint policy-refactor\nChange-Id: I52e9fbe5ef0e7c3021076d84e9b3df89012c45b9\n\nChange-Id: If425d6263854e3bdb505fbdcb4eecf2e9d98d725\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/664418a1976393cde8d347b5953430dc107624f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/664418a1976393cde8d347b5953430dc107624f6"}]},"branch":"refs/heads/master"},"3a40678e173581a6f79d211fc43382cd6667b2ef":{"kind":"REWORK","_number":2,"created":"2021-08-06 16:47:05.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/96/802996/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/96/802996/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/96/802996/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/96/802996/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/96/802996/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/96/802996/2"}}},"commit":{"parents":[{"commit":"3c35d52b6ef953d191baef78e7871f16b4bf05a1","subject":"Refactor gateway auth layer for member APIs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3c35d52b6ef953d191baef78e7871f16b4bf05a1"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-08-05 16:00:04.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-08-06 16:41:44.000000000","tz":0},"subject":"PoC Move member policy checks to API layer","message":"PoC Move member policy checks to API layer\n\nThis change enforce member policy checks in API layer and disables\nread-only checks at auth layer and disables policy layer as well.\n\nThis patch also changes default RBAC policy defined for \u0027get_member\u0027,\n\u0027get_members\u0027 and \u0027modify_member\u0027.\n\nToDo:\n1. Provide additional test coverage\n2. Fix glance-temepest-plugin \u0027test_update_image_member\u0027 as earlier\nit was allowing acceptance of membership by the image owner\n\nPartially implements: blueprint policy-refactor\n\nChange-Id: If425d6263854e3bdb505fbdcb4eecf2e9d98d725\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3a40678e173581a6f79d211fc43382cd6667b2ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3a40678e173581a6f79d211fc43382cd6667b2ef"}]},"branch":"refs/heads/master"},"a91cd5116256dad519ddde8ac9998e1c43d45fa6":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2021-08-06 18:50:18.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/96/802996/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/96/802996/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/96/802996/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/96/802996/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/96/802996/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/96/802996/3"}}},"commit":{"parents":[{"commit":"3c35d52b6ef953d191baef78e7871f16b4bf05a1","subject":"Refactor gateway auth layer for member APIs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3c35d52b6ef953d191baef78e7871f16b4bf05a1"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-08-05 16:00:04.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-08-06 18:50:18.000000000","tz":0},"subject":"PoC Move member policy checks to API layer","message":"PoC Move member policy checks to API layer\n\nThis change enforce member policy checks in API layer and disables\nread-only checks at auth layer and disables policy layer as well.\n\nThis patch also changes default RBAC policy defined for \u0027get_member\u0027,\n\u0027get_members\u0027 and \u0027modify_member\u0027.\n\nToDo:\n1. Provide additional test coverage\n2. Fix glance-tempest-plugin \u0027test_update_image_member\u0027 as earlier\nit was allowing acceptance of membership by the image owner\n\nPartially implements: blueprint policy-refactor\n\nChange-Id: If425d6263854e3bdb505fbdcb4eecf2e9d98d725\nhttps://review.opendev.org/c/openstack/glance-tempest-plugin/+/803801\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a91cd5116256dad519ddde8ac9998e1c43d45fa6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a91cd5116256dad519ddde8ac9998e1c43d45fa6"}]},"branch":"refs/heads/master","description":"Edit commit message"},"91becc83ecb87296ae27583fb0f2f1fee14a015d":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2021-08-06 18:52:57.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/96/802996/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/96/802996/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/96/802996/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/96/802996/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/96/802996/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/96/802996/4"}}},"commit":{"parents":[{"commit":"3c35d52b6ef953d191baef78e7871f16b4bf05a1","subject":"Refactor gateway auth layer for member APIs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3c35d52b6ef953d191baef78e7871f16b4bf05a1"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-08-05 16:00:04.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-08-06 18:52:57.000000000","tz":0},"subject":"PoC Move member policy checks to API layer","message":"PoC Move member policy checks to API layer\n\nThis change enforce member policy checks in API layer and disables\nread-only checks at auth layer and disables policy layer as well.\n\nThis patch also changes default RBAC policy defined for \u0027get_member\u0027,\n\u0027get_members\u0027 and \u0027modify_member\u0027.\n\nToDo:\n1. Provide additional test coverage\n2. Fix glance-tempest-plugin \u0027test_update_image_member\u0027 as earlier\nit was allowing acceptance of membership by the image owner\n\nPartially implements: blueprint policy-refactor\n\nChange-Id: If425d6263854e3bdb505fbdcb4eecf2e9d98d725\nDepends-On: https://review.opendev.org/c/openstack/glance-tempest-plugin/+/803801\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/91becc83ecb87296ae27583fb0f2f1fee14a015d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/91becc83ecb87296ae27583fb0f2f1fee14a015d"}]},"branch":"refs/heads/master","description":"Edit commit message"},"3fa167361a3cea021bf265d5e757bc34ac39dcd7":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2021-08-06 18:54:36.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/96/802996/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/96/802996/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/96/802996/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/96/802996/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/96/802996/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/96/802996/5"}}},"commit":{"parents":[{"commit":"3c35d52b6ef953d191baef78e7871f16b4bf05a1","subject":"Refactor gateway auth layer for member APIs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3c35d52b6ef953d191baef78e7871f16b4bf05a1"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-08-05 16:00:04.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-08-06 18:54:36.000000000","tz":0},"subject":"PoC Move member policy checks to API layer","message":"PoC Move member policy checks to API layer\n\nThis change enforce member policy checks in API layer and disables\nread-only checks at auth layer and disables policy layer as well.\n\nThis patch also changes default RBAC policy defined for \u0027get_member\u0027,\n\u0027get_members\u0027 and \u0027modify_member\u0027.\n\nToDo:\n1. Provide additional test coverage\n2. Fix glance-tempest-plugin \u0027test_update_image_member\u0027 as earlier\nit was allowing acceptance of membership by the image owner\n(https://review.opendev.org/c/openstack/glance-tempest-plugin/+/803801)\n\nPartially implements: blueprint policy-refactor\n\nChange-Id: If425d6263854e3bdb505fbdcb4eecf2e9d98d725\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3fa167361a3cea021bf265d5e757bc34ac39dcd7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3fa167361a3cea021bf265d5e757bc34ac39dcd7"}]},"branch":"refs/heads/master","description":"Edit commit message"},"0fa6bb9c192e2d98ebcb06e3cbb95f1378fa9ebd":{"kind":"REWORK","_number":6,"created":"2021-08-08 19:53:01.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/96/802996/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/96/802996/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/96/802996/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/96/802996/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/96/802996/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/96/802996/6"}}},"commit":{"parents":[{"commit":"3c35d52b6ef953d191baef78e7871f16b4bf05a1","subject":"Refactor gateway auth layer for member APIs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3c35d52b6ef953d191baef78e7871f16b4bf05a1"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-08-05 16:00:04.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-08-08 19:47:42.000000000","tz":0},"subject":"Move member policy checks to API layer","message":"Move member policy checks to API layer\n\nThis change enforce member policy checks in API layer and disables\nread-only checks at auth layer and disables policy layer as well.\n\nThis patch also changes default RBAC policy defined for \u0027get_member\u0027,\n\u0027get_members\u0027 and \u0027modify_member\u0027.\n\nPartially implements: blueprint policy-refactor\n\nChange-Id: If425d6263854e3bdb505fbdcb4eecf2e9d98d725\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0fa6bb9c192e2d98ebcb06e3cbb95f1378fa9ebd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0fa6bb9c192e2d98ebcb06e3cbb95f1378fa9ebd"}]},"branch":"refs/heads/master"},"ad6450b42040d320b8b13bbb628fff6e3f1970be":{"kind":"REWORK","_number":7,"created":"2021-08-10 05:54:33.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/96/802996/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/96/802996/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/96/802996/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/96/802996/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/96/802996/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/96/802996/7"}}},"commit":{"parents":[{"commit":"3c35d52b6ef953d191baef78e7871f16b4bf05a1","subject":"Refactor gateway auth layer for member APIs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3c35d52b6ef953d191baef78e7871f16b4bf05a1"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-08-05 16:00:04.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-08-10 05:54:27.000000000","tz":0},"subject":"Move member policy checks to API layer","message":"Move member policy checks to API layer\n\nThis change enforce member policy checks in API layer and disables\nread-only checks at auth layer and disables policy layer as well.\n\nThis patch also changes default RBAC policy defined for \u0027get_member\u0027,\n\u0027get_members\u0027 and \u0027modify_member\u0027.\n\nPartially implements: blueprint policy-refactor\n\nChange-Id: If425d6263854e3bdb505fbdcb4eecf2e9d98d725\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ad6450b42040d320b8b13bbb628fff6e3f1970be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ad6450b42040d320b8b13bbb628fff6e3f1970be"}]},"branch":"refs/heads/master"},"4fdb90354aff1952e1d3411e17691eef6321d243":{"kind":"REWORK","_number":8,"created":"2021-08-11 09:11:22.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/96/802996/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/96/802996/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/96/802996/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/96/802996/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/96/802996/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/96/802996/8"}}},"commit":{"parents":[{"commit":"d9b8023410348e769f8a69c168461126ce7ff6fc","subject":"Refactor gateway auth layer for member APIs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d9b8023410348e769f8a69c168461126ce7ff6fc"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-08-05 16:00:04.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-08-11 09:07:34.000000000","tz":0},"subject":"Move member policy checks to API layer","message":"Move member policy checks to API layer\n\nThis change enforce member policy checks in API layer and disables\nread-only checks at auth layer and disables policy layer as well.\n\nThis patch also changes default RBAC policy defined for \u0027get_member\u0027,\n\u0027get_members\u0027 and \u0027modify_member\u0027.\n\nPartially implements: blueprint policy-refactor\n\nChange-Id: If425d6263854e3bdb505fbdcb4eecf2e9d98d725\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4fdb90354aff1952e1d3411e17691eef6321d243"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4fdb90354aff1952e1d3411e17691eef6321d243"}]},"branch":"refs/heads/master"},"39de45e4cc78b28a8223da28dbe96ad89f9caa5a":{"kind":"REWORK","_number":9,"created":"2021-08-12 10:45:23.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/96/802996/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/96/802996/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/96/802996/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/96/802996/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/96/802996/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/96/802996/9"}}},"commit":{"parents":[{"commit":"99331d7efacb556effc9dda2a7db06bdc6af0319","subject":"Move metadef namespace policy checks in the API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/99331d7efacb556effc9dda2a7db06bdc6af0319"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-08-12 09:06:53.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-08-12 10:41:31.000000000","tz":0},"subject":"Move member policy checks to API layer","message":"Move member policy checks to API layer\n\nThis change enforce member policy checks in API layer and disables\nread-only checks at auth layer and disables policy layer as well.\n\nThis patch also changes default RBAC policy defined for \u0027get_member\u0027,\n\u0027get_members\u0027 and \u0027modify_member\u0027.\n\nPartially implements: blueprint policy-refactor\n\nChange-Id: If425d6263854e3bdb505fbdcb4eecf2e9d98d725\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/39de45e4cc78b28a8223da28dbe96ad89f9caa5a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/39de45e4cc78b28a8223da28dbe96ad89f9caa5a"}]},"branch":"refs/heads/master"},"a1ac21e76f16a82b6e29f42884afe3c906d75759":{"kind":"REWORK","_number":10,"created":"2021-08-12 19:40:46.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/96/802996/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/96/802996/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/96/802996/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/96/802996/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/96/802996/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/96/802996/10"}}},"commit":{"parents":[{"commit":"8e656a91e6c5809401a70142cab21d4d9dcc5462","subject":"Move metadef object policy checks in the API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8e656a91e6c5809401a70142cab21d4d9dcc5462"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-08-12 09:06:53.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-08-12 19:32:09.000000000","tz":0},"subject":"Move member policy checks to API layer","message":"Move member policy checks to API layer\n\nThis change enforce member policy checks in API layer and disables\nread-only checks at auth layer and disables policy layer as well.\n\nThis patch also changes default RBAC policy defined for \u0027get_member\u0027,\n\u0027get_members\u0027 and \u0027modify_member\u0027.\n\nPartially implements: blueprint policy-refactor\n\nChange-Id: If425d6263854e3bdb505fbdcb4eecf2e9d98d725\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a1ac21e76f16a82b6e29f42884afe3c906d75759"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a1ac21e76f16a82b6e29f42884afe3c906d75759"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
