)]}'
{"id":"openstack%2Fglance~803240","triplet_id":"openstack%2Fglance~master~I111fc36dcf3bc203f21377a3095a2170d00bb9bf","project":"openstack/glance","branch":"master","topic":"803240","attention_set":{},"removed_from_attention_set":{"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2021-11-24 12:09:50.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I111fc36dcf3bc203f21377a3095a2170d00bb9bf","subject":"Refactor glance cinder tests","status":"MERGED","created":"2021-08-02 17:43:53.000000000","updated":"2021-11-29 09:36:54.000000000","submitted":"2021-11-24 12:09:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"803240-803240","meta_rev_id":"f9c9c58dfd6ce35ba936d083661afc1b2fe640a7","_number":803240,"virtual_id_number":803240,"owner":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-11-24 12:09:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2021-11-24 08:00:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2021-11-24 08:00:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]}],"REVIEWER":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-02 18:18:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-03 00:27:25.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-11-24 08:00:50.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"}],"messages":[{"id":"7aeecc3c64d60dfc49e317d43c393ac9f0da347e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2021-08-02 17:43:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f931955f22f6faf779b578ef9ccb4d5762f80970","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-02 18:18:20.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/986346dfea224569986fc81b9d326d77 : SUCCESS in 5m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/097c4b78887540fb889164163a8f7fc4 : SUCCESS in 6m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f58f4ef8f0f34c69aeee143925205c88 : SUCCESS in 5m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/015b485384ba4ee086304e37c313aa99 : SUCCESS in 7m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/537b4e2b231c4e7a8cf4f59cf1ba335d : SUCCESS in 6m 13s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/61990cfa95c646ad8ccf8a0f227a8d7f : FAILURE in 8m 05s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/12b4c9090b664dc08115a9729eaaf6a0 : FAILURE in 9m 22s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/210e8288822345a286277a20cccb1556 : FAILURE in 9m 00s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/a81b91cd68f24a189f8598c99de6dea1 : SUCCESS in 5m 47s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/40c19d4fc0a84a21829321fcc959acd2 : SUCCESS in 29m 59s","accounts_in_message":[],"_revision_number":1},{"id":"1f6f9575dd41d28f2f0102166287787fe9533093","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-08-03 00:27:25.000000000","message":"Patch Set 1:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/803240/803240/1/check-not-vote/ext-glance-zuul/e3cc5f0 : SUCCESS in 34m 10s","accounts_in_message":[],"_revision_number":1},{"id":"cd36020735deb8774882f964f2e5309803165126","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2021-08-10 13:35:58.000000000","message":"Patch Set 1: Workflow-1\n\nThis will pass gate when glance client releases","accounts_in_message":[],"_revision_number":1},{"id":"8379e0f2da5ef1c8cf672e7394fffe6e6388de90","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2021-11-24 07:10:18.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"dfb6e60d011cf7af1d519323f1ca2f05da67963f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-24 07:55:31.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0dbe3e0bcfda4505af4aa4340767896a : SUCCESS in 6m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ac62df793cdf4a17851f310724647b18 : SUCCESS in 9m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d8a08c8ca60a4ad5be1176dca1cf77df : SUCCESS in 5m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/956215ddc1e34b6882a04e295209621b : SUCCESS in 5m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8785110ab294715bdf75d81e2ae25f7 : SUCCESS in 6m 13s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f3ba8b6f45294b6397395e1ba3c54f40 : SUCCESS in 8m 02s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/15d1c816a668413db4fe0c57c79b4e7f : SUCCESS in 8m 44s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/630c55fb3444419696c93d8612c1353b : SUCCESS in 9m 55s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/3ac293f057f34a9cb0d5dface6e2d343 : SUCCESS in 5m 06s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/33b3ace2facd407a99f7552e859cd4fb : SUCCESS in 38m 12s","accounts_in_message":[],"_revision_number":2},{"id":"980b17485b52b8721e5ca9940ce14f9d8ed96570","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-11-24 08:00:50.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"722710105c299464ebe13d05eb5a0972585c7735","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-24 08:01:42.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"af88dc548726ff48856c5775c4adaff36af751df","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-24 12:09:49.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e7a472f94608456393732dea4121570e : SUCCESS in 7m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8a58eeb7af0848bf9f94fce20ee30188 : SUCCESS in 10m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4143d6a6a038446282c4ace71b234b57 : SUCCESS in 8m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eff5d604bebc48d5847d691a59ef44fa : SUCCESS in 7m 01s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2fa385cb04f5451ca041b2193133b0db : SUCCESS in 10m 36s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/50a7d41061dd47bd8c1f12db6b503d5d : SUCCESS in 12m 39s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/cca0df1016b64bae933dbcfb6bf05e18 : SUCCESS in 36m 08s","accounts_in_message":[],"_revision_number":2},{"id":"b8a953792c76a281c72fd6138c7ee9c9cd3a70a5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-24 12:09:50.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"e58506e8d1c588b3e7394303d5076dc834c1d7bf","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-24 12:12:03.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2661a003b0fc4f2eb9f992250c2a836e : SUCCESS in 1m 14s","accounts_in_message":[],"_revision_number":2},{"id":"f9c9c58dfd6ce35ba936d083661afc1b2fe640a7","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-11-29 09:36:54.000000000","message":"Patch Set 2:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/803240/803240/2/check-not-vote/ext-glance-zuul/ddaba25 : FAILURE in 1m 47s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6fd949a966fea443bb4b0b9351b3fa04662a7ecb","revisions":{"21cace850d7e7b38527ee43df9deb6e4277c884a":{"kind":"REWORK","_number":1,"created":"2021-08-02 17:43:53.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/40/803240/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/40/803240/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/40/803240/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/40/803240/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/40/803240/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/40/803240/1"}}},"commit":{"parents":[{"commit":"429f16124b7cd69c85e6d764bcf1e47687780286","subject":"Remove dead 403-\u003e404 code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/429f16124b7cd69c85e6d764bcf1e47687780286"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2021-08-02 17:40:10.000000000","tz":-240},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2021-08-02 17:43:48.000000000","tz":-240},"subject":"Refactor glance cinder tests","message":"Refactor glance cinder tests\n\nLegacy attachment API calls were removed/replaced from the code with\nnew attachment API support in glance store with change[1]\nThis patch removes the same methods mocked in the tests.\n\n[1] https://review.opendev.org/c/openstack/glance_store/+/782200\n\nChange-Id: I111fc36dcf3bc203f21377a3095a2170d00bb9bf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/21cace850d7e7b38527ee43df9deb6e4277c884a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/21cace850d7e7b38527ee43df9deb6e4277c884a"}]},"branch":"refs/heads/master"},"6fd949a966fea443bb4b0b9351b3fa04662a7ecb":{"kind":"REWORK","_number":2,"created":"2021-11-24 07:10:18.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/40/803240/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/40/803240/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/40/803240/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/40/803240/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/40/803240/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/40/803240/2"}}},"commit":{"parents":[{"commit":"1db6e66f93535120fdd974fce7e6490063f3c863","subject":"Merge \"Remove broken tempest-full-py3-opensuse15 job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1db6e66f93535120fdd974fce7e6490063f3c863"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2021-08-02 17:40:10.000000000","tz":-240},"committer":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2021-11-24 07:09:49.000000000","tz":-300},"subject":"Refactor glance cinder tests","message":"Refactor glance cinder tests\n\nLegacy attachment API calls were removed/replaced from the code with\nnew attachment API support in glance store with change[1]\nThis patch removes the same methods mocked in the tests.\n\n[1] https://review.opendev.org/c/openstack/glance_store/+/782200\n\nChange-Id: I111fc36dcf3bc203f21377a3095a2170d00bb9bf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6fd949a966fea443bb4b0b9351b3fa04662a7ecb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6fd949a966fea443bb4b0b9351b3fa04662a7ecb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
