)]}'
{"id":"openstack%2Fglance~803590","triplet_id":"openstack%2Fglance~master~I65b82d60fb31c8004e44c7a054e3d24cb28654fd","project":"openstack/glance","branch":"master","topic":"policy-refactor","hashtags":[],"change_id":"I65b82d60fb31c8004e44c7a054e3d24cb28654fd","subject":"Move Tasks policy checks in the API","status":"ABANDONED","created":"2021-08-05 12:09:40.000000000","updated":"2021-08-05 20:19:58.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"c9303a6aa4f092ac883d87cee7a146878be7d18b","_number":803590,"virtual_id_number":803590,"owner":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2021-08-05 13:44:23.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-05 13:44:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b69c308bd098b2915332fe28262acba7e2ca23a1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-05 12:09:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cf20c0890d46a6ce32eac982923e479de2072e1d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-05 13:44:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/2ae0df6b26cc451794a9b6016c826651 : SUCCESS in 1h 19m 21s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/28214dd70a6441bea7ea9c8efe0948ad : SUCCESS in 58m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb0e9e33201c47ff900c26f55a63942e : SUCCESS in 5m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a28a8f7f8d634ee5927287074d72cb02 : SUCCESS in 5m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/41f7a53d7ac945d288a34ddb4eff938d : SUCCESS in 6m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/081a1390eda54ab5aeaded295217d225 : SUCCESS in 6m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4b42c5ca9034dd7afc036ea2fbad776 : SUCCESS in 7m 58s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d77676e09e7141899f7e47fe2696c227 : SUCCESS in 8m 43s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/5fea89a2a99d438e9498bf2d81be35fb : SUCCESS in 8m 51s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/f7e1d28d23b74f39b7685869dae7c0b0 : SUCCESS in 8m 13s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/307e1ad1d46c4190bba404b8a02e0799 : SUCCESS in 6m 00s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/439564ad584340ae9cdefe598934bad4 : SUCCESS in 1h 08m 49s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/d9a3965359b04e279a3aacc0b6768b95 : SUCCESS in 1h 02m 55s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/f9a09dd653054498b4a8519400964c77 : SUCCESS in 1h 04m 42s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/d1385fc4dd114f18b4ac1287425533a0 : SUCCESS in 1h 31m 42s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e2be56612b3d4133a3ac6863f9f0f03e : SUCCESS in 56m 23s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/fb1318896d454d758d951d994f7e7745 : SUCCESS in 1h 03m 18s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/7a88ba040c404892832931bb0a078544 : SUCCESS in 30m 47s","accounts_in_message":[],"_revision_number":1},{"id":"c9303a6aa4f092ac883d87cee7a146878be7d18b","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-05 20:19:58.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7fbd5a7236a57106e550f6dbba930c5f4d65ea6a","revisions":{"7fbd5a7236a57106e550f6dbba930c5f4d65ea6a":{"kind":"REWORK","_number":1,"created":"2021-08-05 12:09:40.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/90/803590/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/90/803590/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/90/803590/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/90/803590/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/90/803590/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/90/803590/1"}}},"commit":{"parents":[{"commit":"300f59d83b1d896452fca79dc201d48ec08e81bd","subject":"Deprecate task specific policies","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/300f59d83b1d896452fca79dc201d48ec08e81bd"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-07-26 07:34:02.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-08-05 12:06:17.000000000","tz":0},"subject":"Move Tasks policy checks in the API","message":"Move Tasks policy checks in the API\n\nThis change uses new policy module to enforce tasks_api_access\npolicy check and excludes policy enforcement from the policy layer.\n\nNOTE: Since we are excluding policy enforcement at policy layer, removed\nbelow unit tests as those were checking controller methods and we no longer\nenforcing policies at policy layer.\ntest_create_task_unauthorized\ntest_index_unauthorized\ntest_get_unauthorized\n\nPartially implements: blueprint policy-refactor\nChange-Id: I670c076527b7bce55696b294d2283121dab5c707\n\nChange-Id: I65b82d60fb31c8004e44c7a054e3d24cb28654fd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7fbd5a7236a57106e550f6dbba930c5f4d65ea6a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7fbd5a7236a57106e550f6dbba930c5f4d65ea6a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
