)]}'
{"id":"openstack%2Fglance~803619","triplet_id":"openstack%2Fglance~stable%2Fvictoria~I9356bceb914327f526da7b727fa58522ae18856e","project":"openstack/glance","branch":"stable/victoria","topic":"ssl_doc-vic","hashtags":[],"change_id":"I9356bceb914327f526da7b727fa58522ae18856e","subject":"Remove SSL configuration section from docs","status":"MERGED","created":"2021-08-05 15:05:46.000000000","updated":"2021-09-09 16:27:23.000000000","submitted":"2021-09-09 16:25:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"803619-ssl_doc-vic","meta_rev_id":"591adc9dd6643bec8cbf71dcbfee4b96c3d81266","_number":803619,"virtual_id_number":803619,"owner":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-09 16:25:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":2,"date":"2021-09-09 14:22:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":1,"date":"2021-09-09 15:31:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-05 15:25:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-09 14:22:55.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"}],"messages":[{"id":"76305ee4fcfafe63f40a4bb3affe3a0188a13146","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2021-08-05 15:05:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6b9b6c1c46fb6cff53c3f0c8aaa5c74cb297119d","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2021-08-05 15:09:34.000000000","message":"Patch Set 1: Workflow-1\n\noops ... mistake on the merge conflict handling","accounts_in_message":[],"_revision_number":1},{"id":"607dc36e78dc474b2b6c42db8c1300306519059f","tag":"autogenerated:gerrit:abandon","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2021-08-05 15:10:23.000000000","message":"Abandoned\n\nneed to redo the backport","accounts_in_message":[],"_revision_number":1},{"id":"fb180eef427c436d8d241a86dd0bc961b8eb989b","tag":"autogenerated:gerrit:restore","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2021-08-05 15:14:46.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"265fad27c2ca3151555047f86d7d84b4e6d8db2c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2021-08-05 15:14:57.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e880ab4ac541fd59fa1722d582797e7b18b996b8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-05 15:25:43.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/603dd74b2c3344b5a861da6ab5e9a949 : SUCCESS in 6m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0313a8e0cf2548729f66568d804acb47 : SUCCESS in 6m 40s","accounts_in_message":[],"_revision_number":2},{"id":"aeff22eff81aea934133714090f6efb71f7fd242","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-09-09 14:22:55.000000000","message":"Patch Set 2: Code-Review+2\n\nLooks good, will approve once this stable/wallaby backport merges.","accounts_in_message":[],"_revision_number":2},{"id":"59b62731f60d8753bb2a8c5a7386a7d1e8883cc1","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-09-09 15:31:08.000000000","message":"Patch Set 2: Workflow+1\n\nGood to go, merged in stable/wallaby!","accounts_in_message":[],"_revision_number":2},{"id":"8f365e4d7b7cd093a709e3d88b61770a5a3bf1b1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-09 15:31:36.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"02f1c29847d729fa57f3476589c2b6ace2a8bed0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-09 16:25:35.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84a23c8f41da4ad3acd7a9372afec0f5 : SUCCESS in 6m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f51db61b67e441928b9f1aa93ace2339 : SUCCESS in 7m 20s","accounts_in_message":[],"_revision_number":2},{"id":"b298bcc395ed38fb1d0a000a98cbb0ad0183b952","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-09 16:25:35.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"591adc9dd6643bec8cbf71dcbfee4b96c3d81266","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-09 16:27:23.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50f34dae722c476380c2a2091267ffc4 : SUCCESS in 1m 21s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"2d77038b2da039c1a56d00fc46a62295c93b7f98","revisions":{"7f5d46481eaaa99f520e338d7052c7186c1d3bab":{"kind":"REWORK","_number":1,"created":"2021-08-05 15:05:46.000000000","uploader":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"ref":"refs/changes/19/803619/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/19/803619/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/19/803619/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/19/803619/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/19/803619/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/19/803619/1"}}},"commit":{"parents":[{"commit":"958840f29a6294bd0e1932b0a3298282bdc2d8b6","subject":"Merge \"Add housekeeping module and staging cleaner\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/958840f29a6294bd0e1932b0a3298282bdc2d8b6"}]}],"author":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2021-07-29 12:02:20.000000000","tz":60},"committer":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2021-08-05 15:05:14.000000000","tz":60},"subject":"Remove SSL configuration section from docs","message":"Remove SSL configuration section from docs\n\nSince supporting only PY3 (Ussuri) Glance has not been supporting\ntermination of encrypted connection to the service [0]. The\nsection was left behind on the configuring doc.\n\n[0] https://docs.openstack.org/releasenotes/glance/ussuri.html#security-issues\n\nChange-Id: I9356bceb914327f526da7b727fa58522ae18856e\nCloses-Bug: #1930597\n(cherry picked from commit 652780d0299d5734cbc6d21eea74c4fd11797cd1)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7f5d46481eaaa99f520e338d7052c7186c1d3bab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7f5d46481eaaa99f520e338d7052c7186c1d3bab"}]},"branch":"refs/heads/stable/victoria"},"2d77038b2da039c1a56d00fc46a62295c93b7f98":{"kind":"REWORK","_number":2,"created":"2021-08-05 15:14:57.000000000","uploader":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"ref":"refs/changes/19/803619/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/19/803619/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/19/803619/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/19/803619/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/19/803619/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/19/803619/2"}}},"commit":{"parents":[{"commit":"958840f29a6294bd0e1932b0a3298282bdc2d8b6","subject":"Merge \"Add housekeeping module and staging cleaner\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/958840f29a6294bd0e1932b0a3298282bdc2d8b6"}]}],"author":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2021-07-29 12:02:20.000000000","tz":60},"committer":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2021-08-05 15:14:21.000000000","tz":60},"subject":"Remove SSL configuration section from docs","message":"Remove SSL configuration section from docs\n\nSince supporting only PY3 (Ussuri) Glance has not been supporting\ntermination of encrypted connection to the service [0]. The\nsection was left behind on the configuring doc.\n\n[0] https://docs.openstack.org/releasenotes/glance/ussuri.html#security-issues\n\nChange-Id: I9356bceb914327f526da7b727fa58522ae18856e\nCloses-Bug: #1930597\n(cherry picked from commit 652780d0299d5734cbc6d21eea74c4fd11797cd1)\n(cherry picked from commit 62e921ecd9ba2cd33404347ddffcacd345a5ca74)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2d77038b2da039c1a56d00fc46a62295c93b7f98"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2d77038b2da039c1a56d00fc46a62295c93b7f98"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
