)]}'
{"id":"openstack%2Fglance~803937","triplet_id":"openstack%2Fglance~master~I2abd1e60f414fbd68ce84e0b280f8b3e4e791a82","project":"openstack/glance","branch":"master","topic":"pre-cache-uwsgi","hashtags":[],"change_id":"I2abd1e60f414fbd68ce84e0b280f8b3e4e791a82","subject":"[uwsgi] Add missing pefetch periodic job","status":"MERGED","created":"2021-08-09 15:27:32.000000000","updated":"2021-09-14 18:41:18.000000000","submitted":"2021-09-14 18:39:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":14,"unresolved_comment_count":5,"has_review_started":true,"submission_id":"803937-pre-cache-uwsgi","meta_rev_id":"cb6ccde549d948643404ee2302ca92e631628a26","_number":803937,"virtual_id_number":803937,"owner":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-14 18:39:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":2,"date":"2021-09-09 05:14:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":2,"date":"2021-09-14 16:28:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":1,"date":"2021-09-14 16:28:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":1,"date":"2021-09-09 05:14:13.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-09 15:29:31.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2021-08-09 15:29:31.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2021-08-09 16:44:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-10 13:39:37.000000000","updated_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"},{"updated":"2021-09-11 19:19:00.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"29cab406727132b63b2485415921a6b4dda35dc1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-09 15:27:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"49705bced478ce7f808f142a2fd56a3150a71d73","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-09 15:34:57.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"67cc824232e74c05d5ff4421f2ae4a30f3b510b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-09 16:44:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/46a95a22bccf41169d02121ac211dbd9 : SUCCESS in 57m 34s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2a672845228d42a7a8605e0d40a7e828 : SUCCESS in 1h 13m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/00b6b39631cf47efaa383d918accb287 : SUCCESS in 5m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a56f7f68274046878722e88ba6c29149 : SUCCESS in 6m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/90489ff2363a49d9aaac6316e268f107 : SUCCESS in 6m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/015cf0cbaea542a1ad66cdb78836a27d : SUCCESS in 6m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99f6c857da2c49ebbd907966692a359f : SUCCESS in 7m 11s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ce6368b7ba1541ffad00684aab15bee5 : SUCCESS in 7m 53s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/5913808206b9470aa283b8cdbe9f9570 : SUCCESS in 8m 30s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/fe65566fc1574c1f9c92bab8933d5af4 : SUCCESS in 8m 16s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/293be6e3713a4aa3a8d6a49362a5ac0a : SUCCESS in 6m 03s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/5848324f1efc438f9b9742e466b80c91 : SUCCESS in 1h 05m 55s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/f70cffe3c85c410db0c97b3837b3f20d : SUCCESS in 1h 10m 14s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/45ff702a48974864af61f3dd18a7616a : SUCCESS in 1h 07m 30s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/19a4e16a1c944deb804a39c5c4bd536e : SUCCESS in 1h 05m 23s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e9efde6b94b94bafb197a349c9af1722 : SUCCESS in 44m 29s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/5500bb01e2b64cc6b9890ef09e970b7a : SUCCESS in 1h 14m 41s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/367a0f93b269472ca8ef077ac890e4c7 : SUCCESS in 33m 41s","accounts_in_message":[],"_revision_number":1},{"id":"bf56ffea1771719b262f7c7e54234ae99500b0ca","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-09 16:51:50.000000000","message":"Patch Set 1: Workflow-1\n\nBy default glance uwsgi starts with 2 processes which means this periodic job will run twice each time with image_cache_dir shared across them, so might need to add locking here to avoid any issues?","accounts_in_message":[],"_revision_number":1},{"id":"01cf2a53c794a2f379614ad88e9d66f3cd79e56c","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2021-08-10 13:39:37.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)\n\n\u003e Patch Set 1: Workflow-1\n\u003e \n\u003e By default glance uwsgi starts with 2 processes which means this periodic job will run twice each time with image_cache_dir shared across them, so might need to add locking here to avoid any issues?\n\nI don\u0027t think there is any issues running multiple prefetchers at the time (we used to do it also when it was ran by cron if the previous was not finished by the time it was called again). Worth of looking into \u0027though.","accounts_in_message":[],"_revision_number":1},{"id":"4fc5201b4de159ca4cfc4313bbfbcd28b9982347","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-08-10 13:44:27.000000000","message":"Patch Set 1:\n\n\u003e By default glance uwsgi starts with 2 processes which means this periodic job will run twice each time with image_cache_dir shared across them, so might need to add locking here to avoid any issues?\n\nWell, that\u0027s default but real deployments will likely run more. However, people assign a worker count based on API workload, so randomly having one (or more) of those check out to go do prefetches is probably not what they wanted. I\u0027ll have to look at the implementation here, but I would expect that we want one worker to spawn a new native thread to do that thing per cycle.","accounts_in_message":[],"_revision_number":1},{"id":"f6b5c403fcb19f25227d276f83c140a2621c4ce3","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-08-10 13:57:28.000000000","message":"Patch Set 1: Code-Review-1\n\n(3 comments)\n\nDoing this from inside a wsgi worker is really not the best design in general, especially since people expect WSGI workers to be handling API requests and be somewhat expendable (which is why the wsgi container will kill them off regularly).\n\nIf we\u0027re going to keep this model, I think it probably makes sense to use some sort of external lock and timer to make sure that we\u0027re not running more than one in parallel (as this seems to do) nor more often than we want.\n\nSomething that uses a file in the cache directory as both a lock and a timer (i.e. don\u0027t start a caching run unless the mtime on that file is older than the cache interval) would be better I think.\n\nRegardless, in the short term, the pool model stuff needs to not use GreenPool I think. If we\u0027re going to fix that and merge this so that something works, we might want a reno or doc update saying that all the workers will run this, so adjust accordingly.","accounts_in_message":[],"_revision_number":1},{"id":"b533ae2ed2a9dcc5b5ec4358abbb611ebe2d1c06","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-10 14:15:22.000000000","message":"Patch Set 1:\n\n(3 comments)\n\n\u003e Patch Set 1: Code-Review-1\n\u003e \n\u003e (3 comments)\n\u003e \n\u003e Doing this from inside a wsgi worker is really not the best design in general, especially since people expect WSGI workers to be handling API requests and be somewhat expendable (which is why the wsgi container will kill them off regularly).\n\u003e \n\u003e If we\u0027re going to keep this model, I think it probably makes sense to use some sort of external lock and timer to make sure that we\u0027re not running more than one in parallel (as this seems to do) nor more often than we want.\n\u003e \n\u003e Something that uses a file in the cache directory as both a lock and a timer (i.e. don\u0027t start a caching run unless the mtime on that file is older than the cache interval) would be better I think.\n\u003e \n\u003e Regardless, in the short term, the pool model stuff needs to not use GreenPool I think. If we\u0027re going to fix that and merge this so that something works, we might want a reno or doc update saying that all the workers will run this, so adjust accordingly.\n\nAs we are so much overloaded at this moment and also release is near, IMO, we should keep this model with some Release note at this moment and enhance it in next cycle.","accounts_in_message":[],"_revision_number":1},{"id":"41b65456cfd648b274e5b0fecf8500c567703e46","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-31 14:16:59.000000000","message":"Patch Set 1:\n\nThis is postponed to Yoga due to volume of work remaining is not possible to cover in this cycle.","accounts_in_message":[],"_revision_number":1},{"id":"d722736fd3647ace4a8bd5542e78e230a5b111da","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-09-08 16:02:45.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"72ff0b0f571434d69bf7f33c18ad13ae04215fb1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-09-08 17:02:53.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"b1265b08481d13de84a987b8ac5b02eb0796833b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-08 18:22:39.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/1a958014837f47be9f707375cc971be7 : SUCCESS in 1h 01m 02s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d61c8ff1de804daca389502bf20390c1 : SUCCESS in 1h 04m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/448e0bd977444806ba9ab846bfa2b8af : SUCCESS in 5m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0d4a8f5ca75a467783048ed859136ce0 : SUCCESS in 5m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/21ca18150c084b39b74fe362b4316658 : SUCCESS in 6m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/00aa19c50b644117ae47cfe5764fdb58 : SUCCESS in 6m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f0f6a3ba35e4373b53aa5372bc99e3b : SUCCESS in 6m 52s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4f97f2e60ac34de29d713d41ef5a9ced : SUCCESS in 8m 50s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/a361605b3f794c219889cd9d5bb9dabb : SUCCESS in 8m 43s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/a5dadb8d8b86468da01160b7f64376e6 : SUCCESS in 10m 15s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/eaef1b82282948b38d17a8a7a34095f2 : SUCCESS in 5m 52s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/872b5c4aabf34c10bacb0c867ceba6ec : SUCCESS in 1h 15m 31s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/2ae86dd45d3f4feba70c2e70797415aa : SUCCESS in 1h 05m 34s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/96e07b2da0e74add8bae40915048fecd : SUCCESS in 1h 13m 54s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/d792846b405148cab29caa6c238dcf75 : SUCCESS in 1h 10m 31s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1178cd5aa8aa4c3ca8dd08c4f239932f : SUCCESS in 47m 51s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/67807e29c7784c3191457a9c8034cec3 : SUCCESS in 1h 09m 15s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/72cbfc8c3a0c453798d006e06c79015a : SUCCESS in 31m 15s","accounts_in_message":[],"_revision_number":3},{"id":"18d83aa080382089da251cf3a688db368c5bd259","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-09-09 05:14:13.000000000","message":"Patch Set 3: Code-Review+2 Review-Priority+1\n\nRan various functional scenarios in local environment. This is working as expected. We have better unit test coverage with this patch now.\n\nThank you Dan for adding locking mechanism.\n\nI think this should be part of Xena rather than going forward and backport it to stable/xena.","accounts_in_message":[],"_revision_number":3},{"id":"ddf42606387510d49a73aee34f7395ebe0ea5397","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-09-11 19:19:00.000000000","message":"Patch Set 3:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/803937/803937/3/check-not-vote/ext-glance-zuul/685467b : SUCCESS in 40m 02s","accounts_in_message":[],"_revision_number":3},{"id":"480ca82539c6f6ade515ea5c63b968ea249d6e3e","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2021-09-14 16:15:04.000000000","message":"Patch Set 3:\n\n(2 comments)\n\nMinor nitpicking, otherwise lgtm","accounts_in_message":[],"_revision_number":3},{"id":"90aa116f36568076b6728e7dcbd44fdb891ba46c","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-09-14 16:18:03.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"484db5d94d9ca73ea8bb35a1dd729fe7d7345b4f","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-09-14 16:24:25.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"09e4e5e66902cd5ac6fe2ac7d7930ca324280f91","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-09-14 16:27:35.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"859c4a010cc57762774fc5cb91098e40cfc35117","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2021-09-14 16:28:20.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nAfter further testing, it seems I was wrong, so this is good to go!","accounts_in_message":[],"_revision_number":3},{"id":"bd4ded33c7dcdffd4f61c32d28b072248224cddb","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-09-14 16:28:49.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"6b43ec0fcda25a5589b00450f5ad43fe5a359bd1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-14 16:29:18.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"b57f6259b1f5f5aea3794d9f6609a5f87c4a5305","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-14 18:39:28.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/bfaa448c3dc341c7957b8d889334d8c5 : SUCCESS in 1h 02m 04s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/88e852adb316432b8edebc882717707a : SUCCESS in 1h 06m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/04a4fc4e45084bf59cc330a23d5a6893 : SUCCESS in 6m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/634cd64e6a1248bea86902887bbda389 : SUCCESS in 7m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4ce405f310c84758a9047e218e4664fd : SUCCESS in 7m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f217e65980234e13820a9d8f03518c40 : SUCCESS in 8m 19s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/444d3e1d4dce43ac8b89069b37e88a3e : SUCCESS in 10m 36s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c601d78c231646e8b2ff90958456c92d : SUCCESS in 9m 56s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/3c21ee9bdfde4ea1ac342ed4a2061fdb : SUCCESS in 1h 05m 27s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/434220a4a3744f39b7e346563f7fa6ee : SUCCESS in 1h 15m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2898d0ac6a534764ac12349b6ce51a1d : SUCCESS in 50m 13s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/af52e5e458ad4e14adef0431cd58447a : SUCCESS in 1h 09m 20s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/31b9c9eae7394162b48678a2310b5551 : SUCCESS in 30m 25s","accounts_in_message":[],"_revision_number":3},{"id":"22b38cb6e336cd8c3d5c069ebc2406b82e4eb468","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-14 18:39:28.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"cb6ccde549d948643404ee2302ca92e631628a26","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-14 18:41:18.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/657b8e165f8d40879487e0c0665c6c06 : SUCCESS in 1m 16s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"9b385f7b671fd9994d416949ebd10c424d9fc072","revisions":{"8a0c1bea48448c814a392c5da3eef0483ced5e0f":{"kind":"REWORK","_number":1,"created":"2021-08-09 15:27:32.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/37/803937/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/37/803937/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/37/803937/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/37/803937/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/37/803937/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/37/803937/1"}}},"commit":{"parents":[{"commit":"284189222923f60bd883dd95ff0354b80850a4ce","subject":"Merge \"Clean out deprecations from paste.ini\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/284189222923f60bd883dd95ff0354b80850a4ce"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-08-09 07:00:54.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-08-09 15:23:20.000000000","tz":0},"subject":"[uwsgi] Add missing pefetch periodic job","message":"[uwsgi] Add missing pefetch periodic job\n\nGlance was not supporting uwsgi deployment when we added periodic job to\nprefetch images into cache. Later in ussuri uwsgi support was added but\nwe missed to implement periodic job to pre-cache the image.\n\nThis patch add this support for glance + uwsgi.\n\nCloses-Bug: #1939307\n\nChange-Id: I2abd1e60f414fbd68ce84e0b280f8b3e4e791a82\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8a0c1bea48448c814a392c5da3eef0483ced5e0f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8a0c1bea48448c814a392c5da3eef0483ced5e0f"}]},"branch":"refs/heads/master"},"6780848353a01e42e538aba184e003eedf355e68":{"kind":"REWORK","_number":2,"created":"2021-09-08 16:02:45.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/37/803937/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/37/803937/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/37/803937/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/37/803937/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/37/803937/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/37/803937/2"}}},"commit":{"parents":[{"commit":"929190a6afd0a1593e62abb8024c339713b52b98","subject":"Merge \"Add doc support for delete-from-store API\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/929190a6afd0a1593e62abb8024c339713b52b98"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-08-09 07:00:54.000000000","tz":0},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-09-08 17:05:47.000000000","tz":-420},"subject":"[uwsgi] Add missing pefetch periodic job","message":"[uwsgi] Add missing pefetch periodic job\n\nGlance was not supporting uwsgi deployment when we added periodic job\nto prefetch images into cache. Later in ussuri uwsgi support was added\nbut we missed to implement periodic job to pre-cache the image.\n\nThis patch add this support for glance + uwsgi. For WSGI, we run the\nprefetcher with an external lock, which makes sure that multiple API\nworkers will not attempt to cache an image at the same time. In this\ncase, if multiple workers attempt to run at the same time, only one\nwill grab the lock and do the work. When completed, the other worker\nwill grab the lock and either find all the work completed, or complete\nnew work that has been queued since the first one started.\n\nCloses-Bug: #1939307\nCo-Authored-By: Dan Smith \u003cdms@danplanet.com\u003e\nChange-Id: I2abd1e60f414fbd68ce84e0b280f8b3e4e791a82\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6780848353a01e42e538aba184e003eedf355e68"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6780848353a01e42e538aba184e003eedf355e68"}]},"branch":"refs/heads/master"},"9b385f7b671fd9994d416949ebd10c424d9fc072":{"kind":"REWORK","_number":3,"created":"2021-09-08 17:02:53.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/37/803937/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/37/803937/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/37/803937/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/37/803937/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/37/803937/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/37/803937/3"}}},"commit":{"parents":[{"commit":"929190a6afd0a1593e62abb8024c339713b52b98","subject":"Merge \"Add doc support for delete-from-store API\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/929190a6afd0a1593e62abb8024c339713b52b98"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-08-09 07:00:54.000000000","tz":0},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-09-08 18:05:57.000000000","tz":-420},"subject":"[uwsgi] Add missing pefetch periodic job","message":"[uwsgi] Add missing pefetch periodic job\n\nGlance was not supporting uwsgi deployment when we added periodic job\nto prefetch images into cache. Later in ussuri uwsgi support was added\nbut we missed to implement periodic job to pre-cache the image.\n\nThis patch add this support for glance + uwsgi. For WSGI, we run the\nprefetcher with an external lock, which makes sure that multiple API\nworkers will not attempt to cache an image at the same time. In this\ncase, if multiple workers attempt to run at the same time, only one\nwill grab the lock and do the work. When completed, the other worker\nwill grab the lock and either find all the work completed, or complete\nnew work that has been queued since the first one started.\n\nCloses-Bug: #1939307\nCo-Authored-By: Dan Smith \u003cdms@danplanet.com\u003e\nChange-Id: I2abd1e60f414fbd68ce84e0b280f8b3e4e791a82\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9b385f7b671fd9994d416949ebd10c424d9fc072"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9b385f7b671fd9994d416949ebd10c424d9fc072"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
