)]}'
{"id":"openstack%2Fglance~804090","triplet_id":"openstack%2Fglance~master~I129181b2338e09d094ee552a62500c6e460ee621","project":"openstack/glance","branch":"master","hashtags":[],"change_id":"I129181b2338e09d094ee552a62500c6e460ee621","subject":"Fix: glance cinder functional test","status":"MERGED","created":"2021-08-10 13:55:38.000000000","updated":"2021-08-10 15:33:46.000000000","submitted":"2021-08-10 15:31:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"804090","meta_rev_id":"dbb4c002b4f122ac55e209fa71868518ce32a10d","_number":804090,"virtual_id_number":804090,"owner":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-10 15:31:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2021-08-10 14:16:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2021-08-10 14:09:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2021-08-10 14:16:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-10 14:09:42.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2021-08-10 14:16:43.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2021-08-10 14:32:45.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-08-10 14:56:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3dbd4746a89c1f1ae7e78e704dad7220610f86b5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2021-08-10 13:55:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"26543ba4554a234fedc51dde6e4ac8385ec1fc79","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-08-10 14:09:42.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5a20121884554861d3be0426516386d7bece0c61","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2021-08-10 14:15:43.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9d1f4bd7c4dee724eccc8dec052c87a4bf1bbe63","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-10 14:16:43.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"934ed29a4369269dc5af21030b6276ae1642c1e2","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-08-10 14:32:45.000000000","message":"Patch Set 1:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/804090/804090/1/check-not-vote/ext-glance-zuul/e53fa30 : SUCCESS in 36m 40s","accounts_in_message":[],"_revision_number":1},{"id":"b18f1c1ec429d70615c0a152d2e3d7a6f78ad86f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-10 14:56:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c8992874de534833a73fe342950ccee7 : SUCCESS in 5m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cb01cecad2914de5a0ac7ff28aa41753 : SUCCESS in 6m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4e334e02bf9d4539baafe92f36148126 : SUCCESS in 7m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3025dd0e9b7a4af19c3a4b4109e839ce : SUCCESS in 7m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb0ecdf8cfa7497ab1975260f4963bd3 : SUCCESS in 8m 30s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0c4691ea534e456eb68c3df7de49c3b5 : SUCCESS in 9m 10s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/5247c979a42746098fb16a793799470d : SUCCESS in 10m 09s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/644c17cbc4164f83b23b751974638dc5 : SUCCESS in 9m 49s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/dfc27026f833437ea93a96c710b2b894 : SUCCESS in 5m 53s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/5a8a0b2e25ee4ddea6a9a310cfbd4073 : SUCCESS in 33m 27s","accounts_in_message":[],"_revision_number":1},{"id":"095f79bbcbcb8c44079a104d367307dc715d305a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-10 14:56:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3c281e2ce5c36e02aaf28cb1cbfc6b104c9c522b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-10 15:31:53.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/74157ce50d844c1ca59e07371979aaa8 : SUCCESS in 6m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8bb2f914deba44fc9c1bc02c89dfce57 : SUCCESS in 7m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/691756172c6a4079974bd305d6a9a587 : SUCCESS in 7m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc6f634dadfc48f5bbe3b0bb7ac650c7 : SUCCESS in 9m 06s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/418d65205d3c41fe856a1ac6a2f7208a : SUCCESS in 10m 13s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c35d16b206244dc999e433bd6952e4b3 : SUCCESS in 10m 16s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/d8f4b6561fd1453b95be8dd3639e151e : SUCCESS in 29m 54s","accounts_in_message":[],"_revision_number":1},{"id":"522e7f491d933670178e89d7544a80dfa1cae19e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-10 15:31:53.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"dbb4c002b4f122ac55e209fa71868518ce32a10d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-10 15:33:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/88a8ea44ced24be3a3929bd1f80f78a9 : SUCCESS in 1m 23s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"01b5bf5ece9a6904c1c5df4ad557967c927b8683","revisions":{"01b5bf5ece9a6904c1c5df4ad557967c927b8683":{"kind":"REWORK","_number":1,"created":"2021-08-10 13:55:38.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/90/804090/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/90/804090/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/90/804090/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/90/804090/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/90/804090/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/90/804090/1"}}},"commit":{"parents":[{"commit":"284189222923f60bd883dd95ff0354b80850a4ce","subject":"Merge \"Clean out deprecations from paste.ini\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/284189222923f60bd883dd95ff0354b80850a4ce"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2021-08-10 13:48:23.000000000","tz":-240},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2021-08-10 13:48:23.000000000","tz":-240},"subject":"Fix: glance cinder functional test","message":"Fix: glance cinder functional test\n\nWe are adding multiattach volume handling capability to glance cinder\nstore with patch[1] which includes checks for multiattach.\nThis patch adds that attribute to the fake volume object we use in\nthe functional tests.\n\n[1] https://review.opendev.org/c/openstack/glance_store/+/786410\n\nChange-Id: I129181b2338e09d094ee552a62500c6e460ee621\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/01b5bf5ece9a6904c1c5df4ad557967c927b8683"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/01b5bf5ece9a6904c1c5df4ad557967c927b8683"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
