)]}'
{"id":"openstack%2Fglance~804407","triplet_id":"openstack%2Fglance~master~I293919d6dc93bf221eae9109fd4a658d332bdca8","project":"openstack/glance","branch":"master","topic":"sqlalchemy-20","hashtags":[],"change_id":"I293919d6dc93bf221eae9109fd4a658d332bdca8","subject":"setup.cfg: Fix errant key","status":"MERGED","created":"2021-08-12 15:06:10.000000000","updated":"2021-08-12 22:40:01.000000000","submitted":"2021-08-12 22:38:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"804407-sqlalchemy-20","meta_rev_id":"ade151e795c573ed577e7574f4738e6316a092b5","_number":804407,"virtual_id_number":804407,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-12 22:38:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2021-08-12 21:22:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2021-08-12 21:22:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-12 16:57:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-12 21:22:42.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"}],"messages":[{"id":"3f472f4d3e7149cc6172568c3fb9b19c7c4be0bd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-08-12 15:06:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c0972dd6cb0f3f01e26a0d0ce6450106823adcae","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-12 15:40:02.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"517c1b0e9b469408aeff2deb85184524b8999833","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-12 15:40:14.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"3747244b9c6dda08ad3be73417d92606a90b6e11","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-08-12 16:11:24.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"d694b28f31dc8fde29d5f6588246538b684a6d98","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-12 16:57:36.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/7e387254d8b740fd9cdb1e2f0fe66eff : SUCCESS in 3m 19s\n- grenade https://zuul.opendev.org/t/openstack/build/295e4468de3b40ff81513e3dae40ebe9 : SUCCESS in 55m 23s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/38c560a87d334cb9a8a48d899bfe489b : SUCCESS in 1h 05m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/09129fd162174b958a99fff0eb90e4da : SUCCESS in 5m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/db2a544197f84d8289e11495665b6e07 : SUCCESS in 5m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9589a554d5344b27b8b1bbcf8dee8aeb : SUCCESS in 6m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ed8cff46702a4995b0d704572a1320b1 : SUCCESS in 6m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a3691ea5d5245c3aa4826661071f937 : SUCCESS in 7m 06s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/22b28084b0f14436aaaefb74206f8299 : SUCCESS in 9m 43s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/6ec51bd05e2642da8164317d3fdd6b0f : SUCCESS in 8m 32s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/bd431b69f58d4ad69aa2407b81e1aa91 : SUCCESS in 8m 25s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/27a3869725ed4d63bf17b21f49057a7b : SUCCESS in 1h 02m 18s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/a341bbf79ca54c348a70f20bdfded371 : SUCCESS in 1h 11m 27s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/05cc737c782c4d1cb745476575908d09 : SUCCESS in 1h 07m 37s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/7d4126eea0f0417e91045525a22bcb97 : SUCCESS in 58m 50s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/62e168b4e78b422389b9b2cda56b6cfc : SUCCESS in 47m 08s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/3ce72be0500d465d9a5727b94cd87979 : SUCCESS in 1h 15m 23s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/3d355ce3f2b04e25a42beaae924b3649 : SUCCESS in 32m 19s","accounts_in_message":[],"_revision_number":3},{"id":"ae6603213e04d9d5e2d8586cc95f09d112acdba1","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-12 21:22:42.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nThank you, ninja approving as a straight forward fix.","accounts_in_message":[],"_revision_number":3},{"id":"e9debdcbf066383cd87177aa06f26e6dbf804637","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-12 21:25:16.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"70243a1096f8c312d99fe89d1e23cc2048a7a479","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-12 22:38:38.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/6dcfe5134e5a420e9bd8888569392043 : SUCCESS in 2m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/1cd7657838e14fb8b3ab0dddf8a78722 : SUCCESS in 1h 10m 22s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6d70c44c7cd545bab50dc18219c46291 : SUCCESS in 1h 04m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0658adff4d704b2ea43e1603e0563175 : SUCCESS in 5m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c2afa414ae1e4eedbd05e3d8cfc8ef9d : SUCCESS in 5m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cff0aa6c8bcf48208216a1cd72d867d1 : SUCCESS in 5m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/620301b3ae7d4cf79a82f2150325ef31 : SUCCESS in 6m 44s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e1d2d4a00dc742678e6228a5f32062b3 : SUCCESS in 8m 27s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/334edf646b5b4220befa6273a9fb4b93 : SUCCESS in 8m 52s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/f4389db3a2c647d4b10c138244930a56 : SUCCESS in 1h 09m 42s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/056682a977724af4923c82ec98329714 : SUCCESS in 1h 07m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b6bdf19ceee442c3b4e9faebbb12296d : SUCCESS in 47m 47s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/687548601f574965a2d7065af98514c5 : SUCCESS in 59m 25s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/363c5db06c1542458f91c257a93bdb54 : SUCCESS in 32m 46s","accounts_in_message":[],"_revision_number":3},{"id":"ec2661378e5bdb47c1f48240c09ca396f109c93a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-12 22:38:39.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"ade151e795c573ed577e7574f4738e6316a092b5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-12 22:40:01.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2673e36f7da4862ba93ef1e9d82020b : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"870f13098948783341d51e097882759d788d20a3","revisions":{"0001f5466749393796ec18e83ea4d9e60171a228":{"kind":"REWORK","_number":1,"created":"2021-08-12 15:06:10.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/07/804407/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/07/804407/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/07/804407/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/07/804407/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/07/804407/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/07/804407/1"}}},"commit":{"parents":[{"commit":"56813e09da27edb4e2a3815c472a546b424295dc","subject":"Resolve compatibility with oslo.db future","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/56813e09da27edb4e2a3815c472a546b424295dc"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2021-08-12 15:04:02.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2021-08-12 15:05:32.000000000","tz":60},"subject":"setup.cfg: Fix errant key","message":"setup.cfg: Fix errant key\n\nThe author of change I08c014e10c3fc54046930ea8be2b13775428e51b fat\nfingered the replacement of \u0027python-requires\u0027 with \u0027python_requires\u0027\nresulting in the following warning:\n\n  UserWarning: Usage of dash-separated \u0027python-r_quires\u0027 will not be\n  supported in future versions. Please use the underscore name\n  \u0027python_r_quires\u0027 instead\n\nNot sure how setuptools was still identifying but resolve it\nnonetheless.\n\nChange-Id: I293919d6dc93bf221eae9109fd4a658d332bdca8\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0001f5466749393796ec18e83ea4d9e60171a228"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0001f5466749393796ec18e83ea4d9e60171a228"}]},"branch":"refs/heads/master"},"7cff4ebe3cc8c5aa2008b6630f8f237a0eb4a116":{"kind":"NO_CHANGE","_number":2,"created":"2021-08-12 15:40:02.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/07/804407/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/07/804407/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/07/804407/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/07/804407/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/07/804407/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/07/804407/2"}}},"commit":{"parents":[{"commit":"f374022359793dbcd7346743e4345ab4ab9fcb79","subject":"Resolve compatibility with oslo.db future","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f374022359793dbcd7346743e4345ab4ab9fcb79"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2021-08-12 15:04:02.000000000","tz":60},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-08-12 15:40:02.000000000","tz":0},"subject":"setup.cfg: Fix errant key","message":"setup.cfg: Fix errant key\n\nThe author of change I08c014e10c3fc54046930ea8be2b13775428e51b fat\nfingered the replacement of \u0027python-requires\u0027 with \u0027python_requires\u0027\nresulting in the following warning:\n\n  UserWarning: Usage of dash-separated \u0027python-r_quires\u0027 will not be\n  supported in future versions. Please use the underscore name\n  \u0027python_r_quires\u0027 instead\n\nNot sure how setuptools was still identifying but resolve it\nnonetheless.\n\nChange-Id: I293919d6dc93bf221eae9109fd4a658d332bdca8\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7cff4ebe3cc8c5aa2008b6630f8f237a0eb4a116"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7cff4ebe3cc8c5aa2008b6630f8f237a0eb4a116"}]},"branch":"refs/heads/master","description":"Rebase"},"870f13098948783341d51e097882759d788d20a3":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2021-08-12 15:40:14.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/07/804407/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/07/804407/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/07/804407/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/07/804407/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/07/804407/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/07/804407/3"}}},"commit":{"parents":[{"commit":"f374022359793dbcd7346743e4345ab4ab9fcb79","subject":"Resolve compatibility with oslo.db future","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f374022359793dbcd7346743e4345ab4ab9fcb79"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2021-08-12 15:04:02.000000000","tz":60},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-08-12 15:40:14.000000000","tz":0},"subject":"setup.cfg: Fix errant key","message":"setup.cfg: Fix errant key\n\nThe author of change I08c014e10c3fc54046930ea8be2b13775428e51b fat\nfingered the replacement of \u0027python-requires\u0027 with \u0027python_requires\u0027\nresulting in the following warning:\n\n  UserWarning: Usage of dash-separated \u0027python-r_quires\u0027 will not be\n  supported in future versions. Please use the underscore name\n  \u0027python_r_quires\u0027 instead\n\nNot sure how setuptools was still identifying but resolve it\nnonetheless.\n\nRelated-Bug: #1939716\n\nChange-Id: I293919d6dc93bf221eae9109fd4a658d332bdca8\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/870f13098948783341d51e097882759d788d20a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/870f13098948783341d51e097882759d788d20a3"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
