)]}'
{"id":"openstack%2Fglance~804558","triplet_id":"openstack%2Fglance~master~Iba990e59c454a2e9f37a0c659d8634583d3a8bf7","project":"openstack/glance","branch":"master","topic":"policy-refactor","hashtags":[],"change_id":"Iba990e59c454a2e9f37a0c659d8634583d3a8bf7","subject":"Check policies for staging operation in API","status":"MERGED","created":"2021-08-14 10:40:55.000000000","updated":"2021-08-21 05:45:20.000000000","submitted":"2021-08-21 00:31:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":20,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"804558-policy-refactor","meta_rev_id":"32b8408ea22071633465cf63ea6417339c424ebf","_number":804558,"virtual_id_number":804558,"owner":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-21 00:31:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"recommended":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"all":[{"value":1,"date":"2021-08-19 16:34:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":2,"date":"2021-08-20 16:33:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":2,"date":"2021-08-19 16:39:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"all":[{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":1,"date":"2021-08-20 16:33:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-14 11:59:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-16 17:55:15.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"CC"},{"updated":"2021-08-16 21:16:03.000000000","updated_by":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"reviewer":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"state":"CC"},{"updated":"2021-08-17 04:58:02.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-08-17 16:26:56.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"CC"},{"updated":"2021-08-17 21:37:34.000000000","updated_by":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"reviewer":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"state":"REVIEWER"},{"updated":"2021-08-19 16:39:02.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2021-08-20 16:33:14.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"}],"messages":[{"id":"9d2312dde5cb45b50199a8bbea2470eec549ab3b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-14 10:40:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b763665687cffd28c9d7bb3cc640618a49642dab","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-14 10:42:46.000000000","message":"Topic policy-refactoring removed","accounts_in_message":[],"_revision_number":1},{"id":"f09469180fe26d22c4df7a46d0dc0f8f9f0e9fc8","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-14 10:42:55.000000000","message":"Topic set to policy-refactor","accounts_in_message":[],"_revision_number":1},{"id":"0067001f470a86a1f2a346ada7c7a365b4120ca0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-14 11:59:16.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/145999c7343d4187ae9d09fa28fa1149 : SUCCESS in 1h 05m 42s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/326f57b842cc404d91a50bb89448a442 : SUCCESS in 58m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1309cad4ce3f45b089e6c307e75400a6 : FAILURE in 5m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b0c32e31264d4a26b7034a519ec0de53 : SUCCESS in 5m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/11145d5523244b749ebd7e418e33740b : SUCCESS in 5m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7732948689c6427694f2f22938a5cfdd : SUCCESS in 6m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5b540c4c68d43f3bab37bef7ff219b9 : SUCCESS in 6m 43s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/33603034d7b34a83a8d751ec97418a48 : SUCCESS in 8m 49s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/5be16f118ce9490da16aa2b610a605e2 : SUCCESS in 8m 12s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/43310dda3f7749aab2d2759fb0ef2c8f : SUCCESS in 8m 34s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/14714052dbf3466a9ced5cb9546aee28 : SUCCESS in 4m 55s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/b68bb34b5d714b63b2ffd4e616117b2b : SUCCESS in 1h 16m 18s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/f919a7ee6c3f4119aa62e94c02c83622 : SUCCESS in 1h 04m 22s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/2a03905bd7d34c9fb910f5d86b2d4ec4 : SUCCESS in 1h 04m 43s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/caf22f36b06d44338f952ef4eb22043a : SUCCESS in 1h 09m 18s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6430d358a6a84d118f482a97d572bd99 : SUCCESS in 41m 45s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/ec33baf742c44ffa98c320e9076fc9aa : SUCCESS in 1h 02m 12s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/0e5c85197a0f489f8b184887c6999390 : SUCCESS in 27m 08s","accounts_in_message":[],"_revision_number":1},{"id":"6745a661ecabf2330a183168663698a846775386","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-14 20:45:06.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c31b2dea3fe3baaafe0ec56e4cc450af78522ebe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-14 22:14:41.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/1a13505730fd4d0dbb3a51c20436e9af : SUCCESS in 50m 33s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6aa0d171192349bbbe19146cabbd4078 : SUCCESS in 1h 08m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d753ca03677c4e4f98ae1a4c22da456e : SUCCESS in 7m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/008da21d62814ad3851c01cca878ccfe : SUCCESS in 8m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5fe6ac14431e46649760a706d3b3a458 : SUCCESS in 7m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b7926c3ccb454079b900f8e69c4699fe : SUCCESS in 8m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5909eee8c334385bf890e8a108ba70e : SUCCESS in 6m 01s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0143b7f17dcf469ea85a92391e55fe42 : TIMED_OUT in 31m 20s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/56f04f04c7084af38319dced4115c227 : SUCCESS in 8m 13s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/c3721f06d0c84c60bcc3dbd9fb48243d : SUCCESS in 10m 58s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/2248aa10116c4eaaa251190d70fbeab0 : SUCCESS in 5m 11s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/1fbd447a1d804bffa8219384f777b3b6 : SUCCESS in 1h 07m 55s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/bdb840824cf9414a8e6efe5d0124bc27 : SUCCESS in 1h 03m 25s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/a25bcaee545c4b6e9766f0a2c43b192b : SUCCESS in 55m 54s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/adb5f9818470448fbbaa001c5ffe492a : SUCCESS in 1h 27m 19s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e7b9cfa681984ce2811f6824716c570c : SUCCESS in 41m 39s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/8601605d9dae4623ae0175572d5afe69 : SUCCESS in 1h 13m 59s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/b647a310c6a442e2a62b772045a9cf63 : SUCCESS in 30m 42s","accounts_in_message":[],"_revision_number":2},{"id":"87e0daad1861dc5a6448414df017f1365a8711a0","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-15 04:45:04.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"a2cc5b542d65a3e0b2316b24ea99ac1f37d74866","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-15 06:10:14.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/433d2a5e8f1d4bb1985a6a7289a97d6a : SUCCESS in 1h 04m 56s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8318d740a6f34a749c8c6321feb03349 : SUCCESS in 1h 18m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db6ce3e774ae4517ad6391bb5a32d247 : SUCCESS in 9m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ed1b25c088fc4fa4a257264c8fe0ad29 : SUCCESS in 6m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4c23845637c84132adc292807bec8772 : SUCCESS in 6m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4656e833c0464a2faaeab12284f52fcb : SUCCESS in 6m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7e74819c7c24b3c949dca14f6b87d58 : SUCCESS in 6m 32s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/275ae71a2e624a5eb6bf83df31dea374 : SUCCESS in 8m 36s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/0aa4d875598448a88b6d35e72d1ae3fb : SUCCESS in 8m 54s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/ab09cf9dc677451b9405cecd0cbbbd5b : SUCCESS in 8m 25s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/0f517cf3fa7041a5ad1bdf8266fb7fa6 : SUCCESS in 5m 07s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/3d027eb6178843ba97fc14d5cfc86a96 : SUCCESS in 1h 15m 53s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/c76fba027ca548f297a23862c2f88dd9 : SUCCESS in 1h 07m 51s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/fe2ebb26d2cc4559b789cbba3acd0a02 : SUCCESS in 1h 12m 51s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/a0e3803f33434cc1a514208c6122afc9 : SUCCESS in 1h 18m 38s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/270e7210399243f88c089e4ed4bd8d37 : SUCCESS in 48m 45s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/0e683a354e464d869b49c68192f7968c : SUCCESS in 1h 13m 04s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/eee0e04688474d7285462e7af9ba96b7 : SUCCESS in 32m 11s","accounts_in_message":[],"_revision_number":2},{"id":"3b54e08701d15f7ae4cfad7109f60ef2bee83cb2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-15 09:09:30.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"ad287abc7b065e0eef196cf3042372141358ce61","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-15 10:18:03.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/c646840141f947cb82e2b381dbda5967 : SUCCESS in 1h 02m 52s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3867727c01eb4455a59815fead1151c0 : SUCCESS in 1h 03m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a166feb06c7442a80142c6bc4c8fdde : SUCCESS in 5m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7b857df9f3ec4ec9a73a238a2200927c : SUCCESS in 6m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a259e2f728574311a3e0bf65f2706e74 : SUCCESS in 6m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9adc9129d442431ca553a4df72e3feca : SUCCESS in 6m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79a7ec2a9a1c4dde81aaea5e89e0a23c : SUCCESS in 6m 50s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e910b6fd42004c5eae102231c648ba7b : SUCCESS in 9m 17s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/5e5f3df32301437c9ae2df8bda240774 : SUCCESS in 8m 31s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/4df1a2a6d02d4ab7a98f618b7ef54072 : SUCCESS in 8m 56s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/ae7bdbcbe939400aa23207b8d8f875eb : SUCCESS in 5m 33s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/c838da4739a244b693c2e7202afe555a : SUCCESS in 1h 04m 40s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/fd9bcbe549c84aa8b4a3bc3b8ac81d19 : SUCCESS in 1h 03m 44s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/e148c007ab6b4b3496967c9e191616be : SUCCESS in 1h 04m 56s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/f524896045f84abbada1dd4e1c258174 : SUCCESS in 1h 02m 46s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d74e21c48cec41088ce9e94178b94357 : SUCCESS in 41m 05s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/787a47a5ecb949c7ba9902d2b43a1b37 : SUCCESS in 1h 07m 02s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/32f7511153b14bfc9152a790c337553c : SUCCESS in 29m 29s","accounts_in_message":[],"_revision_number":3},{"id":"a812b2f219997a0d6efc94aaff0021bff33998e5","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-08-16 17:55:15.000000000","message":"Patch Set 3:\n\n(4 comments)","accounts_in_message":[],"_revision_number":3},{"id":"08898c5c5bb85031fcdd7990cc64ced9a0ad67a3","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-16 18:06:14.000000000","message":"Patch Set 3:\n\n(4 comments)","accounts_in_message":[],"_revision_number":3},{"id":"6ca81b391059c86293d39966a098921161533c92","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-08-16 18:20:05.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"b406439c607ec423094f775d717b6601265f38e2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-16 19:54:36.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"ca996a438fe7adc0da11e4a5fe69be5bc1c90877","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-16 21:07:54.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/08235d2de2344b1db5ec1f7a029d594c : SUCCESS in 56m 38s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d5168dee5ac3473ba70f115768f342e5 : SUCCESS in 1h 02m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/91fbf5dfa3ce4b5dbc8462dd94dafe7d : SUCCESS in 5m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0b485b54a613414c84404cf553500e3c : SUCCESS in 6m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eab0c6ba781d468f918a9206f46dc3da : SUCCESS in 6m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a8e3e0b66637412986fc3c648b71b577 : SUCCESS in 6m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bfa3a4e95ef64986945197d7e91ef493 : SUCCESS in 7m 03s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/bc2e5278cb5d4e5c80a99dc323528c45 : TIMED_OUT in 31m 23s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/480a3fb1cc304922bceb98df852cea3c : SUCCESS in 9m 07s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/698273bc290e44b7a8e9c620cfc430b5 : SUCCESS in 10m 49s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/5359673173894e5997572640bbb0d266 : SUCCESS in 5m 40s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/ad1306338a864e6fbc7466751f6433fc : SUCCESS in 1h 11m 15s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/4c51da0a6b9c4df9ad388098dc67fb8d : SUCCESS in 1h 05m 47s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/c606300811c14b49b320e36e9595c978 : SUCCESS in 1h 06m 51s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/31acc6738f784c348380634167d09bdd : SUCCESS in 1h 11m 34s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/af59619c82e54b67a1375e382f792766 : SUCCESS in 41m 41s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/6e08a1f9010644c5830c05f5d583bdf4 : SUCCESS in 1h 06m 00s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/721d2038da694645b430e1357ef4dc5e : SUCCESS in 31m 48s","accounts_in_message":[],"_revision_number":4},{"id":"3795fbfccd9c7470a7d13699c1632bbb3b02c439","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-08-16 21:16:03.000000000","message":"Patch Set 4:\n\nrecheck timeout","accounts_in_message":[],"_revision_number":4},{"id":"d722913ee227aa2f2304bfa6ab1e22d0e964359e","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-08-16 21:23:58.000000000","message":"Patch Set 4:\n\n(3 comments)\n\nOnly minor comments and questions inline.","accounts_in_message":[],"_revision_number":4},{"id":"c6accc729105d8b6d1f9afb18ecc8567f515cb66","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-16 22:17:30.000000000","message":"Patch Set 4:\n\n(4 comments)","accounts_in_message":[],"_revision_number":4},{"id":"701c8d7a20ce2cf568304755a2ee06519e401eb0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-16 22:43:47.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/bcd5f03851e542d6b4a6d2fb27d9cfcd : SUCCESS in 43m 54s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/285f70549be74c268602b0080b50250c : SUCCESS in 1h 14m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa22773cb90c4e08bd3abe9e89d70152 : SUCCESS in 4m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5dfd5d390c9a41168deab80078ad1d96 : SUCCESS in 5m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2dd4263e30d8484c8d59f08faf41c967 : SUCCESS in 6m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/04d4a49aa77a4e62a4f112a6640a86f2 : SUCCESS in 6m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec56285795194657909f1e7848bc916a : SUCCESS in 6m 37s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9869ddefbbfe42e8944b3c2e38530e49 : SUCCESS in 9m 31s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/bbad62710a0b4f09a54c2ce8052c9d35 : SUCCESS in 9m 16s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/9a4a3052141e45c8a2cd0a1ea45f9c18 : SUCCESS in 9m 05s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/64447306e50f43479af321e1d9b30bc1 : SUCCESS in 5m 34s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/54b52f5dc9424df3afc7f52416b18646 : SUCCESS in 1h 10m 13s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/d9fe625b4a2e4a588b8bbb06587207d0 : SUCCESS in 59m 04s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/ccb564b09afb4fb3a80b7977cf94cd06 : SUCCESS in 1h 00m 24s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/3324b585b6a0480d9a725f0e108d0cdd : SUCCESS in 1h 07m 09s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/716765f8e6c3484d88abf282aa920cd4 : SUCCESS in 39m 02s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/4e98bbce6a614b8bae3612c91e40c154 : SUCCESS in 1h 00m 06s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/f529c93ee7d14e319f4556b4bf71de28 : SUCCESS in 32m 14s","accounts_in_message":[],"_revision_number":4},{"id":"70fc151bf179b5fc4602fe58386d86776c1d783d","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-08-17 04:58:02.000000000","message":"Patch Set 4:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/804558/804558/4/check-not-vote/ext-glance-zuul/3bdcc27 : SUCCESS in 39m 13s","accounts_in_message":[],"_revision_number":4},{"id":"1c94bf0a08ab20cde05a7d9586c621171deff20f","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-17 15:03:50.000000000","message":"Topic stage-policy-refactor removed","accounts_in_message":[],"_revision_number":4},{"id":"b19c190f57b6c8a341ef41801fd7015a813c39d6","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-17 15:03:57.000000000","message":"Topic set to policy-refactor","accounts_in_message":[],"_revision_number":4},{"id":"d7f9fe579a2c1e3a966b8ede2a9a4ee0636c18c7","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2021-08-17 16:26:56.000000000","message":"Patch Set 4:\n\nI second Lance\u0027s small requests, maybe we can get another patchset while the parent patches get through the gates?","accounts_in_message":[],"_revision_number":4},{"id":"4ca179f1bc4dbd8a379bdb4d32f74a3ac53d6065","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-17 17:38:20.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"b19fbea4c08545f547a8d3b601f501ed3c1829ca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-17 19:09:22.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/8e28029d13f84f169cfe2a912c65e098 : SUCCESS in 53m 50s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/45c3e493a71b46f0a76f0beef0989c2a : SUCCESS in 1h 23m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a261f5528a1b4f06b0c5f84c1cff67a9 : SUCCESS in 7m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2ebf1186d06943f0824fab6221241e3e : SUCCESS in 7m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/183acdea65614b07a7abcff3f9d84ad9 : SUCCESS in 6m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ae01a76de04c47b483413cdc615f88ad : SUCCESS in 5m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d7fe44e8efd748cf9d95bc811aa20ae0 : SUCCESS in 6m 41s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ad373ed9199141f4a6d11952bfb4610e : TIMED_OUT in 31m 33s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/38ca1488094e485f9dd107c43badbfeb : SUCCESS in 10m 26s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/2b84322fda3e405fa2af1f3864a6bb8f : SUCCESS in 8m 34s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/a8cf97db82aa49238f68482fa1a49992 : SUCCESS in 5m 19s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/418e98cc35bf4a47a2315ad1a6df04b3 : SUCCESS in 1h 14m 49s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/b6b5fd94a4c3415297333d15cbadd191 : SUCCESS in 1h 04m 21s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/5bb8c6a04a7c443ebe9436d776e32dec : SUCCESS in 1h 02m 23s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/17c817c0314e4090a0899c4b2b00451f : SUCCESS in 1h 23m 20s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/174375ff30ee474c842ac99c9ff7159f : SUCCESS in 46m 16s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/6a109aaf62cf4f7ab5174a30a071dba5 : SUCCESS in 1h 22m 04s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/cc2beb88971c4975becabdeba390f99b : SUCCESS in 30m 22s","accounts_in_message":[],"_revision_number":5},{"id":"d22e1dad05dd42c6154f6f73481f58fb1af5f53f","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-17 20:47:29.000000000","message":"Patch Set 5:\n\nrecheck attempt 2","accounts_in_message":[],"_revision_number":5},{"id":"9a51e818f8c2e238417ddbc42acc65bc0689ea75","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-08-17 21:34:57.000000000","message":"Patch Set 5:\n\n(3 comments)","accounts_in_message":[],"_revision_number":5},{"id":"576579a61048768f30658e4813624a0e83a12cf4","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-08-17 21:37:34.000000000","message":"Patch Set 5: Code-Review+1\n\nI wonder if there are some issues with opendev?\n\nRUN END RESULT_TIMED_OUT: [untrusted : opendev.org/zuul/zuul-jobs/playbooks/tox/run.yaml@master]","accounts_in_message":[],"_revision_number":5},{"id":"c7c2bbe903d7b2304ce4759acfc2492cf326b40e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-18 00:22:32.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/6a03ebe9f2a74faba54125a768a00f90 : SUCCESS in 1h 00m 40s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c89c6d8167c74951b8df04fc8f0ee5c2 : SUCCESS in 1h 08m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/08d201858a954a0595b1f355c355a8d8 : SUCCESS in 7m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/07ce0c394fb24c308b60b4404913c23e : SUCCESS in 5m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/02663e1baeef4a4dbe9b0a00ff99db38 : SUCCESS in 5m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ef63b0fe9aba44a7acf669aebc9b81a2 : SUCCESS in 7m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eac8305d295140278816c4cf19309e74 : SUCCESS in 10m 02s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/646a90b434264145ba4529b99ca52b7f : TIMED_OUT in 31m 02s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/d47cce8df4af4a7bb262694074423882 : SUCCESS in 8m 26s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/a0d92a74eb4346b294b146b321c78436 : SUCCESS in 10m 12s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/4965532c06b943f88a5ce5f410a53756 : SUCCESS in 11m 35s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/4730382dc7db482f97f8cf300beb25f5 : SUCCESS in 1h 21m 08s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/e41092d753db493bba60f4046fe3e548 : SUCCESS in 1h 14m 04s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/8ec08dbeebdf4d83bb62fc1546a58549 : SUCCESS in 1h 06m 50s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/a98b6342219e433da30f33b532971490 : FAILURE in 2h 05m 12s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9722a046d0ca44449826a0310e8a38d6 : SUCCESS in 46m 10s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/93e5ea194ebe45babcca9daa9385e9cb : SUCCESS in 1h 03m 57s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/667f271152ac4847a7bf23415b7a3d17 : SUCCESS in 50m 34s","accounts_in_message":[],"_revision_number":5},{"id":"f3f2fda38d7457750d0642464fd69a7f0ec09c8f","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-18 06:10:18.000000000","message":"Patch Set 5:\n\nrecheck timeout","accounts_in_message":[],"_revision_number":5},{"id":"b7550bc3205776f54a25b177786a74e87cbf4af6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-18 08:19:39.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/c27914a33f0b4f20bb2d54176a203238 : SUCCESS in 1h 02m 35s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9a9cf164919f4dd6888e8385bd756028 : SUCCESS in 1h 22m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f9911421bd34100a1d322750b8dd705 : SUCCESS in 5m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fea3e63adcad45cb9237eddf093f7862 : SUCCESS in 6m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5b8475edffb4460d9eb2b877c691c4af : SUCCESS in 6m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/92521c2c63464623b0db10450f5c7eb1 : SUCCESS in 6m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f1b07cbbc744d2c8fc6db6635d5395f : SUCCESS in 7m 23s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d938c086c7f6497c90a377a50e770159 : SUCCESS in 8m 08s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/f3cd2c27b4a4496396f76181911809ac : SUCCESS in 8m 10s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/4bc40758b1d1465d8517f12eef2cfbcc : SUCCESS in 8m 24s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/111c5f6e4fd5449885f5fc0bf4c03dbe : SUCCESS in 6m 20s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/e87c7b1cd24b40a6a87078150ac8ef15 : SUCCESS in 1h 12m 19s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/7b68cd108a5645408bf3f83837c0099e : SUCCESS in 1h 03m 44s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/9e92a0b8fd724396b02e098caeee12e8 : SUCCESS in 1h 02m 25s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/8051937abd4b46609a8a2dd11c7dcd4a : SUCCESS in 1h 11m 14s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/48b970b2e8564fb092d43121457dfd7e : SUCCESS in 43m 37s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/d92d7e3f659947058621519993184f46 : SUCCESS in 1h 11m 50s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/08d58b64f40e49b2a705fb72b43c1db5 : SUCCESS in 42m 36s","accounts_in_message":[],"_revision_number":5},{"id":"292598aa9523eda1b835a5eb7d84d0caf739848d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-18 09:27:56.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"e0e17e544b2db458df97917c1b232ecd10d2eb77","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-18 11:01:56.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/da1f2f8da073419fb66758244afb27b5 : SUCCESS in 1h 08m 01s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/67a5f12ddc5941b288baf7e64cd3152b : SUCCESS in 1h 13m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2df50fe91b014aaf8d7c656af25cca40 : SUCCESS in 5m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/43d4b83c2192456ea7aab5cec1b59ee3 : SUCCESS in 6m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bbf3e5b789104110a6a647f70f97ba62 : SUCCESS in 5m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3044a89e3ef945ed96eb5caf1aaa7218 : SUCCESS in 6m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8dd1c48240b41799d313c39a6a3a17e : SUCCESS in 6m 37s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0797d9ecb86246babc02bcdac87a955d : TIMED_OUT in 31m 07s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/0b460665c30d42ef951067f569daef91 : SUCCESS in 8m 34s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/1f55096007f6416c9ec4cd71ed521dab : SUCCESS in 8m 44s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/d3a7009dce2d42a69e1c603b7b1d58ab : SUCCESS in 5m 45s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/910a6f105c904e6994a279437caa7d12 : SUCCESS in 1h 15m 20s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/7ba24ae942e94fd28a6061d0169227c2 : SUCCESS in 1h 25m 14s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/626c2485e22c4796a2812d8fd3198b62 : SUCCESS in 1h 19m 32s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/9f28dd7423744302b2cfe96d8ef2ed66 : SUCCESS in 1h 16m 01s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fa3f9995e9444cc587b9ab0007345cf6 : SUCCESS in 43m 18s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/34e0106cebac45be84d79d310399e8b7 : SUCCESS in 1h 24m 48s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/4cb45e92baa5493493964a58d99e816c : SUCCESS in 31m 44s","accounts_in_message":[],"_revision_number":6},{"id":"0716c6aefdd69fc7bf71061a5fb74112de53efdb","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-18 11:04:02.000000000","message":"Patch Set 6:\n\nrecheck timeout","accounts_in_message":[],"_revision_number":6},{"id":"2fde3882a9288d2bc20e8ff5957e9cdc74db56b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-18 12:27:13.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/7743b771ec5f4cad9855705de754eaa4 : SUCCESS in 1h 03m 01s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9bd30e578be54870bd9ea536c85783e3 : SUCCESS in 1h 14m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/756ff30e2c454b628f38193da0b2e171 : SUCCESS in 5m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ef08b3485a104f68b1d07a0d755e410b : SUCCESS in 5m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dbf9fd3c81b7497fab4d617f10f8d5c3 : SUCCESS in 6m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/81fe4b37685c433d9f65cd826ce52035 : SUCCESS in 7m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2923eacd94d74457bbace39504a7b638 : SUCCESS in 6m 52s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3a2b7e32d16f47cd9b042f2f62bf782b : SUCCESS in 9m 08s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/18b47f9743164dc18db8cd6a592a2ced : SUCCESS in 11m 06s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/9d0ba21591314d49bf9c70b78f0306c1 : SUCCESS in 10m 45s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/8ec5fcb6304346c3aa1555b2d4a104d4 : SUCCESS in 5m 52s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/2654482a1276403896141fd399af80c9 : SUCCESS in 1h 05m 19s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/543a4dad32b5473096cee2d5fc961675 : SUCCESS in 1h 12m 51s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/f0826b185bcc4af8b39e030af900be75 : SUCCESS in 1h 03m 49s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/9effd0146fcd465cbc49e7a556de7153 : SUCCESS in 1h 18m 50s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4c005e38a7974228a129bd8fb2cb951f : SUCCESS in 40m 08s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/a95db11698004550ae35ffb48a262745 : SUCCESS in 1h 10m 21s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/8c37b1a738c041138e1c6c44dcb00dc2 : SUCCESS in 35m 04s","accounts_in_message":[],"_revision_number":6},{"id":"c3f0132dc980c1718f440a97a5679693646e9a21","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-08-18 18:29:41.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"dd2788c97e8ecf4876f9ef6c7ab8f2f8c679c6cf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-19 15:59:54.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"ffc0b17405d98fdcc78d06c7bd940e2bef2a1523","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-08-19 16:34:07.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"a4ad1bed6d6d14fda450416452c322e5c44dd661","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2021-08-19 16:39:02.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"2a1e41daa855b2f9cf35a43b00c134dd6058c922","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-19 18:07:05.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/9437658f67cd4baf8e8b907ae3e8b9b4 : SUCCESS in 1h 12m 18s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/af08f2ab24274c91b253b7ae29874bbc : SUCCESS in 1h 15m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/815979f3c1d449eda0ffeb5b2bf05353 : SUCCESS in 6m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eb06e00b83b949a38e9013e46421545d : SUCCESS in 5m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8aa78abdf0de47f6a3e13259803b4509 : SUCCESS in 5m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8e2aefa71211400ba33be6f6a75e02fd : SUCCESS in 6m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36d107c3f1a44640bc4b010880ccff62 : SUCCESS in 6m 49s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3d946c38290b4c2fb527ff6588a18c13 : TIMED_OUT in 31m 15s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/89be1c9566154a6cbc97a690c54b7369 : SUCCESS in 8m 45s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/a952e3ea40e94306a2bf6ef4003fd6f9 : SUCCESS in 8m 20s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/0fbe53a4c3a949dfb17779bc2696b452 : SUCCESS in 5m 25s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/7fb7063c11584e6ab319b7e9c2fd0272 : TIMED_OUT in 2h 04m 32s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/4254cc1df1514554abcd7c195316c912 : SUCCESS in 1h 14m 05s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/4e85ba1fbac248f8a5f51fc66e76b4c6 : SUCCESS in 1h 07m 44s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/3a0648d44b06488e8d88134bceabf208 : SUCCESS in 1h 18m 30s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0450b6d2ba2e459e881810432c6fa1bc : SUCCESS in 45m 18s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/ce5b389045df4de1b4da3db80115f484 : SUCCESS in 1h 10m 47s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/0b31250e90544ca0ae9230ceff1e410b : SUCCESS in 36m 54s","accounts_in_message":[],"_revision_number":7},{"id":"80e2dd7591f87ffd5a6494728fdf9460e5de73e4","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-19 18:52:28.000000000","message":"Patch Set 7:\n\nrecheck timeout","accounts_in_message":[],"_revision_number":7},{"id":"bb117a212aa7524b3e5e8bbb112232cf386d92d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-19 20:12:52.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/3a4e79206795470e96e17a0729efcb1a : SUCCESS in 1h 01m 16s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/56983eed699041508f3a0b37998773dc : SUCCESS in 1h 10m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/72d30090a4a448a89a544e4674a36232 : SUCCESS in 8m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/931aaddb2df64cb78a59ff273c42d7ca : SUCCESS in 6m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9c66e22a177540c6ac35e21706496a5f : SUCCESS in 6m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/afd1aa5773754db9a542e37e0306ea24 : SUCCESS in 7m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/035cd621906246c3bfebd4088739faec : SUCCESS in 6m 58s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/39e76b18001a4f7c9c666a2848dfdc0a : SUCCESS in 10m 28s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/25f9b8b4e86d4b3e92f3eb07c66e9667 : SUCCESS in 9m 44s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/603d6402b3db4ee0851820d1d63bbcd2 : SUCCESS in 9m 24s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/61fe293124214a0c9131d8641341060b : SUCCESS in 5m 57s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/a6d11434768d4d649f062cba51c31d3e : SUCCESS in 1h 12m 03s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/28bb08fd32444f74809b5ec9b9441642 : SUCCESS in 1h 11m 46s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/d4aac816352e411386b67c92f81b0ff5 : SUCCESS in 1h 09m 11s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/742c652409dc432d90c01493ac93efa4 : FAILURE in 1h 03m 37s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5decd91a72fe4972b6df0cd54433468d : SUCCESS in 47m 25s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/bd407bc05a4943f2a8767871e7a47d93 : SUCCESS in 1h 17m 40s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/32a2016861cb488ba90839b9af681f54 : SUCCESS in 23m 41s","accounts_in_message":[],"_revision_number":7},{"id":"0cea92747308c1d7bce60dfd4627c77c72551aed","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-08-20 01:27:07.000000000","message":"Patch Set 7:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/804558/804558/7/check-not-vote/ext-glance-zuul/04a2a12 : SUCCESS in 33m 59s","accounts_in_message":[],"_revision_number":7},{"id":"0cb82c990898ca7eb02fc86eddae9e813f804d3c","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-08-20 16:33:14.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"e60ebef907dab18513c6f9e9220e96868811cc83","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-20 16:34:22.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"de557dd335864e7ab42249c71a83b635d1f4e2bb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-20 19:22:35.000000000","message":"Patch Set 7: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/801ef1be0cba49be81636fa0ad07886a : SUCCESS in 1h 01m 14s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/bc9e1538b12543ee9612a27940c17f16 : FAILURE in 1h 01m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/55ed800a5eea4dd4ba13f5e3c3ff6d38 : SUCCESS in 6m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d9468579624d44ddbdad25f46da68aaa : SUCCESS in 7m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3d2dfde4d36b4c77a0039b393e3d66e4 : SUCCESS in 7m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2abe6b3f90404b1aa88c3e4d524eb0d4 : SUCCESS in 8m 06s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8d7ee897a7524614b00b920a1902739d : SUCCESS in 10m 18s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/285b22eaee654d4987b0ea54ae4c6a73 : SUCCESS in 9m 43s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/250b6f6a2600419dab6243b422171226 : SUCCESS in 1h 01m 02s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/b365fd896ca945de91a1a347f570d7df : SUCCESS in 1h 03m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ab6450fd31c74026998828f5f9c99663 : SUCCESS in 46m 39s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/d933830fd0d048e8ba5d6287efd33636 : SUCCESS in 1h 18m 55s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/9846916eab0746f999e2a55eb403a235 : SUCCESS in 31m 04s","accounts_in_message":[],"_revision_number":7},{"id":"6e48fea533a3ea841909f0453e2c7716b436638e","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-20 19:33:33.000000000","message":"Patch Set 7:\n\nrecheck tempest failure","accounts_in_message":[],"_revision_number":7},{"id":"06cb63d85ae516cc8a2de8196262f133ab5625b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-20 21:54:29.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/bbff32dc67e942df83895c6e6b85c919 : SUCCESS in 1h 04m 01s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ea1455f82507476db6dfbf38a74a4464 : SUCCESS in 1h 04m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee63b3ef2c1945108e99e545892d5379 : SUCCESS in 6m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/db2194bf4e9d412c8ce72b0dd5b64b56 : SUCCESS in 7m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/470f2d25087c411fb8b062dbe3a55bf9 : SUCCESS in 6m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5bfabb8dd2644752bf5b2f94b5fd6dca : SUCCESS in 7m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1dbd65b1b5f94d24bf2dfcc9265593d6 : SUCCESS in 8m 32s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5be1beba6339475a9f36c8a13ed000dc : SUCCESS in 9m 48s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/2c4643a1b80140efa7cdf5ac7c3b2a9e : SUCCESS in 9m 03s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/d5678f1c84dd4c30941715f83cf34549 : SUCCESS in 8m 46s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/9f728f3affbb46caaa65ec73b50492e8 : SUCCESS in 9m 10s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/6d1ff3f8536e4e46941b3b0fe23e0ab0 : SUCCESS in 1h 18m 15s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/4b95ec0c50b541d3925c5a0866299197 : SUCCESS in 1h 05m 17s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/d49469782fe04442ab47e64beee98509 : SUCCESS in 1h 06m 01s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/c9b42b5148ff4aabaaf065c649f30c7c : SUCCESS in 1h 09m 57s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1ec3cb7ce43f423980741199992f44e5 : SUCCESS in 58m 30s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/78d0467ab9e7425ab5f34cf49555d907 : SUCCESS in 1h 21m 42s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/0a873003f2944bf39fb9546b9398b942 : SUCCESS in 26m 42s","accounts_in_message":[],"_revision_number":7},{"id":"864acfb0bf044d21a79dc436be3c86fa2cb4ed69","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-20 21:55:31.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"92a496f063aa1dc8f3c16e6f6d39224ad27040a0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-21 00:31:04.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/8aa545e10c3440578fe90afe4f976555 : SUCCESS in 1h 05m 04s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5b6b81da96a34c7bb7275085a73c97ad : SUCCESS in 1h 08m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/536c977621284ff9b8aba6506441de67 : SUCCESS in 6m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9f9378c52a16489bbce87642555ca382 : SUCCESS in 6m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0d7f5df8b7d64a4496baf3d5e0d6ff7b : SUCCESS in 6m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c81e0835b1ee4e9186830ab120d0e0fe : SUCCESS in 7m 30s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4a12e82206b24df7abe9399f0fec00ac : SUCCESS in 9m 22s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/d680e7eb27c9423496e4f86d76b2441e : SUCCESS in 9m 17s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/cb3a212b154446e0840fe32d84e11ed4 : SUCCESS in 1h 02m 16s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/854fe362a52548cc85ada473bafd6726 : SUCCESS in 1h 01m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c4dfce6017da456eafb48fd9aefd0f20 : SUCCESS in 41m 15s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/433d9e27da9648b489659855a21353f1 : SUCCESS in 1h 35m 50s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/0468f65852b841f0907c4cadc9351753 : SUCCESS in 24m 53s","accounts_in_message":[],"_revision_number":7},{"id":"b225fffb5ac63222acd351427293f3323bb467f1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-21 00:31:05.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":7},{"id":"a9c8a3c7a72ee15c0be5c53c769f42538f45061b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-21 00:32:22.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ff2f6b9b74b46f2ae6a83d4b4b6ddb8 : SUCCESS in 56s","accounts_in_message":[],"_revision_number":7},{"id":"32b8408ea22071633465cf63ea6417339c424ebf","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-08-21 05:45:20.000000000","message":"Patch Set 7:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/804558/804558/7/check-not-vote/ext-glance-zuul/c86781e : SUCCESS in 38m 34s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"fa6f871e2031505ccb420dc436f293815b87088c","revisions":{"3c87cad71aaf20f105a32afb37637d942cbb4332":{"kind":"REWORK","_number":1,"created":"2021-08-14 10:40:55.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/58/804558/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/58/804558/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/58/804558/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/58/804558/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/58/804558/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/58/804558/1"}}},"commit":{"parents":[{"commit":"35c60ba7a5d47692fa864090f0dde27b51af5e08","subject":"Check download_image policy in the API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/35c60ba7a5d47692fa864090f0dde27b51af5e08"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-08-14 09:39:07.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-08-14 10:34:37.000000000","tz":0},"subject":"PoC Check policies for staging operation in API","message":"PoC Check policies for staging operation in API\n\nToDo: Add functional and unit test coverage\n\nPartially-Implements: blueprint policy-refactor\n\nChange-Id: Iba990e59c454a2e9f37a0c659d8634583d3a8bf7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3c87cad71aaf20f105a32afb37637d942cbb4332"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3c87cad71aaf20f105a32afb37637d942cbb4332"}]},"branch":"refs/heads/master"},"6231c8707b3d1349f45d30d2c138a6176ac148bf":{"kind":"REWORK","_number":2,"created":"2021-08-14 20:45:06.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/58/804558/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/58/804558/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/58/804558/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/58/804558/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/58/804558/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/58/804558/2"}}},"commit":{"parents":[{"commit":"35c60ba7a5d47692fa864090f0dde27b51af5e08","subject":"Check download_image policy in the API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/35c60ba7a5d47692fa864090f0dde27b51af5e08"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-08-14 09:39:07.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-08-14 20:38:42.000000000","tz":0},"subject":"Check policies for staging operation in API","message":"Check policies for staging operation in API\n\nThis patch enforces policy checks required for staging image data\nin API layer. Image mutability check is maintained if secure RBAC\nis not enabled in deployments.\n\nPartially-Implements: blueprint policy-refactor\n\nChange-Id: Iba990e59c454a2e9f37a0c659d8634583d3a8bf7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6231c8707b3d1349f45d30d2c138a6176ac148bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6231c8707b3d1349f45d30d2c138a6176ac148bf"}]},"branch":"refs/heads/master"},"64bf872e52fcec6b018b5cce87779e546a73ec96":{"kind":"REWORK","_number":3,"created":"2021-08-15 09:09:30.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/58/804558/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/58/804558/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/58/804558/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/58/804558/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/58/804558/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/58/804558/3"}}},"commit":{"parents":[{"commit":"35c60ba7a5d47692fa864090f0dde27b51af5e08","subject":"Check download_image policy in the API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/35c60ba7a5d47692fa864090f0dde27b51af5e08"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-08-14 09:39:07.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-08-15 09:04:54.000000000","tz":0},"subject":"Check policies for staging operation in API","message":"Check policies for staging operation in API\n\nThis patch enforces policy checks required for staging image data\nin API layer. Image mutability check is maintained if secure RBAC\nis not enabled in deployments.\n\nPartially-Implements: blueprint policy-refactor\n\nChange-Id: Iba990e59c454a2e9f37a0c659d8634583d3a8bf7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/64bf872e52fcec6b018b5cce87779e546a73ec96"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/64bf872e52fcec6b018b5cce87779e546a73ec96"}]},"branch":"refs/heads/master"},"7c6cc5f9a869cea8a88823d39f37c1dde4f3ff1b":{"kind":"REWORK","_number":4,"created":"2021-08-16 19:54:36.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/58/804558/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/58/804558/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/58/804558/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/58/804558/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/58/804558/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/58/804558/4"}}},"commit":{"parents":[{"commit":"004402ed675ec8003973522429b03a9d4e7da336","subject":"Check download_image policy in the API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/004402ed675ec8003973522429b03a9d4e7da336"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-08-14 09:39:07.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-08-16 19:45:44.000000000","tz":0},"subject":"Check policies for staging operation in API","message":"Check policies for staging operation in API\n\nThis patch enforces policy checks required for staging image data\nin API layer. Image mutability check is maintained if secure RBAC\nis not enabled in deployments.\n\nPartially-Implements: blueprint policy-refactor\n\nChange-Id: Iba990e59c454a2e9f37a0c659d8634583d3a8bf7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7c6cc5f9a869cea8a88823d39f37c1dde4f3ff1b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7c6cc5f9a869cea8a88823d39f37c1dde4f3ff1b"}]},"branch":"refs/heads/master"},"d85546d4773dcf68d278de12a5fdfe70ad91f377":{"kind":"REWORK","_number":5,"created":"2021-08-17 17:38:20.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/58/804558/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/58/804558/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/58/804558/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/58/804558/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/58/804558/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/58/804558/5"}}},"commit":{"parents":[{"commit":"8f2b5ec5699df2df3e8e7ec264a67a21a0249433","subject":"Check download_image policy in the API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8f2b5ec5699df2df3e8e7ec264a67a21a0249433"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-08-14 09:39:07.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-08-17 17:33:46.000000000","tz":0},"subject":"Check policies for staging operation in API","message":"Check policies for staging operation in API\n\nThis patch enforces policy checks required for staging image data\nin API layer. Image mutability check is maintained if secure RBAC\nis not enabled in deployments.\n\nPartially-Implements: blueprint policy-refactor\n\nChange-Id: Iba990e59c454a2e9f37a0c659d8634583d3a8bf7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d85546d4773dcf68d278de12a5fdfe70ad91f377"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d85546d4773dcf68d278de12a5fdfe70ad91f377"}]},"branch":"refs/heads/master"},"f164cb77b7bd7f879ad52f6a07e1026d0ccc040b":{"kind":"REWORK","_number":6,"created":"2021-08-18 09:27:56.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/58/804558/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/58/804558/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/58/804558/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/58/804558/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/58/804558/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/58/804558/6"}}},"commit":{"parents":[{"commit":"25333dc7dff2ba065a2114a6b42d2d51daeb5f34","subject":"Check download_image policy in the API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/25333dc7dff2ba065a2114a6b42d2d51daeb5f34"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-08-14 09:39:07.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-08-18 09:13:05.000000000","tz":0},"subject":"Check policies for staging operation in API","message":"Check policies for staging operation in API\n\nThis patch enforces policy checks required for staging image data\nin API layer. Image mutability check is maintained if secure RBAC\nis not enabled in deployments.\n\nPartially-Implements: blueprint policy-refactor\n\nChange-Id: Iba990e59c454a2e9f37a0c659d8634583d3a8bf7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f164cb77b7bd7f879ad52f6a07e1026d0ccc040b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f164cb77b7bd7f879ad52f6a07e1026d0ccc040b"}]},"branch":"refs/heads/master"},"fa6f871e2031505ccb420dc436f293815b87088c":{"kind":"REWORK","_number":7,"created":"2021-08-19 15:59:54.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/58/804558/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/58/804558/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/58/804558/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/58/804558/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/58/804558/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/58/804558/7"}}},"commit":{"parents":[{"commit":"6ffd80f9c0b15d3c2fdf17476c1a1348ea0705ce","subject":"Check download_image policy in the API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6ffd80f9c0b15d3c2fdf17476c1a1348ea0705ce"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-08-14 09:39:07.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-08-19 15:50:25.000000000","tz":0},"subject":"Check policies for staging operation in API","message":"Check policies for staging operation in API\n\nThis patch enforces policy checks required for staging image data\nin API layer. Image mutability check is maintained if secure RBAC\nis not enabled in deployments.\n\nPartially-Implements: blueprint policy-refactor\n\nChange-Id: Iba990e59c454a2e9f37a0c659d8634583d3a8bf7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/fa6f871e2031505ccb420dc436f293815b87088c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/fa6f871e2031505ccb420dc436f293815b87088c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
