)]}'
{"id":"openstack%2Fglance~804588","triplet_id":"openstack%2Fglance~master~I7b5bcbce9375df5b0826120df5cdf5d4cb788af2","project":"openstack/glance","branch":"master","topic":"policy-refactor","hashtags":[],"change_id":"I7b5bcbce9375df5b0826120df5cdf5d4cb788af2","subject":"Check policies for Image Tags in API","status":"MERGED","created":"2021-08-15 17:26:22.000000000","updated":"2021-08-26 13:34:32.000000000","submitted":"2021-08-26 04:45:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":14,"unresolved_comment_count":5,"has_review_started":true,"submission_id":"804588-policy-refactor","meta_rev_id":"77f0b97b50876294571e9d0fd6e4501e62301f61","_number":804588,"virtual_id_number":804588,"owner":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-26 04:45:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"all":[{"value":2,"date":"2021-08-25 17:55:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":2,"date":"2021-08-25 18:44:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":1,"date":"2021-08-25 18:44:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]}],"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-15 18:38:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-16 19:33:31.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"CC"},{"updated":"2021-08-21 08:29:55.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-08-24 14:06:35.000000000","updated_by":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"reviewer":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"state":"CC"},{"updated":"2021-08-24 16:16:36.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2021-08-25 15:10:55.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"CC"},{"updated":"2021-08-25 17:55:43.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"}],"messages":[{"id":"af256775c4f8c8c8b9039d4b31f2cb68ad95a7f6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-15 17:26:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e043cd2df0f50b454732a88724895c3ac7eeb6f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-15 18:38:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/3cb0f092ee15430391dac5188dc0a831 : SUCCESS in 52m 41s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/bc4a0ee47ad242a39eca27bbb8fff4a5 : SUCCESS in 59m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/68a75391056647228e102cbd9cf0ca11 : SUCCESS in 6m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9de233b110e1408fb8c5162abff4fcb4 : SUCCESS in 7m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d5505bc0d3b04d349f87db6065e9b5b9 : SUCCESS in 6m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/83b44484744841c0b5878f01b69a0c7f : SUCCESS in 7m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/12bd9e5ccf644c939beebcfb0f6ad68e : SUCCESS in 7m 32s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/545ce680c1ee4d6185eca4d190a38813 : SUCCESS in 9m 57s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/4062f8a774c74fc1ab38e0147c9b57f6 : SUCCESS in 9m 51s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/d0c7965387ab43cb946b80829581ecd9 : FAILURE in 9m 33s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/8f09c42d77f9470da2fc12ec09efc88c : SUCCESS in 6m 00s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/3f0d1a833f444eb595029a81bef2b5ba : SUCCESS in 1h 06m 47s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/c43c9208ad004db98e881e08937d2116 : SUCCESS in 1h 07m 36s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/4196a4b95ea44f9084786861920e9e94 : SUCCESS in 1h 03m 04s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/3c9c70cf9535458d89d0901755b2b3da : SUCCESS in 1h 10m 36s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d1daa4788c1f4a4aa8cb69ffbb21f4e5 : SUCCESS in 43m 20s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/d857dc5739744ccbbd0dd338a640c0d4 : SUCCESS in 1h 04m 15s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/d6c5d278af6e4a00b3e856ccc35b2d79 : SUCCESS in 29m 01s","accounts_in_message":[],"_revision_number":1},{"id":"5d04f19a7a17afaa584b0a4579a29546d2f173b4","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2021-08-16 19:33:31.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"91256bfacba00b5e76fd2571c1ecafee203ebcff","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-18 16:56:28.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"30ef4f41dc35aeb48eb8604ec88293b8c392d8a7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-20 06:43:22.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b42293e810386919baff2a55632b91ed1b95aafe","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-20 06:50:05.000000000","message":"Topic image-tag-policy-refactor removed","accounts_in_message":[],"_revision_number":2},{"id":"f4befd988d02e2952d645fcd4ba2d5f1cbc0dbac","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-20 06:50:13.000000000","message":"Topic set to policy-refactor","accounts_in_message":[],"_revision_number":2},{"id":"c56e765aa23cfe49725ca80b0dbd4b831b6c3026","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-20 08:33:22.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/51ce31ffb6064f2a89e9931d41256255 : FAILURE in 51m 06s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/83b1dbfc7daa445aaa9f0a971cbb019d : SUCCESS in 1h 06m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b31621d46ef4c94bd223ed7c3378eaf : FAILURE in 6m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d692e6271c774989a021c7b6dbd1ce95 : FAILURE in 6m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7bff475a57c84dffa5549529b8715094 : SUCCESS in 7m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/aacfc761059d4425b5b6fed295cf7eb7 : SUCCESS in 7m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f4c6626112f4a488de00809080d63f1 : SUCCESS in 8m 55s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/94f03bccd396494caf92c959f78146eb : SUCCESS in 9m 54s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/cc333d98c4cd463e807d4cd90c878bb6 : FAILURE in 12m 34s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/615992798d0048049b5c977cea8270f3 : FAILURE in 11m 44s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/2bafab644edb4fd9b0289b36f856d371 : SUCCESS in 6m 17s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/b6e6a47ddc3e48ecab95f4bd87d54cf0 : SUCCESS in 1h 24m 17s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/15a3298df9cd440e923e8c6a4d942318 : SUCCESS in 55m 03s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/908da63e1e6c4a57abff20d99c980a0e : POST_FAILURE in 16m 08s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/3d6d3c717be54c629bbf162f7519439a : POST_FAILURE in 15m 46s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3e4132f2846d4c13a296e8112abe4af8 : FAILURE in 15m 34s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/e3ec463aa62141d498116b123036e9b3 : FAILURE in 15m 50s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/c32d14ffc7f54ecdbc660bff3578d276 : FAILURE in 16m 10s","accounts_in_message":[],"_revision_number":2},{"id":"549dbf68501697945d60c473f2388906aaa12870","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-20 19:16:32.000000000","message":"Patch Set 2:\n\nrecheck mirror failure","accounts_in_message":[],"_revision_number":2},{"id":"3519c04d0ebb29eecd8f6666e76661c518b5eca6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-20 19:56:00.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"67db64f3106c3e04011eab7de05b5bce223a9577","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-20 22:09:47.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/df9a17db42784de3924e4bc8c9fbf316 : SUCCESS in 1h 03m 04s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4fdee9294ab04c7bb229cfa7473a5a9a : SUCCESS in 1h 04m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c4b73cdfc2ed46e79098749f12ee0e8b : SUCCESS in 6m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/75e09ef164dd4e72b5794cccd5bd3345 : SUCCESS in 6m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/488b052ffc2e4df7a8e75e7a00c0384f : SUCCESS in 8m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0b454d38869b43b282a91ccd51c66e4f : SUCCESS in 7m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37f70f053b9d4b3d8727a9f45c18c279 : SUCCESS in 7m 23s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/557c2de73d6c4417bf9e910549678ab4 : SUCCESS in 11m 51s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/6a35b13219c349e6885f537096bd6561 : SUCCESS in 9m 46s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/21b49ba4b63046609315993a14dc598a : SUCCESS in 9m 58s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/2f068f87e72745b2a90cd7a07a882aec : SUCCESS in 6m 08s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/8f6fbf9ed2e7491082e884907318dec6 : SUCCESS in 1h 11m 54s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/2d3411215192406184ea6aaffe6cb191 : SUCCESS in 1h 10m 39s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/3b24226eda774e349d0a5b4b5d38517a : SUCCESS in 1h 07m 04s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/8b40b38db7d64f858eb67de138330e66 : SUCCESS in 1h 24m 54s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/30b08beb628144fda6a0f41f1f66c280 : SUCCESS in 47m 51s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/6d017e3a569e4c8b838568f308ce1a13 : SUCCESS in 1h 07m 20s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/e82898b40bc2495787a569a422d2a55b : SUCCESS in 30m 36s","accounts_in_message":[],"_revision_number":3},{"id":"54477b2f17c8e7166d6a666a050aad4211b5160a","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-08-21 08:29:55.000000000","message":"Patch Set 3:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/804588/804588/3/check-not-vote/ext-glance-zuul/eb09b52 : SUCCESS in 34m 33s","accounts_in_message":[],"_revision_number":3},{"id":"53b898a81cdf58d6c063d39a7f061d905cb83d77","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-23 06:23:24.000000000","message":"Topic image-tag-policy-refactor removed","accounts_in_message":[],"_revision_number":3},{"id":"1f92206baa731f896d481cbdb381aa946549e877","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-23 06:23:32.000000000","message":"Topic set to policy-refactor","accounts_in_message":[],"_revision_number":3},{"id":"592fc3c879fcd68bb2bfe4fc5d89525435c68bbb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-23 07:15:26.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"6fad4457cc246584588630d45aab6e1f0fc6262a","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-08-23 07:53:27.000000000","message":"Patch Set 4:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/804588/804588/4/check-not-vote/ext-glance-zuul/38ed43f : SUCCESS in 37m 06s","accounts_in_message":[],"_revision_number":4},{"id":"ba0c926e83a567b2bcb5db3b2ea960c3d701e0b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-23 08:31:27.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/697b5014593b4181973bfcc8311cdb84 : SUCCESS in 1h 00m 04s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c07e1357028c4d468c3fcf29bab1f0d8 : SUCCESS in 1h 07m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8988b436d1ba4faabe824957bcc93444 : SUCCESS in 6m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/30ef80e800a54326944e2af090728367 : SUCCESS in 5m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cbb60e79ecb34e22b50ac6e7d5ee788a : SUCCESS in 6m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1f1d20b58de749e1b76836793393ea79 : SUCCESS in 7m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/89420829023d4b2d93c9fa69ded0d2d1 : SUCCESS in 7m 49s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6cf04d9946914e21a9f2103e4057e919 : SUCCESS in 9m 58s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/41e20536d4614a9981a0187af87e00d9 : SUCCESS in 10m 06s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/eb304b100f6846ae8201cc8c8763a748 : SUCCESS in 10m 38s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/ceaed2215dd649b4bb2191ccd19c09bb : SUCCESS in 5m 44s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/3ddfbe3f0821456c8cf518c7643d96b1 : SUCCESS in 1h 08m 41s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/1eea570013544718972dd0a277b37b3c : SUCCESS in 1h 09m 58s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/6030e12789424cd0ae5d24d44909aea7 : SUCCESS in 1h 07m 53s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/9c85613418a946ae853c849998807086 : SUCCESS in 1h 14m 29s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/151d48959f0945fc83fb7ca45d9053ae : SUCCESS in 50m 26s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/e92ede99ea5144bbb35d0bd26f17dce2 : SUCCESS in 1h 13m 52s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/fbd5a343cd86433aa0292699ce934422 : SUCCESS in 27m 25s","accounts_in_message":[],"_revision_number":4},{"id":"1002d555e356a048365718cce1d32af16e5562c3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-24 06:15:20.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"fa3b082985d2e50b292a1f5815c7f41313ef1028","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-24 08:24:14.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/a3537e3d1ef74cc4aea8175e88468755 : SUCCESS in 1h 05m 40s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a26d5d6d49684f6c998299695dcda58a : SUCCESS in 1h 19m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7794e63828e44ed98bfe9b895c279377 : SUCCESS in 7m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5ac83180afb64a43b6ffda7ced70b4eb : SUCCESS in 7m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/69400577e9524957a0cc62ade9e7121a : SUCCESS in 6m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f5c3c069080a4b7c94dc54ac7eb7f3dd : SUCCESS in 8m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/392675fdb8a04a17870adbda6bc3018e : SUCCESS in 10m 46s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/302cca2e48d04ce7ac277e7dcc525bd8 : SUCCESS in 11m 45s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/8280a1ddf95b4a35aed48c365fc65533 : SUCCESS in 10m 13s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/62a24590777f43aa82375403699b2eea : SUCCESS in 10m 21s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/12250c505e3741c5b95c9236df41d84b : SUCCESS in 8m 53s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/1e2f7c4e47b547fe827e551052cf1b5d : SUCCESS in 1h 15m 40s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/4ab42f8fcead4371871c6169fb228b25 : SUCCESS in 1h 02m 40s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/e227f59fe1b144a89789f930e2f8d0ee : SUCCESS in 1h 10m 28s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/cacc36ce21c941f798119a761312d7fa : SUCCESS in 1h 11m 56s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bb3e842310094c558b2962fc561d0584 : SUCCESS in 45m 28s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/878213c8e6be4d73afa9c66a1f0680bb : SUCCESS in 1h 14m 47s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/b27e13cfc21148bc9e6db85c831f2ac9 : SUCCESS in 30m 37s","accounts_in_message":[],"_revision_number":5},{"id":"d09c7ed839a78d98cca17cb0b7c6b718b9e9edf2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-24 09:36:10.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"dcdc1dee5fa9eb64db2628b5e64f85afe745c25d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-24 11:16:47.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/96c11a96de384d7587f79343b16a154f : SUCCESS in 1h 06m 24s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d3f9a5ba577147d3ae362d9a8db746ab : SUCCESS in 1h 06m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f377bb53b581447bb5aec4efc60cadba : SUCCESS in 9m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/150558e4a15943b586172797ee6e7ec0 : SUCCESS in 8m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1004ce53eddf4e5abbbf1b460388c79d : SUCCESS in 10m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fd669cfb5b1940ec9f65f60ec056a42f : SUCCESS in 8m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0acb59f82a494813961703b26902f0e7 : SUCCESS in 9m 58s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/13106ede83154c8a86bb8149d6f9fce3 : SUCCESS in 11m 36s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/33ff04187f064d14b7b8191b7a27b83f : SUCCESS in 11m 14s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/2ce97461105d474895bdeaa46b4cd18f : SUCCESS in 11m 28s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/53cf0b909b514c629e346c3db22ee6a2 : SUCCESS in 7m 47s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/2a220a7ce5a84d99aa23e192ecad1d7c : FAILURE in 30m 04s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/ccf3927300694a5eb3fe7c871e82780f : SUCCESS in 1h 10m 41s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/b0fe80b8680a4a07b6f8ce61f618ecad : SUCCESS in 1h 09m 54s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/55e3bcec929d4c95af6480f04127bfc5 : FAILURE in 1h 01m 23s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/601dc60b8a0f4cedb41624ebc4dd31be : SUCCESS in 51m 00s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/6d38553fa32640d4a6894433e4b4fe33 : SUCCESS in 1h 13m 23s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/7c91ef612a54402a9c4318eb619f558f : SUCCESS in 32m 09s","accounts_in_message":[],"_revision_number":6},{"id":"fb02f92e0ebf8fe87f632059c0019e6a98cc28bb","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-08-24 14:06:35.000000000","message":"Patch Set 6:\n\n(1 comment)\n\nLooks good to me, just one question inline.","accounts_in_message":[],"_revision_number":6},{"id":"2bbd5c77fc65532577664e5d352b02978d1580e7","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-24 14:45:44.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"f40224c0929263dcfa29eb861306473684d080e3","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-24 14:57:26.000000000","message":"Topic policy-reafactor removed","accounts_in_message":[],"_revision_number":6},{"id":"11170639ee698b940eea7ed056640cd8ac03d933","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-24 14:57:31.000000000","message":"Topic set to policy-refactor","accounts_in_message":[],"_revision_number":6},{"id":"0a2e504d2ed70fdae73047fc2cde98e6510b259a","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-08-24 15:47:40.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"feb461d30591acb1007285a1ef0b71c90bd81841","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-24 15:51:58.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"1985c3713edf6c05106b39d38228b9bbd127de30","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2021-08-24 16:16:36.000000000","message":"Patch Set 6: Code-Review-1\n\nDon\u0027t we want a test_tag_image_create since we have test_tag_image_{update,delete}?","accounts_in_message":[],"_revision_number":6},{"id":"c844f5666b9fa00bb70f95c445fa092c57d94275","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-24 16:49:29.000000000","message":"Patch Set 6:\n\n\u003e Patch Set 6: Code-Review-1\n\u003e \n\u003e Don\u0027t we want a test_tag_image_create since we have test_tag_image_{update,delete}?\n\nImageTags supports only two APIs put and delete [1]\n\n[1] https://docs.openstack.org/api-ref/image/v2/index.html#image-tags","accounts_in_message":[],"_revision_number":6},{"id":"8fc5dc97f1fc05ec881355be28ce517698eb705e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-24 18:28:15.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"112761c4e35e1a0a15a32c747cb6f77ae197cd3b","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-08-24 18:42:41.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"0ba5560362065fee76082e479929e8c40702d651","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-24 18:54:14.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"214949e15ca3dbc3459ee5e4560cef9ea633fef4","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2021-08-24 19:32:59.000000000","message":"Patch Set 7: Code-Review+2\n\nThanks for the clarification","accounts_in_message":[],"_revision_number":7},{"id":"b83f24852f3c144e5c4b368625d09a66a33d66b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-24 19:55:55.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/6b0b2205a4554ac3bf6d9c78a1cf625a : SUCCESS in 1h 16m 45s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/682f4dfbbc50493e984c8d0b74c6850f : SUCCESS in 1h 03m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/63bfbd6877b1477e951decc50fc5437a : SUCCESS in 5m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8b0694cd0e5945eb85baf862ad34c99b : SUCCESS in 5m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5941df46789243e8a97cfc8af887cce2 : SUCCESS in 6m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9d01f716d8c9400bbcd2091d3c6b9fa1 : SUCCESS in 7m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5218414cec3489c870f28fc1f40208c : SUCCESS in 7m 02s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/132277153f4741dbbd180cf49d2483c0 : SUCCESS in 9m 04s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/706f5547beac4935abfadbecc5a562da : TIMED_OUT in 31m 22s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/bf686f42d8554d079c054107a3326e48 : TIMED_OUT in 31m 14s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/bacf67d00ea04def97b12301d7bf921f : SUCCESS in 5m 29s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/2ea98bee1a22472b86d10caab235f773 : SUCCESS in 1h 10m 29s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/1a484daac5c245888601918e02aaa777 : SUCCESS in 1h 25m 31s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/6cd786a212d74c948017b018ced9e172 : SUCCESS in 58m 43s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/94fa231a3a6c47ca89c821df20e4927c : SUCCESS in 1h 03m 48s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3ed5359640ba4e3fb378e288c2d5b00a : SUCCESS in 1h 12m 15s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/19d405283b6044f29528bcc82ccb4d4c : SUCCESS in 1h 21m 29s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/c1b7c4c04aa14f59bb8f48efba03d146 : SUCCESS in 32m 30s","accounts_in_message":[],"_revision_number":7},{"id":"5f77286a9fd2fce529b31ef3afd0c7aea6415efd","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-08-24 20:14:42.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"cddd4e32a0824ea401bc730c2dfdbeec758395ce","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-24 20:17:05.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"39ad533b62d1f832dbe5f36a20eae73577349b89","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-24 21:56:04.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/1a80432894314544bdee4de06e90841a : SUCCESS in 59m 17s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2afdbef99bcc422abb08475161bbadc6 : SUCCESS in 55m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/67415ce9e2874c5184b1da00cf69452f : SUCCESS in 6m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/269cbbd6480344d2996607d314cb8bb3 : SUCCESS in 6m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c452d6bf3e79452097375625738f99ac : SUCCESS in 7m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5121bf714f58482796090f42960cdac3 : SUCCESS in 6m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc99b6e6853042ea93acd01532ee8ce0 : SUCCESS in 8m 49s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/235a7fde34bf46a6b26bd536c0b0a306 : SUCCESS in 9m 15s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/6dbd51f165b64631a3a67e97098ba01a : TIMED_OUT in 31m 20s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/a37b40dfaabc49cfb2da352fb3a86e29 : SUCCESS in 9m 21s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/cc834b9e4b3c48ffb9c7bc9fd8fdbb0d : SUCCESS in 5m 24s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/2e469a20cd9140488cb0b5508ed01f84 : SUCCESS in 1h 32m 02s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/b317069b324e42259e7eef673e19a6f8 : SUCCESS in 1h 02m 12s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/c5266ff31f5e44b6b63ca754c7bf35b8 : SUCCESS in 1h 04m 06s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/78f756eea21e40f69f250253e37ec667 : SUCCESS in 1h 09m 13s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b25b470d89244cb6b14655cbb6645286 : SUCCESS in 49m 53s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/1a4e85c348324e55a22af46d5bec3c5e : SUCCESS in 1h 04m 54s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/63b6958f791346d29180d8706e1465c7 : SUCCESS in 28m 36s","accounts_in_message":[],"_revision_number":8},{"id":"aa1e537ef8b1b272494d3bfebaf715c5dfa6b93a","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-24 21:57:16.000000000","message":"Patch Set 8:\n\nrecheck timeout","accounts_in_message":[],"_revision_number":8},{"id":"684584cecce15285782fdfcaba4f938b7e51977b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-24 23:40:33.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/6eb4371c8fb84ff2be66cb75f7d68059 : SUCCESS in 1h 28m 53s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1cfa59c7c87f489bb2545486c0df1751 : SUCCESS in 1h 25m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d19bc16878448cfb937024ec0c888f6 : SUCCESS in 5m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b247e5adc9464512b45c936b73c65924 : SUCCESS in 6m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/359595ceb0db42f8808d088c69d0be06 : SUCCESS in 6m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7d54475e173343749956ecaeedba61ee : SUCCESS in 7m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8fa211046bc347749e9e79217b229e0a : SUCCESS in 7m 59s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8975da8aceed46d6b3f6e7fd93e2b46b : SUCCESS in 10m 42s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/f13ea2bf354a4824a4a6cc28a8c8c64c : SUCCESS in 9m 12s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/caecbf2d0ee0434088d555630a115e86 : SUCCESS in 9m 03s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/74d7f27aeb5541e6bf9c287de8cb54b4 : SUCCESS in 5m 24s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/91ecfc19a43f4fc08ecbcc62c7abb672 : SUCCESS in 1h 12m 42s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/4f41eaf586d4417d8188228dd5bb03bf : POST_FAILURE in 16m 30s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/1ad6f793b87f41589df96eab0e92c1e4 : SUCCESS in 1h 03m 32s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/7990c2115fab4c439ea2d26a8cadb39c : FAILURE in 1h 03m 11s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5470f550a21e40f591cee66bbc5671ad : SUCCESS in 45m 53s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/39edb0d4cd5a4f8fb6de0baade192565 : SUCCESS in 1h 09m 58s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/9efce5c9b1f346ffa3739d52afdd3500 : SUCCESS in 27m 22s","accounts_in_message":[],"_revision_number":8},{"id":"e17822a1ddb34466876a4fac42533def85589935","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-25 05:33:41.000000000","message":"Patch Set 8:\n\nrecheck timeout","accounts_in_message":[],"_revision_number":8},{"id":"a77dcd3ed767063c530fc6b1ea22f514b03de33e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-25 06:37:35.000000000","message":"Uploaded patch set 9: New patch set was added with same tree, parent, and commit message as Patch Set 8.","accounts_in_message":[],"_revision_number":9},{"id":"071be05a69875cacad504bc9f09a8d82512da30a","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-08-25 08:33:16.000000000","message":"Patch Set 9:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/804588/804588/9/check-not-vote/ext-glance-zuul/d19ad50 : SUCCESS in 34m 52s","accounts_in_message":[],"_revision_number":9},{"id":"861698a80dd38ef458a5b501fd2d292301b63313","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-25 08:53:06.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/cac9464262494fdb8c585e2270a1e780 : SUCCESS in 1h 05m 08s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f48cea9bbc324cbb816516d22fa50950 : SUCCESS in 1h 34m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b66578ee82124a5d9e9934b9d5da9761 : SUCCESS in 7m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2c6032f19e68416294a7df01c88acaa1 : SUCCESS in 8m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c465e17ae5ec46528ed78a9ce927a851 : SUCCESS in 8m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f27b34d81bd5468585db6432caaa8d2b : SUCCESS in 9m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/812f04ada7204bc5829635f5e3b72414 : SUCCESS in 9m 32s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a2354121dc4e492cba805c58c9a480e5 : SUCCESS in 9m 20s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/79c246253c5b432cbb928e0a51ceb4f7 : SUCCESS in 11m 45s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/d10f6600f9be4dfb8ae9d97709b45790 : SUCCESS in 13m 31s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/b74a134cc8c740ba8a19b370c5ef4aa4 : SUCCESS in 9m 13s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/0d489fa366f44126982d3c9339d770eb : SUCCESS in 1h 19m 58s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/8cdd06a3379449eb83195ba3583feee1 : SUCCESS in 1h 25m 42s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/98694f26a2984657bff6652723fcf236 : SUCCESS in 1h 06m 29s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/60333d08e57f498f8ec6bb58d3df6522 : SUCCESS in 1h 19m 53s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8ddb8de39a4e4c0d9cb1e2db1411fc0f : SUCCESS in 42m 19s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/969fd0a15f0a4268af04218a6467c9dd : SUCCESS in 1h 54m 23s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/cda8eb34c2634a82a0b2f71215965f7d : SUCCESS in 34m 06s","accounts_in_message":[],"_revision_number":9},{"id":"28ca2bd50a88a16b7374339bcff5abc822cc84ff","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-08-25 13:38:09.000000000","message":"Patch Set 9:\n\n(2 comments)","accounts_in_message":[],"_revision_number":9},{"id":"003a702f3cd484a4ceb52f01e9d20d7fba1f5258","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2021-08-25 14:00:36.000000000","message":"Patch Set 9: Code-Review+2","accounts_in_message":[],"_revision_number":9},{"id":"2eeeddf72ee37fa6ec2333a6da3d7a3f9ff8b954","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-08-25 15:10:55.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"3aedd646c5f02da2833be673bc97eafa2915f6ff","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-08-25 15:16:16.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"e2db8cd3f855a7dc5d395b6c927f6a805d5b4355","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-08-25 15:16:56.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"fbad082f7d9335008dfa4e40a2df9e46a4dacbeb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-25 15:52:23.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"1aa9d93710d6e0356b584a49b81400704ddfb2cb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-25 17:22:04.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/0a47f0d0e2a8483fad4071b6a11d7521 : SUCCESS in 1h 01m 57s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/477d2916bb2d403197dde262c6713fa3 : SUCCESS in 1h 17m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/185eab5f094d4e0fbe5b9a3e4bc9955d : SUCCESS in 7m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/439bc6bedecc4ce0944eb9d57e0df650 : SUCCESS in 8m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/683c80b39a8e4dbaaa24a0f28ddbc829 : SUCCESS in 10m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0b5e169cbd9d4a4387b540503dcd10a8 : SUCCESS in 8m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f1f9359e4cd84aa5bdb838a894b7d30c : SUCCESS in 11m 24s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5755cdee83ab47dc8d535e2862e74250 : SUCCESS in 10m 19s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/db5871b52f64416a8dc9056e59704140 : SUCCESS in 11m 30s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/f265790d5a2f4da492e9a41a493cbb06 : SUCCESS in 10m 35s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/2b9b356e3ba747beb77b2ec5c064aa2d : SUCCESS in 6m 17s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/9ba11b85c14f4e23a6e912edbbe2440f : SUCCESS in 1h 15m 29s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/921b0f94657543e98a04e56d9b559756 : SUCCESS in 1h 05m 11s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/c3d5cdc41b6f45ab80ff33b825c959b3 : SUCCESS in 1h 11m 30s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/340514a33a574ecbaa2c174a42cef468 : SUCCESS in 1h 21m 10s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4f4e27ff795e40a3be91714a79e3ac1f : SUCCESS in 49m 26s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/7c9dc6f3b3a14b74886867bd09c40b27 : SUCCESS in 1h 12m 13s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/6dff441e733c4f0b844f44f0e3651fad : SUCCESS in 44m 22s","accounts_in_message":[],"_revision_number":10},{"id":"c98416375aacc207771f60ce90392384d55a50a8","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-08-25 17:55:43.000000000","message":"Patch Set 10: Code-Review+2","accounts_in_message":[],"_revision_number":10},{"id":"87c71c7041896269efe5fb9c5408f62c5e7f359a","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2021-08-25 18:44:03.000000000","message":"Patch Set 10: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":10},{"id":"7b3cc0f491f6bc1af77ff2b5fb709cb558058946","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-25 18:45:21.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"631e1aa0aa7a5447f7aa491e0f5bb684828cc120","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-25 22:32:59.000000000","message":"Patch Set 10: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/ae15409895c04af5bc35ce6e36c50319 : SUCCESS in 1h 07m 18s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/eb09e8375a2345d2bda80f097b882a5d : SUCCESS in 1h 20m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e93f3bb6ac549a59e878dfe2ce1a106 : SUCCESS in 6m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/771d6a01b1b747eb9b42ab5080a65d0d : SUCCESS in 7m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/af74992e2dbb45dd8681b9065b379dd4 : SUCCESS in 6m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9bf82c10b8f54d118f05d2a8dfc54f9f : SUCCESS in 8m 17s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/68ffeed0f5aa40c79b78f8415bfe5753 : SUCCESS in 11m 15s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/df0234f1bee54f8a959ec1de7fa612ed : SUCCESS in 9m 20s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/6122193aee274bf7a42f0eaf8ba64098 : SUCCESS in 1h 02m 37s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/e6ed4c2faace4980969a060ac6363238 : SUCCESS in 1h 12m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/73c612ee332b4c29acec9ab505b7c774 : SUCCESS in 57m 33s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/3296385365b9462d90421894bc4c3276 : FAILURE in 1h 52m 22s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/ab47811da9224258aef63108811c1261 : SUCCESS in 27m 01s","accounts_in_message":[],"_revision_number":10},{"id":"bb9077b6e109ba83b7f96277af6b9b2f85ebde0e","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-25 22:36:29.000000000","message":"Patch Set 10:\n\nrecheck unrelated issue","accounts_in_message":[],"_revision_number":10},{"id":"23e00a0d4512f7c64ed3fedd59530c63bac037a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-26 00:09:26.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/05c34ad820154d93a3619b6255a1f274 : SUCCESS in 1h 01m 14s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c45e9df6d3f94d6e953ec3124ef520aa : SUCCESS in 1h 15m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/475cf6a1a00249d7b4b337e971493488 : SUCCESS in 5m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c3c89748fb2b4505be2b9fd90a6f9b92 : SUCCESS in 5m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/55ea28e81d7b4ebe860c50f17e8cb261 : SUCCESS in 7m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8efadb67c83d42a99e68bb7a796c02d9 : SUCCESS in 6m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec105eac23964a8198119c44cd9efcfd : SUCCESS in 6m 52s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/06ca4a9b5b7749e7b75966a26400fb43 : SUCCESS in 8m 57s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/cb3fce0acde642838042c5aa8df941b9 : SUCCESS in 8m 46s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/cd7dd63e0dac4b2aa4058bb497d40a4b : SUCCESS in 8m 18s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/597dcefe325144a8abd74e23a93d7573 : SUCCESS in 5m 13s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/502c3db6933b45deb9ccf88513b21b04 : FAILURE in 1h 25m 39s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/07c8105cf777484693172a2211d32d93 : SUCCESS in 1h 05m 04s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/5f17e24ef944415c98b0009b2c3c4256 : SUCCESS in 1h 14m 11s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/b5f5fc6a544d43b4ace261764c34038d : SUCCESS in 1h 17m 18s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/55042c2f33fc4ff78a19c6829ee8a72a : SUCCESS in 53m 28s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/a45e8c647b954596b64e9eae8a81f16a : SUCCESS in 1h 11m 18s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/884ba5a686c84fd08f0a0155ba305ec4 : SUCCESS in 33m 06s","accounts_in_message":[],"_revision_number":10},{"id":"75d597a8375024c923e80c2ee79aba425ac1be11","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-26 00:10:09.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"0abde6dd97bcb9f18e5a879b0b413c0df1aeab53","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-26 04:45:51.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/a2b1e8d065c945d5bc5afa6aaa9d8d7e : SUCCESS in 1h 13m 54s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/13fcce1b97b14cddbd40b41ea315f644 : SUCCESS in 1h 29m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0782ae2ec38a4757902ef6badc58d7eb : SUCCESS in 5m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2ae5342c09f24a42bb61e7f2f2333dc1 : SUCCESS in 6m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fb63303ebdab483b89f7f7255ef30902 : SUCCESS in 6m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41b9fd9aca224c79a77a1c77c687b901 : SUCCESS in 6m 44s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/173e2afdc3ec4c36ad97b0a46247b07f : SUCCESS in 10m 03s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/d6787d86e8d14add8e775cc1f822b6a7 : SUCCESS in 9m 39s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/a0a2056382434804822ed3519cd4bc88 : SUCCESS in 56m 41s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/e17c6a45a2ac48bb9f3126d16c63a729 : SUCCESS in 1h 23m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bf6c04d360d340a580d20c5429d45a8e : SUCCESS in 45m 30s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/07d3eddda876435d87308ff52f63ea3d : SUCCESS in 1h 20m 40s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/a445dd4f86a94b1f82e15d8aa765d270 : SUCCESS in 26m 49s","accounts_in_message":[],"_revision_number":10},{"id":"6558bd0881ec07ace366111ceced780efb7ddfd1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-26 04:45:51.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":10},{"id":"2889c70a14efb5756801556db633402a4584eb62","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-26 04:47:28.000000000","message":"Patch Set 10:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f2710ca746c042638a8e119a489b0586 : SUCCESS in 1m 17s","accounts_in_message":[],"_revision_number":10},{"id":"77f0b97b50876294571e9d0fd6e4501e62301f61","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-08-26 13:34:32.000000000","message":"Patch Set 10:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/804588/804588/10/check-not-vote/ext-glance-zuul/3257134 : SUCCESS in 38m 25s","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"869a39d9e55849c7de7b3ee086cea1017d4f60d5","revisions":{"5f5a5e8d7e330bb1b136b15fc40e92314990ec4c":{"kind":"REWORK","_number":1,"created":"2021-08-15 17:26:22.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/88/804588/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/88/804588/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/88/804588/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/88/804588/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/88/804588/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/88/804588/1"}}},"commit":{"parents":[{"commit":"ffe9f1456bbfc50aaa4b8d8b273af307a50f0d68","subject":"Merge \"setup.cfg: Fix errant key\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ffe9f1456bbfc50aaa4b8d8b273af307a50f0d68"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-08-15 17:19:45.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-08-15 17:22:01.000000000","tz":0},"subject":"Check policies for Image Tags in API","message":"Check policies for Image Tags in API\n\nThis patch enforces policy checks required for adding/deleting tags\nfor image in API layer.\n\nPartially-Implements: blueprint policy-refactor\n\nChange-Id: I7b5bcbce9375df5b0826120df5cdf5d4cb788af2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5f5a5e8d7e330bb1b136b15fc40e92314990ec4c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5f5a5e8d7e330bb1b136b15fc40e92314990ec4c"}]},"branch":"refs/heads/master"},"c9099e270055e87929f5fdd3ca423a1bc2e54493":{"kind":"REWORK","_number":2,"created":"2021-08-20 06:43:22.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/88/804588/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/88/804588/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/88/804588/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/88/804588/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/88/804588/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/88/804588/2"}}},"commit":{"parents":[{"commit":"6ffd80f9c0b15d3c2fdf17476c1a1348ea0705ce","subject":"Check download_image policy in the API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6ffd80f9c0b15d3c2fdf17476c1a1348ea0705ce"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-08-15 17:19:45.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-08-20 06:25:36.000000000","tz":0},"subject":"Check policies for Image Tags in API","message":"Check policies for Image Tags in API\n\nThis patch enforces policy checks required for adding/deleting tags\nfor image in API layer.\n\nPartially-Implements: blueprint policy-refactor\n\nChange-Id: I7b5bcbce9375df5b0826120df5cdf5d4cb788af2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c9099e270055e87929f5fdd3ca423a1bc2e54493"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c9099e270055e87929f5fdd3ca423a1bc2e54493"}]},"branch":"refs/heads/master"},"2c1b6dae7dda2d27a98b7e2915ba3e650a82936d":{"kind":"REWORK","_number":3,"created":"2021-08-20 19:56:00.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/88/804588/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/88/804588/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/88/804588/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/88/804588/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/88/804588/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/88/804588/3"}}},"commit":{"parents":[{"commit":"6ffd80f9c0b15d3c2fdf17476c1a1348ea0705ce","subject":"Check download_image policy in the API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6ffd80f9c0b15d3c2fdf17476c1a1348ea0705ce"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-08-15 17:19:45.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-08-20 19:51:24.000000000","tz":0},"subject":"Check policies for Image Tags in API","message":"Check policies for Image Tags in API\n\nThis patch enforces policy checks required for adding/deleting tags\nfor image in API layer.\n\nPartially-Implements: blueprint policy-refactor\n\nChange-Id: I7b5bcbce9375df5b0826120df5cdf5d4cb788af2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2c1b6dae7dda2d27a98b7e2915ba3e650a82936d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2c1b6dae7dda2d27a98b7e2915ba3e650a82936d"}]},"branch":"refs/heads/master"},"f55c44a260004e771ddc0bf5718a263b16dfcc80":{"kind":"REWORK","_number":4,"created":"2021-08-23 07:15:26.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/88/804588/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/88/804588/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/88/804588/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/88/804588/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/88/804588/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/88/804588/4"}}},"commit":{"parents":[{"commit":"a269847761d793b9e8fb4f89c829c31ea7da7130","subject":"Check policies for image import operation in API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a269847761d793b9e8fb4f89c829c31ea7da7130"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-08-15 17:19:45.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-08-23 06:21:31.000000000","tz":0},"subject":"Check policies for Image Tags in API","message":"Check policies for Image Tags in API\n\nThis patch enforces policy checks required for adding/deleting tags\nfor image in API layer.\n\nPartially-Implements: blueprint policy-refactor\n\nChange-Id: I7b5bcbce9375df5b0826120df5cdf5d4cb788af2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f55c44a260004e771ddc0bf5718a263b16dfcc80"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f55c44a260004e771ddc0bf5718a263b16dfcc80"}]},"branch":"refs/heads/master"},"56d7a8e57e87f6c7cb9eb264b6093171b15c9de7":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2021-08-24 06:15:20.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/88/804588/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/88/804588/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/88/804588/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/88/804588/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/88/804588/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/88/804588/5"}}},"commit":{"parents":[{"commit":"7df1324844c7e0bf73644be7479e606a9f9653f6","subject":"Check policies for image import operation in API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7df1324844c7e0bf73644be7479e606a9f9653f6"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-08-15 17:19:45.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-08-24 06:05:36.000000000","tz":0},"subject":"Check policies for Image Tags in API","message":"Check policies for Image Tags in API\n\nThis patch enforces policy checks required for adding/deleting tags\nfor image in API layer.\n\nPartially-Implements: blueprint policy-refactor\n\nChange-Id: I7b5bcbce9375df5b0826120df5cdf5d4cb788af2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/56d7a8e57e87f6c7cb9eb264b6093171b15c9de7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/56d7a8e57e87f6c7cb9eb264b6093171b15c9de7"}]},"branch":"refs/heads/master"},"d7ae622853007fe1841eb4dbeebe49c162233656":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2021-08-24 09:36:10.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/88/804588/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/88/804588/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/88/804588/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/88/804588/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/88/804588/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/88/804588/6"}}},"commit":{"parents":[{"commit":"5821db41a009c6098d813bb43735d85548c7a2ae","subject":"Check policies for image import operation in API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5821db41a009c6098d813bb43735d85548c7a2ae"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-08-15 17:19:45.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-08-24 09:27:29.000000000","tz":0},"subject":"Check policies for Image Tags in API","message":"Check policies for Image Tags in API\n\nThis patch enforces policy checks required for adding/deleting tags\nfor image in API layer.\n\nPartially-Implements: blueprint policy-refactor\n\nChange-Id: I7b5bcbce9375df5b0826120df5cdf5d4cb788af2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d7ae622853007fe1841eb4dbeebe49c162233656"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d7ae622853007fe1841eb4dbeebe49c162233656"}]},"branch":"refs/heads/master"},"acd22a2588a348c386e6ef1c1a3ee61641050cbb":{"kind":"REWORK","_number":7,"created":"2021-08-24 18:28:15.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/88/804588/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/88/804588/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/88/804588/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/88/804588/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/88/804588/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/88/804588/7"}}},"commit":{"parents":[{"commit":"6801fa9245cf9eff59e748ecbe87569ab228233f","subject":"Check policies for image import operation in API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6801fa9245cf9eff59e748ecbe87569ab228233f"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-08-15 17:19:45.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-08-24 18:20:36.000000000","tz":0},"subject":"Check policies for Image Tags in API","message":"Check policies for Image Tags in API\n\nThis patch enforces policy checks required for adding/deleting tags\nfor image in API layer.\n\nPartially-Implements: blueprint policy-refactor\n\nChange-Id: I7b5bcbce9375df5b0826120df5cdf5d4cb788af2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/acd22a2588a348c386e6ef1c1a3ee61641050cbb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/acd22a2588a348c386e6ef1c1a3ee61641050cbb"}]},"branch":"refs/heads/master"},"abf34a17413332369817f92602d596843587c604":{"kind":"REWORK","_number":8,"created":"2021-08-24 20:17:05.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/88/804588/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/88/804588/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/88/804588/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/88/804588/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/88/804588/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/88/804588/8"}}},"commit":{"parents":[{"commit":"6c87a18d4cb43aa4af6e953af19f1655f7bb3bf9","subject":"Check policies for image import operation in API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6c87a18d4cb43aa4af6e953af19f1655f7bb3bf9"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-08-15 17:19:45.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-08-24 20:09:24.000000000","tz":0},"subject":"Check policies for Image Tags in API","message":"Check policies for Image Tags in API\n\nThis patch enforces policy checks required for adding/deleting tags\nfor image in API layer.\n\nPartially-Implements: blueprint policy-refactor\n\nChange-Id: I7b5bcbce9375df5b0826120df5cdf5d4cb788af2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/abf34a17413332369817f92602d596843587c604"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/abf34a17413332369817f92602d596843587c604"}]},"branch":"refs/heads/master"},"7ea1427d38553e168d4cfe1ec667da1f1976482b":{"kind":"NO_CHANGE","_number":9,"created":"2021-08-25 06:37:35.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/88/804588/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/88/804588/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/88/804588/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/88/804588/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/88/804588/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/88/804588/9"}}},"commit":{"parents":[{"commit":"6c87a18d4cb43aa4af6e953af19f1655f7bb3bf9","subject":"Check policies for image import operation in API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6c87a18d4cb43aa4af6e953af19f1655f7bb3bf9"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-08-15 17:19:45.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-08-25 06:27:47.000000000","tz":0},"subject":"Check policies for Image Tags in API","message":"Check policies for Image Tags in API\n\nThis patch enforces policy checks required for adding/deleting tags\nfor image in API layer.\n\nPartially-Implements: blueprint policy-refactor\n\nChange-Id: I7b5bcbce9375df5b0826120df5cdf5d4cb788af2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7ea1427d38553e168d4cfe1ec667da1f1976482b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7ea1427d38553e168d4cfe1ec667da1f1976482b"}]},"branch":"refs/heads/master"},"869a39d9e55849c7de7b3ee086cea1017d4f60d5":{"kind":"REWORK","_number":10,"created":"2021-08-25 15:52:23.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/88/804588/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/88/804588/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/88/804588/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/88/804588/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/88/804588/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/88/804588/10"}}},"commit":{"parents":[{"commit":"f094b7d2b2ac188f8d047736d51a86f82cd27410","subject":"Merge \"tests: Remove use of \u0027oslo_db.sqlalchemy.test_base\u0027\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f094b7d2b2ac188f8d047736d51a86f82cd27410"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-08-15 17:19:45.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-08-25 15:38:24.000000000","tz":0},"subject":"Check policies for Image Tags in API","message":"Check policies for Image Tags in API\n\nThis patch enforces policy checks required for adding/deleting tags\nfor image in API layer.\n\nPartially-Implements: blueprint policy-refactor\n\nChange-Id: I7b5bcbce9375df5b0826120df5cdf5d4cb788af2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/869a39d9e55849c7de7b3ee086cea1017d4f60d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/869a39d9e55849c7de7b3ee086cea1017d4f60d5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
