)]}'
{"id":"openstack%2Fglance~804590","triplet_id":"openstack%2Fglance~master~I18a5187d80bf76c0dc6f22dd8c96a8ffa0f46dc1","project":"openstack/glance","branch":"master","topic":"policy-refactor","hashtags":[],"change_id":"I18a5187d80bf76c0dc6f22dd8c96a8ffa0f46dc1","subject":"Check policies for image import operation in API","status":"MERGED","created":"2021-08-15 20:08:42.000000000","updated":"2021-08-25 06:24:54.000000000","submitted":"2021-08-25 06:20:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":17,"unresolved_comment_count":5,"has_review_started":true,"submission_id":"804590-policy-refactor","meta_rev_id":"0181f91bd4280ce7edd68ba1338bbef168394a63","_number":804590,"virtual_id_number":804590,"owner":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-25 06:20:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"all":[{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2021-08-25 06:24:54.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":2,"date":"2021-08-25 06:24:54.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":1,"date":"2021-08-25 04:48:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]}],"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-15 21:27:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-16 19:21:28.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2021-08-19 19:43:13.000000000","updated_by":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"reviewer":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"state":"REVIEWER"},{"updated":"2021-08-20 13:17:18.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-08-24 16:15:23.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"}],"messages":[{"id":"8eea1c58d49f65b092ca290a609bd66754e6b390","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-15 20:08:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1b1d82c808534582c055001197ccb7394ad16cf0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-15 21:27:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/e0c3ed762d504f6ea7e8300eb1b9b08e : SUCCESS in 1h 02m 01s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/995dbe0f36924a51beec7ac4f44bed90 : SUCCESS in 1h 00m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a569ecee3749463891eb5012e6e82c48 : SUCCESS in 5m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6ccbbadb2b86454f8ced673263ce49b2 : SUCCESS in 5m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b5b52e4bbffe4dd89d168e3158631ee8 : SUCCESS in 6m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a82f67b08fea4dce948953cb4a7e649b : SUCCESS in 6m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/488667128f85493994b6cc6d58a3deb8 : SUCCESS in 7m 23s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/728f753b70f741209c18cfe18783561f : SUCCESS in 8m 34s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/5a8077f5d6b4402ab25656fd7648ca32 : SUCCESS in 9m 16s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/136cddb3144e48b5b79f38545aa8f792 : SUCCESS in 7m 50s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/621813f87b2849df8cd030bcfff16ae4 : SUCCESS in 5m 26s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/f1b953a0944f47e49a6c79e7b0c6bd4c : SUCCESS in 54m 43s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/d7af452cae094780b1720bdc8543c50c : SUCCESS in 59m 54s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/2aadd1d190e94bebaf42261a31390775 : SUCCESS in 1h 00m 31s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/19944577770645688a98526beb6b4ce7 : SUCCESS in 1h 15m 30s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/567b32f889cc4a899f23e7f8918a2e22 : SUCCESS in 1h 00m 22s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/9bc37547684a4c5a95cdcfde7c8c728f : SUCCESS in 1h 16m 53s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/82fac57883bd4ce59faf211df3c83525 : SUCCESS in 30m 05s","accounts_in_message":[],"_revision_number":1},{"id":"eb440e752cc770bb5f39cb8e04adac00b1f7d65d","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2021-08-16 19:21:28.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c80a94f44ab5990f31e4acd0ae41d5e744aaa9a0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-19 17:21:56.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8c7bf807874f0a0a340c0341246fff7952ff1b10","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-19 17:23:20.000000000","message":"Topic import-policy-refactor removed","accounts_in_message":[],"_revision_number":2},{"id":"e91c15874b4d975c58cfe162774b98c05ec9dc2e","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-19 17:23:27.000000000","message":"Topic set to policy-refactor","accounts_in_message":[],"_revision_number":2},{"id":"1759786b622733499a497c20a23ec3925ba674aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-19 18:48:46.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/ce8955f587eb47efacd983e7713b0488 : SUCCESS in 1h 06m 47s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f823e2003d504183878783056905b3cb : SUCCESS in 1h 16m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d2e2201a1dce43bd87286c38ea5a9bf6 : SUCCESS in 6m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fb438fb00edf404f97333b582c354949 : SUCCESS in 7m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4d7c5c5aa97c40eca49c255e71e357dc : SUCCESS in 6m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bfd21e8c5a804ea4b3ea40a8da64c68d : SUCCESS in 8m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76d7a52b79974714b7469bb9c64df177 : SUCCESS in 7m 56s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d3fcb3da626b4473b392c452d22f45fe : SUCCESS in 11m 39s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/2638d731fcb5458aa2ed79b5e83944ff : SUCCESS in 11m 29s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/54ef9e2f3e6b4f038fc7516360c70982 : SUCCESS in 8m 38s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/e47551435b444bca81d41f9973d9e175 : SUCCESS in 5m 35s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/4db234664b1844488c76bf8f1866c204 : SUCCESS in 1h 08m 44s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/98443a44fef14e128b527657fed66d30 : SUCCESS in 1h 05m 38s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/c2c1c286924c404abc74e45123df085d : SUCCESS in 1h 01m 14s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/27db1ac35d2348e6a64621c25596573a : SUCCESS in 1h 19m 19s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fabd7ca9c9fe42d49f45c28277269a04 : SUCCESS in 49m 33s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/995dc5e8bd004f6cb4864a08a4ed0347 : SUCCESS in 1h 10m 55s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/d01f950c0efb4c0bbacec41fcaa43eac : SUCCESS in 33m 04s","accounts_in_message":[],"_revision_number":2},{"id":"348fc7553553b74e3fc46a6b7ed32034ccb3d22b","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-08-19 19:43:13.000000000","message":"Patch Set 2: Code-Review+1\n\n(2 comments)\n\nMinor comments inline, otherwise I think this looks good.","accounts_in_message":[],"_revision_number":2},{"id":"9c7a481b27d2deb99217faaacb453defd1a5eec3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-19 19:53:39.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"bb847be7d465a4f7bf94c39bf8da0bac2c284679","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-19 19:54:24.000000000","message":"Patch Set 3:\n\n(2 comments)\n\nThank you lance, nice catch!!","accounts_in_message":[],"_revision_number":3},{"id":"fd1d72ca2f7fd0cb743e02f6a80d93e29c9cc72d","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-08-19 20:34:08.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"dfaa10e53284224d3f249ba1cee32db5a32cbc51","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-19 20:37:18.000000000","message":"Topic import-policy-refactor removed","accounts_in_message":[],"_revision_number":3},{"id":"a3c45076d45c545058bbbe00c5da8fdf828df088","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-19 20:37:27.000000000","message":"Topic set to policy-refactor","accounts_in_message":[],"_revision_number":3},{"id":"7d3be7a8af05503b3d2e145c481763d565c2d493","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-19 21:28:31.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/fb14beca235c42a0bc22a9a5d79a24aa : SUCCESS in 1h 16m 53s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/29b0bbe619a34070a6be46520cc16835 : SUCCESS in 1h 04m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/be753c5272b541e9a057a95a0a014366 : SUCCESS in 5m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8d7876358d194745aa4912dd441d003e : SUCCESS in 7m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/46413d2976eb44fdb3dcc3ad94154dd6 : SUCCESS in 5m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9a9c2e0a26d742ba885acd98e42ce418 : SUCCESS in 6m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e47909997d04410096ff65c26b4b0b22 : SUCCESS in 6m 27s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ad09ae398b3b412abf053e69b5e2784f : SUCCESS in 9m 16s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/9cfe57e411c44cc4bc445216688f0c84 : SUCCESS in 8m 23s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/9708fb529b254b50846d3cbb0ee880c9 : SUCCESS in 8m 24s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/69514535ce76442ea7ebe4c99b0f3ba6 : SUCCESS in 5m 16s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/a7a544e209294928828a5d6cc7442457 : SUCCESS in 1h 09m 22s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/0ffde37e549943859724363078e29a0b : SUCCESS in 1h 11m 30s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/fc031604c28a439280c3d4f243ce0bef : SUCCESS in 1h 17m 16s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/a08c0aaf30684eff8b11aacf777c4a08 : SUCCESS in 1h 29m 57s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/33344f00236d4a4081d837d0ca8b4d17 : SUCCESS in 50m 59s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/b1b1ab92777e48388839fbe9ede05f63 : SUCCESS in 1h 26m 12s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/278a22410519439cb8a8fd448654e028 : SUCCESS in 33m 24s","accounts_in_message":[],"_revision_number":3},{"id":"cf2ca47e26c9fcf93758ea6c6faff6c4808d7ab3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-20 06:13:29.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"e8879cb3bdfa13eee975c4321b2db8684d03b629","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-20 08:25:20.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/485376adfb5c48b59586303e07621a7f : FAILURE in 53m 15s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/88af48f19fa44660a9fde5b6639a50b2 : FAILURE in 15m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2748caeba39040af8ae04b36d490d883 : FAILURE in 6m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1f68a664543a472d8aae5464b864d97e : FAILURE in 6m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d113c65a48204969aa3f28f23817207e : FAILURE in 7m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5ee49cfc5dfa4665b546947047f1e568 : FAILURE in 8m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c6e693ac154f44e29f92183f5a386288 : SUCCESS in 9m 42s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a1c0e405f2054a85ab625268f12c1b91 : FAILURE in 5m 46s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/85251bb7fb6c4953912e270fd242957b : FAILURE in 6m 37s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/968b98e3c3bd4b21b9015b2ba129b129 : SUCCESS in 9m 37s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/8c5e5fd920e84403b7dfd6f44503e7d9 : SUCCESS in 6m 41s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/ab19fbc43899431b9f90243ae3b36f7a : SUCCESS in 1h 19m 02s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/e1f0c6288ccf4880b2d4d7722e43514e : SUCCESS in 1h 08m 45s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/47f206d4724f4ebe9f055fc2d5e10fdf : SUCCESS in 1h 11m 36s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/e37da64f324f4df4861b584c2de907b7 : SUCCESS in 1h 21m 55s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/55ddd96cee184d9e8bd14f94bd39dd59 : SUCCESS in 50m 27s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/1704b937a09a4150bf4a0aa986cd4784 : SUCCESS in 1h 20m 51s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/ae091182c48b4493a591cba6521a2116 : SUCCESS in 34m 31s","accounts_in_message":[],"_revision_number":4},{"id":"29f9f2b50df5164ce1e1c03ee92d8f014b26375e","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-08-20 13:17:18.000000000","message":"Patch Set 3:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/804590/804590/3/check-not-vote/ext-glance-zuul/7a6b5cc : FAILURE in 16m 01s","accounts_in_message":[],"_revision_number":3},{"id":"adcac70fdea1071c162888ec2e893fe3290d5767","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-08-20 16:10:39.000000000","message":"Patch Set 4: Code-Review+1\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"bac1a3ee84d959b70e8fa94bae95628e55f0fc88","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-20 17:57:59.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/b8e2c26b105c4d6892448aa8ec2501cc : SUCCESS in 1h 05m 13s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b079e8701ac24ec4adc713c7a570e276 : SUCCESS in 1h 17m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/592a92a9f78b4e3a99a306e9fb8c3bb2 : SUCCESS in 7m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1ee2954d68074cb5b0c0e78b736ad935 : SUCCESS in 6m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fc333e09e79e48da806f5c132a5a4d9f : SUCCESS in 7m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3af4791fc8ed4279add00c43662d2031 : SUCCESS in 7m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1272588498a9495580e1631db27af1ab : SUCCESS in 7m 30s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0ef301897802492b88c66980dd468fca : SUCCESS in 8m 59s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/9778e88c374643899847fa77d911d2a8 : SUCCESS in 10m 45s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/60f255802f5a45c08ab60e63fbbe2f74 : SUCCESS in 8m 59s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/2e371b71a8484c609b35fd42eedc6c44 : SUCCESS in 6m 57s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/359d29c9dde34d05a31322d067e69eb2 : SUCCESS in 1h 12m 20s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/9ef13668fe144690bec1d5729336b278 : SUCCESS in 1h 05m 46s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/61aa243da224420bb087a9cbf77fc97c : SUCCESS in 1h 13m 30s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/374fb641a0cb449ca6258d158377b3a2 : SUCCESS in 1h 36m 19s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/71b056763d2749168e95adef68ff4df3 : SUCCESS in 50m 10s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/59eca0d71c3f47e9b7a868dcda377e30 : SUCCESS in 1h 09m 14s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/0cb084ad0d214e33bb2dd5fe322643d8 : SUCCESS in 31m 18s","accounts_in_message":[],"_revision_number":4},{"id":"b4ffb42811c49e97159b85c3212999fb3cf7027c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-23 06:20:52.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"4c11addca51ba4f979c7f4515abd0c55320f5eee","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-23 06:22:19.000000000","message":"Topic import-policy-refactor removed","accounts_in_message":[],"_revision_number":5},{"id":"e8e37f27e0fc9292ed29f88f7621054f90b96a28","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-23 06:22:26.000000000","message":"Topic set to policy-refactor","accounts_in_message":[],"_revision_number":5},{"id":"506d15d18e2d0907bd7b66dae9ef7940e39d164d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-23 08:24:31.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/9bf1033eec4e4ad9b4f858e5b68151aa : SUCCESS in 55m 35s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/abbc3ec02dad4005aafb13578a827279 : SUCCESS in 1h 19m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd52d780626e44b7bb5fff660fcf89b4 : SUCCESS in 7m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/56877b8a61e74d56b2f2593f1585c81f : SUCCESS in 8m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1694a5c8a78a450093e45983c61a368a : SUCCESS in 8m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c5b8c58603154e6eaca555a5b6c4b2c2 : SUCCESS in 8m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6c7945ea296428eac692d24c139dd90 : SUCCESS in 8m 36s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/826e6f637e3d4a2d8943bd7c44362a06 : SUCCESS in 10m 18s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/8d4676da96344a5aa10dd3f134e92867 : SUCCESS in 9m 13s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/fb8a820dbf024be79ed7f486e239868d : SUCCESS in 10m 13s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/e174f1884d6042f4995f91fcaee9f971 : SUCCESS in 7m 07s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/4c4123adf0ba4b778d26972d2f58d949 : SUCCESS in 1h 16m 19s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/c3ba0cbc418d4e3b89ca107cb8dec744 : SUCCESS in 1h 04m 34s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/e658104524914b4e84ac1575f966671d : SUCCESS in 1h 04m 03s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/bba8bfec191142d78cd2ce89ed227260 : SUCCESS in 1h 06m 24s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3e4c447e0b6f4dad8edebb6307efcb08 : SUCCESS in 43m 02s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/83aeee5eb3cf4f86bae9e1d90e38d017 : SUCCESS in 1h 31m 38s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/86c8f063cffd424692cf7976923ed82a : SUCCESS in 31m 08s","accounts_in_message":[],"_revision_number":5},{"id":"b440e53b66abee1e4d4f784399a1dc6be257e61e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-24 06:15:20.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"427bdcbb45fb6e2227762e77064ceded68742859","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-24 08:22:15.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/f30b95c64a6d4d4f838b4559393f6e78 : SUCCESS in 1h 06m 08s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/19ec3189b8c84db881ab5430a66bcc79 : SUCCESS in 1h 00m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f6d92ff8cb44431af2083a2f4aa1486 : SUCCESS in 6m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2e6c0370138f458cb55bf33a5f6480e0 : SUCCESS in 6m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/92089a44e8df4d8db3212e2fe1e13f26 : SUCCESS in 9m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/10a694a8f4e44358a205d5b39b52a240 : SUCCESS in 7m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b786e5f8d22d4332a2fd7abafd6faf9b : SUCCESS in 7m 47s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b86bf4666ed3476eb64fc495d5450613 : SUCCESS in 11m 05s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/467650bb787445579e119bd09657a6fd : SUCCESS in 12m 03s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/cc8dfaae98c14a2f8bf784b2b9f5b2b1 : SUCCESS in 10m 16s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/07e810a9ecd74ddda0ed95c2991ea423 : SUCCESS in 7m 39s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/f6b1af7e229d472fae1e803249ddb83d : SUCCESS in 1h 08m 32s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/0a0236f99bcd42279e7866e5287647e6 : SUCCESS in 1h 11m 55s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/87d9a657ea714064b9cb35f7aab46fb7 : SUCCESS in 1h 08m 58s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/bc015d8e21564bd78f2bfbaa87fdfbd8 : SUCCESS in 1h 24m 09s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/95ccf99c1f6e41199e35dc2d51975dad : SUCCESS in 53m 04s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/ad04fbc3d1474b1e9614e35bc6015851 : SUCCESS in 1h 17m 17s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/3b99bd141d72490dacbe0e2a572f7d35 : SUCCESS in 33m 43s","accounts_in_message":[],"_revision_number":6},{"id":"fc28e19874240559b19e10653fbc34e42967b171","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-08-24 08:48:27.000000000","message":"Patch Set 6:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/804590/804590/6/check-not-vote/ext-glance-zuul/84112d1 : SUCCESS in 37m 05s","accounts_in_message":[],"_revision_number":6},{"id":"e34f084f030305d1417999a09357f03b511ff203","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-24 09:36:10.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"5dd7c4243bd83b3e2de0715c40cc8153883639db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-24 11:30:36.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/57026360fd3b40a795cb895f212dfbb5 : SUCCESS in 1h 02m 33s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/eb6480bf9a794e2badfc9631a0e5c6c5 : SUCCESS in 1h 12m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a3dab57313094c98aa5455dfc306b26a : SUCCESS in 7m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5f5567ba9dfe4a01b3f098324355f5fc : SUCCESS in 7m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/18ab8801c2784217addf2c2521a528df : SUCCESS in 9m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fb8f7a90bf9f4022ad040c39ea8e9d03 : SUCCESS in 9m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/881410b4d864476f8dc347879930992d : SUCCESS in 8m 50s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/74c8e792c6c543cb8e7131f5d422292f : SUCCESS in 11m 35s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/1cfa99407b864a6c8abbe03a39e1f218 : SUCCESS in 10m 35s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/bdfb1aa305c04edfa608adbedba36718 : SUCCESS in 11m 30s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/47411bb648524695979e99571a44eb2c : SUCCESS in 7m 19s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/ef8fe06429084f4289e0a3c31843f3db : SUCCESS in 1h 17m 53s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/8f063d6ae0a84fcb8d7aeae32350630b : SUCCESS in 1h 08m 23s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/7f0a31bf2d414f00a61365116dec51a2 : SUCCESS in 1h 03m 42s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/415030b53c524beba63aa9a6072e30bd : SUCCESS in 1h 16m 19s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f210be18ed714708987ccadbc24105f9 : SUCCESS in 47m 42s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/fa80ddc5bb1e4d9781bce844231e43ea : SUCCESS in 1h 26m 49s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/f21e22777cfd42b6bedf3b966df3b6dc : SUCCESS in 34m 10s","accounts_in_message":[],"_revision_number":7},{"id":"31dc0c27d6202661e0cc5147ceda10c5c66ddcd3","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-08-24 12:45:23.000000000","message":"Patch Set 7:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/804590/804590/7/check-not-vote/ext-glance-zuul/0e21743 : SUCCESS in 36m 11s","accounts_in_message":[],"_revision_number":7},{"id":"c199606dfcaf360f88f88018fd8f0906acaecb66","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-08-24 14:00:46.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"ccbe80153cef76fb049dc093cc9c409f802e441e","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-24 14:57:59.000000000","message":"Topic policy-reafactor removed","accounts_in_message":[],"_revision_number":7},{"id":"695f3d523ca949c592b9286c479f33ade083f3d6","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-24 14:58:12.000000000","message":"Topic set to policy-refactor","accounts_in_message":[],"_revision_number":7},{"id":"4093d6e4efb08c79ec5b02a34b4abc337e2e6e45","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2021-08-24 15:46:13.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"824f031156fdd680b3210ed4a47d72e78a082a2d","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-08-24 16:15:23.000000000","message":"Patch Set 7: Code-Review-1\n\n(5 comments)\n\nI think we\u0027re missing a test case and maybe testing something we don\u0027t really intend to.","accounts_in_message":[],"_revision_number":7},{"id":"1834af8d048283f861034506e1f4262d16bd900a","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-24 16:45:36.000000000","message":"Patch Set 7:\n\n(5 comments)","accounts_in_message":[],"_revision_number":7},{"id":"46b432864bfb708769643f88e8c9e150a19248b6","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-08-24 17:27:43.000000000","message":"Patch Set 7:\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"ada72adcd0b5a0672a7c6d7ad27e2308d2c59118","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-24 18:28:15.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"260d91295e7922f7db16c079870eadcf767960d0","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-08-24 19:04:26.000000000","message":"Patch Set 8:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/804590/804590/8/check-not-vote/ext-glance-zuul/76ab33c : SUCCESS in 35m 27s","accounts_in_message":[],"_revision_number":8},{"id":"56f07582c3204e13e43119713f5b79fe7ec24e51","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-08-24 19:05:32.000000000","message":"Patch Set 8: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"989fa0ce040004d0b0ca4c76a3531cd53af3443f","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2021-08-24 19:40:18.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"c221358786393739f59d4e1fd4cf55b0d93807ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-24 19:44:18.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/25cf517760cb4a40834c2bc3c295c123 : SUCCESS in 59m 27s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5bc963eb1fbc4d74a726a3d84df00b8d : SUCCESS in 1h 08m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed02ac2f522641ab911212f975279cd0 : SUCCESS in 6m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5f72b257d2ca4cb1a6f5c9b92a003b4a : SUCCESS in 7m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1a3b63cdf55545a69e794036b4dcdb3b : SUCCESS in 6m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/120e611a8c7d43f5902d7e912f50b78a : SUCCESS in 8m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6c83514c6af42f799789ebe16e42308 : SUCCESS in 7m 42s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e4c3595ce0ad47b792bf6373aeea9953 : SUCCESS in 10m 16s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/887e0f788fc442d693a85752470b5008 : SUCCESS in 10m 27s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/9e5ac22e35a04d0eb1f65185eaa69bd5 : TIMED_OUT in 31m 32s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/150f928577474cada44acb4ec69af8de : SUCCESS in 6m 05s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/01c5b860a9b14f4fbfac8f7cf42c95ac : SUCCESS in 1h 12m 55s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/6ed10acafb524d489a964d026ccef75e : SUCCESS in 1h 09m 20s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/e64505999c85419daa014e1b23a59fd4 : SUCCESS in 1h 11m 20s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/8353b875a5d644dda00fc5b1116ba050 : FAILURE in 1h 00m 15s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/041a2d81fbd8400abe57dfb2a54245e0 : SUCCESS in 1h 09m 44s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/aae33f6cdea04d1a8de398478e1450ca : SUCCESS in 1h 05m 33s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/e32f6bdfdb9a44c0b1b2531ab3330004 : SUCCESS in 28m 28s","accounts_in_message":[],"_revision_number":8},{"id":"7327bade714fed5e4ffe0379db1646e3246943e6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-24 19:45:02.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"fbe27a31e038fd91cfea05b449403e3074dde828","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-24 20:17:05.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.","accounts_in_message":[],"_revision_number":9},{"id":"09d8aab242840794c669bdf9afff0579132a8729","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-24 21:41:41.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/7e578563071743e89336e287c3d242cb : SUCCESS in 1h 10m 34s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6216d162f2434321aec9b3f7a5bd35ca : SUCCESS in 1h 17m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea1cdd80115d47de825245f9d7ac9f23 : SUCCESS in 5m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ddcfb10950ca4fcd98d2185d2ecbd63f : SUCCESS in 5m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d4d85dfc68a644899cfead38ea27e30b : SUCCESS in 6m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0c09ea36f066406ba53453c8552caaa8 : SUCCESS in 6m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ec7edb144064ce7896c6a6b459a152c : SUCCESS in 6m 49s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a44aa9d073064033abe0a7f3b21e82bc : SUCCESS in 8m 56s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/4eb057e50c1f40279fc435edfc8fdfba : TIMED_OUT in 31m 18s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/4bedc88fe768442598478481348a60b3 : SUCCESS in 9m 13s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/df1b1ed9093941ff8e55e09e9b0f09eb : SUCCESS in 5m 34s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/efd53988780a490493fb23d02f42abcc : SUCCESS in 1h 14m 14s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/b91785cad8f84639b67b3bdca8319cdb : SUCCESS in 1h 03m 56s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/1ef6424be6c64949b10018a0b92170e9 : SUCCESS in 1h 14m 40s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/7670fb8e75d64fcb9567af184296157d : SUCCESS in 1h 16m 34s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7b8e2db194c14c5bb394ff53ec636bbd : SUCCESS in 48m 07s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/20785df250124ac69f67acd65f5d320a : SUCCESS in 1h 13m 32s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/3cd8541ab5424e34b986c170f027f996 : SUCCESS in 25m 15s","accounts_in_message":[],"_revision_number":9},{"id":"6270fbcccccb6d12a9e5166d83d79f6db3aed460","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-24 21:42:49.000000000","message":"Patch Set 9:\n\nrecheck timeout","accounts_in_message":[],"_revision_number":9},{"id":"74fc81668319a36a129b20f2b65ab9ace0e9648c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-24 23:01:33.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/cd1fe1bfb1e041a2a522f518d3608b6e : SUCCESS in 56m 27s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f5536c975aff4b2f8f19a883302c4f19 : SUCCESS in 1h 11m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/57329c8f37094b35aa097a20ff0e4929 : SUCCESS in 5m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d6f021d6b03d469a92967756b72151b6 : SUCCESS in 6m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a90bd32d07564e0bb419c538692de081 : SUCCESS in 8m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ce88b226768f4bafb5a0e62414ab4132 : SUCCESS in 6m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ccd88d0c349419d85b8b95665352f91 : SUCCESS in 6m 55s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/97727d0e77134db0acdbb38b0036a298 : SUCCESS in 10m 16s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/8ca027d6e15a4112978984f6cbf62fe3 : TIMED_OUT in 31m 21s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/f84d976b0d5c4a7e8c03903cf6d3ced5 : TIMED_OUT in 31m 23s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/89a22aece5ec4d578d2e9a8be49e57fc : SUCCESS in 5m 50s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/8c8e58e80b464f2db77daf5c43a4a123 : SUCCESS in 1h 16m 30s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/58d962f0e8864086876b0c60e152b91e : SUCCESS in 59m 32s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/32dc14c82a4440a884800e1b01c072a4 : SUCCESS in 1h 02m 23s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/e0a018a94a184783852daa8ed567e144 : SUCCESS in 1h 15m 01s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d20888f20959453bb87ec6edd8aaebe5 : SUCCESS in 57m 52s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/0310fd7687004e569ae5c4bb152f53e7 : SUCCESS in 1h 14m 13s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/02f52fcb384b4298bdc37c12893d7a88 : SUCCESS in 24m 40s","accounts_in_message":[],"_revision_number":9},{"id":"086eeaf0e9d221ccd4e0f7a1801672e8bb9f12d6","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-08-24 23:45:24.000000000","message":"Patch Set 9:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/804590/804590/9/check-not-vote/ext-glance-zuul/cdd72a3 : SUCCESS in 37m 55s","accounts_in_message":[],"_revision_number":9},{"id":"546fbc12556884450db980cb4c0c13231c0c4503","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-25 02:48:03.000000000","message":"Patch Set 9:\n\nrecheck timeout","accounts_in_message":[],"_revision_number":9},{"id":"efa39d03eee614ec5c2af1b7606c9d528e54e2e3","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-08-25 03:24:10.000000000","message":"Patch Set 9:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/804590/804590/9/check-not-vote/ext-glance-zuul/bc0541c : SUCCESS in 35m 34s","accounts_in_message":[],"_revision_number":9},{"id":"088debb5e5ed3cbdc4ea823192c0eadf7b4021c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-25 04:34:58.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/1ded3fae89cb4e169d7070cd0526e99f : SUCCESS in 1h 43m 44s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c6682406e8834923bc37ad9bda3cf48d : SUCCESS in 1h 07m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d30c4c53ac7f479891f6aca2adec6234 : SUCCESS in 5m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3ac88aabfe4c431db3ed67543e0be436 : SUCCESS in 5m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/547536baeb6b467588bf0ab3f716a3a9 : SUCCESS in 5m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d81585df276a4aaa98be9c7db11b7d11 : SUCCESS in 6m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/daa4db71bfe44c21bced51befe0cddb8 : SUCCESS in 7m 09s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/120ebfe5b2764843b16173d42d766758 : SUCCESS in 8m 49s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/01d3f81dd1e5491782a13bca5548dd6a : SUCCESS in 8m 11s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/aac00fa29e66496b92244a783ddad4b5 : TIMED_OUT in 31m 07s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/f482b91c6d9c4c739f4d4f7f583552f4 : SUCCESS in 5m 37s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/9d09ca545a3a4603b4aa8dc5debece8e : SUCCESS in 1h 14m 19s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/e18ee43cb637462da0e793f0012110fd : SUCCESS in 1h 03m 09s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/0cbb5be7e7de4f36973467a08b7ab4bd : SUCCESS in 1h 06m 37s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/93336908e6fe41c48ef1e24acea597a0 : SUCCESS in 1h 07m 40s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8f8b18a27e1b4038b6f7799c8b614a48 : SUCCESS in 45m 51s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/e34007da718e4fa098b3c6e107dcb58b : SUCCESS in 1h 10m 19s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/9f08175ba79543088549d38d4b210e04 : SUCCESS in 41m 18s","accounts_in_message":[],"_revision_number":9},{"id":"f78e2dba3293eb99835b90a46039d1cf976102ae","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-25 04:48:13.000000000","message":"Patch Set 9: Workflow+1\n\nRestoring workflow as it was already approved but rebase happen in between.","accounts_in_message":[],"_revision_number":9},{"id":"d7347f1ec702c4a912f01103508b511bb9f90ea8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-25 04:48:57.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"d560fb43e6885159bb64e80dd34406d5cb1e8c57","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-25 06:20:44.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/91559bacba4d45de9ae2c6db94cf113e : SUCCESS in 53m 45s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/fac9b5afecec4c54ba6e67a052798848 : SUCCESS in 56m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c0d8d85d68554cd0bbb08b0ea72128ff : SUCCESS in 6m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cd13870e7a6840e4b8f5421b2c85de36 : SUCCESS in 6m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cec4158d38124faca6146aa73ede4a9a : SUCCESS in 7m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/787e97af2e7348399e2c13b9e324ede2 : SUCCESS in 7m 32s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7019a4f5191c415b98743918e365165a : SUCCESS in 9m 05s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/45ebfebe200a46f3bbe8361f93c7724d : SUCCESS in 10m 14s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/9eca6f84c2a44d05a64b6d4ba4a36716 : SUCCESS in 1h 09m 31s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/50745c8a0404452ba8913d134afb657d : SUCCESS in 1h 14m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dc54ac5a1c9d4292bb57559353a8615c : SUCCESS in 37m 32s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/6ba7cc876bbc485abc1c84e440f4cb07 : SUCCESS in 1h 12m 02s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/339d27f8726448ca83b82fecd191b7de : SUCCESS in 23m 55s","accounts_in_message":[],"_revision_number":9},{"id":"e4eee584901c8375deb0b124ae4ddf8b850b51d8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-25 06:20:44.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":9},{"id":"4fc5e802ecd6598ea5891ef9fa1ef38cad9f21eb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-25 06:24:54.000000000","message":"Patch Set 9:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab653532573d4cd0b3b5537e7595a39e : SUCCESS in 3m 32s","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"6c87a18d4cb43aa4af6e953af19f1655f7bb3bf9","revisions":{"208e1700691d5465ea8f0f325f0891bc80bcddbb":{"kind":"REWORK","_number":1,"created":"2021-08-15 20:08:42.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/90/804590/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/90/804590/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/90/804590/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/90/804590/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/90/804590/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/90/804590/1"}}},"commit":{"parents":[{"commit":"d035553871b351234029ef1850eccacd859a5deb","subject":"Refactor gateway auth layer for task APIs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d035553871b351234029ef1850eccacd859a5deb"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-08-15 20:01:54.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-08-15 20:03:50.000000000","tz":0},"subject":"Check policies for image import operation in API","message":"Check policies for image import operation in API\n\nThis patch enforces policy checks required for importing/copying image\ndata to store in API layer.\n\nPartially-Implements: blueprint policy-refactor\n\nChange-Id: I18a5187d80bf76c0dc6f22dd8c96a8ffa0f46dc1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/208e1700691d5465ea8f0f325f0891bc80bcddbb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/208e1700691d5465ea8f0f325f0891bc80bcddbb"}]},"branch":"refs/heads/master"},"3351fdc2bb5ec9d920402da1c7e9b4b8d8e7e43d":{"kind":"REWORK","_number":2,"created":"2021-08-19 17:21:56.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/90/804590/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/90/804590/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/90/804590/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/90/804590/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/90/804590/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/90/804590/2"}}},"commit":{"parents":[{"commit":"621457823a67dfae424f23e924e18a97cb3d23bb","subject":"Move Tasks policy checks in the API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/621457823a67dfae424f23e924e18a97cb3d23bb"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-08-15 20:01:54.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-08-19 17:07:28.000000000","tz":0},"subject":"Check policies for image import operation in API","message":"Check policies for image import operation in API\n\nThis patch enforces policy checks required for importing/copying image\ndata to store in API layer.\n\nPartially-Implements: blueprint policy-refactor\n\nChange-Id: I18a5187d80bf76c0dc6f22dd8c96a8ffa0f46dc1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3351fdc2bb5ec9d920402da1c7e9b4b8d8e7e43d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3351fdc2bb5ec9d920402da1c7e9b4b8d8e7e43d"}]},"branch":"refs/heads/master"},"3fc9485ca52deee95a923f4d51e466aebb3c597a":{"kind":"REWORK","_number":3,"created":"2021-08-19 19:53:39.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/90/804590/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/90/804590/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/90/804590/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/90/804590/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/90/804590/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/90/804590/3"}}},"commit":{"parents":[{"commit":"621457823a67dfae424f23e924e18a97cb3d23bb","subject":"Move Tasks policy checks in the API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/621457823a67dfae424f23e924e18a97cb3d23bb"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-08-15 20:01:54.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-08-19 19:48:47.000000000","tz":0},"subject":"Check policies for image import operation in API","message":"Check policies for image import operation in API\n\nThis patch enforces policy checks required for importing/copying image\ndata to store in API layer.\n\nPartially-Implements: blueprint policy-refactor\n\nChange-Id: I18a5187d80bf76c0dc6f22dd8c96a8ffa0f46dc1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3fc9485ca52deee95a923f4d51e466aebb3c597a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3fc9485ca52deee95a923f4d51e466aebb3c597a"}]},"branch":"refs/heads/master"},"7713b88c2d962f9925a91cb97cfb87284c73a1ac":{"kind":"REWORK","_number":4,"created":"2021-08-20 06:13:29.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/90/804590/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/90/804590/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/90/804590/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/90/804590/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/90/804590/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/90/804590/4"}}},"commit":{"parents":[{"commit":"6e9ee8cbc1e932ff37e361caf9727ca140b4e15f","subject":"Move Tasks policy checks in the API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6e9ee8cbc1e932ff37e361caf9727ca140b4e15f"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-08-15 20:01:54.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-08-20 06:04:56.000000000","tz":0},"subject":"Check policies for image import operation in API","message":"Check policies for image import operation in API\n\nThis patch enforces policy checks required for importing/copying image\ndata to store in API layer.\n\nPartially-Implements: blueprint policy-refactor\n\nChange-Id: I18a5187d80bf76c0dc6f22dd8c96a8ffa0f46dc1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7713b88c2d962f9925a91cb97cfb87284c73a1ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7713b88c2d962f9925a91cb97cfb87284c73a1ac"}]},"branch":"refs/heads/master"},"a269847761d793b9e8fb4f89c829c31ea7da7130":{"kind":"REWORK","_number":5,"created":"2021-08-23 06:20:52.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/90/804590/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/90/804590/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/90/804590/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/90/804590/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/90/804590/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/90/804590/5"}}},"commit":{"parents":[{"commit":"23b653c10c2858a4820036554f8382ae73f1f39e","subject":"Move Tasks policy checks in the API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/23b653c10c2858a4820036554f8382ae73f1f39e"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-08-15 20:01:54.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-08-23 06:05:21.000000000","tz":0},"subject":"Check policies for image import operation in API","message":"Check policies for image import operation in API\n\nThis patch enforces policy checks required for importing/copying image\ndata to store in API layer.\n\nPartially-Implements: blueprint policy-refactor\n\nChange-Id: I18a5187d80bf76c0dc6f22dd8c96a8ffa0f46dc1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a269847761d793b9e8fb4f89c829c31ea7da7130"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a269847761d793b9e8fb4f89c829c31ea7da7130"}]},"branch":"refs/heads/master"},"7df1324844c7e0bf73644be7479e606a9f9653f6":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2021-08-24 06:15:20.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/90/804590/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/90/804590/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/90/804590/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/90/804590/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/90/804590/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/90/804590/6"}}},"commit":{"parents":[{"commit":"646e15da442a51fd13a9bb4864d04295287c0680","subject":"Merge \"Check add_image policy in the API\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/646e15da442a51fd13a9bb4864d04295287c0680"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-08-15 20:01:54.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-08-24 06:03:25.000000000","tz":0},"subject":"Check policies for image import operation in API","message":"Check policies for image import operation in API\n\nThis patch enforces policy checks required for importing/copying image\ndata to store in API layer.\n\nPartially-Implements: blueprint policy-refactor\n\nChange-Id: I18a5187d80bf76c0dc6f22dd8c96a8ffa0f46dc1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7df1324844c7e0bf73644be7479e606a9f9653f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7df1324844c7e0bf73644be7479e606a9f9653f6"}]},"branch":"refs/heads/master"},"5821db41a009c6098d813bb43735d85548c7a2ae":{"kind":"REWORK","_number":7,"created":"2021-08-24 09:36:10.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/90/804590/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/90/804590/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/90/804590/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/90/804590/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/90/804590/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/90/804590/7"}}},"commit":{"parents":[{"commit":"646e15da442a51fd13a9bb4864d04295287c0680","subject":"Merge \"Check add_image policy in the API\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/646e15da442a51fd13a9bb4864d04295287c0680"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-08-15 20:01:54.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-08-24 09:27:00.000000000","tz":0},"subject":"Check policies for image import operation in API","message":"Check policies for image import operation in API\n\nThis patch enforces policy checks required for importing/copying image\ndata to store in API layer.\n\nPartially-Implements: blueprint policy-refactor\n\nChange-Id: I18a5187d80bf76c0dc6f22dd8c96a8ffa0f46dc1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5821db41a009c6098d813bb43735d85548c7a2ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5821db41a009c6098d813bb43735d85548c7a2ae"}]},"branch":"refs/heads/master"},"6801fa9245cf9eff59e748ecbe87569ab228233f":{"kind":"REWORK","_number":8,"created":"2021-08-24 18:28:15.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/90/804590/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/90/804590/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/90/804590/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/90/804590/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/90/804590/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/90/804590/8"}}},"commit":{"parents":[{"commit":"646e15da442a51fd13a9bb4864d04295287c0680","subject":"Merge \"Check add_image policy in the API\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/646e15da442a51fd13a9bb4864d04295287c0680"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-08-15 20:01:54.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-08-24 18:18:30.000000000","tz":0},"subject":"Check policies for image import operation in API","message":"Check policies for image import operation in API\n\nThis patch enforces policy checks required for importing/copying image\ndata to store in API layer.\n\nPartially-Implements: blueprint policy-refactor\n\nChange-Id: I18a5187d80bf76c0dc6f22dd8c96a8ffa0f46dc1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6801fa9245cf9eff59e748ecbe87569ab228233f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6801fa9245cf9eff59e748ecbe87569ab228233f"}]},"branch":"refs/heads/master"},"6c87a18d4cb43aa4af6e953af19f1655f7bb3bf9":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2021-08-24 20:17:05.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/90/804590/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/90/804590/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/90/804590/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/90/804590/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/90/804590/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/90/804590/9"}}},"commit":{"parents":[{"commit":"3e2474676dedfa6677873b2d36c88e3c379a21bc","subject":"Merge \"Move metadef resource type association policy checks in the API\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3e2474676dedfa6677873b2d36c88e3c379a21bc"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-08-15 20:01:54.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-08-24 20:09:24.000000000","tz":0},"subject":"Check policies for image import operation in API","message":"Check policies for image import operation in API\n\nThis patch enforces policy checks required for importing/copying image\ndata to store in API layer.\n\nPartially-Implements: blueprint policy-refactor\n\nChange-Id: I18a5187d80bf76c0dc6f22dd8c96a8ffa0f46dc1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6c87a18d4cb43aa4af6e953af19f1655f7bb3bf9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6c87a18d4cb43aa4af6e953af19f1655f7bb3bf9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
