)]}'
{"id":"openstack%2Fglance~804898","triplet_id":"openstack%2Fglance~master~I7504582dc9751fb2e0fad914d2ee435d0a06f1b7","project":"openstack/glance","branch":"master","topic":"hail-mary","hashtags":[],"change_id":"I7504582dc9751fb2e0fad914d2ee435d0a06f1b7","subject":"DNM [Upload policy] functional-py36 timeout test","status":"ABANDONED","created":"2021-08-17 17:02:43.000000000","updated":"2021-09-15 07:02:39.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"3c2c4cd655dc34a5e28ecde20ec213e10ef725c1","_number":804898,"virtual_id_number":804898,"owner":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-08-19 13:16:43.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-17 17:03:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-17 17:07:45.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2021-08-17 17:07:45.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2021-08-17 23:21:17.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"97a6e4bdf9f7b0d86c8e49a99ce97b2bf746eab0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-17 17:02:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"563ddfe631c14c6f57ba0b43150b0c15fd53b8ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-17 17:03:06.000000000","message":"Patch Set 1: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/glance on branch master.  The error was:\n\n  found undefined alias\n  in \"openstack/glance/.zuul.yaml@master\", line 314, column 31","accounts_in_message":[],"_revision_number":1},{"id":"984949b4243cba1a49e0cc4f861f67d1286354e7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-17 17:05:30.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"697544187aeb0a27c1ef7dbcc0429cdadf7187ee","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-17 17:07:45.000000000","message":"Patch Set 2: Workflow-1\n\nSetting workflow to -1","accounts_in_message":[],"_revision_number":2},{"id":"e2103a59d3d74d477cc261828a68fc0873359d10","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-17 17:16:42.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ca02734035344583835fe692bc374c4d : SUCCESS in 9m 54s","accounts_in_message":[],"_revision_number":2},{"id":"3c12f8ae5acfe25c13a13befa48c826fbabf8a60","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-17 17:19:12.000000000","message":"Patch Set 2:\n\nrecheck attempt 2","accounts_in_message":[],"_revision_number":2},{"id":"e2647886ec616571f83e8e280fc6a6161f217fc4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-17 17:27:37.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/68052ed8ff6b4cd094a225a55f9ca344 : SUCCESS in 7m 37s","accounts_in_message":[],"_revision_number":2},{"id":"fe8305f747e34a59a91367c3f09b08c0038d6859","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-17 18:04:38.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"5927a994ce9b15524c1c0b0f53e0f58f45f2fb20","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-17 18:36:40.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0176d678b476406ebbb5a49d7eda4027 : TIMED_OUT in 31m 20s","accounts_in_message":[],"_revision_number":3},{"id":"152c47e9d7522f717db4409f60d95bebdfb5469c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-17 18:42:47.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"649e0e7f1e9bbd6c90a6b45c591b68d6cfb19301","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-17 19:18:28.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1e58af478fab49ef9a5240ea147e6ff7 : TIMED_OUT in 31m 06s","accounts_in_message":[],"_revision_number":4},{"id":"1301e82a06fda34b4dcfdf180eef6396084ee5e5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-17 19:22:41.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"a4b5f008468a2e6b88532d45b1c386118f2bda07","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-17 19:54:45.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"161f4a288a737e45cdbd3ad1d22f9947b1efbfee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-17 20:10:48.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ad03726cb02a42b983f6ff508cf6cd19 : FAILURE in 7m 07s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/ba8fff77beaa4d3db9654a12b0aad4a4 : FAILURE in 11m 05s","accounts_in_message":[],"_revision_number":6},{"id":"70106028ff6f25227ad36f64cd401fd3c2f2402a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-17 20:10:59.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"6cfb5f7d2a6df551c0b584641afa7ecc3a58e4a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-17 20:31:41.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/71af87234ffc4d30b2c12037e3d3db3d : SUCCESS in 8m 45s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/7ba53b965c91448e9117aa1b21e16494 : SUCCESS in 8m 34s","accounts_in_message":[],"_revision_number":7},{"id":"9fcf4d0fc946d4de402a8a6c81a458ee832bfd07","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-17 20:32:25.000000000","message":"Patch Set 7:\n\nrecheck attempt 2","accounts_in_message":[],"_revision_number":7},{"id":"029bcc90c3cf2e5d1a8b7f2dcfd038fbb16d936e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-17 20:48:18.000000000","message":"Patch Set 7:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f9ae3ff17a74434ca016bf8ea0af024d : SUCCESS in 8m 17s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/2ee36ee2777a4677a97154af4a32f920 : SUCCESS in 8m 09s","accounts_in_message":[],"_revision_number":7},{"id":"58fe88c3084f11435cc0cdb4e213039c43f041ea","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-17 20:49:45.000000000","message":"Patch Set 7:\n\nrecheck attempt 3","accounts_in_message":[],"_revision_number":7},{"id":"1209d8fc618f33fba4d1d5a58db2aa357c9e8a1f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-17 21:02:45.000000000","message":"Patch Set 7:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/894d42f5f5394e4cb9b92bdbbbe209bf : SUCCESS in 8m 42s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/bb85112fd87945a0b9611cdbeac9e421 : SUCCESS in 8m 35s","accounts_in_message":[],"_revision_number":7},{"id":"6b8079fd10cd7136cc3adc4ea75e56a277f550a1","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-17 21:09:16.000000000","message":"Patch Set 7:\n\nrecheck attempt 4","accounts_in_message":[],"_revision_number":7},{"id":"3c65b849de38d2670162c38e4f2cb7e430b2f975","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-17 21:22:37.000000000","message":"Patch Set 7:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/91813c86bcbe4baab376c19cafff5312 : SUCCESS in 8m 23s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/522e435dac834f359ea72e1a27416970 : SUCCESS in 8m 01s","accounts_in_message":[],"_revision_number":7},{"id":"6ee43408fca27c8c17277c1371a1afc7a5b8197e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-17 21:28:58.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"35f7a8ab706142d6eb285f3170b6b5bce85ee25c","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-17 22:11:55.000000000","message":"Patch Set 8:\n\nrecheck attempt 1 after failure","accounts_in_message":[],"_revision_number":8},{"id":"1f072ee99b29475538cc89b0a4731856315efba4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-17 22:44:28.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/deaccfb821ce416abc491f0174a8a9a3 : TIMED_OUT in 31m 30s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/ae809fb456704262a3fd9bcd10a09136 : SUCCESS in 11m 55s","accounts_in_message":[],"_revision_number":8},{"id":"2a697c81d8b9c0531814da95871e52d6b1fd8e1c","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-08-17 23:21:17.000000000","message":"Patch Set 8:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/804898/804898/8/check-not-vote/ext-glance-zuul/4f57c8c : FAILURE in 11m 21s","accounts_in_message":[],"_revision_number":8},{"id":"5b3a259a6cc20fc330404ce8f8b4154589f26655","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-18 05:44:04.000000000","message":"Patch Set 8:\n\nrecheck attempt 2 after failure","accounts_in_message":[],"_revision_number":8},{"id":"4f0ec27eb7892bf04dc719ee882f1a1c74430b61","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-18 05:58:05.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/bc8fd5a31ff34ebb88217b0725d845bd : SUCCESS in 9m 31s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b301c909871e4decb10ca88dfd8b8d64 : SUCCESS in 7m 33s","accounts_in_message":[],"_revision_number":8},{"id":"5846889214f78176b97e5b872844c7a5166858f3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-18 06:04:10.000000000","message":"Uploaded patch set 9: Commit message was updated.","accounts_in_message":[],"_revision_number":9},{"id":"5e5485700a05f6bbf858f05a1531f26dd59bc18e","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-18 06:05:22.000000000","message":"Patch Set 9:\n\nrecheck success attempt 2","accounts_in_message":[],"_revision_number":9},{"id":"bfe56ac7ac3e31b3881670055acaef610b66c320","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-08-18 06:22:59.000000000","message":"Patch Set 9:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/804898/804898/9/check-not-vote/ext-glance-zuul/c1744f0 : FAILURE in 34s","accounts_in_message":[],"_revision_number":9},{"id":"9fb662cfad1ef6363078566509d7081cb60e572c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-18 07:01:26.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c84c6312564b443796b5a1af2e5cde49 : TIMED_OUT in 31m 39s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/a95846bf1cd3410e9ec8f34e1294f076 : SUCCESS in 9m 29s","accounts_in_message":[],"_revision_number":9},{"id":"7c272c25685f7046dfb3a99cc339dbc09dc77244","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-18 07:02:04.000000000","message":"Patch Set 9:\n\nrecheck attempt 3 after failure","accounts_in_message":[],"_revision_number":9},{"id":"5d13a20420b3343eda44f1421a49b16c96f1bb1d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-18 07:16:23.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7add9bbd37ef4b969fb823b9b85d9234 : SUCCESS in 8m 14s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/32b6ef318c9749c0b594fba4b713c118 : SUCCESS in 8m 28s","accounts_in_message":[],"_revision_number":9},{"id":"fe580070a03006e14e54d6afa610e87dc5e669d4","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-18 07:17:16.000000000","message":"Patch Set 9:\n\nrecheck success attempt 3","accounts_in_message":[],"_revision_number":9},{"id":"d7dc5c90db2a66918bcb9af5f89f53362f91beb0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-18 07:54:48.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a7dc73e4556f465ca2ae0f6558a0ba13 : TIMED_OUT in 31m 21s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/2e12423424464516b6a98326b8c6ebb2 : SUCCESS in 9m 02s","accounts_in_message":[],"_revision_number":9},{"id":"ffcc31fdc0afd920ab643fc61b3cca7dc1cfb331","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-18 07:57:13.000000000","message":"Patch Set 9:\n\nrecheck attempt 4 after failur","accounts_in_message":[],"_revision_number":9},{"id":"8e267c76b566a55d2cb854efa5dd0274095f4596","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-18 08:29:13.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b698d009c30a4268b16bd3bc84406145 : TIMED_OUT in 31m 18s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/1ebf564492b14615a129f65ff5f849e2 : SUCCESS in 9m 14s","accounts_in_message":[],"_revision_number":9},{"id":"cefd1a6dd1648fcdf01a840a77648bcda4723f4f","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-18 08:35:15.000000000","message":"Patch Set 9:\n\nrecheck attempt 5 after failure","accounts_in_message":[],"_revision_number":9},{"id":"ff834c3ab476f09fbf1f5c6f2b19e4a28f55b282","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-08-18 08:38:33.000000000","message":"Patch Set 9:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/804898/804898/9/check-not-vote/ext-glance-zuul/4a06e05 : SUCCESS in 36m 39s","accounts_in_message":[],"_revision_number":9},{"id":"c94e558317296a34707e0f7c907432d3ca949fe3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-18 09:11:23.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1458ed54bef345d88a98025b93649867 : TIMED_OUT in 31m 01s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b78404e5f97049c79051f0b51fb55cc4 : SUCCESS in 8m 44s","accounts_in_message":[],"_revision_number":9},{"id":"1a15f7b3810159bd922a45f699380a4e14888901","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-08-18 15:04:26.000000000","message":"Uploaded patch set 10: Patch Set 9 was rebased.","accounts_in_message":[],"_revision_number":10},{"id":"7064f59da12c956c502f6d8327b51b634a9bda60","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-18 15:18:56.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d2ae0a350e4347e29a7e71325ee54bc5 : SUCCESS in 10m 47s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/4d7f3af9c48c4b18a97cbbc5d132be9e : SUCCESS in 8m 45s","accounts_in_message":[],"_revision_number":10},{"id":"62961d275bad9e5bcad2afddb7f6b6d72e12fd36","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-08-18 19:38:32.000000000","message":"Patch Set 10:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/804898/804898/10/check-not-vote/ext-glance-zuul/7d92a3e : SUCCESS in 38m 58s","accounts_in_message":[],"_revision_number":10},{"id":"46eb5eeb70e0e2867897c1e36a9e76e2293b7e05","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-19 09:45:01.000000000","message":"Patch Set 10:\n\nrecheck success attempt 2","accounts_in_message":[],"_revision_number":10},{"id":"a570434d6510a893e26fa15813f78bf15697fb0d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-19 09:58:25.000000000","message":"Patch Set 10:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d960abf03ae841d992e6baf02e4f3ea8 : SUCCESS in 11m 08s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c36df1b727c94626903c401f5e4381aa : SUCCESS in 10m 54s","accounts_in_message":[],"_revision_number":10},{"id":"6a9f7cc26c703d4f16e3ccaa3c4dc86301b353f1","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-19 10:02:10.000000000","message":"Patch Set 10:\n\nrecheck success attempt 3","accounts_in_message":[],"_revision_number":10},{"id":"257f6aa84506fcf37c4bd2aff01e4997e86caedc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-19 10:23:08.000000000","message":"Patch Set 10:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/afdead03b98e407697a9e4de4819fc78 : SUCCESS in 9m 40s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/ce73923c2f654d2c82107041728ea65e : SUCCESS in 8m 20s","accounts_in_message":[],"_revision_number":10},{"id":"66020de9a7934bfa495f4d5d9585a868d356cfe3","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-19 10:26:07.000000000","message":"Patch Set 10:\n\nrecheck success attempt 4","accounts_in_message":[],"_revision_number":10},{"id":"cf3597dc7aa836d7838ee6bc481b203bbed17955","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-19 10:36:38.000000000","message":"Patch Set 10:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/aa50aceb79c84b9bb53e55f4cd9facad : SUCCESS in 8m 43s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/1e7ef97bd8964cf3ad83361a6bdab2ab : SUCCESS in 8m 56s","accounts_in_message":[],"_revision_number":10},{"id":"3134a332c152c142f05a099d0dc0bd15f60542d3","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-19 10:37:54.000000000","message":"Patch Set 10:\n\nrecheck success attempt 5","accounts_in_message":[],"_revision_number":10},{"id":"fe764be70ff8d2ef2efebfc42637e4fd6332b55d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-19 10:50:53.000000000","message":"Patch Set 10:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b25743a6312d4e4785502f037461ebc4 : SUCCESS in 8m 43s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/657afc5fb3a8420a8cd1b403012323c0 : SUCCESS in 8m 38s","accounts_in_message":[],"_revision_number":10},{"id":"c829c2b84a006b9a76ec5c57683d7d458ebccea6","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-19 11:29:52.000000000","message":"Patch Set 10:\n\nrecheck success attempt 6","accounts_in_message":[],"_revision_number":10},{"id":"d10123bcaffb594ffc6a7a296ccbd4bdc9cba235","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-19 11:44:31.000000000","message":"Patch Set 10:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/68af9c559af04ad6a243804adc5c0e5a : SUCCESS in 8m 48s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/0fcc0e1a56234689801361013c60007e : SUCCESS in 8m 34s","accounts_in_message":[],"_revision_number":10},{"id":"5af877eb85959e8db2c5d48997bef70608832d82","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-19 12:26:04.000000000","message":"Patch Set 10:\n\nrecheck success attempt 7","accounts_in_message":[],"_revision_number":10},{"id":"c76423a0e1bb0a0ae6e8664474d6484b690354e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-19 12:41:47.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/26a0b872c85d4fb19674569225773a51 : FAILURE in 9m 53s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/98dfb7c31e304df3955a09d6abf29770 : SUCCESS in 8m 43s","accounts_in_message":[],"_revision_number":10},{"id":"ac93a163201ebfaf02f4c4aedfa6fb20451266a9","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-19 13:02:07.000000000","message":"Patch Set 10:\n\nrecheck no timeout attempt 8","accounts_in_message":[],"_revision_number":10},{"id":"5c2831b94aceb9057330ab0def4d42fcba47c516","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-19 13:16:43.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/da02d9bf924a46369cb5f1b73ef0e6ed : SUCCESS in 8m 53s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/135af8b46a3b43528acea7b337638c84 : SUCCESS in 8m 06s","accounts_in_message":[],"_revision_number":10},{"id":"39f2eb4be48914816959dfeeadefb0a576bf7651","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-19 13:27:50.000000000","message":"Patch Set 10:\n\nrecheck no timeout attempt 9","accounts_in_message":[],"_revision_number":10},{"id":"e9b593f8c4fbd9b58e79c0d22a9a89af8a6cb30e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-19 14:09:18.000000000","message":"Patch Set 10:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/281730121bdd4a20b3c79af97d018282 : SUCCESS in 9m 29s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/3bbd552818ed46d99854359e94262b0c : SUCCESS in 9m 57s","accounts_in_message":[],"_revision_number":10},{"id":"f7c8de9b31229a9968b0bed7b62867e214caef29","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-08-19 17:33:15.000000000","message":"Patch Set 10:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/804898/804898/10/check-not-vote/ext-glance-zuul/2cb8174 : SUCCESS in 37m 51s","accounts_in_message":[],"_revision_number":10},{"id":"3c2c4cd655dc34a5e28ecde20ec213e10ef725c1","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-09-15 07:02:39.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"da73a0a0d079d89578164dcfd16642631e81cc8c","revisions":{"42051661b944c0917e7bfe40771e696dafe89dc7":{"kind":"REWORK","_number":1,"created":"2021-08-17 17:02:43.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/98/804898/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/98/804898/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/98/804898/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/98/804898/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/98/804898/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/98/804898/1"}}},"commit":{"parents":[{"commit":"9e002a77f2131d3594a2a4029a147beaf37f5b17","subject":"Resolve compatibility with oslo.db future (redux)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9e002a77f2131d3594a2a4029a147beaf37f5b17"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-08-17 16:57:55.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-08-17 16:57:55.000000000","tz":0},"subject":"DNM functional-py36 timeout test","message":"DNM functional-py36 timeout test\n\nChange-Id: I7504582dc9751fb2e0fad914d2ee435d0a06f1b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/42051661b944c0917e7bfe40771e696dafe89dc7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/42051661b944c0917e7bfe40771e696dafe89dc7"}]},"branch":"refs/heads/master"},"65f8cc83914024596f4fd9e6dda391da6fc94a90":{"kind":"REWORK","_number":2,"created":"2021-08-17 17:05:30.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/98/804898/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/98/804898/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/98/804898/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/98/804898/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/98/804898/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/98/804898/2"}}},"commit":{"parents":[{"commit":"9e002a77f2131d3594a2a4029a147beaf37f5b17","subject":"Resolve compatibility with oslo.db future (redux)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9e002a77f2131d3594a2a4029a147beaf37f5b17"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-08-17 16:57:55.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-08-17 17:01:17.000000000","tz":0},"subject":"DNM functional-py36 timeout test","message":"DNM functional-py36 timeout test\n\nChange-Id: I7504582dc9751fb2e0fad914d2ee435d0a06f1b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/65f8cc83914024596f4fd9e6dda391da6fc94a90"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/65f8cc83914024596f4fd9e6dda391da6fc94a90"}]},"branch":"refs/heads/master"},"f62bb1d9d6c4e68d811091551c5c75c43ba05c75":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-08-17 18:04:38.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/98/804898/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/98/804898/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/98/804898/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/98/804898/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/98/804898/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/98/804898/3"}}},"commit":{"parents":[{"commit":"4407d6ceb5e08e3da99c236766b755d3c1a3d8ce","subject":"Check upload_image policy in the API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4407d6ceb5e08e3da99c236766b755d3c1a3d8ce"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-08-17 16:57:55.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-08-17 17:59:25.000000000","tz":0},"subject":"DNM functional-py36 timeout test","message":"DNM functional-py36 timeout test\n\nChange-Id: I7504582dc9751fb2e0fad914d2ee435d0a06f1b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f62bb1d9d6c4e68d811091551c5c75c43ba05c75"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f62bb1d9d6c4e68d811091551c5c75c43ba05c75"}]},"branch":"refs/heads/master"},"ae6ace3e35e0600072854a7ef98a1eaa7784558e":{"kind":"REWORK","_number":4,"created":"2021-08-17 18:42:47.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/98/804898/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/98/804898/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/98/804898/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/98/804898/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/98/804898/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/98/804898/4"}}},"commit":{"parents":[{"commit":"4407d6ceb5e08e3da99c236766b755d3c1a3d8ce","subject":"Check upload_image policy in the API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4407d6ceb5e08e3da99c236766b755d3c1a3d8ce"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-08-17 16:57:55.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-08-17 18:38:05.000000000","tz":0},"subject":"DNM functional-py36 timeout test","message":"DNM functional-py36 timeout test\n\nChange-Id: I7504582dc9751fb2e0fad914d2ee435d0a06f1b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ae6ace3e35e0600072854a7ef98a1eaa7784558e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ae6ace3e35e0600072854a7ef98a1eaa7784558e"}]},"branch":"refs/heads/master"},"10febd037f0c99f7d78b634056be9ca844e6daa9":{"kind":"REWORK","_number":5,"created":"2021-08-17 19:22:41.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/98/804898/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/98/804898/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/98/804898/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/98/804898/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/98/804898/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/98/804898/5"}}},"commit":{"parents":[{"commit":"4407d6ceb5e08e3da99c236766b755d3c1a3d8ce","subject":"Check upload_image policy in the API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4407d6ceb5e08e3da99c236766b755d3c1a3d8ce"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-08-17 16:57:55.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-08-17 19:18:06.000000000","tz":0},"subject":"DNM functional-py36 timeout test","message":"DNM functional-py36 timeout test\n\nReverted chagnes from image_data.py\nReverted functional test case changes\n\nChange-Id: I7504582dc9751fb2e0fad914d2ee435d0a06f1b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/10febd037f0c99f7d78b634056be9ca844e6daa9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/10febd037f0c99f7d78b634056be9ca844e6daa9"}]},"branch":"refs/heads/master"},"19f9aaeb5f549f07a0499b04120a7706c7ed69e4":{"kind":"REWORK","_number":6,"created":"2021-08-17 19:54:45.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/98/804898/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/98/804898/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/98/804898/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/98/804898/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/98/804898/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/98/804898/6"}}},"commit":{"parents":[{"commit":"4407d6ceb5e08e3da99c236766b755d3c1a3d8ce","subject":"Check upload_image policy in the API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4407d6ceb5e08e3da99c236766b755d3c1a3d8ce"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-08-17 16:57:55.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-08-17 19:49:58.000000000","tz":0},"subject":"DNM functional-py36 timeout test","message":"DNM functional-py36 timeout test\n\nReverted chagnes from image_data.py\nReverted functional test case changes\n\nChange-Id: I7504582dc9751fb2e0fad914d2ee435d0a06f1b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/19f9aaeb5f549f07a0499b04120a7706c7ed69e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/19f9aaeb5f549f07a0499b04120a7706c7ed69e4"}]},"branch":"refs/heads/master"},"d2020599846fec6961f30271eabf4bebda060ae9":{"kind":"REWORK","_number":7,"created":"2021-08-17 20:10:59.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/98/804898/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/98/804898/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/98/804898/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/98/804898/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/98/804898/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/98/804898/7"}}},"commit":{"parents":[{"commit":"4407d6ceb5e08e3da99c236766b755d3c1a3d8ce","subject":"Check upload_image policy in the API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4407d6ceb5e08e3da99c236766b755d3c1a3d8ce"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-08-17 16:57:55.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-08-17 20:06:14.000000000","tz":0},"subject":"DNM functional-py36 timeout test","message":"DNM functional-py36 timeout test\n\nReverted chagnes from image_data.py\nReverted functional test case changes\n\nChange-Id: I7504582dc9751fb2e0fad914d2ee435d0a06f1b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d2020599846fec6961f30271eabf4bebda060ae9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d2020599846fec6961f30271eabf4bebda060ae9"}]},"branch":"refs/heads/master"},"83d4b2d0675afb67ef326d2f0f77946362405c59":{"kind":"REWORK","_number":8,"created":"2021-08-17 21:28:58.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/98/804898/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/98/804898/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/98/804898/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/98/804898/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/98/804898/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/98/804898/8"}}},"commit":{"parents":[{"commit":"4407d6ceb5e08e3da99c236766b755d3c1a3d8ce","subject":"Check upload_image policy in the API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4407d6ceb5e08e3da99c236766b755d3c1a3d8ce"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-08-17 16:57:55.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-08-17 21:22:38.000000000","tz":0},"subject":"DNM functional-py36 timeout test","message":"DNM functional-py36 timeout test\n\nChange-Id: I7504582dc9751fb2e0fad914d2ee435d0a06f1b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/83d4b2d0675afb67ef326d2f0f77946362405c59"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/83d4b2d0675afb67ef326d2f0f77946362405c59"}]},"branch":"refs/heads/master"},"16f8b858b558c308a2df35b1d6cc1a411e1edc21":{"kind":"NO_CODE_CHANGE","_number":9,"created":"2021-08-18 06:04:10.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/98/804898/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/98/804898/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/98/804898/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/98/804898/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/98/804898/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/98/804898/9"}}},"commit":{"parents":[{"commit":"4407d6ceb5e08e3da99c236766b755d3c1a3d8ce","subject":"Check upload_image policy in the API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4407d6ceb5e08e3da99c236766b755d3c1a3d8ce"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-08-17 16:57:55.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-08-18 05:58:44.000000000","tz":0},"subject":"DNM [Upload policy] functional-py36 timeout test","message":"DNM [Upload policy] functional-py36 timeout test\n\nChange-Id: I7504582dc9751fb2e0fad914d2ee435d0a06f1b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/16f8b858b558c308a2df35b1d6cc1a411e1edc21"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/16f8b858b558c308a2df35b1d6cc1a411e1edc21"}]},"branch":"refs/heads/master"},"da73a0a0d079d89578164dcfd16642631e81cc8c":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2021-08-18 15:04:26.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/98/804898/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/98/804898/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/98/804898/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/98/804898/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/98/804898/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/98/804898/10"}}},"commit":{"parents":[{"commit":"dd3155516cec2cabf8f74963a44ab642d507384b","subject":"Merge \"Load options from the castellan library\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/dd3155516cec2cabf8f74963a44ab642d507384b"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-08-17 16:57:55.000000000","tz":0},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-08-18 16:07:00.000000000","tz":-420},"subject":"DNM [Upload policy] functional-py36 timeout test","message":"DNM [Upload policy] functional-py36 timeout test\n\nChange-Id: I7504582dc9751fb2e0fad914d2ee435d0a06f1b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/da73a0a0d079d89578164dcfd16642631e81cc8c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/da73a0a0d079d89578164dcfd16642631e81cc8c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
