)]}'
{"id":"openstack%2Fglance~804966","triplet_id":"openstack%2Fglance~master~I29448746b14c542e5fbf0283011968ae1516642e","project":"openstack/glance","branch":"master","topic":"bp/append-tags","attention_set":{},"removed_from_attention_set":{"33765":{"account":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"last_update":"2022-02-17 16:01:16.000000000","reason":"Change was submitted"},"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2022-02-15 14:18:14.000000000","reason":"Abhishek Kekane replied on the change"},"4393":{"account":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"last_update":"2022-02-17 16:01:16.000000000","reason":"Change was submitted"},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2022-02-17 14:45:37.000000000","reason":"Cyril Roelandt replied on the change"},"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2022-02-16 10:24:33.000000000","reason":"Rajat Dhasmana replied on the change"}},"hashtags":[],"change_id":"I29448746b14c542e5fbf0283011968ae1516642e","subject":"Fix overwriting of existing tags while creating new tags","status":"MERGED","created":"2021-08-18 08:33:22.000000000","updated":"2022-02-17 16:03:32.000000000","submitted":"2022-02-17 16:01:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":81,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"804966-bp/append-tags","meta_rev_id":"00c52270aeecf0b38af24e74854e0e834b7dd448","_number":804966,"virtual_id_number":804966,"owner":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":0,"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-17 16:01:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"recommended":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"all":[{"value":1,"date":"2022-02-17 14:05:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":0,"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},{"value":2,"date":"2022-02-16 17:07:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":2,"date":"2022-02-17 14:45:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":0,"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":0,"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":1,"date":"2022-02-17 14:45:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":0,"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"}],"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-18 08:49:16.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-08-18 10:02:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-18 10:08:55.000000000","updated_by":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"reviewer":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"state":"REVIEWER"},{"updated":"2021-08-19 07:43:00.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2021-08-23 14:12:22.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2021-09-08 17:40:17.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2021-09-20 13:58:51.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"CC"},{"updated":"2022-02-15 23:24:27.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"}],"messages":[{"id":"769f75613f3a7c5d918702fd5b94e04c2bd18102","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"date":"2021-08-18 08:33:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"716a761356aba3ddac5a9b688e343fe97f3745be","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-08-18 08:49:16.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- http://p3-ci-int.eng.vmware.com:8080/job/ext-glance-zuul/1829/ : ABORTED in 24s","accounts_in_message":[],"_revision_number":1},{"id":"7d4588d735f94a7f4d1ead563dc0adab0016755a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-18 10:02:23.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/0b284d8b5c8d48ac8451bbf528ed5b81 : SUCCESS in 58m 40s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5d4b411273a446038087b54f185e9f1c : SUCCESS in 1h 05m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/beea3cd474d3409295661017343f7e1e : SUCCESS in 6m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aff32f55c7a0486f814b15d2be14ee55 : FAILURE in 7m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/55409759301d4bcb9e885a23bc00f046 : FAILURE in 5m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bc058a3628634b5e9bdf692b79faaa92 : FAILURE in 8m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a45ed1a10b584b04a86cea212d4d1db6 : SUCCESS in 6m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c0f778dbc29d42ecbae137da3ba5eb48 : SUCCESS in 7m 42s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c07758ec69c749ccb036caab26c4c363 : FAILURE in 11m 53s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/a0ef575979b54153b53b7052c37f40d9 : FAILURE in 10m 08s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/9e050fcea1814fa9b4cbe93309b91745 : FAILURE in 9m 59s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/a32797b1ee4b4395bc936463056d9b09 : SUCCESS in 5m 45s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/fe1e1c6790cd45098bdd4403c25a8ba3 : FAILURE in 1h 23m 26s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/c846e4e431214a15b86f62b0dd42c5ef : SUCCESS in 1h 11m 03s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/951783e3d27d475abdb2f65480a01e76 : SUCCESS in 58m 45s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/5cbf3187bc55428db659d5569e19e9b1 : SUCCESS in 1h 11m 42s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5c1c787c8a71409c929d9dfdb846e3ca : SUCCESS in 41m 24s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/31c78e1641ee4d17803e8013bb2109ff : SUCCESS in 1h 08m 52s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/99b77c0c31054b87af86c15189922254 : SUCCESS in 28m 11s","accounts_in_message":[],"_revision_number":1},{"id":"0ca0f8338afdc55b7f42ffb2399048fab9b64ab5","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2021-08-18 10:08:55.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)\n\nThanks for this fix.\nSome existing unit tests also needs to modify.","accounts_in_message":[],"_revision_number":1},{"id":"629afbf3f12c87b83166fc904baa4feb37722052","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2021-08-19 07:43:00.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a8bc7b2fb5bfa4999e9ff4c084493ea85c2b3592","author":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"date":"2021-08-20 08:16:27.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"4e43e36aa2b3b27b6ed9e42a9871ef16aa04824d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"date":"2021-08-23 14:04:03.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d4f2466d7a677231fb9100b12ca873df1a6cb3f9","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-23 14:12:22.000000000","message":"Patch Set 2: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":2},{"id":"ee6aabe5b044bad93b427f094374db22ec2ddb9e","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-08-23 15:40:54.000000000","message":"Patch Set 2:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/804966/804966/2/check-not-vote/ext-glance-zuul/b4b65af : SUCCESS in 38m 39s","accounts_in_message":[],"_revision_number":2},{"id":"2ee4979a3cd807d478fb91734d3128b305088395","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-23 15:41:14.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/bf809d79c0b54f1bafa32236ccfe9d20 : SUCCESS in 57m 24s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/01547957bbde48a1b5c36f771b63b073 : SUCCESS in 1h 27m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e3c0dd38562d4a6685d33f4fc3635af6 : SUCCESS in 6m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e403170172f34c98a3b61b4c88e25793 : SUCCESS in 6m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c19891e41a1947ae99d2f26af7675791 : SUCCESS in 6m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3026ea6fc07a4bb4b8cce2ad74c34b97 : SUCCESS in 6m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4510ee044a5c4fffb34c882b9fd48741 : SUCCESS in 7m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8c27f8c0b8d34860a662b5a2401f2e1f : SUCCESS in 7m 08s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3a7c7199221c403c8a84f54598e557ee : SUCCESS in 10m 15s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/834a9a9966b1427397ba1c218f5f0ba0 : SUCCESS in 9m 15s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/07d9badada674469b946e311fba4f257 : SUCCESS in 9m 33s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/226106c7350c4a49a03395c3b15cbbed : SUCCESS in 5m 39s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/88d77c31e08743ef9e36767c10ea8dbf : SUCCESS in 1h 08m 02s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/50272450598448cf82452a7e0ce3b96a : SUCCESS in 1h 02m 48s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/da8d426c96f94032ae8e62cbae7ed2a3 : SUCCESS in 1h 02m 34s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/7ddeeab964d24606b0360e3207d6743a : SUCCESS in 1h 23m 24s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bc0082076c1b4ed7a474436b07248380 : SUCCESS in 52m 35s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/d3de4d4806ad4ad389b89060ba7b74e3 : SUCCESS in 1h 13m 16s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/584e44f806104ac4a9c617f554358526 : SUCCESS in 35m 17s","accounts_in_message":[],"_revision_number":2},{"id":"3f89795e3e51718dcae1de2a7742766bb5fc178e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"date":"2021-08-24 05:48:49.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"97415b624f16bf7655bde18d81da6be65ef2776c","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-08-24 06:24:30.000000000","message":"Patch Set 3:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/804966/804966/3/check-not-vote/ext-glance-zuul/cf178fd : SUCCESS in 35m 13s","accounts_in_message":[],"_revision_number":3},{"id":"3569010ddc55cdac9e1b405309ee7914e98dd8bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-24 07:16:53.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/330caa1c3df040418efb178c34f32f27 : SUCCESS in 1h 08m 49s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f5cfd5b118d44acab4f750c574c4a053 : SUCCESS in 1h 20m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/851fe2a258284bfaa8e60ee905e9c61f : SUCCESS in 5m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3262a2667f304deeb2edeef658b60c04 : SUCCESS in 6m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0cae6c9469a540fa96797a64e67b651c : SUCCESS in 6m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2b0c86855a124e9d85aa0d48077c8a8f : SUCCESS in 6m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01bfe9e1f9134539b0d36a873d295f99 : SUCCESS in 7m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a243cbb177074bcd91ad81af4159911b : SUCCESS in 6m 11s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/df3739deb7974173835e3345deccf0de : SUCCESS in 9m 00s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/cdf83fe51e2440e1a684ff7ca3c02d09 : SUCCESS in 11m 32s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/8c0664c282f342dabdc3a4cb127804c7 : SUCCESS in 10m 20s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/b78524255bcf48dea98625f45abad592 : SUCCESS in 6m 43s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/b10c48f28ece49449b3ea808716c8eac : SUCCESS in 1h 14m 39s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/8b636b1fad1049d486ec91b647e25f6e : SUCCESS in 1h 04m 27s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/0caa6ac64a6b425ea7e5580aaf66b963 : SUCCESS in 1h 09m 41s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/f3df6569eac940fc992f9f31aa778047 : SUCCESS in 1h 13m 02s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1184cd73697f454982555b0d490f34db : SUCCESS in 47m 15s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/33eb674d151240a0a09f7093dfb58586 : SUCCESS in 1h 18m 09s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/8a34c8fa4b034333bdc4a83045452c68 : SUCCESS in 33m 01s","accounts_in_message":[],"_revision_number":3},{"id":"47987cd9ac96ce45970cccbe40738100cb6d2845","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-25 05:59:32.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)\n\nChanges looks good, we need additional test to verify this change.","accounts_in_message":[],"_revision_number":3},{"id":"9bd57c1e8e2ee65ea658877bb0371b40046b7756","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"date":"2021-08-26 04:40:27.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"6ebdf3bd2b1086df754125284b7c24badbecc682","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"date":"2021-08-26 05:25:06.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"c066c38cdb4183591f315ee7ce7e9a99d838794e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-26 06:56:06.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/7c7ee28d856f4b7aacf62d166043824a : SUCCESS in 1h 22m 30s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9415acfe2a28428ea13ef8f30245c02d : SUCCESS in 1h 15m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9405ddc9dd3c46aaab79b56b8d6ab756 : SUCCESS in 5m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/78faead0a0b945b0a8d85129eb5a4511 : SUCCESS in 5m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7b77a5aae4454b8fae2d5fe0420c8a6e : SUCCESS in 5m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9d7a985867014a9087d40323db8a1700 : SUCCESS in 6m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b05267451434593bf5af6e7ac401b40 : SUCCESS in 7m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a68b2a5b314447658d8e7c81eb8034ec : SUCCESS in 7m 02s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8dbc95469f944ff48dc2228ed15a43db : SUCCESS in 9m 37s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/d6ed444814464e7f96c64fb0796049c9 : SUCCESS in 8m 40s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/a7c92a6364bd4e7ea818bc4cab7da385 : SUCCESS in 9m 05s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/110b907d090f4341b62c9941be3e5785 : SUCCESS in 5m 03s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/bc17e5ff5bea4c8d825937c57a78feab : SUCCESS in 1h 16m 47s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/4e0293c2c74d4ecc8288a42637a079a8 : FAILURE in 1h 09m 04s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/aec2a42827c04b5b901503ff2b156749 : SUCCESS in 1h 11m 39s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/0cb35beb2aa74af1a5a8d0dd69b97459 : SUCCESS in 1h 15m 55s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/24abfbc8a8584969957ac012a3685ad6 : SUCCESS in 45m 25s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/6e9b6876568d44e68b5a5e0b5445d4c0 : FAILURE in 1h 14m 28s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/647618a53bf14765a8001b6e6b6495cc : SUCCESS in 33m 00s","accounts_in_message":[],"_revision_number":5},{"id":"cf33f5c6b96b8a6aee9b4fc47a8fd752e45a660b","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2021-08-26 09:57:52.000000000","message":"Patch Set 5:\n\nrecheck\n\nunrelated failures","accounts_in_message":[],"_revision_number":5},{"id":"c600471ffe6bfd025a23397934c2be02b981b1a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-26 11:31:25.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/a85314e9ce804fee9bfcb2c341320a39 : SUCCESS in 1h 04m 18s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/af9b60aa6e8242239d78be28709a7a3c : SUCCESS in 1h 06m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f541f36acdc4045a5370e5d3bed3f78 : SUCCESS in 6m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c01f0349271446c88c768cd25e3e0816 : SUCCESS in 6m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a711286c76be44fca16a6b2e24a03da8 : SUCCESS in 5m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d796703130a2483ab2bf04106d9d78cf : SUCCESS in 8m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46fa6e24294141acb297755a36c33855 : SUCCESS in 7m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b56bc89d0f584f4cba9c5ebbef496e07 : SUCCESS in 7m 46s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d037934df0d14b3ba4b71b89a4884809 : SUCCESS in 11m 18s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/bda07034b77841488b30e09c4c45b896 : SUCCESS in 10m 50s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/505df93e52d042ee9b3ec773e7ae26b8 : SUCCESS in 9m 14s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/b6beb584fc0f42ca891f3d8300ea8c72 : SUCCESS in 5m 15s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/c9d6e5cddc2949e99e77e6cbccf6607e : FAILURE in 1h 23m 41s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/fa7935ba396c42a8a15b1da93c29cbba : SUCCESS in 1h 08m 44s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/7bb720f960b24ce4a2c8eeb92c28ef6c : SUCCESS in 1h 04m 48s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/4a65f53e16724347ae879f46c5c04d70 : SUCCESS in 1h 09m 21s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7a301eff2d0d4f618e51b4c716137dbc : SUCCESS in 48m 09s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/6ce691a00af340bd924b5dab92f570b1 : SUCCESS in 1h 09m 27s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/359d26d1c866426cbd83b8200f9351f5 : SUCCESS in 32m 03s","accounts_in_message":[],"_revision_number":5},{"id":"15a1b1e6c456de1145caf7e1609f7ea9d6ee0d2c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"date":"2021-08-26 13:23:55.000000000","message":"Uploaded patch set 6: New patch set was added with same tree, parent, and commit message as Patch Set 5.","accounts_in_message":[],"_revision_number":6},{"id":"246fbdfee057259f45d92cb3c3bfb835f0337120","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2021-08-26 13:35:46.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"95db345e671c8cf92c94662d4a4048594c2ace32","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-26 17:09:12.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/21f34d1fb0154e5da40a814be9896f3d : SUCCESS in 1h 09m 24s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/df7e0325057242b0a4aa91efe563fb1b : SUCCESS in 1h 10m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b970ef5fd5374003a9a6116d05919967 : SUCCESS in 7m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/945fc89ffced4b3082a3d0b1f2bdfc9a : SUCCESS in 6m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fff142cd0cc74ccb8c643e649d8c77ab : SUCCESS in 6m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6108800c52f9493996f34a95b41bfdc9 : SUCCESS in 7m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c1207b5ee4f468ab1ba6ac46f706769 : SUCCESS in 7m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8b2e28d8d9a3495cb900c88d1753b9bd : SUCCESS in 7m 25s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e24cfcfada454ef3b604db39609a1e60 : SUCCESS in 9m 54s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/77943c6ce6e148908a3159ada24b5497 : SUCCESS in 10m 37s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/879f08357891448eb5887afccfb82bcd : SUCCESS in 9m 26s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/b38b5be824914a5eb45bead640360148 : SUCCESS in 7m 52s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/75490aca1eb24b55a074c22d3a6fb818 : SUCCESS in 1h 18m 17s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/ee881e52635f4cee9ff0f1ee63af7b3d : SUCCESS in 1h 17m 36s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/819f94303643448d880f9051235caba7 : SUCCESS in 1h 13m 33s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/cfcb57ecda3f4ee8b5e7f6532b4a4e89 : SUCCESS in 1h 19m 59s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/32dcc8e799a24ed8827beff7cf0e7c7d : SUCCESS in 53m 51s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/c1e532f6e7ba48f7a96da31f89d2a693 : FAILURE in 1h 09m 34s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/93de1f8dc41d400dab6cd9105d5fbc6e : SUCCESS in 29m 23s","accounts_in_message":[],"_revision_number":7},{"id":"aeef3d87f521a18ea1a2d0e0434c7322813aeef5","author":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"date":"2021-08-27 06:12:41.000000000","message":"Patch Set 7:\n\n(8 comments)\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"cb66c6330dfa3b97c4035b4a4cc62606bcfdbf31","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2021-08-27 06:38:37.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"8d3297709b9de7eb1aff10dbd1ecc3c1eedcfe8f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-27 08:42:42.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/c182a21f43524d948674678684c78953 : SUCCESS in 56m 15s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3ff92ad09d3b49bf978aa69df70ade5f : SUCCESS in 1h 12m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/444a759662d141758892aeee238a331f : SUCCESS in 7m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/de1d950008044fcc9006fb613b167ace : SUCCESS in 7m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ed76b92fc1914b4d9205609272b18714 : SUCCESS in 8m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/71ff1398ab6941fe8243958dc4758e90 : SUCCESS in 8m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64bf1caf5cdd4d56b628e0e1024d7251 : SUCCESS in 8m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a90081ade94b449bb79ab1dfeca603d7 : SUCCESS in 9m 45s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/71842a82b8ab4f4a9f29bafc3c92c7cb : SUCCESS in 11m 51s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/208db9c1386d4af294d49ec5290eb12f : SUCCESS in 10m 48s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/95de2ccef17747a497c6b8c6a607b715 : SUCCESS in 10m 19s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/fbb9ff1085334da4ae23d448844635c9 : SUCCESS in 8m 25s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/9d497ae5a0504f6fb11a4f904fb87071 : FAILURE in 1h 23m 10s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/2c7c7ad986bd430aa149a24087dc0334 : FAILURE in 1h 02m 29s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/03eecae8908f44e092b793ee2e50f551 : FAILURE in 56m 16s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/743ab888d3164777a286d33ab2853350 : SUCCESS in 1h 12m 15s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9763bcbdef09423f9b409be0237a414e : SUCCESS in 49m 16s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/a923da898afc444889ec1fc85ce34814 : SUCCESS in 1h 25m 15s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/d537aac81b1842ca8b42033efb3477e0 : SUCCESS in 32m 24s","accounts_in_message":[],"_revision_number":7},{"id":"6574040e27cd09efa2f985dee7416f1aeadb6bf8","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-08-27 13:51:51.000000000","message":"Patch Set 7:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/804966/804966/7/check-not-vote/ext-glance-zuul/f06e5c2 : SUCCESS in 38m 23s","accounts_in_message":[],"_revision_number":7},{"id":"571078797621e0d189e1aa1451c2e2832ff56d76","author":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"date":"2021-08-30 10:26:29.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"4b07b330247c2a4019ddf896a261fcae37ec8ef1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-30 11:46:01.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/320d67435bc24f1ea6caf20d82a4d19f : SUCCESS in 1h 02m 40s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d4a6527a8d574c41b4413c6eae136184 : SUCCESS in 1h 02m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da603a64fcde47c2907fec41b0e0670e : SUCCESS in 5m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/786064b0569645fdae21000d3961ed37 : SUCCESS in 7m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/36ec24c8b1e14b5d8a82ec9e5905f495 : SUCCESS in 6m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dfd262ef205946a0a8988ff0137a381e : SUCCESS in 7m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ecde2a41887b4221801fc72be0385745 : SUCCESS in 6m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cbc7e340ac5049e097c9fea3d9bfd1c8 : SUCCESS in 6m 58s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c7d4756ee5e1486b9e3428651ac71757 : SUCCESS in 10m 09s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/ad5613a536604b10a80bb44ac9ff3802 : SUCCESS in 10m 29s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/1667c3926bed46bf869bed7744fb024a : SUCCESS in 8m 59s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/5bdf5d09cc0b47788d58e5328f417ac0 : SUCCESS in 6m 41s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/d622d117677e4240840e3d11e1c1b3af : SUCCESS in 1h 12m 00s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/677f9abed7d848d68ed8a9585d688432 : SUCCESS in 1h 07m 19s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/bcd5f2e8cca049f3b522beec59eafeb4 : SUCCESS in 1h 06m 13s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/534d06de70964c5e991da41ccdf6df6c : FAILURE in 1h 00m 56s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8eafb82b99ba4ddfb343d25ca3d2c1a2 : SUCCESS in 54m 46s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/8fd63a3994d941809f4b6e7f10912af7 : SUCCESS in 1h 11m 17s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/0be34d12433e48c88f9c8e4321704375 : SUCCESS in 33m 42s","accounts_in_message":[],"_revision_number":7},{"id":"b296e09e1d9ad1ef70724634472dd57f0d44b765","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-31 08:53:00.000000000","message":"Patch Set 7: Code-Review+2\n\nThank you for working on this!","accounts_in_message":[],"_revision_number":7},{"id":"7cf81536a3f6406e7dc11d3cd76a861b41ff0123","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2021-08-31 09:32:26.000000000","message":"Patch Set 7: Code-Review+1\n\nThanks for this fix.","accounts_in_message":[],"_revision_number":7},{"id":"3101d8acbf3593df8ccb9776185bcb5a853e86e7","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-08-31 17:45:41.000000000","message":"Patch Set 7:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/804966/804966/7/check-not-vote/ext-glance-zuul/c0a3cbb : SUCCESS in 37m 42s","accounts_in_message":[],"_revision_number":7},{"id":"14ba705e1719749a62e2eece61b143f0a333f98f","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-09-08 17:40:17.000000000","message":"Patch Set 7: Code-Review-1\n\nI totally agree that this is how the API _should_ behave. However, it\u0027s a departure from how it _does_ behave today. People code to the APIs as they behave, not always as they are documented, and especially if they are not documented at all. The API WG guidelines cover this pretty clearly:\n\n “The existing API is not well documented.”\n\n If an API’s behavior isn’t adequately documented, then developers using the API have no choice but to\n go by what they observe the behavior to be. A change that will violate those observations is a change \n that requires a version.\n\nFrom https://specs.openstack.org/openstack/api-wg/guidelines/api_interoperability.html#interoperability in the examples section.\n\nIn this case, we would be breaking anyone that currently assumes that they are passing the eventual set of tags. Something that adds a tag and removes another by passing the full set would suddenly stop removing the tag they don\u0027t pass. If glance had microversion support, we could change this behavior in the new microversion and continue to honor the old behavior, but we don\u0027t.\n\nSo I think that the proper way to fix this (aside from adopting microversions, which we should!) is to add a new optional parameter indicating the intent, which defaults to the old behavior. Something like \"?append\u003dFalse\" which can be overridden to \"?append\u003dTrue\" to get the new behavior would be compatible. We could make glanceclient default to the new behavior, but the REST API itself would remain unchanged for people who don\u0027t opt-in to this new behavior.","accounts_in_message":[],"_revision_number":7},{"id":"aff8c589d51802a375583b1e224bf7e255894665","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-09-08 17:47:27.000000000","message":"Patch Set 7: -Code-Review\n\nThank you Dan for pointing out API WG guidelines.\nI think we should maintain the old behavior and as suggested by Dan we can use new input parameter and document default as well as new behavior this time.","accounts_in_message":[],"_revision_number":7},{"id":"d6d111ac4554c0d546e76c6afab1a8212066168d","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2021-09-20 13:58:51.000000000","message":"Patch Set 7:\n\nI agree with Dan, but do we want the new behaviour to be Glanceclient\u0027s default? If users are used to the old behaviour, aren\u0027t they going to see this as a bug? What if the client is used in scripts and it breaks them?\n\n\nAlso, even though that is not our responsabilitiy, we should think of the OpenStack client. What is the default behaviour for OSC? It would be nice if it were consistent with Glanceclient\u0027s.","accounts_in_message":[],"_revision_number":7},{"id":"69fe6e6f5ec3c6f60d38bc3c5d95dfdb59d65d69","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-09-20 14:06:37.000000000","message":"Patch Set 7:\n\n\u003e Patch Set 7:\n\u003e \n\u003e I agree with Dan, but do we want the new behaviour to be Glanceclient\u0027s default? If users are used to the old behaviour, aren\u0027t they going to see this as a bug? What if the client is used in scripts and it breaks them?\n\u003e \n\u003e \n\u003e Also, even though that is not our responsabilitiy, we should think of the OpenStack client. What is the default behaviour for OSC? It would be nice if it were consistent with Glanceclient\u0027s.\n\nOSC don\u0027t have support for metadef APIs 😄\nhttps://docs.openstack.org/python-openstackclient/latest/cli/decoder.html\n\nI think we will keep default behavior in glance client as well.","accounts_in_message":[],"_revision_number":7},{"id":"adef691d8251ea18413f9381f070fb47ca597052","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"date":"2021-10-12 05:53:39.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"f1a89b1e1675ed79756778b2fd983ecbff30e6b1","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2021-10-12 06:41:51.000000000","message":"Patch Set 8: Code-Review-1\n\n(6 comments)","accounts_in_message":[],"_revision_number":8},{"id":"ceefe8c389030dcb4adb672b09dd95752c7af6cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-12 07:15:26.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/cb8c9be04e944360b2f8b632b759d164 : SUCCESS in 55m 05s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5a8a6b4ddc2c49bf80f0bed037e56ca2 : SUCCESS in 1h 02m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1793cc8457f5456796e4a0a87889ce72 : SUCCESS in 6m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6684f9cbc1614e46bd4ad17eca769de0 : SUCCESS in 5m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bc3ab016818e41c19e52124e1f7721ba : SUCCESS in 6m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b045b231867e40758d05e237ed751d04 : SUCCESS in 6m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/106dce3f07fb4c99bdb49dd7cb99bfc5 : SUCCESS in 7m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8e11183e7dc3407d954823f9e32580c9 : SUCCESS in 6m 00s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/43adeb0c18334766a0ea25d4a523899b : FAILURE in 11m 20s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/3a841c9862c44bacae93b22ccdc5b6aa : FAILURE in 8m 31s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/5a1a2e1d107c4da78fcf80605b64d5b3 : FAILURE in 10m 39s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/06272e1815a5489091dc0c4adf28c3ec : SUCCESS in 5m 07s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/924bfab7e51640549472a32c0409d04d : SUCCESS in 1h 20m 10s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/5127720704fb46488796b21250bb02d5 : SUCCESS in 1h 03m 50s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/0ab258c5667c4ddd918920b665fd0a3b : SUCCESS in 1h 13m 53s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/6a8d84ae9d064bafbdfdd8294288c9f7 : SUCCESS in 1h 16m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/93871633f7484cd4ad1544d13723cd32 : SUCCESS in 47m 03s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/3e0d167b2e6b405aad5e2fdba7a13f12 : SUCCESS in 1h 18m 18s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/a1a781c7190c422d86c2551ce3771443 : SUCCESS in 35m 57s","accounts_in_message":[],"_revision_number":8},{"id":"edb29e59a2304aa34964e3c5360713830f49a806","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-10-12 13:48:06.000000000","message":"Patch Set 8: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"37e91cc3d083ae7b526cc785576b41e028c84e3d","author":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"date":"2021-10-14 06:44:04.000000000","message":"Patch Set 8:\n\n(5 comments)","accounts_in_message":[],"_revision_number":8},{"id":"80ee4a68f03152333298413905b39b18efcc57c3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"date":"2021-10-14 06:46:22.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"ae947958d27ab9ad91c46463bdd1126a1c0d45e1","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"date":"2021-10-14 06:47:45.000000000","message":"Topic 804966 removed","accounts_in_message":[],"_revision_number":9},{"id":"d59d752ea85df57c6e01aa6d28cded91ef012b14","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-14 12:54:59.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/fe1f52ccb7e4403f8c21d230c53b364f : SUCCESS in 58m 06s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4d0ffd9eb4904f91be9b879c37cb26bd : SUCCESS in 1h 15m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84bc4935f13e4a1e9322fca0f79620b0 : SUCCESS in 6m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1f46b5f4df0f4515bdbf062f62290d6c : SUCCESS in 7m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6360575cf55948248eca566cd4388ecd : SUCCESS in 6m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/07bf0f708c714551bab3c2b5257ca3a9 : SUCCESS in 7m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb9cc3b4e3f347fda1bd41be50c5725b : SUCCESS in 9m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1608a6f200f84a1482bb97ff701c0a9c : SUCCESS in 6m 55s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6d73fea1ae6247d0ab961598ca69aa5e : SUCCESS in 9m 03s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/4ff7263dd0a248caad15896d1b01b506 : SUCCESS in 9m 44s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/628694f90cfb4da0a2f61616957ac949 : SUCCESS in 10m 20s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/ad6b2160f5734867bf7cd38dd592d03f : SUCCESS in 7m 45s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/ee623d86a3a846349a2be247a27c483c : SUCCESS in 1h 31m 23s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/31c783fcbc2843269c6dd3a66ead1958 : SUCCESS in 58m 02s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/7e186fae1a3d41f2b2b66e84128ee81e : SUCCESS in 1h 10m 17s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/c3812cf868ac441884a611b180c8b63b : SUCCESS in 1h 12m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/56e925728f254b00acc4052c75ea3c21 : SUCCESS in 53m 37s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/acb78d5cfb444969ad3cce87e13ad7f7 : SUCCESS in 1h 19m 27s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/a1b75a76bbb94e39b7ee6cabd5e720df : SUCCESS in 38m 10s","accounts_in_message":[],"_revision_number":9},{"id":"76cba33829eb662defdbc78752aa116513e25066","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"date":"2022-01-03 12:07:46.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"1ac88da4aeff15efb251fdcfb235bdf5069f0f3f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-03 13:47:42.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/25ab07cb53bd4b5ebf7fa26f22c0831e : SUCCESS in 1h 00m 29s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e6329003deab4d5180a8b2a627f6e7b8 : SUCCESS in 1h 37m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ced9d0229e2540028bfddd91c7752c12 : SUCCESS in 54m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb6ad47073f84e0d90c0bdfa7f79cd26 : SUCCESS in 5m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b374035ec11142b38073d59f9577f4ef : SUCCESS in 6m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f0d2d947983a44e3b1dde5508155ebac : SUCCESS in 7m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d8a2d4aac854c0eb62b68b733d09144 : SUCCESS in 6m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/955c5f5951914d4e9114d57212b93532 : SUCCESS in 5m 50s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/827601bb4cd040888e38626df20518a7 : SUCCESS in 9m 26s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/6984d09084d54fe7bbf57e31868a202f : SUCCESS in 7m 37s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/135b731044dd49ff8d64431487b46b48 : SUCCESS in 9m 46s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/be5c37c0f7114c5cb55f5b08b9b6f9f7 : SUCCESS in 4m 49s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/73ea1ec66b8e4598b8e94bd0a23e210c : FAILURE in 1h 15m 09s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/659193cdb64e4e48bda02a0effc85602 : SUCCESS in 1h 11m 36s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/f98b1a717fcc4ffaad201b53eed6eb8d : SUCCESS in 1h 17m 43s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/0b4d56a1e06746e7b515be296ab87734 : FAILURE in 1h 18m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/da7a285463bc49ec9f446f771f0966f3 : SUCCESS in 56m 53s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/241a82e78c154f9ea4810c12ef8011a0 : SUCCESS in 1h 21m 40s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/ef02b382a7ce4c258afecbabc517dcf5 : SUCCESS in 30m 39s","accounts_in_message":[],"_revision_number":10},{"id":"fdd8353458be01d2d63e72197afb60ec5e9672b0","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-01-04 04:30:10.000000000","message":"Patch Set 10: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":10},{"id":"bd2d51204682c58598313bccd8d8e78df594c885","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-01-05 03:20:19.000000000","message":"Patch Set 10:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/804966/804966/10/check-not-vote/ext-glance-zuul/2219631 : FAILURE in 17s","accounts_in_message":[],"_revision_number":10},{"id":"11b8b5f9f09f9a3de74005f93e3ff8420434a54d","author":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"date":"2022-01-05 13:34:16.000000000","message":"Patch Set 10:\n\n(3 comments)","accounts_in_message":[],"_revision_number":10},{"id":"ea297ea34aa0c23a00d27ea39f42df4fddc4dfaa","author":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"date":"2022-01-10 11:33:22.000000000","message":"Patch Set 10:\n\n(2 comments)","accounts_in_message":[],"_revision_number":10},{"id":"bc1459a7fa1578888dbf263742433209688956b4","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-01-10 13:27:08.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"dc542f9e16a69d3711f098de2b3a75d83ddbba9b","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2022-01-17 16:22:54.000000000","message":"Patch Set 10: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":10},{"id":"04de63229ad10353c96af0f396628143bc854efb","author":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"date":"2022-01-27 15:54:02.000000000","message":"Patch Set 10:\n\n(3 comments)","accounts_in_message":[],"_revision_number":10},{"id":"6f6346e6f813d6612779e051c0f3a2375b69200a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"date":"2022-01-27 15:58:37.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"b90972212ced58079f88d1f7e9ba584a1e4c9e65","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"date":"2022-01-27 16:00:28.000000000","message":"Topic 804966 removed","accounts_in_message":[],"_revision_number":11},{"id":"da82ab61f63bfa90c97f1f46c5cb9330604976cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-27 17:23:28.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/23ed77e415a64f139f24863d12a50b26 : SUCCESS in 8m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/5537a958d6504338a891b5bf71e55459 : SUCCESS in 50m 45s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/59d3f2d6243b48abb8f696b535ae9b7a : SUCCESS in 1h 19m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/289d9fe7e5a742f4be1aea65fa590748 : SUCCESS in 40m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ca0a76982714281b9e26716557a879c : SUCCESS in 7m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f376d6b696894b81a97b917e10f9040c : SUCCESS in 5m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d51516c8b0f340e78ae7534ce0b6e44b : SUCCESS in 6m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3501186934e241f493183c0d37d14b3a : SUCCESS in 6m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e8d0bda5974a466ba3b545c19be6890f : SUCCESS in 7m 36s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d017d7b8f2854d239b83d8c411de8f9b : SUCCESS in 9m 08s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f93a96740e614c8a8d1ac8ae03ec773d : SUCCESS in 9m 48s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/14802ec5d40449a1ad915b78cc6eb871 : FAILURE in 9m 33s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/0973abe487d043ab9169af923061b608 : SUCCESS in 5m 46s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/7373d9f53421426f86b638e6618f16b1 : SUCCESS in 1h 02m 24s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/5ae2b9a3b47e406aa25a9e960f9a94fa : SUCCESS in 1h 16m 36s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/6743dd3163354c26bd1949fa251c3374 : SUCCESS in 1h 00m 29s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/31b7b9a28a624838a18b9b48b4005b84 : FAILURE in 1h 18m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bde10a0da78d450caa0f18586f4ef4da : SUCCESS in 41m 21s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/d0781814a92843868d52b8753a7f7269 : SUCCESS in 1h 14m 11s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/8d729177fb10417da1e1c56c8f030ef4 : SUCCESS in 27m 30s","accounts_in_message":[],"_revision_number":11},{"id":"131e256b9aa9b486ad1016d057f278ca40326ed4","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-01-28 03:29:46.000000000","message":"Patch Set 11:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/804966/804966/11/check-not-vote/ext-glance-zuul/899a439 : FAILURE in 23s","accounts_in_message":[],"_revision_number":11},{"id":"7a08b2fd3ca4870180cb02090b604255aa32002e","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-01-28 05:17:05.000000000","message":"Patch Set 11: Code-Review-1\n\n(7 comments)","accounts_in_message":[],"_revision_number":11},{"id":"7039c726949337a7ea70165041eb6e787dafc37f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"date":"2022-01-31 14:23:05.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"b7d165d170a84a600af1a35212886f0a1114c88d","author":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"date":"2022-01-31 14:23:37.000000000","message":"Patch Set 12:\n\n(6 comments)","accounts_in_message":[],"_revision_number":12},{"id":"47de81dc88ebd5201603f919708801ff76a3bc52","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"date":"2022-01-31 14:23:42.000000000","message":"Topic 804966 removed","accounts_in_message":[],"_revision_number":12},{"id":"0126e7e1ff64c24dd1793fd685edd705a1188924","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-01-31 14:31:08.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"378457489b7835fe0c7de82e26f50b8de8cca94d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-31 15:57:00.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/2871fe580a77488695e68b3ffaf169f1 : SUCCESS in 4m 29s\n- grenade https://zuul.opendev.org/t/openstack/build/563e36030e7947a89c3025227095ec13 : SUCCESS in 51m 43s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2afce387640841c5b5a40ec1c4a6c282 : SUCCESS in 54m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/106eab811c774930b59777880f8846b5 : SUCCESS in 38m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1869862e36f04b0696c45ffa485be060 : SUCCESS in 4m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e6bd88fbccc54209a44cef25a98e8b96 : SUCCESS in 5m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a7617c67ee3344cbac79af999cfb6454 : SUCCESS in 5m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7e93179023a43e28ed252d9cba273da : SUCCESS in 5m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/62f5680d62ae4057894785aabdb7d3ce : SUCCESS in 6m 10s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6b689a16bf1c426690330c5961a9c68f : SUCCESS in 7m 50s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c7902c68d3f741d5846d60a05f28c494 : SUCCESS in 10m 32s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/644c88441a3240428ed3d63661238571 : SUCCESS in 11m 44s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/193ea796dc7541cb831bde3ecf478fc2 : SUCCESS in 4m 28s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/3fdfdd9490664fe7aecc167342bd619e : SUCCESS in 1h 30m 36s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/24f7980b3a1641839607277381551c17 : SUCCESS in 1h 17m 18s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/9df1fcd3552e487c835f4de16945135f : SUCCESS in 1h 27m 54s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/4cb96ca3dd4d465a844e1d7802d1abac : SUCCESS in 1h 26m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/eac7de2083e44f5d8507408d15888eab : SUCCESS in 46m 57s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/c354ff70ef8448c6a55e986c11a9f817 : SUCCESS in 1h 16m 28s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/8efb5cdc6a884a499032a5ff2d4e126c : SUCCESS in 39m 18s","accounts_in_message":[],"_revision_number":12},{"id":"41d898ee2bbe64a26c24bcc494814b86b3cb8cb8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"date":"2022-02-10 09:56:24.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"9a08f07f22b9a7e6f2404d6f3e8440858e6b86a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-10 11:26:45.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/c8aa25ad35124f968740f28110357d93 : SUCCESS in 5m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/2bd1e2bbaeec43bb96941431272f1d2f : SUCCESS in 1h 00m 51s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/184c8cc05d9444548cd438ae2f068f7c : SUCCESS in 1h 07m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0e009db2c82647fbaef9c9b654e6c234 : SUCCESS in 42m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56425d6963a04e728aedb7415908edb8 : SUCCESS in 4m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5c96d3a9c0c24d0f96e67d67dc26d823 : SUCCESS in 6m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1d05c0e99f774e788075230ba8050e41 : SUCCESS in 6m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0be811985ee4801966a2648ec95d149 : SUCCESS in 6m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/44445bc6f63846f3be60c6d44e9affcb : SUCCESS in 5m 05s\n- openstack-tox-functional-py36-fips https://zuul.opendev.org/t/openstack/build/d58e0530aebb41d7ab59d071caabed07 : SUCCESS in 11m 31s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/2920c4d16a5a4a028c462c9da0e8fb6a : SUCCESS in 9m 26s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/3d342d28bfa6463cb6a9fc45764a5402 : SUCCESS in 8m 33s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/a64189df99654807942258974fd6ec0f : SUCCESS in 5m 02s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/7ecb7bc1d4b0493895d38c6ead87c24a : SUCCESS in 54m 31s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/860a799def83414c8abe776fb2616693 : SUCCESS in 1h 07m 19s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/f3da27ccde67440680cd5d829bcc63c0 : SUCCESS in 1h 03m 10s\n- glance-multistore-cinder-import-fips https://zuul.opendev.org/t/openstack/build/29973c73fd124794977cc0db03493f4a : FAILURE in 1h 26m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/413946cb31764f7392349438491e467d : SUCCESS in 42m 24s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/c43ca417e6c5429fa1e6963a9bfc3e57 : SUCCESS in 1h 00m 25s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/bf1c0b7901f446798667aeba4f1c785c : SUCCESS in 29m 23s","accounts_in_message":[],"_revision_number":13},{"id":"368d0d81162d2d861488cb50ca883bea0f95af77","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-02-11 14:11:24.000000000","message":"Patch Set 13: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":13},{"id":"f527f96e90e46df7746a8c07353ac65429e3cb19","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"date":"2022-02-14 09:09:33.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"1465bd0e2e383d58d8cf8b8c5b86bb45dffaf5c0","author":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"date":"2022-02-14 09:09:41.000000000","message":"Patch Set 13:\n\n(2 comments)","accounts_in_message":[],"_revision_number":13},{"id":"bd2fd02427efb36fcde8e26222e1a101b90fb3bb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-14 11:01:32.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/40abbd44fa9e4b65b8c421577a13d541 : SUCCESS in 4m 39s\n- grenade https://zuul.opendev.org/t/openstack/build/601557d16b9e408ba4c07fb231433747 : SUCCESS in 45m 02s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/401fb144429c44cf8aef2e93f2adf5a1 : SUCCESS in 57m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e582941943684b36b70c384f461ef47e : SUCCESS in 40m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c0b47f152e4d451ba06a278732915749 : SUCCESS in 4m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3afd04ae222845c783d00250fe963bf6 : SUCCESS in 5m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dd2f99038e944a498f4f2aaaf2968941 : SUCCESS in 5m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c50e4ba7f9242419bb737b29f8f6de0 : SUCCESS in 6m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3c9e4ac347ca47c6af212dca6a8c8bbb : SUCCESS in 5m 24s\n- openstack-tox-functional-py36-fips https://zuul.opendev.org/t/openstack/build/22168652644843f88976645f545edb6e : SUCCESS in 11m 50s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/6b23ecb3629b4e9e97b5426c176d40f8 : SUCCESS in 9m 11s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/1c22d0677b494e1c90577cb294b7b215 : TIMED_OUT in 31m 07s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/b76f07003e844d6f89ac7fd4f66016b2 : SUCCESS in 4m 43s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/7c6867848b9f4e4588ef8909ff9657af : SUCCESS in 1h 12m 35s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/8c7aa16cc52f48ecad1ac285aaf7f604 : SUCCESS in 1h 03m 29s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/d44cd54da1d0411a9b0cb3790f497866 : SUCCESS in 1h 19m 15s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/390e5242f2134efe919be8f34e779654 : SUCCESS in 1h 24m 53s\n- glance-multistore-cinder-import-fips https://zuul.opendev.org/t/openstack/build/1b2bbf4a36a84d1884aa78f74400353a : FAILURE in 1h 44m 59s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cd9591ca28cc4957b1cab6eb2f431731 : SUCCESS in 44m 58s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/3e8150507dd9446e9f8fff9a50b7c31f : SUCCESS in 1h 18m 23s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/1488ae81992f463ca3270bc14c5db59d : SUCCESS in 28m 58s","accounts_in_message":[],"_revision_number":14},{"id":"d422d8f869aeffb751d9fe92f7dff93fc54e3496","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-02-14 14:06:56.000000000","message":"Patch Set 14: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":14},{"id":"e8d9bde49cfbce19ac843d03d022d6cc45c501c7","author":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"date":"2022-02-15 07:03:29.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"33c4b4be44160f8ab0ed1cff3e8b34a9aec9f851","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"date":"2022-02-15 07:04:07.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"d873422ca391742c31806184739600b486105fa9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-15 08:24:25.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/eaa9450f1f5f4d15859366633945cd44 : SUCCESS in 5m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/30884f1c6ab74bd5ad6db8e67415c694 : SUCCESS in 47m 38s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0c20883c2fd44b0f83e84c9a6c200931 : SUCCESS in 1h 09m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2d24873d0a864210b26ef64849fb15d2 : SUCCESS in 45m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7773b160486b4d0c8265f0df8a2937b0 : SUCCESS in 5m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fa2a3cd9b1be4fe19d0570a5fc5e06c1 : SUCCESS in 5m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e3cc91ddb4534c4c8ba8979ac249ea8e : SUCCESS in 6m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7eeec91dd15949fc9f5fa5a73696298a : SUCCESS in 6m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ac201c5762fd4617ae5d0905588f98c2 : SUCCESS in 4m 56s\n- openstack-tox-functional-py36-fips https://zuul.opendev.org/t/openstack/build/312c92b3e2cf4809a496b9f3c09f2254 : SUCCESS in 12m 08s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/3ef53d3f6fe24e57a8fa607e40dab821 : SUCCESS in 8m 41s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/a722cb5401f24c40be963b86b3c0c023 : SUCCESS in 10m 02s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/9bdca83f56b6402bb0e84219a15231ff : SUCCESS in 5m 22s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/41df1bd5a6c5457c85aa7f26e265a76a : SUCCESS in 1h 02m 54s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/c9f43b1c02c648c389adc416c5f637bf : SUCCESS in 1h 13m 59s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/81de72c7b28a46278ecf2b810c7c1238 : SUCCESS in 1h 00m 16s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/347f1147674f4bca8a38cdd6cb47842e : FAILURE in 1h 16m 49s\n- glance-multistore-cinder-import-fips https://zuul.opendev.org/t/openstack/build/b7120a0d36544c9a8d69d67f0b7b8a77 : FAILURE in 1h 16m 58s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/317205491d494e69ab757642c159a02f : SUCCESS in 49m 39s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/2f8f9704e24f4fa69cdfbe0906c11f48 : SUCCESS in 1h 15m 33s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/c0b96052450f4b0c8be995298744fcc0 : SUCCESS in 28m 49s","accounts_in_message":[],"_revision_number":15},{"id":"6892a0f0ea4172cf162ea6552a6f72e15333199e","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-02-15 14:17:48.000000000","message":"Topic 804966 removed","accounts_in_message":[],"_revision_number":15},{"id":"ebb000d69d6a9757da0a3cbd0060b380ba8be552","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-02-15 14:17:55.000000000","message":"Topic set to bp/append-tags","accounts_in_message":[],"_revision_number":15},{"id":"2bd1db780d9a10896fe9116ce215ad5d0492db0d","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-02-15 14:18:14.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"b92373fc97e186dfb0562bbae2756805a70697be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-15 16:19:40.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/4858677b74e94f78a0f1df1aef0c12f5 : SUCCESS in 5m 17s\n- grenade https://zuul.opendev.org/t/openstack/build/2aaf54fc06bf40c18cf6a55741f5d505 : SUCCESS in 53m 45s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/65be3301edd04d2abda4009103ed5573 : SUCCESS in 1h 29m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/16e07c2faa6f4fedb63ec56e426b9786 : SUCCESS in 1h 01m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ccd3ae6a250d41b7bb3ca2ceed72ab17 : SUCCESS in 5m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/050c40a66ea04ecb82c27ee2c0377703 : SUCCESS in 6m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e1c278ceb0434dba97e8a365a4cf85b8 : SUCCESS in 6m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d113461b29b04c2db32029b9c6f12d19 : SUCCESS in 7m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c6979a27ceb14f7a9629ea6a4d19581a : SUCCESS in 6m 24s\n- openstack-tox-functional-py36-fips https://zuul.opendev.org/t/openstack/build/fb5bae96535e4bd4b5476bcb2430547e : SUCCESS in 11m 37s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/62972acb70114e62bc6cfc889e7e56e6 : SUCCESS in 10m 26s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/93cf3970e73441cbad9941b91924b244 : SUCCESS in 9m 33s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/e9d596c310bc45c286942d529c76a543 : SUCCESS in 5m 17s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/09c05438a1064b4895e1241fc8d7389b : FAILURE in 1h 15m 17s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/e5ec3074c4694912be18df59261b5e0b : SUCCESS in 1h 10m 12s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/a1379e6cf98449f9950548d8e464b011 : SUCCESS in 1h 09m 46s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/3531a4a4dc4241c6b46ea3881e2d53a7 : SUCCESS in 1h 36m 48s\n- glance-multistore-cinder-import-fips https://zuul.opendev.org/t/openstack/build/4d3754c866e641e8a72afec92f3ef00b : FAILURE in 1h 51m 17s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b3ad108fd0684c8f88aef25d03ccb026 : SUCCESS in 1h 06m 36s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/d89756729bc6469193a746e637591c5e : SUCCESS in 1h 25m 30s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/8a97cbcdaa47489d82e3f7133f5511cb : SUCCESS in 31m 32s","accounts_in_message":[],"_revision_number":15},{"id":"1038ec114a430f5b9547ea2461c5880710ff0272","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2022-02-15 23:24:27.000000000","message":"Patch Set 15: Code-Review+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":15},{"id":"5d62ece163b8d8f30bd0353a0143d908633f391c","author":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"date":"2022-02-16 05:48:44.000000000","message":"Patch Set 15:\n\n(2 comments)","accounts_in_message":[],"_revision_number":15},{"id":"5b7e3f74ab7b5e71cf7efa761de5aef9099f07d4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"date":"2022-02-16 05:54:21.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"3f1df898240641c74f448c86222f0e05a3cf0dbb","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-02-16 05:56:42.000000000","message":"Topic 804966 removed","accounts_in_message":[],"_revision_number":16},{"id":"cd18bcd9b906d6e6b54ea665acb76e9758bc76b7","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-02-16 05:56:50.000000000","message":"Topic set to bp/append-tags","accounts_in_message":[],"_revision_number":16},{"id":"552a5b4dca6df1a47b362add41f6ca11d5e7f650","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-16 07:32:05.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/b08b2687b0b741f48c639f19179a5a52 : SUCCESS in 5m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/20314ab6c576449bb6c50adbdf4537c0 : SUCCESS in 56m 34s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c12f79228ace47f3afa79892a8e2a942 : SUCCESS in 1h 14m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/53cdffa7c7b349d99f8727aa5095e9e0 : SUCCESS in 47m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/24d32ad346934f40b38088bb798cc359 : SUCCESS in 4m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/30a722a3fca348f88cfaee6f294a24a1 : SUCCESS in 5m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4bf603c1f49948c5a3c71768a590a84a : SUCCESS in 5m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17c9481d39e244db98da0f210d6958c9 : SUCCESS in 5m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a02339e88ad34728b1212b91182d89d7 : SUCCESS in 5m 13s\n- openstack-tox-functional-py36-fips https://zuul.opendev.org/t/openstack/build/432aff789b84447ba3166404c2e101ab : SUCCESS in 12m 00s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b7a5f03caa584e61b51eb2c652b9a43b : SUCCESS in 8m 54s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/1f3792d8105041599de651f85428eef7 : SUCCESS in 8m 34s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/59d09eaf5eb04545b00a28800edb01b0 : SUCCESS in 4m 46s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/997347b675db43c0aeea9e8662f69cc3 : SUCCESS in 1h 08m 59s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/3133b5f704b447e68d70af329890012f : SUCCESS in 1h 10m 08s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/1a58214b20704e9da611421182af2580 : SUCCESS in 1h 12m 05s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/75f4e6aae81f478b9f7f8c9812ae4e97 : FAILURE in 1h 17m 43s\n- glance-multistore-cinder-import-fips https://zuul.opendev.org/t/openstack/build/51a13927fcb74f669c58c16c1ac514ca : FAILURE in 1h 33m 06s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/86b5ff0ff0804d9485af404fe1c54598 : SUCCESS in 1h 00m 40s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/aa7b8455dd034c159412607bbf6576b6 : SUCCESS in 1h 05m 43s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/323d452cac2a4bf69e1afcaded988996 : SUCCESS in 28m 46s","accounts_in_message":[],"_revision_number":16},{"id":"c4f666ea8450d677797eedcf558fe185a1dff8a4","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-02-16 07:48:48.000000000","message":"Patch Set 16:\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"7a1db6b40fd57ea73e8fbcedac5cbda10cb62ff5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-16 10:02:56.000000000","message":"Patch Set 16:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/62da813de65c490dbf989877bda94fdd : SUCCESS in 6m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/baec4667db0e405681ee78e9a821e88a : SUCCESS in 1h 00m 53s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/83447a92e4f14f8fadb1bbf65dc4c7cd : SUCCESS in 1h 09m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2a76d9220e2c4ef58d3b1d09645c909d : SUCCESS in 44m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a96f8d7fbc1a415d947680077a5a79bb : SUCCESS in 5m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9f12569b67b84b4eb8632e16459e06ff : SUCCESS in 6m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/55323f4e1aea4c07b986c5987ac1836e : SUCCESS in 5m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9a337abf8924afbb734434ac8e043fc : SUCCESS in 7m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4c0b52ccfef6427faa02fe270a7b7554 : SUCCESS in 4m 36s\n- openstack-tox-functional-py36-fips https://zuul.opendev.org/t/openstack/build/380fdc1cfdb5469ea9719d6e181db3ae : SUCCESS in 10m 03s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/47891527266544f5b4889f66f84d1e2d : SUCCESS in 7m 53s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/360da0c87c0044b3a5459651faae91eb : SUCCESS in 11m 14s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/74f291b05c214aa5ac832208a1b3762f : SUCCESS in 5m 47s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/3383c85bb1aa407bb9154d8b31437247 : SUCCESS in 1h 14m 10s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/ce74c840bec64d3788d1feb60f3ef27f : SUCCESS in 59m 55s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/6fdef3301c68435ab2061dc86ddc5830 : SUCCESS in 56m 54s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/40d0015d18ea40e597dc294a579e8221 : FAILURE in 18m 59s\n- glance-multistore-cinder-import-fips https://zuul.opendev.org/t/openstack/build/4b25ae1a0d5d49529659976908c15b8e : TIMED_OUT in 2h 08m 04s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8cbbb2a040b547dda39b2b30912e5c22 : SUCCESS in 41m 43s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/8cdb9797af464a5aa4be89709ac50479 : SUCCESS in 1h 07m 25s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/aebd375775c947bfbdeb314f649cf1f6 : SUCCESS in 26m 49s","accounts_in_message":[],"_revision_number":16},{"id":"c8feb36f0c5ee7893e07be7567f21f921a4ec9cc","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2022-02-16 10:24:33.000000000","message":"Patch Set 16:\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"4b6d75df84229fa774d8520bb32a52727c93e822","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-16 12:04:19.000000000","message":"Patch Set 16:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/a7bdde97a48c44fd919fc75ec3f88a29 : SUCCESS in 5m 27s\n- grenade https://zuul.opendev.org/t/openstack/build/83ce686ef6344f86a6db52556129332e : SUCCESS in 49m 29s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/605aa408bde2414e9b1467779505d974 : SUCCESS in 1h 06m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b555d10d085644118c151db1acd02175 : SUCCESS in 47m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e31400dd7d7f45698e61573d7ed9eecd : SUCCESS in 5m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/750db97e6f0b4c48b6fa396588442d54 : SUCCESS in 5m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/83f41105707f454d89aa85534f157159 : SUCCESS in 6m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a58c6d48561541be9cdd20a57196f08f : SUCCESS in 7m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7a9f34a196bf4433b7ee3db0ae6b4b33 : SUCCESS in 6m 23s\n- openstack-tox-functional-py36-fips https://zuul.opendev.org/t/openstack/build/d5d5ee6360ad401792509c9bee6c8c77 : SUCCESS in 12m 29s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/082c3ea5c96c481a9944e0404fb1599e : SUCCESS in 10m 57s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/330211a971474cba8213dfa4ea09499d : SUCCESS in 9m 41s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/d4e126f28f96449aa1f11a4c774e61df : SUCCESS in 5m 57s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/1644d66dac9f45fa853d70590ff1de87 : SUCCESS in 1h 11m 02s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/c7d9a6ad51e54cbc8e53eaa2df338736 : SUCCESS in 1h 04m 36s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/28febaa81f0d4821a9a3dea59fbe6ecd : SUCCESS in 1h 18m 49s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/294aa7883aa148db8706491de6c37457 : FAILURE in 27m 04s\n- glance-multistore-cinder-import-fips https://zuul.opendev.org/t/openstack/build/f2d581207337479ea2050f99596fe5c5 : FAILURE in 1h 37m 26s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2a2052c408cb421d86d3819b73e418b3 : SUCCESS in 55m 07s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/ac2c97f88d324ddfa8f3caf89eb7d834 : SUCCESS in 1h 32m 32s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/8d44994bdcce40ae91e172db5a82c8a7 : SUCCESS in 29m 50s","accounts_in_message":[],"_revision_number":16},{"id":"07e4fce753960b0a5ef5dee5f33a465eb618fad0","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-02-16 13:47:20.000000000","message":"Patch Set 16:\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"dcdea032d2ca094aa915872ef68e10e8721d9db0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-16 15:16:33.000000000","message":"Patch Set 16:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/7935a9217f734fb39ac2d37f661a873e : SUCCESS in 5m 28s\n- grenade https://zuul.opendev.org/t/openstack/build/862eb6e3f6e44ea986edba58a706eef0 : SUCCESS in 45m 12s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/bad59a7184e9407fb29f22020cef5da2 : SUCCESS in 1h 15m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cf2195bd4fdf4932a47721d0ebdaa0f1 : SUCCESS in 41m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6bce2cfa6a2d4b9084644509bb927f84 : SUCCESS in 5m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8e7c514f12ee4d949632350843341b92 : SUCCESS in 4m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c09a58f1d51846339c28ec00e88942da : SUCCESS in 6m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c02ded66eafb4f73b715e9f79f7cd1f4 : SUCCESS in 7m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7f929c46d31f42ed9e2b42974e119c93 : SUCCESS in 6m 13s\n- openstack-tox-functional-py36-fips https://zuul.opendev.org/t/openstack/build/4b33958b5d7544ee9eab60eaafc73653 : SUCCESS in 10m 41s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f4de8a1d10cc459186c3ab4b6895af59 : SUCCESS in 9m 21s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/3cd3887750ae4ac59e143a8d63a14713 : SUCCESS in 10m 16s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/1106de1eb60c4d9189f7018ed7a12438 : SUCCESS in 5m 16s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/412ca8ae70ba4d3aae9ed23c87199f89 : SUCCESS in 1h 05m 46s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/b5c7cb220c2d4e16ba95e25c3be9bf5f : SUCCESS in 1h 23m 20s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/2f275920eaf64616b844064416ed21c5 : SUCCESS in 1h 20m 25s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/ec70cbbbe2564cfea8452302f3a19421 : FAILURE in 1h 10m 54s\n- glance-multistore-cinder-import-fips https://zuul.opendev.org/t/openstack/build/6d9cafe768c64404bf8579d61fb3eebd : FAILURE in 1h 19m 48s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9c961190f070433b8b02dcc08e465df0 : SUCCESS in 52m 40s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/54ef90635af94a4492394d0dfa8c5143 : SUCCESS in 1h 10m 54s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/df06f307e3d84e039dcb8b9a82c2acef : SUCCESS in 27m 54s","accounts_in_message":[],"_revision_number":16},{"id":"5ccd4b620dfa5b46e1894e02852a90b30b87d691","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-02-16 15:20:58.000000000","message":"Patch Set 16:\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"82a0bb86129d07a2410989e386948848fb343b42","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-16 16:57:40.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/e37a70be3af4434fb602f66314540a36 : SUCCESS in 6m 12s\n- grenade https://zuul.opendev.org/t/openstack/build/901ec5f9566a47b394f11e83e816f0d7 : SUCCESS in 51m 51s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e53f75c258144fb3938cf92db2c1de34 : SUCCESS in 1h 15m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1361d38985324415862bed58fccc43ad : SUCCESS in 49m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c7c9e8c364114401ab0e835288f99a10 : SUCCESS in 5m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/48cb95f9a68e497a9d2cfa8d6f440994 : SUCCESS in 5m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dacee8520f314822a1cb256c9486d16f : SUCCESS in 6m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38c70f0507034fa8b1d6f9580db72e27 : SUCCESS in 7m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8a909658f7a54bb89818ecb133599728 : SUCCESS in 5m 28s\n- openstack-tox-functional-py36-fips https://zuul.opendev.org/t/openstack/build/49dffad507f646b0b1064d7a9da18a1f : SUCCESS in 11m 13s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1279d18d015d4bd4ae5decb25a3bcd89 : SUCCESS in 9m 11s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/3389b45b73ae49dbaf8f749793ae1269 : SUCCESS in 8m 52s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/a7c883054670464589b0ea645945b7d4 : SUCCESS in 5m 42s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/1cc59a88346142fcab3b7ee77fe201b0 : SUCCESS in 1h 07m 32s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/ad5788e71c904eacbd95901f458f3724 : SUCCESS in 1h 03m 51s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/3000593ed4f04720ba00df2d030dac23 : SUCCESS in 1h 12m 46s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/f69c7aa45d1043ad91dae5abe13d7ad5 : SUCCESS in 1h 30m 30s\n- glance-multistore-cinder-import-fips https://zuul.opendev.org/t/openstack/build/d4daf469a9ba4d9eac65ac9ce8c33c69 : FAILURE in 1h 22m 20s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d785f08964ef4e7d98d82d723fd583bc : SUCCESS in 43m 38s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/091abff0868c47779fed98f818a28390 : SUCCESS in 1h 18m 10s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/8e3aad65280442c8af1cb5089c487e90 : SUCCESS in 25m 51s","accounts_in_message":[],"_revision_number":16},{"id":"ea50ed81bf23cefbaac69a886bb22940511a96a4","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-02-16 17:07:12.000000000","message":"Patch Set 16: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"e4807191a39d6586806a6d53d50e8be55a71bd42","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2022-02-17 14:05:45.000000000","message":"Patch Set 16: Code-Review+1","accounts_in_message":[],"_revision_number":16},{"id":"af42a43c419774a5babc4027fc32b3b5b13a0585","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2022-02-17 14:45:37.000000000","message":"Patch Set 16: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":16},{"id":"4a03ac8ab6d554a0b5877ef355f8dd626df26d33","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 14:46:20.000000000","message":"Patch Set 16: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":16},{"id":"bf4dd6117bc3845a303104b99b964fc100ee303c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 16:01:16.000000000","message":"Patch Set 16: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/8c2ea50b61c24233ba8cc2051bf06fc8 : SUCCESS in 5m 55s\n- grenade https://zuul.opendev.org/t/openstack/build/ebf925f60a3a420b950014982034db52 : SUCCESS in 1h 02m 58s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/66dd3d6c49f541c5bf74418ce74c50a7 : SUCCESS in 1h 14m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/44be700d250643568eb8ef2156e31815 : SUCCESS in 45m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/45c190a1cdc54f50827188b7553fcf75 : SUCCESS in 5m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7a7f5b8a5639492ca30df31d9bd97c6d : SUCCESS in 5m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/61f0eab29c9940eeb2c329f1d5161a02 : SUCCESS in 5m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37eab89dce9f4358914000b7b0feaa30 : SUCCESS in 6m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a581f7e8a18d4162a502bec627f04c5b : SUCCESS in 6m 14s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/42b6e408c2cb4f79879c940393794ab6 : SUCCESS in 10m 06s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d89dfaeba1e7454d8ebf28d308874532 : SUCCESS in 11m 10s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/b6d56599a8264318862a03d924309b38 : SUCCESS in 1h 02m 53s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/b1fee2c0da0a43d9a3c421c036cebfc6 : SUCCESS in 1h 09m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4d1c710049d844adbe1494ed58d64ddc : SUCCESS in 43m 23s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/48533d1a6c0e45a3b5d18ed01edfe52b : SUCCESS in 1h 09m 18s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/44d24d2a24534a66b1aeb15839cc2841 : SUCCESS in 28m 50s","accounts_in_message":[],"_revision_number":16},{"id":"9337f0f9a4cdcf2de659010834508026c7b6b394","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 16:01:16.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":16},{"id":"00c52270aeecf0b38af24e74854e0e834b7dd448","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 16:03:32.000000000","message":"Patch Set 16:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/265d69b17acd4b3aaaa7a6c0a5a4dc8f : SUCCESS in 1m 22s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e24d6cb465ed4d06947405e5541c8273 : SUCCESS in 1m 29s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dda5c055cffd4c4c87acbc2be2e7f2c8 : SUCCESS in 1m 16s","accounts_in_message":[],"_revision_number":16}],"current_revision_number":16,"current_revision":"2a9a4c8e0e095f3cd788e2f24b2dce508538e0d5","revisions":{"8ef00d2a465293b3bb4eab4ceda40bac4b886d23":{"kind":"REWORK","_number":1,"created":"2021-08-18 08:33:22.000000000","uploader":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"ref":"refs/changes/66/804966/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/66/804966/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/804966/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/804966/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/804966/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/66/804966/1"}}},"commit":{"parents":[{"commit":"284189222923f60bd883dd95ff0354b80850a4ce","subject":"Merge \"Clean out deprecations from paste.ini\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/284189222923f60bd883dd95ff0354b80850a4ce"}]}],"author":{"name":"Mridula Joshi","email":"mrjoshi@redhat.com","date":"2021-08-18 08:21:53.000000000","tz":0},"committer":{"name":"Mridula Joshi","email":"mrjoshi@redhat.com","date":"2021-08-18 08:31:08.000000000","tz":0},"subject":"It was observed that md-tag-create-multiple (/v2/metadefs/namespaces/{namespace_name}/tags)  API overwrites existing tags for specified namespace rather than creating new one in addition to the existing tags. This patch resolves the issue by not deleting the previous data and adding the new tags to existing ones.","message":"It was observed that md-tag-create-multiple\n(/v2/metadefs/namespaces/{namespace_name}/tags)  API overwrites existing tags\nfor specified namespace rather than creating new one in addition to the existing tags.\nThis patch resolves the issue by not deleting the previous data and\nadding the new tags to existing ones.\n\nCloses-Bug: #1939169\nChange-Id: I29448746b14c542e5fbf0283011968ae1516642e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8ef00d2a465293b3bb4eab4ceda40bac4b886d23"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8ef00d2a465293b3bb4eab4ceda40bac4b886d23"}]},"branch":"refs/heads/master"},"6dda5a39fd596205e003001f764fb4847a189924":{"kind":"REWORK","_number":2,"created":"2021-08-23 14:04:03.000000000","uploader":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"ref":"refs/changes/66/804966/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/66/804966/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/804966/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/804966/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/804966/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/66/804966/2"}}},"commit":{"parents":[{"commit":"284189222923f60bd883dd95ff0354b80850a4ce","subject":"Merge \"Clean out deprecations from paste.ini\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/284189222923f60bd883dd95ff0354b80850a4ce"}]}],"author":{"name":"Mridula Joshi","email":"mrjoshi@redhat.com","date":"2021-08-18 08:21:53.000000000","tz":0},"committer":{"name":"Mridula Joshi","email":"mrjoshi@redhat.com","date":"2021-08-23 13:59:28.000000000","tz":0},"subject":"Fix overwriting of existing tags while creating new tags","message":"Fix overwriting of existing tags while creating new tags\n\nIt was observed that md-tag-create-multiple\n(/v2/metadefs/namespaces/{namespace_name}/tags)  API overwrites existing tags\nfor specified namespace rather than creating new one in addition to the existing tags.\nThis patch resolves the issue by not overwriting the previous tags but appending the\nnew tags to existing ones.\n\nCloses-Bug: #1939169\nChange-Id: I29448746b14c542e5fbf0283011968ae1516642e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6dda5a39fd596205e003001f764fb4847a189924"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6dda5a39fd596205e003001f764fb4847a189924"}]},"branch":"refs/heads/master"},"57c2c4dddcc93a096304cbf577bbc608fdcd52d3":{"kind":"REWORK","_number":3,"created":"2021-08-24 05:48:49.000000000","uploader":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"ref":"refs/changes/66/804966/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/66/804966/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/804966/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/804966/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/804966/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/66/804966/3"}}},"commit":{"parents":[{"commit":"284189222923f60bd883dd95ff0354b80850a4ce","subject":"Merge \"Clean out deprecations from paste.ini\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/284189222923f60bd883dd95ff0354b80850a4ce"}]}],"author":{"name":"Mridula Joshi","email":"mrjoshi@redhat.com","date":"2021-08-18 08:21:53.000000000","tz":0},"committer":{"name":"Mridula Joshi","email":"mrjoshi@redhat.com","date":"2021-08-24 05:38:08.000000000","tz":0},"subject":"Fix overwriting of existing tags while creating new tags","message":"Fix overwriting of existing tags while creating new tags\n\nIt was observed that md-tag-create-multiple\n(/v2/metadefs/namespaces/{namespace_name}/tags) API overwrites\nexisting tags for specified namespace rather than creating new one\nin addition to the existing tags.\nThis patch resolves the issue by not overwriting the previous tags\nbut appending the new tags to existing ones.\n\nCloses-Bug: #1939169\nChange-Id: I29448746b14c542e5fbf0283011968ae1516642e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/57c2c4dddcc93a096304cbf577bbc608fdcd52d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/57c2c4dddcc93a096304cbf577bbc608fdcd52d3"}]},"branch":"refs/heads/master"},"9299739790e6d417e517ed118d1bcb47dce0146e":{"kind":"REWORK","_number":4,"created":"2021-08-26 04:40:27.000000000","uploader":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"ref":"refs/changes/66/804966/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/66/804966/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/804966/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/804966/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/804966/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/66/804966/4"}}},"commit":{"parents":[{"commit":"284189222923f60bd883dd95ff0354b80850a4ce","subject":"Merge \"Clean out deprecations from paste.ini\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/284189222923f60bd883dd95ff0354b80850a4ce"}]}],"author":{"name":"Mridula Joshi","email":"mrjoshi@redhat.com","date":"2021-08-18 08:21:53.000000000","tz":0},"committer":{"name":"Mridula Joshi","email":"mrjoshi@redhat.com","date":"2021-08-26 04:36:17.000000000","tz":0},"subject":"Fix overwriting of existing tags while creating new tags","message":"Fix overwriting of existing tags while creating new tags\n\nIt was observed that md-tag-create-multiple\n(/v2/metadefs/namespaces/{namespace_name}/tags) API overwrites\nexisting tags for specified namespace rather than creating new one\nin addition to the existing tags.\nThis patch resolves the issue by not overwriting the previous tags\nbut appending the new tags to existing ones.\n\nCloses-Bug: #1939169\nChange-Id: I29448746b14c542e5fbf0283011968ae1516642e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9299739790e6d417e517ed118d1bcb47dce0146e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9299739790e6d417e517ed118d1bcb47dce0146e"}]},"branch":"refs/heads/master"},"553e98d42b6422672ff34c258e1137328a97fcf5":{"kind":"REWORK","_number":5,"created":"2021-08-26 05:25:06.000000000","uploader":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"ref":"refs/changes/66/804966/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/66/804966/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/804966/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/804966/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/804966/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/66/804966/5"}}},"commit":{"parents":[{"commit":"284189222923f60bd883dd95ff0354b80850a4ce","subject":"Merge \"Clean out deprecations from paste.ini\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/284189222923f60bd883dd95ff0354b80850a4ce"}]}],"author":{"name":"Mridula Joshi","email":"mrjoshi@redhat.com","date":"2021-08-18 08:21:53.000000000","tz":0},"committer":{"name":"Mridula Joshi","email":"mrjoshi@redhat.com","date":"2021-08-26 05:23:41.000000000","tz":0},"subject":"Fix overwriting of existing tags while creating new tags","message":"Fix overwriting of existing tags while creating new tags\n\nIt was observed that md-tag-create-multiple\n(/v2/metadefs/namespaces/{namespace_name}/tags) API overwrites\nexisting tags for specified namespace rather than creating new one\nin addition to the existing tags.\nThis patch resolves the issue by not overwriting the previous tags\nbut appending the new tags to existing ones.\n\nCloses-Bug: #1939169\nChange-Id: I29448746b14c542e5fbf0283011968ae1516642e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/553e98d42b6422672ff34c258e1137328a97fcf5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/553e98d42b6422672ff34c258e1137328a97fcf5"}]},"branch":"refs/heads/master"},"1139c51567bfb9f69be80c23b210077dffb9ab27":{"kind":"NO_CHANGE","_number":6,"created":"2021-08-26 13:23:55.000000000","uploader":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"ref":"refs/changes/66/804966/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/66/804966/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/804966/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/804966/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/804966/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/66/804966/6"}}},"commit":{"parents":[{"commit":"284189222923f60bd883dd95ff0354b80850a4ce","subject":"Merge \"Clean out deprecations from paste.ini\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/284189222923f60bd883dd95ff0354b80850a4ce"}]}],"author":{"name":"Mridula Joshi","email":"mrjoshi@redhat.com","date":"2021-08-18 08:21:53.000000000","tz":0},"committer":{"name":"Mridula Joshi","email":"mrjoshi@redhat.com","date":"2021-08-26 05:50:30.000000000","tz":0},"subject":"Fix overwriting of existing tags while creating new tags","message":"Fix overwriting of existing tags while creating new tags\n\nIt was observed that md-tag-create-multiple\n(/v2/metadefs/namespaces/{namespace_name}/tags) API overwrites\nexisting tags for specified namespace rather than creating new one\nin addition to the existing tags.\nThis patch resolves the issue by not overwriting the previous tags\nbut appending the new tags to existing ones.\n\nCloses-Bug: #1939169\nChange-Id: I29448746b14c542e5fbf0283011968ae1516642e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1139c51567bfb9f69be80c23b210077dffb9ab27"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1139c51567bfb9f69be80c23b210077dffb9ab27"}]},"branch":"refs/heads/master"},"6a67034c100ccc23f82c671798e7d9c9a4e35337":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2021-08-26 13:35:46.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/66/804966/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/66/804966/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/804966/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/804966/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/804966/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/66/804966/7"}}},"commit":{"parents":[{"commit":"869a39d9e55849c7de7b3ee086cea1017d4f60d5","subject":"Check policies for Image Tags in API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/869a39d9e55849c7de7b3ee086cea1017d4f60d5"}]}],"author":{"name":"Mridula Joshi","email":"mrjoshi@redhat.com","date":"2021-08-18 08:21:53.000000000","tz":0},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2021-08-26 13:35:46.000000000","tz":0},"subject":"Fix overwriting of existing tags while creating new tags","message":"Fix overwriting of existing tags while creating new tags\n\nIt was observed that md-tag-create-multiple\n(/v2/metadefs/namespaces/{namespace_name}/tags) API overwrites\nexisting tags for specified namespace rather than creating new one\nin addition to the existing tags.\nThis patch resolves the issue by not overwriting the previous tags\nbut appending the new tags to existing ones.\n\nCloses-Bug: #1939169\nChange-Id: I29448746b14c542e5fbf0283011968ae1516642e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6a67034c100ccc23f82c671798e7d9c9a4e35337"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6a67034c100ccc23f82c671798e7d9c9a4e35337"}]},"branch":"refs/heads/master","description":"Rebase"},"59583eefb635f22d3630a57bc69aebc53d72dd47":{"kind":"REWORK","_number":8,"created":"2021-10-12 05:53:39.000000000","uploader":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"ref":"refs/changes/66/804966/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/66/804966/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/804966/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/804966/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/804966/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/66/804966/8"}}},"commit":{"parents":[{"commit":"56cc683055a84f364986507d30078ae30de2ef29","subject":"Merge \"trivial: Double quote check_str for consistency\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/56cc683055a84f364986507d30078ae30de2ef29"}]}],"author":{"name":"Mridula Joshi","email":"mrjoshi@redhat.com","date":"2021-08-18 08:21:53.000000000","tz":0},"committer":{"name":"Mridula Joshi","email":"mrjoshi@redhat.com","date":"2021-10-12 05:45:07.000000000","tz":0},"subject":"Fix overwriting of existing tags while creating new tags","message":"Fix overwriting of existing tags while creating new tags\n\nIt was observed that md-tag-create-multiple\n(/v2/metadefs/namespaces/{namespace_name}/tags) API overwrites\nexisting tags for specified namespace rather than creating new one\nin addition to the existing tags.\nThis patch resolves the issue by introducing a query parameter \u0027append\u0027\nwhich on being True will append the new tags to existing ones and\nif False will continue to overwrite the tags.\n\nCloses-Bug: #1939169\nChange-Id: I29448746b14c542e5fbf0283011968ae1516642e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/59583eefb635f22d3630a57bc69aebc53d72dd47"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/59583eefb635f22d3630a57bc69aebc53d72dd47"}]},"branch":"refs/heads/master"},"2f5a600af443f9168bdb318682fedf9b27e773c1":{"kind":"REWORK","_number":9,"created":"2021-10-14 06:46:22.000000000","uploader":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"ref":"refs/changes/66/804966/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/66/804966/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/804966/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/804966/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/804966/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/66/804966/9"}}},"commit":{"parents":[{"commit":"56cc683055a84f364986507d30078ae30de2ef29","subject":"Merge \"trivial: Double quote check_str for consistency\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/56cc683055a84f364986507d30078ae30de2ef29"}]}],"author":{"name":"Mridula Joshi","email":"mrjoshi@redhat.com","date":"2021-08-18 08:21:53.000000000","tz":0},"committer":{"name":"Mridula Joshi","email":"mrjoshi@redhat.com","date":"2021-10-14 06:42:53.000000000","tz":0},"subject":"Fix overwriting of existing tags while creating new tags","message":"Fix overwriting of existing tags while creating new tags\n\nIt was observed that md-tag-create-multiple\n(/v2/metadefs/namespaces/{namespace_name}/tags) API overwrites\nexisting tags for specified namespace rather than creating new one\nin addition to the existing tags.\nThis patch resolves the issue by introducing a query parameter \u0027append\u0027\nwhich on being True will append the new tags to existing ones and\nif False will continue to overwrite the tags.\n\nCloses-Bug: #1939169\nChange-Id: I29448746b14c542e5fbf0283011968ae1516642e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2f5a600af443f9168bdb318682fedf9b27e773c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2f5a600af443f9168bdb318682fedf9b27e773c1"}]},"branch":"refs/heads/master"},"06a353cc6f3f45e3a5e539656b4a69c5aa4cf168":{"kind":"REWORK","_number":10,"created":"2022-01-03 12:07:46.000000000","uploader":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"ref":"refs/changes/66/804966/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/66/804966/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/804966/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/804966/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/804966/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/66/804966/10"}}},"commit":{"parents":[{"commit":"56cc683055a84f364986507d30078ae30de2ef29","subject":"Merge \"trivial: Double quote check_str for consistency\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/56cc683055a84f364986507d30078ae30de2ef29"}]}],"author":{"name":"Mridula Joshi","email":"mrjoshi@redhat.com","date":"2021-08-18 08:21:53.000000000","tz":0},"committer":{"name":"Mridula Joshi","email":"mrjoshi@redhat.com","date":"2022-01-03 12:02:18.000000000","tz":0},"subject":"Fix overwriting of existing tags while creating new tags","message":"Fix overwriting of existing tags while creating new tags\n\nIt was observed that md-tag-create-multiple\n(/v2/metadefs/namespaces/{namespace_name}/tags) API overwrites\nexisting tags for specified namespace rather than creating new one\nin addition to the existing tags.\nThis patch resolves the issue by introducing a header \u0027X-Openstack-Append\u0027\nwhich on being True will append the new tags to existing ones and\nif False will continue to overwrite the tags.\n\nCloses-Bug: #1939169\nChange-Id: I29448746b14c542e5fbf0283011968ae1516642e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/06a353cc6f3f45e3a5e539656b4a69c5aa4cf168"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/06a353cc6f3f45e3a5e539656b4a69c5aa4cf168"}]},"branch":"refs/heads/master"},"a345495de43136e1f3d673001f18457037cf5fe3":{"kind":"REWORK","_number":11,"created":"2022-01-27 15:58:37.000000000","uploader":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"ref":"refs/changes/66/804966/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/66/804966/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/804966/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/804966/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/804966/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/66/804966/11"}}},"commit":{"parents":[{"commit":"56cc683055a84f364986507d30078ae30de2ef29","subject":"Merge \"trivial: Double quote check_str for consistency\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/56cc683055a84f364986507d30078ae30de2ef29"}]}],"author":{"name":"Mridula Joshi","email":"mrjoshi@redhat.com","date":"2021-08-18 08:21:53.000000000","tz":0},"committer":{"name":"Mridula Joshi","email":"mrjoshi@redhat.com","date":"2022-01-27 15:42:14.000000000","tz":0},"subject":"Fix overwriting of existing tags while creating new tags","message":"Fix overwriting of existing tags while creating new tags\n\nIt was observed that md-tag-create-multiple\n(/v2/metadefs/namespaces/{namespace_name}/tags) API overwrites\nexisting tags for specified namespace rather than creating new one\nin addition to the existing tags.\nThis patch resolves the issue by introducing a header \u0027X-Openstack-Append\u0027\nwhich on being True will append the new tags to existing ones and\nif False will continue to overwrite the tags.\n\nImplements: blueprint append-tags\nCloses-Bug: #1939169\nChange-Id: I29448746b14c542e5fbf0283011968ae1516642e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a345495de43136e1f3d673001f18457037cf5fe3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a345495de43136e1f3d673001f18457037cf5fe3"}]},"branch":"refs/heads/master"},"8ec1d27ccfd25d22a7e1f71a27c09dba8f662594":{"kind":"REWORK","_number":12,"created":"2022-01-31 14:23:05.000000000","uploader":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"ref":"refs/changes/66/804966/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/66/804966/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/804966/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/804966/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/804966/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/66/804966/12"}}},"commit":{"parents":[{"commit":"56cc683055a84f364986507d30078ae30de2ef29","subject":"Merge \"trivial: Double quote check_str for consistency\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/56cc683055a84f364986507d30078ae30de2ef29"}]}],"author":{"name":"Mridula Joshi","email":"mrjoshi@redhat.com","date":"2021-08-18 08:21:53.000000000","tz":0},"committer":{"name":"Mridula Joshi","email":"mrjoshi@redhat.com","date":"2022-01-31 14:14:36.000000000","tz":0},"subject":"Fix overwriting of existing tags while creating new tags","message":"Fix overwriting of existing tags while creating new tags\n\nIt was observed that md-tag-create-multiple\n(/v2/metadefs/namespaces/{namespace_name}/tags) API overwrites\nexisting tags for specified namespace rather than creating new one\nin addition to the existing tags.\nThis patch resolves the issue by introducing a header \u0027X-Openstack-Append\u0027\nwhich on being True will append the new tags to existing ones and\nif False will continue to overwrite the tags.\n\nImplements: blueprint append-tags\nCloses-Bug: #1939169\nChange-Id: I29448746b14c542e5fbf0283011968ae1516642e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8ec1d27ccfd25d22a7e1f71a27c09dba8f662594"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8ec1d27ccfd25d22a7e1f71a27c09dba8f662594"}]},"branch":"refs/heads/master"},"51ebf1566f9db473024c052aa55f48165d716bdc":{"kind":"REWORK","_number":13,"created":"2022-02-10 09:56:24.000000000","uploader":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"ref":"refs/changes/66/804966/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/66/804966/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/804966/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/804966/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/804966/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/66/804966/13"}}},"commit":{"parents":[{"commit":"a42fda92dc8fe160bea1dd4d22cb1135fab89022","subject":"Merge \"[APIImpact] Quota usage API\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a42fda92dc8fe160bea1dd4d22cb1135fab89022"}]}],"author":{"name":"Mridula Joshi","email":"mrjoshi@redhat.com","date":"2021-08-18 08:21:53.000000000","tz":0},"committer":{"name":"Mridula Joshi","email":"mrjoshi@redhat.com","date":"2022-02-10 09:46:42.000000000","tz":0},"subject":"Fix overwriting of existing tags while creating new tags","message":"Fix overwriting of existing tags while creating new tags\n\nIt was observed that md-tag-create-multiple\n(/v2/metadefs/namespaces/{namespace_name}/tags) API overwrites\nexisting tags for specified namespace rather than creating new one\nin addition to the existing tags.\nThis patch resolves the issue by introducing a header \u0027X-Openstack-Append\u0027\nwhich on being True will append the new tags to existing ones and\nif False will continue to overwrite the tags.\n\nImplements: blueprint append-tags\nCloses-Bug: #1939169\nChange-Id: I29448746b14c542e5fbf0283011968ae1516642e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/51ebf1566f9db473024c052aa55f48165d716bdc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/51ebf1566f9db473024c052aa55f48165d716bdc"}]},"branch":"refs/heads/master"},"febc7c3385163bf01e10aba1b635862ea6335635":{"kind":"REWORK","_number":14,"created":"2022-02-14 09:09:33.000000000","uploader":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"ref":"refs/changes/66/804966/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/66/804966/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/804966/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/804966/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/804966/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/66/804966/14"}}},"commit":{"parents":[{"commit":"a42fda92dc8fe160bea1dd4d22cb1135fab89022","subject":"Merge \"[APIImpact] Quota usage API\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a42fda92dc8fe160bea1dd4d22cb1135fab89022"}]}],"author":{"name":"Mridula Joshi","email":"mrjoshi@redhat.com","date":"2021-08-18 08:21:53.000000000","tz":0},"committer":{"name":"Mridula Joshi","email":"mrjoshi@redhat.com","date":"2022-02-14 09:00:24.000000000","tz":0},"subject":"Fix overwriting of existing tags while creating new tags","message":"Fix overwriting of existing tags while creating new tags\n\nIt was observed that md-tag-create-multiple\n(/v2/metadefs/namespaces/{namespace_name}/tags) API overwrites\nexisting tags for specified namespace rather than creating new one\nin addition to the existing tags.\nThis patch resolves the issue by introducing a header \u0027X-Openstack-Append\u0027\nwhich on being True will append the new tags to existing ones and\nif False will continue to overwrite the tags.\n\nImplements: blueprint append-tags\nCloses-Bug: #1939169\nChange-Id: I29448746b14c542e5fbf0283011968ae1516642e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/febc7c3385163bf01e10aba1b635862ea6335635"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/febc7c3385163bf01e10aba1b635862ea6335635"}]},"branch":"refs/heads/master"},"e2296fee755cb811e3312741e68b6feada7483be":{"kind":"REWORK","_number":15,"created":"2022-02-15 07:04:07.000000000","uploader":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"ref":"refs/changes/66/804966/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/66/804966/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/804966/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/804966/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/804966/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/66/804966/15"}}},"commit":{"parents":[{"commit":"a42fda92dc8fe160bea1dd4d22cb1135fab89022","subject":"Merge \"[APIImpact] Quota usage API\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a42fda92dc8fe160bea1dd4d22cb1135fab89022"}]}],"author":{"name":"Mridula Joshi","email":"mrjoshi@redhat.com","date":"2021-08-18 08:21:53.000000000","tz":0},"committer":{"name":"Mridula Joshi","email":"mrjoshi@redhat.com","date":"2022-02-15 06:54:50.000000000","tz":0},"subject":"Fix overwriting of existing tags while creating new tags","message":"Fix overwriting of existing tags while creating new tags\n\nIt was observed that md-tag-create-multiple\n(/v2/metadefs/namespaces/{namespace_name}/tags) API overwrites\nexisting tags for specified namespace rather than creating new one\nin addition to the existing tags.\nThis patch resolves the issue by introducing a header \u0027X-Openstack-Append\u0027\nwhich on being True will append the new tags to existing ones and\nif False will continue to overwrite the tags.\n\nImplements: blueprint append-tags\nCloses-Bug: #1939169\nChange-Id: I29448746b14c542e5fbf0283011968ae1516642e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e2296fee755cb811e3312741e68b6feada7483be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e2296fee755cb811e3312741e68b6feada7483be"}]},"branch":"refs/heads/master"},"2a9a4c8e0e095f3cd788e2f24b2dce508538e0d5":{"kind":"REWORK","_number":16,"created":"2022-02-16 05:54:21.000000000","uploader":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"ref":"refs/changes/66/804966/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/66/804966/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/804966/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/804966/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/804966/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/66/804966/16"}}},"commit":{"parents":[{"commit":"a42fda92dc8fe160bea1dd4d22cb1135fab89022","subject":"Merge \"[APIImpact] Quota usage API\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a42fda92dc8fe160bea1dd4d22cb1135fab89022"}]}],"author":{"name":"Mridula Joshi","email":"mrjoshi@redhat.com","date":"2021-08-18 08:21:53.000000000","tz":0},"committer":{"name":"Mridula Joshi","email":"mrjoshi@redhat.com","date":"2022-02-16 05:45:06.000000000","tz":0},"subject":"Fix overwriting of existing tags while creating new tags","message":"Fix overwriting of existing tags while creating new tags\n\nIt was observed that md-tag-create-multiple\n(/v2/metadefs/namespaces/{namespace_name}/tags) API overwrites\nexisting tags for specified namespace rather than creating new one\nin addition to the existing tags.\nThis patch resolves the issue by introducing a header \u0027X-Openstack-Append\u0027\nwhich on being True will append the new tags to existing ones and\nif False will continue to overwrite the tags.\n\nImplements: blueprint append-tags\nCloses-Bug: #1939169\nChange-Id: I29448746b14c542e5fbf0283011968ae1516642e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2a9a4c8e0e095f3cd788e2f24b2dce508538e0d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2a9a4c8e0e095f3cd788e2f24b2dce508538e0d5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
