)]}'
{"id":"openstack%2Fglance~805065","triplet_id":"openstack%2Fglance~master~Ibe33dff25782a48b90bccb3f99472fcec0f2f5c2","project":"openstack/glance","branch":"master","topic":"bp/policy-refactor","hashtags":[],"change_id":"Ibe33dff25782a48b90bccb3f99472fcec0f2f5c2","subject":"Refactor gateway auth layer for image factory","status":"MERGED","created":"2021-08-18 17:31:10.000000000","updated":"2021-08-19 06:53:15.000000000","submitted":"2021-08-19 06:51:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"805065-bp/policy-refactor","meta_rev_id":"96e0ed3b2a90d99f9349bfb10d8f17eff1a0c1e8","_number":805065,"virtual_id_number":805065,"owner":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-19 06:51:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2021-08-19 05:20:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2021-08-19 05:20:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-18 18:47:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-19 05:20:53.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"}],"messages":[{"id":"84dfbc29f9265c5e066358805ff4ab3d8b16c7e5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-08-18 17:31:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f6181caa49ae19d96dff7d92fb2fde24a0d02b6c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-18 18:47:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/dc655324a3834d9585331aa405f96359 : SUCCESS in 59m 19s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/146ec1478872462ba0818d09c7fdd8d5 : SUCCESS in 1h 03m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/619a505d90ff4ee49fa64fe9b679efad : SUCCESS in 5m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c65111e2d7de4eb28694b9ecb79fbaaa : SUCCESS in 6m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5479f4ee30b04dda8b301a96ae414559 : SUCCESS in 5m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/72d1f608bb9449b29148c9fca0fae8a5 : SUCCESS in 6m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b167d5fc0abc4a1e88ebdc7929aa024a : SUCCESS in 8m 22s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/14c5d7744814456886e316ba2116d353 : SUCCESS in 9m 13s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/85175db67db64439a0e1d1edec5beba5 : SUCCESS in 9m 11s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/c121df8594f44b34a7f0b3cfc8290267 : SUCCESS in 8m 56s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/ed082c44868c437eac6a1e2bfc357aae : SUCCESS in 6m 30s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/9684e13657ed45beba9c361c125fb4e2 : SUCCESS in 1h 09m 45s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/9510c4ca0e7544d2bbe6ec53e2ddd4a1 : SUCCESS in 58m 22s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/e42973d9007947fdb9cac0dcac3fcd32 : SUCCESS in 1h 06m 12s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/e992d56532dd48e0b3c70d24660af849 : SUCCESS in 1h 12m 42s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/040d375d8a7c41119fadab1791dca35f : SUCCESS in 43m 24s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/09abfc034a4c44aa8fb4817b89746102 : SUCCESS in 1h 13m 03s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/64807ff138dd442d8ec7a2279935b72f : SUCCESS in 30m 14s","accounts_in_message":[],"_revision_number":1},{"id":"a654e35498f81fc2e389317cf4e4c3286c847fb7","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-19 05:20:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nStraight forward change, so ninja approving it!","accounts_in_message":[],"_revision_number":1},{"id":"f31ac15692b641bb2a779c80fc2a779edb45a4a6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-19 05:21:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"378dd00678eb495130c66782f154e4b4459a1ccb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-19 06:51:09.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/4603baf2e35d4e3c911b7128d261ef23 : SUCCESS in 57m 54s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e58d1c11b3df4d1ea495e33fad6306d4 : SUCCESS in 1h 12m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a5800719b8dc41d28ef96a650c70061e : SUCCESS in 5m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/302f51d71e9d4a1c8192525b032973ce : SUCCESS in 5m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c13c6614a53e40d78cc675a3722e00fb : SUCCESS in 6m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f6af951abf24ceeb9a1fcf080a75995 : SUCCESS in 6m 21s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2fde127ed6fd4846ab14111e5d255404 : SUCCESS in 9m 26s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b81c85ce4ca14a3da31b0cc06cccab73 : SUCCESS in 9m 57s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/3d1136cb93c0459695b8e1cb23a94067 : SUCCESS in 1h 19m 41s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/04a5e3669a604f1da68795470eb8c011 : SUCCESS in 1h 07m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/88ce65648592437aa57a64d27e86bb22 : SUCCESS in 49m 38s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/c10fdf37efab445aa912e7a7000a0444 : SUCCESS in 1h 28m 47s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/8db2fde1f71e46908befde46ff478154 : SUCCESS in 30m 30s","accounts_in_message":[],"_revision_number":1},{"id":"8b5f487ac476cdd8ca21b6c9f6ac7c84ee822fcf","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-19 06:51:10.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"96e0ed3b2a90d99f9349bfb10d8f17eff1a0c1e8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-19 06:53:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b2a052320594c95b14b512787f34d8a : SUCCESS in 1m 32s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a9009f4eb1696b856f9214b8e5a361cdf0b9c212","revisions":{"a9009f4eb1696b856f9214b8e5a361cdf0b9c212":{"kind":"REWORK","_number":1,"created":"2021-08-18 17:31:10.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/65/805065/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/65/805065/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/65/805065/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/65/805065/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/65/805065/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/65/805065/1"}}},"commit":{"parents":[{"commit":"dd3155516cec2cabf8f74963a44ab642d507384b","subject":"Merge \"Load options from the castellan library\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/dd3155516cec2cabf8f74963a44ab642d507384b"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-08-18 17:02:25.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-08-18 17:12:16.000000000","tz":-420},"subject":"Refactor gateway auth layer for image factory","message":"Refactor gateway auth layer for image factory\n\nChange-Id: Ibe33dff25782a48b90bccb3f99472fcec0f2f5c2\nPartially-Implements: blueprint policy-refactor\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a9009f4eb1696b856f9214b8e5a361cdf0b9c212"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a9009f4eb1696b856f9214b8e5a361cdf0b9c212"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
