)]}'
{"id":"openstack%2Fglance~805256","triplet_id":"openstack%2Fglance~master~Idfaaccde42f43fab1953012a83ec0f259412ba10","project":"openstack/glance","branch":"master","hashtags":[],"change_id":"Idfaaccde42f43fab1953012a83ec0f259412ba10","subject":"trivial: Double quote check_str for consistency","status":"MERGED","created":"2021-08-19 17:18:39.000000000","updated":"2021-10-06 23:22:39.000000000","submitted":"2021-10-06 23:20:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"805256","meta_rev_id":"b301c9b6d443722961e11e5375d90d7e7efae07f","_number":805256,"virtual_id_number":805256,"owner":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-10-06 23:20:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2021-09-27 15:38:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2021-09-21 12:52:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2021-09-27 15:38:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]}],"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-19 18:42:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-20 07:06:01.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-09-20 14:22:47.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"CC"},{"updated":"2021-09-21 12:52:44.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2021-09-27 15:38:02.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"}],"messages":[{"id":"3af0972de0aba28b6fcb9658c6b888c19878d3e4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-08-19 17:18:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"11c4e85d6eff103d39641aae3d9915988ffd776b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-08-19 17:18:58.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"514277a2eca9549bd7802435d18fa6c5a1e4185e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-19 18:42:38.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/3afd5f9632db4103b7f97e0ed0e98e04 : SUCCESS in 1h 03m 30s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/70d41dd1aaa94780943013b8d7527681 : SUCCESS in 1h 05m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a1f2980a86c48e087da75eee6c03163 : SUCCESS in 6m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a038b08ccee54073a77cc59023ec877d : SUCCESS in 6m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/60ba8cd245c741058414f52c946c4956 : SUCCESS in 8m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9cbfcca280d24afebc09bd73f686bbd6 : SUCCESS in 9m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b5f7990d66146b183aa85dd7d747eb8 : SUCCESS in 9m 59s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/02b4714f08674bfda704fa53bd3bf9d2 : TIMED_OUT in 32m 14s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/77d365c49de94a818d6390f58bb86a57 : SUCCESS in 12m 02s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/15f0210b267646dd9a8c7815fa271d1d : SUCCESS in 11m 56s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/eab45888211d4d14b782bf926c201aa6 : SUCCESS in 7m 31s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/b056d600866b4e42b8fa028f90b73de9 : SUCCESS in 1h 17m 45s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/6b624e51c638431ebb757d240361066f : SUCCESS in 1h 06m 14s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/d6e47b1f498e4f50a4ec84778dd34a82 : SUCCESS in 1h 17m 39s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/50e638587b954c75a19da7722e97f758 : SUCCESS in 1h 11m 47s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4ff3dcb49a3a491097e04ddf6ddc0976 : SUCCESS in 44m 09s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/025ca535dd864202a1a3ee4ed4326935 : SUCCESS in 1h 05m 01s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/5f2bc1e46cc84857804c9854c6d858d9 : SUCCESS in 31m 15s","accounts_in_message":[],"_revision_number":2},{"id":"8cb2873f3b6c8d168e2070a93068d5955982b110","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-08-20 07:06:01.000000000","message":"Patch Set 2:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/805256/805256/2/check-not-vote/ext-glance-zuul/7b7ee19 : FAILURE in 11m 01s","accounts_in_message":[],"_revision_number":2},{"id":"4e78067b5445928858fdfad438d6c53892e3a40b","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2021-09-20 14:22:47.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"8e9670fd0a27e7097d8b7f468845bbe265d1190e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-20 15:37:30.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/4c1312a5879741b383d1390e95318f25 : SUCCESS in 1h 06m 35s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1096227868b94ee8ba78174b3c5f7dc2 : SUCCESS in 1h 10m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/939a5d6b070044568ba3cb1cba5b0f9a : SUCCESS in 4m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/40d19880a0c64d39bf1b975604a7b07b : SUCCESS in 6m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9f6f1fca684749ff8d7837980865af19 : SUCCESS in 6m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/db82ea3082744477b1a5377324592c64 : SUCCESS in 6m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a904cc70d1c145e79fc2793c522c98a5 : SUCCESS in 7m 23s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7147918364c84c879545d002ce9961a6 : SUCCESS in 11m 49s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/382b76bcf2534a3f9defba6346de474e : SUCCESS in 9m 57s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/156b7a5a089c444d8e0d232d027b9052 : SUCCESS in 9m 51s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/f81e0d40754d4724a3a1731fe3e03c06 : SUCCESS in 5m 41s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/c5d8de8f11694a6e914c9b81635eaf3f : SUCCESS in 1h 11m 13s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/606e8a159d10452395a82d7edad11873 : SUCCESS in 59m 23s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/7e7fc5ff639b44aa80e3652cc406b134 : SUCCESS in 1h 01m 57s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/79361992f681407485671967ebce8381 : SUCCESS in 1h 12m 47s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/05c6e1d5fb7544da99589e2abaae60a1 : SUCCESS in 38m 10s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/ad5bb55e1c0f429d8371a298bbe7a01d : SUCCESS in 1h 05m 58s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/13c0d708a66d405fbb4b742ff417b62a : SUCCESS in 33m 07s","accounts_in_message":[],"_revision_number":2},{"id":"024103e0a68493213c10102dfdc6a9fbc9cff9fc","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2021-09-21 12:52:44.000000000","message":"Patch Set 2: Code-Review+2\n\nI\u0027d really rather have single quotes everywhere since that is more Pythonic, but apparently double quotes are in fashion ever since Black came out *sigh*. Anyway, consistency matters, so +2. Thanks for addressing our comments from the other revieww!","accounts_in_message":[],"_revision_number":2},{"id":"d084b82d1d0748ed80fe8d512755837e142a1adb","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-09-27 15:38:02.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"4d0c95fb2975d8b8b7fda2e3b13d52e57e453d18","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-27 15:39:48.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"84e0a6f618d50820578fc5038dd3d0770bccfa63","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-27 18:27:08.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/c8bfb74328df42158123e8559cb87350 : FAILURE in 45m 28s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/77ffd59bb5e04841a1c8887472fce8ec : TIMED_OUT in 2h 07m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f7170ba537644deb69820116e8640f6 : SUCCESS in 7m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a3b59cb33dd148dfa52c5f9627b7d2b2 : SUCCESS in 7m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/99dc06bbd30044a69a37811a74a3fe0a : SUCCESS in 7m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/971e9fa76a9048629c7d4c6db6a6ea73 : SUCCESS in 7m 29s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e558776fc91e41a2a08a99404a02309a : SUCCESS in 15m 37s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c66741cf37c64171942352c8e1e73a0d : SUCCESS in 9m 55s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/b6352b931ed84f41bfc58c9da8025961 : SUCCESS in 1h 07m 33s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/a250145fd9034e53ac25cf06b5b56e48 : TIMED_OUT in 2h 07m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/69cbcdaf54354a37a2ff418a04a16b3c : SUCCESS in 44m 37s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/a097b57a0c92431aa4a94e5ae42c2a2d : SUCCESS in 1h 06m 37s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/14367944fe3d44cb946ccc5f4b7e5c39 : SUCCESS in 55m 24s","accounts_in_message":[],"_revision_number":2},{"id":"3aa3f40027e27be411294a8e50524614daf1c2a3","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-09-28 14:48:43.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"12790e1a37e508c2a3b71817430de434e4da496b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-28 15:20:00.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/84656aa543f447e5aa3a25aed7c74703 : FAILURE in 23m 35s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2efc2077deb34e04bff4c8074bec04c7 : FAILURE in 22m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c5fe602497e542acb5a741afc8c7b59b : SUCCESS in 5m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/642dc301f3b74a45b759184e26355c80 : SUCCESS in 5m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7614fcedbf7641cfbccb7c9681a6966f : SUCCESS in 6m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/798d286da55e455eafe56ab7c73c8b87 : SUCCESS in 6m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d4af0e784614ae382a92850c034ebb1 : SUCCESS in 7m 12s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1f1a362055d743f6b545f39b379e29e4 : SUCCESS in 8m 14s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/1e95420676cb4dd8bcb249909473c827 : SUCCESS in 10m 22s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/69ac21ae2a8148ae9de46fc41234769d : SUCCESS in 9m 32s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/3c0e64c7b3d74fd99bd54a13601fc1cb : SUCCESS in 5m 23s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/b0603c0eca7c44939ea7a9e10a0c34e8 : FAILURE in 23m 48s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/5be90279b74b4670b89b34ac5558567e : FAILURE in 25m 27s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/5a42bad331cd4d3482f355b66593a940 : FAILURE in 21m 47s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/863c0b951af348648dac767e67c5f853 : FAILURE in 24m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6d9c397a388941ceb10fb0d1b4b303cd : FAILURE in 23m 49s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/aa79a8eb6e8045b2b49d611d217e5251 : FAILURE in 23m 34s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/851c5810ff044a96b673d316a448f553 : FAILURE in 25m 06s","accounts_in_message":[],"_revision_number":2},{"id":"5f62a37104ff385fd22a119707b3dfad21f44ece","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-09-29 16:52:03.000000000","message":"Patch Set 2:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/805256/805256/2/check-not-vote/ext-glance-zuul/6e83453 : SUCCESS in 37m 33s","accounts_in_message":[],"_revision_number":2},{"id":"63ab0ec02b5720f5691876a8275b54409afb7c94","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2021-10-06 14:28:06.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"4d3fc1f852bc2c3cf59a8a3a9f9fe161604c1a7c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-06 14:32:39.000000000","message":"Patch Set 2:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":2},{"id":"d25376b1bff61a4c32bed015ebf7176b75393fe0","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-10-06 15:01:02.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"cbcb5300cfa1f2c1aa9851f8af941a277a3bf7aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-06 16:52:29.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/91e8695dab49402086084fb68860ab9f : SUCCESS in 1h 08m 00s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9942eb11c6814eaf8b23623aa964598b : SUCCESS in 1h 42m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/473171445e5b4a44a629b5be1a8a2bc6 : SUCCESS in 6m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b1a731188bea4eb1a5f48787713e4b44 : SUCCESS in 7m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/97a42db6feb7437190f955e283c40304 : SUCCESS in 6m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/56e9f7a5ae104470bce75b2b4cebcbe4 : SUCCESS in 7m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25aad376cd9745e1b572549b9a7edcf0 : SUCCESS in 7m 10s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/44100cd22fd1435faec8c471d0893d3c : SUCCESS in 9m 33s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/cd60da7ce6da497981f64ad95c66c9a9 : SUCCESS in 10m 46s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/f7b53fd56fbc43ddb2b806aa6c0c3896 : SUCCESS in 8m 45s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/3e051bb5931449378e5760175e07adea : SUCCESS in 6m 12s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/301d1df5e6e74349945821cc29e4bb13 : SUCCESS in 1h 11m 25s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/9f7fc2e955c042efb72fb10ffdd654b3 : SUCCESS in 1h 06m 20s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/02aa3f4065854cc5b2fffdecdcb5445c : SUCCESS in 1h 06m 06s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/d1c85bcb285b4e95b0e1f3b2323fd2bb : SUCCESS in 1h 09m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ad9b461d2c2246b0ad4b1a23e89ed218 : SUCCESS in 46m 25s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/2c9d455a55294a22b51faa46d86c1e8b : SUCCESS in 1h 18m 09s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/8fea4d99806a4596805cd5558dc32b45 : SUCCESS in 34m 10s","accounts_in_message":[],"_revision_number":2},{"id":"d49f6cdbfec815250d8b2d5d83a0046b9d5ff7aa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-06 16:53:59.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"699355d64caca8b1ad70ca85da27501d06f1b854","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-06 23:20:29.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/009361c9767246ee93e87e58056b0b4e : SUCCESS in 56m 28s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8f89d5b6d7424ad9ab05f7ef250ce492 : SUCCESS in 46m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6220dc3279ec4753bbde160e3c2b68a8 : SUCCESS in 7m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cecd9c0eef794eaab00a6fb95bc14682 : SUCCESS in 7m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/791ce6e32eee4320965da093317efac1 : SUCCESS in 7m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a85f21f29744744abdf9e776b8f0d98 : SUCCESS in 9m 01s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1aad2260638a4c0d986e57c6161298db : SUCCESS in 9m 15s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/1e4285b9429b4fdc83de644b5dd4fb97 : SUCCESS in 10m 12s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/e3b55fba71dd46fe9fd39b4388c23778 : SUCCESS in 1h 11m 53s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/b55d7a7a3c7f4a9fbb300870c9f37801 : SUCCESS in 1h 14m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9c9c1c45477d4b9c9491e9d104f5a8f2 : SUCCESS in 53m 12s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/7eb65c4f08024ebb8308652d1e6a3e8e : SUCCESS in 1h 07m 01s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/95cf70ea204e49c7af99114f2eb2a432 : SUCCESS in 32m 48s","accounts_in_message":[],"_revision_number":2},{"id":"0bdaeeafc7a063563ad379a3149ced1e19126aad","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-06 23:20:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"b301c9b6d443722961e11e5375d90d7e7efae07f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-06 23:22:39.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1129b924a949425f86d3e495e8194aa1 : SUCCESS in 1m 32s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"fb3d3fa8d371f594f257cedfd9ebccdaf3fa1fad","revisions":{"211104e499ad22d449283572238f4f1b8d6a4c7d":{"kind":"REWORK","_number":1,"created":"2021-08-19 17:18:39.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/56/805256/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/56/805256/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/56/805256/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/56/805256/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/56/805256/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/56/805256/1"}}},"commit":{"parents":[{"commit":"cb3369002ade102508bf9c0f776ed7377751789e","subject":"Merge \"Deprecate task specific policies\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/cb3369002ade102508bf9c0f776ed7377751789e"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2021-08-19 17:17:13.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2021-08-19 17:17:13.000000000","tz":0},"subject":"trivial: Double quote check_string for consistency","message":"trivial: Double quote check_string for consistency\n\nThis is a follow up to comments made in an earlier review:\n\n  https://review.opendev.org/c/openstack/glance/+/804585/4/glance/policies/image.py\n\nChange-Id: Idfaaccde42f43fab1953012a83ec0f259412ba10\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/211104e499ad22d449283572238f4f1b8d6a4c7d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/211104e499ad22d449283572238f4f1b8d6a4c7d"}]},"branch":"refs/heads/master"},"fb3d3fa8d371f594f257cedfd9ebccdaf3fa1fad":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-08-19 17:18:58.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/56/805256/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/56/805256/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/56/805256/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/56/805256/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/56/805256/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/56/805256/2"}}},"commit":{"parents":[{"commit":"cb3369002ade102508bf9c0f776ed7377751789e","subject":"Merge \"Deprecate task specific policies\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/cb3369002ade102508bf9c0f776ed7377751789e"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2021-08-19 17:17:13.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2021-08-19 17:18:04.000000000","tz":0},"subject":"trivial: Double quote check_str for consistency","message":"trivial: Double quote check_str for consistency\n\nThis is a follow up to comments made in an earlier review:\n\n  https://review.opendev.org/c/openstack/glance/+/804585/4/glance/policies/image.py\n\nChange-Id: Idfaaccde42f43fab1953012a83ec0f259412ba10\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/fb3d3fa8d371f594f257cedfd9ebccdaf3fa1fad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/fb3d3fa8d371f594f257cedfd9ebccdaf3fa1fad"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
