)]}'
{"id":"openstack%2Fglance~805590","triplet_id":"openstack%2Fglance~master~I8f59b8405d9c55b5f69294d3f9bd7bcbc064203f","project":"openstack/glance","branch":"master","topic":"policy-refactor","hashtags":[],"change_id":"I8f59b8405d9c55b5f69294d3f9bd7bcbc064203f","subject":"Check policies for image tasks information in API","status":"MERGED","created":"2021-08-23 07:46:57.000000000","updated":"2021-08-31 13:34:32.000000000","submitted":"2021-08-30 17:16:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"805590-policy-refactor","meta_rev_id":"b3e9a2c9a2e34d10b4ec029fc16f149eab8fb382","_number":805590,"virtual_id_number":805590,"owner":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-30 17:16:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"recommended":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"all":[{"value":1,"date":"2021-08-30 14:50:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":2,"date":"2021-08-30 08:37:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2021-08-30 13:35:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":1,"date":"2021-08-30 14:23:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-23 08:24:18.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-08-23 09:11:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-24 14:07:14.000000000","updated_by":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"reviewer":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"state":"REVIEWER"},{"updated":"2021-08-24 16:22:38.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2021-08-25 17:57:10.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"}],"messages":[{"id":"a2f8ad528091c98418fc706d6eb336b831d6ad47","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-23 07:46:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"287a4001c48f47d6adc5402d965094eb4f4ad36f","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-08-23 08:24:18.000000000","message":"Patch Set 1:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/805590/805590/1/check-not-vote/ext-glance-zuul/b3976db : SUCCESS in 36m 18s","accounts_in_message":[],"_revision_number":1},{"id":"6aa8f1ed0a7ccdde60a1bcb01445fe167d446c03","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-23 09:11:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/dcd95919acf341db8195d6fc15457971 : SUCCESS in 1h 00m 50s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ea23b7eb379a4a558f444092fe1a907f : SUCCESS in 1h 05m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a7f2a6b06b7b42eca59c05494b0411ff : SUCCESS in 5m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1ca6297b49184f6e96381e12a296f05e : SUCCESS in 6m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e918c43d068246e382e21899e422c448 : SUCCESS in 6m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/87115875774241a4968c171e804b0265 : SUCCESS in 6m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a311fcadb6c34881a9d0d741adea9126 : SUCCESS in 7m 05s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c01fb230649f48e3a1963048364ed020 : SUCCESS in 9m 00s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/6ff55ba65bfd4c6d820cf40fc4baa4f3 : SUCCESS in 8m 57s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/f0471c83a5d94c238e5273e458e86f37 : SUCCESS in 8m 58s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/a4465a14af964f07b27f6d3d541a2767 : SUCCESS in 5m 36s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/9a0a394cfc75494ea3f9f05c2214cf3d : SUCCESS in 1h 04m 24s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/252d3ea82dd14a30b31fab40b00fe677 : SUCCESS in 1h 09m 42s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/9393530242b44fecb6e4449ab4b9c4f1 : SUCCESS in 1h 05m 52s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/7e9c91c091b4483ab8217e34a969a02d : SUCCESS in 1h 14m 26s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ec4e8edcb21744b4a5925584d8c63f4e : SUCCESS in 48m 35s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/cdf4b5a414594ed8a062bd8c3b6bde9f : SUCCESS in 1h 23m 58s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/6b438b34ceb44bc7b67f523eb641bc6b : SUCCESS in 37m 21s","accounts_in_message":[],"_revision_number":1},{"id":"95f220028e1dada2c4c558c3aa9ac66bccc0b362","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-24 06:15:20.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"4ee5688ef32b40c7a2f69b24c5e5c5464b42a951","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-24 08:27:36.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/2b829a99db9a4f689d502dfe2bc546fa : SUCCESS in 1h 07m 49s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/123410114168478baf0794f2ace00a86 : SUCCESS in 1h 07m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3908340536a34b9dbe13410df810733d : SUCCESS in 6m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/45b0b77bb0174e6c89911341dc3e4abd : SUCCESS in 7m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/74f35f5ce684436cb43fe24d6fb017aa : SUCCESS in 6m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/324a1e3663144a34b91e6f34edd9043a : SUCCESS in 8m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b4a7b35051b49edaa054b4e635e3a19 : SUCCESS in 7m 44s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a63a1a580dd047e08f8abbcd04a62083 : SUCCESS in 10m 11s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/f0d7a63140544a6e97643e373b954243 : SUCCESS in 14m 26s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/26ab6d8895f840fe8103338d0d52ef2e : SUCCESS in 12m 43s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/841b672609854fd79fe12c0b46508964 : SUCCESS in 7m 33s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/ce49dce8a2964c9788e605cfad5f3e26 : SUCCESS in 1h 14m 15s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/6e08f6265aa841518cdbb4a3642f484e : SUCCESS in 1h 08m 41s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/ca50615e6fe2457389128ee9f055ea5b : SUCCESS in 1h 06m 26s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/b02f22432ed445f8916cdddc5a84bf66 : SUCCESS in 1h 20m 20s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f2b3e449b3d040cf98e63784d1bd6d55 : SUCCESS in 52m 01s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/f5db3a8282354d0491fd108445f47aa3 : SUCCESS in 1h 15m 01s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/1d1f00510bad4b8c9874845c036ee423 : SUCCESS in 38m 20s","accounts_in_message":[],"_revision_number":2},{"id":"f5175926299232b605ed7a542b12debc08bca3a8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-24 09:36:10.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"3077d3ed79827abd572805dee63cc0bb2f55c857","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-24 11:39:01.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/7945a056128a4d36bdeab15ce948fce4 : SUCCESS in 59m 39s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4014303647b14c6fab3dd49a14f6f249 : SUCCESS in 1h 15m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/77b1ae54b29a4e98b3ad1054bb4f7171 : SUCCESS in 7m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/76c886c131234abe84550a20ccc60935 : SUCCESS in 9m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5ca9d11fac6443c7b530dafc156d53c0 : SUCCESS in 9m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f52a54ccd3704c8a87e242e2de091191 : SUCCESS in 8m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/956d79d9832f42e5965de71c66ed7c97 : SUCCESS in 9m 55s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f1e5635721e5464696b75100eac43d87 : SUCCESS in 10m 01s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/cd656b2d873f4f00811c3e47d5777e27 : SUCCESS in 14m 51s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/192db194b66a4c2fb56ae6933b0ac5fe : SUCCESS in 15m 13s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/c6f72394f7264ec6b2ff99a0cea8ad86 : SUCCESS in 7m 33s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/2d68b97f09cb491790ec2133ed7f3467 : SUCCESS in 1h 09m 26s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/f0f31c3436344728a52ce64425ee3150 : SUCCESS in 1h 13m 30s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/420092ac68b44f76a9457b17db429f9a : SUCCESS in 1h 15m 25s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/92927e0d21124888bca302cdecf6f7c4 : SUCCESS in 1h 28m 56s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8acb93e6ab8f4e2799f9e9a39f442b51 : SUCCESS in 51m 46s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/c4bcd74981b44de3b6a3613204861963 : SUCCESS in 1h 18m 45s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/be06c845770b439db3d5aab79a331483 : SUCCESS in 33m 03s","accounts_in_message":[],"_revision_number":3},{"id":"68ab3174fa48524b8ada3ba778314ea12ed2c4be","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-08-24 14:07:14.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"084a2553985c27eb0beefae9f1858637dfad4e56","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-24 14:57:49.000000000","message":"Topic policy-reafactor removed","accounts_in_message":[],"_revision_number":3},{"id":"7b75519708da4f01962b8bcbbde2e4edcab61d35","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-24 14:57:55.000000000","message":"Topic set to policy-refactor","accounts_in_message":[],"_revision_number":3},{"id":"b79ac5f5b97342b11e6c3eb79f62bedfadd17964","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2021-08-24 16:22:38.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"d604e4d7c291d21b81235718633d35a0dd9d4ab6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-24 18:28:15.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"5285795f98c5746b8aa93495a7a6dbd4ea5f1aa4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-24 19:51:29.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/8f37d6de06bc492db3df4cf90438a077 : SUCCESS in 1h 00m 23s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2c5bbc2281e946a6a26bbca572c67638 : SUCCESS in 1h 05m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/11d745da3a3243588a1e00a283d01bfd : SUCCESS in 6m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/72ba7ef6bd524a6cb2133e44c42659b3 : SUCCESS in 7m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c2d321135a47426ea85e86d4abb37eba : SUCCESS in 7m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9ab9767f72e54431b5e2d09add01b089 : SUCCESS in 7m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c4b6f291ee64b5d8a05fcf79e57c699 : SUCCESS in 8m 15s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2261d2d7654d41f58de405f5de65052c : SUCCESS in 10m 50s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/30548aebaf2241d8b38f71eea57dad3b : TIMED_OUT in 31m 28s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/2c0a626c8c0344309b5514af5c2b8683 : SUCCESS in 9m 21s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/fd8be9e8bb0f4299960168fa89c4fce5 : SUCCESS in 5m 47s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/d2389ca3fad749a1a4a957f01533a94e : SUCCESS in 1h 04m 15s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/af9bdbfe4b7b4b20ba3cc8aff93adc87 : SUCCESS in 1h 00m 23s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/81328d12eef345dabf95428f3e3845b5 : SUCCESS in 1h 00m 29s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/a775fdeafb5642d6af9c03decd6ce91f : SUCCESS in 1h 05m 25s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0631875236b24b6d9cb42bdc379e249e : SUCCESS in 44m 43s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/8fb94ea9d0384e9589ec2ef5f2ec63db : SUCCESS in 1h 19m 17s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/a1168a0a682f4e6f81e12a4e99dc06b9 : SUCCESS in 35m 14s","accounts_in_message":[],"_revision_number":4},{"id":"c7ecf6e9d32e6cc77b49d8d8451b0b15884e119c","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-08-24 20:15:09.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"65f0f551ca69aad057c386f7e6eb9c7a1149bae8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-24 20:17:05.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"4a42e7377ce6640a0df2ebf0be7a618551728f3c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-24 21:30:12.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/e4868b6d8c6e496ea175de5264ab956a : SUCCESS in 1h 05m 09s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/82db8e757cd54951b870b86e392f9e7a : SUCCESS in 1h 07m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2182195fed394d75879dc74065f9416e : SUCCESS in 6m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bbecaf364b5c40909107e81855b19dc6 : SUCCESS in 6m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c065a80823854ab99fca5fc2c80d82a3 : SUCCESS in 7m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/347f0d623ab64dbe848d2538e690108a : SUCCESS in 6m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71e128a0992948f58acda34593e89862 : SUCCESS in 8m 00s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/efd49a8286bf41e0b40d7d02beec944e : SUCCESS in 9m 56s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/8ad11d632c094e0187a04636c4d3fcfe : SUCCESS in 9m 09s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/86bc5cce1b0649a08be91e8f21509842 : SUCCESS in 9m 19s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/58c8fee5b8284e98938a409e51130b9d : SUCCESS in 5m 36s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/19448a3b885a446293d7e39ba647b32d : SUCCESS in 1h 02m 56s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/e2007a2224fd42fc82be51928508b15b : SUCCESS in 1h 06m 00s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/49870405b5eb42fdafd1abe5fefdbdcb : SUCCESS in 1h 06m 45s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/ad9ea677994e4c77ac9807cf8263f41f : FAILURE in 1h 00m 29s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d843336c76b048fb97bbdc72146eb82b : SUCCESS in 55m 12s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/60c21dcaa717416b8f8950b9d0970928 : SUCCESS in 59m 04s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/5068363e6d8f4331859639984a886365 : SUCCESS in 27m 22s","accounts_in_message":[],"_revision_number":5},{"id":"94d79b5668e5f432ec69d7dea47d446d46b0af88","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-25 06:37:35.000000000","message":"Uploaded patch set 6: New patch set was added with same tree, parent, and commit message as Patch Set 5.","accounts_in_message":[],"_revision_number":6},{"id":"c32bd090adaa50c4a5e61e319530cd8e00198f42","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-25 08:16:02.000000000","message":"Patch Set 6:\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/408a6b33e0d9469d98f6dbaae5f72e58 : SUCCESS in 1h 01m 53s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/64607560d540452884b47994b80b4d97 : SUCCESS in 1h 06m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/44fdbd49a5c54abc8a93f76ccd8d7534 : SUCCESS in 6m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6bffb146bc604597adc938eda677143f : SUCCESS in 8m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5085b78f9bdb4156acf6f61a720aa04f : SUCCESS in 10m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/87dfa17b87d845aaa8e055076707f677 : SUCCESS in 8m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/852a4272fef74b2fb89e4d6b129ec0a7 : SUCCESS in 8m 41s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5dcdb082bc4e4299a0652e5dd9b1d8f0 : SUCCESS in 13m 12s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/90629e5052f34a8ca55810e34e3e32fd : SUCCESS in 12m 17s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/79ee64410cf5406da6afcf5e668747f1 : SUCCESS in 12m 20s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/3fd8f24b21834482ab3f195a5ced7252 : SUCCESS in 7m 31s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/0391f3ba5d3b43ecac3d3b350f32d432 : SUCCESS in 1h 13m 29s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/0ebc5c8d9e784441a2ee7bca1bdd074c : SUCCESS in 1h 03m 44s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/0bea0eaeb70f4a0b985c8fba49965b1d : SUCCESS in 1h 17m 25s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/62bcc3bf2acc4510b84eec610d6e1542 : FAILURE in 1h 07m 50s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ee45972663914ba3893e9bfe486504f6 : SUCCESS in 51m 55s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/5b6bd711d6f1423db854bcdf21a23bc3 : SUCCESS in 1h 11m 42s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/42196367aef14ab088e4fd1f0e9c707f : SUCCESS in 28m 14s","accounts_in_message":[],"_revision_number":6},{"id":"3d0a86c64984b75e7b28df5e428b22c80d0d4c7b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-25 15:52:23.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.","accounts_in_message":[],"_revision_number":7},{"id":"de013cdafc7a4d8c4ef4ed8e363028320d756c2f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-25 17:46:31.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/0e6e58c690464e768361040434281335 : SUCCESS in 1h 02m 50s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/42416ea708a44ceebbc5f489e9b59664 : SUCCESS in 1h 13m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5be9e9b21ec84652bab6be561988f775 : SUCCESS in 7m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/22cbc4d8d6ab4851acacf0cb094f34d7 : SUCCESS in 7m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/77aecc6a97984cec913ad91bd52db9f6 : SUCCESS in 6m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/835f84c06d9a4290b076fb8d0e98c064 : SUCCESS in 6m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3868d8d9bd08410cbe3a4b855989ce80 : SUCCESS in 7m 55s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7df743b102344b889b01570a11c98e76 : SUCCESS in 10m 44s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/2d4f440fc01c4423b73be4d990a470ae : SUCCESS in 11m 20s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/fa746191ce1b4b35b81ad59b226e013f : SUCCESS in 10m 30s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/ea2a4063599e4a4482f66bc4ca26d0e6 : SUCCESS in 6m 16s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/2ae7bef910d34e1db99df72f399057d5 : SUCCESS in 1h 08m 16s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/cc8763cd93404b9698defcfa0c7ee579 : SUCCESS in 1h 03m 09s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/b11125a066c84df2910f041b5ba76221 : SUCCESS in 1h 10m 07s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/5b3b897aad9c4e768c9c91009e69d55c : SUCCESS in 1h 23m 08s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/77a9f835fd1048bab2f175a14450ca57 : SUCCESS in 51m 57s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/a62793f1a805460f9e95221a78e81c3c : SUCCESS in 1h 41m 37s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/9785af94a2524b17bfb07fe40a0d898f : SUCCESS in 26m 36s","accounts_in_message":[],"_revision_number":7},{"id":"4bde8b79e4e12c30433a6932f77d270633bb15a5","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-08-25 17:57:10.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"1a59d7e3552e46a9cf35c233ca619b1b9b218cb0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-25 18:45:30.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"7d56cd6ea31ebb7b2f6a6de2a1108446b9c0ae65","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-25 22:33:02.000000000","message":"Patch Set 7: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":7},{"id":"f78960e065a404961c06ad4e314bbed9424ca95f","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-25 22:36:48.000000000","message":"Patch Set 7:\n\nrecheck parent failed","accounts_in_message":[],"_revision_number":7},{"id":"3fdbc4c38a6ea4f0f34dfde3b9d00231442a5716","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-25 23:58:14.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/f8e7e17e029a4127a864f669f64766e7 : SUCCESS in 1h 09m 34s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4dc78d5345fc42328753a6a4bbee9115 : FAILURE in 59m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f7d4a5199b924be4955c33f0782ec79b : SUCCESS in 5m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f117b1ee782f4f118c503f97070c5bd0 : SUCCESS in 5m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e010164f8f7a440882114318deffebd0 : SUCCESS in 5m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9af9fd74929842ff9cdf9806baaf1bb5 : SUCCESS in 6m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a55b54e2051145b898cf51cbae910718 : SUCCESS in 6m 43s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b4db4bcb36324f6e832fe0e28acadda7 : SUCCESS in 7m 26s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/89c35856938642d0afa2433cac122d3a : SUCCESS in 8m 01s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/6e638feb6b8b4cab9946d8276dfe69fe : SUCCESS in 8m 41s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/dd50c9ab1e3a4e38a685dd7477491cbc : SUCCESS in 4m 59s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/459b03d10c1c4a28b8b4e9b376eea887 : SUCCESS in 1h 06m 14s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/ade83ca612594fd7876d6b81e2de3c76 : FAILURE in 58m 29s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/16f87ea1ca774865b6137bcf10c7ca1c : SUCCESS in 1h 04m 18s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/ec126779c4fd450cbf36443af67c1f0b : SUCCESS in 1h 15m 11s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0870ea9f825d40a896cc011d49dd2b76 : SUCCESS in 48m 14s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/02e3f2d3d6fa4ff7909af311fb059cef : SUCCESS in 1h 15m 52s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/32c2ff947872437b87af85cd31e6b5cf : SUCCESS in 31m 28s","accounts_in_message":[],"_revision_number":7},{"id":"a34221d098bb39733782bd8f5f1a91ac6fe6c3a0","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-26 03:33:54.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"7c882c2208373bb681194cfc2720c0bee08d2d4c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-26 05:16:54.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/5a15767c213b47b1bbb9d90dd647fd95 : SUCCESS in 1h 05m 41s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/438ab7d0fb644b3d82995ea9ab14fdfd : FAILURE in 1h 16m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b31d71d6a165477b868d3cce47a458fa : SUCCESS in 5m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a020929f3f9843d29557307f42ccf4d1 : SUCCESS in 5m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/40dc6b94da204401969f08b0f1c2449b : SUCCESS in 6m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2e4d43e422104723a83cb14db72c0752 : SUCCESS in 6m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f25de2ae7afa4d3c8145650e51a84192 : SUCCESS in 6m 53s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a2dd14df6459425cafd25c15a585c194 : SUCCESS in 9m 09s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b71cc29655c84067b5a80599e9f14179 : SUCCESS in 9m 43s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/02e3b50ef49342aa80dc221834e3b745 : SUCCESS in 8m 43s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/14d3b56348e84506b88c29db9d1582ae : SUCCESS in 5m 36s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/f8bb360bafb0410cac27a65d1d1aafe7 : SUCCESS in 1h 27m 14s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/2525eca86a9a41b6a2e85afbba40ef4c : FAILURE in 1h 07m 37s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/20c7868558004d2b9b459ca87f37792d : SUCCESS in 1h 10m 23s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/bdf80f1897a3441a80432d5e7df646e0 : SUCCESS in 1h 37m 04s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e3cdc647699d4d07839b4a3aa958a411 : SUCCESS in 1h 02m 37s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/72a34fffc7ed4020b74ce50426c50358 : FAILURE in 1h 20m 37s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/af8639f91b0a40d884e089d2dd1aff1b : SUCCESS in 32m 27s","accounts_in_message":[],"_revision_number":7},{"id":"0234a618a2e7b4897a7c0d2e25e6167c0bf750c3","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-26 05:17:47.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"d3f77076c85dd8cd30ad1e875677c192253eeedc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-26 06:39:33.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/0db98e84541d4eddb146b01a61269b40 : SUCCESS in 1h 02m 15s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0e2b671fe55e4b76bcf5e06eaa7f7932 : FAILURE in 55m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/506f46d15296421eb9fb731db634d0ed : SUCCESS in 5m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b4a72bbec6b24dc7a81d425480ac8020 : SUCCESS in 5m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cf393586409747a58344754d496af16f : SUCCESS in 5m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/799b4b5058384bf4809bddc8a5fadeb5 : SUCCESS in 6m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bbc126ccc0b14e04b917e0a0e3820b33 : SUCCESS in 6m 18s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1fd07e1e353148ac84a24e2dd09d4eda : SUCCESS in 8m 42s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/5a7b48cf322940139afb99741b8b7ca2 : SUCCESS in 7m 48s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/8dc0093b923846589d7c7d84139d935f : SUCCESS in 8m 31s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/73f27988850246779d826e37437ca9d5 : SUCCESS in 16m 51s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/33c800e6b6854ce3bc03d423e5863ae9 : SUCCESS in 1h 09m 13s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/a9ba8221d1d04aa28b6d81f8e270c661 : SUCCESS in 1h 10m 39s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/f5dd240040b54b428b46ca87b1a4fb2f : SUCCESS in 1h 06m 34s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/02383702a1604d08bb0e1509a8ac6074 : SUCCESS in 1h 12m 44s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/21956bf2343c4d3c8c98274766e72c52 : SUCCESS in 49m 11s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/fa6424bd8c294b5ca0b96375b4461252 : FAILURE in 1h 20m 07s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/d8071535d17145e699f93eceb3b9b8b3 : SUCCESS in 26m 28s","accounts_in_message":[],"_revision_number":7},{"id":"4d8c3becd20b0ad784d78d683dabf4e5bfb8f21d","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-26 07:03:41.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"a10274b362cad2e6d2d1e7681f282bf00f9bf74f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-26 08:33:03.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/f47ffd2e604e48548ff387d48f90c805 : SUCCESS in 1h 04m 08s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/aac9a4e782fa42ebb6f320e8d39c66a0 : FAILURE in 58m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/180c8e62e30141fcba2553ab77e21698 : SUCCESS in 6m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3b1ec123f0ea4147aad7e81885a80dcd : SUCCESS in 6m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c8f3f61361be4ed8ad17fd2f8f66014b : SUCCESS in 6m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/22149322319d410b96bc1ca56cae4710 : SUCCESS in 6m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fcb2331ce6fc46ab842dc0b0d5e58a26 : SUCCESS in 6m 44s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3d10c48da3a24176a0b1040a9218a82b : SUCCESS in 9m 31s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/6079ca95fdf1420da4bad502e7aaba9c : SUCCESS in 9m 01s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/44826abdffe14da5a5c0d2b1869d026c : SUCCESS in 10m 06s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/6f10cf95a9e14ec592696fd5a109b7b2 : SUCCESS in 6m 50s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/335ab48d77614194a6b8d915cb8affbd : SUCCESS in 1h 15m 03s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/b7174280c6b9491e8064d3e8e9135cc5 : FAILURE in 55m 48s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/a5d3a280bd484603b8a7fcf894a5dfaf : SUCCESS in 1h 13m 03s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/e49cb35e8c6944aab1c20c612c033d15 : SUCCESS in 1h 11m 46s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a693ee721e8343bcbc9844c824f3544c : SUCCESS in 49m 39s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/e8709d6652d74d96b333bec9060e65ae : SUCCESS in 1h 04m 44s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/122a0dca01f7465298c6d129dbb336ba : SUCCESS in 30m 07s","accounts_in_message":[],"_revision_number":7},{"id":"b97a86b3d04f7a25bd61bdb3c2881347c5e79dbb","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-26 08:34:49.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"7672ec6dbde9ff392e04e91a6d17ed3951ff2f3f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-26 10:05:45.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/a4f995973ee9407b98f6c5fd5d93c2ed : SUCCESS in 1h 03m 30s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/973586b6a8c74dc0bcbfff51a9c625b6 : SUCCESS in 1h 06m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5fd45ad92e7d47548d70bffd30a909db : SUCCESS in 5m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6bf5a464e0dc46a894cb1aad14acda81 : SUCCESS in 6m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bb4927c1d11f455f95bb626d3b350794 : SUCCESS in 6m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b8d18704bd344c7f8f32cc894fef91cd : SUCCESS in 7m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24bbb56702c34439852036db7ec284ad : SUCCESS in 7m 40s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0e29e195091144edb99ad5f45646f526 : SUCCESS in 11m 17s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/26f10d97e6644a2fb37a4d93df720c51 : SUCCESS in 10m 27s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/3a58128f62444edeb152be46ca8933a6 : SUCCESS in 9m 03s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/24312934112e409aafd74585e015ebdb : SUCCESS in 5m 37s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/61bb9580a335472484bf73e57cdbf730 : SUCCESS in 1h 10m 36s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/d4a720a384264ae3885a80f7ee7a2f81 : FAILURE in 1h 25m 42s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/12d1e44ab60c4af59a76623a692eef19 : SUCCESS in 1h 09m 42s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/2eef8d5062e244b89885d884741e0322 : FAILURE in 1h 00m 15s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/794d4828f40e4a12ad5606cb53e74ece : SUCCESS in 51m 43s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/c4c9fa2a56e24405b105f295dd5f254d : FAILURE in 1h 26m 22s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/4b0bdf6f23a94639832b0d08f62ea9c2 : SUCCESS in 32m 59s","accounts_in_message":[],"_revision_number":7},{"id":"5b70480470d676cebdb6578c0afca6f5ee052a3c","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-26 11:49:36.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"96d6bb51bab46ab26b9f40d985a8d0bbf44dba92","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-26 13:15:16.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/b8cd07f8f2814e7e965af755ca50b321 : SUCCESS in 1h 07m 36s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/cb7c958817c9475d90ee4567fb207bfc : SUCCESS in 1h 06m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0fbe862bd3dc49ac8b35b0a2e8f26e18 : SUCCESS in 8m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/39380fb53702487ba41e7ae61e237db8 : SUCCESS in 6m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1a50a554af314269b404401167a18e76 : SUCCESS in 7m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/411a7f5e482148be81aa45e1dfda65a1 : SUCCESS in 7m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b36102af8da440b8d89e5338c68b62e : SUCCESS in 8m 11s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4c43441c800447d1be7e7a7dc407e53a : SUCCESS in 10m 12s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/39fb5377b5f04e91bd229eca5f23ad30 : SUCCESS in 11m 51s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/1a3f8f581da14321951ff20eb6022aeb : SUCCESS in 10m 03s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/829d4747a0ad466aa13da0a44bc08022 : SUCCESS in 7m 29s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/e0fd871bdcb340f5b33c808bf8188108 : SUCCESS in 1h 15m 50s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/ecf5def4f3cc4bf5bb7e1946365e9eb9 : FAILURE in 1h 03m 42s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/e582d8aac1dc4c3ba4927e58854145d4 : SUCCESS in 1h 07m 21s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/5f8b6904911e44ada38aad4579cde000 : FAILURE in 1h 05m 30s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9cd9fca97daa4c9780881c43f94bc57d : SUCCESS in 53m 24s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/c56857a3ed20414ba6d4a335cfd6b7e6 : SUCCESS in 1h 10m 19s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/d2085f7e1e6143ab8229ac132e5e5803 : SUCCESS in 31m 13s","accounts_in_message":[],"_revision_number":7},{"id":"0aeca844cde4fd8f6dff0cf04ed4b9fe81747114","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-08-26 15:28:06.000000000","message":"Patch Set 7:\n\nrecheck compute setup teardown fails","accounts_in_message":[],"_revision_number":7},{"id":"3bac9822b21cfe45ab2959550da3647bfbb9bf66","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-26 18:12:35.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/afea50f6f7d6441d8985e78c3f0e81ab : SUCCESS in 1h 10m 59s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/549b3338aa97459d8015906813c7889e : SUCCESS in 1h 18m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ebd0021791ae4aeb8d1493e757695053 : SUCCESS in 6m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/46ba16c697144ec092a74963536a5750 : SUCCESS in 7m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/242c3a3765e64dafb6a4596a1946c22e : SUCCESS in 6m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/65eafb265966412aa3418ae27359cd25 : SUCCESS in 8m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7c3a3bbdef14454b1f1727489d2626e : SUCCESS in 8m 10s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/cf873f8e2e6944cebf58254c2a746970 : SUCCESS in 10m 39s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/2cf688d0e4554e679b709aeab8a34200 : SUCCESS in 12m 24s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/e6801cf562fc4bc1a74bc35b0eaf5cdf : SUCCESS in 11m 56s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/3a2d1e8d4720429da54f8dbcb0dc2c6e : SUCCESS in 7m 02s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/40a3100da89847caaa87aa7592f12f9d : SUCCESS in 1h 20m 38s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/208ceee501e34219b763fe6a7387c969 : FAILURE in 1h 06m 18s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/92e83780812d4b9e82a48ea8d3c3173e : FAILURE in 1h 10m 40s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/3379b8900213447d901cf98114d22f2f : FAILURE in 1h 02m 39s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0440f9ef8fdb4000a0a1fde6fd3358a6 : SUCCESS in 49m 19s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/9ba8c0581d2b497ebec83309d5349e6e : SUCCESS in 1h 15m 38s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/c3442a38ddf34ef480850ccced294d01 : SUCCESS in 31m 47s","accounts_in_message":[],"_revision_number":7},{"id":"6e2f8d7ea91a730d60c40540c8251bf5930aaa00","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-08-26 18:17:00.000000000","message":"Patch Set 7:\n\nrecheck nova fail again","accounts_in_message":[],"_revision_number":7},{"id":"5d83f05080bf099465bf23b59aa92f1b44e4b529","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-26 20:04:45.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/c693197aec94440aafad89f1e2017258 : SUCCESS in 58m 21s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/cb96b11be2504051a110d059dd3bcc57 : SUCCESS in 1h 03m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/22b1fe0ca35b49e7b635b6d95203d813 : SUCCESS in 5m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4afe23935107472abfa09bae023a5228 : SUCCESS in 6m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/718ff03e0a624924b5da4141d01ea320 : SUCCESS in 6m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/637196ef8d964caa990b2225d6dae6ac : SUCCESS in 7m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d78c720de25d4acc876cd10064a75b8e : SUCCESS in 7m 41s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6efea585bc164a67a03ca71db88611d1 : SUCCESS in 8m 41s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/9fb3516becb3457eb97125ff1b3659a8 : SUCCESS in 13m 36s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/397fcc91783046198a42baa6155c7893 : FAILURE in 13m 07s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/98ccd0f3f3514a24ba4d56f6babd8387 : SUCCESS in 6m 14s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/0c63526928e645c0b87f8a880a1f63fe : SUCCESS in 1h 15m 39s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/c78ed27a5fb04d27b4ff2f163c56aac6 : SUCCESS in 1h 05m 54s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/e3c50fbdba454a27b307f27a51334c04 : SUCCESS in 1h 10m 06s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/73cbadb0076b499da2de4348b9877368 : FAILURE in 1h 07m 48s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3c3417a9ffff4ec5b23c0a516a054668 : SUCCESS in 47m 32s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/9ba77eff11b342ddb3c8f202015d9604 : SUCCESS in 1h 07m 06s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/8f683966408546798f3670e82943a907 : SUCCESS in 30m 37s","accounts_in_message":[],"_revision_number":7},{"id":"1c4017e7be0b86253d63f22a75f70a27cdd13e3a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-26 20:05:27.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"81ec5c9e01d590528290742d07789f5b59fb9695","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-26 22:57:27.000000000","message":"Patch Set 7: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/00325e6f10dc443aae91bc0258c78c78 : SUCCESS in 1h 00m 02s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/21dabe608139430f97c86c743c777305 : SUCCESS in 1h 08m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b175b29541c44ea49f5a267dce048603 : SUCCESS in 7m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/150f9fba64014d109dce3a915517fe57 : SUCCESS in 8m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e48991eb26f74f298cad0b99245f080c : SUCCESS in 7m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b352021577224ac4a259dca33d50291d : SUCCESS in 8m 13s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/20bbfcc4f04940a990028c5e9a1ee1dc : SUCCESS in 12m 40s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/149725f9c5d642279b5b5205fc19ec61 : SUCCESS in 8m 35s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/1842f3e7e2a741f29e6367782fe0e78f : SUCCESS in 1h 15m 53s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/af98f8ca70324c0d911a5ff993be75a7 : SUCCESS in 1h 06m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f660dd691cc745288bf47f294b6dea7c : SUCCESS in 50m 10s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/eebea750629c424594bee971c3bc1208 : FAILURE in 1h 14m 45s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/f5541d7a92194e388eea8efe652751d3 : SUCCESS in 25m 52s","accounts_in_message":[],"_revision_number":7},{"id":"8f413b0006573324a8aa386e848a02c0a790f5b2","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-27 06:11:19.000000000","message":"Patch Set 7:\n\nrecheck nova failure","accounts_in_message":[],"_revision_number":7},{"id":"f889157a110cc1a4e4b29933280f01cec37f77e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-27 08:33:15.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/8c0c1875e858429fa9811de27e272009 : SUCCESS in 59m 53s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ef1df2aeb3a44e51944bb39a9a66f8f3 : SUCCESS in 1h 05m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58d619c49c894a61a6982181be5548cd : SUCCESS in 7m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/34435addd92a490b9a287e87b58d8cb6 : SUCCESS in 8m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/849c818e8bc74ac09fbc09c2cb1e882e : SUCCESS in 8m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9603295e238f49acbca2346d08dd701b : SUCCESS in 9m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c54872720004119ab0c6698159c184a : SUCCESS in 8m 13s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3077cf072f174d7dbb47bf4a86de765a : SUCCESS in 10m 31s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/45d958cf4b9e4ca597de78fdb5b66a53 : SUCCESS in 11m 55s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/14d382bd701244929d54fcaac2441730 : SUCCESS in 11m 04s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/f75722645b654dcc95fed6758fbe8c1d : SUCCESS in 8m 40s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/42e4012df1074647a66d510ea1fa5e10 : SUCCESS in 1h 19m 34s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/7ac62be5d298453584a198556bf6893c : SUCCESS in 1h 14m 12s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/3c10281242674687b9e9aa38e17944e6 : FAILURE in 1h 05m 05s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/028e5e72098043299006f487f4c37c29 : SUCCESS in 1h 15m 50s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/470e96196c424084a48a9e5cbb959cc3 : SUCCESS in 52m 58s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/0779e830b9a245038b3f3f9eae2db4d4 : SUCCESS in 1h 10m 10s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/2d9e61b034ca4b63b6dbd7c1742c76f6 : SUCCESS in 32m 10s","accounts_in_message":[],"_revision_number":7},{"id":"2402d90cf4303fc1e07a39c24f2a5f5be9d16aed","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-08-27 21:27:35.000000000","message":"Patch Set 7:\n\nrecheck same nova problem","accounts_in_message":[],"_revision_number":7},{"id":"b9aa6c896dd7b035c105a70b6ee57dbde22766a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-27 22:46:38.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/bb49b540f3f248c2af43339e1f50383e : SUCCESS in 1h 05m 54s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/211e9321a2e448049869b047566cc60d : SUCCESS in 1h 14m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/965c500e16e2401e8c8de44a59ba1825 : SUCCESS in 5m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d6128564fceb48f58655fdce99955726 : SUCCESS in 6m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9a3731372bb74b55bd50a3d25815180c : SUCCESS in 5m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9aed9056f9724d3c98b5e1c06442419e : SUCCESS in 6m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6823ed2c57ce448c9c7b8e6880ec7550 : SUCCESS in 6m 51s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6c07f71b991a4e87b6b53a70a777fbaa : SUCCESS in 9m 02s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/68f6f4a78c7345c6b776bd627e1fb8d2 : SUCCESS in 9m 13s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/03e72cbbb3f84dd98de1dc3f65581ed2 : SUCCESS in 9m 28s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/51829b9b1e71445d8e22f6b244bfe318 : SUCCESS in 5m 48s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/0785414114b74f85b697f3e33c7ce976 : FAILURE in 1h 11m 10s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/c56be869181348d8af3942bfef937b7e : SUCCESS in 1h 06m 37s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/3630cabdaa744c4bb22fe1b7750c195a : SUCCESS in 1h 12m 23s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/61900b32053b47c7a8cb6602407c2d3a : FAILURE in 27m 21s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ea24620d117e4020b914cd61a5c1765d : SUCCESS in 48m 25s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/b87a5b9ea1a0408f9654c5d811ddd29c : SUCCESS in 1h 01m 29s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/32b101f629194347bcb3d795e1fdff69 : SUCCESS in 31m 50s","accounts_in_message":[],"_revision_number":7},{"id":"d2113f34011a3c0d8e7417fee7d06d857f949f3e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-27 22:47:12.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"cc5639168ebfa92d8b8819f986f1039d01a7d663","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-28 01:20:14.000000000","message":"Patch Set 7: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/a870499f3e7f4bc68cd07377a55e3703 : SUCCESS in 57m 23s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/fb380e1dd9d7445ab5ef11f5f8f938ee : SUCCESS in 1h 19m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc2424ea2f374ee09446186afe913f5a : SUCCESS in 5m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e560b85c9b95493db8efb5d3ced55e45 : SUCCESS in 7m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e8aab7d51228481c903806b92aa0043a : SUCCESS in 6m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f18ad71d93af473ba3e012708bd3e9c4 : SUCCESS in 7m 20s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c02016994ccc4af98b1e97e5e46516f2 : SUCCESS in 9m 37s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/e37b32423e2444888494bb8b4a0ae546 : SUCCESS in 10m 14s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/7f509b35657e4aafad2edf7a227f0cc6 : SUCCESS in 1h 10m 13s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/41cdf61c08f44dd394c0d2031c83164c : SUCCESS in 1h 10m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ce5c2bcdffbc4cb9a4a790febef1fa33 : SUCCESS in 52m 41s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/0a2546b69f43496e8361a0359da14e89 : FAILURE in 1h 12m 14s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/8a95050cd68f46cd882987d3196dd46c : SUCCESS in 27m 04s","accounts_in_message":[],"_revision_number":7},{"id":"ecb44e4ec5a8076314c22a9b43f068158a0eac8d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-30 05:57:15.000000000","message":"Patch Set 8: Commit message was updated.","accounts_in_message":[],"_revision_number":8},{"id":"7716dd0a0b4f5f025e5873da8a99f783652f4565","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-30 06:12:03.000000000","message":"Patch Set 8: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":8},{"id":"7db592d28d1d06481813934f1f38164a2121e124","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-30 06:28:48.000000000","message":"Patch Set 8:\n\nrechec","accounts_in_message":[],"_revision_number":8},{"id":"fa714cc3910e5689fc454888ab267e1239740abe","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-30 06:31:47.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"089bd4ebbe02e7e6ee9b819ea93420d0ed141a24","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-30 08:19:48.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/90478eb4056440449646f5bd9e112405 : SUCCESS in 1h 00m 40s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5644ed40ab7c43fca7be9d62caa653ef : SUCCESS in 1h 01m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6cc9b88ecee24667bb9f792eb39be855 : SUCCESS in 7m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d470987e3bb44277ada7a2deb9292e72 : SUCCESS in 8m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6210f30293034854af27e8ddc2a3c1ba : SUCCESS in 8m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/723c17e021fd4052bdcab03157a311b8 : SUCCESS in 8m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b477d3a2b57c4c3c90121714ec079600 : SUCCESS in 7m 50s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8608268e57fa41a0b36b3835778e65b6 : SUCCESS in 10m 00s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/a684f5a397d34e56979fa445d645f8f6 : SUCCESS in 10m 06s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/2e16c48bc0494d648ebfb62cc4a58658 : SUCCESS in 11m 29s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/672694f4d11f47bc90f4a3d1915c667d : SUCCESS in 7m 39s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/511efee302e14dbf888c32cff7bf3e3b : SUCCESS in 1h 15m 28s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/0c2bc46bbda2405082aa050dbd9178b3 : SUCCESS in 1h 02m 07s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/2c92c63eb90a4b8ab679d5df4ae85873 : SUCCESS in 1h 11m 49s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/f167890c6bce4b48bb216d4ba34450f2 : SUCCESS in 1h 12m 14s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ec7a8a1773124797a5d61d1c0a0d8c68 : SUCCESS in 49m 23s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/fc14a325cd284e62bda24733989faa1f : SUCCESS in 1h 16m 03s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/8ac4024f147a4eff99ca689c78e51304 : SUCCESS in 33m 50s","accounts_in_message":[],"_revision_number":8},{"id":"1ebdbadbaa902b228fbb2bf4f663088d2b5095e5","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-30 08:37:59.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1\n\nCommit message updated to include depends on since last approval.\nNo other changes, so restoring +2 and +w based on that.","accounts_in_message":[],"_revision_number":8},{"id":"5f53b0bd65093aa396dd0490926fa5a7b5cdc968","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-08-30 13:35:28.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"7d78964b3700576e5a4a67ef176ed0aef4da30b9","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-30 14:23:31.000000000","message":"Patch Set 8: -Workflow","accounts_in_message":[],"_revision_number":8},{"id":"0b92d6e5a48b202dc6a159b78a1c10a25e9e2bcd","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-30 14:23:38.000000000","message":"Patch Set 8: Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"ff7aaecb502079523c07493a0572c6caf7c73c48","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-30 14:24:26.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"2f90add0b5c52f8d4cbce0a95717e36985bcd950","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-08-30 14:50:58.000000000","message":"Patch Set 8: Code-Review+1","accounts_in_message":[],"_revision_number":8},{"id":"69f510a9b4ccec2c445f7558dfa55006f4eb1f89","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-30 17:16:06.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/2abaacf89ba04665a6eec5facb73bc9c : SUCCESS in 1h 05m 28s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/bb5f30170bca425e832c99114a03ecc1 : SUCCESS in 1h 08m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/20942d59db274a96a94d930e03f44b45 : SUCCESS in 6m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6b847e456b5f41fea27af4172f31d9ea : SUCCESS in 7m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/81810c03c29a43cba49f8a9197ba5237 : SUCCESS in 9m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d78bb4848ed4e0896c407540b08aa0e : SUCCESS in 10m 04s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f9d650fae25a45658111e20a68062ee8 : SUCCESS in 11m 16s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b94904f164a447659371ebf6b47113e3 : SUCCESS in 9m 45s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/dd252a3414a64e318815b191f1d23946 : SUCCESS in 1h 13m 20s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/72adc67de05a4933935418b51fc1a1c4 : SUCCESS in 1h 09m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b3f6247cd9e14a4a9276016c91a61d65 : SUCCESS in 48m 31s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/8b855953ca704e4988beaf16bc1306dd : SUCCESS in 1h 04m 59s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/da1965c4adea4127b0aa97b9377a2747 : SUCCESS in 36m 37s","accounts_in_message":[],"_revision_number":8},{"id":"e43993cc29690b91cb1ec823d38134840f20308a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-30 17:16:06.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":8},{"id":"8cc12d6bb7e987991c4e94df8700d5e3af2894b5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-30 17:18:02.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69888e6e0d694295967432b536dc6dc3 : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":8},{"id":"b3e9a2c9a2e34d10b4ec029fc16f149eab8fb382","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-08-31 13:34:32.000000000","message":"Patch Set 8:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/805590/805590/8/check-not-vote/ext-glance-zuul/1a2f9c6 : SUCCESS in 37m 38s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"ab0c95da68d992de278fe8bdaf522f25d1a30759","revisions":{"45162cb3ccd4ba0e584e7e85ae209175b92dc8c3":{"kind":"REWORK","_number":1,"created":"2021-08-23 07:46:57.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/90/805590/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/90/805590/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/90/805590/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/90/805590/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/90/805590/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/90/805590/1"}}},"commit":{"parents":[{"commit":"f55c44a260004e771ddc0bf5718a263b16dfcc80","subject":"Check policies for Image Tags in API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f55c44a260004e771ddc0bf5718a263b16dfcc80"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-08-23 07:40:28.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-08-23 07:41:04.000000000","tz":0},"subject":"Check policies for image tasks information in API","message":"Check policies for image tasks information in API\n\nThis patch enforces policy checks required for fetching task\ninformation for image in API layer.\n\nPartially-Implements: blueprint policy-refactor\n\nChange-Id: I8f59b8405d9c55b5f69294d3f9bd7bcbc064203f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/45162cb3ccd4ba0e584e7e85ae209175b92dc8c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/45162cb3ccd4ba0e584e7e85ae209175b92dc8c3"}]},"branch":"refs/heads/master"},"ccfbeb94cddb9ef4337f1502577174b307c55d6b":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-08-24 06:15:20.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/90/805590/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/90/805590/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/90/805590/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/90/805590/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/90/805590/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/90/805590/2"}}},"commit":{"parents":[{"commit":"56d7a8e57e87f6c7cb9eb264b6093171b15c9de7","subject":"Check policies for Image Tags in API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/56d7a8e57e87f6c7cb9eb264b6093171b15c9de7"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-08-23 07:40:28.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-08-24 06:07:13.000000000","tz":0},"subject":"Check policies for image tasks information in API","message":"Check policies for image tasks information in API\n\nThis patch enforces policy checks required for fetching task\ninformation for image in API layer.\n\nPartially-Implements: blueprint policy-refactor\n\nChange-Id: I8f59b8405d9c55b5f69294d3f9bd7bcbc064203f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ccfbeb94cddb9ef4337f1502577174b307c55d6b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ccfbeb94cddb9ef4337f1502577174b307c55d6b"}]},"branch":"refs/heads/master"},"037d7c24b9daac534a564db688d113f4d286ef6e":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-08-24 09:36:10.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/90/805590/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/90/805590/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/90/805590/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/90/805590/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/90/805590/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/90/805590/3"}}},"commit":{"parents":[{"commit":"d7ae622853007fe1841eb4dbeebe49c162233656","subject":"Check policies for Image Tags in API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d7ae622853007fe1841eb4dbeebe49c162233656"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-08-23 07:40:28.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-08-24 09:28:06.000000000","tz":0},"subject":"Check policies for image tasks information in API","message":"Check policies for image tasks information in API\n\nThis patch enforces policy checks required for fetching task\ninformation for image in API layer.\n\nPartially-Implements: blueprint policy-refactor\n\nChange-Id: I8f59b8405d9c55b5f69294d3f9bd7bcbc064203f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/037d7c24b9daac534a564db688d113f4d286ef6e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/037d7c24b9daac534a564db688d113f4d286ef6e"}]},"branch":"refs/heads/master"},"325bf1c79b024d0850ed3d1bf75479dfb60ec660":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2021-08-24 18:28:15.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/90/805590/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/90/805590/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/90/805590/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/90/805590/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/90/805590/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/90/805590/4"}}},"commit":{"parents":[{"commit":"acd22a2588a348c386e6ef1c1a3ee61641050cbb","subject":"Check policies for Image Tags in API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/acd22a2588a348c386e6ef1c1a3ee61641050cbb"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-08-23 07:40:28.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-08-24 18:22:00.000000000","tz":0},"subject":"Check policies for image tasks information in API","message":"Check policies for image tasks information in API\n\nThis patch enforces policy checks required for fetching task\ninformation for image in API layer.\n\nPartially-Implements: blueprint policy-refactor\n\nChange-Id: I8f59b8405d9c55b5f69294d3f9bd7bcbc064203f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/325bf1c79b024d0850ed3d1bf75479dfb60ec660"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/325bf1c79b024d0850ed3d1bf75479dfb60ec660"}]},"branch":"refs/heads/master"},"f7076cb27bee8ad8a8d5b9117333e3f2c32c6a9d":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2021-08-24 20:17:05.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/90/805590/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/90/805590/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/90/805590/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/90/805590/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/90/805590/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/90/805590/5"}}},"commit":{"parents":[{"commit":"abf34a17413332369817f92602d596843587c604","subject":"Check policies for Image Tags in API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/abf34a17413332369817f92602d596843587c604"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-08-23 07:40:28.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-08-24 20:09:24.000000000","tz":0},"subject":"Check policies for image tasks information in API","message":"Check policies for image tasks information in API\n\nThis patch enforces policy checks required for fetching task\ninformation for image in API layer.\n\nPartially-Implements: blueprint policy-refactor\n\nChange-Id: I8f59b8405d9c55b5f69294d3f9bd7bcbc064203f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f7076cb27bee8ad8a8d5b9117333e3f2c32c6a9d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f7076cb27bee8ad8a8d5b9117333e3f2c32c6a9d"}]},"branch":"refs/heads/master"},"a1d52544b9fd55a192a90f6521e554f36e9abc1f":{"kind":"NO_CHANGE","_number":6,"created":"2021-08-25 06:37:35.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/90/805590/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/90/805590/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/90/805590/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/90/805590/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/90/805590/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/90/805590/6"}}},"commit":{"parents":[{"commit":"7ea1427d38553e168d4cfe1ec667da1f1976482b","subject":"Check policies for Image Tags in API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7ea1427d38553e168d4cfe1ec667da1f1976482b"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-08-23 07:40:28.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-08-25 06:28:55.000000000","tz":0},"subject":"Check policies for image tasks information in API","message":"Check policies for image tasks information in API\n\nThis patch enforces policy checks required for fetching task\ninformation for image in API layer.\n\nPartially-Implements: blueprint policy-refactor\n\nChange-Id: I8f59b8405d9c55b5f69294d3f9bd7bcbc064203f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a1d52544b9fd55a192a90f6521e554f36e9abc1f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a1d52544b9fd55a192a90f6521e554f36e9abc1f"}]},"branch":"refs/heads/master"},"2c27351f4ae157cfb7a058db405610e69565cd47":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2021-08-25 15:52:23.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/90/805590/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/90/805590/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/90/805590/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/90/805590/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/90/805590/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/90/805590/7"}}},"commit":{"parents":[{"commit":"869a39d9e55849c7de7b3ee086cea1017d4f60d5","subject":"Check policies for Image Tags in API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/869a39d9e55849c7de7b3ee086cea1017d4f60d5"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-08-23 07:40:28.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-08-25 15:39:53.000000000","tz":0},"subject":"Check policies for image tasks information in API","message":"Check policies for image tasks information in API\n\nThis patch enforces policy checks required for fetching task\ninformation for image in API layer.\n\nPartially-Implements: blueprint policy-refactor\n\nChange-Id: I8f59b8405d9c55b5f69294d3f9bd7bcbc064203f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2c27351f4ae157cfb7a058db405610e69565cd47"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2c27351f4ae157cfb7a058db405610e69565cd47"}]},"branch":"refs/heads/master"},"ab0c95da68d992de278fe8bdaf522f25d1a30759":{"kind":"NO_CODE_CHANGE","_number":8,"created":"2021-08-30 05:57:15.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/90/805590/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/90/805590/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/90/805590/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/90/805590/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/90/805590/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/90/805590/8"}}},"commit":{"parents":[{"commit":"869a39d9e55849c7de7b3ee086cea1017d4f60d5","subject":"Check policies for Image Tags in API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/869a39d9e55849c7de7b3ee086cea1017d4f60d5"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-08-23 07:40:28.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-08-30 05:57:15.000000000","tz":0},"subject":"Check policies for image tasks information in API","message":"Check policies for image tasks information in API\n\nThis patch enforces policy checks required for fetching task\ninformation for image in API layer.\n\nPartially-Implements: blueprint policy-refactor\nDepends-On: https://review.opendev.org/c/openstack/nova/+/688802\n\nChange-Id: I8f59b8405d9c55b5f69294d3f9bd7bcbc064203f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ab0c95da68d992de278fe8bdaf522f25d1a30759"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ab0c95da68d992de278fe8bdaf522f25d1a30759"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
