)]}'
{"id":"openstack%2Fglance~805967","triplet_id":"openstack%2Fglance~stable%2Fwallaby~Ifbdce0d0386b7c6868654711ae2f98125d9a1985","project":"openstack/glance","branch":"stable/wallaby","hashtags":[],"change_id":"Ifbdce0d0386b7c6868654711ae2f98125d9a1985","subject":"Cinder Store: Update legacy images tests","status":"ABANDONED","created":"2021-08-25 08:38:52.000000000","updated":"2021-09-09 14:53:22.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"43f3197db46dbecbc6166bff4956cf8daa8baf38","_number":805967,"virtual_id_number":805967,"owner":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-08-25 15:56:30.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":-1,"date":"2021-09-09 14:18:32.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-25 09:19:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-09 14:18:32.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"}],"messages":[{"id":"4e71ba375eed452c3cce1b94bdd089d3be2cf1a7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2021-08-25 08:38:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"643603104118d2616ff20f48f9becb2624090242","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-25 09:19:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d57b492c09ca4c2faea1febf0b3f4ad0 : SUCCESS in 8m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/914e1a4dbe674f61bcca913fb4488959 : SUCCESS in 7m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/69ac70bcf79e47b5a782205ef1c48298 : SUCCESS in 6m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/63eae08508384a37a8a8dde80cd03d46 : SUCCESS in 6m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5c472522e4af4dda88a2bc2f94d64c6b : SUCCESS in 7m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e3c1b6bf7014f139ed5e2c8801f3468 : SUCCESS in 8m 17s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3cf5041a47394d3aa039ef0891d17a85 : SUCCESS in 8m 59s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/ecdb3c295c1c4d15870291dc1721210d : SUCCESS in 9m 30s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/de609877f0014dd5a4a28c2a78db3ad9 : SUCCESS in 5m 52s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/cb9eddd5131143998023dbe17f7cc8de : SUCCESS in 36m 24s","accounts_in_message":[],"_revision_number":1},{"id":"b7a8d9fac8eaad5d8bb2bf0d6c509e00213ff19b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2021-08-25 15:08:54.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"745930f1e198f7fd24cf44df4e838df0ef644d28","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-25 15:56:30.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5a3bca2b3e374d9e81f64f700e2f87b1 : SUCCESS in 7m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b4a963fc9d243319475a45a9dfb1cd9 : SUCCESS in 7m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a047107acdb44c56a0e8a319c584e358 : SUCCESS in 6m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a0792d21b21d421c8ac72cedec84a8bd : SUCCESS in 6m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/40f41b48febd40ef822fbfe7ed8a34ca : SUCCESS in 7m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2eed6aa92b24ae09d686b356f6b126c : SUCCESS in 8m 37s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/fa36afbfa9e9445c9085840cd446590a : SUCCESS in 11m 13s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/6edb7aac7ef64053b9bbe619ec178760 : SUCCESS in 10m 36s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/3611d5ce9b464f8197ea9420eaa3213e : SUCCESS in 6m 01s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/95d5d833c0db4453b503bb132482f5b4 : SUCCESS in 35m 12s","accounts_in_message":[],"_revision_number":2},{"id":"7828da0e5b5eaf7aa9e3c4a6c1f82e21c2d70d43","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-09-09 14:18:32.000000000","message":"Patch Set 2: Code-Review-1\n\nif this is cherry-picked from xena to stable/wallaby how there are two cherry-picks ?","accounts_in_message":[],"_revision_number":2},{"id":"6e3972e421b18249b3e044a56c0da25af7456e52","tag":"autogenerated:gerrit:abandon","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2021-09-09 14:51:35.000000000","message":"Abandoned\n\nThis is a supporting patch and since we are not planning to backport glance_store attachment changes to stable/wallaby, we don\u0027t need this anymore.","accounts_in_message":[],"_revision_number":2},{"id":"43f3197db46dbecbc6166bff4956cf8daa8baf38","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2021-09-09 14:53:22.000000000","message":"Patch Set 2:\n\n\u003e Patch Set 2: Code-Review-1\n\u003e \n\u003e if this is cherry-picked from xena to stable/wallaby how there are two cherry-picks ?\n\nThis was squashed from two changes but i think there\u0027s a better way to mention that in commit like we have done in cinder with https://review.opendev.org/c/openstack/cinder/+/802781","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e6a0403985902a2a2276aa9ba6a300e4523c43dd","revisions":{"317421178c968de949a62687a5f10609683900a8":{"kind":"REWORK","_number":1,"created":"2021-08-25 08:38:52.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/67/805967/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/67/805967/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/67/805967/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/67/805967/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/67/805967/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/67/805967/1"}}},"commit":{"parents":[{"commit":"c127bbcbdff5575d717437d023f6c7f869311ce8","subject":"Merge \"Revert \"Remove all usage of keystoneclient\"\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c127bbcbdff5575d717437d023f6c7f869311ce8"}]}],"author":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2021-04-05 07:43:34.000000000","tz":-240},"committer":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2021-08-25 08:36:37.000000000","tz":-240},"subject":"Cinder Store: Update legacy images tests","message":"Cinder Store: Update legacy images tests\n\nupdate tests for cinder\u0027s new attachment API code in glance\ncinder store. This leaves a couple of mocks in place for the old\nway of doing things so that we can tolerate the old and proposed\nchanges to glance_store. We can remove those once we require the\nnewer glance_store version.\n\nChange-Id: If4db20affbba37806f8419b6b30050a32bcc2435\n\n(cherry picked from commit 01b5bf5ece9a6904c1c5df4ad557967c927b8683)\n(cherry picked from commit 0e2043d51ebe90acd0e6f284774c5411cb1c886b)\nChange-Id: Ifbdce0d0386b7c6868654711ae2f98125d9a1985\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/317421178c968de949a62687a5f10609683900a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/317421178c968de949a62687a5f10609683900a8"}]},"branch":"refs/heads/stable/wallaby"},"e6a0403985902a2a2276aa9ba6a300e4523c43dd":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-08-25 15:08:54.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/67/805967/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/67/805967/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/67/805967/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/67/805967/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/67/805967/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/67/805967/2"}}},"commit":{"parents":[{"commit":"c127bbcbdff5575d717437d023f6c7f869311ce8","subject":"Merge \"Revert \"Remove all usage of keystoneclient\"\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c127bbcbdff5575d717437d023f6c7f869311ce8"}]}],"author":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2021-04-05 07:43:34.000000000","tz":-240},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2021-08-25 15:08:50.000000000","tz":0},"subject":"Cinder Store: Update legacy images tests","message":"Cinder Store: Update legacy images tests\n\nupdate tests for cinder\u0027s new attachment API code in glance\ncinder store. This leaves a couple of mocks in place for the old\nway of doing things so that we can tolerate the old and proposed\nchanges to glance_store. We can remove those once we require the\nnewer glance_store version.\n\n(cherry picked from commit 01b5bf5ece9a6904c1c5df4ad557967c927b8683)\n(cherry picked from commit 0e2043d51ebe90acd0e6f284774c5411cb1c886b)\nChange-Id: Ifbdce0d0386b7c6868654711ae2f98125d9a1985\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e6a0403985902a2a2276aa9ba6a300e4523c43dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e6a0403985902a2a2276aa9ba6a300e4523c43dd"}]},"branch":"refs/heads/stable/wallaby","description":"Edit commit message"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
