)]}'
{"id":"openstack%2Fglance~805974","triplet_id":"openstack%2Fglance~master~I5afa2541bc28970191ea9fd76bf44eb9aaaa1081","project":"openstack/glance","branch":"master","topic":"805974","attention_set":{},"removed_from_attention_set":{"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2021-11-30 05:59:52.000000000","reason":"Change was submitted"},"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2021-11-30 05:59:52.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5afa2541bc28970191ea9fd76bf44eb9aaaa1081","subject":"Replace FakeObject with mock.MagicMock","status":"MERGED","created":"2021-08-25 09:05:54.000000000","updated":"2021-12-01 18:29:03.000000000","submitted":"2021-11-30 05:59:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"805974-805974","meta_rev_id":"f1b8c9f81cbef45f2bee91dbcb93b06e5a095093","_number":805974,"virtual_id_number":805974,"owner":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-11-30 05:59:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2021-11-30 05:19:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2021-11-29 15:39:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-11-30 01:14:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2021-11-30 05:19:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}],"reviewers":{"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]}],"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-25 09:40:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-25 14:48:12.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-11-24 08:01:51.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2021-11-24 08:01:51.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2021-11-29 15:39:13.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2021-11-30 01:14:03.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"}],"messages":[{"id":"c9bb9f158aff38b1bd2df0eb8d2ac1f1391e692d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2021-08-25 09:05:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fb3947743c1775902e83d9814ca42f26acb4be6c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-25 09:40:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e553922fa2045509085684d7ca6b15f : SUCCESS in 6m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e42212e03163433d9a3426c630c5d028 : SUCCESS in 8m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/132a58494c9e4a6882a09be66e1cb177 : SUCCESS in 6m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ba96794b8cd148ccad9759cec9037018 : SUCCESS in 8m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02e75fd50be8411fad5864c706192156 : SUCCESS in 8m 44s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/756731ef3a984830936e87e35f1a39f3 : SUCCESS in 9m 28s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/1b5c4200714c44d097b4fd5cdb79f213 : SUCCESS in 9m 52s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/a0fe35f3c49144efbfb6a15aef9c8e4e : SUCCESS in 11m 33s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/7a3bcf5fcb8448d89ea1417bbb8e4701 : SUCCESS in 7m 00s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/69f6d2e26d434a1293f1f2d42416bb35 : SUCCESS in 32m 17s","accounts_in_message":[],"_revision_number":1},{"id":"d54861df60a3f3c0f95b0416f74b085d6f8091f1","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-08-25 14:48:12.000000000","message":"Patch Set 1:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/805974/805974/1/check-not-vote/ext-glance-zuul/d645fdb : SUCCESS in 40m 15s","accounts_in_message":[],"_revision_number":1},{"id":"9a54902cbb11b7e6076da22d80b36a64774263b6","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-11-24 08:01:51.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c77bce07f791eef7d23c8458f98168b27830000d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2021-11-26 13:43:38.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5a29f3f7a7cc785fa628124558d62cbc12766610","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-26 14:33:13.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8ca473d427dc45878f2491d3d412f3ae : SUCCESS in 6m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d2b0ee937653408bb8fee6295ca3a64d : SUCCESS in 6m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0e22733db0704c2fbcd7074f5f6e148a : SUCCESS in 7m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a80526ef97664ef2a9ca7fc7cbe3dace : SUCCESS in 6m 25s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/764978a63133454a842997111c4c8325 : SUCCESS in 9m 34s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/23f8e72e625a476eade5b8d92e8a2c4f : SUCCESS in 9m 22s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/e3abbd4cd58f4d2ca1be3004f981aaba : SUCCESS in 9m 36s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/5ea7d01319064f1d97a425f2f1cde081 : SUCCESS in 6m 52s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/7a193bd8b2194af4b29884d0a8a8d982 : SUCCESS in 32m 14s","accounts_in_message":[],"_revision_number":2},{"id":"e6dff4ef3346e59b0394c1e13d29ee7e6713c002","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-11-29 15:39:13.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"fc3b43b6df3cc55e6a318b98a7e273dd4bfa19d1","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2021-11-30 01:14:03.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"49d0782b0b5aa972c5f5990f316103861e4e82b3","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-11-30 05:19:11.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"b1eb2ae3748edb1b919c1eb5b03ddab9c50c1d53","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-30 05:19:44.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"14348e5c9f5479c0b0836fba36725b377c045873","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-30 05:59:51.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3b3bb33c84df4503855232e093504430 : SUCCESS in 4m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/99f4433efc864b85b8cb8f9f595f975f : SUCCESS in 5m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/afa2fb9edc674e6692f0d23318a21a44 : SUCCESS in 5m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5de1c37677894b67bb0788776948e485 : SUCCESS in 6m 12s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1eaaaf0c3d13424d982dddff9946df08 : SUCCESS in 8m 13s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/0f35acddb8604bf7acad46f02ed8041e : SUCCESS in 8m 39s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/92642a63cca449b7a6a5177e664b2973 : SUCCESS in 34m 30s","accounts_in_message":[],"_revision_number":2},{"id":"923c98837a8ecbe85d9c44923f43a9c8c7a86100","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-30 05:59:52.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"b912d392565076bac036d850594c9d4a1bb5a724","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-30 06:01:28.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e793a00f72de40799f2f9f8ca389aba2 : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":2},{"id":"f1b8c9f81cbef45f2bee91dbcb93b06e5a095093","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-12-01 18:29:03.000000000","message":"Patch Set 2:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/805974/805974/2/check-not-vote/ext-glance-zuul/e05c5f7 : FAILURE in 2m 01s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"de8c8502694f1156e2a4240efb44d5c19e04a486","revisions":{"ad9a68a8a3455b7efabdfffa65b920d240947665":{"kind":"REWORK","_number":1,"created":"2021-08-25 09:05:54.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/74/805974/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/74/805974/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/74/805974/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/74/805974/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/74/805974/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/74/805974/1"}}},"commit":{"parents":[{"commit":"6c87a18d4cb43aa4af6e953af19f1655f7bb3bf9","subject":"Check policies for image import operation in API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6c87a18d4cb43aa4af6e953af19f1655f7bb3bf9"}]}],"author":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2021-08-25 08:49:36.000000000","tz":-240},"committer":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2021-08-25 09:03:38.000000000","tz":-240},"subject":"Replace FakeObject with mock.MagicMock","message":"Replace FakeObject with mock.MagicMock\n\nIn glance cinder functional tests, we currently use FakeObject class\nto create mock objects which has limited functionality.\nReplacing it with mock.MagicMock will prevent failures like in change\nI129181b2338e09d094ee552a62500c6e460ee621 and other features as well.\n\nChange-Id: I5afa2541bc28970191ea9fd76bf44eb9aaaa1081\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ad9a68a8a3455b7efabdfffa65b920d240947665"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ad9a68a8a3455b7efabdfffa65b920d240947665"}]},"branch":"refs/heads/master"},"de8c8502694f1156e2a4240efb44d5c19e04a486":{"kind":"REWORK","_number":2,"created":"2021-11-26 13:43:38.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/74/805974/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/74/805974/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/74/805974/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/74/805974/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/74/805974/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/74/805974/2"}}},"commit":{"parents":[{"commit":"6fd949a966fea443bb4b0b9351b3fa04662a7ecb","subject":"Refactor glance cinder tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6fd949a966fea443bb4b0b9351b3fa04662a7ecb"}]}],"author":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2021-08-25 08:49:36.000000000","tz":-240},"committer":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2021-11-26 13:43:06.000000000","tz":-300},"subject":"Replace FakeObject with mock.MagicMock","message":"Replace FakeObject with mock.MagicMock\n\nIn glance cinder functional tests, we currently use FakeObject class\nto create mock objects which has limited functionality.\nReplacing it with mock.MagicMock will prevent failures like in change\nI129181b2338e09d094ee552a62500c6e460ee621 and other features as well.\n\nChange-Id: I5afa2541bc28970191ea9fd76bf44eb9aaaa1081\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/de8c8502694f1156e2a4240efb44d5c19e04a486"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/de8c8502694f1156e2a4240efb44d5c19e04a486"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
