)]}'
{"id":"openstack%2Fglance~806134","triplet_id":"openstack%2Fglance~master~Icb373fd09ebae32edf6b60e6536562ccb9a1342d","project":"openstack/glance","branch":"master","topic":"bug/1924612","hashtags":[],"change_id":"Icb373fd09ebae32edf6b60e6536562ccb9a1342d","subject":"Make \"killed\" images list available using the CLI","status":"ABANDONED","created":"2021-08-26 10:28:02.000000000","updated":"2021-09-14 01:42:02.000000000","total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"d521aa8689329e5e3b63f853816f7072567f188a","_number":806134,"virtual_id_number":806134,"owner":{"_account_id":33786,"name":"HanGuangyu","email":"hanguangyu2@gmail.com","username":"hanguangyu"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-09-07 10:19:27.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"all":[{"value":-1,"date":"2021-09-08 17:23:27.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":19332,"name":"Daniel Smith","email":"daniel.smith@ericsson.com","username":"lmcdasm"}],"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-26 12:07:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-27 07:45:36.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-09-08 06:54:10.000000000","updated_by":{"_account_id":33786,"name":"HanGuangyu","email":"hanguangyu2@gmail.com","username":"hanguangyu"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"CC"},{"updated":"2021-09-08 06:54:10.000000000","updated_by":{"_account_id":33786,"name":"HanGuangyu","email":"hanguangyu2@gmail.com","username":"hanguangyu"},"reviewer":{"_account_id":19332,"name":"Daniel Smith","email":"daniel.smith@ericsson.com","username":"lmcdasm"},"state":"CC"},{"updated":"2021-09-08 17:23:27.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"}],"messages":[{"id":"8391e0d51bac5c98e4dfdfa1306ca741d1d6a204","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33786,"name":"HanGuangyu","email":"hanguangyu2@gmail.com","username":"hanguangyu"},"date":"2021-08-26 10:28:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9c2b85e17aed8f0be175363cb0fe76be5a52d7ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-26 12:07:53.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/90ac5b077c4a4de382161124f950f9c5 : SUCCESS in 1h 04m 44s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e8ada1e777104a329ed26b8da890c6e5 : SUCCESS in 1h 02m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a1cd69d54084875986aea76844a4fc1 : SUCCESS in 7m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/603a0d9cea8149699ce7dba1a8d06ef1 : SUCCESS in 6m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/00713dcf8b464f77b7b1b21adf0a3abf : SUCCESS in 6m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f923c624e44c4296bb4b15130a2b1eba : SUCCESS in 9m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e922ff55c4c422b9dae8c55d5aac062 : SUCCESS in 7m 05s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0404b0e37a8849e4b19a3ac884fa37c0 : SUCCESS in 9m 55s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/f240d027bc884ecd948658ed4d768428 : SUCCESS in 9m 44s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/7eca2e4dc8e44d1d9c3e5989144e59aa : SUCCESS in 9m 19s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/7bacf8557c6740d588e50ed1db242f7d : SUCCESS in 7m 00s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/3007f12c0c7d49919ac6d193f846be92 : SUCCESS in 1h 29m 35s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/e27278888ed74f2b84b2855f6adcafb2 : SUCCESS in 1h 10m 38s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/28ac149cf87b4c388550a9a95a9905cd : FAILURE in 1h 09m 56s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/6965bf47107140068e88c38de82acad4 : FAILURE in 59m 14s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/908bd07d59914d0d82867e7fa03040f3 : SUCCESS in 44m 35s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/6bcaffa4376b4d5d882463ee10f4ac7f : SUCCESS in 1h 15m 59s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/2227ba9b751d4908bbdac4e663626957 : SUCCESS in 31m 21s","accounts_in_message":[],"_revision_number":1},{"id":"528c3fd719ee13632394d70e7a61ff956b50a5c1","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-08-27 07:45:36.000000000","message":"Patch Set 1:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/806134/806134/1/check-not-vote/ext-glance-zuul/f522675 : SUCCESS in 40m 17s","accounts_in_message":[],"_revision_number":1},{"id":"165f62146fea8afa6afe2e7050cf2f5baa91476e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33786,"name":"HanGuangyu","email":"hanguangyu2@gmail.com","username":"hanguangyu"},"date":"2021-09-02 08:58:22.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b20c44c97509e8bc9ad5b3a23d622ab30f30cf2a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33786,"name":"HanGuangyu","email":"hanguangyu2@gmail.com","username":"hanguangyu"},"date":"2021-09-02 09:03:17.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"454f3812fb9653a5df25ed3744bba852072d5df9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-02 10:40:08.000000000","message":"Patch Set 3: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/6e64d22366bb4204afc3dc6af18607bd : SUCCESS in 58m 44s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/31cb2d2abc55423c9907122c833eb173 : SUCCESS in 58m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/951b5beb596f4c56acb231c2c13ba565 : FAILURE in 6m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/18bbb753d80042eaaa80fb3a8b66d1f3 : SUCCESS in 6m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5f69c9d36484468aba0d24d2eed9c91b : SUCCESS in 7m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5a028ff5e6024b71b47a03a9a2d1315d : SUCCESS in 8m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/982e9e2dabb0435cb656971c4b7d0570 : SUCCESS in 11m 55s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/00f33d7287184fe6ad42a762502e72b3 : SUCCESS in 11m 17s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/458951c411584450a838e25fdee0141f : SUCCESS in 8m 35s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/3406e0b92ff645688cceb8d4d9f0b8f5 : SUCCESS in 12m 21s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/3a1ee5843c8b42ac997bfd7f201cbcee : SUCCESS in 6m 31s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/ff0090a657fb4cd19391c7eca1a681d3 : SUCCESS in 1h 16m 19s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/ab071912bbb8417a84cc6006f14985a2 : SUCCESS in 1h 06m 24s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/48275850bd544ff7b5333d2084129520 : SUCCESS in 1h 01m 01s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/db34b48f7b8c4b738ad431f70531ad15 : SUCCESS in 1h 18m 41s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/372b53811d424d80aa26f0a954c3a072 : SUCCESS in 1h 10m 53s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/cf7f12c2f75f4f9cbd70b6b720954140 : SUCCESS in 1h 24m 52s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/fad79c478a6a4210bf47c6d05ed11cf7 : SUCCESS in 34m 30s","accounts_in_message":[],"_revision_number":3},{"id":"7cb6153f555d1cef48f69babc0bdab61315579b4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33786,"name":"HanGuangyu","email":"hanguangyu2@gmail.com","username":"hanguangyu"},"date":"2021-09-03 01:48:31.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"4e1cff689f824a92a4c6a5d7adab696af2d87cef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-03 04:03:32.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/760c51144d3844c88e72b21300010c50 : SUCCESS in 1h 00m 01s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b26a9372bf87415fb987310f34480ce9 : SUCCESS in 1h 06m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cff7f1f9e07d479887c6aaa89d1fbf7e : SUCCESS in 5m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8b3c2341a4974782b5783108493ed70b : SUCCESS in 6m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/85064376a16f4baabb28b1ae48ed157c : SUCCESS in 6m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/04c1729dbbb94ab2bff52553ff5dcd8a : SUCCESS in 6m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e9516d0401c4730b63fcb601771a939 : SUCCESS in 6m 18s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/79d4c02c940146cb9496a57269d2237b : SUCCESS in 8m 12s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/fbb6c22ff40742188c7e599b3851e29a : SUCCESS in 9m 56s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/b0476734273e4b84b76090ccfbf95959 : SUCCESS in 8m 59s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/aba7d4bc391943379571e9a3a094b4e8 : SUCCESS in 5m 12s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/8a147ec76bde4dd9bd304eade047bf5e : SUCCESS in 1h 04m 37s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/b66623e6204d4dd29d0f3b3e249705dc : FAILURE in 2h 03m 37s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/a70072bb22e64fefbee438cc15e03fd0 : SUCCESS in 1h 07m 27s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/205ab68514714c8c9234231e51a73574 : SUCCESS in 1h 23m 18s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/aa539197b01442d687bcfe1b1402e9c1 : SUCCESS in 1h 24m 20s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/06754be3c9a14bc18890ae4cd059ee7c : SUCCESS in 1h 15m 59s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/f5907e9cb6134daebc66a3fcd3278f05 : SUCCESS in 32m 43s","accounts_in_message":[],"_revision_number":4},{"id":"2cd37243aeefc8f664936b34937ffd14b521122d","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-09-06 05:51:32.000000000","message":"Patch Set 4:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/806134/806134/4/check-not-vote/ext-glance-zuul/2ffbbd6 : SUCCESS in 40m 18s","accounts_in_message":[],"_revision_number":4},{"id":"c9f6adb4f4e32790fb8828da2a8563f81d99456f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33786,"name":"HanGuangyu","email":"hanguangyu2@gmail.com","username":"hanguangyu"},"date":"2021-09-07 08:30:52.000000000","message":"Uploaded patch set 5: New patch set was added with same tree, parent, and commit message as Patch Set 4.","accounts_in_message":[],"_revision_number":5},{"id":"ae55f2681c2daacae014c6a3831328ca20e1f797","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-07 10:19:27.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/beb13eda0b1148c58968aef441780f1a : SUCCESS in 1h 03m 40s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/76ba5031946341c2962bfbfe3a53974e : SUCCESS in 1h 09m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6a53bac20af2412d82d0b1d67ae5faaa : SUCCESS in 8m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ac1fbda82dac40dea93ec079882ac0fc : SUCCESS in 6m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eae12a11d6d1406e98bded3d43547a3e : SUCCESS in 6m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9319d83c4f0642d28ee295ae7d8ccbc4 : SUCCESS in 8m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/05147439a2d449c9a4a748ef88eb6bb0 : SUCCESS in 7m 17s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8a8583f0dd46411382e5bf1df165c613 : SUCCESS in 11m 53s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/483d800b1dbd4d678accf28bf8ecb809 : SUCCESS in 11m 47s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/5339550341c14f4c83fcf7f52d60b609 : SUCCESS in 10m 34s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/ee410d972f904bbcbcdf0edc92f3c2b2 : SUCCESS in 5m 21s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/7e2436bfbb56490fbf3a8c4764bf64f1 : SUCCESS in 1h 27m 56s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/3ede69b8fafe42db9263606693e9fa00 : SUCCESS in 1h 14m 22s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/6ec38494ab664a13924108d5911a59ba : SUCCESS in 1h 08m 04s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/7594fe96d92b4cc18b14259346f6f732 : SUCCESS in 1h 12m 24s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/17ca4157665f4cefb87020fbb8809b98 : SUCCESS in 53m 55s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/fa6afbe8f508427596fc7f2c2ef352cb : SUCCESS in 1h 21m 50s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/f3098b932fbe49fb8b6b85871575cdf0 : SUCCESS in 34m 04s","accounts_in_message":[],"_revision_number":5},{"id":"ddbf523c5bf9f132ceb9682d504723672c124584","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-09-08 17:23:27.000000000","message":"Patch Set 5: Code-Review-1\n\n(1 comment)\n\nI think killed status is just a thing in v1, which is long since gone right? Not sure it makes sense to bring killed forward to v2 just because of the potential to clean up old images.\n\nMaybe we need a glance-manage command to clean old killed images from the DB and a reno to tell operators to do that thing at some point?","accounts_in_message":[],"_revision_number":5},{"id":"773e097e251a3019bf8b15d7d483a1655fcac3ac","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-09-10 04:42:20.000000000","message":"Patch Set 5:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/806134/806134/5/check-not-vote/ext-glance-zuul/e73a193 : SUCCESS in 38m 09s","accounts_in_message":[],"_revision_number":5},{"id":"158065682a85e8a0ee25aae48d4d540e22b48bee","author":{"_account_id":33786,"name":"HanGuangyu","email":"hanguangyu2@gmail.com","username":"hanguangyu"},"date":"2021-09-10 08:38:09.000000000","message":"Patch Set 5:\n\n\u003e Patch Set 5: Code-Review-1\n\u003e \n\u003e (1 comment)\n\u003e \n\u003e I think killed status is just a thing in v1, which is long since gone right? Not sure it makes sense to bring killed forward to v2 just because of the potential to clean up old images.\n\nIt is true that the killed status occurs infrequently now. But, the question that image in killed status coun\u0027t be list was still being reported in the ussuri in april this year. \n\nIn fact, some people still experience the killed status. The \"list\" command is set to list \"all of image\" and this patch does not affect other logic. So, mabye this patch could be recognizeded by you？\n\n\n\u003e Maybe we need a glance-manage command to clean old killed images from the DB and a reno to tell operators to do that thing at some point?\n\nI\u0027m agree with you. If it could be make sure, should we can report a bug or blueprint? I can be responsible for making it happen.","accounts_in_message":[],"_revision_number":5},{"id":"94a1c73e115fb541da69f4d9182f66528cb3c008","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-09-10 13:32:45.000000000","message":"Patch Set 5:\n\n\u003e It is true that the killed status occurs infrequently now. But, the question that image in killed status coun\u0027t be list was still being reported in the ussuri in april this year. \n\nThis should be the only way to get into killed state anymore, and this patch will fix it so it doesn\u0027t happen anymore:\n\nhttps://review.opendev.org/c/openstack/glance/+/807932\n\n\u003e In fact, some people still experience the killed status. The \"list\" command is set to list \"all of image\" and this patch does not affect other logic. So, mabye this patch could be recognizeded by you？\n\nAs documented here, killed state should only be a thing in v1:\n\nhttps://docs.openstack.org/glance/wallaby/user/statuses.html\n\nIf we show images in that state in v2, we need to change that graph and probably handle it in a lot of other places. The correct thing to do is to eliminate the remaining accidental transition (so images can never get into killed state again), and provide a way to clean up those images that are already in the database.\n\n\u003e \u003e Maybe we need a glance-manage command to clean old killed images from the DB and a reno to tell operators to do that thing at some point?\n\u003e \n\u003e I\u0027m agree with you. If it could be make sure, should we can report a bug or blueprint? I can be responsible for making it happen.\n\nI think that a bug would be suitable for this, but yes, please file one at least. Abhi could decide that a blueprint is *also* needed, but I would expect just a but would be fine.\n\nThanks!","accounts_in_message":[],"_revision_number":5},{"id":"85b2d84d4d2a9bf031b3c19f1c302a3037266925","author":{"_account_id":33786,"name":"HanGuangyu","email":"hanguangyu2@gmail.com","username":"hanguangyu"},"date":"2021-09-13 01:41:42.000000000","message":"Patch Set 5:\n\nOh, Thanks you very much for your comments. I get my error. Yes, the good way is don\u0027t make a image into killed state. \n\nI will report a bug about the glance-manage command which aims to clean old killed images. And I will hard to achive it.\n\nThanks you again, my friend.","accounts_in_message":[],"_revision_number":5},{"id":"bdf1c532a8255d67c36cd848c660b681025eca95","author":{"_account_id":33786,"name":"HanGuangyu","email":"hanguangyu2@gmail.com","username":"hanguangyu"},"date":"2021-09-13 06:22:15.000000000","message":"Patch Set 5:\n\nHi, I have reported a bug about the glance-manage command to clean old kiled images. Looking forword to your review.","accounts_in_message":[],"_revision_number":5},{"id":"17fc2a4b2b7f5e6f7491d086519c2cfafc03f7c3","author":{"_account_id":33786,"name":"HanGuangyu","email":"hanguangyu2@gmail.com","username":"hanguangyu"},"date":"2021-09-13 06:22:33.000000000","message":"Patch Set 5:\n\n\u003e Patch Set 5:\n\u003e \n\u003e Hi, I have reported a bug about the glance-manage command to clean old kiled images. Looking forword to your review.\n\nhttps://bugs.launchpad.net/glance/+bug/1943418","accounts_in_message":[],"_revision_number":5},{"id":"95607ed5de878f078062da6ea539ccce913bea2b","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-09-13 06:26:34.000000000","message":"Patch Set 5:\n\n\u003e Patch Set 5:\n\u003e \n\u003e \u003e It is true that the killed status occurs infrequently now. But, the question that image in killed status coun\u0027t be list was still being reported in the ussuri in april this year. \n\u003e \n\u003e This should be the only way to get into killed state anymore, and this patch will fix it so it doesn\u0027t happen anymore:\n\u003e \n\u003e https://review.opendev.org/c/openstack/glance/+/807932\n\u003e \n\u003e \u003e In fact, some people still experience the killed status. The \"list\" command is set to list \"all of image\" and this patch does not affect other logic. So, mabye this patch could be recognizeded by you？\n\u003e \n\u003e As documented here, killed state should only be a thing in v1:\n\u003e \n\u003e https://docs.openstack.org/glance/wallaby/user/statuses.html\n\u003e \n\u003e If we show images in that state in v2, we need to change that graph and probably handle it in a lot of other places. The correct thing to do is to eliminate the remaining accidental transition (so images can never get into killed state again), and provide a way to clean up those images that are already in the database.\n\u003e \n\u003e \u003e \u003e Maybe we need a glance-manage command to clean old killed images from the DB and a reno to tell operators to do that thing at some point?\n\u003e \u003e \n\u003e \u003e I\u0027m agree with you. If it could be make sure, should we can report a bug or blueprint? I can be responsible for making it happen.\n\u003e \n\u003e I think that a bug would be suitable for this, but yes, please file one at least. Abhi could decide that a blueprint is *also* needed, but I would expect just a but would be fine.\n\u003e \n\u003e Thanks!\n\nBug will be fine.","accounts_in_message":[],"_revision_number":5},{"id":"60123609515c9403bbc88c97dcaa006ddc0a99d1","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-09-13 06:27:37.000000000","message":"Patch Set 5:\n\n\u003e Patch Set 5:\n\u003e \n\u003e \u003e Patch Set 5:\n\u003e \u003e \n\u003e \u003e Hi, I have reported a bug about the glance-manage command to clean old kiled images. Looking forword to your review.\n\u003e \n\u003e https://bugs.launchpad.net/glance/+bug/1943418\n\nThank you for filing the bug, I think you can abandon this patch now and submit a new patch to add this support in glance-manage mentioning this bug as a reference.","accounts_in_message":[],"_revision_number":5},{"id":"eb4c75f8aa82dc5c3cded2cb21a1d81ab24ca021","author":{"_account_id":33786,"name":"HanGuangyu","email":"hanguangyu2@gmail.com","username":"hanguangyu"},"date":"2021-09-14 01:37:08.000000000","message":"Patch Set 5:\n\n\u003e Patch Set 5:\n\u003e \n\u003e \u003e Patch Set 5:\n\u003e \u003e \n\u003e \u003e \u003e Patch Set 5:\n\u003e \u003e \u003e \n\u003e \u003e \u003e Hi, I have reported a bug about the glance-manage command to clean old kiled images. Looking forword to your review.\n\u003e \u003e \n\u003e \u003e https://bugs.launchpad.net/glance/+bug/1943418\n\u003e \n\u003e Thank you for filing the bug, I think you can abandon this patch now and submit a new patch to add this support in glance-manage mentioning this bug as a reference.\n\nGood, I will submit a new patch.","accounts_in_message":[],"_revision_number":5},{"id":"d521aa8689329e5e3b63f853816f7072567f188a","tag":"autogenerated:gerrit:abandon","author":{"_account_id":33786,"name":"HanGuangyu","email":"hanguangyu2@gmail.com","username":"hanguangyu"},"date":"2021-09-14 01:42:02.000000000","message":"Abandoned\n\nThe only way to get into killed state has been fixed(https://review.opendev.org/c/openstack/glance/+/807932).\nAnd the correct way is to providing a way to clean up those old killed images. I have reported a bug about this glace-manage command(https://bugs.launchpad.net/glance/+bug/1943418). I will realize it.","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"e23250763b4daa5e018f6e6fcad1271a589ed4cd","revisions":{"4546af2027de465e936672d8f2b334abb619f455":{"kind":"REWORK","_number":1,"created":"2021-08-26 10:28:02.000000000","uploader":{"_account_id":33786,"name":"HanGuangyu","email":"hanguangyu2@gmail.com","username":"hanguangyu"},"ref":"refs/changes/34/806134/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/34/806134/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/34/806134/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/34/806134/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/34/806134/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/34/806134/1"}}},"commit":{"parents":[{"commit":"869a39d9e55849c7de7b3ee086cea1017d4f60d5","subject":"Check policies for Image Tags in API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/869a39d9e55849c7de7b3ee086cea1017d4f60d5"}]}],"author":{"name":"Han Guangyu","email":"hanguangyu@uniontech.com","date":"2021-08-26 09:47:10.000000000","tz":480},"committer":{"name":"Han Guangyu","email":"hanguangyu@uniontech.com","date":"2021-08-26 10:27:58.000000000","tz":480},"subject":"Make \"killed\" images list available using the CLI","message":"Make \"killed\" images list available using the CLI\n\nImages in \"killed\" status weren\u0027t able to list by using\nCLI even if using \"--property status\u003dkilled\". Fisrt, It is\nnot reasonable that the argument \"status\u003dkilled\" could be\naccepted but not effective. Besides, when the image_id is\nknown the details can be shown and they can be deleted. It\nalso is a contradiction. But if an user can\u0027t list \"killed\"\nimages, he doesn\u0027t know the image_id.\n\nThe filtering for the \"killed\" status was cancelled.\n\nUsers can list Images in \"killed\" status using CLI——\n\"openstack image list\".\n\nCloses-Bug: #1924612\nChange-Id: Icb373fd09ebae32edf6b60e6536562ccb9a1342d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4546af2027de465e936672d8f2b334abb619f455"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4546af2027de465e936672d8f2b334abb619f455"}]},"branch":"refs/heads/master"},"99d67c18dffcd7cc0991967ef927b2abbad14fe2":{"kind":"REWORK","_number":2,"created":"2021-09-02 08:58:22.000000000","uploader":{"_account_id":33786,"name":"HanGuangyu","email":"hanguangyu2@gmail.com","username":"hanguangyu"},"ref":"refs/changes/34/806134/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/34/806134/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/34/806134/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/34/806134/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/34/806134/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/34/806134/2"}}},"commit":{"parents":[{"commit":"869a39d9e55849c7de7b3ee086cea1017d4f60d5","subject":"Check policies for Image Tags in API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/869a39d9e55849c7de7b3ee086cea1017d4f60d5"}]}],"author":{"name":"Han Guangyu","email":"hanguangyu@uniontech.com","date":"2021-08-26 09:47:10.000000000","tz":480},"committer":{"name":"Han Guangyu","email":"hanguangyu@uniontech.com","date":"2021-09-02 08:57:59.000000000","tz":480},"subject":"Make \"killed\" images list available using the CLI","message":"Make \"killed\" images list available using the CLI\n\nImages in \"killed\" status weren\u0027t able to list by using\nCLI even if using \"--property status\u003dkilled\". Fisrt, It is\nnot reasonable that the argument \"status\u003dkilled\" could be\naccepted but not effective. Besides, when the image_id is\nknown the details can be shown and they can be deleted. It\nalso is a contradiction. But if an user can\u0027t list \"killed\"\nimages, he doesn\u0027t know the image_id.\n\nThe filtering for the \"killed\" status was cancelled.\n\nUsers can list Images in \"killed\" status using CLI——\n\"openstack image list\".\n\nCloses-Bug: #1924612\nChange-Id: Icb373fd09ebae32edf6b60e6536562ccb9a1342d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/99d67c18dffcd7cc0991967ef927b2abbad14fe2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/99d67c18dffcd7cc0991967ef927b2abbad14fe2"}]},"branch":"refs/heads/master"},"265446c0c767acd9cc9c25ae7201c2e5e6850ae8":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2021-09-02 09:03:17.000000000","uploader":{"_account_id":33786,"name":"HanGuangyu","email":"hanguangyu2@gmail.com","username":"hanguangyu"},"ref":"refs/changes/34/806134/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/34/806134/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/34/806134/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/34/806134/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/34/806134/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/34/806134/3"}}},"commit":{"parents":[{"commit":"869a39d9e55849c7de7b3ee086cea1017d4f60d5","subject":"Check policies for Image Tags in API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/869a39d9e55849c7de7b3ee086cea1017d4f60d5"}]}],"author":{"name":"Han Guangyu","email":"hanguangyu@uniontech.com","date":"2021-08-26 09:47:10.000000000","tz":480},"committer":{"name":"Han Guangyu","email":"hanguangyu@uniontech.com","date":"2021-09-02 09:02:39.000000000","tz":480},"subject":"Make \"killed\" images list available using the CLI","message":"Make \"killed\" images list available using the CLI\n\nImages in \"killed\" status weren\u0027t able to list by using\nCLI even if using \"--property status\u003dkilled\". Fisrt, It is\nnot reasonable that the argument \"status\u003dkilled\" could be\naccepted but not effective. Besides, when the image_id is\nknown the details can be shown and they can be deleted. It\nalso is a contradiction. If an user can\u0027t list \"killed\"\nimages, he doesn\u0027t know the image_id.\n\nThe filtering for the \"killed\" status was cancelled.\n\nUsers can list Images in \"killed\" status using CLI——\n\"openstack image list\".\n\nCloses-Bug: #1924612\nChange-Id: Icb373fd09ebae32edf6b60e6536562ccb9a1342d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/265446c0c767acd9cc9c25ae7201c2e5e6850ae8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/265446c0c767acd9cc9c25ae7201c2e5e6850ae8"}]},"branch":"refs/heads/master"},"daf8ca688a6fce8c0ec999624b24ea5489331a38":{"kind":"REWORK","_number":4,"created":"2021-09-03 01:48:31.000000000","uploader":{"_account_id":33786,"name":"HanGuangyu","email":"hanguangyu2@gmail.com","username":"hanguangyu"},"ref":"refs/changes/34/806134/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/34/806134/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/34/806134/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/34/806134/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/34/806134/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/34/806134/4"}}},"commit":{"parents":[{"commit":"869a39d9e55849c7de7b3ee086cea1017d4f60d5","subject":"Check policies for Image Tags in API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/869a39d9e55849c7de7b3ee086cea1017d4f60d5"}]}],"author":{"name":"Han Guangyu","email":"hanguangyu@uniontech.com","date":"2021-08-26 09:47:10.000000000","tz":480},"committer":{"name":"Han Guangyu","email":"hanguangyu@uniontech.com","date":"2021-09-03 01:47:56.000000000","tz":480},"subject":"Make \"killed\" images list available using the CLI","message":"Make \"killed\" images list available using the CLI\n\nImages in \"killed\" status weren\u0027t able to list by using\nCLI even if using \"--property status\u003dkilled\". Fisrt, It is\nnot reasonable that the argument \"status\u003dkilled\" could be\naccepted but not effective. Besides, when the image_id is\nknown the details can be shown and they can be deleted. It\nalso is a contradiction. If an user can\u0027t list \"killed\"\nimages, he doesn\u0027t know the image_id.\n\nThe filtering for the \"killed\" status was cancelled.\n\nUsers can list Images in \"killed\" status using CLI——\n\"openstack image list\".\n\nCloses-Bug: #1924612\nChange-Id: Icb373fd09ebae32edf6b60e6536562ccb9a1342d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/daf8ca688a6fce8c0ec999624b24ea5489331a38"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/daf8ca688a6fce8c0ec999624b24ea5489331a38"}]},"branch":"refs/heads/master"},"e23250763b4daa5e018f6e6fcad1271a589ed4cd":{"kind":"NO_CHANGE","_number":5,"created":"2021-09-07 08:30:52.000000000","uploader":{"_account_id":33786,"name":"HanGuangyu","email":"hanguangyu2@gmail.com","username":"hanguangyu"},"ref":"refs/changes/34/806134/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/34/806134/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/34/806134/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/34/806134/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/34/806134/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/34/806134/5"}}},"commit":{"parents":[{"commit":"869a39d9e55849c7de7b3ee086cea1017d4f60d5","subject":"Check policies for Image Tags in API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/869a39d9e55849c7de7b3ee086cea1017d4f60d5"}]}],"author":{"name":"Han Guangyu","email":"hanguangyu@uniontech.com","date":"2021-08-26 09:47:10.000000000","tz":480},"committer":{"name":"Han Guangyu","email":"hanguangyu@uniontech.com","date":"2021-09-07 08:30:40.000000000","tz":480},"subject":"Make \"killed\" images list available using the CLI","message":"Make \"killed\" images list available using the CLI\n\nImages in \"killed\" status weren\u0027t able to list by using\nCLI even if using \"--property status\u003dkilled\". Fisrt, It is\nnot reasonable that the argument \"status\u003dkilled\" could be\naccepted but not effective. Besides, when the image_id is\nknown the details can be shown and they can be deleted. It\nalso is a contradiction. If an user can\u0027t list \"killed\"\nimages, he doesn\u0027t know the image_id.\n\nThe filtering for the \"killed\" status was cancelled.\n\nUsers can list Images in \"killed\" status using CLI——\n\"openstack image list\".\n\nCloses-Bug: #1924612\nChange-Id: Icb373fd09ebae32edf6b60e6536562ccb9a1342d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e23250763b4daa5e018f6e6fcad1271a589ed4cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e23250763b4daa5e018f6e6fcad1271a589ed4cd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
