)]}'
{"id":"openstack%2Fglance~806774","triplet_id":"openstack%2Fglance~master~Idb2ab1f4ac6a2c3e09cba4ca966b7faff30d34b1","project":"openstack/glance","branch":"master","topic":"dnm/ceph-pacific","hashtags":[],"change_id":"Idb2ab1f4ac6a2c3e09cba4ca966b7faff30d34b1","subject":"DNM: Run glance ceph jobs with \u0027pacific\u0027 version","status":"ABANDONED","created":"2021-08-31 16:19:36.000000000","updated":"2021-09-16 06:01:44.000000000","total_comment_count":3,"unresolved_comment_count":2,"has_review_started":true,"meta_rev_id":"497ae47acfbb52d53de96e7205db58740c269ee4","_number":806774,"virtual_id_number":806774,"owner":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2021-08-31 19:56:37.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]}],"REVIEWER":[{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-31 16:20:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-31 16:20:48.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-08-31 16:27:01.000000000","updated_by":{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},"reviewer":{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},"state":"CC"},{"updated":"2021-08-31 16:27:01.000000000","updated_by":{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},"reviewer":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"state":"REVIEWER"},{"updated":"2021-09-02 12:41:50.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"923d72236e0ce39d9afe93d2d8724ef5bec8cc5e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-31 16:19:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"efdafb172e4818c476f2b0850d7939a9d5f04aa0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-31 16:20:10.000000000","message":"Patch Set 1: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/glance on branch master.  The error was:\n\n  while parsing a block mapping\n  in \"openstack/glance/.zuul.yaml@master\", line 283, column 3\ndid not find expected key\n  in \"openstack/glance/.zuul.yaml@master\", line 288, column 4","accounts_in_message":[],"_revision_number":1},{"id":"fb21ba5e4fc5770c39b9c912636e83b06f3645a1","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2021-08-31 16:20:48.000000000","message":"Patch Set 1:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/glance on branch master.  The error was:\n\n  while parsing a block mapping\n  in \"openstack/glance/.zuul.yaml@master\", line 283, column 3\ndid not find expected key\n  in \"openstack/glance/.zuul.yaml@master\", line 288, column 4","accounts_in_message":[],"_revision_number":1},{"id":"224f02967f1032a921e5097ea619efd8d693f7b3","author":{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},"date":"2021-08-31 16:27:01.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a7c555d442202a060b837fda6b44e8841df7f9ab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-31 16:27:24.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7c712601f264c1a43d4287f94386acc4955b606b","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-31 16:33:36.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0c9326e2e9266d0cc56bcc710b9ce2119e09ca70","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-31 18:42:10.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"b77f8733bf6e84880941323686523c61fc15ec70","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-31 19:01:33.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"823289a3da4832be25ab37f79dafc8f044cb4009","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-31 19:56:37.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5aaf5dc4acaf45df98b5f664d6132085 : SUCCESS in 5m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4b984e4e7ab148ba88e00f3db6eb14dc : SUCCESS in 7m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/087b707ffe7c4ccc93fae4c9b15edcde : SUCCESS in 7m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d949c7d846f44ba998367ce927901002 : SUCCESS in 9m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8dcbe310cff41178a25b835871d1420 : SUCCESS in 7m 03s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/fe507d12d3094ddeabbc7b5e7c2e920d : SUCCESS in 11m 34s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/2fa962fd19d84ca79fd9c7b6527d6942 : SUCCESS in 9m 11s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/473071957f834319b30cef7a0a62f002 : SUCCESS in 9m 23s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/c2a72ee7f7ea442a869c29b33c8137ed : SUCCESS in 1h 07m 29s (non-voting)\n- glance-nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/0cff34343f3c4ff0b43492fea40ff1ba : SUCCESS in 1h 11m 33s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/dfc5338d272b4ddc9ced43e01c0e79cf : SUCCESS in 32m 00s","accounts_in_message":[],"_revision_number":3},{"id":"b82038c93f50b687059d8df4733ab815c80f7e0f","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-09-02 12:41:50.000000000","message":"Patch Set 3:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/806774/806774/3/check-not-vote/ext-glance-zuul/4866d5b : SUCCESS in 39m 43s","accounts_in_message":[],"_revision_number":3},{"id":"497ae47acfbb52d53de96e7205db58740c269ee4","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-09-16 06:01:44.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"8a8ac8d7a34c02de40e833b5155c823148a4dc22","revisions":{"e330f817013c998182b3adf96332676bb2ca7151":{"kind":"REWORK","_number":1,"created":"2021-08-31 16:19:36.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/74/806774/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/74/806774/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/74/806774/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/74/806774/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/74/806774/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/74/806774/1"}}},"commit":{"parents":[{"commit":"25e60aea8e2fe11937e4b4d94e89c35cbb43940e","subject":"Merge \"Check policies for image tasks information in API\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/25e60aea8e2fe11937e4b4d94e89c35cbb43940e"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-08-31 16:12:49.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-08-31 16:14:02.000000000","tz":0},"subject":"DNM: Run glance ceph jobs with \u0027pacific\u0027 version","message":"DNM: Run glance ceph jobs with \u0027pacific\u0027 version\n\nChange-Id: Idb2ab1f4ac6a2c3e09cba4ca966b7faff30d34b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e330f817013c998182b3adf96332676bb2ca7151"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e330f817013c998182b3adf96332676bb2ca7151"}]},"branch":"refs/heads/master"},"1041334bfad16b236505ac61ba042c9fb830cb7e":{"kind":"REWORK","_number":2,"created":"2021-08-31 16:27:24.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/74/806774/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/74/806774/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/74/806774/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/74/806774/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/74/806774/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/74/806774/2"}}},"commit":{"parents":[{"commit":"25e60aea8e2fe11937e4b4d94e89c35cbb43940e","subject":"Merge \"Check policies for image tasks information in API\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/25e60aea8e2fe11937e4b4d94e89c35cbb43940e"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-08-31 16:12:49.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-08-31 16:21:26.000000000","tz":0},"subject":"DNM: Run glance ceph jobs with \u0027pacific\u0027 version","message":"DNM: Run glance ceph jobs with \u0027pacific\u0027 version\n\nChange-Id: Idb2ab1f4ac6a2c3e09cba4ca966b7faff30d34b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1041334bfad16b236505ac61ba042c9fb830cb7e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1041334bfad16b236505ac61ba042c9fb830cb7e"}]},"branch":"refs/heads/master"},"8a8ac8d7a34c02de40e833b5155c823148a4dc22":{"kind":"REWORK","_number":3,"created":"2021-08-31 18:42:10.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/74/806774/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/74/806774/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/74/806774/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/74/806774/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/74/806774/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/74/806774/3"}}},"commit":{"parents":[{"commit":"25e60aea8e2fe11937e4b4d94e89c35cbb43940e","subject":"Merge \"Check policies for image tasks information in API\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/25e60aea8e2fe11937e4b4d94e89c35cbb43940e"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-08-31 16:12:49.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-08-31 18:42:02.000000000","tz":0},"subject":"DNM: Run glance ceph jobs with \u0027pacific\u0027 version","message":"DNM: Run glance ceph jobs with \u0027pacific\u0027 version\n\nChange-Id: Idb2ab1f4ac6a2c3e09cba4ca966b7faff30d34b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8a8ac8d7a34c02de40e833b5155c823148a4dc22"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8a8ac8d7a34c02de40e833b5155c823148a4dc22"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
