)]}'
{"id":"openstack%2Fglance~806775","triplet_id":"openstack%2Fglance~master~Ic469ee50b9a3f162fd75a8290766d1e0bb4b26f5","project":"openstack/glance","branch":"master","topic":"secure-rbac","hashtags":[],"change_id":"Ic469ee50b9a3f162fd75a8290766d1e0bb4b26f5","subject":"Remove duplicate namespace utilities in functional tests","status":"MERGED","created":"2021-08-31 16:22:20.000000000","updated":"2021-09-13 02:47:12.000000000","submitted":"2021-09-10 16:26:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"806775-secure-rbac","meta_rev_id":"d07022ed8d061160cebc1260176059d33961c97a","_number":806775,"virtual_id_number":806775,"owner":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-10 16:26:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2021-09-10 14:12:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":1,"date":"2021-09-10 14:12:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-31 18:48:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-02 09:45:02.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-09-02 14:55:19.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2021-09-02 22:03:30.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"}],"messages":[{"id":"88280405a373414aa9bd3b21958facdb34147090","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-08-31 16:22:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"204ec45ebba17b3fcc6c3738b6d5f3d68bf9ef8d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-08-31 16:54:08.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"2b4751d1f48d37716795fb70199ade5e114d35a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-31 18:48:47.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cdff7d9c6bed4869bc156582a7ab2f4f : SUCCESS in 7m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7421f3f0536f4eafa1b1978f12fbf9bd : SUCCESS in 7m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/978f1dd0bb0044219f784b5db56847f1 : SUCCESS in 6m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/10cc1bce91b045ba8344910f8386309e : SUCCESS in 7m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df39e75683074431916801b3c4fbac8d : SUCCESS in 9m 52s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/dd290d3a7e9941f58bb943794065d999 : SUCCESS in 10m 45s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c43c7a15d9b44e05b16ffb38dd0741d5 : SUCCESS in 8m 44s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/1779fc1cc1ef43a6827e5444db2f7332 : SUCCESS in 10m 09s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/d39d9f09e2a84584afd3e0710bfeb146 : SUCCESS in 5m 25s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/7708b76b268a411684bc54334e920d24 : SUCCESS in 30m 12s","accounts_in_message":[],"_revision_number":2},{"id":"115e3ff615909967ce707a29eac5a5e73ea9691f","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-09-02 09:45:02.000000000","message":"Patch Set 2:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/806775/806775/2/check-not-vote/ext-glance-zuul/a5484a3 : SUCCESS in 37m 16s","accounts_in_message":[],"_revision_number":2},{"id":"d29e3ba54482ba253f6ee270d2d5ba13b8d614e0","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-09-02 14:43:10.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5ac35ba1a69deed0977e6c4e3f32f5e22f09aa6b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-09-02 14:55:19.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"7aba2e6ca2e74257f86c2af4aaf4fd6d17c12f35","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-02 16:07:58.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4401a4f807444706a10376f9129fbe70 : SUCCESS in 6m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f593df2e80844f039d3a7cf9cc1ec8b1 : SUCCESS in 8m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a85b694c561d4aef90dc135e6aaa393f : SUCCESS in 7m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/368c1fa591b74e90b8f021e55434c10d : SUCCESS in 7m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a50946ff781484b9fd4fa18361e90ae : SUCCESS in 7m 31s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7680af4617754aedb799a08fb99704b9 : SUCCESS in 8m 46s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/55dfb56cbf274cebad951012a28e02fa : SUCCESS in 10m 38s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/3967e1b89c034fe39fb139ddf2a2e6c4 : SUCCESS in 10m 35s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/6c91798ade5e458587393aa3a11da490 : SUCCESS in 5m 22s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/55cccf04de944690bb63820e6e67aac2 : SUCCESS in 33m 13s","accounts_in_message":[],"_revision_number":3},{"id":"eebc0822ccf244b6481553b98bd2399ba9d6f789","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-09-02 17:01:14.000000000","message":"Patch Set 3: Code-Review+2\n\nI have just rebased the patch!","accounts_in_message":[],"_revision_number":3},{"id":"de8f6251fcbd0106bd2266862094fa1ec2cd4fd3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-09-02 22:03:30.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"97e0e37c3f515f48d5b6c5e8e511891985df332c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-02 22:33:37.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/431e332bd2c1402a83004b5b1f05770a : SUCCESS in 6m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/747cbc9cf48845daac7f89fe997e1f2a : SUCCESS in 6m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9058c1c57fe54a0394d1e856763be963 : SUCCESS in 6m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d81a21b2d06040428d302d8871784a13 : SUCCESS in 7m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b92d0b89470148cb8e1b5bc9dd322172 : SUCCESS in 7m 30s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7ee498600e434bbf807867e7657b7698 : SUCCESS in 10m 37s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/56276ee9e77c416ca26b5a0349656888 : SUCCESS in 9m 45s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/197d118f051543a3a255aa5f14d87beb : SUCCESS in 11m 04s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/c52bd60d37ef4880a85bf14931c7e829 : SUCCESS in 7m 02s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/c3fcb678d2a940cd9bb7a6d09d964756 : SUCCESS in 27m 57s","accounts_in_message":[],"_revision_number":4},{"id":"ca5b6ce81e8ad9f0d40d2f6c1051d08e0b02f19f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-09-03 06:34:31.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"7836531e38af080261fc2c81267e3b94c9995aad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-03 07:45:02.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b7689ea76de4f00a5e580a92edef509 : SUCCESS in 7m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/41c5951f564f49eda059b52dc665ce89 : SUCCESS in 6m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/af2effd21df140afad5be62acacdf6b5 : SUCCESS in 10m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8a168688059b42a8b42ed1f32ea2986e : SUCCESS in 7m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ee69748328842649435222bf7073dd5 : SUCCESS in 11m 14s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/59b4fc66f7b040c48f52517d364959e2 : SUCCESS in 11m 03s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/3e89cb70b1bb4e8caa202266907fb283 : SUCCESS in 10m 54s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/3d6bb8050dc04ac29bf695895e35c631 : SUCCESS in 13m 13s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/088b9e3dd0b949da91ff832d696958c4 : SUCCESS in 7m 16s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/7fb0da3519044bdd9121005aef04e34d : SUCCESS in 32m 48s","accounts_in_message":[],"_revision_number":5},{"id":"121502a97945f3b3fbc6a15b8ba3483b86209b61","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-09-03 18:25:09.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"91cad4b4949267c1b6d11d4dc5623235ccf78c9d","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-09-06 10:09:02.000000000","message":"Patch Set 5:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/806775/806775/5/check-not-vote/ext-glance-zuul/41baeac : SUCCESS in 46m 45s","accounts_in_message":[],"_revision_number":5},{"id":"6323771dbf06f91ad9a099b1ee470a1209778a36","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-09-10 13:58:15.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"f5041b3fa5b044ae15529762da0cb37dc4a7abe7","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-09-10 14:08:02.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"48b8726b10884524349e157563b47003e268f2d4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-09-10 14:11:54.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"989133f2b601eab7745e66beb2f616462f61bf20","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-09-10 14:12:40.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1\n\nI just moved some code around per previous discussion. Abhi was +2 before, so ninja\u0027ing this in.","accounts_in_message":[],"_revision_number":6},{"id":"4f85151c490bf75b29a73f153b048507f7a82498","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-10 15:00:54.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da80c680dd504929bb8df60072c91e31 : SUCCESS in 6m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f2b99034c0e34bc5b42712d90579867c : SUCCESS in 6m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/03897a8162ee4e0d8bad61ad200d2922 : SUCCESS in 6m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/97e88a0b42994b17a63f81c676734371 : SUCCESS in 7m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ce87ef52ffc4fbf96be48c5ea893576 : SUCCESS in 7m 56s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0b3de70bf84a4b6db0cf962fdf2f93a3 : SUCCESS in 9m 39s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/e2244cebadfe4ada9ef5b5678a70c3f3 : SUCCESS in 9m 45s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/b93e40985b274445a2a61465ce0cd566 : SUCCESS in 10m 05s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/90d6e0e6cafb4f44b981cad3f9a5fb01 : SUCCESS in 5m 39s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/b81c3362c6224778a16e03b6a04ae78d : SUCCESS in 34m 10s","accounts_in_message":[],"_revision_number":6},{"id":"f528d561ac55c3befd0612e3416cff53a9ab0fc5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-10 15:06:41.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"41d2c389ba20fcae3de8f495d891abd129df751e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-10 16:26:29.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c8b5d2d961014671b8da7380ea7dc5e2 : SUCCESS in 9m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/743afdc47b0640b295bedc00d75faed8 : SUCCESS in 9m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cd4ca176b8b74bab9a06c417a0860c1a : SUCCESS in 8m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a6b52a799dd4ff2bc741fb7682b1d01 : SUCCESS in 10m 45s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/76d40bf38b9241918dbe9aceeecca9d0 : SUCCESS in 13m 31s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b59b421973e84efd993502312f874cdf : SUCCESS in 13m 23s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/aa88eb27e6eb475a9919c9125718cad6 : SUCCESS in 44m 02s","accounts_in_message":[],"_revision_number":6},{"id":"53a96c88e227c18f7415f25e807dc79445aee708","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-10 16:26:30.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"d07022ed8d061160cebc1260176059d33961c97a","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-09-13 02:47:12.000000000","message":"Patch Set 6:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/806775/806775/6/check-not-vote/ext-glance-zuul/d354f38 : SUCCESS in 37m 14s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"a8dda8c03c74f0372d3f753043515a69ddb9163b","revisions":{"dc079e18db4a99ab18408c9f8ea1d99e425f2d2d":{"kind":"REWORK","_number":1,"created":"2021-08-31 16:22:20.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/75/806775/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/75/806775/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/75/806775/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/75/806775/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/75/806775/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/75/806775/1"}}},"commit":{"parents":[{"commit":"f942d212d74b6b5b76d20a96855857e3c6833c14","subject":"Implement project personas for metadef tags","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f942d212d74b6b5b76d20a96855857e3c6833c14"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2021-08-31 16:18:28.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2021-08-31 16:18:28.000000000","tz":0},"subject":"Remove duplicate namespace utilities in functional tests","message":"Remove duplicate namespace utilities in functional tests\n\nWe\u0027re in the process of adding test coverage for the metadef API, which\nis sorely needed. This commit introduces two different base classes that\nhelp consolidate some duplication that\u0027s grown across the metadef tests.\n\nSince the metadef API contains a whole bunch of nest resources, the\nfirst class is a set of utilities that are handy for creating\nmetadef resources. This is nice because almost every resources in the\nmetadef API needs a namespace.\n\nThe second class is a subclass of the functional test case, but provides\na layer for metadef-specific test utilities, like assertions. We can\ncontinue using this layer to consolidate test utilities without bloating\nthe base functional test class.\n\nChange-Id: Ic469ee50b9a3f162fd75a8290766d1e0bb4b26f5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/dc079e18db4a99ab18408c9f8ea1d99e425f2d2d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/dc079e18db4a99ab18408c9f8ea1d99e425f2d2d"}]},"branch":"refs/heads/master"},"5c9741a31e1aca81450c72273ce866492e50f5a1":{"kind":"REWORK","_number":2,"created":"2021-08-31 16:54:08.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/75/806775/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/75/806775/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/75/806775/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/75/806775/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/75/806775/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/75/806775/2"}}},"commit":{"parents":[{"commit":"f942d212d74b6b5b76d20a96855857e3c6833c14","subject":"Implement project personas for metadef tags","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f942d212d74b6b5b76d20a96855857e3c6833c14"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2021-08-31 16:18:28.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2021-08-31 16:36:36.000000000","tz":0},"subject":"Remove duplicate namespace utilities in functional tests","message":"Remove duplicate namespace utilities in functional tests\n\nWe\u0027re in the process of adding test coverage for the metadef API, which\nis sorely needed. This commit introduces two different base classes that\nhelp consolidate some duplication that\u0027s grown across the metadef tests.\n\nSince the metadef API contains a whole bunch of nest resources, the\nfirst class is a set of utilities that are handy for creating\nmetadef resources. This is nice because almost every resources in the\nmetadef API needs a namespace.\n\nThe second class is a subclass of the functional test case, but provides\na layer for metadef-specific test utilities, like assertions. We can\ncontinue using this layer to consolidate test utilities without bloating\nthe base functional test class.\n\nChange-Id: Ic469ee50b9a3f162fd75a8290766d1e0bb4b26f5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5c9741a31e1aca81450c72273ce866492e50f5a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5c9741a31e1aca81450c72273ce866492e50f5a1"}]},"branch":"refs/heads/master"},"2dc8b29311770a1b3183f364acf15bc4fa87a5e3":{"kind":"REWORK","_number":3,"created":"2021-09-02 14:55:19.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/75/806775/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/75/806775/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/75/806775/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/75/806775/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/75/806775/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/75/806775/3"}}},"commit":{"parents":[{"commit":"11eab5ef8a3f3d912c609bb0a51500db8ea7169b","subject":"Implement project personas for metadef tags","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/11eab5ef8a3f3d912c609bb0a51500db8ea7169b"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2021-08-31 16:18:28.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-09-02 14:47:59.000000000","tz":0},"subject":"Remove duplicate namespace utilities in functional tests","message":"Remove duplicate namespace utilities in functional tests\n\nWe\u0027re in the process of adding test coverage for the metadef API, which\nis sorely needed. This commit introduces two different base classes that\nhelp consolidate some duplication that\u0027s grown across the metadef tests.\n\nSince the metadef API contains a whole bunch of nest resources, the\nfirst class is a set of utilities that are handy for creating\nmetadef resources. This is nice because almost every resources in the\nmetadef API needs a namespace.\n\nThe second class is a subclass of the functional test case, but provides\na layer for metadef-specific test utilities, like assertions. We can\ncontinue using this layer to consolidate test utilities without bloating\nthe base functional test class.\n\nChange-Id: Ic469ee50b9a3f162fd75a8290766d1e0bb4b26f5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2dc8b29311770a1b3183f364acf15bc4fa87a5e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2dc8b29311770a1b3183f364acf15bc4fa87a5e3"}]},"branch":"refs/heads/master"},"ffb35e71fceb4c57189ad33a5496d7a286675183":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2021-09-02 22:03:30.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/75/806775/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/75/806775/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/75/806775/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/75/806775/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/75/806775/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/75/806775/4"}}},"commit":{"parents":[{"commit":"a2f18e5554afe7316f4b750a4feb6348e9eefad7","subject":"Implement project personas for metadef tags","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a2f18e5554afe7316f4b750a4feb6348e9eefad7"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2021-08-31 16:18:28.000000000","tz":0},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-09-02 23:06:13.000000000","tz":-420},"subject":"Remove duplicate namespace utilities in functional tests","message":"Remove duplicate namespace utilities in functional tests\n\nWe\u0027re in the process of adding test coverage for the metadef API, which\nis sorely needed. This commit introduces two different base classes that\nhelp consolidate some duplication that\u0027s grown across the metadef tests.\n\nSince the metadef API contains a whole bunch of nest resources, the\nfirst class is a set of utilities that are handy for creating\nmetadef resources. This is nice because almost every resources in the\nmetadef API needs a namespace.\n\nThe second class is a subclass of the functional test case, but provides\na layer for metadef-specific test utilities, like assertions. We can\ncontinue using this layer to consolidate test utilities without bloating\nthe base functional test class.\n\nChange-Id: Ic469ee50b9a3f162fd75a8290766d1e0bb4b26f5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ffb35e71fceb4c57189ad33a5496d7a286675183"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ffb35e71fceb4c57189ad33a5496d7a286675183"}]},"branch":"refs/heads/master"},"6029c1f2f55b695e066f9ee8ee01f051fb910366":{"kind":"REWORK","_number":5,"created":"2021-09-03 06:34:31.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/75/806775/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/75/806775/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/75/806775/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/75/806775/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/75/806775/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/75/806775/5"}}},"commit":{"parents":[{"commit":"4473a6053644eec84552780930efe544fd503774","subject":"Implement project personas for metadef tags","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4473a6053644eec84552780930efe544fd503774"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2021-08-31 16:18:28.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-09-03 05:50:46.000000000","tz":0},"subject":"Remove duplicate namespace utilities in functional tests","message":"Remove duplicate namespace utilities in functional tests\n\nWe\u0027re in the process of adding test coverage for the metadef API, which\nis sorely needed. This commit introduces two different base classes that\nhelp consolidate some duplication that\u0027s grown across the metadef tests.\n\nSince the metadef API contains a whole bunch of nest resources, the\nfirst class is a set of utilities that are handy for creating\nmetadef resources. This is nice because almost every resources in the\nmetadef API needs a namespace.\n\nThe second class is a subclass of the functional test case, but provides\na layer for metadef-specific test utilities, like assertions. We can\ncontinue using this layer to consolidate test utilities without bloating\nthe base functional test class.\n\nChange-Id: Ic469ee50b9a3f162fd75a8290766d1e0bb4b26f5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6029c1f2f55b695e066f9ee8ee01f051fb910366"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6029c1f2f55b695e066f9ee8ee01f051fb910366"}]},"branch":"refs/heads/master"},"a8dda8c03c74f0372d3f753043515a69ddb9163b":{"kind":"REWORK","_number":6,"created":"2021-09-10 14:11:54.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/75/806775/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/75/806775/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/75/806775/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/75/806775/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/75/806775/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/75/806775/6"}}},"commit":{"parents":[{"commit":"0db149425c8d06ad3b43f691ece62a4cf255a85b","subject":"Merge \"Replace deprecated assertDictContainsSubset\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0db149425c8d06ad3b43f691ece62a4cf255a85b"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2021-08-31 16:18:28.000000000","tz":0},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-09-10 15:14:11.000000000","tz":-420},"subject":"Remove duplicate namespace utilities in functional tests","message":"Remove duplicate namespace utilities in functional tests\n\nWe\u0027re in the process of adding test coverage for the metadef API, which\nis sorely needed. This commit introduces two different base classes that\nhelp consolidate some duplication that\u0027s grown across the metadef tests.\n\nSince the metadef API contains a whole bunch of nest resources, the\nfirst class is a set of utilities that are handy for creating\nmetadef resources. This is nice because almost every resources in the\nmetadef API needs a namespace.\n\nThe second class is a subclass of the functional test case, but provides\na layer for metadef-specific test utilities, like assertions. We can\ncontinue using this layer to consolidate test utilities without bloating\nthe base functional test class.\n\nChange-Id: Ic469ee50b9a3f162fd75a8290766d1e0bb4b26f5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a8dda8c03c74f0372d3f753043515a69ddb9163b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a8dda8c03c74f0372d3f753043515a69ddb9163b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
