)]}'
{"id":"openstack%2Fglance~806797","triplet_id":"openstack%2Fglance~master~I0b161ad404bdb53e3c1cfe878f78a71c239cf40e","project":"openstack/glance","branch":"master","topic":"secure-rbac","hashtags":[],"change_id":"I0b161ad404bdb53e3c1cfe878f78a71c239cf40e","subject":"Remove duplicate _url() methods from functional test classes","status":"MERGED","created":"2021-08-31 19:51:50.000000000","updated":"2021-09-10 16:26:34.000000000","submitted":"2021-09-10 16:26:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"806797-secure-rbac","meta_rev_id":"37830ace1b8bab27af9b7d44e82714a61e050651","_number":806797,"virtual_id_number":806797,"owner":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-10 16:26:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2021-09-10 16:26:34.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2021-09-10 14:13:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":1,"date":"2021-09-10 14:13:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-31 20:44:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-02 15:53:15.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2021-09-02 22:03:30.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2021-09-04 22:43:31.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"53984f94ffceb8001be98d87cdb04c41d4309fa5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-08-31 19:51:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6d54111c8c89dacbcf3dee0c8dc144e165216635","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-31 20:44:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/21e4507b538647d29ddacc8173595489 : SUCCESS in 6m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/34a0c22451444f3e89f2b13bee98b823 : SUCCESS in 7m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6c433d439ab9464280598214d4cc08c7 : SUCCESS in 6m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4047338eaf9c4f75b8dd43a223aa99b3 : SUCCESS in 6m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92d273874af743ce9f0486dfbf764bd5 : SUCCESS in 7m 51s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4e2d9329d0294f9f983252293aeedc26 : SUCCESS in 9m 31s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c69d03b741b04d30b54c603f79c4eaa1 : SUCCESS in 9m 23s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/812c5bf7e6a44ba1ab0babadcc81e761 : SUCCESS in 9m 58s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/dcf48cae5f7c4ad9a9d9e08169d552f4 : SUCCESS in 6m 15s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/a0012d4f4d4e46ec9899f89e57b8ec5b : SUCCESS in 42m 07s","accounts_in_message":[],"_revision_number":1},{"id":"d03b83af42bcd92483ea0bc08ddd1994f7737411","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-09-02 15:53:15.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"aa95fb6353cd64a3851c736721a03cc4cdcacb08","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-02 16:50:11.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a0292721547f49c8807ea35098ead534 : SUCCESS in 6m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6e7940fd3fe94c3aa4ac244c6fa31b30 : SUCCESS in 7m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a25d7a1a957748558120b271bdebb892 : SUCCESS in 6m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f3d33fa1950d455eafb1a7cfc7cb1173 : SUCCESS in 8m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f39aad50c1e24849ada582eed69ca698 : SUCCESS in 8m 41s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/edffe4ef819b4d0ab20a5fba9db2f18b : SUCCESS in 10m 23s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/91b221f7bb2b441db094b944e4fb2385 : SUCCESS in 11m 19s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/cafcc6ab4f9349e1905917ea571f91b6 : SUCCESS in 11m 42s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/c7875e2249ba4f2ba2a3d8daa7943b36 : SUCCESS in 6m 16s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/32306fb7b8fd4547b811b0918cd0d199 : SUCCESS in 32m 56s","accounts_in_message":[],"_revision_number":2},{"id":"e994baf61e4a5a73b3e7b66735adf803c2b9d644","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-09-02 17:01:26.000000000","message":"Patch Set 2:\n\nI have just rebased the patch!","accounts_in_message":[],"_revision_number":2},{"id":"bb9e0460cab8400d02337d07797c7120d9eebc3a","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-09-02 17:15:10.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"8188bb91f26287d8e873a4e5807fda8a02905e93","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-09-02 22:03:30.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"2ccb0e9cd981613022f15ca435e7b3d868e01f6a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-02 22:42:25.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/268c2af26e4c4762a598b69e4aba55c2 : SUCCESS in 6m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b04b390f575c4b0ca646db74dc4c7a2f : SUCCESS in 6m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/510fd0db44594886b6049cbbe847a55f : SUCCESS in 6m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0cd1eb849bd547bd91776b209e1f7131 : SUCCESS in 7m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64d594def5e5422d976783470a4cdadc : SUCCESS in 7m 29s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/96abb90376a14017921acb86d7de31d7 : SUCCESS in 9m 28s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/3de003a20acb42a6bf594d639a91dcd6 : SUCCESS in 9m 19s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/44c0704685c94461b8b30a814cbd6467 : SUCCESS in 10m 25s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/f57d12e2799648bb883b23cb869fd2c1 : SUCCESS in 5m 09s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/86d6d61f8aee4211b24ee2954cd88a7d : SUCCESS in 27m 03s","accounts_in_message":[],"_revision_number":3},{"id":"b630149ee05de273fbd4473647dabbe58cec251b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-09-03 06:34:31.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"1a3cab99fc0a2645c48dea0ccbf2a032fd3c5aed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-03 07:45:46.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b310b1777d19499ead61569ee192ae2c : SUCCESS in 8m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2822ac144b814179ad26259d0bc294df : SUCCESS in 9m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cdd86729dfeb496da941b886fe396f65 : SUCCESS in 9m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0d6d0bb9abe14687942265ce9252f967 : SUCCESS in 9m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4af4e91fa5e4595ada264287b254404 : SUCCESS in 8m 27s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c8015a36e1e64d1cb86dce03ca91e588 : SUCCESS in 11m 59s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/4267e47c77d94ad49025a1c628d7f1d8 : SUCCESS in 12m 43s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/ac19d8f175c74c2fbb590b41f08a79c2 : SUCCESS in 14m 18s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/58de21703c5948c58e2331977dc03f7a : SUCCESS in 7m 06s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/443878add63141308568defd3ad4c139 : SUCCESS in 34m 33s","accounts_in_message":[],"_revision_number":4},{"id":"0c41dd1098fa02f29d32e38d84f1ce452a81adf5","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-09-04 22:43:31.000000000","message":"Patch Set 2:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":2},{"id":"c723a248cf2de68250eba9f725a7a9e809c20aed","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-09-10 14:11:54.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"c7b21b81190f7937c7c55e4d29f711550c770235","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-09-10 14:13:02.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"945edc4226ff8bee047954812fb945d55f0045df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-10 15:01:23.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c6f7a2187b842199bd6af69f7a1186d : SUCCESS in 7m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5f36493bc54c4c0db5aaa072de85fbe4 : SUCCESS in 6m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6682e222bc8e44f8977a68fadbfaa31a : SUCCESS in 8m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8cdd82c7515a425f8fcd8161747cb4b6 : SUCCESS in 7m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d7438474b974db7ad4bd6a38c12b2d4 : SUCCESS in 8m 19s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/718c3fcab47d4371a66e721f464ccdd7 : SUCCESS in 9m 16s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/64dd4d63707a4b53a639bee1c3e23b60 : SUCCESS in 10m 02s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/55138437a10f4eb18cb8c5509b50f8ef : SUCCESS in 10m 02s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/230ff87779af4e2a82833033f793c875 : SUCCESS in 6m 21s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/49ebbc6bee36476f81d3d7b618f497cd : SUCCESS in 37m 16s","accounts_in_message":[],"_revision_number":5},{"id":"46096ac752c6dbff9f84f0efd23c45cc7d3e5662","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-10 15:07:51.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"eb985fddf92ad2d86ad36667074cabfd2a5ab250","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-10 16:26:34.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c3bee537b68454e833280700aeb8463 : SUCCESS in 7m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/50ba4a93268c44479865266df2c52485 : SUCCESS in 10m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f2197f0d459047c1bebf88c4c79e8752 : SUCCESS in 9m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8a26b25f5354b40bfeebd9efeb91a7b : SUCCESS in 10m 24s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8df8998e4279450f82e0a8378a46aca3 : SUCCESS in 11m 13s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/0509c48f4c214436adfe224a05d450f6 : SUCCESS in 13m 29s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/eef43032d4044ce797c9d82549073118 : SUCCESS in 36m 20s","accounts_in_message":[],"_revision_number":5},{"id":"923f161f4db3751d6bcc3f065c8d11f19450be4e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-10 16:26:34.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"4e68e5da6f4c787e9fc6af76847602a69b0feb96","revisions":{"eb0f873bf79029bffce1e5b71ce835688eb2a06f":{"kind":"REWORK","_number":1,"created":"2021-08-31 19:51:50.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/97/806797/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/97/806797/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/97/806797/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/97/806797/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/97/806797/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/97/806797/1"}}},"commit":{"parents":[{"commit":"fd687899d0b1afa6d60967d43bb202b6bdddf029","subject":"Move global constants to test module setUp","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/fd687899d0b1afa6d60967d43bb202b6bdddf029"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2021-08-31 19:50:03.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2021-08-31 19:50:03.000000000","tz":0},"subject":"Remove duplicate _url() methods from functional test classes","message":"Remove duplicate _url() methods from functional test classes\n\nMost of the functional tests contain a version of _url(). The\nimplementations don\u0027t really differ, outside of one that accepts a\nprotocol and another that is specific to IPV6 for a single test case.\n\nThis commit moves the common method to the functional base class and\nremoves the duplicated logic across the functional test implementations.\n\nChange-Id: I0b161ad404bdb53e3c1cfe878f78a71c239cf40e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/eb0f873bf79029bffce1e5b71ce835688eb2a06f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/eb0f873bf79029bffce1e5b71ce835688eb2a06f"}]},"branch":"refs/heads/master"},"f412d0baa54ab4089156a2f34f4be366644274a0":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-09-02 15:53:15.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/97/806797/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/97/806797/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/97/806797/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/97/806797/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/97/806797/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/97/806797/2"}}},"commit":{"parents":[{"commit":"3a596339d38830042db36c1b44976c17b70f309d","subject":"Move global constants to test module setUp","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3a596339d38830042db36c1b44976c17b70f309d"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2021-08-31 19:50:03.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-09-02 15:19:44.000000000","tz":0},"subject":"Remove duplicate _url() methods from functional test classes","message":"Remove duplicate _url() methods from functional test classes\n\nMost of the functional tests contain a version of _url(). The\nimplementations don\u0027t really differ, outside of one that accepts a\nprotocol and another that is specific to IPV6 for a single test case.\n\nThis commit moves the common method to the functional base class and\nremoves the duplicated logic across the functional test implementations.\n\nChange-Id: I0b161ad404bdb53e3c1cfe878f78a71c239cf40e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f412d0baa54ab4089156a2f34f4be366644274a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f412d0baa54ab4089156a2f34f4be366644274a0"}]},"branch":"refs/heads/master"},"5454bb493956299483e266f8b8f507775497e0f8":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-09-02 22:03:30.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/97/806797/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/97/806797/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/97/806797/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/97/806797/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/97/806797/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/97/806797/3"}}},"commit":{"parents":[{"commit":"98082e4c5e2c645d5b7f2b49dfb5773f7bc43877","subject":"Move global constants to test module setUp","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/98082e4c5e2c645d5b7f2b49dfb5773f7bc43877"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2021-08-31 19:50:03.000000000","tz":0},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-09-02 23:06:13.000000000","tz":-420},"subject":"Remove duplicate _url() methods from functional test classes","message":"Remove duplicate _url() methods from functional test classes\n\nMost of the functional tests contain a version of _url(). The\nimplementations don\u0027t really differ, outside of one that accepts a\nprotocol and another that is specific to IPV6 for a single test case.\n\nThis commit moves the common method to the functional base class and\nremoves the duplicated logic across the functional test implementations.\n\nChange-Id: I0b161ad404bdb53e3c1cfe878f78a71c239cf40e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5454bb493956299483e266f8b8f507775497e0f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5454bb493956299483e266f8b8f507775497e0f8"}]},"branch":"refs/heads/master"},"3c1c1ff497a732b95a77c9aaf6ca6fda9c415a64":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2021-09-03 06:34:31.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/97/806797/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/97/806797/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/97/806797/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/97/806797/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/97/806797/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/97/806797/4"}}},"commit":{"parents":[{"commit":"29e47933e09091edb5b353b7505d5e50cd3a9b99","subject":"Move global constants to test module setUp","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/29e47933e09091edb5b353b7505d5e50cd3a9b99"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2021-08-31 19:50:03.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-09-03 06:03:27.000000000","tz":0},"subject":"Remove duplicate _url() methods from functional test classes","message":"Remove duplicate _url() methods from functional test classes\n\nMost of the functional tests contain a version of _url(). The\nimplementations don\u0027t really differ, outside of one that accepts a\nprotocol and another that is specific to IPV6 for a single test case.\n\nThis commit moves the common method to the functional base class and\nremoves the duplicated logic across the functional test implementations.\n\nChange-Id: I0b161ad404bdb53e3c1cfe878f78a71c239cf40e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3c1c1ff497a732b95a77c9aaf6ca6fda9c415a64"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3c1c1ff497a732b95a77c9aaf6ca6fda9c415a64"}]},"branch":"refs/heads/master"},"4e68e5da6f4c787e9fc6af76847602a69b0feb96":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2021-09-10 14:11:54.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/97/806797/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/97/806797/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/97/806797/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/97/806797/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/97/806797/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/97/806797/5"}}},"commit":{"parents":[{"commit":"82a894588517a212228bb9ce1808326ba4bdf359","subject":"Move global constants to test module setUp","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/82a894588517a212228bb9ce1808326ba4bdf359"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2021-08-31 19:50:03.000000000","tz":0},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-09-10 15:14:52.000000000","tz":-420},"subject":"Remove duplicate _url() methods from functional test classes","message":"Remove duplicate _url() methods from functional test classes\n\nMost of the functional tests contain a version of _url(). The\nimplementations don\u0027t really differ, outside of one that accepts a\nprotocol and another that is specific to IPV6 for a single test case.\n\nThis commit moves the common method to the functional base class and\nremoves the duplicated logic across the functional test implementations.\n\nChange-Id: I0b161ad404bdb53e3c1cfe878f78a71c239cf40e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4e68e5da6f4c787e9fc6af76847602a69b0feb96"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4e68e5da6f4c787e9fc6af76847602a69b0feb96"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
