)]}'
{"id":"openstack%2Fglance~807242","triplet_id":"openstack%2Fglance~master~Ib0d69c875ed6291b0ab2b82e737d201f7041bbc8","project":"openstack/glance","branch":"master","topic":"m3-config-refresh","hashtags":[],"change_id":"Ib0d69c875ed6291b0ab2b82e737d201f7041bbc8","subject":"Refresh Glance example configs for Xena milestone 3","status":"MERGED","created":"2021-09-03 07:26:11.000000000","updated":"2021-09-08 03:49:09.000000000","submitted":"2021-09-03 17:05:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"807242-m3-config-refresh","meta_rev_id":"3b73fe3c6097e4aa319d7dbe4c93f8b0bed6aa15","_number":807242,"virtual_id_number":807242,"owner":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-03 17:05:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2021-09-03 14:17:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":1,"date":"2021-09-03 15:43:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-03 07:38:19.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2021-09-03 08:41:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-08 03:49:09.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"5f7d09f030312ee853b62f39960b51dde54e6928","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-09-03 07:26:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d06fbe59a042fcc0e09c5e9ac8edd65b0e79e22e","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-09-03 07:38:45.000000000","message":"Patch Set 1: Workflow-1\n\nLooks like something is wrong with config generator, needs to verify","accounts_in_message":[],"_revision_number":1},{"id":"d65402cf20487d6c5cdd5ba092b300fe9697b02d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-03 08:41:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/a5e18f6672334f6ba146c742d8136141 : SUCCESS in 1h 07m 07s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4a2beb86c60541498a8836ce9f9b9ffe : SUCCESS in 1h 07m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1cfda9926dd74fbbb71fb27b9cee8bed : SUCCESS in 5m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e86be269e5ef4bdba4be134cecdaf701 : SUCCESS in 6m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e6d8a273dd08456999f0a4e93550a2ff : SUCCESS in 6m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7d8c586a6c16453fa6faa54ebcc0b85b : SUCCESS in 6m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/982cb7f97a374377a8a82ef3aac627b3 : SUCCESS in 7m 52s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d85057bdf7df4a69b7fd835fa9359ae7 : SUCCESS in 14m 39s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/0f1e3a8db62a44e89ad32c5a08a3c22f : SUCCESS in 9m 32s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/24773ff1e07c4d3fa544525295eff19c : SUCCESS in 16m 18s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/7a87f97d1e604ee6acf8495c31780dae : SUCCESS in 1h 10m 15s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/8e6d994a0a314cc9bf325043fef41ed8 : SUCCESS in 1h 13m 45s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/096e596063834329962ae6aa31021ec3 : SUCCESS in 1h 10m 52s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/f1294ff76e85425da342fb023d819f90 : SUCCESS in 1h 13m 52s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6d4ff7fe1bc44dd2a61a614e9855870f : SUCCESS in 45m 06s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/fa153a8e52ab4d7db80468200b76c044 : SUCCESS in 1h 12m 54s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/52f846aed4f14b03bf7b550d430e6852 : SUCCESS in 31m 47s","accounts_in_message":[],"_revision_number":1},{"id":"b7d1a1e6563df65c1db10fcf7d58b8473e011d5e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-09-03 13:44:34.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1681c632059c91e58b32f80e5f2cce0de515bd89","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-09-03 14:17:57.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"a0074cc976a5816c703db8dbfecf3d8f88b23947","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-03 15:09:57.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/a04704cbc88a452e80308c5a0a7994d8 : SUCCESS in 59m 55s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/af31cb53ca274872b077cf3e03755efb : SUCCESS in 1h 15m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/66f4c4cb66564bee94d35aff63738a52 : SUCCESS in 8m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/051925d3ad6d4d94ad18e8ab4f96b9bf : SUCCESS in 7m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e90e0619ccc346a08f3e317930bab723 : SUCCESS in 7m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9f3763f1af3349a1a0943f38d2177a1d : SUCCESS in 8m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/098ea00e125f45c08d71891b0fda4464 : SUCCESS in 7m 38s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b06b4a911bab44e8a3d72ad273bc70a7 : SUCCESS in 10m 11s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/ed25afa1b2b940b7af8ff415ce75bdaf : SUCCESS in 12m 35s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/e16fe34408a0462db359dd4966559258 : SUCCESS in 11m 14s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/124bb917ab55421c9de3f90db8134c8d : SUCCESS in 1h 19m 54s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/c5b85681a86445388fed9f4e03403d2b : SUCCESS in 1h 06m 33s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/a71a59f932674af39f0c63fbc28406b4 : SUCCESS in 1h 16m 30s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/a8aef5389e1b46f6be89fddf42bf3f60 : SUCCESS in 1h 15m 16s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/664f47e5bfee47b6a249e1a2571bef0c : SUCCESS in 48m 33s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/c4463e6cd31348a5a8a72789582aa299 : SUCCESS in 1h 06m 33s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/90f0d2fd15234e7c8ba97364819dcacd : SUCCESS in 31m 26s","accounts_in_message":[],"_revision_number":2},{"id":"ae7dc04d51896cd3e986215c830d1eec6bc34765","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-09-03 15:43:49.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"c0561dad25fddeb0f52797937d66b2d19795588b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-03 15:44:42.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"87cbcb278731c4069ca7a0dcb7fe9fdf7d0278a8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-03 17:05:13.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/c8ae8cc226974f9990cf54952965b7f3 : SUCCESS in 1h 00m 31s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/33914d9e9ae848c58196bab65bc19935 : SUCCESS in 1h 08m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/40f6deefee404d5eb858a0b54253db66 : SUCCESS in 7m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dd1141d397d448009676f3c16b224569 : SUCCESS in 9m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/25420ceadd3e4f038687692afc96531c : SUCCESS in 7m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1fc93c1f8a14e7ab837305d3475c5f5 : SUCCESS in 9m 40s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4da88fd1d8b7494da5c07d3987b9736b : SUCCESS in 11m 15s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/49763ff35d1645bd83a9ffcbb81e0aea : SUCCESS in 11m 29s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/93dd8a6b0fcd47bf93979f42e0d974dd : SUCCESS in 1h 07m 59s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/e4de428e77c94530acd88a5499d290d7 : SUCCESS in 1h 14m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/606012b75f1046208255d32a5b0f149c : SUCCESS in 51m 38s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/5f5ab9eb90c1400e815b84cc4d5ca09d : SUCCESS in 1h 12m 05s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/1abcbb1ae67a45a7af14dd1daf4fe94d : SUCCESS in 29m 36s","accounts_in_message":[],"_revision_number":2},{"id":"a691711b6f373b71b6f29c3d32165c43492f09d8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-03 17:05:13.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"13bf151db80277e50071dafbb1091a03db648236","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-03 17:07:21.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb5298fff503448a9e4b5dc5bbdc8074 : SUCCESS in 1m 44s","accounts_in_message":[],"_revision_number":2},{"id":"3b73fe3c6097e4aa319d7dbe4c93f8b0bed6aa15","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-09-08 03:49:09.000000000","message":"Patch Set 2:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/807242/807242/2/check-not-vote/ext-glance-zuul/76308e3 : SUCCESS in 38m 05s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0bd3791e15d4328b587b6339dcd84ea682c562fd","revisions":{"dafb753ff16c26a563c51e4b1f0f1b2be627031d":{"kind":"REWORK","_number":1,"created":"2021-09-03 07:26:11.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/42/807242/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/42/807242/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/42/807242/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/42/807242/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/42/807242/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/42/807242/1"}}},"commit":{"parents":[{"commit":"7af5597043ae2020dc7a4435d06f3ab0c3ce257b","subject":"Merge \"Implement project personas for metadef resource-types\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7af5597043ae2020dc7a4435d06f3ab0c3ce257b"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-09-03 07:17:52.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-09-03 07:17:52.000000000","tz":0},"subject":"Refresh Glance example configs for Xena milestone 3","message":"Refresh Glance example configs for Xena milestone 3\n\nChange-Id: Ib0d69c875ed6291b0ab2b82e737d201f7041bbc8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/dafb753ff16c26a563c51e4b1f0f1b2be627031d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/dafb753ff16c26a563c51e4b1f0f1b2be627031d"}]},"branch":"refs/heads/master"},"0bd3791e15d4328b587b6339dcd84ea682c562fd":{"kind":"REWORK","_number":2,"created":"2021-09-03 13:44:34.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/42/807242/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/42/807242/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/42/807242/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/42/807242/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/42/807242/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/42/807242/2"}}},"commit":{"parents":[{"commit":"7af5597043ae2020dc7a4435d06f3ab0c3ce257b","subject":"Merge \"Implement project personas for metadef resource-types\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7af5597043ae2020dc7a4435d06f3ab0c3ce257b"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-09-03 13:36:57.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-09-03 13:38:41.000000000","tz":0},"subject":"Refresh Glance example configs for Xena milestone 3","message":"Refresh Glance example configs for Xena milestone 3\n\nChange-Id: Ib0d69c875ed6291b0ab2b82e737d201f7041bbc8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0bd3791e15d4328b587b6339dcd84ea682c562fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0bd3791e15d4328b587b6339dcd84ea682c562fd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
