)]}'
{"id":"openstack%2Fglance~807827","triplet_id":"openstack%2Fglance~master~I7fea410a1bdee883e965b0dca4a5ffe8b36ff067","project":"openstack/glance","branch":"master","topic":"bug/1942977","hashtags":[],"change_id":"I7fea410a1bdee883e965b0dca4a5ffe8b36ff067","subject":"Remove single quotes in exception message","status":"ABANDONED","created":"2021-09-08 06:40:51.000000000","updated":"2021-09-21 01:38:04.000000000","total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"96111a0a5fe2c6819c64485be1a82adf07076f26","_number":807827,"virtual_id_number":807827,"owner":{"_account_id":23950,"name":"Eric Xie","email":"eric_xiett@163.com","username":"ericxie"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-09-11 05:03:54.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"all":[{"value":-1,"date":"2021-09-20 13:50:12.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":-1,"date":"2021-09-20 13:48:34.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-08 08:26:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-08 17:17:20.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2021-09-13 18:05:18.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-09-20 13:48:34.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"}],"messages":[{"id":"3e2060096e3d5d0b91bb0df9f97ecead399d64af","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23950,"name":"Eric Xie","email":"eric_xiett@163.com","username":"ericxie"},"date":"2021-09-08 06:40:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1a33c1567536e60756232d7939f4a187db060e02","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-08 08:26:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/bd8f4c6429d4458faaf258e66fcee1f4 : SUCCESS in 1h 00m 50s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3e8907f91ff541d49918f0c61fe187b3 : SUCCESS in 1h 03m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db9806987b354b0f847296d62f111c99 : SUCCESS in 6m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2917ee79a18f482aa9c36d85a46b74cc : SUCCESS in 6m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/84a95c4772614d25a32e283b7e2df3fe : SUCCESS in 7m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9842692378f545e389bf1cc4f352477c : SUCCESS in 9m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6b7ef0dadc04796984419688c93d9ac : SUCCESS in 9m 30s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/aff188ee58c34bc2bd46ccffeae65a58 : SUCCESS in 9m 55s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/a1abb2738d7742a6af535d932be58344 : SUCCESS in 10m 06s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/9c94baaca34a453e8cbbba976a020323 : SUCCESS in 8m 51s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/1414105f31014f95b7b092b502af239c : SUCCESS in 6m 07s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/6fcd62c05d074e44b971759b9447fe00 : SUCCESS in 1h 07m 37s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/8f8a5aa592b24f5f96f5dfdee92f05ca : SUCCESS in 1h 06m 08s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/b3f5665a17ee4760a6ba9740e535b470 : SUCCESS in 1h 21m 40s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/f56813c473a14683b12665b8e5a59f88 : SUCCESS in 1h 12m 03s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4d6d8f06436d48a0a10fb882316dffe7 : SUCCESS in 49m 13s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/01458a22937242e7ba9e02df26e2bfbd : SUCCESS in 1h 15m 05s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/ac026689cdbe4a3fa6138e797a237ebb : SUCCESS in 32m 04s","accounts_in_message":[],"_revision_number":1},{"id":"7ad5612b0103cab48b4e5a65cfcc00c17a2fd20a","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-09-08 17:17:20.000000000","message":"Patch Set 1: Code-Review-1\n\n(3 comments)\n\nI\u0027m not sure I see the point of this patch. The bug shows the HTML escapes used, which is exactly what you\u0027d expect in an HTML error message that includes escape-able characters.\n\nWhat does this actually fix? I\u0027m not sure I think the new version of these messages is really better or worth the change, IMHO.","accounts_in_message":[],"_revision_number":1},{"id":"78e41b1150942ea0e65fd2ccdc8647e7fd62b55b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23950,"name":"Eric Xie","email":"eric_xiett@163.com","username":"ericxie"},"date":"2021-09-11 02:17:01.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"64d82b6d6907241fcac92be452fdfe3c04f23708","author":{"_account_id":23950,"name":"Eric Xie","email":"eric_xiett@163.com","username":"ericxie"},"date":"2021-09-11 02:18:18.000000000","message":"Patch Set 1:\n\n(3 comments)\n\nIn fact, this patch has no functional problems, just an improvement of the user experience.","accounts_in_message":[],"_revision_number":1},{"id":"8ee3d91e3dfbdc35f5db01043dfd645ee743349b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-11 05:03:54.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/97482f1748ee41da93c4e832aac9f066 : SUCCESS in 1h 01m 38s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2553218e5b764bcd8aaecc8740f8d5bb : TIMED_OUT in 2h 09m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/00498f3c8bce49d287d29a5eb79c3807 : SUCCESS in 5m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1bce418657384d1aab3a10fb420ff898 : SUCCESS in 6m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f7bb5055cbc6456cb80bdc3d77e864ef : SUCCESS in 7m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7931bdfd914e4544a251ad868391e8f0 : SUCCESS in 6m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d206849c8ddb493082be463c6851c97e : SUCCESS in 7m 56s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e4848c8925444a5b8e553c325cdb0ec6 : SUCCESS in 10m 04s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/fd03a3de6ed64aefbd55ef1852e79344 : SUCCESS in 9m 01s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/b2115130a71a4bc6ad8a21b12f7ccb32 : SUCCESS in 9m 08s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/04840755253241c1bd614217c6ce2346 : SUCCESS in 5m 21s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/dc11a2f35cae4ca1b7e67bc29f0b8141 : SUCCESS in 1h 13m 44s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/2dffec77ffa34f659d74ad27bb161bd2 : SUCCESS in 1h 00m 07s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/47afb6a34a4b43b982ba490f8f034871 : SUCCESS in 1h 11m 16s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/295a0ec71d1b42afa9c5f198179d6886 : TIMED_OUT in 2h 10m 37s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e0a19aa7731d42fa827517adf6cbcef8 : SUCCESS in 1h 32m 08s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/d1a7a6df5da241b3b209741ebaf1985d : SUCCESS in 1h 30m 05s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/f3be86e4e1a8439b8fce437f60cd98be : SUCCESS in 32m 49s","accounts_in_message":[],"_revision_number":2},{"id":"f5d28f2d7c243409d903296cbd6c4686eafdbc90","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-09-13 18:05:18.000000000","message":"Patch Set 2:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/807827/807827/2/check-not-vote/ext-glance-zuul/0c4fad7 : SUCCESS in 42m 39s","accounts_in_message":[],"_revision_number":2},{"id":"268ee6a7a35f4de39d7a7fa7e102c9d95377f364","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2021-09-20 13:48:34.000000000","message":"Patch Set 2: Code-Review-1\n\nI must say I like the current version better. I really don\u0027t understand how the new version is an improvement.","accounts_in_message":[],"_revision_number":2},{"id":"57e129118c00dc9e5706d8adda629a315d497610","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-09-20 13:50:12.000000000","message":"Patch Set 2: Code-Review-1\n\n\u003e I must say I like the current version better. I really don\u0027t understand how the new version is an improvement.\n\nAs noted in my previous review, I agree. Putting my -1 back on here to signal that.","accounts_in_message":[],"_revision_number":2},{"id":"96111a0a5fe2c6819c64485be1a82adf07076f26","tag":"autogenerated:gerrit:abandon","author":{"_account_id":23950,"name":"Eric Xie","email":"eric_xiett@163.com","username":"ericxie"},"date":"2021-09-21 01:38:04.000000000","message":"Abandoned\n\nOK, abandon this patch.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e66fec64053ef78c4e3d5c8ec2c0bf6b23b46cd3","revisions":{"65dc9c759bcddb22b132b3a69d82e2b56f43f47d":{"kind":"REWORK","_number":1,"created":"2021-09-08 06:40:51.000000000","uploader":{"_account_id":23950,"name":"Eric Xie","email":"eric_xiett@163.com","username":"ericxie"},"ref":"refs/changes/27/807827/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/27/807827/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/807827/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/807827/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/807827/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/27/807827/1"}}},"commit":{"parents":[{"commit":"929190a6afd0a1593e62abb8024c339713b52b98","subject":"Merge \"Add doc support for delete-from-store API\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/929190a6afd0a1593e62abb8024c339713b52b98"}]}],"author":{"name":"ericxiett","email":"eric_xiett@163.com","date":"2021-09-08 06:38:00.000000000","tz":480},"committer":{"name":"ericxiett","email":"eric_xiett@163.com","date":"2021-09-08 06:38:00.000000000","tz":480},"subject":"Remove single quotes in excetion message","message":"Remove single quotes in excetion message\n\nThere are escape characters in the returned exception prompt. This patch\n removes these single quotes.\n\nCloses-Bug: #1942977\nChange-Id: I7fea410a1bdee883e965b0dca4a5ffe8b36ff067\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/65dc9c759bcddb22b132b3a69d82e2b56f43f47d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/65dc9c759bcddb22b132b3a69d82e2b56f43f47d"}]},"branch":"refs/heads/master"},"e66fec64053ef78c4e3d5c8ec2c0bf6b23b46cd3":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-09-11 02:17:01.000000000","uploader":{"_account_id":23950,"name":"Eric Xie","email":"eric_xiett@163.com","username":"ericxie"},"ref":"refs/changes/27/807827/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/27/807827/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/807827/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/807827/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/807827/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/27/807827/2"}}},"commit":{"parents":[{"commit":"929190a6afd0a1593e62abb8024c339713b52b98","subject":"Merge \"Add doc support for delete-from-store API\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/929190a6afd0a1593e62abb8024c339713b52b98"}]}],"author":{"name":"ericxiett","email":"eric_xiett@163.com","date":"2021-09-08 06:38:00.000000000","tz":480},"committer":{"name":"ericxiett","email":"eric_xiett@163.com","date":"2021-09-11 02:15:17.000000000","tz":480},"subject":"Remove single quotes in exception message","message":"Remove single quotes in exception message\n\nThere are escape characters in the returned exception message. This patch\nremoves these single quotes.\n\nCloses-Bug: #1942977\nChange-Id: I7fea410a1bdee883e965b0dca4a5ffe8b36ff067\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e66fec64053ef78c4e3d5c8ec2c0bf6b23b46cd3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e66fec64053ef78c4e3d5c8ec2c0bf6b23b46cd3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
