)]}'
{"id":"openstack%2Fglance~807932","triplet_id":"openstack%2Fglance~master~I9330cc42900dd019e9132c1f4e49eb9eff4f3fc9","project":"openstack/glance","branch":"master","topic":"fix-killed","hashtags":[],"change_id":"I9330cc42900dd019e9132c1f4e49eb9eff4f3fc9","subject":"Make signature verification go back to \u0027queued\u0027","status":"MERGED","created":"2021-09-08 17:49:41.000000000","updated":"2021-09-12 03:42:06.000000000","submitted":"2021-09-10 16:26:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"807932-fix-killed","meta_rev_id":"fdaebec2fe8691988e70a06102c87d2944e0c2fd","_number":807932,"virtual_id_number":807932,"owner":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-10 16:26:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"all":[{"value":2,"date":"2021-09-10 14:40:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":2,"date":"2021-09-08 18:08:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"all":[{"value":1,"date":"2021-09-10 14:40:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]}],"REVIEWER":[{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-08 18:08:13.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2021-09-08 19:25:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-10 14:40:24.000000000","updated_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"},{"updated":"2021-09-12 03:42:06.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"82f3520301eed54d6f0f7aa8f527b53b6909aba1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-09-08 17:49:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"854028efd1b110bfda2a2c4ede62d07d5563720d","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-09-08 18:08:13.000000000","message":"Patch Set 1: Code-Review+2\n\nThank you for the quick fix.\nAs per [1] \u0027killed\u0027 is definitely part of V2 image flow.\n\n[1] https://docs.openstack.org/glance/wallaby/user/statuses.html","accounts_in_message":[],"_revision_number":1},{"id":"ea05355f0c4b8aeefafac8b236166687de740344","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-08 19:25:25.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/3b9392f9056e4a94b076b985ff6a832b : SUCCESS in 1h 02m 44s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4ef3a8e442fa4490be107474877cd7cd : SUCCESS in 1h 30m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a84e63a647be4d45859e3b55fb739600 : SUCCESS in 5m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0e0fea8b3324454aa3a6f3dc2a65ac10 : SUCCESS in 6m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2e5933da3454469c982e36de84aa221a : SUCCESS in 6m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/aa3f6fd302f3435c9f1b359c9a0e9936 : SUCCESS in 6m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac2cfd5e962042768fdd3ea081f0d849 : SUCCESS in 7m 09s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0b0e2abb64a74d9eadeb98b3ba17e65b : SUCCESS in 9m 01s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/baf3968ad0464681a9a77b61bf812439 : SUCCESS in 8m 35s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/14d226ce04ca4f4e8b556dfe013a8104 : SUCCESS in 8m 44s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/4a5991fb20374d44ae02fde623dc2ea1 : SUCCESS in 5m 37s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/7b2f69c293a34e65bdbe3e7aaf761a3b : SUCCESS in 1h 08m 52s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/65693f209b2343378070a4ddec5a156e : SUCCESS in 1h 07m 31s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/d69daa001fba4305a279e0b7055dfd9c : SUCCESS in 1h 04m 25s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/45e3ae42dbb9478a92586378aa428e66 : SUCCESS in 1h 18m 49s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f3edf64b5c9240b89c63323bb643a855 : SUCCESS in 49m 13s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/15891f3fde7249619e2865fa4d892967 : FAILURE in 1h 11m 51s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/c46da729e69443b6b86da4bf582caa15 : SUCCESS in 33m 16s","accounts_in_message":[],"_revision_number":1},{"id":"8cd3a853c922b8e706ce36cb0f447deccbe54b15","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-09-08 20:44:30.000000000","message":"Patch Set 1:\n\nrecheck nova failed-to-delete-allocations known thing","accounts_in_message":[],"_revision_number":1},{"id":"95dfe029b041f22c4941c43b231049e466258284","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-08 22:39:59.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/eff13283bbf14e5e88b77318939178c5 : SUCCESS in 58m 50s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8b9aaf65a0cc48a69df33376051a2652 : SUCCESS in 1h 03m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d294e6ce509945d3a63bba89d17fc9a0 : SUCCESS in 5m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9886737c810b46bd9f5d67dc870c9ab4 : SUCCESS in 6m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6884f2718cc34d2b98d29d56eda70db3 : SUCCESS in 7m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5e6c46ef9c134d8d99cfdf9b28a02a4f : SUCCESS in 7m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/667c7f55dd484aacb05abad6b28afd0c : SUCCESS in 7m 25s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/47ad207efc0b464992d0aa9cf49f37a1 : SUCCESS in 9m 44s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/6d788f0fea674dbdbfa11dea274cbb2d : SUCCESS in 11m 02s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/9a64b70b06c44738a8037d1c6d3e7e4e : SUCCESS in 9m 12s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/aa30e9deafe74dd7b5a30ef28fa603db : SUCCESS in 6m 21s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/c425cb0760244f4c96abdb47dcf1170c : SUCCESS in 1h 12m 39s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/145f61de350f4a1ab69a6653632306ae : SUCCESS in 1h 05m 00s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/ebc2bef9b802405c872df0ebf978ad03 : FAILURE in 1h 03m 11s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/b9769110822d463f9e297e76c06d4eed : SUCCESS in 1h 11m 03s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/56398d0e7ea1401491028497f40ec8af : SUCCESS in 50m 44s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/5e9dbcb5cf69431fa874ca5af06293f9 : SUCCESS in 1h 09m 05s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/816645cf0a4042698e85ad486ca4dac1 : SUCCESS in 32m 40s","accounts_in_message":[],"_revision_number":1},{"id":"50c747b137d1b0f51f395d801801f36e7f17d5ef","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-09-08 23:31:32.000000000","message":"Patch Set 1:\n\nrecheck more nova fail :(","accounts_in_message":[],"_revision_number":1},{"id":"496da5ba6bca0b8be8af767968711a226e6f8a13","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-09 01:01:40.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/0f2e4f17f900441f86b79854da15b7fa : SUCCESS in 1h 08m 13s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/fca59c15c455450dbeaa527cd4272026 : FAILURE in 58m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/76ea9ebab477424698f3b4300b18558f : SUCCESS in 5m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fd76938ea8994abbb0c21e82b1b37a1b : SUCCESS in 5m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cc400a61252e4c3e8c8cdd7924bf201a : SUCCESS in 6m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/baf4b8b15c3c410da0d4a7d2a5d7865c : SUCCESS in 7m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2da7d433dd7149c8a2b6e62ea0179e8a : SUCCESS in 6m 41s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e1e717a4ae894573a35b621f50a364f4 : SUCCESS in 8m 30s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/26abd6d69e8b4bf1a08692c954538dd3 : SUCCESS in 10m 51s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/e4400b0a36544602a49b2b5a8e7efaeb : SUCCESS in 14m 08s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/5ca760820773447c98b0cd02c7f9785d : SUCCESS in 5m 27s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/b5a4ab8f7a264838bf0148e69227b4a6 : SUCCESS in 1h 10m 09s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/acb9c7b3327c4c9a988b70cf980a8b88 : SUCCESS in 58m 36s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/a8ad713f960e460f9c00bd791e2c370d : SUCCESS in 1h 05m 25s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/d1675979c5744211976e4fae08b1128b : SUCCESS in 1h 22m 56s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4e0a383e07614917bcf2a43e3ef21848 : SUCCESS in 44m 57s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/6950cb192e08496b992188740597be91 : SUCCESS in 1h 07m 52s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/fc2a829e08574ddd8254e7b12ff6870a : SUCCESS in 42m 13s","accounts_in_message":[],"_revision_number":1},{"id":"6f58ffdb67730cd2b71ac5e6e11cd79fe74af48d","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-09-09 05:29:13.000000000","message":"Patch Set 1:\n\nrecheck another nova failure","accounts_in_message":[],"_revision_number":1},{"id":"ec6cc2a7520d0f3a4c41be03349eb16f1d88469d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-09 07:08:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/1bb17becf4ad4eeca11ef364e96bd92f : SUCCESS in 1h 16m 35s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f26c38b6a52349d993e4d1d0fb2a8a4e : SUCCESS in 1h 14m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da152caee8b242a498828e13b58cb96f : SUCCESS in 6m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c73bb7cc12c84bbc95e113b3ac365d3e : SUCCESS in 5m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4c1d6f7fca2347f7b1ce821fa156db5a : SUCCESS in 6m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5683da4e4be04aa0be213f9533e550ba : SUCCESS in 6m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1505b785b9d48b891f73134011ea619 : SUCCESS in 7m 30s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/255fd052adaf4645838b5cbe4a1a9593 : SUCCESS in 9m 27s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/871b8d3111fb4571bc08788d411db960 : SUCCESS in 8m 37s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/1439965a4f5e42859c91c2dec43c766d : SUCCESS in 10m 07s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/67d050b70816461c9cd8503b76cd7898 : SUCCESS in 5m 57s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/e15bbcdf13be474db2001d371323ae90 : SUCCESS in 1h 11m 12s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/cc827e36c07b469faf960318591b856e : SUCCESS in 1h 08m 55s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/f7a38b19426c4290bd61f1e5e8e502f8 : SUCCESS in 1h 10m 48s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/33387af78137409dbfdbf8184292af60 : SUCCESS in 1h 24m 25s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6bc883d4a0184f9fa44d03f2baafdec7 : SUCCESS in 52m 34s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/5c40e68073ae407184e2f298f993f57a : SUCCESS in 1h 15m 43s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/921839e20aa84d83b7a7c0aa67f1c720 : SUCCESS in 34m 16s","accounts_in_message":[],"_revision_number":1},{"id":"6d5c168915e34bae07d2e213bef417dd947bb424","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2021-09-10 14:40:24.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nNice one!","accounts_in_message":[],"_revision_number":1},{"id":"656dd79e0fd310d5ef926dd32b101893dd2be364","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-10 15:01:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"03af815e0f0b8821517a1d2fca4cb6efbd625d0c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-10 16:26:27.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/4aec99d962bb447983fcc7646f56f8ba : SUCCESS in 1h 09m 49s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/42c73208141c4deab12f54c3577c29a6 : SUCCESS in 1h 15m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42408f16c8e948febfcfb64ce936c66b : SUCCESS in 7m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/35bfa3c5ccaf48a1be278180011507fc : SUCCESS in 8m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e43a87853ecf4b30ae65a8109d7b2a75 : SUCCESS in 9m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b675acd4c47414499fb409fd58ca2a6 : SUCCESS in 10m 25s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/367e37a82e5b4a93abcbce9570b126d4 : SUCCESS in 12m 42s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/1878edc12933441490455c530d8f0a21 : SUCCESS in 13m 29s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/a460b0b0ad4540dd9173b452947e1043 : SUCCESS in 1h 15m 41s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/832cc0178d614cad97461ca9a3f0d032 : SUCCESS in 1h 08m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5a1c796f68fb4ff9a461de9ca35c83a6 : SUCCESS in 58m 20s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/1c3736c89bec41f18b7d0ff830efb836 : SUCCESS in 1h 15m 43s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/9e37012af28a4dd29ffba370204f82e8 : SUCCESS in 33m 59s","accounts_in_message":[],"_revision_number":1},{"id":"8e352caf0d59bbd2f5f311d5c749ee592140e373","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-10 16:26:27.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"87570857c2d1c4abde81670e0cc0433fed4e6726","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-10 16:30:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/778b4c84c2eb47adace76c28b4110fd2 : SUCCESS in 2m 54s","accounts_in_message":[],"_revision_number":1},{"id":"fdaebec2fe8691988e70a06102c87d2944e0c2fd","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-09-12 03:42:06.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/807932/807932/1/check-not-vote/ext-glance-zuul/b66a68c : FAILURE in 11m 35s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4ee400accae414b9d4d08a3388c2e23be1b00bd4","revisions":{"4ee400accae414b9d4d08a3388c2e23be1b00bd4":{"kind":"REWORK","_number":1,"created":"2021-09-08 17:49:41.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/32/807932/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/32/807932/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/32/807932/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/32/807932/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/32/807932/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/32/807932/1"}}},"commit":{"parents":[{"commit":"929190a6afd0a1593e62abb8024c339713b52b98","subject":"Merge \"Add doc support for delete-from-store API\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/929190a6afd0a1593e62abb8024c339713b52b98"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-09-08 18:50:17.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-09-08 18:50:17.000000000","tz":-420},"subject":"Make signature verification go back to \u0027queued\u0027","message":"Make signature verification go back to \u0027queued\u0027\n\nThe image state diagram shows that for v2, any image upload failure\nresults in the image going back to \u0027queued\u0027 state. The old v1 path\nused the \u0027killed\u0027 state, which is no longer value in image list\noperations for v2. The signature verification error was sending the\nimage back to killed state still, which results in the user losing\nvisibility to the image entirely.\n\nThis removes the one last way we can get an image into killed state,\nbut there may still be images in the DB that have been orphaned in\nthis way, so some cleanup routine will be necesssary.\n\nChange-Id: I9330cc42900dd019e9132c1f4e49eb9eff4f3fc9\nRelated-Bug: #1924612\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4ee400accae414b9d4d08a3388c2e23be1b00bd4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4ee400accae414b9d4d08a3388c2e23be1b00bd4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
