)]}'
{"id":"openstack%2Fglance~809906","triplet_id":"openstack%2Fglance~master~Ib2e267eb1ac5ef4da711b6f5bccc5eccfb4b6954","project":"openstack/glance","branch":"master","attention_set":{},"removed_from_attention_set":{"32291":{"account":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"last_update":"2022-07-07 15:01:32.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib2e267eb1ac5ef4da711b6f5bccc5eccfb4b6954","subject":"Drop support for SQLite \u003c 3.7","status":"MERGED","created":"2021-09-18 02:16:46.000000000","updated":"2022-07-07 15:02:54.000000000","submitted":"2022-07-07 15:01:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"809906","meta_rev_id":"b2096ce56bf4f9ba9e4caef6a422a3fdeac7b94c","_number":809906,"virtual_id_number":809906,"owner":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-07 15:01:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":2,"date":"2021-09-20 14:16:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":1,"date":"2022-07-07 14:14:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-18 03:06:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-20 14:16:46.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"}],"messages":[{"id":"ef369feec97212afb86184f2936fe85c93ac7fea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"date":"2021-09-18 02:16:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2cc5c3700c10104976cc7617515f55ad63ef62d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-18 03:06:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a926652c5cb41fdad39f910481d9e70 : SUCCESS in 6m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b6885c406c1343b2b795637959761d26 : SUCCESS in 10m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e92c1c6147c143fd9d5d5b5318e9e6ff : SUCCESS in 11m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/515217be41e54ef492fc189b6e4d0804 : SUCCESS in 6m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6987f26dc1514f55a9bbb6ca4fc5daf2 : SUCCESS in 7m 14s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/127ab49a8f0f447eb6fc524a02353a77 : SUCCESS in 9m 39s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/7d463451a6134969884efa579df1e9bf : SUCCESS in 8m 34s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/db742217dc5e4bb0972c2388af39cb30 : SUCCESS in 8m 21s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/bef6c6c022cd4649b45bc5f3b3b0bba8 : SUCCESS in 5m 29s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/2d0184a8d1164fe5bf665a6a8f084e75 : SUCCESS in 35m 23s","accounts_in_message":[],"_revision_number":1},{"id":"498ce995524c4deca9131e01577c038c599705c4","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2021-09-20 14:16:46.000000000","message":"Patch Set 1: Code-Review+2\n\nFrom what I understand, Debian provides at least 3.16 (in stretch). SQLite 3.7 was released in 2010 (https://www.sqlite.org/changes.html), so it should be available everywhere.","accounts_in_message":[],"_revision_number":1},{"id":"f7c73de150ab82b68bfffb545f841689f9702a16","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2022-07-07 14:14:57.000000000","message":"Patch Set 1: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2cba0edd13f9ae647d92e4b3107a82382610d544","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-07 14:15:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0d8ff02973cd63098df85c84789596b7ad743edb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-07 15:01:32.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d5cabf012a644dfa890034ad6d32e4bd\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/041f1b9450f94c35a5d373ea7d27ea3b : SUCCESS in 44m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/edf2d463c914466b8156ab3cc5934376 : SUCCESS in 6m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f40083c16f204f5aac4d000b77a95433 : SUCCESS in 4m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d782ac5726714ea2b6b1b85d67a711ad : SUCCESS in 6m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57103f03d15243fd8bce003aad391d72 : SUCCESS in 7m 10s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/f8129f68c8dc4981b18504e5c3778c08 : SUCCESS in 9m 28s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/3c3331690cee4999bd073dcefd89e9a5 : SUCCESS in 9m 43s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/51a08a4591bc4088923ad47ab2e029f1 : SUCCESS in 30m 10s","accounts_in_message":[],"_revision_number":1},{"id":"b4a997b05167b2a2913ad0bf2e82698d039cbee6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-07 15:01:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"b2096ce56bf4f9ba9e4caef6a422a3fdeac7b94c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-07 15:02:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eb2da48d938c4290a2f78c915937c222\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3852242c68f34521b40f51954c4a8686 : SUCCESS in 58s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3d204f5f138da792d78ee20729fac4f04777a7b7","revisions":{"3d204f5f138da792d78ee20729fac4f04777a7b7":{"kind":"REWORK","_number":1,"created":"2021-09-18 02:16:46.000000000","uploader":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"ref":"refs/changes/06/809906/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/06/809906/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/06/809906/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/06/809906/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/06/809906/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/06/809906/1"}}},"commit":{"parents":[{"commit":"489298849157f76512358185edd4be9e3d8c07df","subject":"Bump oslo.log to version 4.5.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/489298849157f76512358185edd4be9e3d8c07df"}]}],"author":{"name":"wu.shiming","email":"wushiming@yovole.com","date":"2021-09-18 02:16:11.000000000","tz":480},"committer":{"name":"wu.shiming","email":"wushiming@yovole.com","date":"2021-09-18 02:16:11.000000000","tz":480},"subject":"Drop support for SQLite \u003c 3.7","message":"Drop support for SQLite \u003c 3.7\n\nBoth Ubuntu 18.04 [1] and CentOS 8 [2] provide sufficiently new versions\nof SQLite that make this check unnecessary now.\n\n[1] https://packages.ubuntu.com/bionic/sqlite3\n[2] http://mirror.centos.org/centos/7/os/x86_64/Packages/\n\nChange-Id: Ib2e267eb1ac5ef4da711b6f5bccc5eccfb4b6954\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3d204f5f138da792d78ee20729fac4f04777a7b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3d204f5f138da792d78ee20729fac4f04777a7b7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
