)]}'
{"id":"openstack%2Fglance~811467","triplet_id":"openstack%2Fglance~master~Iba9b5c660628bae77708911ef0459992e01c93c4","project":"openstack/glance","branch":"master","topic":"bug/1870205","attention_set":{},"removed_from_attention_set":{"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2021-10-13 17:31:31.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iba9b5c660628bae77708911ef0459992e01c93c4","subject":"Add libpcre3-dev/pcre-devel to bindep.txt","status":"MERGED","created":"2021-09-28 18:12:51.000000000","updated":"2021-10-13 17:33:23.000000000","submitted":"2021-10-13 17:31:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"811467-bug/1870205","meta_rev_id":"ace6c519bdb9a89a1c55ea34ad453114401e5016","_number":811467,"virtual_id_number":811467,"owner":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-10-13 17:31:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2021-10-13 06:51:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2021-10-13 16:03:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]}],"REVIEWER":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-28 18:59:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-30 03:32:48.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-10-13 06:51:29.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"}],"messages":[{"id":"6c224eae529cdc56fd1e788fa02917aaa0c880a7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2021-09-28 18:12:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ccd5ae7291592e01b19ac79cbc4b6e08a8afbf4b","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2021-09-28 18:22:24.000000000","message":"Patch Set 1:\n\nI\u0027m not sure whether we should have a specific test for this, like \"pcre-devel [platform:redhat docs]\". What do you think?","accounts_in_message":[],"_revision_number":1},{"id":"48bbd000b0aa8f29a775b742d0430ccdcc357d40","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-28 18:59:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/bb8c53d2e00e406dabe45f2250a8b39c : SUCCESS in 6m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/b0c2bd58d6354a8a88b2acc02969b06b : FAILURE in 24m 57s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6fb88072afc9428ab311016876ca84cd : FAILURE in 29m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/317f236c532e426e81eee38393d43601 : SUCCESS in 5m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b821a97bb4c24ef2ab6070ea88b7c231 : SUCCESS in 6m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d14a5388c0ef4d05b3237537dd6cbf1f : SUCCESS in 7m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1fef890a1ce64ac2bcb29cf505402c59 : SUCCESS in 7m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cadf7634b3894f6f902c569a59247954 : SUCCESS in 7m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3585b8f509d3406e86fd3c4ef737cc84 : SUCCESS in 8m 05s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/621298b1f377404e86bdf981a5685e97 : SUCCESS in 9m 28s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/021df9bc5e2a4f2fa74816afcbd9b692 : SUCCESS in 9m 08s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/3a6cd149ea514ba280b5a416cdcafebf : SUCCESS in 10m 34s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/0502cff0367342969f7cfaab0ec66f04 : SUCCESS in 6m 03s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/ca39a524d6bc4fc1ae6e7144d32a4761 : FAILURE in 29m 34s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/4ca30e5abfd74a0a8cf51a3d1ead96aa : FAILURE in 26m 20s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/811774a4b2a84dfa892a4b6139ea437b : FAILURE in 32m 37s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/c5bf332c6ed54e0aa710471bdeb70654 : FAILURE in 25m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1b55854e70d44bb18e11484b13fa86e9 : FAILURE in 24m 22s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/43fcafea486946dc86a640a477701675 : FAILURE in 36m 27s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/05448fe5ae5c4e72a324df3911de45b1 : FAILURE in 39m 42s","accounts_in_message":[],"_revision_number":1},{"id":"beae1675a3b97466b3735d90608f0856c063303d","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-09-30 03:32:48.000000000","message":"Patch Set 1:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/811467/811467/1/check-not-vote/ext-glance-zuul/1aadbd6 : SUCCESS in 33m 33s","accounts_in_message":[],"_revision_number":1},{"id":"67c25e5c8d338dcb51c324a59b75db5300eeede1","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2021-10-05 16:45:43.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"536180713c220702cc53ede3f6155bfb7c4f2075","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-05 18:23:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/04972fe304224d14a8c4b610221c98e8 : SUCCESS in 8m 50s\n- grenade https://zuul.opendev.org/t/openstack/build/ea7165f079cb4b8faf4a27c1456daffa : SUCCESS in 48m 26s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ba03d1abfb024fc6b3286de974a28b2b : SUCCESS in 58m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef26b4a5064446ed9dcb616109666c29 : SUCCESS in 7m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/15e4f086e62c4ac3969add3885bf75f2 : SUCCESS in 6m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a8c40308aca649c0972ccb2a508fcfad : SUCCESS in 7m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ac5573f95ad244fbaab4ca59552616f7 : SUCCESS in 7m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/121f0d25de304faa8e34cecc5d135007 : SUCCESS in 8m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/790a8e951875433b9c5886bb24818e45 : SUCCESS in 7m 05s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/757f53d649ee4eb4a6e9b4d0d87c262b : SUCCESS in 9m 57s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/4c5381e0eb654214adf5a0b822365231 : SUCCESS in 10m 03s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/3c2ad36bfecf45aca222464ac1235225 : SUCCESS in 9m 13s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/fedc69f6a45948f492895465ff8bc125 : SUCCESS in 6m 20s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/97386dd23ac54d42b23e9b21020da77c : SUCCESS in 1h 20m 25s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/2b1709bd0ea145b0928486358c4da8d2 : SUCCESS in 1h 16m 42s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/4fe17f47e1034fcfb900cf8e06e9e071 : SUCCESS in 1h 05m 58s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/cc21bca222a443778677384bf9224988 : SUCCESS in 1h 29m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e3c5035ea19f43d59d960f9ee088c68f : SUCCESS in 48m 22s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/ff1c356671464799bfd3e2cf2cdcaf32 : SUCCESS in 1h 16m 03s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/41c9a1034f2046ef8b6a0cac1e8bc43e : SUCCESS in 36m 45s","accounts_in_message":[],"_revision_number":1},{"id":"e45a01d825e5af526f58fcec3ac29468850e9685","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-10-13 06:51:29.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e1309289c1dd1a097bc0c4847440682d980abb18","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-10-13 16:03:25.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"79d60f5c1d4c00fb64bbccc954c0803b0b18a036","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-13 16:05:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0a4924280b84d34413fcc2f5c58505e1e9ba36ec","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-13 17:31:31.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/1da927d8abb44724a54509d2220b678d : SUCCESS in 7m 50s\n- grenade https://zuul.opendev.org/t/openstack/build/092a794b880b44239a1a5917fb66ab4d : SUCCESS in 1h 12m 53s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9022c52b3ede4305b7ab3bb1e43ddc8a : SUCCESS in 1h 10m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f8167ff2583421f9f346e05cde09a1a : SUCCESS in 6m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/11b44f2e889c4e5eaf28524cf1e08345 : SUCCESS in 6m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0a337352b07245ef888b05ce1a45ef57 : SUCCESS in 7m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/854935b4170440ea994baa18f825c05f : SUCCESS in 8m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9e3794dae91946fb82bcbf51fea8c61a : SUCCESS in 6m 36s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6e0f5742e8534983b1c71911b6d3fe50 : SUCCESS in 9m 58s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/5747f2933ca8485588ffe966524ec7f3 : SUCCESS in 10m 23s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/5c99bdc5b9ea4ced97cb46b65332e7b5 : SUCCESS in 1h 09m 37s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/59204c0e983f439ab99f8230a878ee03 : SUCCESS in 1h 00m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/01abe2b434cc4998a16b967cec7c2979 : SUCCESS in 55m 53s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/e84e69ad76954d508b13532288536b5e : SUCCESS in 1h 21m 17s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/e1a8e624880f4745bbb002aae511fb14 : SUCCESS in 36m 28s","accounts_in_message":[],"_revision_number":1},{"id":"6f7d3940136b8460194a7d272b9e40f9b5a64ed3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-13 17:31:31.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"ace6c519bdb9a89a1c55ea34ad453114401e5016","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-13 17:33:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/59475f5d34f74994a8d13709b200d4bf : SUCCESS in 1m 00s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ac73d18cd664b70b6b0d014064f532c : SUCCESS in 1m 12s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f98b36b1750c448a89cd9354a06cc97e : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6bfe0a97370a12862d8243219fcea6a99e595cf9","revisions":{"6bfe0a97370a12862d8243219fcea6a99e595cf9":{"kind":"REWORK","_number":1,"created":"2021-09-28 18:12:51.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/67/811467/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/67/811467/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/67/811467/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/67/811467/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/67/811467/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/67/811467/1"}}},"commit":{"parents":[{"commit":"c19be4cab7fdfe13dcf9c56324e96f871b7c6f9b","subject":"Merge \"Add missing forbidden to not found case for GET namespace API\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c19be4cab7fdfe13dcf9c56324e96f871b7c6f9b"}]}],"author":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2021-09-28 18:08:59.000000000","tz":120},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2021-09-28 18:11:25.000000000","tz":120},"subject":"Add libpcre3-dev/pcre-devel to bindep.txt","message":"Add libpcre3-dev/pcre-devel to bindep.txt\n\nThis is required to run \"tox -edocs\".\n\nChange-Id: Iba9b5c660628bae77708911ef0459992e01c93c4\nCloses-Bug: #1870205\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6bfe0a97370a12862d8243219fcea6a99e595cf9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6bfe0a97370a12862d8243219fcea6a99e595cf9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
