)]}'
{"id":"openstack%2Fglance~813406","triplet_id":"openstack%2Fglance~master~If488ae4299a6cab521eec9c5e9ce4f29f94057e1","project":"openstack/glance","branch":"master","topic":"bug/1946457","attention_set":{},"removed_from_attention_set":{"34120":{"account":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"last_update":"2021-10-13 11:41:34.000000000","reason":"Change was submitted"},"4393":{"account":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"last_update":"2021-10-13 11:41:34.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If488ae4299a6cab521eec9c5e9ce4f29f94057e1","subject":"Update qcow test to use qcow2 instead","status":"MERGED","created":"2021-10-11 10:22:33.000000000","updated":"2021-10-13 11:46:57.000000000","submitted":"2021-10-13 11:41:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"813406-bug/1946457","meta_rev_id":"3baafe41ba261dbd8b47be9625a58a88beb545b5","_number":813406,"virtual_id_number":813406,"owner":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-10-13 11:41:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2021-10-13 11:11:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2021-10-13 11:11:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-11 10:54:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-11 14:02:26.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2021-10-13 06:45:06.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"CC"},{"updated":"2021-10-13 11:11:06.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"}],"messages":[{"id":"90dae42b9bcedb63717de87ff31544fbac989d84","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"date":"2021-10-11 10:22:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"49867486d32393d6a0bf76af7fd12b14f0c4a3af","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-11 10:54:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/44d179340b0b492486b6fd8c9b88cc31 : SUCCESS in 5m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a55df0958a2c483bafe330235d5e85c8 : SUCCESS in 6m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0c171eaa285f4dc1a7b3739d29aa2194 : SUCCESS in 6m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c21ef1d32882485d9a345d8ffc2412fb : SUCCESS in 7m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc738c93bb9b4e668009e4bedadb6437 : SUCCESS in 7m 33s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/88e9fd27c6564437a5c63c0c61b82307 : SUCCESS in 8m 48s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/cfb7231d2c5749b5b935b2dee3bd24b1 : SUCCESS in 9m 45s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/37af38ca01ac47e39f0b194e18fad184 : SUCCESS in 10m 03s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/6cb22445efb74370972531014d50df32 : SUCCESS in 5m 47s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/5c5ff8935c164d5499376f6155ae4410 : SUCCESS in 23m 59s","accounts_in_message":[],"_revision_number":1},{"id":"386a4c2444d24cc82b08cf37db8871fcbd73b531","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-10-11 14:02:26.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"50a1a74666d72c3235a2debb96379adceb56e83b","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-10-11 14:05:32.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"0766dd98f6fb08193a153dc1abe7ce1ee656a8c9","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-10-13 06:45:06.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"442e1b62b6f1ed8ba5cf803917fab4aefde9694b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"date":"2021-10-13 10:31:27.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"036474edb121226b8a22033908f5c208a7ba9949","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-13 11:09:06.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/efb3d43d78464bb49d59378267fc4f8e : SUCCESS in 5m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6b83ac6054914ce891aacfa67fa7c72c : SUCCESS in 6m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/92d96f81ac1c4ae0803db1d5e6fa92a3 : SUCCESS in 6m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/69523d9251b14022b0b97b8bf6b5f44f : SUCCESS in 7m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41debc6fe18e474f99457bac18518924 : SUCCESS in 7m 13s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0b236fa0a1a645079ae3ecb6d6e4ca92 : SUCCESS in 8m 41s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/11fd589bcb4649cabda80ed223b6826a : SUCCESS in 8m 30s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/d70b1adbb186414c8c7b9ec457e896ec : SUCCESS in 8m 54s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/6b8470248c5940c39dcb4d50ce2e5040 : SUCCESS in 6m 21s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/524a608aef544860899e6b00e140828d : SUCCESS in 33m 21s","accounts_in_message":[],"_revision_number":2},{"id":"c238cab3e5c2f6dfbd97ba9653dc095c33613cf1","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-10-13 11:11:06.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8db54f0b790e16461a3df738bf301747af2156ff","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-13 11:11:48.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b8bb56eb021c8871316ead43c25b972e6e35c8b7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-13 11:41:34.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/861819cdd7e34efaba1993015dc21482 : SUCCESS in 6m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/db5cfacfb23c4e789e6badbaf19a72b5 : SUCCESS in 7m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/caf5f079f6a843dba423641f7798f5fb : SUCCESS in 6m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/edc0ab68875048d1b2039888a9e67b22 : SUCCESS in 6m 59s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/dfc10f7d488349f18282f06dad7a00eb : SUCCESS in 9m 17s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/321fdb8024fc4aa791dfebececcf6b70 : SUCCESS in 10m 08s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/bf3e4928244e40bcbdfa891542e0ad07 : SUCCESS in 28m 21s","accounts_in_message":[],"_revision_number":2},{"id":"af846be1da841bb85d399e9c126a4c8b264ab1b4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-13 11:41:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"3baafe41ba261dbd8b47be9625a58a88beb545b5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-13 11:46:57.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2bbc5ea76bf04631a9601ea8f0b06fe8 : SUCCESS in 1m 11s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"947e541fc0f7c0fc2b759898ab8b9401d152f3d0","revisions":{"8b49cf63c5cb73de7ec1a91cf10172a835f8fc12":{"kind":"REWORK","_number":1,"created":"2021-10-11 10:22:33.000000000","uploader":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"ref":"refs/changes/06/813406/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/06/813406/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/06/813406/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/06/813406/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/06/813406/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/06/813406/1"}}},"commit":{"parents":[{"commit":"5e288201950d032178b20843a87609d47b6636fb","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5e288201950d032178b20843a87609d47b6636fb"}]}],"author":{"name":"Andre Aranha","email":"afariasa@redhat.com","date":"2021-10-11 10:20:38.000000000","tz":0},"committer":{"name":"Andre Aranha","email":"afariasa@redhat.com","date":"2021-10-11 10:22:32.000000000","tz":0},"subject":"Update qcow test to use qcow2 instead","message":"Update qcow test to use qcow2 instead\n\nqcow format is not supported since Train, but we still have tests that\nuses qcow instead of a supported format.\n\nFixes bug 1946457\n\nChange-Id: If488ae4299a6cab521eec9c5e9ce4f29f94057e1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8b49cf63c5cb73de7ec1a91cf10172a835f8fc12"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8b49cf63c5cb73de7ec1a91cf10172a835f8fc12"}]},"branch":"refs/heads/master"},"947e541fc0f7c0fc2b759898ab8b9401d152f3d0":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-10-13 10:31:27.000000000","uploader":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"ref":"refs/changes/06/813406/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/06/813406/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/06/813406/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/06/813406/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/06/813406/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/06/813406/2"}}},"commit":{"parents":[{"commit":"5e288201950d032178b20843a87609d47b6636fb","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5e288201950d032178b20843a87609d47b6636fb"}]}],"author":{"name":"Andre Aranha","email":"afariasa@redhat.com","date":"2021-10-11 10:20:38.000000000","tz":0},"committer":{"name":"Andre Aranha","email":"afariasa@redhat.com","date":"2021-10-13 10:31:27.000000000","tz":0},"subject":"Update qcow test to use qcow2 instead","message":"Update qcow test to use qcow2 instead\n\nqemu-img package for Centos 8 doesn\u0027t support qcow disk format anymore,\ndue to this, tests using qcow are failing with\n`Unknown file format \u0027qcow\u0027`.\nThis patch updates the qcow tests for qcow2 instead.\n\nCloses-Bug: #1946457\n\nChange-Id: If488ae4299a6cab521eec9c5e9ce4f29f94057e1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/947e541fc0f7c0fc2b759898ab8b9401d152f3d0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/947e541fc0f7c0fc2b759898ab8b9401d152f3d0"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
