)]}'
{"id":"openstack%2Fglance~813536","triplet_id":"openstack%2Fglance~master~I5e7776324c01e467799b1296c35b84dc0c10cce2","project":"openstack/glance","branch":"master","topic":"typos","attention_set":{},"removed_from_attention_set":{"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2021-10-19 18:06:31.000000000","reason":"Cyril Roelandt replied on the change"}},"hashtags":[],"change_id":"I5e7776324c01e467799b1296c35b84dc0c10cce2","subject":"Fix typos","status":"MERGED","created":"2021-10-11 21:04:34.000000000","updated":"2021-10-20 00:03:29.000000000","submitted":"2021-10-20 00:01:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"813536-typos","meta_rev_id":"19fcb5d1c1a6b945ee394617a7176a6dda03b612","_number":813536,"virtual_id_number":813536,"owner":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-10-20 00:01:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2021-10-13 06:42:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2021-10-13 06:42:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-11 22:22:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-13 06:42:39.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"}],"messages":[{"id":"dc18623edd51e16e8fcce0b9185dbbfb8954a8de","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2021-10-11 21:04:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b5a030fcff3a3cb51e46cfd5f35ad886aa6188ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-11 22:22:04.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/dd861bfe02914f599d78afbdc5fd797d : SUCCESS in 57m 06s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/56fbddc978a9478cb98786add655577e : SUCCESS in 1h 07m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/396e1e33e61f477fbf9861576e1066fe : FAILURE in 5m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e8b78c2334324b1a9e39eabf4d125b70 : SUCCESS in 6m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/73a1b7d5452d4df6b93b3a2843b91407 : SUCCESS in 5m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9710de2e3ab94991a45b8c2ff151c8da : SUCCESS in 6m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/efdeab1fba4348a7b87a72a2156d6a0e : SUCCESS in 6m 20s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f1f179513c014d808d842c319b955b6e : SUCCESS in 8m 46s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c77a93aa4b574d56acbef2acb1e9c284 : SUCCESS in 8m 53s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/0f1a82cc4fec4427ad9a67cc9dafb01c : SUCCESS in 9m 18s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/b7f87753bc9142b4a03a267f5f683cd2 : SUCCESS in 5m 20s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/22a3b47b07a74e18aa94820766ec1797 : SUCCESS in 1h 12m 38s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/13214878379f451ba74e95a181849b06 : SUCCESS in 1h 07m 53s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/30e6172471f844e0aef84d1c5c660fcd : SUCCESS in 1h 04m 14s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/0292b24fed5447a8a28f806ff45af4ba : FAILURE in 1h 00m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0e35837ed83c415993d0eb9def8a4650 : SUCCESS in 46m 45s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/6e882df583594327b8223069d6a3d29e : SUCCESS in 1h 06m 45s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/878bf1c34637473189a167e7f9eb2bd6 : SUCCESS in 33m 46s","accounts_in_message":[],"_revision_number":1},{"id":"c9cee770210028cabb6d6f8e84ef23fda321888a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2021-10-13 01:07:10.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"cd5510702ccfe79c06f3c5466253fa3cfdb56a51","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-13 03:28:06.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/911aba7b9e654b04905cdadc0d4852b8 : SUCCESS in 51m 36s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4dd2d434921b44e69cc4fd37c2331ee9 : SUCCESS in 58m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0d1fe8593c241f4ba5249f8e7d178ac : SUCCESS in 6m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ece71528fedb474e88cb132bb7c320b9 : SUCCESS in 6m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/06c65d2c96804a70b270ea3e75e60904 : SUCCESS in 6m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/964c4da1ae91486ab3e8169393632b8e : SUCCESS in 6m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a44862af386c4470897f7d70b0789180 : SUCCESS in 12m 21s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b9acdf3543634991853745bcb0afaa77 : SUCCESS in 13m 17s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/17b59339e2c54655992c8d0b0d6f8704 : SUCCESS in 10m 08s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/be4f9933df9641048a088bb14e08dd5a : FAILURE in 28m 49s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/c61bc8d241554e7083e1efe396e7a5f5 : SUCCESS in 5m 05s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/9b046941bf05474a86c8fd233c2656aa : TIMED_OUT in 2h 09m 55s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/56e430f2a338452c951b56909592b130 : SUCCESS in 1h 03m 28s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/7194f1c0d1ee4ba8a0e6fac610be9bb7 : SUCCESS in 1h 04m 14s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/8a50049a2c2345788445a3f4ae19ee2d : SUCCESS in 1h 21m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ecfff55374584acca98f9ce4957179ab : SUCCESS in 42m 41s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/4db6860c357a48d2845e3816a1fd762b : SUCCESS in 1h 01m 32s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/c2acf98236024618b14efba737ea7606 : SUCCESS in 28m 57s","accounts_in_message":[],"_revision_number":2},{"id":"22bf0f2974920074b5168bea832a66a25b9f5e9c","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-10-13 06:42:39.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a23ca6d340f36cee4e067c6fb5d4b9f400999639","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-13 06:43:37.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"95e470a1b380a51b932cebc27059fd96e0f6068e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-13 09:04:33.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/21fe673fe6e04c8e87efb8bd59a95795 : SUCCESS in 56m 26s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/81efad1665554e83a490a100116a7b67 : SUCCESS in 1h 17m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/17db34191bd3407b92cba4b257d9d825 : SUCCESS in 9m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ca7e115ee4094f9394e7e9632f51ce74 : SUCCESS in 7m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4be8c737542344b383f1f0c9baa73102 : SUCCESS in 8m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/88516cc0568e425c89ac5f9937da645e : SUCCESS in 10m 23s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d652b9a4925d4772b73340adf7b80af4 : SUCCESS in 11m 28s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c18ff03c0e2542cba64b5988dfa43499 : SUCCESS in 10m 55s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/faf43f2934524495ab4ebbdfcd466b22 : POST_FAILURE in 1h 25m 39s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/72affbbbb4be4a7aafac3129c6e129af : SUCCESS in 1h 01m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7ee73f1afe464fa28fe08f127197d73b : SUCCESS in 48m 03s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/924ccb5f65404be5875f26211dbde4e1 : SUCCESS in 1h 18m 58s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/6b1cd0738d3b434bbe940fbcadf44921 : POST_FAILURE in 38m 01s","accounts_in_message":[],"_revision_number":2},{"id":"0565a1f5086ebc55c295f7a276b52d8bd0f20d8a","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-10-13 09:12:39.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1da4729e0ea1f0fbfca7ea806877d30875fa680c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-13 11:24:56.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/15cdb61a1841481ba9044b6e80c6474c : SUCCESS in 56m 20s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d82fbff0fe6e48bab5793f54612ed553 : SUCCESS in 1h 07m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/769f69eb54b54dd39539380519fe45ad : SUCCESS in 5m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/278b02941af84db196c565c42948e1cd : SUCCESS in 5m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0479efb345bc43e2bd4b8b018142a19d : SUCCESS in 6m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/71a58d62ad164871b2e43f89249ebe5e : SUCCESS in 6m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35d1d7c6aa7e4383a676c16c1e89cc00 : POST_FAILURE in 7m 46s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/125f7cd0621f4b3aa534ca966c68f66e : SUCCESS in 9m 13s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/854c8f82335643f3b95cc14bd65ca63b : SUCCESS in 9m 31s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/729d02b54f45432bb4b0958c961920af : SUCCESS in 9m 22s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/cdfb2b01ff0041c3a37fde61ea154627 : SUCCESS in 8m 32s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/a8799e2cf321437eaa5eb21ba56db4ae : TIMED_OUT in 2h 05m 51s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/2ad53807cfd24e709b9de78c5e4de6d7 : SUCCESS in 1h 10m 48s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/dffa2307b6014d0ca0c03c1ee34e0bc8 : SUCCESS in 1h 05m 54s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/3d2a7cdddcee4996973ea3d3f4ba6ac0 : SUCCESS in 1h 17m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3917192da582479a88ac81cb733ee211 : SUCCESS in 59m 01s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/525f7328506349699202c74c78de705b : SUCCESS in 1h 16m 55s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/87d94b4b54554ae3b7ef962677349215 : SUCCESS in 40m 10s","accounts_in_message":[],"_revision_number":2},{"id":"a42e05eb152cdf3585a9e114b7a8679785e2c275","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2021-10-19 18:06:31.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1adb3aa60b065408ae5bd6a7fd308fb349ecd671","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-19 19:46:14.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/ba019237f80a425ea62dfcabf1957f8a : SUCCESS in 59m 05s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/02d7fb2163484e30b68dc244f305ace0 : SUCCESS in 1h 00m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4bca18d6defc4ca08abed535d4a02fab : SUCCESS in 7m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/060265ab6b234e7a833eb3b03a28896a : SUCCESS in 6m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5c918ae260a54f69aae496f2f33b7450 : SUCCESS in 5m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a8d1e9151e8e45479ddf82962305ab8f : SUCCESS in 6m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fafdb98a4a13467d8fee2b42aba48cdf : SUCCESS in 8m 40s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3f20d5dd424c4d2eadc501c58e514c55 : SUCCESS in 10m 35s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/113561c316a94157b766228b6295c083 : SUCCESS in 11m 02s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/0493beca317f4a76946b624c0135702d : SUCCESS in 10m 13s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/a184820b9c54442bbf9e5ed893f16927 : SUCCESS in 6m 03s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/8dcfaa89459b4ee0a891a43e0ae6951d : SUCCESS in 1h 15m 17s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/f861aec72f104311b97afc692156714d : SUCCESS in 1h 12m 15s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/b72b940ef4fb41f7877be12a837ba615 : SUCCESS in 1h 07m 36s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/2502f74e6c82427b9e1c6a3a6538eace : SUCCESS in 1h 14m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/718362929ffa4fd1b0e105edfd328963 : SUCCESS in 49m 36s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/aed7427693184d9bb725069c6b29cf5e : SUCCESS in 1h 32m 22s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/72fbe65887074719a4c3cd422d8b01fc : SUCCESS in 37m 49s","accounts_in_message":[],"_revision_number":2},{"id":"8e0a99749896ba854d5bba9134a17b247406114f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-19 19:50:33.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"1455ad8e0937951b45603bed0ad4399001dcdd2e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-20 00:01:42.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/b94e6f063d9b4093bcad81fb05af6c36 : SUCCESS in 56m 14s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/da44a5bef5b1440881b0036856e3e092 : SUCCESS in 1h 11m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6bf3adf74db14c74874dd1ebc3c34d45 : SUCCESS in 6m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5c7fdbde96a7444bb41dc1418bbf68cf : SUCCESS in 6m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9722de94e5d3466990ab207de0f2a3aa : SUCCESS in 6m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d27576dd57a24b86a4a34f54fc71551f : SUCCESS in 6m 58s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9a67ac0aa48f4280849651fa1f39de86 : SUCCESS in 10m 37s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/e1c74a3d5fe94f46bba203d769d46d33 : SUCCESS in 10m 36s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/686778af6b674632b9cf40fd6c1eb40c : SUCCESS in 1h 04m 29s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/b1aec71452124f31b881b8134c55495b : SUCCESS in 1h 05m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/af032d5fae6b4a5296a350b82bab060f : SUCCESS in 43m 56s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/0e5c3a80b27c4d05a6e0907a48546af5 : SUCCESS in 1h 22m 10s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/c265308c8af34144bc2212ed864ee03f : SUCCESS in 36m 11s","accounts_in_message":[],"_revision_number":2},{"id":"c7aa9934973f1e1effbf363b37c6d121360c5eac","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-20 00:01:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"19fcb5d1c1a6b945ee394617a7176a6dda03b612","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-20 00:03:29.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/446b6a6d4c2948c395cd9c0c406ec9f9 : SUCCESS in 1m 12s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0ca2f92161e663907e64c53ebf56a6740f0deac9","revisions":{"49e5bbc4fa5e027e8e1707f9a60abec2c71ab50a":{"kind":"REWORK","_number":1,"created":"2021-10-11 21:04:34.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/36/813536/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/36/813536/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/813536/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/813536/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/813536/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/36/813536/1"}}},"commit":{"parents":[{"commit":"56cc683055a84f364986507d30078ae30de2ef29","subject":"Merge \"trivial: Double quote check_str for consistency\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/56cc683055a84f364986507d30078ae30de2ef29"}]}],"author":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2021-10-07 02:11:30.000000000","tz":120},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2021-10-07 02:13:21.000000000","tz":120},"subject":"Fix typos","message":"Fix typos\n\nChange-Id: I5e7776324c01e467799b1296c35b84dc0c10cce2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/49e5bbc4fa5e027e8e1707f9a60abec2c71ab50a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/49e5bbc4fa5e027e8e1707f9a60abec2c71ab50a"}]},"branch":"refs/heads/master"},"0ca2f92161e663907e64c53ebf56a6740f0deac9":{"kind":"REWORK","_number":2,"created":"2021-10-13 01:07:10.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/36/813536/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/36/813536/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/813536/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/813536/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/813536/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/36/813536/2"}}},"commit":{"parents":[{"commit":"56cc683055a84f364986507d30078ae30de2ef29","subject":"Merge \"trivial: Double quote check_str for consistency\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/56cc683055a84f364986507d30078ae30de2ef29"}]}],"author":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2021-10-07 02:11:30.000000000","tz":120},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2021-10-13 01:02:52.000000000","tz":120},"subject":"Fix typos","message":"Fix typos\n\nChange-Id: I5e7776324c01e467799b1296c35b84dc0c10cce2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0ca2f92161e663907e64c53ebf56a6740f0deac9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0ca2f92161e663907e64c53ebf56a6740f0deac9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
