)]}'
{"id":"openstack%2Fglance~813565","triplet_id":"openstack%2Fglance~master~Ib820ac6eb1a3732bacb622e9215f22231ec6dc89","project":"openstack/glance","branch":"master","topic":"zanata/translations","hashtags":[],"change_id":"Ib820ac6eb1a3732bacb622e9215f22231ec6dc89","subject":"Imported Translations from Zanata","status":"MERGED","created":"2021-10-12 06:42:30.000000000","updated":"2021-12-01 10:14:45.000000000","submitted":"2021-11-30 05:59:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"813565-zanata/translations","meta_rev_id":"5f53b366bd47b83c219df648cbc4316aa94e79af","_number":813565,"virtual_id_number":813565,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-11-30 05:59:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2021-11-30 05:24:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2021-11-30 05:24:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]}],"REVIEWER":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-12 07:15:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-18 02:55:24.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-11-30 05:24:26.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"}],"messages":[{"id":"3d14d25f06f783ad76b4ac36ef8e1dbcf14a4826","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-10-12 06:42:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e038ad57daa0a982d3b46a7565d5e707691b5899","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-12 07:15:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3734da0dd6b34ac29286ae859998a962 : SUCCESS in 6m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/854fcc21eb134cb8a25c2dc3e65637fb : SUCCESS in 5m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b63e8af33b1044b182d5920758fcccba : SUCCESS in 5m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c1a66a8f20d840dfa63f6a1333897656 : SUCCESS in 6m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa0db0ba6eb645329fb6821ed3b5bebe : SUCCESS in 9m 01s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6f5836b2ba9a4148a31181bf7ee22715 : SUCCESS in 13m 56s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/1f887b53823c4865bf436b4daf92c245 : SUCCESS in 10m 50s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/bff3bb16c534434d95f90e5caceafca9 : SUCCESS in 9m 43s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/755a72b5afeb4cb0b6508f604005e0d9 : SUCCESS in 5m 41s","accounts_in_message":[],"_revision_number":1},{"id":"d1a03dce0b08121f2e714ac903a71ee473337a28","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-10-13 06:41:15.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8262b11a3494938f7534ce6b2b91ce73ea72a182","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-13 07:19:02.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e316540b46dd4b6aa95fdd940b4ac41c : SUCCESS in 7m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7123886cf4d448c78c0b864f670a3b58 : SUCCESS in 7m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2ce2e0fc404b404486f8ce9aab4096c1 : SUCCESS in 7m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fc8b3f7621e84bd3952712e098162838 : SUCCESS in 7m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ccbb7997f7be452c90d42f7fb96960db : SUCCESS in 8m 39s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f55601bb50474b22becd06c694e4fa31 : SUCCESS in 10m 56s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/7d51ace634c9416bb177f27009423bc6 : SUCCESS in 8m 48s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/f417153a056d43a3973b3eeee91482d5 : SUCCESS in 10m 30s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/1a1f831b88584456af5e7b63e89dd11f : SUCCESS in 5m 51s","accounts_in_message":[],"_revision_number":2},{"id":"5644d6efc09a7e46df27b5f297bf56bd3f02bb5a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-10-14 07:04:41.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"a6a63bb2c1a183b2065be2a41c12b33d43070ce2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-14 08:09:43.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f92946656ca4d18a404e4e735ab3c2b : SUCCESS in 6m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6a39e9edf5e240f9a730846e0bec2d8d : SUCCESS in 7m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2c4e19aa054a47d4bd1df2d4932f2c9e : SUCCESS in 7m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/03773672d0494d2db536fd5c7c207ddf : SUCCESS in 7m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b578528a615046408d32fd638c4b62ed : SUCCESS in 8m 30s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/28a1044c517e41269717bf7ed1ce3824 : SUCCESS in 9m 36s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/389faf723a7f49d6a922df6be562f4b7 : SUCCESS in 11m 18s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/a526ed61a1c04be3858baaac98396053 : SUCCESS in 10m 22s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/0d27be3dc6114f0981314d4eb461d96a : SUCCESS in 6m 18s","accounts_in_message":[],"_revision_number":3},{"id":"7eec482589acf0faa4a079c38a32ac2a9705a1c6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-10-20 06:59:01.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"dab69c394e837f0a111c3bb3bb7fc8a52411c872","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-20 07:58:23.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/62a3f02538d04eeb8a8ddfa46d704aeb : SUCCESS in 7m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/708fda53b8c34bdd878fc959cf0d921c : SUCCESS in 8m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/484b49a6be414aa2ae75e1fa83d00884 : SUCCESS in 7m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/532443435a804ff8a504fd7685f34588 : SUCCESS in 7m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd156e4abbd94caa8244f928aadf4ecc : SUCCESS in 9m 12s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/460de3c9c4c645b88dc92a43eb62e81a : SUCCESS in 11m 29s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/f4eba3b37c1947acb172a100da0c8201 : SUCCESS in 10m 41s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/4fe91f9f2a3446479a5a423d322fab88 : SUCCESS in 10m 25s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/ac35ceecbe1f4d72aeb3913897b3b670 : SUCCESS in 7m 23s","accounts_in_message":[],"_revision_number":4},{"id":"34006760eb184b64293ae3619314cd72c6571b48","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-11-18 02:52:57.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"78bc6fa86e72afdd523af4b84ad3ebfe27453b82","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-11-18 02:55:24.000000000","message":"Patch Set 5:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/813565/813565/5/check-not-vote/ext-glance-zuul/1056121 : FAILURE in 1m 59s","accounts_in_message":[],"_revision_number":5},{"id":"78b507631fb3a5b3eb9cc6f991fecd5bb43509bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-18 03:10:41.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/48bab637cdcf44349f7e90b520a7c5c6 : SUCCESS in 6m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/726da10a60f743c8a1f4784a6b69f62e : SUCCESS in 7m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a6442887eea44c83be4553d25bf34adc : SUCCESS in 5m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/49174a1827b14ba2b38cee5d1f560aa4 : SUCCESS in 6m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a59a91f4df584f16856ce852a7b84e71 : SUCCESS in 7m 07s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9cf72da4466342f8b1261ecea662ccda : SUCCESS in 10m 33s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/e4ba205edfa84c8f83bba202c69ff652 : SUCCESS in 8m 22s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/1fb45ef26dd04aacad1f5d49fa033e2f : SUCCESS in 12m 13s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/39ecb4c884e0420d80d79666dfa0327f : SUCCESS in 5m 04s","accounts_in_message":[],"_revision_number":5},{"id":"91ace949b9fbd5f8635d7ccca5637503ebb6b883","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-11-19 02:54:00.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"e673167ed7e2bd766551a881a2bfe5c7a9b2d584","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-19 03:18:22.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/738d789657dc4448a05bbc71594a4b4c : SUCCESS in 5m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/064e2d3836834448992cf0e7aed2cd5e : SUCCESS in 6m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/42605d8067674416915b2413135e32f5 : SUCCESS in 6m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2522617449544f1993db0926fb6c9308 : SUCCESS in 7m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/107e1e99a1194bdfb28e0fdeca0fa111 : SUCCESS in 7m 19s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8255629a9bba4110861fe03276b083bc : SUCCESS in 9m 43s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/5671b007b09a4642a1b1f699e8f4c66c : SUCCESS in 9m 07s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/80ad3c15016644bf8605b6b6a778bd3e : SUCCESS in 7m 54s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/9865cf6847be4c298de02c4b69bf2ad8 : SUCCESS in 6m 47s","accounts_in_message":[],"_revision_number":6},{"id":"e930fc8961d9ed0188af4f4661ba33e9db48dd9a","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-11-19 06:10:34.000000000","message":"Patch Set 6:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/813565/813565/6/check-not-vote/ext-glance-zuul/c11c928 : FAILURE in 1m 50s","accounts_in_message":[],"_revision_number":6},{"id":"7a5fb47683c4fda10511444ab5e1a249a42182e9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-11-24 02:43:24.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"05aafb7497c4cbae45cde5117ae54b4ca727c242","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-24 03:21:40.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ed3520f5ec747ab9d0210936ed35862 : SUCCESS in 7m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a40d32b7011e40beb408b2bc70bbacc8 : SUCCESS in 8m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e73ee3f5d6b94fd197869a5635ffecd1 : SUCCESS in 8m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/992ac3e8e78247ecaaaaf4de413ba756 : SUCCESS in 8m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e8a9a49cdb449179f3e491ca576b948 : SUCCESS in 6m 53s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8d8ebc7e21f04365831f5cd82a8b7507 : SUCCESS in 9m 34s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b1a89114c79b4b5f95a145b5f00c8186 : SUCCESS in 9m 06s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/8e90a44234c645d3ad97f70fda437b5e : SUCCESS in 11m 45s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/fca4cf693b444c2f84854758051f10fd : SUCCESS in 6m 15s","accounts_in_message":[],"_revision_number":7},{"id":"e8274a343e467e5f7c54f8df2ebb870fdd27cf44","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-11-25 02:27:20.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"fb24c85992a0f0c695c139ec799862601d2d6fcc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-25 03:06:54.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/825c576246d44fc3ba30b1aebf93634f : SUCCESS in 7m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3f36039d93154a5aa05df5710a8935d2 : SUCCESS in 6m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/eb1469d920f845b4aa103659b184375b : SUCCESS in 7m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28bd26c45be74000b79de3db37f45aa9 : SUCCESS in 8m 20s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d848cef90d7d45e097d91fb7d39a231d : SUCCESS in 10m 28s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/4caebdf4135a4d1f8378a50fb4aca07a : SUCCESS in 11m 00s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/91ca9f194fb2426397c7d53feb79aa13 : SUCCESS in 10m 09s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/1b345f087e90419b8fc3930be947340b : SUCCESS in 6m 29s","accounts_in_message":[],"_revision_number":8},{"id":"e8854ea27b27b82cc1d0258ccbb9339ec337ad60","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-11-30 05:24:26.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"abe06f2dffff12b31f4a97afcb9a039f6bf43f36","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-30 05:24:57.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"ee2d8c94075708abe90057030daabb2391ce10b5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-30 05:59:54.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f0827acd96f4a1e9ee387c5400523ad : SUCCESS in 5m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/81876bdd09b84949b859c858a9e9c00b : SUCCESS in 5m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/649229e7e501485b81749cbfe71f4f91 : SUCCESS in 6m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd631d70d66044168326482d695a25a7 : SUCCESS in 7m 04s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e8802db6bc7140f3aba6777c3f130f3a : SUCCESS in 7m 50s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b56d6bb61e3b45d589f459dcae4ce58d : SUCCESS in 7m 04s","accounts_in_message":[],"_revision_number":8},{"id":"c247ec391392bd599b182cedacd93ec25fc35f6f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-30 05:59:54.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":8},{"id":"b9ff839e333e56db07b0ba3f502309b24f508423","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-30 06:02:44.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e444c5bf9b7470b903e81d05a461c92 : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":8},{"id":"5f53b366bd47b83c219df648cbc4316aa94e79af","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-12-01 10:14:45.000000000","message":"Patch Set 8:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/813565/813565/8/check-not-vote/ext-glance-zuul/63200ee : FAILURE in 1m 53s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"89769785e137c65d5315599fc5b0eea9f611cf6d","revisions":{"3161e29044ea75fd549db179f6cb4ed151a3aafb":{"kind":"REWORK","_number":1,"created":"2021-10-12 06:42:30.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/65/813565/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/65/813565/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/65/813565/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/65/813565/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/65/813565/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/65/813565/1"}}},"commit":{"parents":[{"commit":"5e288201950d032178b20843a87609d47b6636fb","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5e288201950d032178b20843a87609d47b6636fb"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-10-12 06:42:28.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-10-12 06:42:28.000000000","tz":0},"subject":"Imported Translations from Zanata","message":"Imported Translations from Zanata\n\nFor more information about this automatic import see:\nhttps://docs.openstack.org/i18n/latest/reviewing-translation-import.html\n\nChange-Id: Ib820ac6eb1a3732bacb622e9215f22231ec6dc89\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3161e29044ea75fd549db179f6cb4ed151a3aafb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3161e29044ea75fd549db179f6cb4ed151a3aafb"}]},"branch":"refs/heads/master"},"453fcaca2c38e5fc66bf9e6a93d266a72d4047e5":{"kind":"REWORK","_number":2,"created":"2021-10-13 06:41:15.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/65/813565/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/65/813565/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/65/813565/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/65/813565/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/65/813565/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/65/813565/2"}}},"commit":{"parents":[{"commit":"46fb37112cb95c1dae8fe732cfa7df69eaee5e93","subject":"Merge \"Use single quotes in image policies\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/46fb37112cb95c1dae8fe732cfa7df69eaee5e93"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-10-13 06:41:10.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-10-13 06:41:10.000000000","tz":0},"subject":"Imported Translations from Zanata","message":"Imported Translations from Zanata\n\nFor more information about this automatic import see:\nhttps://docs.openstack.org/i18n/latest/reviewing-translation-import.html\n\nChange-Id: Ib820ac6eb1a3732bacb622e9215f22231ec6dc89\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/453fcaca2c38e5fc66bf9e6a93d266a72d4047e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/453fcaca2c38e5fc66bf9e6a93d266a72d4047e5"}]},"branch":"refs/heads/master"},"f2353ffc8cddbcc5b4c4ef77b5d8a6e006f77dd8":{"kind":"REWORK","_number":3,"created":"2021-10-14 07:04:41.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/65/813565/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/65/813565/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/65/813565/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/65/813565/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/65/813565/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/65/813565/3"}}},"commit":{"parents":[{"commit":"ff8e3d4a0a00caff4949620e1ddf3e541a522923","subject":"Merge \"Add libpcre3-dev/pcre-devel to bindep.txt\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ff8e3d4a0a00caff4949620e1ddf3e541a522923"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-10-14 07:04:39.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-10-14 07:04:39.000000000","tz":0},"subject":"Imported Translations from Zanata","message":"Imported Translations from Zanata\n\nFor more information about this automatic import see:\nhttps://docs.openstack.org/i18n/latest/reviewing-translation-import.html\n\nChange-Id: Ib820ac6eb1a3732bacb622e9215f22231ec6dc89\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f2353ffc8cddbcc5b4c4ef77b5d8a6e006f77dd8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f2353ffc8cddbcc5b4c4ef77b5d8a6e006f77dd8"}]},"branch":"refs/heads/master"},"66b8218e4f7297de479a56b3bfc12bca0b00f345":{"kind":"REWORK","_number":4,"created":"2021-10-20 06:59:01.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/65/813565/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/65/813565/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/65/813565/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/65/813565/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/65/813565/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/65/813565/4"}}},"commit":{"parents":[{"commit":"a693e066b3a713e0cc76389837593cde472cb9c8","subject":"Merge \"Fix typos\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a693e066b3a713e0cc76389837593cde472cb9c8"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-10-20 06:58:59.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-10-20 06:58:59.000000000","tz":0},"subject":"Imported Translations from Zanata","message":"Imported Translations from Zanata\n\nFor more information about this automatic import see:\nhttps://docs.openstack.org/i18n/latest/reviewing-translation-import.html\n\nChange-Id: Ib820ac6eb1a3732bacb622e9215f22231ec6dc89\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/66b8218e4f7297de479a56b3bfc12bca0b00f345"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/66b8218e4f7297de479a56b3bfc12bca0b00f345"}]},"branch":"refs/heads/master"},"4acf9424f08ff373d9bcf929e1147324e8f2c160":{"kind":"REWORK","_number":5,"created":"2021-11-18 02:52:57.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/65/813565/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/65/813565/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/65/813565/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/65/813565/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/65/813565/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/65/813565/5"}}},"commit":{"parents":[{"commit":"79f285623cba1a60156902903dc1e6b23ce11f46","subject":"Merge \"Add missing packages to test dependencies\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/79f285623cba1a60156902903dc1e6b23ce11f46"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-11-18 02:52:54.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-11-18 02:52:54.000000000","tz":0},"subject":"Imported Translations from Zanata","message":"Imported Translations from Zanata\n\nFor more information about this automatic import see:\nhttps://docs.openstack.org/i18n/latest/reviewing-translation-import.html\n\nChange-Id: Ib820ac6eb1a3732bacb622e9215f22231ec6dc89\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4acf9424f08ff373d9bcf929e1147324e8f2c160"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4acf9424f08ff373d9bcf929e1147324e8f2c160"}]},"branch":"refs/heads/master"},"a6912b5cf6ec8cd33ae8b57653fdfa86243aef2a":{"kind":"REWORK","_number":6,"created":"2021-11-19 02:54:00.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/65/813565/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/65/813565/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/65/813565/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/65/813565/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/65/813565/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/65/813565/6"}}},"commit":{"parents":[{"commit":"7231a44b18c0e3fda7915ab29599df26a60f9719","subject":"Merge \"Add \"bochs\" as a possible video model\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7231a44b18c0e3fda7915ab29599df26a60f9719"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-11-19 02:53:56.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-11-19 02:53:56.000000000","tz":0},"subject":"Imported Translations from Zanata","message":"Imported Translations from Zanata\n\nFor more information about this automatic import see:\nhttps://docs.openstack.org/i18n/latest/reviewing-translation-import.html\n\nChange-Id: Ib820ac6eb1a3732bacb622e9215f22231ec6dc89\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a6912b5cf6ec8cd33ae8b57653fdfa86243aef2a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a6912b5cf6ec8cd33ae8b57653fdfa86243aef2a"}]},"branch":"refs/heads/master"},"356eac10af619b68715009beea6e6c1c01cdcc7a":{"kind":"REWORK","_number":7,"created":"2021-11-24 02:43:24.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/65/813565/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/65/813565/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/65/813565/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/65/813565/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/65/813565/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/65/813565/7"}}},"commit":{"parents":[{"commit":"1db6e66f93535120fdd974fce7e6490063f3c863","subject":"Merge \"Remove broken tempest-full-py3-opensuse15 job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1db6e66f93535120fdd974fce7e6490063f3c863"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-11-24 02:43:22.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-11-24 02:43:22.000000000","tz":0},"subject":"Imported Translations from Zanata","message":"Imported Translations from Zanata\n\nFor more information about this automatic import see:\nhttps://docs.openstack.org/i18n/latest/reviewing-translation-import.html\n\nChange-Id: Ib820ac6eb1a3732bacb622e9215f22231ec6dc89\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/356eac10af619b68715009beea6e6c1c01cdcc7a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/356eac10af619b68715009beea6e6c1c01cdcc7a"}]},"branch":"refs/heads/master"},"89769785e137c65d5315599fc5b0eea9f611cf6d":{"kind":"REWORK","_number":8,"created":"2021-11-25 02:27:20.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/65/813565/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/65/813565/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/65/813565/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/65/813565/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/65/813565/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/65/813565/8"}}},"commit":{"parents":[{"commit":"6fd949a966fea443bb4b0b9351b3fa04662a7ecb","subject":"Refactor glance cinder tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6fd949a966fea443bb4b0b9351b3fa04662a7ecb"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-11-25 02:27:15.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-11-25 02:27:15.000000000","tz":0},"subject":"Imported Translations from Zanata","message":"Imported Translations from Zanata\n\nFor more information about this automatic import see:\nhttps://docs.openstack.org/i18n/latest/reviewing-translation-import.html\n\nChange-Id: Ib820ac6eb1a3732bacb622e9215f22231ec6dc89\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/89769785e137c65d5315599fc5b0eea9f611cf6d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/89769785e137c65d5315599fc5b0eea9f611cf6d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
