)]}'
{"id":"openstack%2Fglance~813691","triplet_id":"openstack%2Fglance~master~Ia9811d93b1c0b67a1b29a9e8ee3e0fc098c57d4a","project":"openstack/glance","branch":"master","topic":"bug/1702445","attention_set":{},"removed_from_attention_set":{"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2022-05-13 09:41:16.000000000","reason":"Abhishek Kekane replied on the change"},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2022-06-07 18:33:35.000000000","reason":"Cyril Roelandt replied on the change"},"32927":{"account":{"_account_id":32927,"name":"mitya-eremeev-2","display_name":"Mitya Eremeev","email":"mitossvyaz@mail.ru","username":"mitya-eremeev-2"},"last_update":"2022-06-07 19:47:54.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia9811d93b1c0b67a1b29a9e8ee3e0fc098c57d4a","subject":"glance-manage can purge all deleted rows","status":"MERGED","created":"2021-10-12 18:57:45.000000000","updated":"2022-06-07 19:49:14.000000000","submitted":"2022-06-07 19:47:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":23,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"813691-bug/1702445","meta_rev_id":"da62fec4493c9db03514946f7f95f65a7e52f841","_number":813691,"virtual_id_number":813691,"owner":{"_account_id":32927,"name":"mitya-eremeev-2","display_name":"Mitya Eremeev","email":"mitossvyaz@mail.ru","username":"mitya-eremeev-2"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-06-07 19:47:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2022-05-13 09:41:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2022-06-07 18:33:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":1,"date":"2022-06-07 18:33:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-12 20:26:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-13 06:47:44.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2022-05-04 15:02:48.000000000","updated_by":{"_account_id":32927,"name":"mitya-eremeev-2","display_name":"Mitya Eremeev","email":"mitossvyaz@mail.ru","username":"mitya-eremeev-2"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"}],"messages":[{"id":"16802acf46a059c1b5e0d254642d5310cf003928","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32927,"name":"mitya-eremeev-2","display_name":"Mitya Eremeev","email":"mitossvyaz@mail.ru","username":"mitya-eremeev-2"},"date":"2021-10-12 18:57:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1587f1ce376890f6f13ea464e32a55d1c7bcfd65","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-12 20:26:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/7323c6c4cfbd42f1aee650250031a18b : SUCCESS in 1h 03m 32s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d9b5ad87511842d9b03de49d83cccca7 : SUCCESS in 1h 09m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25108624f2124394bacad8021ae7b2bd : SUCCESS in 5m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4ad92f1b97874e30b0d6301a0400dec3 : SUCCESS in 6m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/895869692c6c45e48a63b34e11e3246d : SUCCESS in 6m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2bad141e672b43f1b203b53e767ce3a6 : SUCCESS in 6m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e9e0fabc1e684077b6c002606ad68cba : SUCCESS in 6m 50s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a224220c71c64df8a995c834930807ac : SUCCESS in 12m 56s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/cfe96b389f914fcba645698edfc6d8eb : SUCCESS in 9m 19s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/a8eb4aa1e74646c0a974f19278624ef0 : SUCCESS in 10m 00s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/57f61a61cf5740828574761aed2a58bc : SUCCESS in 5m 35s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/7cb5375fb80d4b20b798d1c87caaa7b4 : FAILURE in 1h 26m 27s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/a7568f451cdd4f338ed0325828ca38bc : SUCCESS in 1h 00m 48s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/3a0070899d5b465aae07f834049705e3 : SUCCESS in 1h 22m 00s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/3c7169fc67634ee5a50db0ecb8f095d0 : SUCCESS in 1h 17m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c6deb11e4a074dfebbb5bb142beecae6 : SUCCESS in 52m 26s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/460461ae2d51478e834f51f2aef0c90a : SUCCESS in 1h 11m 48s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/d2683f98faae4f979f755ed4e5d90ad4 : SUCCESS in 38m 01s","accounts_in_message":[],"_revision_number":1},{"id":"e56fdf8b266d47d2a148c8dc9e8fa9271d2a9ec4","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-10-13 06:47:44.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"1fd09d9ccd790c433fb81a8ee0c14c7a7e1bda74","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32927,"name":"mitya-eremeev-2","display_name":"Mitya Eremeev","email":"mitossvyaz@mail.ru","username":"mitya-eremeev-2"},"date":"2021-10-13 15:20:46.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7f5517b5d7e73bd04661d7427e2e0ef4386f3ec8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-13 16:45:15.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/2b38450adfea43c0a133744acb0d4ad5 : SUCCESS in 1h 02m 42s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e30977781f904ca6b1b855e3341477ba : SUCCESS in 1h 18m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/889f9ca328fe401992851f0160dadb75 : SUCCESS in 5m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5052c0dec0c649439da6e38b31b92c7b : SUCCESS in 6m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e46220d09b2c47e2b66b792a8fa88b4a : SUCCESS in 6m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/37a5a3ebd54e4fa7a112124a9203c1c5 : SUCCESS in 6m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e741f2a3b95140a6bb0bc598de267ab6 : SUCCESS in 6m 53s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ec0bab5da8374b169dff491336d1ccb9 : SUCCESS in 8m 53s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/ccc9c65c1a8c4f7ea37c9a0788bbf119 : SUCCESS in 9m 00s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/e528088304cc468ca28e5bb6ace6ebea : SUCCESS in 8m 42s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/c93e2ba832a6471d9c3496b4b5b1caea : SUCCESS in 5m 12s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/96ba03ac1ad34ea6967fbe2b4513657a : SUCCESS in 1h 08m 55s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/aaffb5853edb413d9c6b22f9227fd8ea : SUCCESS in 1h 05m 52s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/7361a34cb4d64b76941858b446c6aed0 : SUCCESS in 57m 29s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/1cef862802ac47ddbf3ddd8ae57797bb : SUCCESS in 1h 07m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/297a494b503543209c6b0283e5f4de0a : SUCCESS in 44m 55s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/fcede8a7892b48029e7c1f63193091be : SUCCESS in 1h 07m 28s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/cc960e11df764bcfb70eed055297c131 : SUCCESS in 29m 46s","accounts_in_message":[],"_revision_number":2},{"id":"46899ed2b447b58f42cf8e5bbbd0fc7929bab480","author":{"_account_id":32927,"name":"mitya-eremeev-2","display_name":"Mitya Eremeev","email":"mitossvyaz@mail.ru","username":"mitya-eremeev-2"},"date":"2021-10-13 16:50:22.000000000","message":"Patch Set 2:\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"8d503910b0f03ff111dab9ee7233571299ab21c9","author":{"_account_id":32927,"name":"mitya-eremeev-2","display_name":"Mitya Eremeev","email":"mitossvyaz@mail.ru","username":"mitya-eremeev-2"},"date":"2021-11-15 08:24:14.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a02bc60d7d7519972d557484019401ac1935e3ad","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-11-15 08:31:58.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5bc5263ed89768a314550c179802ecc068fb94f4","author":{"_account_id":32927,"name":"mitya-eremeev-2","display_name":"Mitya Eremeev","email":"mitossvyaz@mail.ru","username":"mitya-eremeev-2"},"date":"2021-11-15 12:44:44.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d177fcbf2b38c0aab92792014b5656431e949691","author":{"_account_id":32927,"name":"mitya-eremeev-2","display_name":"Mitya Eremeev","email":"mitossvyaz@mail.ru","username":"mitya-eremeev-2"},"date":"2021-11-15 15:33:25.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6f8964a9eec34f1abee9c37db3c67ceb93191b80","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32927,"name":"mitya-eremeev-2","display_name":"Mitya Eremeev","email":"mitossvyaz@mail.ru","username":"mitya-eremeev-2"},"date":"2022-05-04 12:21:12.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"05c9a71469e6d9194d82bb3701f510ca9b7386d4","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-05-04 14:16:32.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"68dfc79b0dd5ff4270b8dca159d0df8f1a5ca4b9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-04 14:20:14.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/158feee42b3d43e98c4b15685d5092af : SUCCESS in 59m 25s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/069ed60d3be149d7888219842f330e4e : SUCCESS in 1h 27m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cfb6ee4d49714c1d8b12ba780dc3c86b : SUCCESS in 57m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5be9bd2a4594121b541531aa44e848c : SUCCESS in 5m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d5bb96738ddd4d9fa2797359bb36748a : FAILURE in 5m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/41f12eceb8e94925baa2a0fb5dbaef27 : FAILURE in 5m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6af352fdf28947e290d5d70a7e8343ab : SUCCESS in 7m 06s\n- openstack-tox-functional-py36-fips https://zuul.opendev.org/t/openstack/build/ca800056d1d1482aa9de58fce0cff8fd : SUCCESS in 10m 50s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a3057c3d0d8f4c8ea220a99589ddfa1b : SUCCESS in 9m 19s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/4870ea6930384c708df57d30d76758ec : SUCCESS in 9m 08s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/a45187f76a3b4206b885415e073acf6e : SUCCESS in 4m 57s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/bb23b0cafacf4b5e8e249a1c4b97910c : SUCCESS in 1h 13m 45s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/42b64601099f49b1bc45e9854b0de242 : SUCCESS in 1h 18m 24s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/c75c48de2baa46a79ced5b9e9867d722 : SUCCESS in 1h 20m 03s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/9cdfa7651d0c421c96d55e41a8641a75 : SUCCESS in 1h 53m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1a00a31829814c828a99780fce8e0253 : SUCCESS in 53m 37s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/d2cf20e252e045998c987e3f155f6ce3 : SUCCESS in 1h 14m 20s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/dc7970c5b5c540d1bc38dfbd3bb80b26 : SUCCESS in 38m 22s","accounts_in_message":[],"_revision_number":3},{"id":"7fe65c05e5a85ead863a814bc83621df99dbd907","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32927,"name":"mitya-eremeev-2","display_name":"Mitya Eremeev","email":"mitossvyaz@mail.ru","username":"mitya-eremeev-2"},"date":"2022-05-04 14:56:56.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"96c1483ac31c5bb2801cd9bffc8f43f7aa0f9376","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-04 16:43:21.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/dd4c832a8211418eb13622a1dd9b7407 : SUCCESS in 54m 18s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/899b68f169754f35a22dd9459f318b47 : SUCCESS in 1h 16m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7712000854184f2b855c6770a833e195 : SUCCESS in 55m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f88f3055d7f4c34a6902b5671e1b280 : SUCCESS in 4m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/48ed9e7d0f8b4112a681f7f72a72f25d : SUCCESS in 5m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b1629d244f67400398cc10e9f6fda659 : SUCCESS in 5m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a59001858c6842f4b0c9555b80b72c77 : SUCCESS in 6m 10s\n- openstack-tox-functional-py36-fips https://zuul.opendev.org/t/openstack/build/6ed04d52f9694c2480aa24d280260993 : SUCCESS in 10m 54s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/172f25b735d74961aa7f4d377ec91f47 : SUCCESS in 9m 04s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/a179ea32a97b45a2998e75502eb4b514 : SUCCESS in 9m 46s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/a1e0f559308f4f34a855cc7cd0ed3fe6 : SUCCESS in 5m 01s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/05d6d5d1317247fcb709ca92598e0413 : FAILURE in 1h 12m 28s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/e0d5bb932f6b4bd5bbb9ee0102f8a578 : SUCCESS in 1h 31m 41s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/f090fdb49d024ee5b87d02f2988edb54 : SUCCESS in 1h 23m 52s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/e7e733e502bd44ebb7fe078c525cd0cc : FAILURE in 1h 39m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7eb1d190d318406394251926b0d6978e : SUCCESS in 58m 51s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/67aac48230bb49ddb48ca20f51140707 : SUCCESS in 1h 05m 14s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/f3171ac8e05b4faabc1f4c8911c3401d : SUCCESS in 38m 46s","accounts_in_message":[],"_revision_number":4},{"id":"41f6e0b4c5aaab4deeee4a8cf021e45a4135f919","author":{"_account_id":32927,"name":"mitya-eremeev-2","display_name":"Mitya Eremeev","email":"mitossvyaz@mail.ru","username":"mitya-eremeev-2"},"date":"2022-05-05 09:44:15.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"f4fe931a5175d1e84835a82d0151cdbd6dd00131","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-05 11:29:00.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/28f1d434ab3046069e3fd61070bc9238 : SUCCESS in 1h 07m 13s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0138a3e8f4cc46408117788a8165ef39 : SUCCESS in 1h 33m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/17a660cd62204d45b476c3cb925d3b16 : SUCCESS in 57m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/981c7380e61e4836b631e5d8ddc09bc2 : SUCCESS in 5m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4637e22ec4b745bfb0ee151b89009409 : SUCCESS in 6m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/21ae4d2f7a444ffc97abb241e37e850a : SUCCESS in 6m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cafb22492c9e4684a699ad83dc4ef8cd : SUCCESS in 6m 14s\n- openstack-tox-functional-py36-fips https://zuul.opendev.org/t/openstack/build/ce81946948a4441696604e4a3ef01d1e : SUCCESS in 11m 33s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/113c5f5ce4b2490bada13f8dd2965659 : SUCCESS in 8m 57s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/20559c0028404e5d8b2861b4bcef2321 : SUCCESS in 9m 36s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/01736d07df8a41a09a38b02589c88950 : SUCCESS in 4m 42s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/cad6ca96ace74ecca806361af0869008 : SUCCESS in 1h 38m 38s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/36283c1fc2f24ada99062bab8b80a529 : SUCCESS in 1h 27m 00s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/8b5aef9048ac42df8097b6f85a54329c : SUCCESS in 1h 20m 02s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/81338fec44d44a268ff5893ec7356cd4 : FAILURE in 1h 31m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4b64f58c7a184478bf3de4f091db76a4 : SUCCESS in 59m 18s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/be536a17cb574ae4a36b64eac1a19e19 : SUCCESS in 1h 15m 16s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/97c9a5a1f8064b2f99458a17cfb8f26b : SUCCESS in 40m 17s","accounts_in_message":[],"_revision_number":4},{"id":"04bcc2382bd7b32a00bbac62f7f14d309a7c10ca","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-05-05 13:54:32.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"f75f1a148d968af2dea9a26d74f6d74112c808b1","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-05-05 13:55:11.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"6c69774f5f5adbae579c80159efb9fab62c4fd81","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32927,"name":"mitya-eremeev-2","display_name":"Mitya Eremeev","email":"mitossvyaz@mail.ru","username":"mitya-eremeev-2"},"date":"2022-05-05 14:36:07.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"6ce44211121a20677b631d1ccbaec947eb2442f3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-05 16:03:29.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/d55f05b746be4302bc6b70298d4194ff : SUCCESS in 1h 03m 57s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a3d2ac29c3f94164becfcec79f078c37 : SUCCESS in 1h 20m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/826186c334d74d60b67adbc243da11b0 : SUCCESS in 51m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d0a5d3e9b0d4720adc87e4b97cb202f : SUCCESS in 4m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9b63f13878534838bcf1510128411dfa : SUCCESS in 6m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/527c7dfc71604edbbb205ccacba82200 : SUCCESS in 6m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6da00f6d0507444c85bacb36e43bccda : SUCCESS in 8m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/01eeb9d80ac448858b996a3d1cd64859 : SUCCESS in 5m 42s\n- openstack-tox-functional-py36-fips https://zuul.opendev.org/t/openstack/build/de5a347a23984641a8f4db34408565c9 : SUCCESS in 11m 27s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/58f6ee44275745ae99501f900c67cf8b : SUCCESS in 8m 58s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/c1937ebe44234f87a3aff64d542a2cea : SUCCESS in 11m 35s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/40097f7a0f1b4a4399781cd065514279 : SUCCESS in 4m 56s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/1b09bbdfe2644891873b7210eaaadc26 : FAILURE in 1h 18m 47s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/3aead8f63b21433f890246e4db43b21c : SUCCESS in 1h 18m 46s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/90b22b7213984062909154e85ab34082 : SUCCESS in 1h 17m 13s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/a214f6d1f41347af88d089d24046af1e : FAILURE in 1h 11m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/30454ee0ee1b48f8a2778991a920cac7 : SUCCESS in 57m 48s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/a02aeb05bb5e4db38d97468526ff622b : SUCCESS in 1h 02m 13s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/5ef9decceaf94133954f1318894d5b87 : SUCCESS in 39m 32s","accounts_in_message":[],"_revision_number":5},{"id":"4923a3fafe3a407126a4e02bc8925263a99ba3fb","author":{"_account_id":32927,"name":"mitya-eremeev-2","display_name":"Mitya Eremeev","email":"mitossvyaz@mail.ru","username":"mitya-eremeev-2"},"date":"2022-05-06 08:22:16.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"76a392178aad8f15d8dd73dd058974e55671d133","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-06 10:12:04.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/28bc40e6fc654ec480e2fbb37b575779 : SUCCESS in 50m 46s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/14363f00c1544fa797fc605c76d96f93 : SUCCESS in 1h 11m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8f5835fe078242c4892d1cb0abf6bc25 : SUCCESS in 43m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ab3de1ca8364a09b864f2cc9a3e9f5e : SUCCESS in 4m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1f8b139d8f2d499e965ae09cd89074e6 : SUCCESS in 5m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/090a3d9f7c254f03a304b5c3f0697cbd : SUCCESS in 5m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aef6e3dad1a24fbdbaa79ff9493f5d5f : SUCCESS in 6m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7d38f2ef56d8439f94d0a29f56353fa2 : SUCCESS in 5m 24s\n- openstack-tox-functional-py36-fips https://zuul.opendev.org/t/openstack/build/8f4b7d99bf46492dac52d027ef23d6a1 : SUCCESS in 10m 10s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/9995876910304fa2b43a0a29e65e1b7d : FAILURE in 9m 51s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/2e3cb049096047cd9f1c8ff9a2523451 : SUCCESS in 9m 00s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/8eeb74eb01f84a5e8acac5febadfa050 : SUCCESS in 5m 38s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/3d76c1d79af247c68793f921bb4b4ea5 : FAILURE in 1h 10m 32s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/ff99e9ae954b4b6eab1779eedabc9ccb : SUCCESS in 1h 24m 41s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/9bca21ffe2774cf48caeb8191771752a : SUCCESS in 1h 21m 51s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/d1e33e40746a4c61bb8ffa712810183f : SUCCESS in 1h 39m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/75b85ccc23024451908c6869bc46294a : SUCCESS in 59m 01s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/7d4cc4d7378a465faeeedac3e1b7f435 : SUCCESS in 1h 03m 00s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/6d50c50fb5b44500a567acc718c9013e : SUCCESS in 35m 08s","accounts_in_message":[],"_revision_number":5},{"id":"6d95259a48f21e05b571fb21e7a3f4871151ee35","author":{"_account_id":32927,"name":"mitya-eremeev-2","display_name":"Mitya Eremeev","email":"mitossvyaz@mail.ru","username":"mitya-eremeev-2"},"date":"2022-05-11 07:02:24.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"27ce56075d4fea3a9f35e8656e9796bd9245ae09","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-11 08:32:33.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/d798238fa603402d8a15c85c88f71c4f : SUCCESS in 1h 13m 03s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9f4af348abf74f4c8003d85a89b3d444 : SUCCESS in 1h 29m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ddf7aadce45b4390aa5fd0e6ddf114bf : SUCCESS in 52m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b48545f3e114b6c885049f850acef2d : SUCCESS in 4m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b876a1ee4a664ce494cb23d5e2abcdeb : SUCCESS in 6m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7daa354fb61548329eaa1380495fb8db : SUCCESS in 5m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59d43b800ec746e990cfe50bdad1a10d : SUCCESS in 7m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ad3fe871ae7648bab2ed6ab801e07ec0 : SUCCESS in 5m 43s\n- openstack-tox-functional-py36-fips https://zuul.opendev.org/t/openstack/build/1b07ba54240e4dbeb71c1268099ebe52 : SUCCESS in 10m 43s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ad28c635cfbd45d9a1aada4974d8f73b : SUCCESS in 9m 19s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/c26be12fc9dc4c5baf8d1d1108159bff : SUCCESS in 8m 45s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/904d63cc181e49edad7a2faae9408342 : SUCCESS in 4m 23s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/82dd334fdb174abc98564845659f2914 : SUCCESS in 1h 22m 45s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/a66beaf045f840e8bbf71c1e1169c28b : SUCCESS in 1h 16m 10s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/9b19912c0c5042edb67abc6138cd9f8f : SUCCESS in 1h 07m 21s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/95ee19a76c42481797c8da30d7dbac85 : FAILURE in 1h 17m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d61d1d2ae13b4fa7aaa17e8cefdd6a08 : SUCCESS in 45m 31s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/b11302dbc1f348a99d0b51f362e560bf : SUCCESS in 1h 08m 34s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/0801c98c3b2040cf8fa0d6b3738f51e0 : SUCCESS in 29m 57s","accounts_in_message":[],"_revision_number":5},{"id":"7c5ceb9d35fcb519442e9f7513dff4f4ea969df6","author":{"_account_id":32927,"name":"mitya-eremeev-2","display_name":"Mitya Eremeev","email":"mitossvyaz@mail.ru","username":"mitya-eremeev-2"},"date":"2022-05-12 07:06:27.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"a69f7529f4ea85e58b4c6122b118ff9feb87bfd8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-12 08:46:24.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/ead6facdd76e49e981377a920598c00a : SUCCESS in 52m 28s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4b199a673ce44d9ca32b9fe2eed482ea : SUCCESS in 1h 04m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ce98babead2249379ba5fd76d4d669e3 : SUCCESS in 55m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/20cbc273ef64419a8e647fa2821d68e6 : SUCCESS in 5m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8500584abcd04ff4b445a01c157f033e : SUCCESS in 5m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d1507d08a71a4d8b97462ef52801207b : SUCCESS in 6m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb04feea4ae44ec297f2c840e0e09d15 : SUCCESS in 7m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8830f57ecbde49ddbbb2d453b8969735 : SUCCESS in 6m 09s\n- openstack-tox-functional-py36-fips https://zuul.opendev.org/t/openstack/build/89f7cdb2af764bd59df6b00d482ddb61 : SUCCESS in 10m 01s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/33fff3268dd842f0b610eb4c0cd6d6c1 : TIMED_OUT in 31m 00s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/9b1cbaae49f141e48358695d0aad1c61 : SUCCESS in 9m 37s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/67f6d8feedc34295a94f964155d71ac2 : SUCCESS in 4m 55s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/bba2845572eb45228521f0099dd63291 : SUCCESS in 1h 18m 43s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/6b149e72ac914912a1f888ab30b2b55b : SUCCESS in 1h 28m 35s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/3b2be0f8522040d4bf2b01a3b454baab : SUCCESS in 1h 24m 58s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/8ba923bc774b46769e311b848c3acc08 : SUCCESS in 1h 38m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b93adb99a4fb499690c4969552371ffe : SUCCESS in 53m 58s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/9128e84944f04c02b6c53db2ecfd7596 : SUCCESS in 1h 22m 17s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/a0c4222998734130a8fa04b2adffe37f : SUCCESS in 31m 49s","accounts_in_message":[],"_revision_number":5},{"id":"b094ef0db2e8ff2ba018e616096d95b6591fa637","author":{"_account_id":32927,"name":"mitya-eremeev-2","display_name":"Mitya Eremeev","email":"mitossvyaz@mail.ru","username":"mitya-eremeev-2"},"date":"2022-05-13 07:05:00.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"ff98013c3a3fa8cc4fe39e0b59159616b48b5d5b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-13 08:54:10.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/c07c4261192842699158d7f116df5ede : SUCCESS in 1h 02m 55s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/cfecbb07c2b84d0699f61c259ba2e618 : SUCCESS in 1h 20m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a14be12a2d90410c9239da54a06dadd3 : SUCCESS in 55m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d6d5f19b8e3240479a647447e1a72123 : SUCCESS in 4m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/535402a0bbd44c56bcaab16dbeaf948f : SUCCESS in 5m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e6186e6cc1234fe2af64a294abe1cd85 : SUCCESS in 6m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3448f7d286045dcb49d9c5f8e9671d9 : SUCCESS in 5m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fa77ce4a80d04ee19736b6cc0de98383 : SUCCESS in 5m 41s\n- openstack-tox-functional-py36-fips https://zuul.opendev.org/t/openstack/build/64c0ff2c06a6405fa903daa2cd9eb002 : SUCCESS in 11m 49s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/87e64307e69f4b838fe52c8018e478f8 : SUCCESS in 9m 28s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/2a1b499b02d54c6297e3c01928f9c740 : SUCCESS in 8m 19s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/c7f24f7580584aa18ae8d4277109047c : SUCCESS in 4m 03s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/cdbfc5e0c2d04203b5af5a7d0fc50346 : FAILURE in 1h 08m 32s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/2c8efa2a242647aa99a18f5a27f03354 : SUCCESS in 1h 22m 05s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/0caa03d344314a8c85e810ce7a4f937a : SUCCESS in 1h 13m 42s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/a0c52e4448fe40d7a678327e7bbd94a3 : SUCCESS in 1h 44m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/23eeddeaaab64d9d964309161431bd4e : SUCCESS in 59m 46s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/bc980d642149496bb737759e8c7af8da : SUCCESS in 1h 00m 40s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/6e1471284b6744cca19c3489334f7d5e : SUCCESS in 32m 20s","accounts_in_message":[],"_revision_number":5},{"id":"b106ace94ab770e8a956272a19212e80c7014b15","author":{"_account_id":32927,"name":"mitya-eremeev-2","display_name":"Mitya Eremeev","email":"mitossvyaz@mail.ru","username":"mitya-eremeev-2"},"date":"2022-05-13 09:16:44.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"8dfadf9dca4c0f8d64dbbcd481c1fd0a7047c10c","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-05-13 09:41:16.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"73ca5b6ef358c4f29417150a3c593a999a9f3739","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2022-05-31 16:49:57.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"c19717bad200b8afe8ef91c0894d6eb001ddc881","author":{"_account_id":32927,"name":"mitya-eremeev-2","display_name":"Mitya Eremeev","email":"mitossvyaz@mail.ru","username":"mitya-eremeev-2"},"date":"2022-05-31 18:06:02.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"819e523bb8ae88f658e536339b089a3f0a9b79c3","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2022-06-07 18:33:35.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"166a96c84e6bdd139eb2de615b9118b3d0b23426","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-07 18:34:26.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"28b6bffe7b66864d39e58053141f3318fa509c83","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-07 19:47:54.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/2b0bb5f6a4d94c388cf425040058cec8 : SUCCESS in 52m 35s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c5acba6031d8420890194df72cd7c409 : SUCCESS in 1h 07m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cb7883642bbf46aca0bf6d3a35e1764a : SUCCESS in 39m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf82e266257a4a8992c1b4c9eced059b : SUCCESS in 5m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/79cc169034534db6a5791d889549023e : SUCCESS in 5m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8a48edf0ca944141a66c65512e91867f : SUCCESS in 5m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ddac3d8b78d648f4b8dc913111b1244f : SUCCESS in 6m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f72b9f2d6e554b8d9fa4c194dc4f69d2 : SUCCESS in 5m 55s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/4d84b58b585b4ffc89caf678f1273b06 : SUCCESS in 9m 00s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b7ec5f7457d944eea9b169898ba29e58 : SUCCESS in 9m 53s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/db77bc323cff426eab26797e3fe15335 : SUCCESS in 1h 12m 36s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/a1d13cd1a152489fad53aa06dbbf1129 : SUCCESS in 1h 08m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7b6e46c271184b2d937f96351092421e : SUCCESS in 46m 59s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/f7b29053dc1f46f8af56983c21c550d4 : SUCCESS in 59m 12s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/be27948e6a6d4e0faa603def890aedce : SUCCESS in 30m 32s","accounts_in_message":[],"_revision_number":5},{"id":"147f240b544abfe50e00e9a95164fe35a9f4e633","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-07 19:47:54.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"da62fec4493c9db03514946f7f95f65a7e52f841","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-07 19:49:14.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/baab3e16f30c4da6a2ed91c84564dce7 : SUCCESS in 53s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6817040eb6164699bb407224c8de8044 : SUCCESS in 57s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"360dc45680df0c968994d60f87446e1a93e2d792","revisions":{"f4e514db6ebe4e11e2ce178065a797ced4c5ff81":{"kind":"REWORK","_number":1,"created":"2021-10-12 18:57:45.000000000","uploader":{"_account_id":32927,"name":"mitya-eremeev-2","display_name":"Mitya Eremeev","email":"mitossvyaz@mail.ru","username":"mitya-eremeev-2"},"ref":"refs/changes/91/813691/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/91/813691/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/91/813691/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/91/813691/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/91/813691/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/91/813691/1"}}},"commit":{"parents":[{"commit":"5e288201950d032178b20843a87609d47b6636fb","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5e288201950d032178b20843a87609d47b6636fb"}]}],"author":{"name":"Mitya_Eremeev","email":"mitossvyaz@mail.ru","date":"2021-10-12 16:01:43.000000000","tz":180},"committer":{"name":"Mitya_Eremeev","email":"mitossvyaz@mail.ru","date":"2021-10-12 16:01:43.000000000","tz":180},"subject":"glance-manage can purge all deleted rows.","message":"glance-manage can purge all deleted rows.\n\nWhen we want to purge all deleted rows\nwe use \"glance-manage db purge\" or\n\"glance-manage purge_images_table\"\nthen we never know how many deleted rows are still in a table.\nSo we need to launch the command many times until\nthe command reports that 0 rows were deleted in every table.\nIt\u0027s inconvenient.\nThe patch adds new command-line option \"--purge_all\".\nThe option purges all deleted rows in cycles.\nBatch size in cycle equals \"--max_rows\" (100 by default).\n\nCloses-Bug: 1702445\nChange-Id: Ia9811d93b1c0b67a1b29a9e8ee3e0fc098c57d4a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f4e514db6ebe4e11e2ce178065a797ced4c5ff81"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f4e514db6ebe4e11e2ce178065a797ced4c5ff81"}]},"branch":"refs/heads/master"},"e462b251eae00212dd1bce63e5414d33e10c7bab":{"kind":"REWORK","_number":2,"created":"2021-10-13 15:20:46.000000000","uploader":{"_account_id":32927,"name":"mitya-eremeev-2","display_name":"Mitya Eremeev","email":"mitossvyaz@mail.ru","username":"mitya-eremeev-2"},"ref":"refs/changes/91/813691/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/91/813691/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/91/813691/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/91/813691/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/91/813691/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/91/813691/2"}}},"commit":{"parents":[{"commit":"5e288201950d032178b20843a87609d47b6636fb","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5e288201950d032178b20843a87609d47b6636fb"}]}],"author":{"name":"Mitya_Eremeev","email":"mitossvyaz@mail.ru","date":"2021-10-12 16:01:43.000000000","tz":180},"committer":{"name":"Mitya_Eremeev","email":"mitossvyaz@mail.ru","date":"2021-10-13 15:19:29.000000000","tz":180},"subject":"glance-manage can purge all deleted rows.","message":"glance-manage can purge all deleted rows.\n\nWhen we want to purge all deleted rows\nwe use \"glance-manage db purge\" or\n\"glance-manage purge_images_table\"\nthen we never know how many deleted rows are still in a table.\nSo we need to launch the command many times until\nthe command reports that 0 rows were deleted in every table.\nIt\u0027s inconvenient.\nThe patch adds new command-line option \"--purge_all\".\nThe option purges all deleted rows in cycles.\nBatch size in cycle equals \"--max_rows\" (100 by default).\n\nCloses-Bug: 1702445\nChange-Id: Ia9811d93b1c0b67a1b29a9e8ee3e0fc098c57d4a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e462b251eae00212dd1bce63e5414d33e10c7bab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e462b251eae00212dd1bce63e5414d33e10c7bab"}]},"branch":"refs/heads/master"},"ec834a3047838126cdb2119df6646ab7a4df662f":{"kind":"REWORK","_number":3,"created":"2022-05-04 12:21:12.000000000","uploader":{"_account_id":32927,"name":"mitya-eremeev-2","display_name":"Mitya Eremeev","email":"mitossvyaz@mail.ru","username":"mitya-eremeev-2"},"ref":"refs/changes/91/813691/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/91/813691/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/91/813691/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/91/813691/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/91/813691/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/91/813691/3"}}},"commit":{"parents":[{"commit":"8271cde1e2850ff1870b310801e3fa8c7fbbf998","subject":"[api-ref] update cache manage API info","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8271cde1e2850ff1870b310801e3fa8c7fbbf998"}]}],"author":{"name":"Mitya_Eremeev","email":"mitossvyaz@mail.ru","date":"2022-05-04 12:16:29.000000000","tz":180},"committer":{"name":"Mitya_Eremeev","email":"mitossvyaz@mail.ru","date":"2022-05-04 12:16:29.000000000","tz":180},"subject":"glance-manage can purge all deleted rows","message":"glance-manage can purge all deleted rows\n\nWhen we want to purge all deleted rows\nwe use \"glance-manage db purge\" or\n\"glance-manage purge_images_table\"\nthen we never know how many deleted rows are still in a table.\nSo we need to launch the command many times until\nthe command reports that 0 rows were deleted in every table.\nIt\u0027s inconvenient.\nThe patch introduced new valid value for \"max_rows\".\nIf \"max_rows\" equals -1 all deleted rows are purged.\n\nCloses-Bug: 1702445\nChange-Id: Ia9811d93b1c0b67a1b29a9e8ee3e0fc098c57d4a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ec834a3047838126cdb2119df6646ab7a4df662f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ec834a3047838126cdb2119df6646ab7a4df662f"}]},"branch":"refs/heads/master"},"4f968c4687029566e33bfd00e8d56b85755bb5ea":{"kind":"REWORK","_number":4,"created":"2022-05-04 14:56:56.000000000","uploader":{"_account_id":32927,"name":"mitya-eremeev-2","display_name":"Mitya Eremeev","email":"mitossvyaz@mail.ru","username":"mitya-eremeev-2"},"ref":"refs/changes/91/813691/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/91/813691/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/91/813691/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/91/813691/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/91/813691/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/91/813691/4"}}},"commit":{"parents":[{"commit":"8271cde1e2850ff1870b310801e3fa8c7fbbf998","subject":"[api-ref] update cache manage API info","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8271cde1e2850ff1870b310801e3fa8c7fbbf998"}]}],"author":{"name":"Mitya_Eremeev","email":"mitossvyaz@mail.ru","date":"2022-05-04 12:16:29.000000000","tz":180},"committer":{"name":"Mitya_Eremeev","email":"mitossvyaz@mail.ru","date":"2022-05-04 14:56:32.000000000","tz":180},"subject":"glance-manage can purge all deleted rows","message":"glance-manage can purge all deleted rows\n\nWhen we want to purge all deleted rows\nwe use \"glance-manage db purge\" or\n\"glance-manage purge_images_table\"\nthen we never know how many deleted rows are still in a table.\nSo we need to launch the command many times until\nthe command reports that 0 rows were deleted in every table.\nIt\u0027s inconvenient.\nThe patch introduced new valid value for \"max_rows\".\nIf \"max_rows\" equals -1 all deleted rows are purged.\n\nCloses-Bug: 1702445\nChange-Id: Ia9811d93b1c0b67a1b29a9e8ee3e0fc098c57d4a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4f968c4687029566e33bfd00e8d56b85755bb5ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4f968c4687029566e33bfd00e8d56b85755bb5ea"}]},"branch":"refs/heads/master"},"360dc45680df0c968994d60f87446e1a93e2d792":{"kind":"REWORK","_number":5,"created":"2022-05-05 14:36:07.000000000","uploader":{"_account_id":32927,"name":"mitya-eremeev-2","display_name":"Mitya Eremeev","email":"mitossvyaz@mail.ru","username":"mitya-eremeev-2"},"ref":"refs/changes/91/813691/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/91/813691/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/91/813691/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/91/813691/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/91/813691/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/91/813691/5"}}},"commit":{"parents":[{"commit":"8271cde1e2850ff1870b310801e3fa8c7fbbf998","subject":"[api-ref] update cache manage API info","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8271cde1e2850ff1870b310801e3fa8c7fbbf998"}]}],"author":{"name":"Mitya_Eremeev","email":"mitossvyaz@mail.ru","date":"2022-05-04 12:16:29.000000000","tz":180},"committer":{"name":"Mitya_Eremeev","email":"mitossvyaz@mail.ru","date":"2022-05-05 14:31:35.000000000","tz":180},"subject":"glance-manage can purge all deleted rows","message":"glance-manage can purge all deleted rows\n\nWhen we want to purge all deleted rows\nwe use \"glance-manage db purge\" or\n\"glance-manage purge_images_table\"\nthen we never know how many deleted rows are still in a table.\nSo we need to launch the command many times until\nthe command reports that 0 rows were deleted in every table.\nIt\u0027s inconvenient.\nThe patch introduced new valid value for \"max_rows\".\nIf \"max_rows\" equals -1 all deleted rows are purged.\n\nCloses-Bug: 1702445\nChange-Id: Ia9811d93b1c0b67a1b29a9e8ee3e0fc098c57d4a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/360dc45680df0c968994d60f87446e1a93e2d792"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/360dc45680df0c968994d60f87446e1a93e2d792"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
