)]}'
{"id":"openstack%2Fglance~813787","triplet_id":"openstack%2Fglance~master~I99ab96a7f8b6fdf6a27faf1fe0ea0328f89b154f","project":"openstack/glance","branch":"master","topic":"bug/1947146","attention_set":{},"removed_from_attention_set":{"34120":{"account":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"last_update":"2021-11-17 08:06:16.000000000","reason":"Change was submitted"},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2021-11-04 15:01:54.000000000","reason":"Cyril Roelandt replied on the change"}},"hashtags":[],"change_id":"I99ab96a7f8b6fdf6a27faf1fe0ea0328f89b154f","subject":"Add missing packages to test dependencies","status":"MERGED","created":"2021-10-13 09:20:49.000000000","updated":"2021-11-17 12:09:20.000000000","submitted":"2021-11-17 08:06:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":13,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"813787-bug/1947146","meta_rev_id":"fd818dd398a853d9e10302579810cdbf9e69ca2a","_number":813787,"virtual_id_number":813787,"owner":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-11-17 08:06:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2021-11-16 14:21:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},{"value":2,"date":"2021-11-04 15:01:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2021-11-16 14:21:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"}],"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-13 10:46:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-13 13:20:15.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"CC"},{"updated":"2021-10-13 16:05:52.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"CC"},{"updated":"2021-10-20 15:25:33.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2021-11-04 14:05:58.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-11-16 14:21:52.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"}],"messages":[{"id":"018f15cdade6b8c5bf852fdf344c0e8c53871755","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"date":"2021-10-13 09:20:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f779b36b007a4782c72e66a8e3236c82c5b78efd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-13 10:46:00.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/0da8c36ebddd451f9c17f9b537ce6230 : SUCCESS in 6m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/50a576bb842c4538831a38a932966ff3 : SUCCESS in 52m 36s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e60665f6f74341e1ad41f123d43a46c6 : SUCCESS in 1h 06m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8532af53a2ce4d0595da87d04ed84f33 : SUCCESS in 6m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/464e6bd88271449eb109cc82665aa5cd : SUCCESS in 7m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ba7433d97b0e416f981d9b6f641a237c : SUCCESS in 6m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/66da8d7bf06445b9940fb06e98a2e3d5 : SUCCESS in 7m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/05cfce1953924d169b27edf6b492851d : SUCCESS in 7m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/023536ab143c458b88d34629046cb6ed : SUCCESS in 9m 06s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4204e1ddca5f4305a37bb98f8a36bfd9 : SUCCESS in 9m 28s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/5ca3f5ae4934410e8f7ad9538ca33eef : SUCCESS in 12m 21s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/7f942f2d742c4eeab2417b970743ff95 : SUCCESS in 9m 38s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/477af0ae57de408ba3691d50ef95d1d6 : SUCCESS in 5m 41s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/4732679c1c5d4a0791af7ee940226241 : SUCCESS in 1h 01m 46s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/e5ebb1d69fba46949156b835bb5b9052 : SUCCESS in 56m 43s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/b7796d855455442c8ac7666c936502be : SUCCESS in 1h 04m 35s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/f253b145cb9b430383c98f7a48b29447 : FAILURE in 1h 04m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a3efa230aa204aabbe555b0693511db8 : SUCCESS in 48m 57s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/ff217b37b83f4cbe9a3d761616628835 : SUCCESS in 1h 20m 29s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/c71bf62295f5401ab220afb684ab00a9 : SUCCESS in 34m 11s","accounts_in_message":[],"_revision_number":1},{"id":"64173d1d859701dc5da20f9cffab2487453e668c","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2021-10-13 13:20:15.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4e2b67ea06a2103ac4d21230345a0bc431c99cef","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-10-13 16:05:52.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"de09e3ef3b4a69a52d8cc92b1a9aaa2effd10d9e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-13 17:35:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/97c2140edb344e0ca9b73de164914d2c : SUCCESS in 5m 12s\n- grenade https://zuul.opendev.org/t/openstack/build/788484d89ad148efa7119a935670a780 : SUCCESS in 43m 59s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6f69bb79dd594ced9ef6f79c28f7eab0 : SUCCESS in 1h 06m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e91f5e3091fb46e0ba142b408f9dbb61 : SUCCESS in 5m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a11753002eab4c2fb4aee8fdc9f8985c : SUCCESS in 5m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4d93da03a8ca4d15b056a2e3a68b536c : SUCCESS in 7m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1e5701e5760e4692b72b11362831680d : SUCCESS in 6m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d1659a843af4980a94431fa153d9c77 : SUCCESS in 8m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/813db1709edb46829563f583772fb363 : SUCCESS in 6m 30s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5f7876b402ce4a118e49c8bcb919b5f6 : SUCCESS in 10m 03s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/e23fdde39a8244e1b03c99e252f00be5 : SUCCESS in 8m 45s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/d2fd1a22419044d48bdf41d0786b1113 : SUCCESS in 8m 06s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/8b8143c9d8394282be0ea81d6637d316 : SUCCESS in 6m 26s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/df041b8e429c43408fc3b8a6fcb2525b : FAILURE in 39m 34s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/534ed4e871f04517883e7a0a531c88ea : SUCCESS in 1h 00m 45s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/a098a907e13249b9901fd42dd2364623 : SUCCESS in 1h 05m 15s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/7a76ee3a64e943eb8632e14595b4623a : SUCCESS in 1h 23m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cd931100431b4ce8be249d5b72a2d4b8 : SUCCESS in 46m 06s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/bbb15b9710204954aedd5c535b58bced : SUCCESS in 1h 06m 34s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/688c65ad50914aeb97facb9b12ccaa7f : SUCCESS in 35m 00s","accounts_in_message":[],"_revision_number":1},{"id":"96d0d5cc0850f4ed145ff8b585933bb24579d31b","author":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"date":"2021-10-14 11:15:07.000000000","message":"Patch Set 1: Code-Review-1 Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8fd8c21da3e6abca932219c8007ae6b170782ecb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"date":"2021-10-14 11:18:50.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"fc6f8ba79408c0bf544c6fa8c50b9eca7ba08742","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-14 13:54:18.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/c36c66c55f8b4f3399fe4bd67e9773c4 : SUCCESS in 6m 29s\n- grenade https://zuul.opendev.org/t/openstack/build/cb3aba938f1249658aab5d7ba8fd6d0b : SUCCESS in 59m 12s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1891c6a7fd7345ff903ef75c41c138a2 : SUCCESS in 1h 05m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e73c525e0776445f9a59e3b8e4676a11 : SUCCESS in 7m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/51105c73e3a04467a41b862766fc58d4 : SUCCESS in 5m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/628775b375da4519b741f9f4078f10dd : SUCCESS in 6m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2b7fa0155b5941ec8db5e6080e197899 : SUCCESS in 7m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4856ac946d449f2a05490ab99c00d04 : SUCCESS in 8m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9061870e796f4ed1ae754fc11f8d5b1b : SUCCESS in 6m 59s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/10af762158bc4acabda89fb72ee778d6 : SUCCESS in 8m 17s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/dde9cd690826481480cf0e759f5e6b9b : SUCCESS in 9m 37s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/82ba61e724d64254b1d15318909768fd : SUCCESS in 8m 41s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/e6dfb3b1a3ba4ba39848580b41f43434 : SUCCESS in 6m 22s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/9e8ea9cae50a4dd8bc1103ceb1a73a33 : SUCCESS in 1h 18m 18s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/ae1bb7ae224c4df9bae31c93a6002f23 : SUCCESS in 1h 06m 49s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/f81d48066ed94372846a597fe229e511 : SUCCESS in 1h 13m 24s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/6d4ad8fa829e48a1a418dceab505d11b : SUCCESS in 1h 26m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/560e6bef1aca4eb3be350087ee642903 : SUCCESS in 46m 30s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/dd926a382c4741b7b7a7d322cf56f3d9 : SUCCESS in 1h 38m 36s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/62d4373389154354944c967ec4aca6d0 : SUCCESS in 38m 21s","accounts_in_message":[],"_revision_number":2},{"id":"297f64502c6f2033b649b734f0410a18a360c0c2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"date":"2021-10-14 15:45:10.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"14da3b405786a7170c058b3bdcdce790d03f475f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-14 17:15:40.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/0472cd7c09274a208c985aab27147fc7 : SUCCESS in 6m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/bedd4b1b784142a4a3d4087634a684ad : SUCCESS in 58m 00s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/cd50574791ee419bb555ebc3b88c1394 : SUCCESS in 1h 20m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e8530ef45e9a46afb6bb0def57715e5d : SUCCESS in 8m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5527d17adaef46d487c0d49b17c91d11 : SUCCESS in 7m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ae91a4855e5446399e37ecd2f5b437ee : SUCCESS in 5m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c71baf6d7c7f473f98cecc1a1747f8d7 : SUCCESS in 7m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33f00bbaa5214411a4ab12e18c97a5bf : SUCCESS in 6m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4085273615804d85b85f59ded1d3e41a : SUCCESS in 7m 52s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/028fe36910f84dbfa17d8445814cb617 : SUCCESS in 13m 16s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/1733915232394373b8aaa5d07f6222dd : SUCCESS in 10m 06s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/d78b136c885b4572867e21fe67b78d9c : SUCCESS in 8m 49s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/ceaf4588268f492fa37a49e8d09784f7 : SUCCESS in 4m 49s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/df65e53cf3494695be4e92ca0e0684ef : SUCCESS in 1h 26m 15s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/30251de860884c969d85c1a17e6d752c : SUCCESS in 50m 25s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/0c15beda49e74402910212ffb0e82024 : SUCCESS in 46m 12s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/ad12cb58465c4ea39eab1e3283ccb36f : SUCCESS in 53m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ce7725bb1c914f78b931ec5a31bed202 : SUCCESS in 48m 07s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/66de5d2b1cee4e939e98a2eb24958b4d : SUCCESS in 1h 16m 04s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/4410b4a7dee643b0b822fbd88c957bf7 : SUCCESS in 41m 04s","accounts_in_message":[],"_revision_number":3},{"id":"897a9269ec810305dda985ad8c0034ab8b58169c","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2021-10-20 15:25:33.000000000","message":"Patch Set 3: Code-Review+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"5c8ea6cff684f8555b498d6aa0422aad1ab72aac","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2021-11-03 22:37:28.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"f6fff2d3bac30b733fb5446f6fe8a7199296dcd8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"date":"2021-11-04 11:47:17.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"27ee06ee179cbb5c5c6f5150a3391513820a2687","author":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"date":"2021-11-04 11:47:46.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"85444cba01aaeb84e3daee93e47bc7fb54632504","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-04 13:19:09.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/1ec55b509ced47bbb4feeaa926341297 : SUCCESS in 5m 55s\n- grenade https://zuul.opendev.org/t/openstack/build/be3925c24e8444569068cdc6883b2e46 : SUCCESS in 1h 00m 10s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/639a5b93f99f4deda2ed8ad89794b556 : SUCCESS in 1h 27m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3034ed3009d94f4a9017c553b3597670 : SUCCESS in 5m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/82cc9069dcd74aeba7e4a23e0adfb054 : SUCCESS in 6m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0a32a46142e64968991941d20a51cec6 : SUCCESS in 6m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/246e889ffd614e449b738db872ee90ba : SUCCESS in 7m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/062015afa72d44d198bc8d664e55b0ab : SUCCESS in 6m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e7c0edac03304f2b9b6deae1c60aa505 : SUCCESS in 6m 55s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/cdd1a7da05154110a4c7c730da295345 : SUCCESS in 9m 48s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/4aea5012dcfd46cdaa0522f06fcb3659 : SUCCESS in 9m 07s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/cf3680a75fd64995bf16abf2b841f03b : SUCCESS in 9m 26s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/02359a1820b840e98b3496046c88e486 : SUCCESS in 6m 05s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/d01c8e3e11f344bbbe6728fad6c5716a : SUCCESS in 1h 10m 07s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/dbd7d3858b3b4abbb7138133104a573b : SUCCESS in 1h 09m 19s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/49cbb6b2284b4166b91579696fc76530 : SUCCESS in 1h 03m 05s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/4d4bb544182f47eda11719be9a9f26c1 : SUCCESS in 1h 26m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/60b299d33afb4a149088d1dd2c0cb8f9 : SUCCESS in 38m 22s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/28f04f2b78494e3f8b7ef001d69f90e4 : SUCCESS in 1h 12m 38s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/613d5745133745e3bae1ae4593c7a082 : SUCCESS in 26m 45s","accounts_in_message":[],"_revision_number":4},{"id":"1596590791bbdc07e40693041f319a3d6a21c802","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-11-04 14:05:58.000000000","message":"Patch Set 4:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/813787/813787/4/check-not-vote/ext-glance-zuul/8510d4c : FAILURE in 1m 45s","accounts_in_message":[],"_revision_number":4},{"id":"454f6dc36cbc4bad7f5a9df88cfa9c0bf7d47f56","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2021-11-04 15:01:54.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"a0ac20fe23f7e0be018acceedca099da6e4329e7","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-11-16 14:21:52.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"1340fae79ee10fe1d6c587afbc5f0b658227bb23","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-16 14:23:11.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"90d1c5742a9fde65a08337b319ab8057d11b53d6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-17 04:55:21.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/988dfdf3ec4c459499520ee31466acdc : SUCCESS in 6m 09s\n- grenade https://zuul.opendev.org/t/openstack/build/55afaabbf27c4b22b37734e80c7d5461 : SUCCESS in 53m 20s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/462f68434d404792a39e8ff755b02515 : SUCCESS in 1h 14m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d6453761a006469db41750452fe21b78 : SUCCESS in 6m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/123e3024b775427097bf6c2e067cadfe : SUCCESS in 8m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/400f531e15db4ac68377d537988910dd : SUCCESS in 8m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d28ff068bdfe417cb56309a069e12343 : SUCCESS in 8m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ce2fdfcd6c874ece9ce43718a7979a3e : SUCCESS in 6m 42s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d8a030f2ffae44338a3b89c52834df4d : TIMED_OUT in 32m 37s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/18e9a9244f2c47c79a6ba40951bdc89f : SUCCESS in 9m 25s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/f60c706b9cb844e8be34ad1f5780d66f : SUCCESS in 1h 06m 58s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/4b31fba548ec4bba91599f6863960494 : SUCCESS in 1h 09m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d687a4dbb19d410d996bd77fff3ea966 : SUCCESS in 52m 05s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/4e6fd1e054664fba977e047e1ca0a0ed : SUCCESS in 1h 19m 28s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/e8d9471e5bd64d3fb416b28f4bd6be59 : SUCCESS in 37m 32s","accounts_in_message":[],"_revision_number":4},{"id":"e992037fce5e3a8fb39b15331ac35c7c3f2a8258","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-11-17 05:13:38.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"cac7fdf06144217fa33f872e4ae6b4028e7c8036","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-17 06:29:10.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/c5ff449eba204e2f8d11e24dd18ac40d : SUCCESS in 6m 02s\n- grenade https://zuul.opendev.org/t/openstack/build/2e00dd74b9a7478cb662fdef5a1c5074 : SUCCESS in 39m 58s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6cddbdad750c4620bec9abc4c704d6ac : SUCCESS in 1h 11m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c38b0c625f447cab400ccc25ca50974 : SUCCESS in 5m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ec1a7a6ae54f4f358a79e3c077932da1 : SUCCESS in 6m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/caa1c996059743c3b8e79b48a522b20d : SUCCESS in 6m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b252dd2031004b86b6cad9110029f51d : SUCCESS in 6m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f853e1e957e84066b10a9d0810594840 : SUCCESS in 7m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4c61ec9fdf334c71862ee802860046da : SUCCESS in 5m 25s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c88cb30cf4404afebb5fbb1b0b3e51d1 : SUCCESS in 9m 28s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/124102de23a84804aaccab9a1e104716 : SUCCESS in 9m 11s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/ddfc40ffc86944c9b0cca1fe98f44f8a : SUCCESS in 9m 23s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/6d08090f64034a03bb87f99ead5c1617 : SUCCESS in 5m 38s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/cc8cf0996d904426bf26c3c739c7af43 : SUCCESS in 1h 03m 57s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/76201f6b8e904ba18a996025f3c723ec : SUCCESS in 56m 33s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/a636bfe810d34b26a2ea4d2ca5835567 : SUCCESS in 1h 07m 36s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/136c707e20924bac9e41b7a25d520327 : SUCCESS in 1h 13m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f8314d850dc24bf08e514bd44b62efa1 : SUCCESS in 59m 08s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/db258b28ee0e41c5aecc403643428fa6 : SUCCESS in 1h 04m 47s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/fbb7875a352843dd8cc189d56339b7a9 : SUCCESS in 35m 54s","accounts_in_message":[],"_revision_number":4},{"id":"b49a38e81826076e9b6e44af9a55a2fd2b8e2875","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-17 06:29:56.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"7192dc36e740d36e4251ec0d6d6f774d29668fe2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-17 08:06:15.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/c573cb7ae9344c8daa8524e2b6504da2 : SUCCESS in 4m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/0787fa7c1fa4462682888307335ebfd7 : SUCCESS in 1h 01m 46s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b0f00ae8fdd148f192e74f11a67443b5 : SUCCESS in 1h 17m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/67ccc3910f6f4b959a148199d0fa436c : SUCCESS in 5m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c7cea5699d7145d4bd73210ea0f99942 : SUCCESS in 5m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ebae4b2eca294614820d13604abd740c : SUCCESS in 5m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b72bdadcd009406295390e10c58f7167 : SUCCESS in 6m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/916e365349604e35ba7765e32e59a775 : SUCCESS in 5m 48s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/695995f6b8aa431db1e55850084813c3 : SUCCESS in 8m 59s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/447efa705e3143b498d9649e4ebcf0db : SUCCESS in 8m 52s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/f81809a1f7644a1b8c12544bf6c8bce0 : SUCCESS in 1h 25m 13s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/1194c21e169744c5a49bffeaa8cdc7e6 : SUCCESS in 1h 23m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/02dfade9016448d68564ccb0090fa2ae : SUCCESS in 57m 48s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/f8f283a95c0b45a8ad5c17ef1746fdb7 : SUCCESS in 1h 30m 04s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/f49e6e60980d48edb12b5a1b0f6e0859 : SUCCESS in 37m 24s","accounts_in_message":[],"_revision_number":4},{"id":"93472a19f3662b24b26f3f43e62dd99074fc9d79","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-17 08:06:16.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"699abdd7c02b19e07ef0cffeae9e76e0fdd5c22d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-17 08:08:06.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/b873362290144aae9c4c3798c9be3172 : SUCCESS in 1m 11s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/efe691b83f224918adfb6b7d54eb0edd : SUCCESS in 1m 13s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e206ce74698c4da6b2105bd72931738d : SUCCESS in 59s","accounts_in_message":[],"_revision_number":4},{"id":"fd818dd398a853d9e10302579810cdbf9e69ca2a","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-11-17 12:09:20.000000000","message":"Patch Set 4:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/813787/813787/4/check-not-vote/ext-glance-zuul/8a8d1a2 : FAILURE in 2m 03s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"474c600b8afb6e1b8b7e0a2cdadd1907d42e3a04","revisions":{"d2a7fbb5b0ff84eedbd4d147b62961c942b411aa":{"kind":"REWORK","_number":1,"created":"2021-10-13 09:20:49.000000000","uploader":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"ref":"refs/changes/87/813787/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/87/813787/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/87/813787/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/87/813787/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/87/813787/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/87/813787/1"}}},"commit":{"parents":[{"commit":"9b30e7fe0e66a05d6f51a77cdcc4ef72bce92c55","subject":"Merge \"Use singular they instead of \"he/she\" where possible\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9b30e7fe0e66a05d6f51a77cdcc4ef72bce92c55"}]}],"author":{"name":"Andre Aranha","email":"afariasa@redhat.com","date":"2021-10-13 09:17:56.000000000","tz":0},"committer":{"name":"Andre Aranha","email":"afariasa@redhat.com","date":"2021-10-13 09:17:56.000000000","tz":0},"subject":"Add python3-devel to test dependencies","message":"Add python3-devel to test dependencies\n\nThe tests are failing when running on Centos due to lack of\npython3-devel package.\nThis patch is adding the missing ackage to bindep.txt, using as\nreference ironic bindep.txt, that uses python3-devel as well.\n\nChange-Id: I99ab96a7f8b6fdf6a27faf1fe0ea0328f89b154f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d2a7fbb5b0ff84eedbd4d147b62961c942b411aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d2a7fbb5b0ff84eedbd4d147b62961c942b411aa"}]},"branch":"refs/heads/master"},"701e97193e8349dcf8cfbfd802c667c61e06a3db":{"kind":"REWORK","_number":2,"created":"2021-10-14 11:18:50.000000000","uploader":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"ref":"refs/changes/87/813787/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/87/813787/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/87/813787/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/87/813787/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/87/813787/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/87/813787/2"}}},"commit":{"parents":[{"commit":"9b30e7fe0e66a05d6f51a77cdcc4ef72bce92c55","subject":"Merge \"Use singular they instead of \"he/she\" where possible\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9b30e7fe0e66a05d6f51a77cdcc4ef72bce92c55"}]}],"author":{"name":"Andre Aranha","email":"afariasa@redhat.com","date":"2021-10-13 09:17:56.000000000","tz":0},"committer":{"name":"Andre Aranha","email":"afariasa@redhat.com","date":"2021-10-14 11:17:35.000000000","tz":0},"subject":"Add qemu-img to test dependencies","message":"Add qemu-img to test dependencies\n\nThe tests are failing when running on Centos due to lack of\nqemu-img package.\nThis patch is adding the missing ackage to bindep.txt.\n\nCloses-Bug: #1947146\n\nChange-Id: I99ab96a7f8b6fdf6a27faf1fe0ea0328f89b154f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/701e97193e8349dcf8cfbfd802c667c61e06a3db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/701e97193e8349dcf8cfbfd802c667c61e06a3db"}]},"branch":"refs/heads/master"},"36fb2ab3fcad3cc1aaeecb57d3d32f8998e659df":{"kind":"REWORK","_number":3,"created":"2021-10-14 15:45:10.000000000","uploader":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"ref":"refs/changes/87/813787/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/87/813787/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/87/813787/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/87/813787/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/87/813787/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/87/813787/3"}}},"commit":{"parents":[{"commit":"9b30e7fe0e66a05d6f51a77cdcc4ef72bce92c55","subject":"Merge \"Use singular they instead of \"he/she\" where possible\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9b30e7fe0e66a05d6f51a77cdcc4ef72bce92c55"}]}],"author":{"name":"Andre Aranha","email":"afariasa@redhat.com","date":"2021-10-13 09:17:56.000000000","tz":0},"committer":{"name":"Andre Aranha","email":"afariasa@redhat.com","date":"2021-10-14 15:44:33.000000000","tz":0},"subject":"Add missing packages to test dependencies","message":"Add missing packages to test dependencies\n\nThe tests are failing when running on Centos due to lack of\nqemu-img package and python3-devel.\nThis patch is adding the missing packages to bindep.txt.\n\nCloses-Bug: #1947146\n\nChange-Id: I99ab96a7f8b6fdf6a27faf1fe0ea0328f89b154f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/36fb2ab3fcad3cc1aaeecb57d3d32f8998e659df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/36fb2ab3fcad3cc1aaeecb57d3d32f8998e659df"}]},"branch":"refs/heads/master"},"474c600b8afb6e1b8b7e0a2cdadd1907d42e3a04":{"kind":"REWORK","_number":4,"created":"2021-11-04 11:47:17.000000000","uploader":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"ref":"refs/changes/87/813787/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/87/813787/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/87/813787/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/87/813787/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/87/813787/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/87/813787/4"}}},"commit":{"parents":[{"commit":"9b30e7fe0e66a05d6f51a77cdcc4ef72bce92c55","subject":"Merge \"Use singular they instead of \"he/she\" where possible\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9b30e7fe0e66a05d6f51a77cdcc4ef72bce92c55"}]}],"author":{"name":"Andre Aranha","email":"afariasa@redhat.com","date":"2021-10-13 09:17:56.000000000","tz":0},"committer":{"name":"Andre Aranha","email":"afariasa@redhat.com","date":"2021-11-04 11:46:06.000000000","tz":0},"subject":"Add missing packages to test dependencies","message":"Add missing packages to test dependencies\n\nThe tests are failing when running on Centos due to lack of\nqemu-img package and python3-devel.\nThis patch is adding the missing packages to bindep.txt.\n\nCloses-Bug: #1947146\n\nChange-Id: I99ab96a7f8b6fdf6a27faf1fe0ea0328f89b154f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/474c600b8afb6e1b8b7e0a2cdadd1907d42e3a04"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/474c600b8afb6e1b8b7e0a2cdadd1907d42e3a04"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
