)]}'
{"id":"openstack%2Fglance~817970","triplet_id":"openstack%2Fglance~stable%2Fxena~I2026a81d3611e069fbcd1a79c2384d18cb5a653d","project":"openstack/glance","branch":"stable/xena","topic":"yoga/xena","attention_set":{},"removed_from_attention_set":{"6962":{"account":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"last_update":"2022-07-09 05:43:39.000000000","reason":"Change was submitted"},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2022-07-09 05:43:39.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I2026a81d3611e069fbcd1a79c2384d18cb5a653d","subject":"Add \"bochs\" as a possible video model","status":"MERGED","created":"2021-11-15 13:49:09.000000000","updated":"2022-07-09 05:44:47.000000000","submitted":"2022-07-09 05:43:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"817970-yoga/xena","cherry_pick_of_change":817950,"cherry_pick_of_patch_set":1,"meta_rev_id":"b3a55895aa6a179a44cdc70c836c31af9a91f394","_number":817970,"virtual_id_number":817970,"owner":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-09 05:43:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"date":"2022-05-16 10:56:56.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":2,"date":"2022-07-07 15:08:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2022-07-07 15:07:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":1,"date":"2022-07-07 15:08:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-15 13:58:42.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2021-11-15 15:46:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-18 14:15:43.000000000","updated_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"},{"updated":"2022-07-07 15:07:51.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"}],"messages":[{"id":"ffe400e54f3d26f58ffcc293181fd610ac76b327","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"date":"2021-11-15 13:49:09.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"f23e928344d9ad06b6b5f750a09a8d4bf21a5e03","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-11-15 13:58:42.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fc8f5a917b5cc3672a7375673a7ee6917e15077c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-15 15:46:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/bdf8563915c440fdb3a462f77ba13c31 : SUCCESS in 1h 02m 29s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/af30cd9eafeb48788b5174dc71d4cb4e : SUCCESS in 1h 09m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea46b4ca4f31413f82ea4de9c72fdbc5 : SUCCESS in 5m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/80ab46c20df648918b8e6bb9ace1681f : SUCCESS in 6m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/75aa73ef184a468a8317156c935c7af8 : SUCCESS in 6m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/59f617e16b8e47c98c27105a5c84bf22 : SUCCESS in 6m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82cba3c92e044690b4c8159a0812d41c : SUCCESS in 8m 00s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d3973bcb06714fe9ad22c7c6b1577d14 : SUCCESS in 8m 14s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/f5a75c097a0544a7b18e98ebdab946b2 : SUCCESS in 8m 21s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/80200e57205f45eea1a06f1f2aeb116a : SUCCESS in 8m 29s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/d13490c4b6054649b9098f88c1d6bcb2 : SUCCESS in 1h 52m 58s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/e1a52d35ce964fb3a1251931864d81b6 : SUCCESS in 1h 04m 09s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/394c104fda5e442ebe742894474d35d1 : SUCCESS in 58m 19s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/dd4457e552834365bc4da6d153ff6454 : SUCCESS in 1h 14m 01s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e074ee8e1e2747eea8188496c489d093 : SUCCESS in 1h 04m 03s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/3e666d2245b44409ad88795cc84bab22 : SUCCESS in 1h 15m 16s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/23535437314147eeadd24bb305b4c914 : SUCCESS in 34m 11s","accounts_in_message":[],"_revision_number":1},{"id":"d1a048389343deca74452e348fe7292f2f2cc52e","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2021-11-18 14:15:43.000000000","message":"Patch Set 1: Code-Review-2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d08b2e5eab75ad3dfacd43473171bcd9ce41da04","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2022-05-06 15:40:35.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e562f8277d3c15195c8f5b31a8f7a8109364551b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-06 16:55:03.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/ecfb0e022b9a42dbbb4f711bb1ec7e8b : SUCCESS in 51m 59s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/8db8e640a1174c1c8160db6ba766a838 : SUCCESS in 58m 25s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e081d42584e448f386c878630730d62e : SUCCESS in 54m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/382aeb6a9847400f9af9af45d4cae51d : SUCCESS in 46m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db539f302e83400e89203c1afafd2c6b : SUCCESS in 5m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4fcd749a40b240acb5b33208bc4c0bc8 : SUCCESS in 5m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/61f3861255034ed0935f0ae4c61b4771 : SUCCESS in 5m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9ceb1215d15f41fea5c9cc0467194640 : SUCCESS in 6m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8dc9e1dd60ad434eb6e7b79838ee4312 : SUCCESS in 6m 05s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0c8aa5b4734f424a9cfd7afd78d52013 : SUCCESS in 8m 05s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/4b516eaa9945458cae33b1b150d24c19 : SUCCESS in 7m 32s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/bb10c515341a4713a1be8911cf4ddc64 : SUCCESS in 8m 52s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/15eb174ab51a4d369817f86c31738a14 : SUCCESS in 1h 07m 06s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/c157b49a87ac4280973222b167c21d95 : SUCCESS in 1h 11m 53s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/586aaaa12080424fa69bb06995b63465 : SUCCESS in 58m 56s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/291ddfd8a1a242bdad9d530c0cbbc505 : SUCCESS in 1h 11m 58s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c6cc022e694c422b8801f472d98e848a : SUCCESS in 40m 26s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/3c44da9897cf4a0e8dba8ca61914e5f8 : SUCCESS in 1h 02m 19s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/5f6f0cb2a0ff46d4b2cd4d738452766d : SUCCESS in 32m 10s","accounts_in_message":[],"_revision_number":2},{"id":"1a1d7d4e707f5adc2ffbac491b46b52b759a26c0","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2022-05-16 10:56:56.000000000","message":"Patch Set 2: -Code-Review","accounts_in_message":[],"_revision_number":2},{"id":"ba1742dbd7000b453a1ff09ac95d212c83221490","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2022-07-07 15:07:51.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"8277688616b6fe94c79cf7ef507757b8111b8551","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-07-07 15:08:15.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f872207f2858efdeb5cc07fd87db90d8273999cd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-07 15:09:05.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"1e4fef07a0a5e6f9528b3d5c99d348300b5f3d15","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-07 17:15:14.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6d76cfd40c96445780a457d64568514f\n\n- grenade https://zuul.opendev.org/t/openstack/build/7ade512f0d4a45fc8cd2a47ee5962baf : SUCCESS in 57m 40s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/bf6127e729b74a63aa41b4cebd2cc06e : SUCCESS in 1h 16m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fc8135398f1b41f18812770049af5242 : SUCCESS in 43m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1496d9704e774c3fa734ea62bff431d8 : SUCCESS in 5m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cbd0df0278534b04873258b0b2dbedda : SUCCESS in 5m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1224bb73a5094db396c4505d0525ab91 : SUCCESS in 6m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad5608d4f265435a8b938571ae57c135 : SUCCESS in 7m 31s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7ce2b823a85840cc8e4f1d7719e2f44e : TIMED_OUT in 31m 06s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/a4a5041638d5415f9d6ed628dfb09c7b : SUCCESS in 10m 33s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/c97a1453e8f64da4ba92894fc7e8e6bc : SUCCESS in 1h 19m 59s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/6e314eae8dd84d3f9250f1f8f1610586 : SUCCESS in 1h 09m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3332c9400d7a49e49afc4a6ae7828cd1 : SUCCESS in 47m 54s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/0226628b939941cebf7e747f1aed4d99 : SUCCESS in 1h 12m 46s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/f80e49efca9340789d08cc9684a78d95 : FAILURE in 31m 27s","accounts_in_message":[],"_revision_number":2},{"id":"8b09aa613793a27f702eca33cb2d72976acd2e1a","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-07-07 17:57:37.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6cd32649f01bbd0f0483d03cfa92c910ea291825","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-07 19:19:51.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fcac4c424f6b4cfe950283609457c223\n\n- grenade https://zuul.opendev.org/t/openstack/build/e00309a58af94fcca01d295e6f5fe05e : SUCCESS in 1h 02m 31s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/196e0f45f8304933852e294a5194142d : SUCCESS in 1h 07m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a5799569087049c9a806c6ca03d4575a : SUCCESS in 43m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/195fe4511aaf4bf1930f97b49fcef13e : SUCCESS in 4m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7150a02fcd77471ab3b13f6c3a1f5f68 : SUCCESS in 5m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f7f3c4fa42d64700b512720f58da3a7d : SUCCESS in 5m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/654a27c35c4f4d3d850af68632d7efd3 : SUCCESS in 6m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae536687f6294473921e9631ab283921 : SUCCESS in 6m 38s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7c0f71cb91084b4c8b3315ebd32423ed : SUCCESS in 10m 39s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c171107ad17d4c3784ce06c193e9caa1 : SUCCESS in 8m 13s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/63edf86ed6fa4e4e848f2aa200ebf323 : SUCCESS in 9m 22s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/b96a365cec684541b0e5ee4fcabef427 : SUCCESS in 1h 20m 17s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/1db4751e47f04111975834e3580e3a88 : SUCCESS in 1h 16m 10s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/beed27041501446696a76572a1b9c607 : SUCCESS in 1h 07m 12s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/a6a64f37903b43569ddeae9c5e193b32 : SUCCESS in 1h 18m 59s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/65c2505384064ddcaafcd342b7d9acaf : SUCCESS in 43m 09s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/f72a6531dcd14ee993ca75121f9a739e : SUCCESS in 1h 06m 50s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/34f7e5a6585349179f9f7b78eb7b011d : FAILURE in 28m 12s","accounts_in_message":[],"_revision_number":2},{"id":"6c34b414b5645e7c4225013ac48595e26b21f0b2","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-07-09 03:03:59.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ce4d027aa718d97b44b946993a59c2495b13ff83","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-09 04:27:31.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6eb4a50c880c4829bac1dcc6d456517c\n\n- grenade https://zuul.opendev.org/t/openstack/build/6721a9de7c6e469f8ad514ddcbf97386 : SUCCESS in 1h 02m 35s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/63b130d4d4a74ce8885cdfaaed6bc193 : SUCCESS in 1h 09m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f77c4a55f9934e01a6eeb7416fd08975 : SUCCESS in 46m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7de271d54aca41579b75eb9ac9356eb0 : SUCCESS in 4m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d6c18f98358d4aa7bd8e02af6d9296b4 : SUCCESS in 5m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/21030960d5884784848495e34f9fe208 : SUCCESS in 4m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2b676e2edb9e4215bf2f8d3ecf0b30f2 : SUCCESS in 5m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc4d94198a7b48b1b3299e3d27765451 : SUCCESS in 5m 40s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5bf80f3a37a546ac8e290bb8ec44c4b8 : SUCCESS in 7m 32s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/474ee8e3beb54f9489655f816cd21748 : SUCCESS in 7m 23s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/873f5da02e4d43468db6fa311e7db287 : SUCCESS in 7m 11s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/c865e6354048489e8439930a81655774 : SUCCESS in 1h 18m 39s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/6c3b5f94a7c84d3caa2076ff784e1b04 : SUCCESS in 1h 09m 02s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/906081ccb66d459e8940a91aa0e75a4f : SUCCESS in 1h 05m 41s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/1021e8a91e53425eb96358b3b7261549 : SUCCESS in 1h 17m 47s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/525b80c6ece945f2a6518624dfeee1b4 : SUCCESS in 52m 06s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/d31fc78660c84592ae86870820fbea14 : SUCCESS in 1h 00m 44s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/39cc81fb2b0b4ddf898689f9d6900190 : SUCCESS in 29m 00s","accounts_in_message":[],"_revision_number":2},{"id":"f8e95426fc38ceceb8e776f072cc59e2bb14deab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-09 04:28:09.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"3eac244fa732d1f7ed47c4a5fa1367cb53af0432","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-09 05:43:39.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4086b0e618bf490c8198376b5f50429a\n\n- grenade https://zuul.opendev.org/t/openstack/build/2883863a1745465995ee15ec505c44e2 : SUCCESS in 56m 42s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d5fe455466ea469a89f8c237bd1fee33 : SUCCESS in 1h 11m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9a1687be365a4beea6d7cccb3bac4d43 : SUCCESS in 58m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/54c1eb5e20c840d3910290d358381177 : SUCCESS in 4m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d9d0e67b37044d3d8afbd774af5704f0 : SUCCESS in 5m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6609b4b7a4794d9981ae8669e79a6bd8 : SUCCESS in 4m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e516e6629050466695ebc3cef9262514 : SUCCESS in 5m 30s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b52b8814b39a4bdc9c1a34fb08e6ef87 : SUCCESS in 8m 28s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/f07aa94aca45492ea1b0099ec6b4a579 : SUCCESS in 6m 51s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/d88f4cfa9be741acbec87e0f757d7ddf : SUCCESS in 1h 08m 11s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/aad110e2dea14221a7857790fa796d8d : SUCCESS in 1h 04m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ef00976132ba43e9a0c67202ceae61a3 : SUCCESS in 45m 13s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/609b3c97c11a40a1b840e74a5208b819 : SUCCESS in 1h 10m 14s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/e5f0bfee48254609ac93c3b55bebcdbf : SUCCESS in 31m 57s","accounts_in_message":[],"_revision_number":2},{"id":"c721827ebbad0a330feceb25a26638eb5eb55baa","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-09 05:43:39.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"b3a55895aa6a179a44cdc70c836c31af9a91f394","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-09 05:44:47.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c8b2587646b74a20a36cbcaa8c53137a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6eb88358c24e47f9bbfd976e78b1b71c : SUCCESS in 49s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"965a97837a1e85bb935d0e4201922201eaf84a05","revisions":{"d78b99c959aa37a950bd71c9119abe9bcdabbc72":{"kind":"REWORK","_number":1,"created":"2021-11-15 13:49:09.000000000","uploader":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"ref":"refs/changes/70/817970/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/70/817970/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/70/817970/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/70/817970/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/70/817970/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/70/817970/1"}}},"commit":{"parents":[{"commit":"aa7c4acdade2071084a35e42b35c1ec5a8c9990f","subject":"Bump oslo.policy to version 3.8.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/aa7c4acdade2071084a35e42b35c1ec5a8c9990f"}]}],"author":{"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","date":"2021-11-15 13:42:49.000000000","tz":60},"committer":{"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","date":"2021-11-15 13:49:09.000000000","tz":0},"subject":"Add \"bochs\" as a possible video model","message":"Add \"bochs\" as a possible video model\n\nIn Xena, this commit[1] support for the \"bochs\" video model (the device\nis recommended for UEFI guest), but neglected to update Glance\ndocumentation.  Correct that now.\n\n[1] https://opendev.org/openstack/nova/commit/c590843974\n    \"Allow \u0027bochs\u0027 as a display device option\"\n\nChange-Id: I2026a81d3611e069fbcd1a79c2384d18cb5a653d\nSigned-off-by: Kashyap Chamarthy \u003ckchamart@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d78b99c959aa37a950bd71c9119abe9bcdabbc72"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d78b99c959aa37a950bd71c9119abe9bcdabbc72"}]},"branch":"refs/heads/stable/xena"},"965a97837a1e85bb935d0e4201922201eaf84a05":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":2,"created":"2022-05-06 15:40:35.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/70/817970/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/70/817970/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/70/817970/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/70/817970/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/70/817970/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/70/817970/2"}}},"commit":{"parents":[{"commit":"881b24a2ad882ba06e16acd8e5a54eeb9d30e1a5","subject":"Update qcow test to use qcow2 instead","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/881b24a2ad882ba06e16acd8e5a54eeb9d30e1a5"}]}],"author":{"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","date":"2021-11-15 13:42:49.000000000","tz":60},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2022-05-06 15:40:05.000000000","tz":120},"subject":"Add \"bochs\" as a possible video model","message":"Add \"bochs\" as a possible video model\n\nIn Xena, this commit[1] support for the \"bochs\" video model (the device\nis recommended for UEFI guest), but neglected to update Glance\ndocumentation.  Correct that now.\n\n[1] https://opendev.org/openstack/nova/commit/c590843974\n    \"Allow \u0027bochs\u0027 as a display device option\"\n\nChange-Id: I2026a81d3611e069fbcd1a79c2384d18cb5a653d\nSigned-off-by: Kashyap Chamarthy \u003ckchamart@redhat.com\u003e\n(cherry picked from commit bfc7a178e39ba9e79a6706d7c03edc736eca28dd)\nCloses-Bug: #1972002\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/965a97837a1e85bb935d0e4201922201eaf84a05"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/965a97837a1e85bb935d0e4201922201eaf84a05"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
