)]}'
{"id":"openstack%2Fglance~818966","triplet_id":"openstack%2Fglance~master~Ibd9563fb854f888b4a2f5a5dcf6abce0cfab0f04","project":"openstack/glance","branch":"master","topic":"818966","attention_set":{},"removed_from_attention_set":{"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2021-11-30 05:59:00.000000000","reason":"Abhishek Kekane replied on the change"},"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2022-05-10 20:05:32.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ibd9563fb854f888b4a2f5a5dcf6abce0cfab0f04","subject":"Fix tests for logging connection info","status":"MERGED","created":"2021-11-23 16:05:05.000000000","updated":"2022-05-10 20:06:49.000000000","submitted":"2022-05-10 20:05:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"818966-818966","meta_rev_id":"7f103cbf262408a1fa46fd2b4b103fa6abf1e45a","_number":818966,"virtual_id_number":818966,"owner":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-10 20:05:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"recommended":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"all":[{"value":1,"date":"2021-12-20 09:20:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":2,"date":"2021-11-30 05:59:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2022-05-10 19:06:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":1,"date":"2022-05-10 19:07:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"}],"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-23 16:57:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-26 09:58:42.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2021-11-26 09:58:54.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2021-11-30 05:20:28.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"CC"},{"updated":"2021-11-30 05:59:00.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2021-12-02 14:21:30.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-12-20 09:20:59.000000000","updated_by":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"reviewer":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"state":"REVIEWER"},{"updated":"2022-05-10 19:06:42.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"}],"messages":[{"id":"5462ee0f5e5a0bcf985a2d795ed7df69661184d0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2021-11-23 16:05:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f5cafe60b63f7d534e263e1e2bf21509f8e3afbd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-23 16:57:00.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c20874b7e244c3793dd5e58de2e70ad : SUCCESS in 5m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/68b950c2e373466d9bfd421e52b7ef5a : SUCCESS in 5m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2119c08f199f4cab8bfa436f74bce11a : SUCCESS in 5m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d993951c89af4713a33802aada6cd3a1 : SUCCESS in 6m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/246c5a8669bc45198a56ca33bef62f8d : SUCCESS in 6m 55s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/41ec5e7907c44bab81b85dcc895cee54 : FAILURE in 5m 14s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/223a12a76a72497fa90f65c37fa207a7 : FAILURE in 4m 51s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/f917f743a84747f0b31c289adf585002 : FAILURE in 4m 46s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/c9fdf05181484d8a816f4f7bf4e8652a : SUCCESS in 4m 56s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/8f4addc7210a47b2ae1385064e956263 : SUCCESS in 41m 43s","accounts_in_message":[],"_revision_number":1},{"id":"01ea5cd02824a69b1c865eca6f2ceffc42aae4c5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2021-11-26 09:58:27.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"59a901ce259e2ee244386c139f555dbc4591fa3e","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-11-26 09:58:42.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n\nWarning:\n  Dependency cycle detected","accounts_in_message":[],"_revision_number":2},{"id":"77d68b919606fdf9f7a3db457f914dd74c888d42","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-11-26 09:58:54.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n\nWarning:\n  Dependency cycle detected","accounts_in_message":[],"_revision_number":2},{"id":"f368ae5f086ece9677b080c4852f9d329cc3c510","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-26 09:58:59.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n\nWarning:\n  Dependency cycle detected","accounts_in_message":[],"_revision_number":2},{"id":"2660ca6684eb709ab915cfed7455bb11eb27b5a6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2021-11-29 15:32:19.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"3a0f36f9c4f67c1e3e260954da69f3f1964cb3c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-29 16:07:14.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69d1f0b574be439aa4a2e8332116a651 : SUCCESS in 5m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2f4795da59d44c71b00cc9876b7b2e8f : SUCCESS in 6m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/429825f2e0f5412a8bf5bf14b6f9db15 : SUCCESS in 6m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/15aa3e71c0894f9faf8b3fafbf73d110 : SUCCESS in 6m 55s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/831b18759c5d41e59634d95b937fdfb0 : SUCCESS in 11m 44s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/2a41a21524f148c8b88f40ad7c892818 : SUCCESS in 11m 12s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/5c0706d1ed304b5b9599a4d1ed086660 : SUCCESS in 11m 20s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/591c311b06df4aefb0198968ab8e17bb : SUCCESS in 5m 37s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/8ea59443737f4396a0fda19904460f34 : SUCCESS in 31m 15s","accounts_in_message":[],"_revision_number":3},{"id":"0fe68f40b004ac1d1e44ce9d58a6c8e595839581","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-11-30 05:20:28.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"70435d1556d139fccd389951ca52743f1f3044b3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2021-11-30 05:41:47.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"c9e63f0785aee1d48549fdcb012dd74dd9bb7644","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2021-11-30 05:42:22.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"3f8ec4aa40968b065db267dfa754cf0892ab8113","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-11-30 05:59:00.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"a61264407ce134bb16abaa5e738222b54ffaefdd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-30 06:23:47.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e26d4cba7df34ec89ac62382efc6e1d1 : SUCCESS in 4m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e9f73b85040549709b2ba0e3c81a791f : SUCCESS in 5m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7e65f14833b644fbb44c4468a5debbfb : SUCCESS in 5m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9eadbbe08fa4825a5a4bccee08cb3da : SUCCESS in 6m 03s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6715dcc1e5d1462d8846ea6bbb694a63 : SUCCESS in 11m 06s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/8bdfc9d7fda84d20b1637ab13d01a684 : SUCCESS in 7m 50s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/f8bccbfee38b469d89344a6b661a8da7 : SUCCESS in 8m 01s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/7ebdd82bdad645cbbaaf576cd60cae87 : SUCCESS in 4m 21s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/e9c14977f37a42a99cd81005277b3d06 : SUCCESS in 32m 42s","accounts_in_message":[],"_revision_number":4},{"id":"b0dc650c2b02e22cfc86acff011e878e4821b231","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-12-02 14:21:30.000000000","message":"Patch Set 4:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/818966/818966/4/check-not-vote/ext-glance-zuul/a9f940f : FAILURE in 1m 44s","accounts_in_message":[],"_revision_number":4},{"id":"b763568632686a4c098c6e4674d8d7a06c4302c5","author":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"date":"2021-12-20 09:20:59.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"86d5c4b257ed888d164b8d2516ac83ec8e382595","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2022-05-10 19:06:42.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"cf8835b97a48a1738e5e7431cb4408fe5fcbd1e7","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2022-05-10 19:07:11.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"90ff86a589882c998b10ea47fda6404e209b2191","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-10 19:08:10.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"14651a6ed9f6b278e4a89523d8c4a7907562015c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-10 20:05:31.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/841f88d6faed41ebbd84e5e47819dab6 : SUCCESS in 49m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/151a9bce22654910a5231275b57d504b : SUCCESS in 7m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aec6d552bc9f4ac48d23d6d2aa6cac6f : SUCCESS in 5m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e0a6dc6152f34379b13406e03b44097b : SUCCESS in 7m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51846d4e0f5f426594c5458900ddc6fb : SUCCESS in 7m 11s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d5159c29799b464baef8c7098f450997 : SUCCESS in 12m 56s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/50dc68777e904568a9bd28e1c3ef01fe : SUCCESS in 10m 07s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/ce3239a0660348e8b9306fbe268d81eb : SUCCESS in 26m 53s","accounts_in_message":[],"_revision_number":4},{"id":"3d4bcef3612e3ea5859c51cccc7bef02145dd76b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-10 20:05:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"7f103cbf262408a1fa46fd2b4b103fa6abf1e45a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-10 20:06:49.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7015fccb424b4b8e9787d17abe304b1f : SUCCESS in 54s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"acc6f624cdb1b2e4a3900e7d226bbc5156e8fa17","revisions":{"3780899ebc8f3b08a0f1a3a1a3e9d8d3d61d8681":{"kind":"REWORK","_number":1,"created":"2021-11-23 16:05:05.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/66/818966/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/66/818966/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/818966/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/818966/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/818966/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/66/818966/1"}}},"commit":{"parents":[{"commit":"1db6e66f93535120fdd974fce7e6490063f3c863","subject":"Merge \"Remove broken tempest-full-py3-opensuse15 job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1db6e66f93535120fdd974fce7e6490063f3c863"}]}],"author":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2021-11-23 16:01:44.000000000","tz":-300},"committer":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2021-11-23 16:05:00.000000000","tz":-300},"subject":"Fix tests for logging connection info","message":"Fix tests for logging connection info\n\nwhile logging connection info in the glance cinder store driver, we\nneed to mask passwords for which we use oslo_utils\u0027s mask_dict_password\nmethod.\nSince this is not mocked in the glance cinder tests, we encounter a failure\nwith review[1].\nThis patch adds the required mocking to pass the tests.\n\n[1] https://review.opendev.org/c/openstack/glance_store/+/818722\n\nChange-Id: Ibd9563fb854f888b4a2f5a5dcf6abce0cfab0f04\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3780899ebc8f3b08a0f1a3a1a3e9d8d3d61d8681"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3780899ebc8f3b08a0f1a3a1a3e9d8d3d61d8681"}]},"branch":"refs/heads/master"},"43ffc201c2961ceacab6907c72a29c714920878d":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-11-26 09:58:27.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/66/818966/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/66/818966/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/818966/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/818966/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/818966/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/66/818966/2"}}},"commit":{"parents":[{"commit":"1db6e66f93535120fdd974fce7e6490063f3c863","subject":"Merge \"Remove broken tempest-full-py3-opensuse15 job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1db6e66f93535120fdd974fce7e6490063f3c863"}]}],"author":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2021-11-23 16:01:44.000000000","tz":-300},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2021-11-26 09:58:27.000000000","tz":0},"subject":"Fix tests for logging connection info","message":"Fix tests for logging connection info\n\nWhile logging connection info in the glance cinder store driver, we\nneed to mask passwords for which we use oslo_utils\u0027s mask_dict_password\nmethod.\nSince this is not mocked in the glance cinder tests, we encounter a failure\nwith the change in depends on.\nThis patch adds the required mocking to pass the tests.\n\nDepends-On: https://review.opendev.org/c/openstack/glance_store/+/818722\n\nChange-Id: Ibd9563fb854f888b4a2f5a5dcf6abce0cfab0f04\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/43ffc201c2961ceacab6907c72a29c714920878d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/43ffc201c2961ceacab6907c72a29c714920878d"}]},"branch":"refs/heads/master","description":"Edit commit message"},"464c7ce503341c35979b9f8f0fa59bfc5b15f1a0":{"kind":"REWORK","_number":3,"created":"2021-11-29 15:32:19.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/66/818966/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/66/818966/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/818966/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/818966/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/818966/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/66/818966/3"}}},"commit":{"parents":[{"commit":"1db6e66f93535120fdd974fce7e6490063f3c863","subject":"Merge \"Remove broken tempest-full-py3-opensuse15 job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1db6e66f93535120fdd974fce7e6490063f3c863"}]}],"author":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2021-11-23 16:01:44.000000000","tz":-300},"committer":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2021-11-29 15:31:30.000000000","tz":-300},"subject":"Fix tests for logging connection info","message":"Fix tests for logging connection info\n\nWhile logging connection info in the glance cinder store driver, we\nneed to mask passwords for which we use oslo_utils\u0027s mask_dict_password\nmethod.\nSince this is not mocked in the glance cinder tests, we encounter a failure\nwith the change in depends on.\nThis patch adds the required mocking to pass the tests.\n\nChange-Id: Ibd9563fb854f888b4a2f5a5dcf6abce0cfab0f04\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/464c7ce503341c35979b9f8f0fa59bfc5b15f1a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/464c7ce503341c35979b9f8f0fa59bfc5b15f1a0"}]},"branch":"refs/heads/master"},"acc6f624cdb1b2e4a3900e7d226bbc5156e8fa17":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2021-11-30 05:41:47.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/66/818966/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/66/818966/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/818966/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/818966/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/818966/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/66/818966/4"}}},"commit":{"parents":[{"commit":"1db6e66f93535120fdd974fce7e6490063f3c863","subject":"Merge \"Remove broken tempest-full-py3-opensuse15 job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1db6e66f93535120fdd974fce7e6490063f3c863"}]}],"author":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2021-11-23 16:01:44.000000000","tz":-300},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2021-11-30 05:41:46.000000000","tz":0},"subject":"Fix tests for logging connection info","message":"Fix tests for logging connection info\n\nWhile logging connection info in the glance cinder store driver, we\nneed to mask passwords for which we use oslo_utils\u0027s mask_dict_password\nmethod.\nSince this is not mocked in the glance functional tests, we encounter a failure\nin cross-glance-tox-functional job in the change[1].\nThis patch adds the required mocking to pass the tests.\n\n[1] https://review.opendev.org/c/openstack/glance_store/+/818722\n\nChange-Id: Ibd9563fb854f888b4a2f5a5dcf6abce0cfab0f04\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/acc6f624cdb1b2e4a3900e7d226bbc5156e8fa17"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/acc6f624cdb1b2e4a3900e7d226bbc5156e8fa17"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
